)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~663356","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~I823534e9596b535f8371335260e53c622265fdca","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"remove-multithread","hashtags":[],"change_id":"I823534e9596b535f8371335260e53c622265fdca","subject":"Remove sleep time in favor of checking members","status":"ABANDONED","created":"2019-06-05 15:11:19.000000000","updated":"2019-06-25 12:23:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"50ca593159aad5586776968b803b234827ae047d","_number":663356,"virtual_id_number":663356,"owner":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-06-14 12:37:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-06-21 08:07:40.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-14 12:37:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1a3c5c0fe974913cb788cb99a3bb0baba0384322","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-05 15:11:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"45bd2fda486d67b9e9e899e4f97c11d554ff1859","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-05 15:31:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e990ca7974b48198474f7f4320260df3b28bcc7e","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-05 15:31:50.000000000","message":"Patch Set 2:\n\ncheck experimental","accounts_in_message":[],"_revision_number":2},{"id":"a76511a08135b8ab58b866926b19d8ee91755cd4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 18:46:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/56/663356/2/check/openstack-tox-docs/9dbc17a/html/ : SUCCESS in 5m 06s\n- openstack-tox-pep8 http://logs.openstack.org/56/663356/2/check/openstack-tox-pep8/ced90e2/ : FAILURE in 4m 52s\n- kuryr-kubernetes-tempest http://logs.openstack.org/56/663356/2/check/kuryr-kubernetes-tempest/25480c0/ : SUCCESS in 1h 18m 43s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/56/663356/2/check/kuryr-kubernetes-tempest-openshift/83ab502/ : TIMED_OUT in 3h 09m 25s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/56/663356/2/check/kuryr-kubernetes-tempest-containerized/99996d6/ : SUCCESS in 1h 31m 05s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/56/663356/2/check/kuryr-kubernetes-tempest-containerized-ovn/1408763/ : FAILURE in 1h 36m 02s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/56/663356/2/check/kuryr-kubernetes-tempest-py36/beb4086/ : SUCCESS in 1h 39m 10s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/56/663356/2/check/kuryr-kubernetes-tempest-containerized-py36/d8f8d5a/ : SUCCESS in 1h 15m 38s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/56/663356/2/check/kuryr-kubernetes-tempest-multinode-containerized/9e6d8e1/ : SUCCESS in 1h 44m 04s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/56/663356/2/check/kuryr-kubernetes-tempest-centos-7/3cc1e13/ : SUCCESS in 1h 19m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b1f0c862f9563974e56cfc5615a32cb8406517bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 19:25:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/56/663356/2/experimental/kuryr-kubernetes-tempest-containerized-openshift/c9b531c/ : FAILURE in 1h 41m 55s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/56/663356/2/experimental/kuryr-kubernetes-tempest-containerized-l2/fe128f4/ : FAILURE in 2h 00m 01s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/56/663356/2/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/c1ae39e/ : TIMED_OUT in 3h 06m 50s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/56/663356/2/experimental/kuryr-kubernetes-tempest-containerized-network-policy/a8d7629/ : SUCCESS in 1h 07m 55s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/56/663356/2/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/d5ea2fa/ : FAILURE in 34m 27s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/56/663356/2/experimental/kuryr-kubernetes-tempest-ovn/5053962/ : FAILURE in 1h 59m 14s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/56/663356/2/experimental/kuryr-kubernetes-tempest-openshift-ingress/52c1d80/ : TIMED_OUT in 3h 05m 47s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/56/663356/2/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/b6e4987/ : TIMED_OUT in 3h 03m 05s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/56/663356/2/experimental/kuryr-kubernetes-tempest-multinode-ha/51ec95c/ : TIMED_OUT in 2h 16m 04s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/56/663356/2/experimental/kuryr-kubernetes-tempest-containerized-crio/0976f4e/ : FAILURE in 28m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a24dc70cde508412a0cd52d7e1e2df33dd3df425","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-05 21:49:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"405d4b230621b8b9400b48c70006fb5be822542d","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-05 21:49:19.000000000","message":"Patch Set 3:\n\ncheck experimental","accounts_in_message":[],"_revision_number":3},{"id":"befd7567038da877e3387ec9c43c854c5d2c6b56","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 00:14:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/56/663356/3/check/openstack-tox-docs/8afd992/html/ : SUCCESS in 4m 30s\n- openstack-tox-pep8 http://logs.openstack.org/56/663356/3/check/openstack-tox-pep8/d97d9be/ : SUCCESS in 5m 58s\n- kuryr-kubernetes-tempest http://logs.openstack.org/56/663356/3/check/kuryr-kubernetes-tempest/d2a52c4/ : SUCCESS in 1h 30m 55s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/56/663356/3/check/kuryr-kubernetes-tempest-openshift/340f747/ : FAILURE in 1h 33m 26s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/56/663356/3/check/kuryr-kubernetes-tempest-containerized/9326813/ : SUCCESS in 1h 19m 39s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/56/663356/3/check/kuryr-kubernetes-tempest-containerized-ovn/b228f3c/ : FAILURE in 1h 29m 05s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/56/663356/3/check/kuryr-kubernetes-tempest-py36/8dcd627/ : SUCCESS in 1h 12m 08s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/56/663356/3/check/kuryr-kubernetes-tempest-containerized-py36/3356d58/ : SUCCESS in 1h 20m 44s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/56/663356/3/check/kuryr-kubernetes-tempest-multinode-containerized/27696a0/ : SUCCESS in 1h 59m 28s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/56/663356/3/check/kuryr-kubernetes-tempest-centos-7/6918be1/ : FAILURE in 1h 42m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b78a727fc1759260a32ec57d0c658ce4d9310c6e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 01:07:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/56/663356/3/experimental/kuryr-kubernetes-tempest-containerized-openshift/7956762/ : FAILURE in 1h 21m 40s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/56/663356/3/experimental/kuryr-kubernetes-tempest-containerized-l2/75f8a75/ : FAILURE in 2h 17m 53s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/56/663356/3/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/c75b476/ : TIMED_OUT in 3h 04m 49s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/56/663356/3/experimental/kuryr-kubernetes-tempest-containerized-network-policy/df02c2d/ : SUCCESS in 57m 06s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/56/663356/3/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/ae27b00/ : FAILURE in 2h 42m 02s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/56/663356/3/experimental/kuryr-kubernetes-tempest-ovn/6f7d616/ : FAILURE in 1h 31m 42s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/56/663356/3/experimental/kuryr-kubernetes-tempest-openshift-ingress/5d58444/ : FAILURE in 1h 04m 48s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/56/663356/3/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/0dc4f75/ : FAILURE in 1h 50m 41s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/56/663356/3/experimental/kuryr-kubernetes-tempest-multinode-ha/d8d47b8/ : TIMED_OUT in 2h 15m 28s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/56/663356/3/experimental/kuryr-kubernetes-tempest-containerized-crio/bd4dcb4/ : FAILURE in 26m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"342319f95aeda05aac4c882977f80b38a646839c","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-10 08:40:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"37127c2b984815e9b9be6a7759053e5537bf3106","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-10 08:42:03.000000000","message":"Patch Set 4:\n\ncheck experimental","accounts_in_message":[],"_revision_number":4},{"id":"b3c427cfefece72271a9ac3a7c000e75101021e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 10:47:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/56/663356/4/check/openstack-tox-docs/41c215e/html/ : SUCCESS in 5m 20s\n- openstack-tox-pep8 http://logs.openstack.org/56/663356/4/check/openstack-tox-pep8/24a9fe3/ : SUCCESS in 5m 34s\n- kuryr-kubernetes-tempest http://logs.openstack.org/56/663356/4/check/kuryr-kubernetes-tempest/8572a1e/ : SUCCESS in 1h 24m 12s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/56/663356/4/check/kuryr-kubernetes-tempest-openshift/247babe/ : FAILURE in 1h 43m 45s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/56/663356/4/check/kuryr-kubernetes-tempest-containerized/c9246cb/ : SUCCESS in 1h 23m 35s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/56/663356/4/check/kuryr-kubernetes-tempest-containerized-ovn/b14e476/ : FAILURE in 1h 35m 01s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/56/663356/4/check/kuryr-kubernetes-tempest-py36/e13e479/ : FAILURE in 1h 13m 50s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/56/663356/4/check/kuryr-kubernetes-tempest-containerized-py36/a3ff867/ : SUCCESS in 1h 22m 40s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/56/663356/4/check/kuryr-kubernetes-tempest-multinode-containerized/e4f04d4/ : SUCCESS in 1h 43m 15s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/56/663356/4/check/kuryr-kubernetes-tempest-centos-7/8c33283/ : SUCCESS in 1h 20m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3d3c5a1ce124ff8550cb49e40924f0bb75d0c91b","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-10 11:04:07.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"9759dc1e43a3ca041c5f537602ea6f4fca4ab1b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 11:48:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/56/663356/4/experimental/kuryr-kubernetes-tempest-containerized-openshift/bbc2ae1/ : FAILURE in 1h 44m 14s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/56/663356/4/experimental/kuryr-kubernetes-tempest-containerized-l2/3b929dc/ : FAILURE in 1h 41m 19s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/56/663356/4/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/f774b2f/ : TIMED_OUT in 3h 05m 36s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/56/663356/4/experimental/kuryr-kubernetes-tempest-containerized-network-policy/2e1f1b0/ : SUCCESS in 1h 03m 57s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/56/663356/4/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/7aec91a/ : FAILURE in 2h 05m 37s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/56/663356/4/experimental/kuryr-kubernetes-tempest-ovn/7c43548/ : FAILURE in 2h 01m 04s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/56/663356/4/experimental/kuryr-kubernetes-tempest-openshift-ingress/ed0260f/ : FAILURE in 1h 01m 56s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/56/663356/4/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/ee18af1/ : FAILURE in 1h 29m 08s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/56/663356/4/experimental/kuryr-kubernetes-tempest-multinode-ha/cbd73e9/ : TIMED_OUT in 2h 15m 47s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/56/663356/4/experimental/kuryr-kubernetes-tempest-containerized-crio/dcae64f/ : FAILURE in 34m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c5507995b80e24c518e0c3adca19be5d75cd837b","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-10 12:31:41.000000000","message":"Patch Set 4:\n\nseems to have help a bit and some of the errors seems unrelated. Though still many things to chase down...","accounts_in_message":[],"_revision_number":4},{"id":"ff1f26f629625a944efc43f3d6d5db08cd8a6841","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-10 12:31:48.000000000","message":"Patch Set 4:\n\ncheck experimental","accounts_in_message":[],"_revision_number":4},{"id":"98242b943ff8b40a32b19014121a76c9b3c812cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 13:16:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/56/663356/4/check/openstack-tox-docs/28a4df9/html/ : SUCCESS in 5m 24s\n- openstack-tox-pep8 http://logs.openstack.org/56/663356/4/check/openstack-tox-pep8/9be1f4c/ : SUCCESS in 6m 04s\n- kuryr-kubernetes-tempest http://logs.openstack.org/56/663356/4/check/kuryr-kubernetes-tempest/fa24e80/ : SUCCESS in 1h 27m 40s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/56/663356/4/check/kuryr-kubernetes-tempest-openshift/3eae272/ : FAILURE in 1h 46m 52s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/56/663356/4/check/kuryr-kubernetes-tempest-containerized/de2b517/ : SUCCESS in 1h 24m 53s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/56/663356/4/check/kuryr-kubernetes-tempest-containerized-ovn/5371985/ : FAILURE in 1h 31m 45s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/56/663356/4/check/kuryr-kubernetes-tempest-py36/63bc73a/ : SUCCESS in 1h 19m 51s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/56/663356/4/check/kuryr-kubernetes-tempest-containerized-py36/38f2f67/ : SUCCESS in 1h 52m 51s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/56/663356/4/check/kuryr-kubernetes-tempest-multinode-containerized/184a0b2/ : FAILURE in 1h 55m 56s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/56/663356/4/check/kuryr-kubernetes-tempest-centos-7/dd9135a/ : SUCCESS in 1h 35m 12s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2f7c4a1b47271122a2544cdda76294f6978e9421","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 15:42:29.000000000","message":"Patch Set 4:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/56/663356/4/experimental/kuryr-kubernetes-tempest-containerized-openshift/044e2c1/ : FAILURE in 2h 00m 33s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/56/663356/4/experimental/kuryr-kubernetes-tempest-containerized-l2/0b62502/ : FAILURE in 1h 44m 19s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/56/663356/4/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/b32bc49/ : TIMED_OUT in 3h 05m 02s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/56/663356/4/experimental/kuryr-kubernetes-tempest-containerized-network-policy/483b9a5/ : SUCCESS in 1h 01m 41s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/56/663356/4/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/61acaf6/ : FAILURE in 2h 26m 50s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/56/663356/4/experimental/kuryr-kubernetes-tempest-ovn/cca2e77/ : FAILURE in 1h 37m 16s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/56/663356/4/experimental/kuryr-kubernetes-tempest-openshift-ingress/ae64280/ : FAILURE in 59m 35s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/56/663356/4/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/7c452fb/ : FAILURE in 1h 40m 18s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/56/663356/4/experimental/kuryr-kubernetes-tempest-multinode-ha/b723371/ : TIMED_OUT in 2h 15m 40s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/56/663356/4/experimental/kuryr-kubernetes-tempest-containerized-crio/1a404a8/ : FAILURE in 26m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c2620bd02707fb12cc04222a8491c4136619b962","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-10 16:40:29.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d9cbe37f257af4a23961859aba8456dbca2b4ea7","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-10 16:40:48.000000000","message":"Patch Set 5:\n\ncheck experimental","accounts_in_message":[],"_revision_number":5},{"id":"c85602e7900bc6c20d0467899de429fba544abc5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 18:52:52.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/56/663356/5/check/openstack-tox-docs/a2db926/html/ : SUCCESS in 4m 00s\n- openstack-tox-pep8 http://logs.openstack.org/56/663356/5/check/openstack-tox-pep8/b3c5822/ : SUCCESS in 7m 25s\n- kuryr-kubernetes-tempest http://logs.openstack.org/56/663356/5/check/kuryr-kubernetes-tempest/7feaf85/ : SUCCESS in 1h 45m 38s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/56/663356/5/check/kuryr-kubernetes-tempest-openshift/abe09d8/ : FAILURE in 1h 33m 27s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/56/663356/5/check/kuryr-kubernetes-tempest-containerized/0388476/ : SUCCESS in 1h 30m 53s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/56/663356/5/check/kuryr-kubernetes-tempest-containerized-ovn/22a3047/ : FAILURE in 1h 40m 38s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/56/663356/5/check/kuryr-kubernetes-tempest-py36/bc98a00/ : SUCCESS in 1h 35m 21s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/56/663356/5/check/kuryr-kubernetes-tempest-containerized-py36/c95b893/ : FAILURE in 1h 47m 37s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/56/663356/5/check/kuryr-kubernetes-tempest-multinode-containerized/dcd9f57/ : SUCCESS in 1h 59m 47s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/56/663356/5/check/kuryr-kubernetes-tempest-centos-7/0a4e8c6/ : SUCCESS in 1h 42m 46s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5b553aae6ab9c62bd22322ec5c5f20134ca2981d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 19:59:08.000000000","message":"Patch Set 5:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/56/663356/5/experimental/kuryr-kubernetes-tempest-containerized-openshift/3f0986c/ : FAILURE in 1h 34m 34s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/56/663356/5/experimental/kuryr-kubernetes-tempest-containerized-l2/f7ed65e/ : FAILURE in 1h 32m 46s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/56/663356/5/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/946e4b8/ : TIMED_OUT in 3h 05m 10s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/56/663356/5/experimental/kuryr-kubernetes-tempest-containerized-network-policy/1cc5996/ : SUCCESS in 1h 09m 19s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/56/663356/5/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/c274bbd/ : FAILURE in 2h 34m 44s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/56/663356/5/experimental/kuryr-kubernetes-tempest-ovn/5a98203/ : FAILURE in 1h 33m 00s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/56/663356/5/experimental/kuryr-kubernetes-tempest-openshift-ingress/d5bdfa7/ : FAILURE in 1h 06m 58s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/56/663356/5/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/aef833d/ : FAILURE in 2h 01m 45s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/56/663356/5/experimental/kuryr-kubernetes-tempest-multinode-ha/24c4011/ : TIMED_OUT in 2h 16m 04s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/56/663356/5/experimental/kuryr-kubernetes-tempest-containerized-crio/9db2191/ : FAILURE in 27m 39s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4341ee9725b80e76acc7a2b4aeeef384d24beffc","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-11 07:07:48.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"bca3b30c7203239908cb8b89401054128186ac76","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-11 07:08:36.000000000","message":"Patch Set 6:\n\ncheck experimental","accounts_in_message":[],"_revision_number":6},{"id":"32ba7b2b5fde1a1964f549884fe711aa18f2acec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 09:08:52.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/56/663356/6/check/openstack-tox-docs/52f7336/html/ : SUCCESS in 4m 28s\n- openstack-tox-pep8 http://logs.openstack.org/56/663356/6/check/openstack-tox-pep8/fcf61c3/ : SUCCESS in 4m 32s\n- kuryr-kubernetes-tempest http://logs.openstack.org/56/663356/6/check/kuryr-kubernetes-tempest/d5c9fb0/ : SUCCESS in 1h 36m 59s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/56/663356/6/check/kuryr-kubernetes-tempest-openshift/b1a1879/ : FAILURE in 1h 49m 40s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/56/663356/6/check/kuryr-kubernetes-tempest-containerized/0f66834/ : FAILURE in 1h 33m 02s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/56/663356/6/check/kuryr-kubernetes-tempest-containerized-ovn/21a4276/ : FAILURE in 1h 41m 10s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/56/663356/6/check/kuryr-kubernetes-tempest-py36/f1b4d9b/ : SUCCESS in 1h 27m 10s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/56/663356/6/check/kuryr-kubernetes-tempest-containerized-py36/9a8394e/ : SUCCESS in 1h 26m 32s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/56/663356/6/check/kuryr-kubernetes-tempest-multinode-containerized/39c6a02/ : SUCCESS in 1h 46m 35s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/56/663356/6/check/kuryr-kubernetes-tempest-centos-7/408d1b5/ : FAILURE in 1h 39m 05s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0a06099a270aac6d109a103d242b5409fe90b4b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 10:24:37.000000000","message":"Patch Set 6:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/56/663356/6/experimental/kuryr-kubernetes-tempest-containerized-openshift/2e542c2/ : TIMED_OUT in 3h 08m 57s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/56/663356/6/experimental/kuryr-kubernetes-tempest-containerized-l2/72d4b4c/ : FAILURE in 1h 35m 15s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/56/663356/6/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/656fe8b/ : TIMED_OUT in 3h 07m 46s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/56/663356/6/experimental/kuryr-kubernetes-tempest-containerized-network-policy/947ec33/ : SUCCESS in 1h 14m 02s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/56/663356/6/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/ff9d584/ : TIMED_OUT in 3h 04m 42s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/56/663356/6/experimental/kuryr-kubernetes-tempest-ovn/5b92b59/ : FAILURE in 1h 50m 23s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/56/663356/6/experimental/kuryr-kubernetes-tempest-openshift-ingress/d0cc773/ : FAILURE in 59m 57s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/56/663356/6/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/7d2b716/ : FAILURE in 1h 52m 24s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/56/663356/6/experimental/kuryr-kubernetes-tempest-multinode-ha/1494c13/ : TIMED_OUT in 2h 15m 36s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/56/663356/6/experimental/kuryr-kubernetes-tempest-containerized-crio/c398dbd/ : FAILURE in 27m 26s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"be92c04fd45d511400325001cbab4997f2c7a244","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-13 07:09:40.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"a9650f2e46ee1e6a5aa0127edf1d758687c72d52","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-13 07:10:33.000000000","message":"Patch Set 7:\n\ncheck experimental","accounts_in_message":[],"_revision_number":7},{"id":"67a537278e100e98110778ea0edf70278f43217e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 09:22:06.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/56/663356/7/check/openstack-tox-docs/ec357ca/html/ : SUCCESS in 4m 29s\n- openstack-tox-pep8 http://logs.openstack.org/56/663356/7/check/openstack-tox-pep8/6d7c0a5/ : SUCCESS in 4m 35s\n- kuryr-kubernetes-tempest http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest/ecf9d20/ : SUCCESS in 1h 43m 11s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest-openshift/010afaf/ : FAILURE in 1h 43m 22s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest-containerized/8924034/ : SUCCESS in 1h 27m 53s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest-containerized-ovn/b8f3981/ : SUCCESS in 1h 52m 01s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest-py36/c5127a2/ : SUCCESS in 1h 33m 16s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest-containerized-py36/a550fb7/ : FAILURE in 40m 05s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest-multinode-containerized/028e82b/ : SUCCESS in 1h 41m 57s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest-centos-7/660b5bd/ : SUCCESS in 1h 38m 10s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"059545d578542b19313faac0f85a35ecc0fef723","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-13 09:30:45.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"dcb89fbfb202a07b71a099cc090fa74eac195371","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 09:38:49.000000000","message":"Patch Set 7:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":7},{"id":"e14efe24e5fd15e6cadebffebaa3b72723a01a72","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 09:38:50.000000000","message":"Patch Set 7:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":7},{"id":"642eb584b934fc9564a3c85e9c620001e60f05d3","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-13 14:15:43.000000000","message":"Patch Set 7:\n\ncheck experimental","accounts_in_message":[],"_revision_number":7},{"id":"45c269a0cefa03e368add220187769532559840f","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-13 15:54:14.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"03e727010d943f08b2e72e6082337a18281d9e27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 19:21:56.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/56/663356/7/check/openstack-tox-docs/b0ab1ed/html/ : SUCCESS in 8m 41s\n- openstack-tox-pep8 http://logs.openstack.org/56/663356/7/check/openstack-tox-pep8/12e4e04/ : SUCCESS in 4m 21s\n- kuryr-kubernetes-tempest http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest/787f631/ : FAILURE in 1h 30m 05s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest-openshift/a8a756a/ : FAILURE in 1h 26m 56s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest-containerized/ed6458f/ : SUCCESS in 1h 54m 11s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest-containerized-ovn/28bbfc8/ : SUCCESS in 1h 33m 38s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest-py36/487a1dd/ : SUCCESS in 1h 23m 48s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest-containerized-py36/2a5965f/ : SUCCESS in 1h 32m 40s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest-multinode-containerized/8c2ae02/ : SUCCESS in 1h 46m 14s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest-centos-7/41d4651/ : SUCCESS in 1h 35m 30s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"edab03127f53e15d42fc88ac481ae9eb77ee3b16","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-14 09:21:43.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"fcaeebd70e91b3aabb49cdb9f29fe7ea1b983111","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-14 09:22:07.000000000","message":"Patch Set 7:\n\ncheck experimental","accounts_in_message":[],"_revision_number":7},{"id":"60521611b93d067ead86fbf214ee59bae0886b36","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 12:37:58.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/56/663356/7/check/openstack-tox-docs/512cb32/html/ : SUCCESS in 4m 10s\n- openstack-tox-pep8 http://logs.openstack.org/56/663356/7/check/openstack-tox-pep8/1da7ae0/ : SUCCESS in 5m 14s\n- kuryr-kubernetes-tempest http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest/fa6bb13/ : SUCCESS in 1h 23m 12s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest-openshift/0f275e9/ : FAILURE in 1h 09m 04s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest-containerized/b7b7145/ : FAILURE in 1h 11m 49s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest-containerized-ovn/4671351/ : TIMED_OUT in 3h 05m 09s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest-py36/f55a00e/ : SUCCESS in 1h 20m 03s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest-containerized-py36/f8db595/ : SUCCESS in 1h 32m 04s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest-multinode-containerized/26b6a86/ : SUCCESS in 1h 44m 48s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest-centos-7/b5de80f/ : SUCCESS in 1h 55m 02s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6cf67b525474dd3ea51421b5c334830c3601d6dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 12:38:08.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/56/663356/7/check/openstack-tox-docs/512cb32/html/ : SUCCESS in 4m 10s\n- openstack-tox-pep8 http://logs.openstack.org/56/663356/7/check/openstack-tox-pep8/1da7ae0/ : SUCCESS in 5m 14s\n- kuryr-kubernetes-tempest http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest/fa6bb13/ : SUCCESS in 1h 23m 12s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest-openshift/0f275e9/ : FAILURE in 1h 09m 04s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest-containerized/b7b7145/ : FAILURE in 1h 11m 49s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest-containerized-ovn/4671351/ : TIMED_OUT in 3h 05m 09s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest-py36/f55a00e/ : SUCCESS in 1h 20m 03s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest-containerized-py36/f8db595/ : SUCCESS in 1h 32m 04s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest-multinode-containerized/26b6a86/ : SUCCESS in 1h 44m 48s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest-centos-7/b5de80f/ : SUCCESS in 1h 55m 02s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"69f140469033eaec8e8ad176c0c17a8a4b702289","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 12:38:23.000000000","message":"Patch Set 7:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-containerized-openshift/292d1f2/ : TIMED_OUT in 3h 06m 16s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-containerized-l2/56ade9b/ : FAILURE in 2h 07m 41s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/8daba4d/ : TIMED_OUT in 3h 04m 39s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-containerized-network-policy/aea9e22/ : SUCCESS in 1h 05m 12s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/8913959/ : FAILURE in 1h 45m 08s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-ovn/835dcca/ : SUCCESS in 1h 10m 32s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-openshift-ingress/507c589/ : FAILURE in 1h 04m 00s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/f666f21/ : TIMED_OUT in 3h 04m 27s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-multinode-ha/9914dc7/ : TIMED_OUT in 2h 15m 50s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-containerized-crio/61640fe/ : FAILURE in 30m 56s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5f3fc66187ff7bbe0d03caf90cedc8cca4c178aa","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-17 06:39:15.000000000","message":"Patch Set 7:\n\ncheck experimental","accounts_in_message":[],"_revision_number":7},{"id":"fa001ffd8b37c9ebfc500ac9b522ed57c02e3456","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 08:04:44.000000000","message":"Patch Set 7:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-containerized-openshift/8e4a8f5/ : FAILURE in 1h 04m 29s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-containerized-l2/90f271a/ : FAILURE in 1h 03m 11s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/1df5c12/ : FAILURE in 1h 00m 22s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-containerized-network-policy/80b08f6/ : FAILURE in 58m 11s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/8cbe1c8/ : FAILURE in 57m 19s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-ovn/d59701e/ : FAILURE in 52m 29s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-openshift-ingress/ddd462e/ : FAILURE in 50m 36s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/a934fed/ : FAILURE in 51m 11s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-multinode-ha/e48f408/ : FAILURE in 1h 19m 55s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-containerized-crio/4514355/ : FAILURE in 29m 07s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b1aae5a1db11f15344d36fb2d0353d62216b1f02","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-17 10:50:40.000000000","message":"Patch Set 7:\n\ncheck experimental","accounts_in_message":[],"_revision_number":7},{"id":"19ac4ad4acad809bd50fc21b7708d2800458d598","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 14:01:01.000000000","message":"Patch Set 7:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-containerized-openshift/b63e3ea/ : FAILURE in 1h 42m 40s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-containerized-l2/2e52ead/ : FAILURE in 1h 44m 54s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/4405a82/ : TIMED_OUT in 3h 05m 11s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-containerized-network-policy/920855e/ : SUCCESS in 58m 57s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/62341b6/ : TIMED_OUT in 3h 04m 01s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-ovn/6a16317/ : SUCCESS in 1h 35m 26s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-openshift-ingress/29cfb01/ : FAILURE in 1h 04m 21s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/17bac80/ : FAILURE in 1h 54m 41s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-multinode-ha/edb791e/ : TIMED_OUT in 2h 15m 46s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-containerized-crio/493de29/ : FAILURE in 28m 35s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ca747435af30919b7e82cb572f9a5625e14eadf1","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-18 10:17:43.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"bdc4a48828386352295d57e79d23f3feecf27f87","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 12:56:34.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/56/663356/7/check/openstack-tox-docs/b311d5e/html/ : SUCCESS in 7m 53s\n- openstack-tox-pep8 http://logs.openstack.org/56/663356/7/check/openstack-tox-pep8/70b315a/ : SUCCESS in 4m 55s\n- kuryr-kubernetes-tempest http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest/3a13d84/ : SUCCESS in 1h 28m 54s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest-openshift/f990653/ : FAILURE in 1h 37m 30s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest-containerized/2887485/ : SUCCESS in 1h 31m 03s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest-containerized-ovn/df48fe1/ : SUCCESS in 1h 26m 26s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest-py36/e04c5b4/ : SUCCESS in 1h 30m 44s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest-containerized-py36/a32257d/ : SUCCESS in 1h 34m 43s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest-multinode-containerized/ffd6990/ : SUCCESS in 2h 35m 15s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/56/663356/7/check/kuryr-kubernetes-tempest-centos-7/26106de/ : FAILURE in 24m 49s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"7d94116e96a8106f865054a68e8c71aab89d905e","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-19 06:58:06.000000000","message":"Patch Set 7:\n\ncheck experimental","accounts_in_message":[],"_revision_number":7},{"id":"cf756ff3621e419e5606047fe9c99a99e71359a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 10:25:24.000000000","message":"Patch Set 7:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-containerized-openshift/4629a87/ : FAILURE in 1h 35m 03s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-containerized-l2/d85569c/ : FAILURE in 2h 24m 39s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/a34fe5d/ : TIMED_OUT in 3h 04m 56s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-containerized-network-policy/8e4df13/ : SUCCESS in 57m 26s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/4b874c6/ : FAILURE in 2h 31m 32s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-ovn/99e388b/ : FAILURE in 1h 30m 55s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-openshift-ingress/1707927/ : FAILURE in 57m 38s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/664b483/ : TIMED_OUT in 3h 04m 31s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-multinode-ha/9b6f0e4/ : FAILURE in 1h 28m 15s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-containerized-crio/43ac998/ : FAILURE in 26m 58s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"acce04db9a236519e6b1a1614d216bf97f3f7aaa","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-19 14:58:59.000000000","message":"Patch Set 7:\n\ncheck experimental","accounts_in_message":[],"_revision_number":7},{"id":"d6ad5abbb5bea97a4d9c3d173b8f823871e9ba73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 18:11:15.000000000","message":"Patch Set 7:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-containerized-openshift/7794aca/ : FAILURE in 1h 22m 21s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-containerized-l2/ba8461c/ : FAILURE in 2h 03m 50s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/f0865ef/ : TIMED_OUT in 3h 04m 58s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-containerized-network-policy/be6b566/ : SUCCESS in 1h 17m 39s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/2c23c62/ : FAILURE in 2h 13m 04s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-ovn/d2dfe7c/ : SUCCESS in 1h 33m 42s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-openshift-ingress/ee0416c/ : FAILURE in 1h 01m 03s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/7cd3901/ : FAILURE in 1h 30m 48s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-multinode-ha/95761dd/ : TIMED_OUT in 2h 16m 07s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-containerized-crio/a8e60ba/ : FAILURE in 27m 26s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"4bdc0866dd9c859ac42ea4deb74b55f996e388b1","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-21 08:07:40.000000000","message":"Patch Set 7:\n\ncheck experimental","accounts_in_message":[],"_revision_number":7},{"id":"5bbbfcf6d91ddd92a8832770fb86db912fae0b32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 11:17:47.000000000","message":"Patch Set 7:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-containerized-openshift/bc90743/ : FAILURE in 2h 09m 02s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-containerized-l2/d3741d0/ : FAILURE in 1h 45m 04s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/f6105aa/ : TIMED_OUT in 3h 08m 13s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-containerized-network-policy/afde57c/ : SUCCESS in 1h 15m 39s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/1a9685c/ : FAILURE in 2h 34m 08s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-ovn/8a12f03/ : SUCCESS in 1h 14m 27s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-openshift-ingress/ed8a124/ : FAILURE in 1h 52m 33s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/77d0abc/ : FAILURE in 2h 23m 52s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-multinode-ha/afc1019/ : TIMED_OUT in 2h 14m 39s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/56/663356/7/experimental/kuryr-kubernetes-tempest-containerized-crio/2efed85/ : FAILURE in 26m 24s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0839d7f3dd6454b1a0456432b48a0fe6cf5df8b2","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-25 12:23:14.000000000","message":"Abandoned\n\nThis was merged into https://review.opendev.org/#/c/636329/16","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"6fe8fc5b9fe86a366fca67fcb2210e2df486e22f","revisions":{"247941baee14abffd258301f0ee11f229e2d2ec7":{"kind":"REWORK","_number":1,"created":"2019-06-05 15:11:19.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/56/663356/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/56/663356/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/663356/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/663356/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/663356/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/663356/1"}}},"commit":{"parents":[{"commit":"1cef7e4c23702778efec872711fe414c233ad9c0","subject":"Fixing service connectivity testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/1cef7e4c23702778efec872711fe414c233ad9c0"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2019-06-05 15:07:45.000000000","tz":120},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2019-06-05 15:07:45.000000000","tz":120},"subject":"Remove sleep time in favor of checking members","message":"Remove sleep time in favor of checking members\n\nThis patch removes the sleep time in favor of checking when kuryr\nwrites the information about the members of the services\n\nChange-Id: I823534e9596b535f8371335260e53c622265fdca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/247941baee14abffd258301f0ee11f229e2d2ec7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/247941baee14abffd258301f0ee11f229e2d2ec7"}]},"branch":"refs/heads/master"},"09e006b0e7c9ce9f02106943bfea4b77983de736":{"kind":"REWORK","_number":2,"created":"2019-06-05 15:31:19.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/56/663356/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/56/663356/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/663356/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/663356/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/663356/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/663356/2"}}},"commit":{"parents":[{"commit":"48050152c1ac0f5b363c8ed250b1c64c73b6d646","subject":"Fixing service connectivity testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/48050152c1ac0f5b363c8ed250b1c64c73b6d646"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2019-06-05 15:07:45.000000000","tz":120},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2019-06-05 15:30:50.000000000","tz":120},"subject":"Remove sleep time in favor of checking members","message":"Remove sleep time in favor of checking members\n\nThis patch removes the sleep time in favor of checking when kuryr\nwrites the information about the members of the services\n\nChange-Id: I823534e9596b535f8371335260e53c622265fdca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/09e006b0e7c9ce9f02106943bfea4b77983de736"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/09e006b0e7c9ce9f02106943bfea4b77983de736"}]},"branch":"refs/heads/master"},"c7f0b022ba061e10e9202fc550433324cba2a7b0":{"kind":"REWORK","_number":3,"created":"2019-06-05 21:49:04.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/56/663356/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/56/663356/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/663356/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/663356/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/663356/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/663356/3"}}},"commit":{"parents":[{"commit":"48050152c1ac0f5b363c8ed250b1c64c73b6d646","subject":"Fixing service connectivity testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/48050152c1ac0f5b363c8ed250b1c64c73b6d646"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2019-06-05 15:07:45.000000000","tz":120},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2019-06-05 21:48:49.000000000","tz":120},"subject":"Remove sleep time in favor of checking members","message":"Remove sleep time in favor of checking members\n\nThis patch removes the sleep time in favor of checking when kuryr\nwrites the information about the members of the services\n\nChange-Id: I823534e9596b535f8371335260e53c622265fdca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c7f0b022ba061e10e9202fc550433324cba2a7b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c7f0b022ba061e10e9202fc550433324cba2a7b0"}]},"branch":"refs/heads/master"},"4aa6cc6887849d2f191a0200a9f5034f8a4e5267":{"kind":"REWORK","_number":4,"created":"2019-06-10 08:40:53.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/56/663356/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/56/663356/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/663356/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/663356/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/663356/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/663356/4"}}},"commit":{"parents":[{"commit":"48050152c1ac0f5b363c8ed250b1c64c73b6d646","subject":"Fixing service connectivity testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/48050152c1ac0f5b363c8ed250b1c64c73b6d646"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2019-06-05 15:07:45.000000000","tz":120},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2019-06-10 08:40:33.000000000","tz":120},"subject":"Remove sleep time in favor of checking members","message":"Remove sleep time in favor of checking members\n\nThis patch removes the sleep time in favor of checking when kuryr\nwrites the information about the members of the services\n\nChange-Id: I823534e9596b535f8371335260e53c622265fdca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/4aa6cc6887849d2f191a0200a9f5034f8a4e5267"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/4aa6cc6887849d2f191a0200a9f5034f8a4e5267"}]},"branch":"refs/heads/master"},"ef155d96ebbddb3d53ff4840bb757dca5db61545":{"kind":"REWORK","_number":5,"created":"2019-06-10 16:40:29.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/56/663356/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/56/663356/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/663356/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/663356/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/663356/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/663356/5"}}},"commit":{"parents":[{"commit":"48050152c1ac0f5b363c8ed250b1c64c73b6d646","subject":"Fixing service connectivity testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/48050152c1ac0f5b363c8ed250b1c64c73b6d646"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2019-06-05 15:07:45.000000000","tz":120},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2019-06-10 16:40:02.000000000","tz":120},"subject":"Remove sleep time in favor of checking members","message":"Remove sleep time in favor of checking members\n\nThis patch removes the sleep time in favor of checking when kuryr\nwrites the information about the members of the services\n\nChange-Id: I823534e9596b535f8371335260e53c622265fdca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ef155d96ebbddb3d53ff4840bb757dca5db61545"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ef155d96ebbddb3d53ff4840bb757dca5db61545"}]},"branch":"refs/heads/master"},"d92ef7013ede60ada070cd7ea200d872460fb85b":{"kind":"REWORK","_number":6,"created":"2019-06-11 07:07:48.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/56/663356/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/56/663356/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/663356/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/663356/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/663356/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/663356/6"}}},"commit":{"parents":[{"commit":"48050152c1ac0f5b363c8ed250b1c64c73b6d646","subject":"Fixing service connectivity testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/48050152c1ac0f5b363c8ed250b1c64c73b6d646"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2019-06-05 15:07:45.000000000","tz":120},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2019-06-11 06:52:18.000000000","tz":120},"subject":"Remove sleep time in favor of checking members","message":"Remove sleep time in favor of checking members\n\nThis patch removes the sleep time in favor of checking when kuryr\nwrites the information about the members of the services\n\nChange-Id: I823534e9596b535f8371335260e53c622265fdca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d92ef7013ede60ada070cd7ea200d872460fb85b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d92ef7013ede60ada070cd7ea200d872460fb85b"}]},"branch":"refs/heads/master"},"6fe8fc5b9fe86a366fca67fcb2210e2df486e22f":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2019-06-13 07:09:40.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/56/663356/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/56/663356/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/663356/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/663356/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/663356/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/663356/7"}}},"commit":{"parents":[{"commit":"75df7b1384b3fe763fce4cbff8a04569e9a3efed","subject":"Fixing service connectivity testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/75df7b1384b3fe763fce4cbff8a04569e9a3efed"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2019-06-05 15:07:45.000000000","tz":120},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2019-06-13 07:08:44.000000000","tz":120},"subject":"Remove sleep time in favor of checking members","message":"Remove sleep time in favor of checking members\n\nThis patch removes the sleep time in favor of checking when kuryr\nwrites the information about the members of the services\n\nDepends-On: I35f3be8fc16d71d0d2ac01c5451d398dd631f118\n\nChange-Id: I823534e9596b535f8371335260e53c622265fdca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/6fe8fc5b9fe86a366fca67fcb2210e2df486e22f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/6fe8fc5b9fe86a366fca67fcb2210e2df486e22f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
