)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~667670","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~I1e539802426e530ba00213f02d839d5e220ad3dc","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"remove-test-something","hashtags":[],"change_id":"I1e539802426e530ba00213f02d839d5e220ad3dc","subject":"Remove test_kuryr_tempest_plugin.py","status":"MERGED","created":"2019-06-26 16:10:04.000000000","updated":"2019-07-03 08:32:42.000000000","submitted":"2019-07-03 08:32:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667670-1562142762863-2d6c1d39","meta_rev_id":"3e8326a68ee246bdb2f69df23da8aed1a3c19c8a","_number":667670,"virtual_id_number":667670,"owner":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":2,"date":"2019-07-03 08:32:42.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"recommended":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"all":[{"value":0,"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":1,"date":"2019-06-27 05:51:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-27 05:57:10.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"all":[{"value":0,"date":"2019-06-27 05:57:16.000000000","_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-03 06:41:57.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-27 05:51:00.000000000","updated_by":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"reviewer":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"state":"REVIEWER"},{"updated":"2019-06-27 05:57:16.000000000","updated_by":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"reviewer":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"state":"REVIEWER"},{"updated":"2019-07-03 06:41:57.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2019-07-03 08:32:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7ae109ca40938e691884bee5a961a3a7caf6c9c4","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-06-26 16:10:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a680073943b4ab26f1c71231959fc503ea8326ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 18:07:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/70/667670/1/check/openstack-tox-docs/58c6213/html/ : SUCCESS in 2m 49s\n- openstack-tox-pep8 http://logs.openstack.org/70/667670/1/check/openstack-tox-pep8/ba959d6/ : SUCCESS in 2m 30s\n- kuryr-kubernetes-tempest http://logs.openstack.org/70/667670/1/check/kuryr-kubernetes-tempest/1be8cac/ : SUCCESS in 1h 19m 02s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/70/667670/1/check/kuryr-kubernetes-tempest-openshift/ab99ebe/ : FAILURE in 1h 27m 22s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/70/667670/1/check/kuryr-kubernetes-tempest-containerized/9bde834/ : SUCCESS in 1h 44m 06s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints http://logs.openstack.org/70/667670/1/check/kuryr-kubernetes-tempest-containerized-lower-constraints/839cdfe/ : SUCCESS in 1h 35m 10s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/70/667670/1/check/kuryr-kubernetes-tempest-containerized-ovn/391b45c/ : SUCCESS in 1h 50m 51s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/70/667670/1/check/kuryr-kubernetes-tempest-py36/82a738f/ : SUCCESS in 1h 28m 27s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/70/667670/1/check/kuryr-kubernetes-tempest-containerized-py36/bc38747/ : SUCCESS in 1h 30m 13s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/70/667670/1/check/kuryr-kubernetes-tempest-multinode-containerized/daffff9/ : SUCCESS in 1h 47m 27s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/70/667670/1/check/kuryr-kubernetes-tempest-centos-7/d11ce2f/ : SUCCESS in 1h 35m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0c8305a669e30a7bf01e67529895f5695d87df24","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-06-27 05:51:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4111d734ac1267c8dc931c01e20418ad0cd5edb0","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-27 05:57:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3a5c043d0be48962e29d25c53184e5c4661c1b61","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-07-03 06:41:57.000000000","message":"Patch Set 1: Workflow+1\n\nmerging as it is a trivial change","accounts_in_message":[],"_revision_number":1},{"id":"0a77ee037a6cd1d747416132ae3b59d7edc7e315","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 06:42:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"06a6fe67861c67ca79c80608ca386df0bb46806c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 08:32:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b8829ab7bcd41b814f7a7a6636da93d75dabf738","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 08:32:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/70/667670/1/gate/openstack-tox-docs/86c4607/html/ : SUCCESS in 2m 55s\n- openstack-tox-pep8 http://logs.openstack.org/70/667670/1/gate/openstack-tox-pep8/4930cdf/ : SUCCESS in 3m 51s\n- kuryr-kubernetes-tempest http://logs.openstack.org/70/667670/1/gate/kuryr-kubernetes-tempest/a0ec00b/ : SUCCESS in 1h 35m 27s\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/70/667670/1/gate/kuryr-kubernetes-tempest-py36/619522a/ : SUCCESS in 1h 33m 27s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/70/667670/1/gate/kuryr-kubernetes-tempest-containerized-py36/1576417/ : SUCCESS in 1h 44m 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ee968d17998dab8919f624a0396a4764f97249fd","revisions":{"ee968d17998dab8919f624a0396a4764f97249fd":{"kind":"REWORK","_number":1,"created":"2019-06-26 16:10:04.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/70/667670/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/70/667670/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/70/667670/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/70/667670/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/70/667670/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/70/667670/1"}}},"commit":{"parents":[{"commit":"11463dda9e4feb26eac938aee31d203aea1e8675","subject":"Removing threading option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/11463dda9e4feb26eac938aee31d203aea1e8675"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2019-06-26 16:09:16.000000000","tz":120},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2019-06-26 16:09:40.000000000","tz":120},"subject":"Remove test_kuryr_tempest_plugin.py","message":"Remove test_kuryr_tempest_plugin.py\n\nI assume we don\u0027t need a noop test_something() anymore.\n\nChange-Id: I1e539802426e530ba00213f02d839d5e220ad3dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ee968d17998dab8919f624a0396a4764f97249fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ee968d17998dab8919f624a0396a4764f97249fd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
