)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~668737","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~I082355b0414ee5308418c683fb133d67ea414d3f","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"test-recreate-ns-pod","hashtags":[],"change_id":"I082355b0414ee5308418c683fb133d67ea414d3f","subject":"Test recreate namespace and pod","status":"MERGED","created":"2019-07-02 18:41:22.000000000","updated":"2019-09-09 13:46:46.000000000","submitted":"2019-09-09 13:45:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668737-1568036739690-98feb54a","meta_rev_id":"7fb971f9ca37ee2faa868c30aadd8fe0e38af737","_number":668737,"virtual_id_number":668737,"owner":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":0,"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":2,"date":"2019-09-09 13:45:39.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2019-09-09 12:09:49.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":0,"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-05 08:30:40.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":1,"date":"2019-09-09 12:09:49.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"date":"2019-09-09 07:33:00.000000000","_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":0,"date":"2019-09-09 07:32:56.000000000","_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-04 08:56:54.000000000","updated_by":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"reviewer":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"state":"REVIEWER"},{"updated":"2019-09-05 08:30:40.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2019-09-09 07:32:56.000000000","updated_by":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"reviewer":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"state":"REVIEWER"},{"updated":"2019-09-09 07:33:00.000000000","updated_by":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"reviewer":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"state":"REVIEWER"},{"updated":"2019-09-09 12:09:49.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2019-09-09 13:45:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"941cfaace21a1ad415eb8cdd861fc615d8ed7e48","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-07-02 18:41:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac448ccbc39e8bd6d6ea547196c762e4a18e7d9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 20:37:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/37/668737/1/check/openstack-tox-docs/82c19a3/html/ : SUCCESS in 2m 49s\n- openstack-tox-pep8 http://logs.openstack.org/37/668737/1/check/openstack-tox-pep8/f1b35f0/ : SUCCESS in 4m 12s\n- kuryr-kubernetes-tempest http://logs.openstack.org/37/668737/1/check/kuryr-kubernetes-tempest/1db3f33/ : SUCCESS in 1h 36m 12s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/37/668737/1/check/kuryr-kubernetes-tempest-openshift/c92fea8/ : FAILURE in 1h 54m 15s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/37/668737/1/check/kuryr-kubernetes-tempest-containerized/a1d6eff/ : SUCCESS in 1h 36m 56s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints http://logs.openstack.org/37/668737/1/check/kuryr-kubernetes-tempest-containerized-lower-constraints/f843ede/ : FAILURE in 1h 24m 22s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/37/668737/1/check/kuryr-kubernetes-tempest-containerized-ovn/36df791/ : SUCCESS in 1h 44m 43s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/37/668737/1/check/kuryr-kubernetes-tempest-py36/831893b/ : SUCCESS in 1h 30m 58s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/37/668737/1/check/kuryr-kubernetes-tempest-containerized-py36/8e4975c/ : SUCCESS in 1h 28m 57s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/37/668737/1/check/kuryr-kubernetes-tempest-multinode-containerized/6844551/ : SUCCESS in 1h 43m 09s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/37/668737/1/check/kuryr-kubernetes-tempest-centos-7/588757f/ : SUCCESS in 1h 37m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b6043b61e1a07fd66dd843ce312cce24caa01f44","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-07-03 06:29:22.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"dbb223c0b7697515b31d5037377e71e49aeda0f6","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-07-03 06:46:51.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)\n\nJust a couple of nits! Thanks for adding this case!","accounts_in_message":[],"_revision_number":1},{"id":"f06c8f894d7c8f5678f69fe39359230a0ba65660","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-07-03 09:25:30.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"016c00d14a6d998ba7c51d8541cd70044fa2dd76","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-07-03 11:56:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"23bd6ef5fb8e2ffd540e3fb2bb227c17614185e9","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-07-03 13:00:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"afd1431fe533ae20bc6b94496bc8675904ba46ba","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-07-03 13:01:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2d7fc8032b02ff6bb0f009ad1015b379afdf3068","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 14:56:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/37/668737/2/check/openstack-tox-docs/e63165f/html/ : SUCCESS in 4m 17s\n- openstack-tox-pep8 http://logs.openstack.org/37/668737/2/check/openstack-tox-pep8/7978806/ : SUCCESS in 2m 56s\n- kuryr-kubernetes-tempest http://logs.openstack.org/37/668737/2/check/kuryr-kubernetes-tempest/d686ae5/ : SUCCESS in 1h 23m 12s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/37/668737/2/check/kuryr-kubernetes-tempest-openshift/ba74ea1/ : FAILURE in 1h 29m 31s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/37/668737/2/check/kuryr-kubernetes-tempest-containerized/05a8a53/ : SUCCESS in 1h 45m 14s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints http://logs.openstack.org/37/668737/2/check/kuryr-kubernetes-tempest-containerized-lower-constraints/1db4112/ : SUCCESS in 1h 30m 01s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/37/668737/2/check/kuryr-kubernetes-tempest-containerized-ovn/79e41e4/ : SUCCESS in 1h 48m 23s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/37/668737/2/check/kuryr-kubernetes-tempest-py36/9b66a84/ : SUCCESS in 1h 34m 24s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/37/668737/2/check/kuryr-kubernetes-tempest-containerized-py36/65248a3/ : SUCCESS in 2h 54m 12s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/37/668737/2/check/kuryr-kubernetes-tempest-multinode-containerized/bfdac82/ : SUCCESS in 2h 29m 38s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/37/668737/2/check/kuryr-kubernetes-tempest-centos-7/b22c6b2/ : SUCCESS in 1h 24m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dc77c27174578f6935cc4bc1795483ff3bcb347e","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-07-04 06:30:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ba3d9c5f49a85e97304ef1690184967c2bc61f5a","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-07-04 08:10:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1d88b1ca00cc7d3a2eaa9e8be855092f35554d5c","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-07-04 08:53:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"46ab84274b78818e29233090a3cd04227c3ad11b","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-07-04 08:56:54.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d2b237acf537c1a70010d6f8936eb4efd789e8dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 12:31:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/37/668737/3/check/openstack-tox-docs/4a0dd85/html/ : SUCCESS in 2m 27s\n- openstack-tox-pep8 http://logs.openstack.org/37/668737/3/check/openstack-tox-pep8/e7fc68b/ : SUCCESS in 3m 08s\n- kuryr-kubernetes-tempest http://logs.openstack.org/37/668737/3/check/kuryr-kubernetes-tempest/a1850aa/ : SUCCESS in 1h 18m 03s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/37/668737/3/check/kuryr-kubernetes-tempest-openshift/1ea29a1/ : TIMED_OUT in 3h 05m 01s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/37/668737/3/check/kuryr-kubernetes-tempest-containerized/f45811c/ : FAILURE in 1h 54m 47s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints http://logs.openstack.org/37/668737/3/check/kuryr-kubernetes-tempest-containerized-lower-constraints/8860864/ : SUCCESS in 1h 46m 59s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/37/668737/3/check/kuryr-kubernetes-tempest-containerized-ovn/8149fe0/ : FAILURE in 1h 35m 35s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/37/668737/3/check/kuryr-kubernetes-tempest-py36/1884400/ : SUCCESS in 1h 32m 54s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/37/668737/3/check/kuryr-kubernetes-tempest-containerized-py36/50843b7/ : SUCCESS in 1h 21m 44s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/37/668737/3/check/kuryr-kubernetes-tempest-multinode-containerized/f246606/ : SUCCESS in 2h 21m 08s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/37/668737/3/check/kuryr-kubernetes-tempest-centos-7/e9d6fea/ : SUCCESS in 2h 03m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"88951a2b4c19264fc0b59eb9372a6a5f89ad2251","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-07-05 07:32:10.000000000","message":"Patch Set 3: Code-Review+2\n\nlooks good. let check experimental","accounts_in_message":[],"_revision_number":3},{"id":"20c99334d840974c60b92ae149e023d19a58bec5","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-07-05 07:32:17.000000000","message":"Patch Set 3:\n\ncheck experimental","accounts_in_message":[],"_revision_number":3},{"id":"efb8c3b2843406eefc11a5eabce080560f64790c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 10:43:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/37/668737/3/experimental/kuryr-kubernetes-tempest-containerized-openshift/2c0d7ef/ : TIMED_OUT in 3h 05m 07s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/37/668737/3/experimental/kuryr-kubernetes-tempest-containerized-l2/88ad3c1/ : FAILURE in 1h 58m 20s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/37/668737/3/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/7117dfb/ : TIMED_OUT in 3h 07m 33s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/37/668737/3/experimental/kuryr-kubernetes-tempest-containerized-network-policy/289762c/ : SUCCESS in 1h 03m 38s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/37/668737/3/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/9e47d98/ : FAILURE in 2h 54m 57s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/37/668737/3/experimental/kuryr-kubernetes-tempest-ovn/fa4bfc4/ : SUCCESS in 1h 21m 08s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/37/668737/3/experimental/kuryr-kubernetes-tempest-openshift-ingress/f302124/ : FAILURE in 1h 29m 14s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/37/668737/3/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/faaa03a/ : FAILURE in 1h 56m 08s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/37/668737/3/experimental/kuryr-kubernetes-tempest-multinode-ha/03249f4/ : TIMED_OUT in 2h 16m 08s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/37/668737/3/experimental/kuryr-kubernetes-tempest-containerized-crio/7d3967f/ : FAILURE in 30m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a1f9f157b0435d6a53fbb3b84103a56a359cf3a0","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-07-17 08:22:14.000000000","message":"Patch Set 3: Code-Review+1\n\nrechecking as none of the gates using it worked","accounts_in_message":[],"_revision_number":3},{"id":"e7ceeffd144620a836613553a15a8dd550d016e5","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-07-17 08:22:18.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5e12962bd3fec94cc11c492725d5361638898831","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-07-17 08:22:23.000000000","message":"Patch Set 3:\n\ncheck experimental","accounts_in_message":[],"_revision_number":3},{"id":"4b5f92c4a8b49136fb21b96bb4e641f41c91e9eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 10:15:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/37/668737/3/check/openstack-tox-docs/2bd76ce/html/ : SUCCESS in 2m 33s\n- openstack-tox-pep8 http://logs.openstack.org/37/668737/3/check/openstack-tox-pep8/79fd6ab/ : SUCCESS in 3m 18s\n- kuryr-kubernetes-tempest http://logs.openstack.org/37/668737/3/check/kuryr-kubernetes-tempest/20115f9/ : SUCCESS in 1h 28m 30s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/37/668737/3/check/kuryr-kubernetes-tempest-openshift/1cb37d2/ : FAILURE in 1h 24m 21s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/37/668737/3/check/kuryr-kubernetes-tempest-containerized/a5101d7/ : SUCCESS in 1h 50m 44s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints http://logs.openstack.org/37/668737/3/check/kuryr-kubernetes-tempest-containerized-lower-constraints/a16dfdf/ : SUCCESS in 1h 46m 22s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/37/668737/3/check/kuryr-kubernetes-tempest-containerized-ovn/b06af11/ : SUCCESS in 1h 38m 15s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/37/668737/3/check/kuryr-kubernetes-tempest-py36/1918066/ : SUCCESS in 1h 20m 38s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/37/668737/3/check/kuryr-kubernetes-tempest-containerized-py36/ae218e3/ : FAILURE in 42m 29s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/37/668737/3/check/kuryr-kubernetes-tempest-multinode-containerized/7f25e89/ : SUCCESS in 1h 37m 09s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/37/668737/3/check/kuryr-kubernetes-tempest-centos-7/1fba02e/ : SUCCESS in 1h 36m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2524d1de14ed99c94a25fa627e7c707f3f21ee61","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 11:35:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/37/668737/3/experimental/kuryr-kubernetes-tempest-containerized-openshift/f30d1bd/ : FAILURE in 2h 07m 41s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/37/668737/3/experimental/kuryr-kubernetes-tempest-containerized-l2/932bff5/ : FAILURE in 1h 39m 43s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/37/668737/3/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/46d35b8/ : FAILURE in 1h 18m 28s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/37/668737/3/experimental/kuryr-kubernetes-tempest-containerized-network-policy/be3ba26/ : SUCCESS in 1h 05m 08s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/37/668737/3/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/259aa92/ : TIMED_OUT in 3h 06m 02s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/37/668737/3/experimental/kuryr-kubernetes-tempest-ovn/c47ef5d/ : SUCCESS in 1h 25m 14s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/37/668737/3/experimental/kuryr-kubernetes-tempest-openshift-ingress/06cdb17/ : TIMED_OUT in 3h 05m 52s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/37/668737/3/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/cda0637/ : FAILURE in 1h 50m 35s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/37/668737/3/experimental/kuryr-kubernetes-tempest-multinode-ha/dfde0c0/ : TIMED_OUT in 2h 17m 05s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/37/668737/3/experimental/kuryr-kubernetes-tempest-containerized-crio/0257066/ : FAILURE in 31m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0f3c1c2f66fe87490c1baf415f46932e08113c8a","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-07-31 08:54:45.000000000","message":"Patch Set 3:\n\ncheck experimental","accounts_in_message":[],"_revision_number":3},{"id":"dd48926070fbb6b0595fb1a52d605c520a184dde","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 11:44:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/37/668737/3/experimental/kuryr-kubernetes-tempest-containerized-openshift/901021d/ : FAILURE in 1h 39m 45s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/37/668737/3/experimental/kuryr-kubernetes-tempest-containerized-l2/032ba08/ : FAILURE in 1h 38m 36s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/37/668737/3/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/9e66b44/ : FAILURE in 1h 58m 20s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/37/668737/3/experimental/kuryr-kubernetes-tempest-containerized-network-policy/76e8053/ : SUCCESS in 59m 55s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/37/668737/3/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/6876f08/ : FAILURE in 2h 43m 31s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/37/668737/3/experimental/kuryr-kubernetes-tempest-ovn/aad7dab/ : SUCCESS in 1h 37m 24s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/37/668737/3/experimental/kuryr-kubernetes-tempest-openshift-ingress/830f639/ : FAILURE in 1h 24m 21s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/37/668737/3/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/32b6b38/ : FAILURE in 1h 36m 36s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/37/668737/3/experimental/kuryr-kubernetes-tempest-multinode-ha/1800e49/ : TIMED_OUT in 2h 15m 59s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/37/668737/3/experimental/kuryr-kubernetes-tempest-containerized-crio/53615c4/ : FAILURE in 36m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a271f231a7eb1e0a902ac57e7fe4d80ceed82bf0","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-08-12 07:14:35.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"3105b06ca8cf6b5b14ccf94273e80e928cb9e64e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 09:12:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/37/668737/3/check/openstack-tox-docs/1275832/html/ : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://logs.opendev.org/37/668737/3/check/openstack-tox-pep8/9903979/ : SUCCESS in 3m 13s\n- kuryr-kubernetes-tempest https://logs.opendev.org/37/668737/3/check/kuryr-kubernetes-tempest/17d6b95/ : SUCCESS in 1h 23m 31s\n- kuryr-kubernetes-tempest-openshift https://logs.opendev.org/37/668737/3/check/kuryr-kubernetes-tempest-openshift/c5a9654/ : FAILURE in 1h 37m 09s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://logs.opendev.org/37/668737/3/check/kuryr-kubernetes-tempest-containerized/ed663f0/ : SUCCESS in 1h 39m 48s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://logs.opendev.org/37/668737/3/check/kuryr-kubernetes-tempest-containerized-lower-constraints/0af60f8/ : SUCCESS in 1h 25m 57s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://logs.opendev.org/37/668737/3/check/kuryr-kubernetes-tempest-containerized-ovn/cd99b83/ : SUCCESS in 1h 35m 47s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://logs.opendev.org/37/668737/3/check/kuryr-kubernetes-tempest-py36/5dfda63/ : SUCCESS in 1h 20m 40s\n- kuryr-kubernetes-tempest-containerized-py36 https://logs.opendev.org/37/668737/3/check/kuryr-kubernetes-tempest-containerized-py36/059a35f/ : SUCCESS in 1h 35m 51s\n- kuryr-kubernetes-tempest-multinode-containerized https://logs.opendev.org/37/668737/3/check/kuryr-kubernetes-tempest-multinode-containerized/f759dbc/ : SUCCESS in 1h 51m 19s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://logs.opendev.org/37/668737/3/check/kuryr-kubernetes-tempest-centos-7/5840575/ : SUCCESS in 1h 25m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"aef6d5df4a4a40523a6553925e370d90247858e8","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-08-12 09:23:08.000000000","message":"Patch Set 3:\n\ncheck experimental","accounts_in_message":[],"_revision_number":3},{"id":"3a4c9e66d9b5269a4ff589f0e85b9dd321cc4d3e","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-08-12 09:24:22.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f9e1f65b0d92f12f83a5468195dc641a65af81af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 12:10:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://logs.opendev.org/37/668737/3/experimental/kuryr-kubernetes-tempest-containerized-openshift/bf73b45/ : FAILURE in 1h 54m 49s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://logs.opendev.org/37/668737/3/experimental/kuryr-kubernetes-tempest-containerized-l2/3dff905/ : FAILURE in 2h 19m 35s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://logs.opendev.org/37/668737/3/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/11c2ebb/ : FAILURE in 2h 10m 06s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://logs.opendev.org/37/668737/3/experimental/kuryr-kubernetes-tempest-containerized-network-policy/5e8bda1/ : SUCCESS in 52m 23s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://logs.opendev.org/37/668737/3/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/4557ccf/ : FAILURE in 2h 19m 45s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://logs.opendev.org/37/668737/3/experimental/kuryr-kubernetes-tempest-ovn/7eab35b/ : SUCCESS in 1h 27m 52s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://logs.opendev.org/37/668737/3/experimental/kuryr-kubernetes-tempest-openshift-ingress/8f3f1e2/ : FAILURE in 2h 18m 54s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://logs.opendev.org/37/668737/3/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/590902c/ : FAILURE in 2h 36m 59s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://logs.opendev.org/37/668737/3/experimental/kuryr-kubernetes-tempest-multinode-ha/71d3164/ : TIMED_OUT in 2h 16m 21s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://logs.opendev.org/37/668737/3/experimental/kuryr-kubernetes-tempest-containerized-crio/967196d/ : FAILURE in 27m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"40d1848f8135a03b88804fdfc8bdaac4607a47e5","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-08-12 12:23:45.000000000","message":"Patch Set 3:\n\nI\u0027m not sure there is a single gate passing the new test...Lets recheck","accounts_in_message":[],"_revision_number":3},{"id":"78116d71c4c924c4b4eb809a3ac04a99be2f69c9","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-08-12 12:23:54.000000000","message":"Patch Set 3:\n\ncheck experimental","accounts_in_message":[],"_revision_number":3},{"id":"e5acef71d32e9a55c73148867bc2a52e54239d2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 15:13:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://logs.opendev.org/37/668737/3/experimental/kuryr-kubernetes-tempest-containerized-openshift/34388b4/ : FAILURE in 1h 41m 45s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://logs.opendev.org/37/668737/3/experimental/kuryr-kubernetes-tempest-containerized-l2/7ed8efc/ : FAILURE in 1h 56m 42s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://logs.opendev.org/37/668737/3/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/9a18866/ : FAILURE in 2h 08m 09s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://logs.opendev.org/37/668737/3/experimental/kuryr-kubernetes-tempest-containerized-network-policy/7062ac3/ : SUCCESS in 55m 57s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://logs.opendev.org/37/668737/3/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/6a78031/ : FAILURE in 2h 41m 53s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://logs.opendev.org/37/668737/3/experimental/kuryr-kubernetes-tempest-ovn/f90ed23/ : SUCCESS in 1h 28m 40s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://logs.opendev.org/37/668737/3/experimental/kuryr-kubernetes-tempest-openshift-ingress/217b0b2/ : FAILURE in 1h 22m 18s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://logs.opendev.org/37/668737/3/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/be5ddea/ : FAILURE in 1h 49m 36s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://logs.opendev.org/37/668737/3/experimental/kuryr-kubernetes-tempest-multinode-ha/6f6de7c/ : TIMED_OUT in 2h 16m 08s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://logs.opendev.org/37/668737/3/experimental/kuryr-kubernetes-tempest-containerized-crio/37dd2e9/ : FAILURE in 28m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"76fa1aa7562652328f99aca43aa2fd146cf5299e","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-08-12 15:21:36.000000000","message":"Patch Set 3: Code-Review-1\n\nseems the problem is real","accounts_in_message":[],"_revision_number":3},{"id":"23e685d7b697aa07d45dbd7b019099b80e18e867","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-08-26 08:34:21.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"484a1275fb6a9436d5d7aee11c59b409a27cbba3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 10:49:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/207fd42e7c9e45d1a792b603d56c8aea : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fec04833acf54abb8de9a7a2b51f45f5 : SUCCESS in 5m 18s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/6c9a2055fe32417d84261576a3b50331 : SUCCESS in 1h 02m 36s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/7fd66303139d4f708e300da7eae7047d : FAILURE in 2h 04m 37s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/afc22ce03e794079b696fcb407afbc64 : SUCCESS in 1h 30m 42s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/00bc09a729c64b84a2b131bf00056ff0 : POST_FAILURE in 1h 30m 24s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/0bb4a697ff4643eba8595957744b49ab : SUCCESS in 1h 34m 22s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/7a59a1e0ad09432794f60993683fad1d : FAILURE in 1h 39m 03s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/be2f541919a844f09312e0f15cf103f3 : SUCCESS in 1h 26m 20s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/43c521df846c4937bd3c481780cd9496 : SUCCESS in 1h 39m 36s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/144eb756b6754d53b6eda916ceb35d17 : SUCCESS in 1h 34m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"da756b55c52e67beccf8d7636a068ce3884fd770","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-08-28 08:36:02.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0db3ccc6befbc3e6020b708feaafacb4f9594972","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 10:59:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82493e4a1d1d4e3ba0812426cb161a1c : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e794675682e436789d9c50af2c69b7b : SUCCESS in 4m 03s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/221f753c3b2946df9bbb0c8a7c2723c7 : SUCCESS in 1h 19m 22s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/f83ff67d335e42a68495e03e579a52e4 : FAILURE in 1h 29m 24s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/bae3f8f8ea294415afaaff01e0bd281c : SUCCESS in 1h 34m 36s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/7f22a50a8f8b4a68b107a12c24502f42 : SUCCESS in 2h 07m 18s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/30618fd8bf4e41c2a8cd43dbd9794be6 : SUCCESS in 1h 17m 55s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/4013b22e234d4db0a75068a898ceb8b1 : SUCCESS in 1h 12m 25s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/fcc9aa86dcdb48c58a5a178ebbc96ce3 : SUCCESS in 1h 31m 17s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/200b839186e445a3865aca2a3bd72afc : SUCCESS in 2h 16m 34s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/1dac14a488af40f691df7ccc681b791f : SUCCESS in 1h 31m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8b75e76d2cd92199d7e9587596444632506938c0","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-08-28 11:25:08.000000000","message":"Patch Set 4:\n\ncheck experimental","accounts_in_message":[],"_revision_number":4},{"id":"1a8892bb3cae72b0947c5799883dce6b6a9c5f19","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-08-28 11:29:03.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"f43c3688380fe87d8eab65b400fbe46056dbc2af","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-08-28 14:04:11.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)\n\nI\u0027ve checked and test is passing fine. Some small improvements listed inline.","accounts_in_message":[],"_revision_number":4},{"id":"a29b9662907fa26cb2378095339518e532249b83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 14:38:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/0a3c9d552dcd4297bd7ae9c02234a360 : TIMED_OUT in 3h 06m 40s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/87aaca8cec224123ab5a0523033642da : FAILURE in 1h 38m 45s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/56eab8bdd4d14a348ea6d5078e9f2ad8 : FAILURE in 1h 42m 13s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/88c0d472a7fa48e49edb03ae75c3df0b : SUCCESS in 50m 26s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/488902b297674c10a62afc8de11454e8 : FAILURE in 2h 15m 10s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/08bb22188b034a27b4890ed4bf13a629 : SUCCESS in 1h 16m 16s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/b225fc43586443029f5d67e1dffb3338 : FAILURE in 1h 48m 26s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/ea83feac39f94d57969d2376e95b2d66 : FAILURE in 1h 33m 58s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/b34c2287f9274e89b68569cbad3c46f2 : TIMED_OUT in 2h 19m 40s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/a78170dc923c446e8526dd783fd30177 : FAILURE in 26m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"18a1e2389e349747c323bd2c07b5a2e340953da0","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-09-05 08:16:04.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c85cdbb236f649dc636218c13b60d4c6c5c2ff18","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-09-05 08:30:22.000000000","message":"Patch Set 5:\n\ncheck experimental","accounts_in_message":[],"_revision_number":5},{"id":"f65c5f53d04c2ff6d69a250fa486cd08a30f16ec","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-09-05 08:30:40.000000000","message":"Patch Set 5: Code-Review+2\n\nlet see the experimental gate results!","accounts_in_message":[],"_revision_number":5},{"id":"99ec0eb0a167b83d075d536f10d540ac813465b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 11:39:47.000000000","message":"Patch Set 5:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/388f109a181040ef815bb944a0ab664b : FAILURE in 1h 34m 09s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/8dd7b3c48ea74b6db9d83e4db0bf4cab : FAILURE in 1h 55m 00s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/c2c2f2b4763544d484c8300b87c15f53 : FAILURE in 2h 08m 52s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/a885a73f134b4fe7880d441a469f5f7c : SUCCESS in 59m 25s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/08ff44f3b4154564a460fd51f0864344 : FAILURE in 2h 27m 18s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/19da255d9e4d4afba2e8c1734f40adaa : RETRY_LIMIT in 50m 47s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/8db84445a5b941f3a71601baccba3b87 : FAILURE in 3h 00m 27s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/cdffe4233b3f4b4f9bf8eb522eb10a4e : FAILURE in 1h 47m 49s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/1bec2792851d41aea4560f5ef7ea0ebe : TIMED_OUT in 2h 22m 55s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/934294412bc84ef29aba2dab1160f8a5 : FAILURE in 27m 14s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6e91ad3889bfaf103bc5b0c3bb68e15acbf1e1c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 11:42:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ed70c01d7d7499382bcf6872dcc5454 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22ccad9ac5a1442dbda56ebee5d42619 : SUCCESS in 2m 59s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/065ed9f1f45146fbb698e7bdc5b61fd2 : SUCCESS in 1h 20m 36s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/091a934c1ec44cb597cf0742f4137c33 : FAILURE in 1h 31m 14s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/e6acd513fec14c07ab8a0ebfdcc9ce3b : FAILURE in 37m 36s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/41f07625c3844de484b2d99939b9316f : SUCCESS in 1h 40m 02s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/7fee848f74bf40428e4d4813b2c01705 : RETRY_LIMIT in 58m 05s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/bb18fd41512f47bbb0ec8a4575eae2e4 : SUCCESS in 1h 14m 18s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/11ad6e37c3ca4f5a86c4e0bfec521d1e : SUCCESS in 1h 22m 08s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/f8697a82e7f34732850d783867eb1f8e : FAILURE in 1h 28m 45s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/25d2effd3db74fd1893d29844ed59516 : SUCCESS in 1h 28m 37s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e370da35104ede209e9e845b2f6138b78caf5b28","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-09-09 12:09:49.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"86ea6ada9294dbaa2811e9a993c8f40ee4da93d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 12:11:07.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"e2916982930f4ce5c26951013f09e896d60e8969","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 13:45:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"a550014bfcf21931b3ef65f635fc74ceca2102c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 13:45:39.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d44191104c04bb291e58eef1d818816 : SUCCESS in 2m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eefba2537a824568a474b9bfe815469a : SUCCESS in 3m 04s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/885eab0219aa4d199fcf04de30527025 : SUCCESS in 1h 11m 48s\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/97b65a097d4a451f88341d7b78d44fd6 : SUCCESS in 1h 14m 12s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/e938844012964572b1ea2b448080d211 : SUCCESS in 1h 27m 44s","accounts_in_message":[],"_revision_number":5},{"id":"34ed0f4c8eec9843907edefc124f5ca1257fb7d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 13:46:46.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a091e986dd64c48900918e4c621d272 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"3d82e550506faf633b404935017f4b10457886aa","revisions":{"77717c542210c45891e8c220a80d02852baaae25":{"kind":"REWORK","_number":1,"created":"2019-07-02 18:41:22.000000000","uploader":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"ref":"refs/changes/37/668737/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/37/668737/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/37/668737/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/37/668737/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/37/668737/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/37/668737/1"}}},"commit":{"parents":[{"commit":"47d978c439e9b8ea448ae74f3144af2e2389c0c5","subject":"Merge \"Ensure Kuryr CRDs are validated\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/47d978c439e9b8ea448ae74f3144af2e2389c0c5"}]}],"author":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2019-07-02 18:37:14.000000000","tz":0},"committer":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2019-07-02 18:37:14.000000000","tz":0},"subject":"Test recreate namespace and pod","message":"Test recreate namespace and pod\n\nThis commit adds test to the recreation of a namespace\nand a pod without waiting for Kuryr generated resources,\nwhich is the actual behavior expected by the user.\n\nChange-Id: I082355b0414ee5308418c683fb133d67ea414d3f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/77717c542210c45891e8c220a80d02852baaae25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/77717c542210c45891e8c220a80d02852baaae25"}]},"branch":"refs/heads/master"},"79ca858523c9b56e5ba37b93af9e4a3dfbf296b7":{"kind":"REWORK","_number":2,"created":"2019-07-03 11:56:16.000000000","uploader":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"ref":"refs/changes/37/668737/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/37/668737/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/37/668737/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/37/668737/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/37/668737/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/37/668737/2"}}},"commit":{"parents":[{"commit":"47d978c439e9b8ea448ae74f3144af2e2389c0c5","subject":"Merge \"Ensure Kuryr CRDs are validated\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/47d978c439e9b8ea448ae74f3144af2e2389c0c5"}]}],"author":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2019-07-02 18:37:14.000000000","tz":0},"committer":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2019-07-03 11:56:04.000000000","tz":0},"subject":"Test recreate namespace and pod","message":"Test recreate namespace and pod\n\nThis commit adds test to the recreation of a namespace\nand a pod without waiting for Kuryr generated resources,\nwhich is the actual behavior expected by the user.\n\nChange-Id: I082355b0414ee5308418c683fb133d67ea414d3f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/79ca858523c9b56e5ba37b93af9e4a3dfbf296b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/79ca858523c9b56e5ba37b93af9e4a3dfbf296b7"}]},"branch":"refs/heads/master"},"7e5b6e60ba66340b9e276b180c0f58f0fea4fa93":{"kind":"REWORK","_number":3,"created":"2019-07-04 08:53:16.000000000","uploader":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"ref":"refs/changes/37/668737/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/37/668737/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/37/668737/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/37/668737/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/37/668737/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/37/668737/3"}}},"commit":{"parents":[{"commit":"47d978c439e9b8ea448ae74f3144af2e2389c0c5","subject":"Merge \"Ensure Kuryr CRDs are validated\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/47d978c439e9b8ea448ae74f3144af2e2389c0c5"}]}],"author":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2019-07-02 18:37:14.000000000","tz":0},"committer":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2019-07-04 08:53:05.000000000","tz":0},"subject":"Test recreate namespace and pod","message":"Test recreate namespace and pod\n\nThis commit adds test to the recreation of a namespace\nand a pod without waiting for Kuryr generated resources,\nwhich is the actual behavior expected by the user.\n\nChange-Id: I082355b0414ee5308418c683fb133d67ea414d3f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/7e5b6e60ba66340b9e276b180c0f58f0fea4fa93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/7e5b6e60ba66340b9e276b180c0f58f0fea4fa93"}]},"branch":"refs/heads/master"},"b5c433950dd5b536e2472086815ab1312398014d":{"kind":"REWORK","_number":4,"created":"2019-08-28 08:36:02.000000000","uploader":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"ref":"refs/changes/37/668737/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/37/668737/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/37/668737/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/37/668737/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/37/668737/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/37/668737/4"}}},"commit":{"parents":[{"commit":"47d978c439e9b8ea448ae74f3144af2e2389c0c5","subject":"Merge \"Ensure Kuryr CRDs are validated\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/47d978c439e9b8ea448ae74f3144af2e2389c0c5"}]}],"author":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2019-07-02 18:37:14.000000000","tz":0},"committer":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2019-08-28 08:35:50.000000000","tz":0},"subject":"Test recreate namespace and pod","message":"Test recreate namespace and pod\n\nThis commit adds test to the recreation of a namespace\nand a pod without waiting for Kuryr generated resources,\nwhich is the actual behavior expected by the user.\n\nChange-Id: I082355b0414ee5308418c683fb133d67ea414d3f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/b5c433950dd5b536e2472086815ab1312398014d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/b5c433950dd5b536e2472086815ab1312398014d"}]},"branch":"refs/heads/master"},"3d82e550506faf633b404935017f4b10457886aa":{"kind":"REWORK","_number":5,"created":"2019-09-05 08:16:04.000000000","uploader":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"ref":"refs/changes/37/668737/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/37/668737/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/37/668737/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/37/668737/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/37/668737/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/37/668737/5"}}},"commit":{"parents":[{"commit":"47d978c439e9b8ea448ae74f3144af2e2389c0c5","subject":"Merge \"Ensure Kuryr CRDs are validated\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/47d978c439e9b8ea448ae74f3144af2e2389c0c5"}]}],"author":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2019-07-02 18:37:14.000000000","tz":0},"committer":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2019-09-05 08:15:51.000000000","tz":0},"subject":"Test recreate namespace and pod","message":"Test recreate namespace and pod\n\nThis commit adds test to the recreation of a namespace\nand a pod without waiting for Kuryr generated resources,\nwhich is the actual behavior expected by the user.\n\nChange-Id: I082355b0414ee5308418c683fb133d67ea414d3f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/3d82e550506faf633b404935017f4b10457886aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/3d82e550506faf633b404935017f4b10457886aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
