)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~674482","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~Ida893c0ca6a340b342d903974ec64b8a8c98565b","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"network_policy_ip_block_allow_except","hashtags":[],"change_id":"Ida893c0ca6a340b342d903974ec64b8a8c98565b","subject":"Check ingress and egress IPBlock rules for NP","status":"MERGED","created":"2019-08-04 10:38:02.000000000","updated":"2020-03-03 10:16:28.000000000","submitted":"2020-03-03 10:14:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":30,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674482-1583230480403-97979107","meta_rev_id":"6a6aae2ed2f71e3b09a8f49e10bb9efc56b22f97","_number":674482,"virtual_id_number":674482,"owner":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"date":"2020-03-03 07:05:18.000000000","_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-03 10:14:40.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2020-03-03 08:31:30.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":1,"date":"2020-03-03 08:17:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-03 08:11:59.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":1,"date":"2020-03-03 08:31:30.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-28 10:37:17.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"},{"updated":"2020-03-03 07:05:18.000000000","updated_by":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"reviewer":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"state":"REVIEWER"},{"updated":"2020-03-03 08:11:59.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2020-03-03 08:31:30.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2020-03-03 10:14:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"32ea31d092b7b26398911c9074914c92d693203e","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-08-04 10:38:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ef5e221dd0ab3501f4edb4fda94b38a5567acc6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-04 12:48:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs finger://ze01.openstack.org/b22b037c043c4a158e4eca30d0f62d4e : RETRY_LIMIT in 2m 42s\n- openstack-tox-pep8 finger://ze05.openstack.org/0463b358b2fe45c683b4f1e02d45ac66 : RETRY_LIMIT in 2m 30s\n- kuryr-kubernetes-tempest https://logs.opendev.org/82/674482/1/check/kuryr-kubernetes-tempest/38d00ed/ : SUCCESS in 1h 18m 38s\n- kuryr-kubernetes-tempest-openshift https://logs.opendev.org/82/674482/1/check/kuryr-kubernetes-tempest-openshift/2f6e75e/ : FAILURE in 1h 50m 01s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://logs.opendev.org/82/674482/1/check/kuryr-kubernetes-tempest-containerized/7d2bac9/ : SUCCESS in 1h 21m 38s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://logs.opendev.org/82/674482/1/check/kuryr-kubernetes-tempest-containerized-lower-constraints/7d2d82f/ : SUCCESS in 1h 26m 01s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://logs.opendev.org/82/674482/1/check/kuryr-kubernetes-tempest-containerized-ovn/02fcc79/ : SUCCESS in 1h 51m 52s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://logs.opendev.org/82/674482/1/check/kuryr-kubernetes-tempest-py36/9da1cba/ : SUCCESS in 1h 06m 21s\n- kuryr-kubernetes-tempest-containerized-py36 https://logs.opendev.org/82/674482/1/check/kuryr-kubernetes-tempest-containerized-py36/846faef/ : SUCCESS in 1h 38m 09s\n- kuryr-kubernetes-tempest-multinode-containerized https://logs.opendev.org/82/674482/1/check/kuryr-kubernetes-tempest-multinode-containerized/b79902e/ : SUCCESS in 2h 08m 41s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://logs.opendev.org/82/674482/1/check/kuryr-kubernetes-tempest-centos-7/5be9c44/ : SUCCESS in 1h 35m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1a63e24fb3e90dc1100ce1a36add00b4ffb8bdb4","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-08-28 15:58:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"585c9a7a6c2ce4be13c7e9fc0a06b905672aa7a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 20:38:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b810b81718b946508b9b7301f131aa12 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5b6c62c56c44317b87c7fb002997a8f : SUCCESS in 5m 31s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/0f2606d4bfc74e578098b3b4e37d9cc4 : SUCCESS in 1h 13m 28s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/bd0c676d54e043f6afceb9a056010033 : FAILURE in 1h 33m 06s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/73f7bad9473a4d78a66aeca61dc8c274 : SUCCESS in 1h 33m 02s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/0e9d69eebe954c8b85d521aea6de3f95 : SUCCESS in 1h 22m 31s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/777b6d52cad34b48b805117dfc4ca03e : SUCCESS in 1h 35m 56s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/fbbaf08a33844edfa03551e4f8bd8645 : SUCCESS in 1h 43m 53s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/fab9939c64ad40ba98f7b2fb9de679ea : SUCCESS in 1h 41m 30s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/ea40038bb3f549bd896eb0ce3f8e1c64 : SUCCESS in 1h 44m 07s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/ac3a5bc13ef64537943d98f64b56168d : POST_FAILURE in 1h 48m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"893c21e4f761a01be9d8910ce215ff43e6fc2bd3","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-08-29 07:50:29.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"2b2c4795032aac29295d83997d449fa86377f6e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 10:24:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/1892f399c04048f2a45a8c7470d0b5c4 : FAILURE in 1h 57m 49s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/70c759c708544873bc89c02a66950ee2 : FAILURE in 1h 38m 10s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/311b5b7b0dc34c2983b9d2607c973bec : FAILURE in 1h 50m 36s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/a96bfd8e90504c839f260878bb8b21e6 : SUCCESS in 1h 13m 27s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/b4511a53338d42a182fe7f85f3566bab : FAILURE in 2h 26m 40s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/558c0b230d3542469f3b88719f48a92b : SUCCESS in 1h 26m 13s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/cf168cf874e5412fbd50e278f7887b73 : FAILURE in 1h 32m 58s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/c7d0039213794f888c379d887036e60f : FAILURE in 1h 39m 55s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/df5960f91d4a48f48b1643fdeb426e1a : TIMED_OUT in 2h 17m 46s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/465470cefc2a4fbe94f55b47888c5f28 : FAILURE in 37m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"211ea2cfc9025559858cf690a0dd5f073fd1e35b","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-08-29 10:57:34.000000000","message":"Patch Set 1: Code-Review-1\n\ntest are not being executed","accounts_in_message":[],"_revision_number":1},{"id":"65d04370fef1b8b0fdee4682b03265899aa2a890","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-09-15 12:20:01.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"149b3e31912f1d1494bc86a636717ac3f4b72eec","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-09-15 12:20:15.000000000","message":"Patch Set 2:\n\ncheck experimental","accounts_in_message":[],"_revision_number":2},{"id":"e56e89ac8cf6c31fba988f3513e45439b5d1890a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 13:05:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/86833df0826f42878e61c3ea5f62d54f : FAILURE in 31m 38s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/053bb89d552844bfbcd0e02871cbb2a0 : FAILURE in 29m 52s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/eb5de4daeb5347e4aeeb23a9c08d95ba : FAILURE in 32m 22s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/3466ec18c1f746c9bfb341824f0275b5 : FAILURE in 30m 55s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/d617baf7f35242f6ae4122c079a819ed : FAILURE in 31m 39s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/2c7fa3071bdd44ec93b9592ead4c49c1 : FAILURE in 33m 27s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/031c9859f87c419981baa0503aee974b : FAILURE in 29m 52s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/f4390cb9f79a4c4bada3705d09c710f7 : FAILURE in 29m 59s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/3efcd785cd9544b699fe0679c6e78f2d : FAILURE in 43m 26s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/ab0c240eec1c4f5d8ccce5dd93881eeb : FAILURE in 24m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5a06009f826e9496600b08475661b97fc55c2995","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 13:12:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbb337ec4da64ebbabbe1bc4e5d511ce : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/848cfefb125f45b18dff3743d3502345 : SUCCESS in 3m 34s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/b19cf0aa7bb14cc1b169517b3d426881 : FAILURE in 34m 42s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/8473db746c624eb6b930e7481bf751ea : FAILURE in 37m 47s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/315305865c834815a2dc9f779cb8cfbf : FAILURE in 39m 36s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/4b520594c0b240b5aa3b01b164c9af4b : FAILURE in 44m 09s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/9ba24b4dc3ba41028825b0a48ad94b69 : FAILURE in 44m 04s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/03952352e414420587d31a49c9929174 : FAILURE in 27m 18s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/e803871b160c46dda38f76370b26e603 : FAILURE in 36m 01s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/4a3f2fe1aa5f446a8214054582ba7f90 : FAILURE in 50m 58s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/840b03e142044fbcb22703b2b10a3e70 : FAILURE in 35m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"13902c8a385da59cca39845eabeabc37d632a0e6","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-09-15 14:09:25.000000000","message":"Patch Set 2:\n\ncheck experimental","accounts_in_message":[],"_revision_number":2},{"id":"bc418e7c49e24838d74a95e2e46e6f8d5246a2bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 14:59:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/3aab590427954a778f02a1777f081f19 : FAILURE in 28m 19s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/d180699ec408455d8b9baaa6d11bc06b : FAILURE in 27m 44s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/e46c3d706b4640ef8c639beb4612443f : FAILURE in 27m 13s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/d474bdd76cda459dacd576ad52a8b0e5 : FAILURE in 27m 34s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/23afef21a3d14dd3a615ef99aa853c1f : FAILURE in 27m 35s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/a9b91bdcc5984682a5c7b8198a98d91b : FAILURE in 28m 42s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/04c74861e17d45018f4b2a04d8f54f43 : FAILURE in 33m 52s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/528952fa802d44908706e686b0fe4be3 : FAILURE in 32m 15s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/ea370cfb24b4466a903290dc2ded4022 : FAILURE in 42m 16s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/567f0d84df1748c1804722c478c5daad : FAILURE in 21m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e8919a48c62dd46e55bf723b1a600f4bf99ce238","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-10-02 11:31:19.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5ad8cc8ff418e97f1df5d6bdabc2cea5daaeab7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 13:23:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1cfa2dbd40d42debe3c8ec187207332 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3f84735fdd241f78619dda335828cf7 : SUCCESS in 3m 43s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/fe709f58c5154974a935e803119ced8d : SUCCESS in 1h 18m 52s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/865e3187fea44a63acd032c5636fe0d7 : FAILURE in 1h 35m 20s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/dc324a4bad9a42238cadbfee258f16c6 : SUCCESS in 1h 34m 19s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/baa9808890964f4699aaff69f28e2d23 : SUCCESS in 1h 46m 03s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/8574a70404b24c9db777df7f0d3b8c38 : FAILURE in 20m 15s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/6c533a38901f4ad4bfa42efa69c5b357 : SUCCESS in 1h 11m 59s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/2cf1073944c54ed0a882ac59cf6cb820 : SUCCESS in 1h 09m 55s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/25e9b1d7a782437fbd57c30be6ff4005 : SUCCESS in 1h 26m 02s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/e1342755d74d49aab08a91a1fb1d8a53 : SUCCESS in 1h 10m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"80cf63f958c95d4d09526cf4712baf1f3d60acdc","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-10-07 11:33:20.000000000","message":"Patch Set 2:\n\ncheck experimental","accounts_in_message":[],"_revision_number":2},{"id":"f5af2fd89d74f0e0ac91927e6ff8a46387066154","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 14:42:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/aa9465174aed4e0fa2b2ce306f3aca91 : FAILURE in 2h 24m 57s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/a6f7bdc78cfa4386b6649bac701d297f : FAILURE in 2h 02m 21s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/15a35071d1e34aa7813e36083c3dac89 : FAILURE in 2h 28m 08s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/4ce78df11cc14e68acbc9b43a564dd2b : SUCCESS in 59m 44s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/627f797c52dc4ffdaf214c6c9501884c : TIMED_OUT in 3h 07m 04s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/f9046fad75a34a2e9d54106d19d8c7cc : FAILURE in 18m 42s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/56cf0f4539b5425c91ad1bd960e5c068 : FAILURE in 1h 59m 06s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/125a042891a642ef8dbb94f5c61f82e1 : FAILURE in 1h 48m 50s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/c18e1e1128ca4ec8bbce517ba14ad8e7 : TIMED_OUT in 2h 20m 01s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/de7c4a47cf00416b89d8fb1056fbf57f : FAILURE in 22m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1da6e05555bd5f1a548d786d871d8f15a23976bd","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-10-11 14:53:23.000000000","message":"Patch Set 2: Code-Review+1\n\nSeems to make sense. The only problem is that kuryr-kubernetes-tempest-containerized-network-policy is not running any tests [1]. I won\u0027t merge this until this gets fixed.\n\n[1] https://zuul.opendev.org/t/openstack/build/4ce78df11cc14e68acbc9b43a564dd2b","accounts_in_message":[],"_revision_number":2},{"id":"441ba2adb9eb7c7e4009872f1ae3fb252d12a066","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-10-31 08:37:26.000000000","message":"Patch Set 2:\n\ncheck experimental","accounts_in_message":[],"_revision_number":2},{"id":"d9be8e95baffa90a162bc9c0d0b157b012e16cee","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 11:56:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/e1a83910f41e479b9a977a3c78e75b0f : FAILURE in 2h 35m 52s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/ddac7c38a676484e8d68677a9b7f936f : FAILURE in 1h 50m 24s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/cb40781d8a80418fb0ddebcc5f204c34 : FAILURE in 1h 50m 00s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/f176595902b64f3cb5e89ff41c91f4df : SUCCESS in 1h 00m 22s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/f16e64a8350145e1a7c5f71edfe06f3f : TIMED_OUT in 3h 05m 12s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/8ba5e2e4ec7c44beac1d9393fe716e12 : SUCCESS in 1h 19m 55s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/481751abd9d14911b72a388f7e5fdd0a : FAILURE in 46m 08s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/2bca73c3c6144d20be82772c9bff94e5 : FAILURE in 55m 17s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/26920fbfd8aa42f4b92176f325fba481 : TIMED_OUT in 2h 16m 22s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/aaab3076f67b445aa206d832f2533248 : FAILURE in 27m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"319565dec8e820cca0c0c35385c34313fdccdc34","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-03 12:09:18.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"472e0755bba98193489dd57feeb439404f3d47b7","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-03 12:09:34.000000000","message":"Patch Set 3:\n\ncheck experimental","accounts_in_message":[],"_revision_number":3},{"id":"ae5f84f8e4893f3b0653104bfc5c146a887696ff","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-03 12:31:20.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"dfeed75f74a4e7b1c6d23ffc9af6adaec9068bc1","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-03 12:31:41.000000000","message":"Patch Set 3:\n\ncheck experimental","accounts_in_message":[],"_revision_number":3},{"id":"d9c79c873c08849985b2fbe7c042d6d65862798c","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-11-03 12:50:09.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"34222aeb8cf241dc960522f12dfa444dda7183d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-03 15:42:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d568ca6d332457babcb4a1389dd4374 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a92578848da244cab0780f64df3620ee : SUCCESS in 2m 40s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/c8ff4eb4c05f47d4bfc7aa50baeed056 : SUCCESS in 1h 27m 59s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/21c72f53ec064c318f82ef235f088c8a : FAILURE in 1h 56m 30s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/af5eaa10ba0044039e5dacca6ab247de : SUCCESS in 1h 18m 33s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/1630b4263f0a48c0a707c38f4f5d3eb0 : TIMED_OUT in 3h 06m 51s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/7737c420f62e4ee0bec154366159df07 : SUCCESS in 1h 11m 30s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/4b523d2033ae4d85a26f8790b6a40f6e : SUCCESS in 1h 35m 14s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/1e32830817784243af10d23b182bec9c : SUCCESS in 1h 37m 37s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/d8f2fae1ae854dcfbd713fe5a2660a63 : SUCCESS in 2h 05m 15s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/775123b18bf34b9cab513c1c6e88fbe8 : SUCCESS in 1h 27m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8622aa0e91f363ca45ba6dab16ffd26985bfb990","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-03 15:42:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/2b7f054d07434ee58120fcb2db6b6c6e : FAILURE in 2h 16m 18s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/c30b81e1898e4db8a01daa0fe3bd66c0 : FAILURE in 2h 07m 02s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/01da3908408a4e8299e039a9842d7683 : FAILURE in 2h 11m 53s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/270b3271b52c4ca9a5fc45706d159ef0 : FAILURE in 59m 48s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/164bf37d402b40e0ba9805c1f291790b : FAILURE in 3h 05m 47s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/8528bfa824bc42028f60726ffc88f213 : SUCCESS in 1h 28m 16s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/59b9f64eab03480da43a78f8a4036e61 : FAILURE in 1h 54m 55s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/ecab6e09ba3640c3a1ac5dec0c49e513 : FAILURE in 1h 32m 38s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/57fb06e6f92b46e9a1dd6d4564b44b28 : SUCCESS in 2h 17m 52s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/b558b18e9012447cb67934609c6ff08c : FAILURE in 29m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"67f6612133c1b54fbbc3c561a1812afa361b3001","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-11-05 08:03:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bcffe1e59329a16c27de49834c7158c67ef85164","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-11-07 22:18:20.000000000","message":"Patch Set 4:\n\n(2 comments)\n\nWhen having the following Network policy the last assert on the test should be working (Ping from 10.0.1.205 to 10.0.1.249), as the following rules were enforced on the pod, but it\u0027s not:\n\nhttp://pastebin.test.redhat.com/812455\n\nEither I missed something or there could to be a bug in Neutron.","accounts_in_message":[],"_revision_number":4},{"id":"2fe077bba05642f154f715b38d3e1625eef34940","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-11-13 13:47:20.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5b42700f2fb2de3c6ce1bd7b71570f482dd89d13","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-11-13 13:48:16.000000000","message":"Patch Set 5:\n\ncheck experimental","accounts_in_message":[],"_revision_number":5},{"id":"bca9caf5384cbe2e1fcd26f1cc5fb18d78b6435b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 16:57:11.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/534594f82249408f85460087d3e3d8b8 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b76aa554aef04c4f93ed4ee61551f334 : FAILURE in 3m 55s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/7b9b7007356c46d59b1a70facfffb12e : SUCCESS in 1h 18m 01s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/2085226026f44610bcb75692947733a8 : FAILURE in 1h 36m 52s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/931fe53cdbb24bd59e5ce784cbff805a : SUCCESS in 1h 14m 57s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/496faf4fd51847688887898e3b62ab11 : TIMED_OUT in 3h 06m 57s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/5b9fb7a3cf4745c2b12f5b6162945be0 : SUCCESS in 1h 18m 41s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/3d28439f54b74082adfc7eb283066309 : SUCCESS in 1h 08m 44s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/7a962c1db74647a8ac5cc2f415fcdb2e : SUCCESS in 1h 19m 28s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/77b76c85f4b744c88005c930b0e4d49a : SUCCESS in 2h 04m 11s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/8a271243d1754a77a04497ca5b9ccfeb : SUCCESS in 1h 17m 59s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"dca7a85c738ec54589ef6330b4aa03892fc3fb98","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 17:00:03.000000000","message":"Patch Set 5:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/0831d602b38a4d3d90a1501b9b1192c5 : FAILURE in 2h 02m 59s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/41c27869c7714dc0aff10a36287eb312 : FAILURE in 2h 09m 30s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/697730bcb1cd4808a266e461081ecd1e : FAILURE in 1h 58m 28s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/2eb10b99249344a394e8584b9ab60441 : FAILURE in 1h 31m 04s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/caa8e9a87a8d4564a3a2e35e74c465d3 : FAILURE in 3h 05m 35s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/00d01d73a0e54c06b8a8c03dc7b83b17 : SUCCESS in 1h 19m 14s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/feb65b49613c4e068bb7b882cd84bdc7 : FAILURE in 2h 20m 50s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/ec8e0c6f6436484e9eb11fd8fd79bcfa : FAILURE in 1h 30m 49s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/b294f3582fa24abe85ffa95ad026606e : TIMED_OUT in 2h 17m 06s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/6a576307f85a42c48c83d2534fb009c6 : FAILURE in 30m 13s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cf11ab96600785e272b771e6d8973067b6f43fbc","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-11-20 15:06:56.000000000","message":"Patch Set 5: Code-Review-1\n\nSeems okay code-wise, but fails in the gate.","accounts_in_message":[],"_revision_number":5},{"id":"fa10888c2940d533d92b085882b9605fa2bd0f20","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-11-20 15:08:15.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5897128e40c8108f196a5b344d3d021862dcc062","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-11-20 15:39:40.000000000","message":"Patch Set 5:\n\n(1 comment)\n\n\u003e Seems okay code-wise, but fails in the gate.\n\nI think we hit a bug here that the moment you block traffic for a specific Pod, all other Pods are blocked as well","accounts_in_message":[],"_revision_number":5},{"id":"aab044a9c3cad131039c67b5d2acab71cec0df7b","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-11-20 15:40:44.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"739eca04db260efcf8f81ace7cad88604d2919bf","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-11-20 15:41:12.000000000","message":"Patch Set 6:\n\ncheck experimental","accounts_in_message":[],"_revision_number":6},{"id":"834a996e538ebc5c769ac7c57c66c9b5790094d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 18:51:40.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a33b0886e494d78940c08b7eab0abde : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba933f25d279465c85dc449bd913e786 : SUCCESS in 3m 13s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/3cf8ccc547694cda8040f93231b8ae56 : SUCCESS in 1h 12m 56s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/3e8b4e8962764823bcdf5097b8551488 : FAILURE in 2h 06m 02s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/76ea4ef4d7c14e249dbafcfbe4e433cb : SUCCESS in 1h 48m 34s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/f4bd034c44934c68b5971d069491369b : TIMED_OUT in 3h 05m 44s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/5665336605904e2c9b1e825d756215c0 : SUCCESS in 1h 21m 08s (non-voting)\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/9285e89aaedf4569bd1f28fcd03b5b3e : POST_FAILURE in 1h 44m 25s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0a41a2efa2b248acdd3fa36421f5f7a0dc4fc030","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 18:54:14.000000000","message":"Patch Set 6:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/48f5ff3177524400a9d9067477a60f79 : FAILURE in 2h 30m 00s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/ab027bef212b426e9a57229566ac34d7 : TIMED_OUT in 3h 09m 41s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/0f244735f3ff4f029d092fd0eda6ee56 : FAILURE in 2h 23m 03s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/a88c9fc9d70c47d6913df237ede6bcea : FAILURE in 1h 40m 34s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/28a54a0d38ca4457989e16c604657fb9 : TIMED_OUT in 3h 04m 06s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/2574e5bc2e5c4f8585eea5f6ac154544 : SUCCESS in 1h 22m 49s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/299a3eff1fc84dc5a4f48b72272fbc51 : FAILURE in 1h 50m 40s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/2cb57c363eb845608deb285cc47c5734 : FAILURE in 1h 36m 26s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/b34ac77e2df44eaaa4b224e9f85838bf : TIMED_OUT in 2h 19m 17s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/eb9b352a98da4fda8e8aa1a0ed457051 : FAILURE in 27m 52s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f1ec4857cd081207f0efa3c5428d40b7b3df8be1","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-02-23 10:02:08.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"3916e072b086850636173e399d4c6bfc0b8d8071","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-23 11:42:01.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ba7a5afd87a4270a54f0a4676718616 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcf9bc863c0e46b28b796d3cbf2727e4 : SUCCESS in 4m 17s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/2290b504e5f04b969c92b2fffbefac0f : SUCCESS in 1h 15m 53s\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/37fd7ed3824d43d18e0096f4402c9847 : FAILURE in 1h 28m 18s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/2310e10265dc458cb657b7edfdec867d : SUCCESS in 1h 20m 28s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/2b7beefdddbc45db85dac3c47291f4b8 : SUCCESS in 1h 34m 52s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/8152d7b84efe4e99ac4cf035a1c27617 : SUCCESS in 1h 32m 04s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/2f813ff178ee4cb9a350b59c60afdc48 : FAILURE in 1h 27m 05s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/c0fc43eae31c46f295eabe7a0e7c7ff2 : SUCCESS in 1h 30m 42s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/649bc04f33e0405fb5f70bbe87ad800a : SUCCESS in 1h 15m 28s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/210e9811cfec47a6b43838402b87cea9 : SUCCESS in 1h 13m 33s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0a8eae2f1aef1ad4db744d60e4e58f8bb1c69827","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-02-25 07:19:18.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"2178b2a41ce22e379b717dc9e18220e000cf057b","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-02-25 08:10:02.000000000","message":"Patch Set 8: Code-Review-1\n\n(5 comments)\n\ncouple of nits here and there. Other than that it looks good to me","accounts_in_message":[],"_revision_number":8},{"id":"ff7d0a800add8cfdbad26fd6dad6ed5923ca95dc","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-02-25 08:31:22.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"064728cf5ea4df98cf0c186ce56aadf22d08b587","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-02-25 08:31:31.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"84b9b8eac942934b9fba8089f7da7a764065b3ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 10:22:13.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c8915f45530493eb22fc94972883030 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/875dc1fba91243b8978379b60ca05259 : SUCCESS in 3m 55s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/1f6ac58c5f924f1c92a0ff3a68ea54f4 : SUCCESS in 1h 40m 30s\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/88760e82adb7480884bde8329b8da4a3 : SUCCESS in 1h 45m 22s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/94b03f454fed465eaa1047b139562f37 : SUCCESS in 1h 21m 27s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/da355989d6b347a0b47d89486f740de3 : SUCCESS in 1h 20m 35s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/019485f534e34af1a9b722df920d5509 : SUCCESS in 1h 34m 13s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/093faa4548f24bd596fa3970e43345d5 : SUCCESS in 1h 44m 39s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/cf8d522de7fb4ad4a4e1cc5afb1f839f : SUCCESS in 1h 30m 29s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/a42d040a1ba240f8aff8941849001d30 : SUCCESS in 1h 21m 14s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/6bdff3b1ffa243c29766619710ac0e8f : SUCCESS in 1h 10m 51s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"6d08f4c1570c0e3fe14f213999438165d29c1729","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2020-02-25 10:31:05.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"c177d09f5177d96610fdf9b5c53abec4365f6ad7","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-02-25 11:02:57.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"49f60de35f59c882220f7ff82079f50a11a3a433","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-02-25 11:03:07.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"e043ed11ccaae4929dc3501f00ad6f52b6e656db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 13:13:36.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bb67cc71fdc4bb786d587d1a48e9df1 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/defa1f3fe024470aae517a0f121b81df : SUCCESS in 5m 28s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/02d6dc62896b417c80ea350b4628001e : SUCCESS in 1h 30m 32s\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/fb6155f182614b3baea4b9c39571b280 : FAILURE in 2h 05m 57s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/a5af488327c94b6ca6f1c100d6201331 : SUCCESS in 1h 34m 47s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/5c47bcc0284042e880c238633926c9f1 : SUCCESS in 1h 41m 05s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/0049ebb990474203aa5bb83045f703c4 : FAILURE in 1h 45m 24s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/d736ed48179142bdb848e90fea504934 : SUCCESS in 1h 34m 30s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/55574938f5c1450ebffa54cae4e007f5 : SUCCESS in 1h 36m 33s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/8694601ea5654b66b949013e980860b6 : SUCCESS in 1h 19m 29s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/c3686e7dbcaf4aec879443b6e19eea3c : SUCCESS in 1h 25m 21s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"43dc238505f0b571e9d82114edfa29430f6457b3","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-02-25 13:30:48.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"9dfb1c4bda66a2a6e5bb72d945b3bf158416093e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 15:36:48.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd1280eab0f740349956b1a2a6d259df : SUCCESS in 8m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc9f257bd18441f98c9348a954c4bb90 : SUCCESS in 6m 23s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/82514d4d154142c1841b37959e3e8644 : SUCCESS in 1h 30m 30s\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/5af84282e6e1400ab9c248d063785cae : SUCCESS in 1h 49m 08s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/7457269994e647ec97716ace924ff619 : SUCCESS in 1h 38m 13s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/220add91d79a42ed8f770de65e85ce5c : SUCCESS in 1h 26m 50s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/3e4c23def0794550a7b694b0aeca2f82 : SUCCESS in 1h 59m 37s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/c20655e2462147d188d777857ef60295 : SUCCESS in 1h 34m 28s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/79596147273d42faa717f1a172b8e77e : SUCCESS in 1h 55m 46s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/90449028728e4260a3f3041db1bdc293 : SUCCESS in 1h 25m 13s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/d2352ecd6595445486557c9e8a106926 : SUCCESS in 1h 22m 30s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"147657c7b2b48279fede723a1577b8f4de840df9","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2020-02-26 07:21:42.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"43ac2d49c0cde8d8c152922cb90a6a95a8490d63","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-02-26 08:48:11.000000000","message":"Patch Set 10: Code-Review-1\n\nplease handle the remaining comments at PS8","accounts_in_message":[],"_revision_number":10},{"id":"112012972378828bede8261184549c74e255a32c","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-02-26 10:12:39.000000000","message":"Patch Set 10: Code-Review-1\n\n(2 comments)\n\nThanks Genadi and Itzik. Just a few more suggestions:","accounts_in_message":[],"_revision_number":10},{"id":"8c1173783f1631ca7a8997da1ec13f92c05cb0bb","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-02-26 10:13:23.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"5774771c222d81eec91848e06265c6d3f0fbb661","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-02-26 11:01:12.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"771c2c9ce50095aed58ea659a6a016e345897c17","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-02-26 11:13:35.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"a5565e1639c04dc379af7ef1493ef09a9ef1f572","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 13:11:00.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19758ed4993949c9aebc3dc0456f2fa3 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9adc90f32d5c41d79a625d8374401bff : SUCCESS in 4m 26s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/d5317b79856943078ff422480eafbe62 : SUCCESS in 1h 15m 30s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/91de04512518477683778c08e09de653 : SUCCESS in 1h 17m 31s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/8ef9c5ba6df44b9cb1d974ae9c18dcee : SUCCESS in 1h 18m 45s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/dffa703326b9414f9cf010aeb3e5f645 : SUCCESS in 1h 50m 25s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/e041f28c8eb14eeea07999ee8ea306a2 : FAILURE in 1h 26m 09s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/50aba15f2fce44dea8f4facd35090986 : SUCCESS in 1h 46m 15s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/a088824164074a598002c5ed6df7e52b : SUCCESS in 1h 19m 16s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/25c8f4f54ecf4d56a92b30ba90d2c765 : SUCCESS in 1h 10m 33s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"b262b3ed0691bca551793d968d854c9b4374353e","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-02-26 13:40:37.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"8a0fde7847ebfcaeba309572969d29d9e917d330","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-02-26 14:36:15.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"0c5cb5b38d7f2cc51eaeacb4818ff4c304011686","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-02-26 16:58:50.000000000","message":"Patch Set 13: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"43bfe6e2f4cb88ce5ef8d97d57c6afcd6e8740da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 17:45:24.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aad7d7882ded4c2d8b60b0deba1c127b : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53272ab52ce7434f822d81490bc1cdde : SUCCESS in 5m 32s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/98d356a1fe04489b8d9da688c48448ee : SUCCESS in 1h 14m 34s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/10e3edefab7d478395996cb255a970a8 : SUCCESS in 1h 31m 14s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/48be00de804f449d879156d75171f350 : TIMED_OUT in 3h 04m 19s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/9d707830e35746a48cfcd51b6f9a708d : SUCCESS in 1h 36m 57s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/0e1035d31dfe4dd3ac2d35d5697942dd : FAILURE in 1h 38m 26s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/b254fde61634428e8994309386199afa : SUCCESS in 1h 37m 08s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/a068b9c476dc489abdd8f8755f3bf546 : FAILURE in 41m 33s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/181b552a69494eaa83ee4b3a89c8656c : SUCCESS in 1h 09m 04s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"b038a2148fad5e50dcda9f4a38c0de9fa4913f77","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-02-27 13:33:18.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"36873d208f63bf721c3ae5ea6a45bbe967dc6c01","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-02-27 13:33:56.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"a6ebcdd9ecd7eccd0e50bee654ee43473b46671b","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-02-27 15:54:13.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"c6d4066edf997753c94e7b1d63e92d5c5be18d9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 17:33:15.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23b645c7228a463cad9f5d960ecb7320 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1d007932b69480b90efabae3f526418 : SUCCESS in 7m 39s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/8709bf2f0f3a49cc9c50c40eb610465a : RETRY_LIMIT in 27m 37s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/ad1805b423c84c85881f27ad156aea2a : RETRY_LIMIT in 19m 46s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/9dcc425b14944670852ff9e3c1936b3a : RETRY_LIMIT in 19m 33s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/919e65289e754a6b9c329656675f39b3 : RETRY_LIMIT in 18m 51s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/6a67895dc11648969faeef7cf3972e54 : RETRY_LIMIT in 17m 48s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/0a3838b4996a4666a1da488f59c06be0 : RETRY_LIMIT in 34m 16s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/d16da8392f86489cb49db1ddea4542ad : RETRY_LIMIT in 29m 45s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/fbe8a897e3de47df8d12e85ddb75fd90 : RETRY_LIMIT in 26m 55s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"36b7629779e9d10c9db290c56da012445b629b61","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-02-27 17:59:26.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"52717356d973a39a2f45909fdfc36be45f75cb79","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2020-02-27 18:52:47.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"de9a8be556811c24e5981072d74cb864de9d3fcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 21:41:03.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13d2c221b42a4ec89aee2588a228fab6 : SUCCESS in 7m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19c9cbbf937e4395906a795965d45ae1 : SUCCESS in 7m 46s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/902931d6935549f286dfe981e9cc7cac : SUCCESS in 1h 22m 03s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/6e5c37c9ab3a4a4e8473fed2f66c5f2b : SUCCESS in 1h 46m 38s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/23f74cb59ec04a97b80cb413c9ac6da9 : SUCCESS in 1h 41m 44s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/9c32b9cb5d654ca6b832f462ec48d37a : SUCCESS in 1h 43m 32s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/60d965bd6eeb47088325a2ab84cca998 : FAILURE in 1h 45m 14s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/780a4e091c7446d69f04f1437f067043 : SUCCESS in 1h 54m 41s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/db9318557c5d4a0e9e33a4c928cdbf32 : SUCCESS in 1h 35m 12s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/0f1b15ed247149d9a67c4b0031f41d1f : SUCCESS in 1h 19m 06s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"7913a8888962859830e759fe5fa3fbfc2af3fc99","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-02-28 08:12:35.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"ae6e023e96ccdd9a8cef27d64502b03d080238c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 10:21:45.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e37f5dc8a114e51aa9430997fe80617 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/505c3b5262d24854a06c8d0c67a7faed : SUCCESS in 4m 08s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/63d0d75bad0f41d88a06683c1f5cbcd5 : SUCCESS in 1h 09m 49s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/57738c95188b450d9a26f19e2ec418bc : SUCCESS in 1h 19m 01s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/45e6f52b14834f4b8623b9a394d1d30c : SUCCESS in 1h 27m 44s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/abae092f463c46419099c51c01af79b2 : SUCCESS in 1h 44m 30s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/fc978090abbf4bc19a0bc91cc9edbcf5 : FAILURE in 1h 57m 26s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/eec2348067f64a1885712a637c382dc3 : SUCCESS in 2h 06m 12s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/94daba5c581144f1a16375f4e2ccd446 : SUCCESS in 1h 25m 47s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/9b5c6e2017a543a99ae7e314f9c46b04 : SUCCESS in 1h 35m 32s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"1ef4120aa4d8afdf20f043a8b2ece3337b8fbd30","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-02-28 10:37:17.000000000","message":"Patch Set 14: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"33c7523460f9415846ec14db9c5fbe9daa20f5fb","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-02-28 11:24:19.000000000","message":"Patch Set 14: Code-Review+1\n\nnetwork policy test failing","accounts_in_message":[],"_revision_number":14},{"id":"4a9ffe15abcd43276e966191e08621f88cc399f0","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-02-28 11:24:23.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"10838b129e5e0595c99185408891f14487228c55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 12:46:42.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/374a2711f0ef4e388c30dd06333eb0c7 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b9eb3a1e3494fed8dffd1c413dd8928 : SUCCESS in 6m 18s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/293ac1c6a9de44ae9e6225c759c94e72 : SUCCESS in 1h 13m 02s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/0c4b6fc5490348d28dccbabf905e09e3 : SUCCESS in 1h 19m 46s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/592440cafecc4145a895cc241a8b5706 : SUCCESS in 1h 39m 11s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/91a76de40619488b8db6136fe5e99587 : SUCCESS in 2h 06m 21s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/105eca2862494f419ac8d3edb011b7aa : FAILURE in 1h 33m 23s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/30cf0bff640c42f89bee18414dc603b5 : SUCCESS in 1h 48m 40s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/c0a7e3c5e9c74f8ebff765714e11531a : SUCCESS in 1h 12m 28s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/8802bf05d21c4849862a45c108f7d277 : SUCCESS in 1h 14m 31s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"32a36e62080f01faeb0dd374ad3553f500a79348","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-02-28 15:40:47.000000000","message":"Patch Set 14:\n\nrecheck\n\nWeird, but seems to be unrelated.","accounts_in_message":[],"_revision_number":14},{"id":"00d2cf15c29b548f0a757729d33a804f4f45d918","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-02-28 15:45:34.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1\n\nThis looks good and while those \"Length too long\" errors seem pretty weird, I bet they\u0027re unrelated.","accounts_in_message":[],"_revision_number":14},{"id":"eb59fd0e3aa748a4ac910c2defc4615701f36544","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 19:57:46.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e29c476e49f84b9a8837a3abf7a458b2 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4f43474b65b4325a5985fd8fdcc9c62 : SUCCESS in 5m 50s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/8c9547e98da04297a2fc8dcda6afec8e : SUCCESS in 1h 41m 08s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/f624881581184d34a83d8a89f3742589 : SUCCESS in 1h 29m 34s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/171ce7477cc34692a9b4ab108ee72c70 : FAILURE in 2h 00m 12s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/1df3e749f6f4409e89b400faa0ebb2de : SUCCESS in 1h 55m 54s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/884ecfb0f2a1495780181b92df5b57bf : FAILURE in 37m 35s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/0d23096832cf4998b9ca8e38cc4840fd : SUCCESS in 1h 43m 02s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/e8cd784fef3646eb91ed0e780f4b494d : SUCCESS in 1h 09m 37s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/1966c70dbeb640459f8690302cc6bf82 : SUCCESS in 1h 10m 58s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"957c2f0b801ad0812e8cb78e9d3f173b8f0a2b2c","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2020-02-29 17:01:26.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"5ba4967b1477d127e70c01e053af72715bc0273d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-29 18:48:47.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4e8d5aa451347e8a7281f9265ad9f92 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bdd981ee0b74a4bbd499070b80b5795 : SUCCESS in 5m 01s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/6a3a6061d2634f92b8fe843ef7735206 : SUCCESS in 1h 20m 27s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/dddd47f07cfe436da2a3689a00eb445d : SUCCESS in 1h 20m 37s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/0b46b4a7a025491487dc3cb90cb2d2ff : SUCCESS in 1h 40m 51s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/dcbeb685f7ae4d77a818aff28552bd09 : SUCCESS in 1h 14m 26s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/9e231cf7e32c4c30b2c69a2dfc1c9826 : FAILURE in 1h 31m 25s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/43d8b9e4fc1049e99245cf293d9aa630 : SUCCESS in 1h 42m 12s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/592735cf8e084cbba2a9d154f9ab9121 : SUCCESS in 1h 09m 03s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/4af3604e1d064f26a289a506d7a72b2b : SUCCESS in 1h 09m 42s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"0ed5762b342f39540c5854a1e6ec0af0feabb973","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-03-02 07:02:13.000000000","message":"Patch Set 14: Code-Review-1 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"e9f6ec5a813ef58acdee4b618cf1aabbc7d22940","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-03-02 08:08:29.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"39e4ac484c78e83fd2e1301042f0f7fa4ddcbb07","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-03-02 09:14:57.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"c14fbe65610ad35bedf67d37e8db5b40c0aa19ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 09:56:52.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcd4b3c01a9d4251b4a1f779b6a4253d : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f73e409b1354e03a4916942f0fa5571 : SUCCESS in 4m 15s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/ce6b3420a68647c58a66206ac120e033 : SUCCESS in 1h 15m 43s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/d73dffc9d4544a2aa36cb69cbc37ffdf : SUCCESS in 1h 30m 33s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/0a42012b3fa9472a86649c2e63c93d80 : SUCCESS in 1h 27m 46s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/672bfa88ec9944d6953d05e009dff3ac : SUCCESS in 1h 28m 55s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/af3fb326553d4ad188223dcca68fe970 : FAILURE in 1h 21m 23s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/6ccc5c20523a4d658cfc5d6c5d666e6e : SUCCESS in 1h 43m 17s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/48f3a6bac2704ce8a0c6d018a1097ce3 : SUCCESS in 1h 19m 13s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/5df2e681efc04b5abe460c13c6a4a400 : SUCCESS in 1h 22m 55s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"0a3028d0ca9a3a96866cc549574c8615b98326f0","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-03-02 10:07:08.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"e38b6bd8d457d10c42678f0845461c4d20a775f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 12:02:45.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82bacdf71f6e48e1b86deb1edbe0b9b9 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed94ff08baa54e2db7530e93cb52f62b : SUCCESS in 4m 22s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/98adb612d6cf4282bbe97dbae6d0fc66 : SUCCESS in 1h 13m 30s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/9523f5dd2ee14a1cbe2099927ca78e4c : SUCCESS in 1h 34m 01s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/bbb6ec7227fe4458a6560c87339ecd4e : SUCCESS in 1h 23m 30s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/d3e66ad883a54ddd9ba1841a1035f6ae : SUCCESS in 1h 46m 17s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/e510cc3900854eaf92464d47ae23fd72 : FAILURE in 1h 33m 34s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/a0b33fbad27f418c80312abb0c10e820 : SUCCESS in 1h 39m 21s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/0343bb7d0f214385bbf2032791b6c4e7 : SUCCESS in 1h 13m 40s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/d04fb4f972984b08818332cc806a8fdd : SUCCESS in 1h 16m 02s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"3bc032f1813235ed24c3bc122488dba2753358f5","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-03-02 13:01:41.000000000","message":"Patch Set 15: Code-Review-1\n\n(1 comment)\n\nOkay, at this point it\u0027s hard to ignore this. The \"Length too long: 10670009\" error happens only in the gate that runs this tests and only on this patch, so I bet it is related. My thought here is that the tests produces so much tempest logs that it overflows some buffer.\n\nLet\u0027s try to fix this by implementing the thing I noted inline.","accounts_in_message":[],"_revision_number":15},{"id":"a3395ce2b5457b7667be49eec1e93b88adb1c1e5","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-03-02 14:22:11.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"1aa65bf1cc6db31a868be29c4f298db91774d6d2","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-03-02 14:22:31.000000000","message":"Patch Set 16: Workflow-1\n\nOkay, let\u0027s see if this helps.","accounts_in_message":[],"_revision_number":16},{"id":"ae05d4fc3172edc69938c807b58a884bdc70aedf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 16:18:51.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a97cf8a66c1044ebbaeceecece9720fe : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04cfb9d874a047a08e0f96d45d60f946 : SUCCESS in 6m 49s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/eefe9c6db8984af9983bba835d760ac9 : SUCCESS in 1h 21m 02s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/0245dae8bcec4c08b5950b7dbc6bc15e : SUCCESS in 1h 36m 34s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/8aa09c08922d487a8123ac874d3f723a : SUCCESS in 1h 28m 50s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/d12458129a59453aba4b4e8f67039b65 : SUCCESS in 1h 51m 39s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/6a8bd2448e49475b81329a7478df820a : SUCCESS in 1h 53m 55s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/8bbc3a97837c465fbcf7d3351d64a6cd : SUCCESS in 1h 35m 02s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/c26b8ac8dd434ca8b7b4a3a739053efd : SUCCESS in 1h 08m 32s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/70390396e610495689e5282eacf09077 : SUCCESS in 1h 30m 12s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"73ef2da3be293ed4c820b5ecea7ff412557effd2","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-03-03 07:05:18.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"6bdd2c9cc118375a3388e14e01b1ac61a59776f5","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-03-03 08:11:59.000000000","message":"Patch Set 16: Code-Review+2","accounts_in_message":[],"_revision_number":16},{"id":"29b69d35642c29efc8587c9352bd6a4edfaad9ce","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2020-03-03 08:17:14.000000000","message":"Patch Set 16: Code-Review+1","accounts_in_message":[],"_revision_number":16},{"id":"71f219ccf746d4ea6aa59ef60b739b758c59b6f9","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-03-03 08:31:30.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"a3094ed30f8118d95c1de777d9d579ab12ec1b13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 08:31:42.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"fc91a91b3f0641d9cf9b1a932528c2281203ea13","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 10:14:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":16},{"id":"4945accd3bf44e84203652653b0647fd4bacbaae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 10:14:40.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a0a65fb7b704a21a6231bf22a28269f : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d750ccfaee654da991485c63c7944b14 : SUCCESS in 5m 20s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/69a5ccd64fbf490798c0c4404f807563 : SUCCESS in 1h 24m 09s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/ad953a2eb91547c6921414fc81d35644 : SUCCESS in 1h 28m 40s\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/56267d6aafcf4fd3995bb1717e84a4c7 : SUCCESS in 1h 41m 10s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/1f690d8b23434154bef8d61fe05a0ce6 : SUCCESS in 1h 18m 34s","accounts_in_message":[],"_revision_number":16},{"id":"158f367ccc4de7e47b3be0661ff9dd383555f89b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 10:16:28.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56858d05ebff4817818631f11c5bbf4f : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"d88e00caa11f9f499af060d5489d3d8c43de965b","revisions":{"8e9e63e761c623a15038e134b4ae3f8f126092ac":{"kind":"REWORK","_number":1,"created":"2019-08-04 10:38:02.000000000","uploader":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"ref":"refs/changes/82/674482/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/82/674482/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/1"}}},"commit":{"parents":[{"commit":"daca767ee91717a2661ce00fc376d14222385777","subject":"Testing basic Network policy IPBlock functionality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/daca767ee91717a2661ce00fc376d14222385777"}]}],"author":{"name":"Genadi Chereshnya","email":"gcheresh@redhat.com","date":"2019-08-04 10:34:20.000000000","tz":180},"committer":{"name":"Genadi Chereshnya","email":"gcheresh@redhat.com","date":"2019-08-04 10:34:20.000000000","tz":180},"subject":"Check ingress and egress IPBlock rules for NP feature","message":"Check ingress and egress IPBlock rules for NP feature\n\nCreate 4 pods\nCheck the connectivity works before applying NP rules\nApply NP by allowing all and blocking pod1 ingress and pod2 egress\nCheck that connectivity from pod4 to pod1 and pod2 is blocked\nCheck that connectivity from pod4 to pod3 is still working\n\nChange-Id: Ida893c0ca6a340b342d903974ec64b8a8c98565b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8e9e63e761c623a15038e134b4ae3f8f126092ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8e9e63e761c623a15038e134b4ae3f8f126092ac"}]},"branch":"refs/heads/master"},"ddc8cc8fea9fd39754bc3e30dceeda623b7b0605":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-09-15 12:20:01.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/82/674482/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/82/674482/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/2"}}},"commit":{"parents":[{"commit":"daca767ee91717a2661ce00fc376d14222385777","subject":"Testing basic Network policy IPBlock functionality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/daca767ee91717a2661ce00fc376d14222385777"}]}],"author":{"name":"Genadi Chereshnya","email":"gcheresh@redhat.com","date":"2019-08-04 10:34:20.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-09-15 12:18:27.000000000","tz":0},"subject":"Check ingress and egress IPBlock rules for NP feature","message":"Check ingress and egress IPBlock rules for NP feature\n\nCreate 4 pods\nCheck the connectivity works before applying NP rules\nApply NP by allowing all and blocking pod1 ingress and pod2 egress\nCheck that connectivity from pod4 to pod1 and pod2 is blocked\nCheck that connectivity from pod4 to pod3 is still working\n\nChange-Id: Ida893c0ca6a340b342d903974ec64b8a8c98565b\nDepends-On: I1109a303ee216ed8061959a81f0ed315bf138c80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ddc8cc8fea9fd39754bc3e30dceeda623b7b0605"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ddc8cc8fea9fd39754bc3e30dceeda623b7b0605"}]},"branch":"refs/heads/master"},"278647fe85335d15c52e85681629ae371b3bc944":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-11-03 12:09:18.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/82/674482/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/82/674482/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/3"}}},"commit":{"parents":[{"commit":"4175a85a1609f093987d24903d7f6d1592ad7c7a","subject":"Testing basic Network policy IPBlock functionality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/4175a85a1609f093987d24903d7f6d1592ad7c7a"}]}],"author":{"name":"Genadi Chereshnya","email":"gcheresh@redhat.com","date":"2019-08-04 10:34:20.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-11-03 12:09:04.000000000","tz":120},"subject":"Check ingress and egress IPBlock rules for NP feature","message":"Check ingress and egress IPBlock rules for NP feature\n\nCreate 4 pods\nCheck the connectivity works before applying NP rules\nApply NP by allowing all and blocking pod1 ingress and pod2 egress\nCheck that connectivity from pod4 to pod1 and pod2 is blocked\nCheck that connectivity from pod4 to pod3 is still working\n\nChange-Id: Ida893c0ca6a340b342d903974ec64b8a8c98565b\nDepends-On: I1109a303ee216ed8061959a81f0ed315bf138c80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/278647fe85335d15c52e85681629ae371b3bc944"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/278647fe85335d15c52e85681629ae371b3bc944"}]},"branch":"refs/heads/master"},"43a17cbf01c5ae563b143751b1fe284b062daccb":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-11-03 12:31:20.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/82/674482/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/82/674482/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/4"}}},"commit":{"parents":[{"commit":"4175a85a1609f093987d24903d7f6d1592ad7c7a","subject":"Testing basic Network policy IPBlock functionality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/4175a85a1609f093987d24903d7f6d1592ad7c7a"}]}],"author":{"name":"Genadi Chereshnya","email":"gcheresh@redhat.com","date":"2019-08-04 10:34:20.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-11-03 12:30:46.000000000","tz":120},"subject":"Check ingress and egress IPBlock rules for NP feature","message":"Check ingress and egress IPBlock rules for NP feature\n\nCreate 4 pods\nCheck the connectivity works before applying NP rules\nApply NP by allowing all and blocking pod1 ingress and pod2 egress\nCheck that connectivity from pod4 to pod1 and pod2 is blocked\nCheck that connectivity from pod4 to pod3 is still working\n\nChange-Id: Ida893c0ca6a340b342d903974ec64b8a8c98565b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/43a17cbf01c5ae563b143751b1fe284b062daccb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/43a17cbf01c5ae563b143751b1fe284b062daccb"}]},"branch":"refs/heads/master"},"05ec54044b5c666c34c64fff2fed4e3b739da153":{"kind":"REWORK","_number":5,"created":"2019-11-13 13:47:20.000000000","uploader":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"ref":"refs/changes/82/674482/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/82/674482/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/5"}}},"commit":{"parents":[{"commit":"4175a85a1609f093987d24903d7f6d1592ad7c7a","subject":"Testing basic Network policy IPBlock functionality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/4175a85a1609f093987d24903d7f6d1592ad7c7a"}]}],"author":{"name":"Genadi Chereshnya","email":"gcheresh@redhat.com","date":"2019-08-04 10:34:20.000000000","tz":180},"committer":{"name":"Genadi Chereshnya","email":"gcheresh@redhat.com","date":"2019-11-13 13:47:06.000000000","tz":120},"subject":"Check ingress and egress IPBlock rules for NP feature","message":"Check ingress and egress IPBlock rules for NP feature\n\nCreate 4 pods\nCheck the connectivity works before applying NP rules\nApply NP by allowing all and blocking pod1 ingress and pod2 egress\nCheck that connectivity from pod4 to pod1 and pod2 is blocked\nCheck that connectivity from pod4 to pod3 is still working\n\nChange-Id: Ida893c0ca6a340b342d903974ec64b8a8c98565b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/05ec54044b5c666c34c64fff2fed4e3b739da153"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/05ec54044b5c666c34c64fff2fed4e3b739da153"}]},"branch":"refs/heads/master"},"88416eaf6a2f69f75370dbb1d4a7c2e21b83686c":{"kind":"REWORK","_number":6,"created":"2019-11-20 15:40:44.000000000","uploader":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"ref":"refs/changes/82/674482/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/82/674482/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/6"}}},"commit":{"parents":[{"commit":"4175a85a1609f093987d24903d7f6d1592ad7c7a","subject":"Testing basic Network policy IPBlock functionality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/4175a85a1609f093987d24903d7f6d1592ad7c7a"}]}],"author":{"name":"Genadi Chereshnya","email":"gcheresh@redhat.com","date":"2019-08-04 10:34:20.000000000","tz":180},"committer":{"name":"Genadi Chereshnya","email":"gcheresh@redhat.com","date":"2019-11-20 15:40:33.000000000","tz":120},"subject":"Check ingress and egress IPBlock rules for NP feature","message":"Check ingress and egress IPBlock rules for NP feature\n\nCreate 4 pods\nCheck the connectivity works before applying NP rules\nApply NP by allowing all and blocking pod1 ingress and pod2 egress\nCheck that connectivity from pod4 to pod1 and pod2 is blocked\nCheck that connectivity from pod4 to pod3 is still working\n\nChange-Id: Ida893c0ca6a340b342d903974ec64b8a8c98565b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/88416eaf6a2f69f75370dbb1d4a7c2e21b83686c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/88416eaf6a2f69f75370dbb1d4a7c2e21b83686c"}]},"branch":"refs/heads/master"},"61e876be1d6a297b6e93af7168e7b7f992047ab3":{"kind":"REWORK","_number":7,"created":"2020-02-23 10:02:08.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/82/674482/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/82/674482/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/7"}}},"commit":{"parents":[{"commit":"681ff7a0cc2295a9d71923ac7a7da999c959a0fc","subject":"Basic IPv6 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/681ff7a0cc2295a9d71923ac7a7da999c959a0fc"}]}],"author":{"name":"Genadi Chereshnya","email":"gcheresh@redhat.com","date":"2019-08-04 10:34:20.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-02-23 10:01:58.000000000","tz":120},"subject":"[WIP] Check ingress and egress IPBlock rules for NP feature","message":"[WIP] Check ingress and egress IPBlock rules for NP feature\n\nCreate 4 pods\nCheck the connectivity works before applying NP rules\nApply NP by allowing all and blocking pod1 ingress and pod2 egress\nCheck that connectivity from pod4 to pod1 and pod2 is blocked\nCheck that connectivity from pod4 to pod3 is still working\n\nChange-Id: Ida893c0ca6a340b342d903974ec64b8a8c98565b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/61e876be1d6a297b6e93af7168e7b7f992047ab3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/61e876be1d6a297b6e93af7168e7b7f992047ab3"}]},"branch":"refs/heads/master"},"3c1eb555117a19c7c7efc51747d3ab2c9367e2ce":{"kind":"REWORK","_number":8,"created":"2020-02-25 07:19:18.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/82/674482/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/82/674482/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/8"}}},"commit":{"parents":[{"commit":"681ff7a0cc2295a9d71923ac7a7da999c959a0fc","subject":"Basic IPv6 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/681ff7a0cc2295a9d71923ac7a7da999c959a0fc"}]}],"author":{"name":"Genadi Chereshnya","email":"gcheresh@redhat.com","date":"2019-08-04 10:34:20.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-02-25 07:17:03.000000000","tz":120},"subject":"Check ingress and egress IPBlock rules for NP","message":"Check ingress and egress IPBlock rules for NP\n\nCreate 4 pods\nCheck the connectivity works before applying NP rules\nApply NP by allowing all and blocking pod1 ingress and pod2 egress\nCheck that connectivity from pod1 to pod4 is blocked\nCheck that connectivity from pod4 to pod2 is blocked\nCheck that connectivity from pod4 to pod3 is still working\n\nChange-Id: Ida893c0ca6a340b342d903974ec64b8a8c98565b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/3c1eb555117a19c7c7efc51747d3ab2c9367e2ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/3c1eb555117a19c7c7efc51747d3ab2c9367e2ce"}]},"branch":"refs/heads/master"},"5b18b867b63a6f42b2bc546d6a38dd2265c4b2ba":{"kind":"REWORK","_number":9,"created":"2020-02-25 08:31:31.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/82/674482/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/82/674482/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/9"}}},"commit":{"parents":[{"commit":"681ff7a0cc2295a9d71923ac7a7da999c959a0fc","subject":"Basic IPv6 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/681ff7a0cc2295a9d71923ac7a7da999c959a0fc"}]}],"author":{"name":"Genadi Chereshnya","email":"gcheresh@redhat.com","date":"2019-08-04 10:34:20.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-02-25 08:31:00.000000000","tz":120},"subject":"Check ingress and egress IPBlock rules for NP","message":"Check ingress and egress IPBlock rules for NP\n\nCreate 4 pods\nCheck the connectivity works before applying NP rules\nApply NP by allowing all and blocking pod1 ingress and pod2 egress\nCheck that connectivity from pod1 to pod4 is blocked\nCheck that connectivity from pod4 to pod2 is blocked\nCheck that connectivity from pod4 to pod3 is still working\n\nChange-Id: Ida893c0ca6a340b342d903974ec64b8a8c98565b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/5b18b867b63a6f42b2bc546d6a38dd2265c4b2ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/5b18b867b63a6f42b2bc546d6a38dd2265c4b2ba"}]},"branch":"refs/heads/master"},"723c5da1c2a05e67f1cfa0bcc3afbe15c81071f0":{"kind":"REWORK","_number":10,"created":"2020-02-25 11:03:07.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/82/674482/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/82/674482/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/10"}}},"commit":{"parents":[{"commit":"681ff7a0cc2295a9d71923ac7a7da999c959a0fc","subject":"Basic IPv6 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/681ff7a0cc2295a9d71923ac7a7da999c959a0fc"}]}],"author":{"name":"Genadi Chereshnya","email":"gcheresh@redhat.com","date":"2019-08-04 10:34:20.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-02-25 11:02:33.000000000","tz":120},"subject":"Check ingress and egress IPBlock rules for NP","message":"Check ingress and egress IPBlock rules for NP\n\nCreate 4 pods\nCheck that the http connection works before applying NP rules\nApply NP by allowing all and blocking pod1 ingress and pod2 egress\nCheck that http connection from pod1 to pod4 is blocked\nCheck that http connection from pod4 to pod2 is blocked\nCheck that http connection from pod4 to pod3 is still working\n\nChange-Id: Ida893c0ca6a340b342d903974ec64b8a8c98565b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/723c5da1c2a05e67f1cfa0bcc3afbe15c81071f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/723c5da1c2a05e67f1cfa0bcc3afbe15c81071f0"}]},"branch":"refs/heads/master"},"fcb806fd2746d32a7764399a9957a51c99806f06":{"kind":"REWORK","_number":11,"created":"2020-02-26 11:01:12.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/82/674482/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/82/674482/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/11"}}},"commit":{"parents":[{"commit":"681ff7a0cc2295a9d71923ac7a7da999c959a0fc","subject":"Basic IPv6 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/681ff7a0cc2295a9d71923ac7a7da999c959a0fc"}]}],"author":{"name":"Genadi Chereshnya","email":"gcheresh@redhat.com","date":"2019-08-04 10:34:20.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-02-26 11:00:55.000000000","tz":120},"subject":"Check ingress and egress IPBlock rules for NP","message":"Check ingress and egress IPBlock rules for NP\n\nCreate 4 pods\nCheck that the http connection works before applying NP rules\nApply NP by allowing all and blocking pod1 ingress and pod2 egress\nCheck that http connection from pod1 to pod4 is blocked\nCheck that http connection from pod4 to pod2 is blocked\nCheck that http connection from pod4 to pod3 is still working\n\nChange-Id: Ida893c0ca6a340b342d903974ec64b8a8c98565b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/fcb806fd2746d32a7764399a9957a51c99806f06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/fcb806fd2746d32a7764399a9957a51c99806f06"}]},"branch":"refs/heads/master"},"16c16907d8f40716705b949acd3f0705912019ad":{"kind":"REWORK","_number":12,"created":"2020-02-26 11:13:35.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/82/674482/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/82/674482/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/12"}}},"commit":{"parents":[{"commit":"681ff7a0cc2295a9d71923ac7a7da999c959a0fc","subject":"Basic IPv6 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/681ff7a0cc2295a9d71923ac7a7da999c959a0fc"}]}],"author":{"name":"Genadi Chereshnya","email":"gcheresh@redhat.com","date":"2019-08-04 10:34:20.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-02-26 11:13:02.000000000","tz":120},"subject":"Check ingress and egress IPBlock rules for NP","message":"Check ingress and egress IPBlock rules for NP\n\nCreate 4 pods\nCheck that the http connection works before applying NP rules\nApply NP by allowing all and blocking pod1 ingress and pod2 egress\nCheck that http connection from pod1 to pod4 is blocked\nCheck that http connection from pod4 to pod2 is blocked\nCheck that http connection from pod4 to pod3 is still working\n\nChange-Id: Ida893c0ca6a340b342d903974ec64b8a8c98565b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/16c16907d8f40716705b949acd3f0705912019ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/16c16907d8f40716705b949acd3f0705912019ad"}]},"branch":"refs/heads/master"},"129746a24e4eafcc06cd8b9fcf9ab8dd43b97af1":{"kind":"REWORK","_number":13,"created":"2020-02-26 14:36:15.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/82/674482/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/82/674482/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/13"}}},"commit":{"parents":[{"commit":"681ff7a0cc2295a9d71923ac7a7da999c959a0fc","subject":"Basic IPv6 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/681ff7a0cc2295a9d71923ac7a7da999c959a0fc"}]}],"author":{"name":"Genadi Chereshnya","email":"gcheresh@redhat.com","date":"2019-08-04 10:34:20.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-02-26 14:34:22.000000000","tz":120},"subject":"Check ingress and egress IPBlock rules for NP","message":"Check ingress and egress IPBlock rules for NP\n\nCreate 4 pods\nCheck that the http connection works before applying NP rules\nApply NP by allowing all and blocking pod1 ingress and pod2 egress\nCheck that http connection from pod1 to pod4 is blocked\nCheck that http connection from pod4 to pod2 is blocked\nCheck that http connection from pod4 to pod3 is still working\n\nChange-Id: Ida893c0ca6a340b342d903974ec64b8a8c98565b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/129746a24e4eafcc06cd8b9fcf9ab8dd43b97af1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/129746a24e4eafcc06cd8b9fcf9ab8dd43b97af1"}]},"branch":"refs/heads/master"},"d5af22838da1b3f5fe65b2ef2d1821b93983c7f2":{"kind":"REWORK","_number":14,"created":"2020-02-27 13:33:18.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/82/674482/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/82/674482/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/14"}}},"commit":{"parents":[{"commit":"681ff7a0cc2295a9d71923ac7a7da999c959a0fc","subject":"Basic IPv6 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/681ff7a0cc2295a9d71923ac7a7da999c959a0fc"}]}],"author":{"name":"Genadi Chereshnya","email":"gcheresh@redhat.com","date":"2019-08-04 10:34:20.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-02-27 13:33:08.000000000","tz":120},"subject":"Check ingress and egress IPBlock rules for NP","message":"Check ingress and egress IPBlock rules for NP\n\nCreate 4 pods\nCheck that the http connection works before applying NP rules\nApply NP by allowing all and blocking pod1 ingress and pod2 egress\nCheck that http connection from pod1 to pod4 is blocked\nCheck that http connection from pod4 to pod2 is blocked\nCheck that http connection from pod4 to pod3 is still working\n\nChange-Id: Ida893c0ca6a340b342d903974ec64b8a8c98565b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d5af22838da1b3f5fe65b2ef2d1821b93983c7f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d5af22838da1b3f5fe65b2ef2d1821b93983c7f2"}]},"branch":"refs/heads/master"},"99fdd4d86127143ba2194d98a5ffa877af9e4892":{"kind":"REWORK","_number":15,"created":"2020-03-02 08:08:29.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/82/674482/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/82/674482/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/15"}}},"commit":{"parents":[{"commit":"681ff7a0cc2295a9d71923ac7a7da999c959a0fc","subject":"Basic IPv6 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/681ff7a0cc2295a9d71923ac7a7da999c959a0fc"}]}],"author":{"name":"Genadi Chereshnya","email":"gcheresh@redhat.com","date":"2019-08-04 10:34:20.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-03-02 08:08:20.000000000","tz":120},"subject":"Check ingress and egress IPBlock rules for NP","message":"Check ingress and egress IPBlock rules for NP\n\nCreate 4 pods\nCheck that the http connection works before applying NP rules\nApply NP by allowing all and blocking pod1 ingress and pod2 egress\nCheck that http connection from pod1 to pod4 is blocked\nCheck that http connection from pod4 to pod2 is blocked\nCheck that http connection from pod4 to pod3 is still working\n\nChange-Id: Ida893c0ca6a340b342d903974ec64b8a8c98565b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/99fdd4d86127143ba2194d98a5ffa877af9e4892"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/99fdd4d86127143ba2194d98a5ffa877af9e4892"}]},"branch":"refs/heads/master"},"d88e00caa11f9f499af060d5489d3d8c43de965b":{"kind":"REWORK","_number":16,"created":"2020-03-02 14:22:11.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/82/674482/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/82/674482/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/674482/16"}}},"commit":{"parents":[{"commit":"681ff7a0cc2295a9d71923ac7a7da999c959a0fc","subject":"Basic IPv6 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/681ff7a0cc2295a9d71923ac7a7da999c959a0fc"}]}],"author":{"name":"Genadi Chereshnya","email":"gcheresh@redhat.com","date":"2019-08-04 10:34:20.000000000","tz":180},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-03-02 14:22:03.000000000","tz":60},"subject":"Check ingress and egress IPBlock rules for NP","message":"Check ingress and egress IPBlock rules for NP\n\nCreate 4 pods\nCheck that the http connection works before applying NP rules\nApply NP by allowing all and blocking pod1 ingress and pod2 egress\nCheck that http connection from pod1 to pod4 is blocked\nCheck that http connection from pod4 to pod2 is blocked\nCheck that http connection from pod4 to pod3 is still working\n\nChange-Id: Ida893c0ca6a340b342d903974ec64b8a8c98565b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d88e00caa11f9f499af060d5489d3d8c43de965b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d88e00caa11f9f499af060d5489d3d8c43de965b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
