)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~676368","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~I497f34d3484b1acb106c3979dba5749c9d1ca0ab","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"restart","hashtags":[],"change_id":"I497f34d3484b1acb106c3979dba5749c9d1ca0ab","subject":"Use label selector to get kuryr controller pods","status":"MERGED","created":"2019-08-14 08:22:23.000000000","updated":"2019-08-26 18:00:15.000000000","submitted":"2019-08-26 17:58:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676368-1566842282185-f1a8958c","meta_rev_id":"d4b61b8ba2fef08f88505beeb2b8daa143e98504","_number":676368,"virtual_id_number":676368,"owner":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"date":"2019-08-26 14:16:09.000000000","_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":2,"date":"2019-08-26 17:57:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2019-08-26 08:22:25.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":1,"date":"2019-08-26 08:54:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-26 08:25:05.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-26 08:25:05.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-26 08:22:25.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2019-08-26 08:25:05.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2019-08-26 08:54:03.000000000","updated_by":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"reviewer":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"state":"REVIEWER"},{"updated":"2019-08-26 17:57:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"475d6b43de73182b3ee76f6a891cb2a41a6e3cc3","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-08-14 08:22:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"213fe67e137faafef72389fb7524a6d7369c2436","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 10:35:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42e2ad089e934125a6c66be96d8fd7b8 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e592e294d96b4f1ba6ef7fa7fed43d5e : SUCCESS in 3m 29s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/a5ff42a51fec4f64b7ae5c922d9353d0 : SUCCESS in 1h 22m 49s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/1c9b9d706a5a489b86fd3b34e56ddc16 : FAILURE in 1h 37m 16s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/d0dcb114bdf14dfa8a54ab6c41725145 : SUCCESS in 1h 30m 08s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/2781ac7c4e0e40ff8d89210bf0ab53d9 : SUCCESS in 1h 38m 11s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/20c979781bb84ca880bcc151ae28b973 : SUCCESS in 1h 30m 21s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/d3d4290beada4e1eb4199693217f3182 : SUCCESS in 1h 29m 04s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/5e8fb08ba7024acaaff58ef0d59ced29 : SUCCESS in 1h 48m 02s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/7d9405f1e57043c29c27d726ee40bafb : SUCCESS in 2h 10m 06s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/1e9dfa7a6f92460c98cc194f1ef1432a : SUCCESS in 1h 45m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cc2802fba2de454de71959eb46b8c1399fa60200","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-08-14 12:12:45.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"0e203b47fc595da650428e907ddaeb8cad44d63e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 15:23:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/01609d7a64754ed5b8a2a0fd33b748ee : TIMED_OUT in 3h 08m 37s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/a88f12769e3041df9b21a668384fbc9d : FAILURE in 1h 34m 14s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/7c881e5adf2b498d80cef6dcfc52e3b6 : FAILURE in 1h 22m 11s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/aceeb43cebf6441b9b162efad936b91b : SUCCESS in 1h 00m 21s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/e52997e719444363a5fe41738487f223 : FAILURE in 2h 32m 01s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/65321701782d4f488128bccf8f73c0fc : SUCCESS in 1h 27m 05s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/cf36b41957534b7aa70290c378f11417 : TIMED_OUT in 3h 04m 11s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/b1364a297f4648b0be4ca68c1b1371f0 : FAILURE in 1h 36m 02s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/397cbbb24cf74c9398982d2051b910c9 : TIMED_OUT in 2h 15m 45s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/fd0c9c3798c84477bb60bb6597ed3fc8 : FAILURE in 31m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"67da5d3389de676ffa57ffde227fe6d92ac3606c","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-08-20 10:04:38.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"7c62f584557db60083577b3ac12864a5562740b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 13:15:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/3eadc752802f4d9fadebe36cfd93a644 : FAILURE in 1h 40m 49s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/9a7d0db3565d4829bc1fc1184490fff6 : FAILURE in 2h 03m 34s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/6a295e6a32c84011bbc05fc758a3a723 : FAILURE in 2h 08m 07s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/f9e59b4dedab47af80337e3361f2ae58 : SUCCESS in 1h 06m 55s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/d66bb5d1766a44048e2f51f961704995 : FAILURE in 2h 38m 37s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/66f1711892ad46d2b634aeddc6dcd675 : SUCCESS in 1h 18m 27s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/58366a180ce94bb1866a6dfb2de89c3e : TIMED_OUT in 3h 05m 47s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/012078559aa54cdb994e8dd587ce35cf : FAILURE in 1h 40m 33s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/748b088b9b52490b8de43c69ab777f57 : TIMED_OUT in 2h 18m 05s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/739f5a5f7704492487b87b1091a32ee3 : FAILURE in 29m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"31332f5aa835f32dbd96eeaeb74be6eb797570f1","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-08-20 13:20:51.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c585fc0e9c778706faf713d8f99fab83495d0cac","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-08-20 13:59:11.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"f64479b39e7e58ae9be73249f8e2a3b676a60451","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-08-20 14:21:54.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4b812e9d5df599e7628e1065353143d580f77f4a","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-08-20 15:34:08.000000000","message":"Patch Set 3: Code-Review+2\n\nLooks fine to me.","accounts_in_message":[],"_revision_number":3},{"id":"996b30e12501fc995ca9e18aef453fb448a4ebe2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 17:10:00.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ff7fb85c8db44eca9876ad4c83e7872 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a820b25f0a324e769612a0727e0013ef : SUCCESS in 5m 12s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/5723007305b643deac52a45099bd7b96 : SUCCESS in 1h 16m 59s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/260c4aad1f414db1896ba452ed46a8db : FAILURE in 1h 52m 41s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/4e5bc79016d14f6c950f43cf208cbc4d : SUCCESS in 1h 31m 12s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/5c76472f677c4d43ac8e9a5a97eceaa0 : SUCCESS in 1h 16m 36s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/279efdd662ca4ee4b39388ef0181dad0 : TIMED_OUT in 3h 05m 11s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/88a78ca1d41e4a41a6a44ba1441a1df2 : FAILURE in 1h 14m 51s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/977dc45f30264588b24dc50f4ced02dc : SUCCESS in 1h 41m 37s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/3f7f0cd2f12f4cb4b80ced0b3a8613d6 : SUCCESS in 1h 59m 40s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/9ca82e9760914d36acc993fc3ac31d66 : SUCCESS in 1h 31m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cc99efe929cd936f0dfb12c7f7ed919ce84f703c","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-08-21 06:02:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8f3e2cb5ba39f7064913aa10faf89187f05d5c84","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-08-21 06:02:13.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4d8b1c999a9833089b0c1d7151bdc74b7be3f5c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 08:09:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82cb5c390ca9451496c33a1384dd05ba : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/132cc45bc7994d9b93332150c72bf0a5 : SUCCESS in 6m 29s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/6672f509f07d414a88e266f98003a95d : SUCCESS in 1h 15m 24s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/451b5ae8de4d4695b8e1a27ce161b441 : FAILURE in 1h 44m 47s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/d906246120ed4febbf0171b613271643 : FAILURE in 45m 07s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/ff275b5e032240aa94c628bbb7da2702 : SUCCESS in 1h 20m 43s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/751e2a0bf95a4cff8d53f0380275dad6 : SUCCESS in 1h 44m 59s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/2b7f0eae932348198ce5719b0770b033 : SUCCESS in 1h 26m 21s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/357c3da5ddbe4a03bc595b1ca791334f : SUCCESS in 1h 42m 12s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/a485c560fbf449e784e1f9b14506d509 : SUCCESS in 1h 53m 41s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/2103629420614487a1c34e15801baba7 : SUCCESS in 1h 31m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"46975f2408c74803dc80f9d5e7f531994ef33142","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-08-21 09:38:27.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)\n\nAre we setting that label on the controller?","accounts_in_message":[],"_revision_number":4},{"id":"e97d5618add408ef34297e3463810ce446dbfd5e","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-08-21 12:43:55.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"c08a5f599cc7e1683b9140a13e89370c011622a1","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-08-21 12:44:46.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"aee4ab5149afedfa3c4e566797aa9c64feae3db1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 15:13:03.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bee57a695e724b3aa7ac4224c5699c04 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2c8943f513b47ae88a7cb31489a6726 : SUCCESS in 4m 40s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/e95398db27754c1fa2b5a5d7a6ad287e : SUCCESS in 1h 04m 46s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/74927dcbfbe24b86a606aad07148245f : FAILURE in 1h 52m 04s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/06ad842df5784268b94bce9279f21967 : SUCCESS in 1h 19m 50s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/feb8e27d24524b16821c15f488df4772 : SUCCESS in 1h 39m 57s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/a8934e12dbe842bc981ca9e387264ef3 : SUCCESS in 1h 39m 37s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/3fdf883dce8049608ba1d824f2f75b2c : SUCCESS in 1h 19m 44s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/41dd75055c8443278e014e5c820dad9d : SUCCESS in 1h 24m 43s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/75dcfbb404bf48dda6a8cb8b67a1d738 : SUCCESS in 1h 55m 29s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/8ef1ef6e98f24d8daab1fcb19d71d81e : SUCCESS in 1h 22m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"dca3bd8fce4a46882fdd11117f2f0458368b8535","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-08-21 19:24:58.000000000","message":"Patch Set 5:\n\ncheck experimental","accounts_in_message":[],"_revision_number":5},{"id":"32b9f2628cd85c05189523261b4c7807bda998ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 22:37:24.000000000","message":"Patch Set 5:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/8bc12db9b4e3419d8db338facb607844 : FAILURE in 2h 04m 32s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/71df7af23aeb4301944d0f0f3b01b24a : FAILURE in 1h 45m 17s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/0b5d42c9f57b4c4ab724af90c83dfad5 : FAILURE in 2h 00m 49s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/396321a8768e427fa740f5bec8040453 : SUCCESS in 58m 53s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/71bc3e34dec64f538f64f7475a8d51ee : FAILURE in 2h 42m 08s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/b6fd08ba82964908b43896fce542de66 : SUCCESS in 1h 12m 57s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/e2999ecd958449d9bb4d0b72bc0b406a : TIMED_OUT in 3h 06m 53s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/5d0a4a7ed9ba441e92aafe3e0e940a89 : FAILURE in 2h 00m 14s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/a695e9179e3e4ce28496a9493912bb21 : TIMED_OUT in 2h 21m 01s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/06d1a5c73cd74e73a73d93e621b181e8 : FAILURE in 34m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"30a6254dd1b73bcc55e78e86d887323d3311eced","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-08-22 07:45:11.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4c10b7b23d13d26832ad4e4519426e6953211771","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-08-26 08:22:25.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"c1dd67307f0a88799fd731dca37cd6e17e541429","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-08-26 08:25:05.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"403059087c28820312a28a19fd1df3bae59d8a09","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 08:25:17.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"6aeae8971a5a68a0638bc12121e6ba077be06379","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-08-26 08:54:03.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"bb394d1e58370d94fe9c63cbfaab61dfa855ab47","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 10:37:27.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4da504731ea543ffa594f5c16c72dc81 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b87a77255103455bbddc0fee33809b9d : SUCCESS in 4m 33s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/fc83285873bc4ea3a184e5600e7b8f6e : SUCCESS in 1h 25m 28s\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/34c88ebe85ed4eacbb03fa9a39fc346c : POST_FAILURE in 2h 04m 40s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/be5ad96dd9634f25ad4c4108991bdd28 : SUCCESS in 1h 20m 10s","accounts_in_message":[],"_revision_number":5},{"id":"129c448a8e6f4a9a68f197e029680e523eedae5d","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-08-26 14:16:09.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"38d925f571b3a45ce4fa5b4a13874c01dd631cf9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 16:07:54.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ccb2ed253d2421da62db6495b8f7063 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00827bec3fb0411e82beb016926ad63f : SUCCESS in 5m 14s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/793dfa45cc51445ab563effe33a9afd2 : SUCCESS in 1h 11m 08s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/148909df1fae43ff89e2ea222a6ce9f6 : FAILURE in 1h 47m 14s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/1a9042e4ba2a4a6daf6cb444ef5de582 : SUCCESS in 1h 31m 19s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/b1faeef77711440295f0bad85cffa9c4 : SUCCESS in 1h 40m 09s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/afaa157ac10a4a3bbab90495ec7f00ce : FAILURE in 1h 41m 07s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/c76a4ad2b2d54aea83f8edc92ab747e9 : SUCCESS in 1h 31m 24s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/88b7a205fb054e6a9390ff89c5d959c9 : SUCCESS in 1h 36m 07s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/91b9313899374fe4a3dd8dd7222d22e9 : SUCCESS in 1h 38m 49s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/998f236deb7741438e5b26496628a2b5 : SUCCESS in 1h 34m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1452bc4144d730432e28084ab64928aa965ce6bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 16:08:10.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b341f3c9ca1668a07bffcd12b9190d81dd897701","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 17:57:58.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd3086bca02841b9970bbf103457d9d9 : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/455db35ff2f1403695c905856445b14b : SUCCESS in 3m 36s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/a15f4cf00f7747b89a870d745f5dfe6e : SUCCESS in 1h 26m 37s\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/a49914ebf1c140049c0303e7028eaeb2 : SUCCESS in 1h 32m 01s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/5b60059673084f65825f32146cb005a3 : SUCCESS in 1h 42m 38s","accounts_in_message":[],"_revision_number":5},{"id":"34b082ef6fd892bf1dc7e767994df38039e04003","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 17:58:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"f79a4395262e8d32700b100481e39ca653f4f176","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 18:00:15.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff0212a3eaaf4c52a9a84e419b830434 : SUCCESS in 1m 39s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"46db32a85df1a16daf9cfa829aee84e600fa4467","revisions":{"5cba64ba74bf4daf2494af6098022e3ea6f2bce0":{"kind":"REWORK","_number":1,"created":"2019-08-14 08:22:23.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/68/676368/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/68/676368/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/68/676368/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/68/676368/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/68/676368/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/68/676368/1"}}},"commit":{"parents":[{"commit":"9450ef731a485c9dd5680c08d4d7ce849b355190","subject":"Check connectivity to an existing pod in test_ha","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/9450ef731a485c9dd5680c08d4d7ce849b355190"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-08-14 08:21:15.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-08-14 08:22:06.000000000","tz":180},"subject":"[WIP] Use label selector to get kuryr controller pods","message":"[WIP] Use label selector to get kuryr controller pods\n\nChange-Id: I497f34d3484b1acb106c3979dba5749c9d1ca0ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/5cba64ba74bf4daf2494af6098022e3ea6f2bce0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/5cba64ba74bf4daf2494af6098022e3ea6f2bce0"}]},"branch":"refs/heads/master"},"b2e7abfae24c4eab0360eab6edcdd1063319d795":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-08-20 13:20:51.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/68/676368/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/68/676368/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/68/676368/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/68/676368/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/68/676368/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/68/676368/2"}}},"commit":{"parents":[{"commit":"9450ef731a485c9dd5680c08d4d7ce849b355190","subject":"Check connectivity to an existing pod in test_ha","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/9450ef731a485c9dd5680c08d4d7ce849b355190"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-08-14 08:21:15.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-08-20 13:20:39.000000000","tz":0},"subject":"Use label selector to get kuryr controller pods","message":"Use label selector to get kuryr controller pods\n\nChange-Id: I497f34d3484b1acb106c3979dba5749c9d1ca0ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/b2e7abfae24c4eab0360eab6edcdd1063319d795"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/b2e7abfae24c4eab0360eab6edcdd1063319d795"}]},"branch":"refs/heads/master"},"f545fe90ed928575e47c5d74c1a8897976680df2":{"kind":"NO_CHANGE","_number":3,"created":"2019-08-20 13:59:11.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/68/676368/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/68/676368/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/68/676368/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/68/676368/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/68/676368/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/68/676368/3"}}},"commit":{"parents":[{"commit":"9450ef731a485c9dd5680c08d4d7ce849b355190","subject":"Check connectivity to an existing pod in test_ha","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/9450ef731a485c9dd5680c08d4d7ce849b355190"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-08-14 08:21:15.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-08-20 13:59:09.000000000","tz":180},"subject":"Use label selector to get kuryr controller pods","message":"Use label selector to get kuryr controller pods\n\nChange-Id: I497f34d3484b1acb106c3979dba5749c9d1ca0ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/f545fe90ed928575e47c5d74c1a8897976680df2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/f545fe90ed928575e47c5d74c1a8897976680df2"}]},"branch":"refs/heads/master"},"381f65f3b8593b595030926cac4cdbe18dcd04fd":{"kind":"REWORK","_number":4,"created":"2019-08-21 06:02:13.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/68/676368/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/68/676368/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/68/676368/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/68/676368/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/68/676368/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/68/676368/4"}}},"commit":{"parents":[{"commit":"9450ef731a485c9dd5680c08d4d7ce849b355190","subject":"Check connectivity to an existing pod in test_ha","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/9450ef731a485c9dd5680c08d4d7ce849b355190"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-08-14 08:21:15.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-08-21 06:01:51.000000000","tz":180},"subject":"Use label selector to get kuryr controller pods","message":"Use label selector to get kuryr controller pods\n\nChange-Id: I497f34d3484b1acb106c3979dba5749c9d1ca0ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/381f65f3b8593b595030926cac4cdbe18dcd04fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/381f65f3b8593b595030926cac4cdbe18dcd04fd"}]},"branch":"refs/heads/master"},"46db32a85df1a16daf9cfa829aee84e600fa4467":{"kind":"REWORK","_number":5,"created":"2019-08-21 12:44:46.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/68/676368/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/68/676368/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/68/676368/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/68/676368/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/68/676368/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/68/676368/5"}}},"commit":{"parents":[{"commit":"9450ef731a485c9dd5680c08d4d7ce849b355190","subject":"Check connectivity to an existing pod in test_ha","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/9450ef731a485c9dd5680c08d4d7ce849b355190"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-08-14 08:21:15.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-08-21 12:44:38.000000000","tz":180},"subject":"Use label selector to get kuryr controller pods","message":"Use label selector to get kuryr controller pods\n\nChange-Id: I497f34d3484b1acb106c3979dba5749c9d1ca0ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/46db32a85df1a16daf9cfa829aee84e600fa4467"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/46db32a85df1a16daf9cfa829aee84e600fa4467"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
