)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~683122","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~Ic79f307d622affbab196ac37611de7d579f69293","project":"openstack/kuryr-tempest-plugin","branch":"master","hashtags":[],"change_id":"Ic79f307d622affbab196ac37611de7d579f69293","subject":"Revert \"Use label selector to get kuryr pods\"","status":"ABANDONED","created":"2019-09-19 12:11:08.000000000","updated":"2019-09-24 12:06:58.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7b881e2d2c3e613f9830b2178f40bcc52d5f8c0c","_number":683122,"virtual_id_number":683122,"owner":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"date":"2019-09-24 08:13:12.000000000","_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-09-24 11:22:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":1,"date":"2019-09-23 13:15:23.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},{"value":-1,"date":"2019-09-24 09:26:33.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"date":"2019-09-23 13:15:23.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-19 12:11:42.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2019-09-19 12:12:13.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"},{"updated":"2019-09-19 12:17:12.000000000","updated_by":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"reviewer":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"state":"REVIEWER"},{"updated":"2019-09-23 11:36:18.000000000","updated_by":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"reviewer":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"state":"REVIEWER"},{"updated":"2019-09-24 09:26:33.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2019-09-24 11:22:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bf919a211351e0f7e66f3be7a3dbdb846c0af512","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-09-19 12:11:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e0db56b902dad9809a88ffb7e3ac775fdd7c963","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-09-19 12:15:30.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"b923cadd92e0d086876fa55293cbcaa0b7d29209","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-09-19 12:17:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6027dba980e620a1c60fab76d277cc3f0fca7839","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 13:04:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/158fe7bc35944fa3aa582b0dcf8557e1 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2efbd25194e4be5a833785ca81d8016 : SUCCESS in 4m 39s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/7d723646ca514195b7ee668b1f60ce80 : FAILURE in 39m 29s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/35fca28e21f941938da3d9b2874c52fe : FAILURE in 37m 46s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/c61b178bb1f948a58cbd225a9d56b869 : FAILURE in 36m 30s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/6a8941aac5ed4ff18739f95d18d557f1 : FAILURE in 33m 39s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/85cb3522b2654d2c922246f253b2c752 : FAILURE in 34m 27s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/aca98e5a6d2640a69e52f610a5354946 : FAILURE in 33m 46s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/09e79db2c83a43b18b59bf42c8c4857b : FAILURE in 30m 48s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/8eec477922834f23b5fd22e6835a9a47 : FAILURE in 49m 30s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/f8a7f06e592d45c6affe49be184c181d : FAILURE in 41m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"27396ed96e613cf5eec68b6c70ad495473e47336","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 13:06:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/9af54d65647941e2b7079406173910d9 : FAILURE in 33m 39s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/127b89b1a24545829165ec3154a50eb3 : FAILURE in 43m 13s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/7fc368d49e02459ba4002c19fdfe78c5 : FAILURE in 34m 29s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/4a7f9f8c4cfe425a914c926d20973da5 : FAILURE in 37m 33s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/cd35cb53a04c422f82648ee3b6612947 : FAILURE in 34m 44s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/498557a26efb493abb7504fcd3fcaa74 : FAILURE in 38m 48s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/4dd8f1005227425483f3966e2bf124cc : FAILURE in 43m 21s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/5414f68ffe344c329e6e872409010fde : FAILURE in 34m 57s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/31da5467fbf145d0b0a16123f8f32912 : FAILURE in 41m 59s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/e1d819ee82fd49e3ad2705720077559b : FAILURE in 34m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5b73f48c7cff71758bba7379e0d677c513148577","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-09-19 16:34:34.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1952d236a9b75a9dcd0eb16748f26e6ba440a0f4","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-09-19 16:34:43.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"62c4e2c411576624f75e0ebff6f32290adb5a9f6","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-09-19 16:34:52.000000000","message":"Patch Set 2:\n\ncheck experimental","accounts_in_message":[],"_revision_number":2},{"id":"13354425a7166c30763288d5d80113af6030a5c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 18:43:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c34a96b30d944c0854df51d2d22aa3e : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b8868a37bae491c8261083b32533258 : SUCCESS in 4m 23s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/41e177dafa9d46f0b573cd645edc19ce : SUCCESS in 1h 07m 40s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/d758e669e2074d86aab9403be251b323 : FAILURE in 38m 54s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/b183c7caddb5436eb981363ff5a3722c : SUCCESS in 1h 45m 46s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/b12eb81763e946f6a498c8f2bc1c80f2 : SUCCESS in 2h 00m 33s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/b359563261a8425596038cec0d2b8418 : SUCCESS in 1h 17m 50s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/a7b6fd697262439eae24f12035ed2435 : SUCCESS in 1h 08m 02s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/6bf2135d2bca40b5a9be5f945dbb9a3d : FAILURE in 1h 36m 39s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/59e920d8bf75439d98b72b393abb5036 : SUCCESS in 1h 45m 47s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/27299c1ba4f04d3bb896842dc9e7fd68 : FAILURE in 23m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"91c5ffb4299b86bbd05498d229ad2ff72a947d28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 18:58:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/48e46be29c5743b9a80820c1eff430e9 : FAILURE in 34m 28s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/c4a4edcb05c7482289801268cac14f1b : FAILURE in 2h 20m 50s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/0507e9465be348b89742ca05cc97a6e0 : FAILURE in 33m 36s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/6d581e7013564823bbdc23259c67b643 : FAILURE in 32m 33s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/3b92aecd5d57443f94c6a127092b56fa : FAILURE in 33m 57s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/c433f0297fa74b1c93ec481cdea550a1 : SUCCESS in 1h 10m 23s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/df663fbbe32748d6bd8c9562052ffc8c : FAILURE in 34m 33s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/5baeb37808b4452892e117e8e67b856f : FAILURE in 32m 36s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/80bb77581c404f34a4e8278b3081ce05 : TIMED_OUT in 2h 16m 22s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/7ad558d9675647d8881a0be330c73553 : FAILURE in 31m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2c83fdf7719a8d4e356a31a59023817e27dbd075","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-09-23 09:12:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"257160e215af1ea8c29ec88ca798676530d1406c","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-09-23 09:12:10.000000000","message":"Patch Set 2:\n\ncheck experimental","accounts_in_message":[],"_revision_number":2},{"id":"663c5470f31b7fd89be5cba4c7811c1e40634258","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 11:00:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ed1a5cb54f14ec0ac6b78173d8f26c0 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/669e25526edd46d090d571c6dff6b4d0 : SUCCESS in 2m 50s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/1934641ead514b4fbca9e11875ae35d6 : SUCCESS in 1h 32m 28s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/dd7b5c47b2c441979e979bc82c1b7ebb : FAILURE in 41m 09s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/c7fa28a2b9854993b3a3d6de8b69a27a : SUCCESS in 1h 25m 57s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/4c2d044211ac45f3bbb916e381ad149f : FAILURE in 1h 38m 46s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/d2eb98bbab924c009dbb0dc663ce5665 : SUCCESS in 1h 25m 20s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/a857ab46546f46d3a845478b55701474 : SUCCESS in 1h 11m 07s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/98f2d8b7401c47e79ae671246647cc72 : SUCCESS in 1h 36m 23s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/14ea0b4b8f3c41359aa381bd53d0467f : SUCCESS in 1h 42m 24s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/ed206bca9d6a4f009aac20d563fd26e2 : SUCCESS in 1h 19m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f33417a471d2ccd638e27a551390b440847864df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 11:49:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/b3b69ba1b1784bb7a63244551801ddbd : FAILURE in 33m 53s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/32be770d20064244ae5ac4d4a2cb14ac : FAILURE in 2h 02m 51s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/32ad4e4cb55542eaa06a37ad27e3e4bb : FAILURE in 41m 59s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/e9f5a227bdc8486e88b70b6256cce353 : FAILURE in 35m 33s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/6a10756dfc1343f599bd89d7d0a9b3b7 : FAILURE in 37m 47s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/29cf81b161354b16946666921d6e6bdc : SUCCESS in 1h 14m 57s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/f9f4df244fd04081b921ac2fefac204c : FAILURE in 37m 26s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/a5cbbb680aee46e381ef5403038c8e31 : FAILURE in 34m 40s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/6d6408bd23334edc8c75e749e5cb2220 : TIMED_OUT in 2h 17m 51s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/715a21224fa54b7695871d816d9419be : FAILURE in 32m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f1ec719c0f5add57974ae7d046227a38a78ac796","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-09-23 13:15:23.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"cd45d0bf60d63dcf59220a1f3d8e1a3a18d870e2","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-09-23 13:15:35.000000000","message":"Patch Set 3:\n\ncheck experimental","accounts_in_message":[],"_revision_number":3},{"id":"433530a6f6104daf790694d642dc52160b999e9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 15:32:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/418c8dad871b4f04a26a996da51f754c : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32c9247804b14a7ea59e91de0aa2f3c4 : SUCCESS in 2m 34s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/c4a1b7f2d0694f219400b564b2f129e7 : SUCCESS in 1h 17m 39s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/8fa479b09fd046b9a88525c8d82f9843 : FAILURE in 1h 53m 21s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/6956c523085a40629aca74844fb9465a : SUCCESS in 1h 40m 48s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/6e464c81a8ee488196df02b1c18f4940 : SUCCESS in 1h 18m 36s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/87a376e366a14d608fb01b460361e264 : SUCCESS in 1h 55m 55s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/d8f286e448134ad193a24abe5af6f697 : SUCCESS in 1h 30m 44s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/36db1549c44546bbb0daa8a771985e18 : FAILURE in 1h 39m 45s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/1bc915a046604e8e9b32ed2cc19126bc : SUCCESS in 1h 40m 28s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/7706bc578e64413cb634a8e1eaa27f89 : SUCCESS in 1h 24m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6dfcd5c15942885be95bd45d93d8806c7c167f50","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 16:45:43.000000000","message":"Patch Set 3:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/e3e124d2f933485ea8caadb0f4c8cf80 : FAILURE in 2h 01m 42s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/7232002e38bd428b86827103a1a0534a : FAILURE in 1h 51m 25s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/11e87b5654504dfd81311f560dffc25d : FAILURE in 2h 26m 30s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/efda584d917648979befd35de1ea4945 : SUCCESS in 57m 46s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/b7b25e2d1cfb4811a7748a0043f375de : TIMED_OUT in 3h 04m 43s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/9d15c980f57942a79ca8bac587e97a17 : SUCCESS in 1h 42m 42s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/e7976292a0324ede8cc1054dc61fd9f8 : FAILURE in 2h 10m 51s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/f607831519804dfa902dfce069f0edc6 : FAILURE in 2h 49m 54s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/b4e8ea1fc9d74218adac72e46414e4f3 : TIMED_OUT in 2h 17m 19s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/1d9c4b5cd7064ecfa147a69127cdbfe0 : FAILURE in 31m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"429b23f7a761db06d1247496b1fb330ecc85649f","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-09-24 05:49:00.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5ba0a9aa38579c199386cf6af68677d9a81920a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 08:07:13.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0934b87395f4a8bb2f4b996387ed445 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5287ab19bb64d6f823cd76e832506eb : SUCCESS in 2m 29s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/f80ad0d7ca814d81a7bb9782f71e6bd4 : SUCCESS in 1h 45m 44s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/913a85f5738445f68fd3c5b37857ad15 : FAILURE in 2h 06m 06s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/783e8e0edf0f4456a3d2fd4c97560e35 : FAILURE in 1h 57m 56s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/d3b90344f3004f70af041d793975268a : SUCCESS in 1h 40m 27s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/5d5392d6fd5b435bae4fbae8f80ac999 : FAILURE in 27m 34s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/929b03457ada48ee9c1aca8d1e1db223 : FAILURE in 1h 01m 22s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/a95fba61426d4784a97d3fcaee59c336 : SUCCESS in 1h 47m 01s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/3498cbdebba44459961086a07258f7cc : SUCCESS in 2h 11m 34s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/4ab048cf3c3c4c8ea337ea11e3f30d35 : SUCCESS in 1h 54m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2fedb50f35ceb16f852dbfc5f154722b0149ad5b","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-09-24 08:13:12.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ce952d5e6aec264dffd7a541978bf7194e246d19","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-09-24 09:26:33.000000000","message":"Patch Set 3: Code-Review-1\n\nWhy not adding the label instead of reverting?","accounts_in_message":[],"_revision_number":3},{"id":"55c639d9318d7f28897e67a4f35e3174f851ec72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 11:22:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6db6879c203148838a2c3000d5765191 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d69f1d0584dd4340af2444da7d75e5ce : SUCCESS in 3m 23s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/754b6ecbdb5848ed826021108b738757 : SUCCESS in 1h 45m 01s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/0a0afd5b7fdf45a49822402de87a165e : FAILURE in 2h 11m 28s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/0eae4a9bd1aa48c9a2ea1a71241698ee : FAILURE in 1h 44m 56s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/7800c9df9d5746368d09e599b7cf6a24 : SUCCESS in 1h 30m 08s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/5c60ce8ff1b74938b1e322da698c867b : TIMED_OUT in 3h 07m 23s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/a7411d078f7742c5a49337f53afd9b66 : SUCCESS in 1h 23m 31s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/958cd1d8ebe94763b08c37bcad1db5d7 : SUCCESS in 1h 34m 13s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/c96ba46f7143490298d45a82ae8bc2dd : FAILURE in 1h 47m 31s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/1d59779b0c9a4d6089052b72a81cdb62 : SUCCESS in 2h 05m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7b881e2d2c3e613f9830b2178f40bcc52d5f8c0c","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-09-24 12:06:58.000000000","message":"Abandoned\n\nAfter talking with Luis - the change will be in openshift-ansible","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8ffc983f84600843725c43ab72acb6968da22731","revisions":{"38c2afcc05fe887eae8be198e4385efec674bcd1":{"kind":"REWORK","_number":1,"created":"2019-09-19 12:11:08.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/22/683122/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/22/683122/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/22/683122/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/22/683122/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/22/683122/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/22/683122/1"}}},"commit":{"parents":[{"commit":"55c980d4e43a30bb3d6610388bd4ad152c8fe422","subject":"Merge \"Test recreate namespace and pod\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/55c980d4e43a30bb3d6610388bd4ad152c8fe422"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-09-19 12:09:25.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-09-19 12:10:03.000000000","tz":180},"subject":"Revert \"Use label selector to get kuryr pods\"","message":"Revert \"Use label selector to get kuryr pods\"\n\nThis reverts commit 46db32a85df1a16daf9cfa829aee84e600fa4467.\nThere is a difference between u/s and OpenShift labels.\nReverting until it\u0027s solved.\n\nChange-Id: Ic79f307d622affbab196ac37611de7d579f69293\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/38c2afcc05fe887eae8be198e4385efec674bcd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/38c2afcc05fe887eae8be198e4385efec674bcd1"}]},"branch":"refs/heads/master"},"8ed202948ab6232b1be9e4278e56d3f2ebc34a33":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-09-19 16:34:34.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/22/683122/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/22/683122/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/22/683122/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/22/683122/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/22/683122/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/22/683122/2"}}},"commit":{"parents":[{"commit":"55c980d4e43a30bb3d6610388bd4ad152c8fe422","subject":"Merge \"Test recreate namespace and pod\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/55c980d4e43a30bb3d6610388bd4ad152c8fe422"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-09-19 12:09:25.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-09-19 16:34:22.000000000","tz":0},"subject":"Revert \"Use label selector to get kuryr pods\"","message":"Revert \"Use label selector to get kuryr pods\"\n\nThis reverts commit 46db32a85df1a16daf9cfa829aee84e600fa4467.\nThere is a difference between u/s and OpenShift labels.\nReverting until it\u0027s solved.\n\nChange-Id: Ic79f307d622affbab196ac37611de7d579f69293\nDepends-On: I53f56160680555c70fc6d1d87796badf8d2d4e37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8ed202948ab6232b1be9e4278e56d3f2ebc34a33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8ed202948ab6232b1be9e4278e56d3f2ebc34a33"}]},"branch":"refs/heads/master"},"8ffc983f84600843725c43ab72acb6968da22731":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-09-23 13:15:23.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/22/683122/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/22/683122/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/22/683122/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/22/683122/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/22/683122/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/22/683122/3"}}},"commit":{"parents":[{"commit":"55c980d4e43a30bb3d6610388bd4ad152c8fe422","subject":"Merge \"Test recreate namespace and pod\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/55c980d4e43a30bb3d6610388bd4ad152c8fe422"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-09-19 12:09:25.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-09-23 13:15:17.000000000","tz":0},"subject":"Revert \"Use label selector to get kuryr pods\"","message":"Revert \"Use label selector to get kuryr pods\"\n\nThis reverts commit 46db32a85df1a16daf9cfa829aee84e600fa4467.\nThere is a difference between u/s and OpenShift labels.\nReverting until it\u0027s solved.\n\nChange-Id: Ic79f307d622affbab196ac37611de7d579f69293\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8ffc983f84600843725c43ab72acb6968da22731"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8ffc983f84600843725c43ab72acb6968da22731"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
