)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~684538","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~Iff3986c37f00a3d4b8bcdda977aaef244d3bc4ad","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"constraints","hashtags":[],"change_id":"Iff3986c37f00a3d4b8bcdda977aaef244d3bc4ad","subject":"Update the constraints url","status":"MERGED","created":"2019-09-25 07:11:10.000000000","updated":"2019-09-27 16:22:10.000000000","submitted":"2019-09-27 16:20:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684538-1569601239364-827234f7","meta_rev_id":"bc527214f73fcd1c6dae3feed7e14deb14c5b327","_number":684538,"virtual_id_number":684538,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-27 16:20:39.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2019-09-25 09:52:31.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-26 08:35:32.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-26 08:35:32.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-25 09:52:31.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2019-09-26 08:35:32.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2019-09-27 16:20:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2338ff17d04fa01be6adad66d4506e3f6a0d4927","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-09-25 07:11:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d2c51b5faac2ffd4ee8f915e27a3190550beabe3","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-09-25 09:52:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c587bb34ac1980e0fe8241c68d234f14a5dfb891","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 10:18:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d95076179a85462796f58d2fb3c47698 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/080043f52b8146c28aae60d14d7afe9c : SUCCESS in 3m 48s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/0b3de47bd76c4147b2be127087345c24 : SUCCESS in 1h 26m 14s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/de1abdea1c6f4167a5915afc719c17af : FAILURE in 1h 58m 03s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/7fcd0a643f7b4a5790b50f27b5867624 : SUCCESS in 1h 52m 40s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/f05de660d94f4ac8b064bbd032c93872 : SUCCESS in 1h 12m 16s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/1756c3b106bf44688330b7a3608ff6e6 : TIMED_OUT in 3h 04m 46s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/e65f4e4d55d740b4a3e33a3475b61384 : SUCCESS in 1h 36m 48s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/cdef217e48f14caca6b06af0844bbabf : SUCCESS in 1h 34m 35s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/29b4db9ca2aa499190aac554d72b3053 : FAILURE in 40m 49s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/31a1fbf014e343f583334633928154f2 : SUCCESS in 1h 43m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"df23aeb43fa7c6dda5a2fb1774487497bbff78a9","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-09-26 08:35:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"38d5ca8b902385bddc66c7ad4bc59cc550dc9243","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 08:35:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"afdbed7b657b9e4ac59f324b572f58d6392cb944","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 10:28:52.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f367be414444532899f12a3b541da17 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0e43ad6070e4556ad045662e73caf43 : SUCCESS in 3m 38s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/4ae59e25eb5d43c7b096abd0277ad0f7 : SUCCESS in 1h 20m 15s\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/b4afccb56e0546c9918327f6fbbced47 : SUCCESS in 1h 32m 52s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/b97e6dff6e5943079b330d422ce03f34 : FAILURE in 1h 46m 45s","accounts_in_message":[],"_revision_number":1},{"id":"db27d8ad86a9403f9d6f3d5f3fa85ce3f4ee6913","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-09-26 10:43:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e95d2a2226f5721f1c449265e7e35ccd9afa538a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 13:14:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cec57105ea304ff38052140f5cafc5eb : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f74f0c6833b4655b033cf8bdff7b25c : SUCCESS in 4m 20s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/e8c2a766c4f24cb6b9cf3fedfb51a5ac : SUCCESS in 1h 14m 24s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/353c288f19b04015aad3666eb466fbe9 : FAILURE in 1h 34m 55s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/84f93189555449fe8b29b085266fc8d5 : FAILURE in 1h 29m 37s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/f2b44a97ffae4653be912f99790298c3 : SUCCESS in 1h 24m 10s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/248d2a6e8e7f4f1ab607b2da715bad53 : FAILURE in 1h 55m 41s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/2cececda94a24d75a6686d20a709e918 : SUCCESS in 1h 33m 23s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/b6f1ef29d932423bb45c15071cea5b04 : FAILURE in 1h 38m 01s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/e24822c3c0bf4d9fabb5dc664d217931 : SUCCESS in 2h 21m 07s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/a4b837c5298b4c0b8567e94b1a0c95b4 : SUCCESS in 1h 23m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"92381cb487c4b1b813a0b4be203d0c087ad569b6","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-09-26 13:22:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"970a83d27f94267fbd8ea032592777437f0b4377","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 16:18:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/526b133681c7424a9bf78629d54ad9a9 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cce250026bf4443fae9fbba1bd449193 : SUCCESS in 4m 21s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/f6f79b6fa0a64586af44a0e79c6ac12f : SUCCESS in 1h 41m 58s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/2b78150b27cd4474b40df32a8b7ffbe2 : FAILURE in 2h 52m 54s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/f9491dbf42694f03b14f56e880efa3a3 : SUCCESS in 1h 26m 06s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/fd0f64e689f94e04bafe07e368359867 : SUCCESS in 1h 32m 23s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/a7f604cadc314c9693dedb62f4cd367c : SUCCESS in 1h 27m 13s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/f75f8ce8272e4f1ba9807d38ca3bbc9b : SUCCESS in 1h 36m 49s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/1e52370a17d8413e8f96efe7fa639e2b : FAILURE in 2h 01m 00s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/38759345d4ba4a4eb083b7d6a9267acf : SUCCESS in 1h 49m 23s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/8397be2e4708422baece65a94e41188d : SUCCESS in 1h 41m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"125be508a6c1e65a69938765e6a7de6a62e0b4b9","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-09-27 08:49:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fe76bb301931aa7f28b0aed48a6184fcf14c4108","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 10:50:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9cf778805ee46e7ad28542ba91f9a4e : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/278c0964f7824a148beb4a18f16cafb7 : SUCCESS in 3m 31s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/8f9968f2e81d4c46be4cd50524a846b9 : SUCCESS in 1h 21m 12s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/9b353282e51f49fa96a53e912ae200c5 : FAILURE in 1h 57m 45s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/1615cdf1e15c4391bc6aba3592803aee : FAILURE in 1h 20m 30s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/e3d10888c40249aa9a93aa1a797717a5 : SUCCESS in 1h 24m 52s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/d333b93ca497488397c14d8829d97139 : SUCCESS in 1h 29m 20s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/e9c5b6d603ef4fb3bfff96b35f43154c : SUCCESS in 1h 17m 55s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/dbccf4c224284673bd79c705bd00f0ff : SUCCESS in 1h 19m 35s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/619948f9d9ae4b019913e38cf3ab089d : SUCCESS in 1h 41m 33s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/3f9ebc85c43d41c6b9dee2bf47c99301 : SUCCESS in 1h 25m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cbb7b27771f616d3b62c3a6ac78f210f4cc0f291","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 10:50:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8d85d48298bdb63055c25b0d55eebd7d91b7d0e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 12:40:07.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a14016b63774723bc437d4ff3fa7ad9 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c8e0e357a99403a98aa2c0775b156d6 : SUCCESS in 4m 23s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/5a3301d4abcd404397fd02320c940f4c : SUCCESS in 1h 28m 35s\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/8ddf0ba7b7a441ecb2168f90fa3370f0 : FAILURE in 1h 43m 25s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/4458fc2f1412461184a2f00f1e47f111 : SUCCESS in 1h 35m 43s","accounts_in_message":[],"_revision_number":1},{"id":"f2de4a6963369d232e600743334804ed4ea66fe1","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-09-27 12:45:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d24effcf77cd4897fa7e6e9b061c434ad3469dfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 14:52:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a60f93aad2a4745ae79aa22ab51ec90 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fbc17a9710b487da5aa1243b86163e3 : SUCCESS in 3m 09s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/f4c93b2e912849a0b5ba78023bdcdb37 : SUCCESS in 1h 15m 23s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/02b468013e3e4d20b2a4f9cd6b78ad30 : FAILURE in 1h 29m 39s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/a789ae6b297e4c0c980a6c9f38ea7098 : SUCCESS in 1h 23m 47s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/27f6159c576146dca7566b9a6561113e : SUCCESS in 1h 38m 32s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/efc0303027384a7f8e8b39f05cbcce18 : SUCCESS in 1h 26m 43s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/7bc1db81611e4f01aae0390af855c01f : SUCCESS in 1h 40m 09s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/809defb2d52d4853b291e75503ef6f83 : SUCCESS in 1h 42m 41s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/ac12769090e94eb983e41bc3721137f8 : SUCCESS in 1h 57m 38s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/152733678fd74ace9b9d4f5fda76c24c : SUCCESS in 1h 31m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"afa7382d6e984e758a3f3582a4d5179294e927f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 14:53:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1d8644448646d55a2f0e18eb5e05e15469030a5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 16:20:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50706a60390146fabc8edf6f40992db1 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d30e51e37b9d49909ba29678044b3a8f : SUCCESS in 4m 04s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/750c902153e24ff4967714393edea814 : SUCCESS in 1h 21m 09s\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/28438840e4a74aafbfe6c106beb74a61 : SUCCESS in 1h 18m 50s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/08a4de5b6fdb40ff9d90c64780d9ed72 : SUCCESS in 1h 21m 56s","accounts_in_message":[],"_revision_number":1},{"id":"3757f7f8b9245a720c3ed7d7c1cd8f9b05bd925b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 16:20:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"14bd8c6bb84ab05786522011f4a38cf81a5b88d1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 16:22:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4956ed892ee4703850ecde38d5606a4 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"764b362dcfa734c6f6b90fb34d12aa639e0ca43c","revisions":{"764b362dcfa734c6f6b90fb34d12aa639e0ca43c":{"kind":"REWORK","_number":1,"created":"2019-09-25 07:11:10.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/38/684538/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/38/684538/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/38/684538/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/38/684538/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/38/684538/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/38/684538/1"}}},"commit":{"parents":[{"commit":"55c980d4e43a30bb3d6610388bd4ad152c8fe422","subject":"Merge \"Test recreate namespace and pod\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/55c980d4e43a30bb3d6610388bd4ad152c8fe422"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-25 07:11:02.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-25 07:11:53.000000000","tz":480},"subject":"Update the constraints url","message":"Update the constraints url\n\nFor more detail, see http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nChange-Id: Iff3986c37f00a3d4b8bcdda977aaef244d3bc4ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/764b362dcfa734c6f6b90fb34d12aa639e0ca43c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/764b362dcfa734c6f6b90fb34d12aa639e0ca43c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
