)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~686988","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~I0e2f3834bde465560b8d4c4ae375c38b41118000","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"port_pool","hashtags":[],"change_id":"I0e2f3834bde465560b8d4c4ae375c38b41118000","subject":"Add subnet_per_namespace config option","status":"MERGED","created":"2019-10-07 09:33:36.000000000","updated":"2019-10-25 14:32:15.000000000","submitted":"2019-10-25 14:30:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686988-1572013846847-5cb63f8a","meta_rev_id":"35eaa6c9309c1082daaf079fe0595c3aa2eaf135","_number":686988,"virtual_id_number":686988,"owner":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"date":"2019-10-25 09:34:01.000000000","_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-25 14:30:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2019-10-23 08:21:29.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-24 13:39:43.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-24 13:39:43.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-07 09:40:22.000000000","updated_by":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"reviewer":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"state":"REVIEWER"},{"updated":"2019-10-14 04:36:13.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-10-14 07:42:27.000000000","updated_by":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"reviewer":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"state":"REVIEWER"},{"updated":"2019-10-23 08:21:29.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2019-10-24 13:39:43.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2019-10-25 14:30:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"99728769840b833e8eb1ede2d972e0884918d89a","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-10-07 09:33:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"daeb2723133e0e871f48864df21aaf85af340ad7","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-10-07 09:39:08.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"d977f9d6e2c7f44b734a2b4be8d0e6eddd2a283e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 11:36:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47e95e3e56b94f6696ca3c707d68b07e : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4a1f7cee3da4b3e95de4cecc11c8692 : SUCCESS in 3m 02s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/68f36fae4afd4810999c99f0c30b769c : SUCCESS in 1h 27m 17s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/990ac3b9658745ae916d4ba727c9ceaa : FAILURE in 1h 57m 24s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/3377be8b1fbb4faba2e684f6f50fe0d3 : SUCCESS in 1h 29m 23s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/e2a5651461c94ce69d08c6bb4bb8bc90 : SUCCESS in 1h 35m 01s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/6a76992d05f146e4b47a8c2bc217405f : FAILURE in 25m 04s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/fa102f9617cf4918b7e1732c49e746d9 : SUCCESS in 1h 22m 08s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/d7fafff34388404b84eb485b25b5e522 : SUCCESS in 1h 11m 54s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/a02f2f2aaa924f8782e57e4da0080699 : SUCCESS in 1h 44m 40s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/deccdbb08b9c4e7b9571c2837b8451d4 : SUCCESS in 1h 40m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"680ca90a3abe51f0d032b016a924eee70c4f1b8f","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 12:48:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/2116334226114879ab07a494b80b8bf2 : FAILURE in 2h 10m 17s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/ce0655abe5054d05bf91e67ca22b5ea2 : FAILURE in 2h 10m 32s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/1505dd5f8f7e4809845b2919a9d2bbef : FAILURE in 1h 44m 03s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/a8ee67f1796f40a89728a0974242b7e6 : SUCCESS in 1h 03m 28s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/c5bdb45fcbaa44f7a860b0d5509e9dd6 : TIMED_OUT in 3h 06m 48s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/3cdbff4a966d4213b62e3b44bf02f9d1 : FAILURE in 20m 59s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/a9c867610ef24af1a6cf9d41b131d88b : FAILURE in 1h 57m 38s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/9a72cf3e6b7445829efa435c52a414da : FAILURE in 1h 39m 42s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/25cc1ae644474006914db7945d0e799f : TIMED_OUT in 2h 16m 03s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/175b8f2e9cad4f8dbc84fc99e82bef7e : FAILURE in 27m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0a0cfe9f7d2eea1ac642a2314c2b6bef3ec7e1ac","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-10-08 08:05:07.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9e10a5d5fc3655762b77c590ac24557e9f62542d","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-10-10 05:47:18.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"cbc54657833b738d9f8087089f1d7b14c2ed1399","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-10-10 07:15:29.000000000","message":"Patch Set 2: Code-Review+1\n\ncheck experimental","accounts_in_message":[],"_revision_number":2},{"id":"3724fcb8b47ed1ea895e0b8deb56ce9e31f7a644","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 07:42:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/861e9c7571e54fdab85fa1339341a251 : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e9814a59a364660aeb6ca7c93a1312d : SUCCESS in 2m 59s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/3f731c7ac044464e84836ca07f7e61c0 : SUCCESS in 59m 04s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/b0c14b86b1a6449581c2502a3754e995 : FAILURE in 1h 45m 30s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/3570f930d67e4fe88bb56ca436e99ac0 : SUCCESS in 1h 36m 33s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/e2c86f012e8b41a690994fd61fdb225d : SUCCESS in 1h 26m 57s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/2ba748957fd94e85a0a98234ec774092 : FAILURE in 21m 57s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/c1a060238a6643b1b68ad9ebd4d6c015 : SUCCESS in 1h 12m 34s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/04aaab5f1cdf4cbab7f3ff719afcac8f : SUCCESS in 1h 12m 32s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/567b4b5493e74ac4873e98f05f96dfb7 : SUCCESS in 1h 49m 31s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/836d30f7474f4b01b50d43a86c3049f6 : SUCCESS in 1h 30m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"64a89c85126a5d18eeaa2b84ac7d57c949417954","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 10:25:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/40b81f4d199e4698b4e92e3a764b2e96 : FAILURE in 2h 15m 16s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/b372b03bda884c3c87e463c90835ed91 : FAILURE in 2h 18m 35s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/4401d56c00424aaa976a0eb3ab5ff13d : FAILURE in 1h 51m 41s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/847cda73d9a74f84ba28e003fa3ab8f8 : SUCCESS in 1h 03m 39s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/ce3618c3740b4542b06ee685dbe09671 : TIMED_OUT in 3h 03m 51s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/d0bd02a822264e46b34c5db71cd5706f : FAILURE in 22m 50s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/832ddd2ab1e84324b04e5ecdeb0f436b : FAILURE in 1h 43m 15s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/831ae4b345564a0a850acd2b126fc9de : FAILURE in 1h 50m 28s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/2db9c02e780f40fa91ab52dbadb9ed26 : TIMED_OUT in 2h 15m 19s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/6e9776b1597344fea218c14d7db79200 : FAILURE in 30m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0da416168b50e3bcaf8256b68bc7d67bd71f3edd","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-10-11 14:17:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"14fb7d6ddf44944b89784d97b4e30a7258a6bec6","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-10-14 04:36:13.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"fa2185a0317cd4c282f108b3e97a385a41a9d85a","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2019-10-14 07:42:27.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1a3862c225af9e5323e4fc8a144c8b61fd76874c","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-10-14 07:58:49.000000000","message":"Patch Set 2: Code-Review-1\n\nSeems ports pool tests are skipt","accounts_in_message":[],"_revision_number":2},{"id":"43efcc83b24fc8fa891764703960cbaef5c444d7","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-10-16 07:22:54.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"375d465d9c2bcfe0143c6d3628bde6399adf2559","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-10-16 07:26:44.000000000","message":"Patch Set 3:\n\ncheck experimental","accounts_in_message":[],"_revision_number":3},{"id":"3a3182f0a2662c3bd15ac762d74fab97aba6ac03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 09:25:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1bdfed33fc34c6c9fa35f552c038dfa : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e94653da6aea4fe9aead28ed13829017 : SUCCESS in 3m 02s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/f71d89113e8441f098292b3c9fefbe35 : SUCCESS in 1h 10m 11s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/f31feabf16684987904c31aa776e44d2 : FAILURE in 1h 42m 58s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/0a36c06cb1704c338a99e9f03113143f : SUCCESS in 1h 22m 32s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/f1850de45d494ce2a3c393bb7a5358d8 : SUCCESS in 1h 27m 25s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/608448aee5dd42619ca8a6656ec310ff : FAILURE in 25m 33s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/9cf52104ccdd4a6c80ee097dd51acf87 : SUCCESS in 1h 08m 40s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/bd8f3d958f5f4d80b121bfb819334a41 : SUCCESS in 1h 32m 11s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/cc9074ad5ad04ef89a8b1e9979a80132 : SUCCESS in 1h 56m 39s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/e752089ae142461fa72eb06d8b5ebde3 : SUCCESS in 1h 14m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b9524d393d0536416b1cc30512103ee517caa83b","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 10:43:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/6bea912e20e74d188ddacb43fb4aaef8 : FAILURE in 2h 54m 42s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/07bf0d146b414dbb91c4af263fdcb368 : FAILURE in 2h 05m 27s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/39d5ba945e5f4a21b2a0ed3a06361a09 : FAILURE in 2h 36m 30s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/5740ad4fbd9440ea8330f7b421cc586d : SUCCESS in 1h 11m 04s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/561e3292f3494b00b0b5ecc6b67be493 : TIMED_OUT in 3h 03m 53s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/049291bcfc224dcb9435ef0e3c07cea6 : FAILURE in 24m 41s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/de6ac0fa38ea439a8d1f40558b78cfc8 : FAILURE in 2h 08m 48s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/bb87030987bc497782e43b9bac7fe268 : FAILURE in 1h 24m 43s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/10470cdb3c364c07a9d11ee6fd6b1af6 : TIMED_OUT in 2h 15m 56s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/aec10041b8a947c891683a1eb915b3b5 : FAILURE in 28m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e2edfb2181fde999f35446df32117d596a055a8b","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-10-17 07:03:19.000000000","message":"Patch Set 3:\n\nTest are not being executed for pools! You need to ensure they are enabled and executed! (on kuryr-kubernetes-tempest-containerized-pools-namespace gate)","accounts_in_message":[],"_revision_number":3},{"id":"483814759fea3bfcec7a852dcf64b14d6a7fc323","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-10-22 12:31:33.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"2895a4f44fee68f960be022b78151478b4d5cc30","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-10-22 12:38:46.000000000","message":"Patch Set 4:\n\ncheck experimental","accounts_in_message":[],"_revision_number":4},{"id":"f4eebcf390ca1e8ea94538c423f326898efc126f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 14:37:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a41c2e40aa742f28d0fae71b4633d0c : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6666b99400764598888f81c435dae869 : SUCCESS in 3m 05s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/b477b6985a804da786f9e902099865ce : SUCCESS in 1h 19m 43s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/832a9a124e624e9bb217f29e64cd8771 : FAILURE in 2h 03m 16s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/6c30ad3d92c740baaf16028b160c755e : SUCCESS in 1h 31m 41s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/bb42d846b81c45d897ee6f6c89e57503 : FAILURE in 1h 56m 07s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/c4e801dbc4104a829602a65dcd2dfe04 : FAILURE in 20m 30s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/99950315e4a4413aa857eeee75e1eae1 : SUCCESS in 1h 08m 12s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/4a3ca1929b7640c9ac502640a001af59 : SUCCESS in 1h 43m 36s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/ef0ab5c82ab742a78e9b42a247316441 : FAILURE in 17m 15s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/7f89cabeab704e398d62269471b0ad2e : SUCCESS in 1h 27m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6f845a4bdc92d389e4b974e0c4f9becc00c67920","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 15:45:40.000000000","message":"Patch Set 4:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/e10c95f15e7644be91b967fd41448546 : FAILURE in 2h 27m 11s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/8b0d11bc3cbc4c47963840201098fd6d : FAILURE in 1h 56m 27s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/f27e29ddfa694d9eb843d3e5d65cbb64 : FAILURE in 1h 53m 01s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/88474c23922844eca41005e0922486e4 : FAILURE in 1h 04m 25s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/0aea946d90484982be4244580cf5778d : TIMED_OUT in 3h 05m 08s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/d7ca57476e7440b18300498b7adfc6f9 : FAILURE in 21m 15s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/fe141a238f444651b4ca0c143cbb24a5 : FAILURE in 2h 04m 00s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/5cf87199afb945a3bb75b2275a259c8b : FAILURE in 1h 54m 14s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/9a2842403d774103ab6cca61b4c054e1 : SUCCESS in 2h 17m 00s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/1caf47dae4b14c3096dbf0b37f8d30ea : FAILURE in 24m 57s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"11f082b43d0a0311135b973fe4f7db37bdd3d351","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-10-23 08:21:29.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"5c3f312b687a4237578a28dfb3438b529ac21f2e","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-10-24 12:18:28.000000000","message":"Patch Set 4: Code-Review-1\n\nSeems that port_pool test is skipt in gates which has subnet per namespace enabled, like the openshift ones","accounts_in_message":[],"_revision_number":4},{"id":"6caa97ada1b1009ee25d4ffc0fbece2e8f5b9fe6","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-10-24 13:39:43.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b71f35c57f30a71a8e13079393d28e0a2bf8756e","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-10-25 09:34:01.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"e9e6073c376a2adf64fd6615a111a0971c33187f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 12:42:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8716ba3e7fe64f3b8b35048279df0699 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0ee7a79579c4406b2afa4873aba89c4 : SUCCESS in 3m 10s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/2c361f2fccb7498cb95ac6c343a5ae24 : SUCCESS in 1h 02m 07s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/ff4b662c2b3040379d274f335332889b : FAILURE in 1h 55m 26s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/ac0605b154b84b78b3546771560fbed5 : SUCCESS in 1h 14m 54s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/229ca3b64a9c4b92a706826f0217ab96 : TIMED_OUT in 3h 06m 44s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/1f031cd20d524f0c87658c38797ac359 : FAILURE in 19m 56s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/04c44e1322c14b7c99da567d9b572f86 : SUCCESS in 1h 15m 16s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/9b2ae80df9f844b3955144218c56cda6 : SUCCESS in 1h 59m 55s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/c15c8fef15ea4fc3add753796f71adc4 : SUCCESS in 1h 40m 53s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/ee082cef8aff470da97fa3b0fb40596f : SUCCESS in 1h 23m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"95852285c9ce35ed48b101ba30b947d435928208","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 12:42:24.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"8d4ac09889d81bab4cdab4c1c7b23a7b4b89108a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 14:30:46.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b00cbc4a8c954052871ea351bca372f2 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46a9075d90cc406690826f21f464bff7 : SUCCESS in 4m 24s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/86fdfd5bc7204b7fb5344296dd549957 : SUCCESS in 1h 20m 10s\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/498829dce67e47479e79ce14ad9ee2ff : SUCCESS in 1h 16m 27s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/eca7b6d8cdf545a799bba813bfd6b4b9 : SUCCESS in 1h 41m 06s","accounts_in_message":[],"_revision_number":4},{"id":"c34d6e75a2cb9dce2b4e6cc12292bc649c60a19b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 14:30:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"6cff46e13d3d583434c8bbb58c62ba67348ec843","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 14:32:15.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ad9db9afcd047de810ff1f467920b63 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"66958972da19086a5ce5ec10b83a1500cd40347c","revisions":{"894dc1134e6d46053a61b6ebe98b2e0d5161099a":{"kind":"REWORK","_number":1,"created":"2019-10-07 09:33:36.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/88/686988/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/88/686988/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/88/686988/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/88/686988/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/88/686988/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/88/686988/1"}}},"commit":{"parents":[{"commit":"fc4ec21306453a008ecdef075ac48eb5907df11f","subject":"Adjust initial ports list in port_port test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/fc4ec21306453a008ecdef075ac48eb5907df11f"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-10-07 09:20:58.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-10-07 09:33:15.000000000","tz":180},"subject":"Add subnet_per_namespace config option","message":"Add subnet_per_namespace config option\n\nChange-Id: I0e2f3834bde465560b8d4c4ae375c38b41118000\nDepends-on: I1109a303ee216ed8061959a81f0ed315bf138c80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/894dc1134e6d46053a61b6ebe98b2e0d5161099a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/894dc1134e6d46053a61b6ebe98b2e0d5161099a"}]},"branch":"refs/heads/master"},"b053fbe314b92efd3e7a8e36b71126983a206f78":{"kind":"REWORK","_number":2,"created":"2019-10-10 05:47:18.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/88/686988/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/88/686988/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/88/686988/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/88/686988/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/88/686988/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/88/686988/2"}}},"commit":{"parents":[{"commit":"fc4ec21306453a008ecdef075ac48eb5907df11f","subject":"Adjust initial ports list in port_port test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/fc4ec21306453a008ecdef075ac48eb5907df11f"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-10-07 09:20:58.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-10-10 05:47:05.000000000","tz":0},"subject":"Add subnet_per_namespace config option","message":"Add subnet_per_namespace config option\n\nChange-Id: I0e2f3834bde465560b8d4c4ae375c38b41118000\nDepends-on: I1109a303ee216ed8061959a81f0ed315bf138c80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/b053fbe314b92efd3e7a8e36b71126983a206f78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/b053fbe314b92efd3e7a8e36b71126983a206f78"}]},"branch":"refs/heads/master"},"c82551374eb108822017f5399813026e20b9e826":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-10-16 07:22:54.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/88/686988/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/88/686988/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/88/686988/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/88/686988/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/88/686988/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/88/686988/3"}}},"commit":{"parents":[{"commit":"42c4d5344169659e358f35c64647937a8e741748","subject":"Adjust initial ports list in port_pool test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/42c4d5344169659e358f35c64647937a8e741748"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-10-07 09:20:58.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-10-16 07:22:37.000000000","tz":180},"subject":"Add subnet_per_namespace config option","message":"Add subnet_per_namespace config option\n\nChange-Id: I0e2f3834bde465560b8d4c4ae375c38b41118000\nDepends-on: I1109a303ee216ed8061959a81f0ed315bf138c80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c82551374eb108822017f5399813026e20b9e826"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c82551374eb108822017f5399813026e20b9e826"}]},"branch":"refs/heads/master"},"66958972da19086a5ce5ec10b83a1500cd40347c":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-10-22 12:31:33.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/88/686988/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/88/686988/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/88/686988/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/88/686988/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/88/686988/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/88/686988/4"}}},"commit":{"parents":[{"commit":"42c4d5344169659e358f35c64647937a8e741748","subject":"Adjust initial ports list in port_pool test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/42c4d5344169659e358f35c64647937a8e741748"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-10-07 09:20:58.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-10-22 12:30:17.000000000","tz":180},"subject":"Add subnet_per_namespace config option","message":"Add subnet_per_namespace config option\n\nChange-Id: I0e2f3834bde465560b8d4c4ae375c38b41118000\nDepends-on: I49e0f59008894078f9e40198873caa78767ef0b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/66958972da19086a5ce5ec10b83a1500cd40347c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/66958972da19086a5ce5ec10b83a1500cd40347c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
