)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~690877","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~Ic0d821979b4277b564705022c0858721ef52348f","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"port_pool","hashtags":[],"change_id":"Ic0d821979b4277b564705022c0858721ef52348f","subject":"Adjust expected number of ports to config values","status":"MERGED","created":"2019-10-24 08:29:34.000000000","updated":"2019-10-25 19:12:30.000000000","submitted":"2019-10-25 19:11:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690877-1572030663848-f546210b","meta_rev_id":"4ce0e8b3fe2ec6f07dc29055f98024bed77a1d0f","_number":690877,"virtual_id_number":690877,"owner":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"date":"2019-10-25 14:15:39.000000000","_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-25 19:11:03.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2019-10-25 09:41:47.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-25 09:31:10.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":1,"date":"2019-10-25 14:17:57.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-24 09:06:34.000000000","updated_by":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"reviewer":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"state":"REVIEWER"},{"updated":"2019-10-25 09:31:10.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2019-10-25 14:17:57.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2019-10-25 19:11:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2de5e3b595c28c67616fd21da22f6f668975ca17","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-10-24 08:29:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f49e709be9785eb299be45381acf84b00ebac13","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-10-24 09:06:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b284953919027f5b170793762cd2725e2cceae9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 10:30:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a61cf7eb1854aa280e1977066bc0571 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe5f31ee8485410eb13ce4bdb4f3d66a : SUCCESS in 3m 26s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/605c91c09616410a821fa4798af77763 : SUCCESS in 1h 30m 11s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/d9ae4b11318a4154acbef77495c063eb : FAILURE in 1h 34m 15s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/efa56adfda8f42bf933f0d041e6c745e : SUCCESS in 1h 43m 05s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/5a00616c8e6b402da7e79cf358c0437e : FAILURE in 38m 28s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/6a91148d882640279f8e0e4279a610fe : FAILURE in 25m 49s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/ff322583f4b14c9085d56e35c53752dd : SUCCESS in 1h 39m 50s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/377b936a785d4afda42c97dba088990c : SUCCESS in 1h 54m 08s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/8d1bdc37af5a41fa9d9d6c1664d557e3 : SUCCESS in 1h 48m 53s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/f9b0d012811746848cfcaa97a16b974f : SUCCESS in 1h 43m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a18b4bf6c341e85cf4ab4138244faec54a3e67ff","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-10-24 11:46:22.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"2b7bda8e8e9603af3a5ce0b00129c4c8ad6e4618","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-10-24 12:16:22.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a82cdc2432e6790b4e779ceb45270bf9f43781e1","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-10-24 12:20:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ce317c22d0beab91113d869a144d34071916cdc7","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-10-24 13:43:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c22470cb08c6919012ca066636b49cb7d59d94b8","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-10-24 13:52:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"92dcf228253ae2c386d5975339088f7b69f7dcd4","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-10-24 13:52:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"78854bd8bba4d2f5cec9ad48f11c0565bde3ca98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 15:55:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9bc3d55f55b4d23b35932e32048c417 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7b4b175167e4dd49d129899a274b89f : SUCCESS in 3m 38s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/7bd9b711a26c4840bb757b12290b1260 : SUCCESS in 1h 18m 05s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/7665638c2c174cf0831b49a353e24542 : FAILURE in 2h 00m 31s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/b3a15fd430704904be99271ee13c1ed7 : SUCCESS in 1h 41m 46s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/bd45018975364c3fa12ca0e82a94863a : FAILURE in 37m 14s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/9cb570f85cf147a49e9c42124b31f9fd : FAILURE in 24m 38s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/e98b877c85c94db189b688171810cc7f : SUCCESS in 1h 15m 31s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/a47d5ea800424135a917fb28beff114f : SUCCESS in 1h 20m 59s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/cc20b4d4e42a457bb52d64d6714f1b89 : SUCCESS in 1h 42m 24s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/b34f9230400c49e4a919ed3bd3c74860 : SUCCESS in 1h 37m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6bbf5ab2b7a87169872163a2c8f92485dcb73bf7","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-10-25 07:44:18.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6e5483e5d5e94be3283a74b810908163c4b97a01","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-10-25 09:19:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9acf016beabe8366f5863fd52418a87d07ae642d","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-10-25 09:26:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6f2969a7813fbb980f3e8fe2cbaceba55f532e9b","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-10-25 09:31:07.000000000","message":"Patch Set 4:\n\ncheck experimental","accounts_in_message":[],"_revision_number":4},{"id":"d2981dd5d538d1feb664d721ce6255c7670be631","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-10-25 09:31:10.000000000","message":"Patch Set 4: Code-Review+2\n\ncheck experimental","accounts_in_message":[],"_revision_number":4},{"id":"9b9093eac25e5e931f6720bc8cb34fb9ec512bde","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-10-25 09:41:47.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"6c5cc17b62e2c8709482f8072a8de31b033e61c6","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-10-25 09:41:53.000000000","message":"Patch Set 4: -Workflow","accounts_in_message":[],"_revision_number":4},{"id":"1c877658c445e68fd7233103ebf53ed383bb69c6","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-10-25 09:42:06.000000000","message":"Patch Set 4:\n\nAh, okay, job first.","accounts_in_message":[],"_revision_number":4},{"id":"96c06a83fd9d9cc1e42e93f0806dad78c86e417e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 12:39:21.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb1f07e9be534c89a4ab7b9b479134ce : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19d399f1104448f48a64b56d24e09cb9 : SUCCESS in 3m 35s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/3a4e342a71de47aabe32d0f35bde43cb : SUCCESS in 1h 30m 05s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/7e282fd1aa7344888704fd0ecdfab30f : FAILURE in 1h 21m 26s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/97de613bc0834e138c12c77be379c76d : SUCCESS in 1h 24m 09s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/f55854d3fd4744ffa46a085fed7c44a7 : TIMED_OUT in 3h 05m 59s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/6b1ea474ec83414fa32c5fb0920b11c6 : FAILURE in 23m 47s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/6f58d46a19bf43ce902de32ac82eaf24 : SUCCESS in 1h 28m 00s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/da7d87d94643456082d702f79ef46db4 : FAILURE in 1h 27m 31s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/13e584d2b4f1437e9f557326d5c6fa05 : SUCCESS in 1h 42m 08s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/114043fabe344913af3dbdc8a60d0872 : SUCCESS in 1h 28m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0db35f35b495fd853270e5579146de3a3398401b","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 12:43:32.000000000","message":"Patch Set 4:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/4a026975293c40379b62cdb82a04a6c8 : FAILURE in 1h 58m 16s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/6458bcefaf0e48f3a92780a56856843f : FAILURE in 1h 30m 07s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/31c2a5f90ec94406bbc767c9f73b6326 : FAILURE in 2h 12m 50s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/9fa720f044c24f4e819d058eb535a231 : FAILURE in 59m 09s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/a203dc02439d4fe49a9d7ae40b224c8e : TIMED_OUT in 3h 06m 04s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/5f80875049414506b61d6c7bf0fb6280 : FAILURE in 21m 10s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/b75bca46ccf941d9a004dbb43de4d935 : FAILURE in 2h 27m 30s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/4c42af9889394875973e524b38a6e378 : FAILURE in 1h 34m 39s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/818e61433ccf4749b83be63a852a3ec8 : SUCCESS in 2h 14m 28s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/3ccf872ee92a45e48b4128e94a491a2c : FAILURE in 27m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1f8f6fbd45c6faaef7bb0e4d5839dd13fa7502dc","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-10-25 14:15:39.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"0ee52d1fd19953aa6b3901a77d9573316d44faa0","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-10-25 14:17:57.000000000","message":"Patch Set 4: Workflow+1\n\nFixed test seems to be passing.","accounts_in_message":[],"_revision_number":4},{"id":"77acfa00d1501a59e31780e1e703950538cc18f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 17:26:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/409af7945f334c2ea423f4ea545cec21 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8ee699512a84ec2af4ade3effe5858e : SUCCESS in 3m 27s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/1447c4c959c3438689233a69542b2406 : SUCCESS in 1h 19m 10s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/6f09edceb08c4c19a37d389feec1f831 : FAILURE in 1h 51m 49s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/e1a7c16846d648cca7641dd0a26c3c34 : SUCCESS in 1h 57m 17s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/0d2169ee71e64c6eb19867a5e78df0ce : TIMED_OUT in 3h 08m 35s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/657ba4242a474180bbc04458151ee979 : FAILURE in 22m 51s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/07a5d65656a5440880cb1bd1127dd4ec : SUCCESS in 1h 14m 29s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/524ba0e334bf467398a4f431a229e73a : SUCCESS in 1h 46m 17s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/20042f11129d4ad4a60774b2f8753b0e : SUCCESS in 1h 47m 56s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/5e6fb4b61571452a98162c2475ed37e9 : FAILURE in 1h 39m 25s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"15789c240b0d7802dd24e7ad11e2ff69613efb3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 17:26:50.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ccfe14456e6292040f1244594a31a0c174eac1d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 19:11:03.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8245d2753994794b29879b0615f2747 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cd346ab6c2e40eba37350a51a0ca984 : SUCCESS in 4m 29s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/df485097c13d4da9bc0dc0dbd8f2e34b : SUCCESS in 1h 25m 05s\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/ac60cc3a7b674ff3a1c83493ba5225fd : SUCCESS in 1h 25m 30s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/c20425a2ec81475cb7d7b1de725526a4 : SUCCESS in 1h 41m 16s","accounts_in_message":[],"_revision_number":4},{"id":"c6e1075b8e9863e9d5dc5676593da3d39e0b7536","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 19:11:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"38988be743edc00dff71f848c268ddea07e35eb2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 19:12:30.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ed08ca64c174886b89a8072297a35d0 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a0775316ac160dadf90304b556eb64f80b836eee","revisions":{"625b44e896f4990977fa924462453e4e5ff099af":{"kind":"REWORK","_number":1,"created":"2019-10-24 08:29:34.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/77/690877/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/77/690877/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/77/690877/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/77/690877/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/77/690877/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/77/690877/1"}}},"commit":{"parents":[{"commit":"42c4d5344169659e358f35c64647937a8e741748","subject":"Adjust initial ports list in port_pool test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/42c4d5344169659e358f35c64647937a8e741748"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-10-24 08:25:42.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-10-24 08:29:24.000000000","tz":180},"subject":"Adjust expected number of ports to config values","message":"Adjust expected number of ports to config values\n\nCheck the number of ports after the third pod creation based on the values\nof ports_pool_batch and ports_pool_min\n\nChange-Id: Ic0d821979b4277b564705022c0858721ef52348f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/625b44e896f4990977fa924462453e4e5ff099af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/625b44e896f4990977fa924462453e4e5ff099af"}]},"branch":"refs/heads/master"},"3e0c9f9a6fcbea9500f052581dc92061a3faba70":{"kind":"REWORK","_number":2,"created":"2019-10-24 13:52:53.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/77/690877/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/77/690877/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/77/690877/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/77/690877/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/77/690877/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/77/690877/2"}}},"commit":{"parents":[{"commit":"42c4d5344169659e358f35c64647937a8e741748","subject":"Adjust initial ports list in port_pool test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/42c4d5344169659e358f35c64647937a8e741748"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-10-24 08:25:42.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-10-24 13:52:08.000000000","tz":180},"subject":"Adjust expected number of ports to config values","message":"Adjust expected number of ports to config values\n\nCheck the number of ports after the third pod creation based on the values\nof ports_pool_batch and ports_pool_min\n\nChange-Id: Ic0d821979b4277b564705022c0858721ef52348f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/3e0c9f9a6fcbea9500f052581dc92061a3faba70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/3e0c9f9a6fcbea9500f052581dc92061a3faba70"}]},"branch":"refs/heads/master"},"6a6f4b54e8d15927d6114a55487baf82b580a4f9":{"kind":"REWORK","_number":3,"created":"2019-10-25 09:19:27.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/77/690877/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/77/690877/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/77/690877/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/77/690877/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/77/690877/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/77/690877/3"}}},"commit":{"parents":[{"commit":"42c4d5344169659e358f35c64647937a8e741748","subject":"Adjust initial ports list in port_pool test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/42c4d5344169659e358f35c64647937a8e741748"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-10-24 08:25:42.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-10-25 09:19:04.000000000","tz":180},"subject":"Adjust expected number of ports to config values","message":"Adjust expected number of ports to config values\n\nCheck the number of ports after the third pod creation based on the values\nof ports_pool_batch and ports_pool_min\n\nChange-Id: Ic0d821979b4277b564705022c0858721ef52348f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/6a6f4b54e8d15927d6114a55487baf82b580a4f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/6a6f4b54e8d15927d6114a55487baf82b580a4f9"}]},"branch":"refs/heads/master"},"a0775316ac160dadf90304b556eb64f80b836eee":{"kind":"REWORK","_number":4,"created":"2019-10-25 09:26:42.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/77/690877/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/77/690877/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/77/690877/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/77/690877/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/77/690877/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/77/690877/4"}}},"commit":{"parents":[{"commit":"42c4d5344169659e358f35c64647937a8e741748","subject":"Adjust initial ports list in port_pool test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/42c4d5344169659e358f35c64647937a8e741748"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-10-24 08:25:42.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-10-25 09:25:47.000000000","tz":180},"subject":"Adjust expected number of ports to config values","message":"Adjust expected number of ports to config values\n\nCheck the number of ports after the third pod creation based on the values\nof ports_pool_batch and ports_pool_min\n\nChange-Id: Ic0d821979b4277b564705022c0858721ef52348f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/a0775316ac160dadf90304b556eb64f80b836eee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/a0775316ac160dadf90304b556eb64f80b836eee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
