)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~692844","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~I4ed2bdbbc664579689763f2759741b0efced6db6","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"fix-np-gate","hashtags":[],"change_id":"I4ed2bdbbc664579689763f2759741b0efced6db6","subject":"Ensure correct failure message is checked on CRD validation","status":"MERGED","created":"2019-11-04 15:25:40.000000000","updated":"2019-11-20 09:24:06.000000000","submitted":"2019-11-20 09:22:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692844-1574241755856-64a7fc44","meta_rev_id":"341600c5344533d25b8d37a550d18a7fd668e120","_number":692844,"virtual_id_number":692844,"owner":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-20 09:22:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2019-11-19 18:27:02.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-19 18:27:02.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-20 07:34:08.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-06 15:19:21.000000000","updated_by":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"reviewer":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"state":"REVIEWER"},{"updated":"2019-11-19 09:45:38.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2019-11-20 07:34:08.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2019-11-20 09:22:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"499b7fa04cd4b9e5cd01fb5124ee7502cd6474b3","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-11-04 15:25:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"077ce6253a438ccd7adb7223438eb3a6d34ec39e","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-11-04 15:26:44.000000000","message":"Patch Set 1:\n\nLet\u0027s wait for the Network Policy gate.","accounts_in_message":[],"_revision_number":1},{"id":"9641dcb8c825eef1dd0de137743969e213b5da13","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-11-04 15:36:02.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b734646fca9ea9a8a22d58377b0fc415cece8467","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 18:46:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0aa57709f42446bc9bb94ad3ecd0c495 : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ba7fe940c474dc99c515b02307511b5 : SUCCESS in 2m 40s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/23c47354335f4e25ba28ce868a1a4c23 : SUCCESS in 1h 03m 12s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/6acfbb6b581d4c46b453a5874daa7b94 : FAILURE in 1h 41m 48s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/b936896c02124e728b6b44b2e345e533 : SUCCESS in 1h 24m 50s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/f4ddba6568a74c9cbae9ee00d41cdec5 : TIMED_OUT in 3h 04m 16s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/d94d68ed572f485a80e3354d570b27ee : SUCCESS in 1h 27m 17s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/2adea339debe4f8583f22ff0811a2340 : SUCCESS in 1h 08m 17s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/62efed3d0daf4204af5e692e578705fc : SUCCESS in 1h 25m 05s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/3b53a8fa24dd41cab7551729991a7dad : SUCCESS in 1h 38m 57s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/b390dae6ad8e418fbc885e9219735aa3 : SUCCESS in 1h 45m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5b8bd739ecf1b720b82e82126b3fd257455cbc35","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-11-04 19:07:21.000000000","message":"Patch Set 2:\n\ncheck experimental","accounts_in_message":[],"_revision_number":2},{"id":"b7cc643c9956575cf520d243414570f0e2512ca2","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 22:16:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/8bba053c5ce940c693b3dcac8c97c2a4 : FAILURE in 2h 11m 10s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/21902b5ebdb84a339a9da40de70e351e : FAILURE in 2h 16m 13s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/24d2d92b12ad4a379fb27a205d00952f : FAILURE in 1h 45m 56s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/4275cfbfaa6e475ab560e4de3589a002 : FAILURE in 1h 26m 03s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/682d95c9308a445a82a5a427dfb69e81 : TIMED_OUT in 3h 07m 11s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/f14d8b2ff19b4c20b11d36235a8c908d : SUCCESS in 1h 15m 12s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/014c571bff974501affa1b3c8597aaec : FAILURE in 2h 05m 03s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/9ffd86d4ed7b463f9e32e9e6f3d68f07 : FAILURE in 1h 22m 22s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/32774a387b3a47139f8d16688370e01e : SUCCESS in 2h 12m 59s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/dbf5e922aeb74817a35b9455d2a1b598 : FAILURE in 24m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"02a1f40c327f1cd7a2cb99f5a15a68a04de6882e","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-11-05 07:39:51.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1af882290b1e3096f8cf9414988b1307ecde082d","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-11-05 07:52:06.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"11d41843e61450af3d482f2fbff888c28f25aecc","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-11-05 08:26:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"acaec02cc02ccc461ffaa5bd8f3cd23748064751","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-11-05 08:42:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b8b4c55d1dcc17edbb4a7eb7c6a0a7a7f300fa8f","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-11-05 08:50:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1142abce6746672374328900d050e8460a412532","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-11-05 09:14:50.000000000","message":"Patch Set 2:\n\nI don\u0027t have strong opinion here. I can move them to another folder if you see fit.","accounts_in_message":[],"_revision_number":2},{"id":"05fbd1f394ba6f49c1c90ee98eaedc9b2aa36f33","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-11-05 09:40:11.000000000","message":"Patch Set 2:\n\n\u003e I don\u0027t have strong opinion here. I can move them to another folder\n \u003e if you see fit.\n\nI think it will be probably better to create additional tree paralel to scenario, maybe called API (as they do in tempest or if you have other idea something else) and put it there\nNot sure if it should be done now or some other patch though","accounts_in_message":[],"_revision_number":2},{"id":"c9757a308fdc976f9ea2695ab02e4c1ca5d7b0ca","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-11-06 15:01:47.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"d258aa8bf051bbb0ea886ada3f5873d283207c62","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-11-06 15:02:44.000000000","message":"Patch Set 3:\n\n\u003e \u003e I don\u0027t have strong opinion here. I can move them to another\n \u003e folder\n \u003e \u003e if you see fit.\n \u003e \n \u003e I think it will be probably better to create additional tree\n \u003e paralel to scenario, maybe called API (as they do in tempest or if\n \u003e you have other idea something else) and put it there\n \u003e Not sure if it should be done now or some other patch though\n\nIf you agree, we can handle this improvement in a follow up patch.","accounts_in_message":[],"_revision_number":3},{"id":"55c03de07757bef1d8be3253e181e4757a32ea1c","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-11-06 15:19:16.000000000","message":"Patch Set 3:\n\n\u003e \u003e \u003e I don\u0027t have strong opinion here. I can move them to another\n \u003e \u003e folder\n \u003e \u003e \u003e if you see fit.\n \u003e \u003e\n \u003e \u003e I think it will be probably better to create additional tree\n \u003e \u003e paralel to scenario, maybe called API (as they do in tempest or\n \u003e if\n \u003e \u003e you have other idea something else) and put it there\n \u003e \u003e Not sure if it should be done now or some other patch though\n \u003e \n \u003e If you agree, we can handle this improvement in a follow up patch.\n\nI agree","accounts_in_message":[],"_revision_number":3},{"id":"f45e245fd5cb866599a9a6e640256e4c818c5e8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 18:11:32.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4ddaeeffab04eefba371a22d3c92c09 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3629e55c4f0843d59d120711c82bc87e : SUCCESS in 3m 02s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/bb26e2bc7c094ac6a3484f105a3e1bb7 : SUCCESS in 1h 13m 15s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/c373142939ca41018b9bc0e1a2abd87d : FAILURE in 1h 33m 08s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/373b2d40826c4538bf92c127b8ce6064 : SUCCESS in 1h 17m 57s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/3bd0af5081e94ccd9140e08c730d4c18 : TIMED_OUT in 3h 04m 23s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/757dcb26730d46edb84e5c74d31c2f0d : SUCCESS in 1h 14m 56s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/41234348a71e448c9f4a1f834011ba74 : FAILURE in 29m 00s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/9418744fee6f4edba233fde7f2b35c81 : FAILURE in 29m 55s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/565ef816ce0943f6a0298d45c8acae0f : SUCCESS in 1h 40m 53s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/59d6403538d947cbae1e26d56adf1006 : SUCCESS in 1h 48m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"813008e18b7616df15f07d327f28914676ed9012","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-11-07 09:27:29.000000000","message":"Patch Set 3:\n\ncheck experimental","accounts_in_message":[],"_revision_number":3},{"id":"c4817557f8082a5a328c42972373d7814a7912f9","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 12:38:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/484f08e104c64d77b196665624db0c0c : FAILURE in 2h 20m 11s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/282d862f869b408bbe395a3ab2eca510 : FAILURE in 1h 30m 23s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/8c141e1178204da49edefcd64d65210e : FAILURE in 1h 58m 17s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/d42fa9712e1e444cb03987b3d3eedf31 : FAILURE in 1h 40m 35s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/ee05af6e07c04bceaa96457293ea634a : TIMED_OUT in 3h 04m 08s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/aedca4eefc604bc9ab6a33be604af4f2 : SUCCESS in 1h 34m 48s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/e92b2267c2d141249d9a888983a9f629 : FAILURE in 2h 05m 25s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/13cbe687e2584a84ab581d0870c10f1e : FAILURE in 1h 30m 47s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/27693539020e4bd48f156b28d770c1e9 : TIMED_OUT in 2h 16m 11s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/6da24f2654aa43b7af1c6f13bf145391 : FAILURE in 27m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"533a2bb7a282a53ffd0e68e814c669a4a9ca3bf1","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-11-07 13:15:58.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"3b948edec83b7aefa2af869977ca223ac35d1e4f","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-11-07 13:16:16.000000000","message":"Patch Set 4:\n\ncheck experimental","accounts_in_message":[],"_revision_number":4},{"id":"e33677b8184a97e32e29baa6647eacbc6f06c856","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 16:23:21.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0ca7496bfc44e85946d0d0aa788236b : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c89c8e7042a4c2c973c1f15d9c86258 : SUCCESS in 3m 36s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/a9d0e78214124fc3b9c3d674feb479f9 : SUCCESS in 1h 51m 59s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/a7189620eac14d7dabf0928c86032491 : FAILURE in 2h 04m 41s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/3901755fb40e4ff39952a0c793ff010e : SUCCESS in 1h 35m 45s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/e53c6090d03d4bbb86beb0fddc2e02fa : TIMED_OUT in 3h 05m 49s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/17ec8a6062c645a7a0f7c45790862cfd : FAILURE in 1h 53m 17s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/ea1966701e894ffe9abee4ac0e8e7aa7 : SUCCESS in 59m 24s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/9a34c5b7c910414f9d026134d15f5eb9 : FAILURE in 2h 00m 55s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/1e362565f0f049beb2ef0b1e5ebef5f1 : FAILURE in 1h 38m 25s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/92995132eea14989bd24cd1df1654dab : SUCCESS in 1h 22m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8ad8f475543e584f3432723c1c8c4cc9f4366384","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 16:24:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/7883fa0a6a7e4c1c94df9ecac27c35f4 : FAILURE in 2h 17m 07s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/78132234b1944008b51963615f77defe : FAILURE in 2h 04m 32s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/56a063b37c8c42c884fc6a3d18576dbf : FAILURE in 1h 59m 36s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/e6103608ef324e998f094455f85afc3c : FAILURE in 1h 49m 00s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/f4aacd4cbdaa4c4f941f6a21f7db2c22 : TIMED_OUT in 3h 06m 43s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/9fa7f79483b94c519c9642d1f3c78c1a : SUCCESS in 1h 28m 22s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/53011c1acc4c4f3b831d980c961a0343 : FAILURE in 1h 44m 40s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/0f248c7a2b6b46d5a62ee81619a3b2fa : FAILURE in 2h 38m 32s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/471342b0303949729fdac34a59720ec2 : TIMED_OUT in 2h 16m 33s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/6a18d54a0ba14a4dbeafe686878c09a1 : FAILURE in 24m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f5c80b7ae28c2a93461616be8c7c84334626b779","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-11-07 21:25:46.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"a885fee9213251183d6bf91e9c83cbfb4ed827d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 23:24:59.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6ca552a465a4a44a76df3213ca96d7b : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62bbe21a1bec4c5298e9304382a47e1f : SUCCESS in 3m 07s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/80cc7390b3354bd38b4673caf92a5c98 : FAILURE in 37m 50s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/f7de1389365d40f398af4b1ccba4dbfa : FAILURE in 1h 57m 35s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/7f2e87fcdac245769da40075bd71158b : SUCCESS in 1h 50m 28s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/9f97215f37d3467f8878728c764e68a9 : FAILURE in 1h 20m 07s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/6bbe88a755cf4ca3805c456742ebfb13 : SUCCESS in 1h 56m 32s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/44322f06da554fc0914f05f5108da696 : FAILURE in 38m 05s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/c3c19b54ff764500b3bcee5044a08e5c : SUCCESS in 1h 56m 06s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/b1a15791a6d245558930c2cc200fec96 : SUCCESS in 1h 31m 57s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/559fa3b10c694698b303775e1f3bb74a : SUCCESS in 1h 24m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"013e34095e1714da0363cc2dbc7fbfb66a2d4ce9","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-11-12 15:08:48.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"4a2b811e0fa80491f21a88b7d88a01bc9e1c1ec5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 18:14:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6eb968a54b5b4f1481d7350795dff12d : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2fe1490127442cfb09f41940cd0aba9 : SUCCESS in 3m 15s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/1974093c2b394bf5a9b173ae6dfc07d1 : SUCCESS in 1h 16m 01s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/87b40f1de42c47818d32986b05dfe923 : FAILURE in 1h 26m 37s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/a2f2a4371ad74af29d30f1c521521ca3 : SUCCESS in 1h 33m 22s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/0e4b30d41c6c4e87a5d17804ba55f5d7 : TIMED_OUT in 3h 04m 58s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/7a896954190a4981b43cc2d6f3810e4f : SUCCESS in 1h 25m 46s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/28f9c23f8f354135b7096211f7d20c70 : SUCCESS in 1h 04m 44s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/49003ba6097b485485979622fe150c12 : SUCCESS in 1h 13m 39s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/1d1e5558db0146c49544d6746b3874f6 : SUCCESS in 1h 58m 22s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/06ff80a32f724524a92b3f7f27a675f0 : SUCCESS in 1h 33m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f430f0672f76e89be5538cc6f8b966f14544cbab","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-11-13 10:29:20.000000000","message":"Patch Set 4:\n\ncheck experimental","accounts_in_message":[],"_revision_number":4},{"id":"0f3adb9c6d2878e1c99193350ac355fcae5c688f","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 13:32:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/4b34294d44844a6e804aa06bba3bf63d : FAILURE in 2h 18m 12s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/0b2b5bd8781b4138b0bacdf587ce2bcd : FAILURE in 1h 01m 54s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/d0e731776fd7409abda5e3c16fbf77ef : FAILURE in 2h 37m 32s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/1ffa5a261b744598865f8fd4fe94d614 : FAILURE in 1h 22m 55s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/a9803c43bf584918a5a2dc82c5d80a68 : FAILURE in 3h 01m 30s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/d23cf6cdd46a4bb1ab53e7c1e2ebdb34 : SUCCESS in 1h 36m 47s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/b9930f1c09004c3fbf8e9e80760fb527 : FAILURE in 1h 50m 32s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/f1655fb08011403995b6b10d4e9398c1 : FAILURE in 1h 32m 01s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/428e303e920b43efb492843d07952f39 : SUCCESS in 2h 07m 39s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/54e0c58e2db6424dae72b04b2bdda8a2 : FAILURE in 26m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9a7d2b039df1151146faf785117f03b336cfc560","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-11-14 10:42:30.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"975d8314bfaa2493870e7b24f735cb77c344cf03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 13:54:05.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ce54a99c1ec4416a61aa10d8c46d2c4 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c04e700de3c49caa8f8d7e510e699ff : SUCCESS in 3m 28s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/b319a81137e34dceb1065dda01b25e33 : SUCCESS in 1h 37m 37s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/a000c7056f624d1baf3617704b923d4e : FAILURE in 1h 26m 14s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/9f0f55a1c63348e0a7d64460baaf6176 : SUCCESS in 1h 28m 34s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/d8be8cba25e343af95d6d517e7b2f64e : TIMED_OUT in 3h 04m 24s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/84d7d26935b6496d8d2b38e2038bb85f : SUCCESS in 1h 34m 27s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/139876d9f8f84df1bf57c318d61ed103 : SUCCESS in 1h 24m 32s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/bea473202f544efa8a8c1791d52dc7f9 : SUCCESS in 1h 35m 09s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/9fa66fea2de24a89adfb9b111f3bb7cf : SUCCESS in 1h 37m 38s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/1073164a7b894d7dbba0fb12c25cde96 : SUCCESS in 1h 25m 03s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cf57cd3af3ee37c4138b5039f989a665e61be0bb","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-11-14 13:54:32.000000000","message":"Patch Set 5:\n\ncheck experimental","accounts_in_message":[],"_revision_number":5},{"id":"790a27a1d013a5d432f2918cb645fced5bc9ed86","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 17:07:56.000000000","message":"Patch Set 5:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/af7f54f7fbd14d5f8dc83d74b8fc7058 : FAILURE in 2h 46m 37s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/997c4eb3e7da41579dfcee8f44056685 : FAILURE in 2h 10m 17s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/0119ce95ecc8461b9547c16b63093349 : FAILURE in 1h 56m 29s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/01eabc27154c4bbdb70eca39a25ee9b1 : FAILURE in 1h 45m 56s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/1a3fa2380dd54dccbac4fd6b30d0aa25 : TIMED_OUT in 3h 09m 59s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/07db5f96c2e04a208290b041943eb070 : SUCCESS in 1h 30m 26s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/d0ab0620aca648149146ccc5f0f73bca : FAILURE in 2h 48m 03s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/2ac9496a71a44db58490020113055040 : FAILURE in 1h 47m 53s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/04063c84a8cd4c9bb5b4ea8ab7de705d : TIMED_OUT in 2h 21m 40s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/f7ee92212210445a84c79d1b0f2c49c2 : FAILURE in 27m 06s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e9e67725728d78c787c657c215fbce52f3779892","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-11-14 17:11:36.000000000","message":"Patch Set 5:\n\ncheck experimental","accounts_in_message":[],"_revision_number":5},{"id":"e901e4c4515d2be925ae736d4bf97b6957e44890","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 20:22:37.000000000","message":"Patch Set 5:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/8c3b69729a904d50bc45fe891df431e4 : FAILURE in 2h 21m 08s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/5c9b8394c6824c33ae19526e17e12f34 : FAILURE in 2h 51m 11s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/a5cfb6c77b7f4cafb84b488f5293dd3c : FAILURE in 2h 42m 01s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/6f78f0c2ba5040f6a67e860c8a1fef0a : SUCCESS in 1h 47m 53s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/d656f6e4420246f9b937f0447217ddda : TIMED_OUT in 3h 04m 35s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/7800e2bf3efc4743b618bf585a78b99f : SUCCESS in 1h 11m 39s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/c187586850b7450283fd2f6355cfec24 : FAILURE in 1h 52m 45s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/004ea26d78dc4bf4b6fbe0ff2e1c7611 : FAILURE in 1h 58m 09s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/258c06d72c7e447f806f0a8c5f5de9f6 : TIMED_OUT in 2h 15m 18s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/9baf6c7055ab414eae3bea5ef0490037 : FAILURE in 34m 56s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b73a851a2cde0287833ea18418e5753898824cd6","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-11-15 09:02:56.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"00267f7d8ee069e0ea869934d4379f6571eb445d","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-11-15 09:03:45.000000000","message":"Patch Set 5: Workflow-1\n\nThis will not work on previous versions. I\u0027m working on it.","accounts_in_message":[],"_revision_number":5},{"id":"78c85e928543e2056dc35727d35090a92a666684","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-11-15 10:30:55.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"da5c2db9aaa410d7b8e87c07d1cc776e60d4db5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 13:36:30.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e1e874268f94f6a82bd42b431c75f8b : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64aae5a43b09444395af35a8047d00e8 : SUCCESS in 3m 10s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/3e907fceb5b14cdcaee0efebdd2e66ad : SUCCESS in 1h 09m 06s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/c04bc11c7799497ead00e5b814bf8836 : FAILURE in 1h 37m 15s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/122e9e1dae784ecf9dc7bb1d619ef479 : SUCCESS in 1h 11m 10s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/729347752a364aa29d42458229245fb3 : TIMED_OUT in 3h 03m 40s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/f5759377b29e451aa0058b3ca71a868c : SUCCESS in 1h 18m 05s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/9fafc77e3edf4102aa2235193b671089 : SUCCESS in 1h 15m 12s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/6a74fbba195b42c4b9c5adc99e7bff96 : SUCCESS in 1h 26m 12s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/b9d390028e5e420f9b9967bf0a0a45e5 : SUCCESS in 1h 27m 36s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/b3d5cb593d284e22bb0a280ef4259b2b : SUCCESS in 1h 38m 10s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ca04df942b37f732bb2b017e74070cbefbbb6b71","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-11-15 13:37:03.000000000","message":"Patch Set 6:\n\ncheck experimental","accounts_in_message":[],"_revision_number":6},{"id":"6f1bae2c146ac59cc67df876499974b5c6b47e57","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 16:45:31.000000000","message":"Patch Set 6:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/2856937824e745598df10ff4b80f95de : FAILURE in 2h 18m 50s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/5372bfaeca6840e5aa3dfcf01e9a4baa : FAILURE in 1h 38m 57s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/e8d4020d64fc4bada1aafe7b52408cdf : FAILURE in 1h 53m 48s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/5ee93a99661a473f8983597ef8b1b6a3 : SUCCESS in 1h 39m 10s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/3e6d693e5c6546c0a398d0f7489e4c79 : TIMED_OUT in 3h 04m 01s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/fe8b0d8f11ab4de7910ecf66bb4cec66 : SUCCESS in 1h 11m 53s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/99298cf0cc9c44b59bcb236f846e69ea : FAILURE in 2h 18m 52s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/c19829f21bf8469284e9ca2adab354d5 : FAILURE in 1h 34m 05s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/aff6e3c8260b4755adfdcd7b05a8bb72 : TIMED_OUT in 2h 17m 03s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/9b3e957092e0419aab11322bd8909869 : FAILURE in 30m 48s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"96108c1f2f14e86eaee958b6599b94cef410c3ec","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-11-18 08:50:11.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)\n\nJust one question!","accounts_in_message":[],"_revision_number":6},{"id":"e81d1a96de884e1e8dc6bc72c09f78b862fd0f0d","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-11-18 09:10:30.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d9b14320ff9809c04101cca54f5f9df703cb51dc","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-11-18 09:21:47.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"f6e612bd65a86347fd32d249b3e832ae8929a6cf","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-11-18 10:00:50.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"4f265ac22591c55623388228d25e5447ea3d7f3a","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-11-18 10:06:16.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"bc9417d0b8ccbda5c27fa89faa47f0367614a8c6","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-11-18 10:34:47.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"04aff6d16e091dca0d1318f5ea3c5921e926f216","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-11-18 10:36:51.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"2110641ad01660d062f01eef1fe85f20c59e21cd","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-11-18 12:31:19.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"ff6040af3715bbc906803271e8d423e30573b948","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-11-18 12:55:50.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"7fe629ef72eb2120460a97a681318cf5ad473ea9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 15:42:10.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccb93434d2984487bdacba396c547a8a : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8795d00fabd45f386e7841f333d994b : SUCCESS in 3m 10s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/fb93a1b0928d416483bddba8db0b5664 : SUCCESS in 1h 23m 05s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/f7b4be645cc54dfe9acd9eda5f69fb62 : FAILURE in 1h 55m 29s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/7ae7b6617603440b807ab5ad61906a46 : SUCCESS in 1h 59m 45s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/b05f7e248c7c464487cb43d47d9fcb78 : TIMED_OUT in 3h 07m 20s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/4f8cc41253f743d89aa9a3f6c11082df : SUCCESS in 2h 09m 51s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/1955c95c93884dfead0bc4ed3696a1f9 : SUCCESS in 1h 34m 40s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/57060fff02144a249772fc9aee9d08e5 : SUCCESS in 1h 57m 03s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/0e983bd11bd94c0481e784e59db13ba6 : SUCCESS in 2h 22m 45s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/f548cbd93a454b8b9b84d491dc89a94a : POST_FAILURE in 1h 45m 55s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"c9c7e67667320db339bb2e89b4b3b318cfbb716c","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-11-18 15:42:32.000000000","message":"Patch Set 9:\n\ncheck experimental","accounts_in_message":[],"_revision_number":9},{"id":"6f33620b4d1d0308037dd9fdc29cf1cf51044824","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 18:55:31.000000000","message":"Patch Set 9:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/7d0e3e9e4f57434ba616bb7033a22fd0 : FAILURE in 2h 15m 55s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/b754455641e8439dbde4e83d6b7510cc : FAILURE in 1h 53m 16s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/bb27a2c98f4c435c9ec76a1a32476c5d : FAILURE in 2h 08m 30s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/c3ca4a54c2cd4d3f874d9b234f8d2fbe : FAILURE in 1h 39m 55s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/a2696aa9809e497380c52d37c8b44a13 : TIMED_OUT in 3h 05m 08s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/5fde66ecb1c247948d4dc2af56ad4920 : SUCCESS in 1h 51m 15s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/2cc631abb5c74afc9e8eaab87429c243 : FAILURE in 1h 55m 00s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/653c1612504949d0b582faf380584eb8 : FAILURE in 1h 39m 58s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/772a774c9f554ec8b563dcdad50d9152 : FAILURE in 1h 27m 51s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/a99f9115905f4895b8e2a66d639a8a05 : FAILURE in 30m 18s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"5f9bf7036fd37e46913fe830d0de8a0d5f7c1e5f","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-11-18 19:37:33.000000000","message":"Patch Set 9:\n\ncheck experimental","accounts_in_message":[],"_revision_number":9},{"id":"8365ef0174f93fa78a1364cc948c23d8f1278781","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 22:49:31.000000000","message":"Patch Set 9:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/b0716cd0a8d046308cbe6bb62ffaf015 : FAILURE in 2h 01m 20s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/217324f9a84f420fb58fa37f7f26b66c : FAILURE in 1h 36m 33s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/6b53d06d18c54b48bcd420c1b54d80b0 : FAILURE in 2h 03m 18s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/24b34615e3a642e7a5b885aba0e8f425 : SUCCESS in 1h 39m 42s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/ea3be65e2d0b43cc9b0fa253d138262e : TIMED_OUT in 3h 08m 17s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/a54375a0cb8d459ab388dc0e8c0c9358 : SUCCESS in 1h 21m 03s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/352fd958c3604a89a2fbdb6c5ca76062 : FAILURE in 1h 03m 37s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/b8fb1e6239244309a7d1f5ea941cceae : FAILURE in 1h 39m 38s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/12b183a1364a457fb1a992cb6ac0b206 : SUCCESS in 2h 10m 29s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/14ecec1495d9424ebb939cb9c41a1889 : FAILURE in 28m 10s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"0f895d2477056efe2b4473eb3b0cd12823d0854f","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-11-19 09:45:38.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"24ce99bd9877f54d4237a863475fad5fb7f7b264","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 09:45:49.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"568fc17a37e5f560a499e25d50dc9c4cc67251a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 11:21:04.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21abd5f153dd4b60996754a3039a668a : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1458520b4d194cf0a8e1ea091e833efc : SUCCESS in 2m 54s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/8810087adef84d8dae77835c972ddd51 : FAILURE in 23m 00s\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/3aa1c6fa25f848c78bf14ddc92caebb5 : SUCCESS in 1h 13m 53s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/9abfa1aeff3040d5ac0579c10279bef9 : SUCCESS in 1h 30m 04s","accounts_in_message":[],"_revision_number":9},{"id":"b565fcb9c717931a9f11275e1ed181e09d4f265c","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-11-19 15:21:01.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"5f691070cfcddf7272545442214a36fe6455a38d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 16:37:45.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb9ac89d89864f7caac00f43c6ce574e : SUCCESS in 2m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d37499afa6b49cb9bdac912d7ce2a15 : SUCCESS in 2m 42s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/4ba6363c497e40b6a219dbb277aa6d23 : FAILURE in 15m 59s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/dee71874026240b59d3f71f419cbaa6e : FAILURE in 13m 33s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/3b45485af5384c5ca0722a9d27fed3e4 : FAILURE in 13m 05s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/f52ac1f51f1549ecae4ba82a27e99315 : FAILURE in 18m 28s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/b4fac4253c344e0083fbdd31b3be2c5f : FAILURE in 18m 42s (non-voting)\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/b2b3bf2fb4d14e1a97a4a869a4bfb1b9 : FAILURE in 24m 44s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"3cd1ed16fd77154f370f19ae7f59b09dd5e63910","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-11-19 16:37:53.000000000","message":"Patch Set 9:\n\nIt\u0027ll need one more recheck, patch fixing gates just got merged.","accounts_in_message":[],"_revision_number":9},{"id":"dfd6626e523943795d443d74f67beb9c1a8b3e2d","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-11-19 16:38:10.000000000","message":"Patch Set 9:\n\nrecheck\n\nHa, exactly on time. xD","accounts_in_message":[],"_revision_number":9},{"id":"782ab1628e4f3f2f2a866e32dea27f797e83c581","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-11-19 16:38:15.000000000","message":"Patch Set 9:\n\ncheck experimental","accounts_in_message":[],"_revision_number":9},{"id":"3440b4c38b8345003051f93198a42054c1f24c50","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-11-19 18:27:02.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"613c37e84a235cefba5d42579310910f303d7c2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 21:40:18.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9d5de6dcf6e40dc80e777bc2e26c6fa : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/506238d0d1c74bffb7abe6f7c714f466 : SUCCESS in 3m 05s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/9e171b2c94db44db88071b0e3a9ff8f5 : SUCCESS in 1h 24m 49s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/089410f59d7449a79e260d6b1e1b35f2 : FAILURE in 1h 44m 54s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/be9a278041b24deda0d220d45aea606e : POST_FAILURE in 1h 29m 36s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/8daed56a02c6427fa28b524298b9590e : TIMED_OUT in 3h 04m 13s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/d970d802eea640b481426fda2f26372f : SUCCESS in 1h 26m 22s (non-voting)\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/f6fccd64b5254b71a8a4520a2877c8fa : SUCCESS in 1h 46m 26s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"257544f07aaa95df76e5d3ae66f18b3c573552db","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-11-20 07:34:08.000000000","message":"Patch Set 10: Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"96824dc9be09cf2fd14a92e952f88a8268714b33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 07:34:19.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"060122a3c1fcdd9b0e2347478757b7e33a8ed154","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 09:22:35.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ef7c0a2383f443ea3ade74c0e24ebb4 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/540588cc170244caa252f9588ff45341 : SUCCESS in 3m 04s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/5283c1a32d0244999c5973b5e4acf09b : SUCCESS in 1h 30m 47s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/8557eecb834649e682e1eb06f7e89924 : SUCCESS in 1h 44m 24s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"21143fb860f2d9f841ead0345cddb57011273916","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 09:22:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"a2a52b1ecbd9149428cc9a605170c0fdb1d6cbd9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 09:24:06.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ffe8656c9e144b1b01475fdaa6e524f : SUCCESS in 53s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"feaef8ed862d80250e9f7777301f79ef18b93b21","revisions":{"c57d2a7837435848d36c1934d86c8b205661fce2":{"kind":"REWORK","_number":1,"created":"2019-11-04 15:25:40.000000000","uploader":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"ref":"refs/changes/44/692844/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/44/692844/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/1"}}},"commit":{"parents":[{"commit":"b197432a1d65415f58e85c9bcc22786c7ea30e79","subject":"Merge \"Adjust expected number of ports to config values\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/b197432a1d65415f58e85c9bcc22786c7ea30e79"}]}],"author":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2019-11-04 11:39:54.000000000","tz":60},"committer":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2019-11-04 15:22:10.000000000","tz":60},"subject":"Ensure proper failure message on CRD validation","message":"Ensure proper failure message on CRD validation\n\nThe error message retuned when a CRD is not validated has changed\non Kubernetes 1.16. We should also update the NP tempests tests\nto match it, as the gate is using Kubernetes 1.16.\n\nChange-Id: I4ed2bdbbc664579689763f2759741b0efced6db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c57d2a7837435848d36c1934d86c8b205661fce2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c57d2a7837435848d36c1934d86c8b205661fce2"}]},"branch":"refs/heads/master"},"884b9b125dee025652d9b84005c2bf38bc06657e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-11-04 15:36:02.000000000","uploader":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"ref":"refs/changes/44/692844/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/44/692844/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/2"}}},"commit":{"parents":[{"commit":"b197432a1d65415f58e85c9bcc22786c7ea30e79","subject":"Merge \"Adjust expected number of ports to config values\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/b197432a1d65415f58e85c9bcc22786c7ea30e79"}]}],"author":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2019-11-04 11:39:54.000000000","tz":60},"committer":{"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","date":"2019-11-04 15:35:56.000000000","tz":0},"subject":"Ensure correct failure message is checked on CRD validation","message":"Ensure correct failure message is checked on CRD validation\n\nThe error message retuned when a CRD is not validated has changed\non Kubernetes 1.16. We should also update the NP tempests tests\nto match it, as the gate is using Kubernetes 1.16.\n\nChange-Id: I4ed2bdbbc664579689763f2759741b0efced6db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/884b9b125dee025652d9b84005c2bf38bc06657e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/884b9b125dee025652d9b84005c2bf38bc06657e"}]},"branch":"refs/heads/master"},"6a4cff1eaa5bc14a23a559d6f9722149775db93e":{"kind":"REWORK","_number":3,"created":"2019-11-06 15:01:47.000000000","uploader":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"ref":"refs/changes/44/692844/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/44/692844/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/3"}}},"commit":{"parents":[{"commit":"b197432a1d65415f58e85c9bcc22786c7ea30e79","subject":"Merge \"Adjust expected number of ports to config values\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/b197432a1d65415f58e85c9bcc22786c7ea30e79"}]}],"author":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2019-11-04 11:39:54.000000000","tz":60},"committer":{"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","date":"2019-11-06 15:01:41.000000000","tz":0},"subject":"Ensure correct failure message is checked on CRD validation","message":"Ensure correct failure message is checked on CRD validation\n\nThe error message retuned when a CRD is not validated has changed\non Kubernetes 1.16. We should also update the NP tempests tests\nto match it, as the gate is using Kubernetes 1.16.\n\nChange-Id: I4ed2bdbbc664579689763f2759741b0efced6db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/6a4cff1eaa5bc14a23a559d6f9722149775db93e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/6a4cff1eaa5bc14a23a559d6f9722149775db93e"}]},"branch":"refs/heads/master"},"c1c0dcdabaefbe83a8ce6989222af15f1c41b9b0":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-11-07 13:15:58.000000000","uploader":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"ref":"refs/changes/44/692844/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/44/692844/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/4"}}},"commit":{"parents":[{"commit":"4175a85a1609f093987d24903d7f6d1592ad7c7a","subject":"Testing basic Network policy IPBlock functionality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/4175a85a1609f093987d24903d7f6d1592ad7c7a"}]}],"author":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2019-11-04 11:39:54.000000000","tz":60},"committer":{"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","date":"2019-11-07 13:15:58.000000000","tz":0},"subject":"Ensure correct failure message is checked on CRD validation","message":"Ensure correct failure message is checked on CRD validation\n\nThe error message retuned when a CRD is not validated has changed\non Kubernetes 1.16. We should also update the NP tempests tests\nto match it, as the gate is using Kubernetes 1.16.\n\nChange-Id: I4ed2bdbbc664579689763f2759741b0efced6db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c1c0dcdabaefbe83a8ce6989222af15f1c41b9b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c1c0dcdabaefbe83a8ce6989222af15f1c41b9b0"}]},"branch":"refs/heads/master"},"760fe72356814d17876aae02ae1a590782eadf0d":{"kind":"REWORK","_number":5,"created":"2019-11-14 10:42:30.000000000","uploader":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"ref":"refs/changes/44/692844/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/44/692844/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/5"}}},"commit":{"parents":[{"commit":"4175a85a1609f093987d24903d7f6d1592ad7c7a","subject":"Testing basic Network policy IPBlock functionality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/4175a85a1609f093987d24903d7f6d1592ad7c7a"}]}],"author":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2019-11-04 11:39:54.000000000","tz":60},"committer":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2019-11-14 10:41:40.000000000","tz":0},"subject":"Ensure correct failure message is checked on CRD validation","message":"Ensure correct failure message is checked on CRD validation\n\nThe error message retuned when a CRD is not validated has changed\non Kubernetes 1.16. We should also update the NP and Namespaces\ntempests tests to match it, as the gate is using Kubernetes 1.16.\n\nChange-Id: I4ed2bdbbc664579689763f2759741b0efced6db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/760fe72356814d17876aae02ae1a590782eadf0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/760fe72356814d17876aae02ae1a590782eadf0d"}]},"branch":"refs/heads/master"},"4b8ae9e15fc1837ae696c632d690ab59c91e74fb":{"kind":"REWORK","_number":6,"created":"2019-11-15 10:30:55.000000000","uploader":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"ref":"refs/changes/44/692844/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/44/692844/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/6"}}},"commit":{"parents":[{"commit":"4175a85a1609f093987d24903d7f6d1592ad7c7a","subject":"Testing basic Network policy IPBlock functionality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/4175a85a1609f093987d24903d7f6d1592ad7c7a"}]}],"author":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2019-11-04 11:39:54.000000000","tz":60},"committer":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2019-11-15 10:30:44.000000000","tz":0},"subject":"Ensure correct failure message is checked on CRD validation","message":"Ensure correct failure message is checked on CRD validation\n\nThe error message retuned when a CRD is not validated has changed\non Kubernetes 1.16. We should also update the NP and Namespaces\ntempests tests to match it, as the gate is using Kubernetes 1.16.\n\nChange-Id: I4ed2bdbbc664579689763f2759741b0efced6db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/4b8ae9e15fc1837ae696c632d690ab59c91e74fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/4b8ae9e15fc1837ae696c632d690ab59c91e74fb"}]},"branch":"refs/heads/master"},"ca6398c1b727309b29e67c2ddc4724fd3f2a0fe7":{"kind":"REWORK","_number":7,"created":"2019-11-18 09:21:47.000000000","uploader":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"ref":"refs/changes/44/692844/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/44/692844/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/7"}}},"commit":{"parents":[{"commit":"4175a85a1609f093987d24903d7f6d1592ad7c7a","subject":"Testing basic Network policy IPBlock functionality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/4175a85a1609f093987d24903d7f6d1592ad7c7a"}]}],"author":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2019-11-04 11:39:54.000000000","tz":60},"committer":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2019-11-18 09:21:37.000000000","tz":0},"subject":"Ensure correct failure message is checked on CRD validation","message":"Ensure correct failure message is checked on CRD validation\n\nThe error message retuned when a CRD is not validated has changed\non Kubernetes 1.16. We should also update the NP and Namespaces\ntempests tests to match it, as the gate is using Kubernetes 1.16.\n\nChange-Id: I4ed2bdbbc664579689763f2759741b0efced6db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ca6398c1b727309b29e67c2ddc4724fd3f2a0fe7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ca6398c1b727309b29e67c2ddc4724fd3f2a0fe7"}]},"branch":"refs/heads/master"},"98136215c435f8ad4724621f788549a8a95c9bae":{"kind":"REWORK","_number":8,"created":"2019-11-18 10:36:51.000000000","uploader":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"ref":"refs/changes/44/692844/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/44/692844/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/8"}}},"commit":{"parents":[{"commit":"4175a85a1609f093987d24903d7f6d1592ad7c7a","subject":"Testing basic Network policy IPBlock functionality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/4175a85a1609f093987d24903d7f6d1592ad7c7a"}]}],"author":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2019-11-04 11:39:54.000000000","tz":60},"committer":{"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","date":"2019-11-18 10:36:48.000000000","tz":0},"subject":"Ensure correct failure message is checked on CRD validation","message":"Ensure correct failure message is checked on CRD validation\n\nThe error message retuned when a CRD is not validated has changed\non Kubernetes 1.16. We should also update the NP and Namespaces\ntempests tests to match it, as the gate is using Kubernetes 1.16.\n\nChange-Id: I4ed2bdbbc664579689763f2759741b0efced6db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/98136215c435f8ad4724621f788549a8a95c9bae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/98136215c435f8ad4724621f788549a8a95c9bae"}]},"branch":"refs/heads/master"},"b702a0697aedc5c06df1ed271fdf3bd13db55df9":{"kind":"REWORK","_number":9,"created":"2019-11-18 12:31:19.000000000","uploader":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"ref":"refs/changes/44/692844/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/44/692844/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/9"}}},"commit":{"parents":[{"commit":"4175a85a1609f093987d24903d7f6d1592ad7c7a","subject":"Testing basic Network policy IPBlock functionality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/4175a85a1609f093987d24903d7f6d1592ad7c7a"}]}],"author":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2019-11-04 11:39:54.000000000","tz":60},"committer":{"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","date":"2019-11-18 12:31:12.000000000","tz":0},"subject":"Ensure correct failure message is checked on CRD validation","message":"Ensure correct failure message is checked on CRD validation\n\nThe error message retuned when a CRD is not validated has changed\non Kubernetes 1.16. We should also update the NP and Namespaces\ntempests tests to match it, as the gate is using Kubernetes 1.16.\n\nChange-Id: I4ed2bdbbc664579689763f2759741b0efced6db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/b702a0697aedc5c06df1ed271fdf3bd13db55df9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/b702a0697aedc5c06df1ed271fdf3bd13db55df9"}]},"branch":"refs/heads/master"},"feaef8ed862d80250e9f7777301f79ef18b93b21":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-11-19 18:27:02.000000000","uploader":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"ref":"refs/changes/44/692844/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/44/692844/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/44/692844/10"}}},"commit":{"parents":[{"commit":"ffae68805fac442a0d3fa1db08d25957c0741646","subject":"Skip Namespace isolation when NP is enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ffae68805fac442a0d3fa1db08d25957c0741646"}]}],"author":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2019-11-04 11:39:54.000000000","tz":60},"committer":{"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","date":"2019-11-19 18:27:02.000000000","tz":0},"subject":"Ensure correct failure message is checked on CRD validation","message":"Ensure correct failure message is checked on CRD validation\n\nThe error message retuned when a CRD is not validated has changed\non Kubernetes 1.16. We should also update the NP and Namespaces\ntempests tests to match it, as the gate is using Kubernetes 1.16.\n\nChange-Id: I4ed2bdbbc664579689763f2759741b0efced6db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/feaef8ed862d80250e9f7777301f79ef18b93b21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/feaef8ed862d80250e9f7777301f79ef18b93b21"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
