)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~693632","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~I6ce0b16985de6cc7182026975a63f03844d9d9ee","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"service","hashtags":[],"change_id":"I6ce0b16985de6cc7182026975a63f03844d9d9ee","subject":"Add a check for ClusterIP IP of LB service","status":"MERGED","created":"2019-11-10 11:41:49.000000000","updated":"2019-12-02 16:28:32.000000000","submitted":"2019-12-02 16:27:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693632-1575304035167-56cf989d","meta_rev_id":"d2998bac5d8ccf79934fca8ab4ed9a33b803ac94","_number":693632,"virtual_id_number":693632,"owner":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"date":"2019-12-02 11:33:11.000000000","_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-02 16:27:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2019-12-02 09:22:45.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":1,"date":"2019-11-27 20:36:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-29 08:20:42.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":1,"date":"2019-12-02 09:22:45.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-11 06:35:18.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"},{"updated":"2019-11-27 20:36:36.000000000","updated_by":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"reviewer":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"state":"REVIEWER"},{"updated":"2019-11-29 08:20:42.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2019-12-02 09:22:45.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2019-12-02 16:27:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ac7ca5bafbef810a6ebb4f75b3d6d4c9c4d74aa0","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-10 11:41:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a4f24480bff190075a0d21f96b51919e79ed7cb","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-10 14:20:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2cfb650d755304cbec264dec6e46b8d60b9fbe05","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-10 14:21:02.000000000","message":"Patch Set 2:\n\ncheck experimental","accounts_in_message":[],"_revision_number":2},{"id":"cf8465ca14fc9ac93e394c97193d7afe28bd5b29","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-10 17:30:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/8b8ab8a59f3043cb9c0f2113ec26b954 : FAILURE in 1h 54m 25s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/f6cf2aefd1b2434ea02b153404e68016 : FAILURE in 2h 07m 08s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/d5d5258b223d47c088dd7eec57beafdb : FAILURE in 1h 05m 05s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/23a76d45a3cb4e9da7cd2ff1f02316d5 : FAILURE in 1h 39m 40s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/6ca182dd9b124e6e9768d00477ef005f : TIMED_OUT in 3h 03m 37s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/b25c99f50b3d4e08a127f3af75e803fd : SUCCESS in 1h 20m 18s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/4e8a6cbf847f43c59f9f941d734f0905 : FAILURE in 1h 49m 26s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/fc42558cb90f47c7ae286d5bded4619d : FAILURE in 1h 28m 59s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/cca72f01f9fc4237adc7c83fc403b60e : FAILURE in 1h 24m 56s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/f4e0d69af2494a0daf6c4b67bf7ffed3 : FAILURE in 25m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dbd30a4eb6cb052743a351bd19186390a7cb8613","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-10 17:31:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03fe2031d1a84056874ef8507314b482 : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/394888e8f3414124bbbbece088903158 : SUCCESS in 2m 40s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/52c593f01764438d8f2f7a779c1cff8a : SUCCESS in 1h 29m 05s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/10d9d06ac46b4db892f32169811a0662 : FAILURE in 1h 29m 34s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/2512e03eb35d4b06bb25207b8fea279d : SUCCESS in 1h 20m 25s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/346cd029da164ecdbc25239952c9494d : TIMED_OUT in 3h 04m 37s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/5df26852557648b8bf329028bc5415ed : SUCCESS in 1h 22m 47s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/1c0e0837a4434b31a4c9ba8869597761 : FAILURE in 2h 06m 35s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/5a80a9c68f9642ef8d956114bcba4392 : SUCCESS in 1h 17m 29s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/8735597687714ac4ab26807730d5f3ac : SUCCESS in 1h 52m 01s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/2945f7af4e294dc98ac17a14758f8ff6 : SUCCESS in 1h 29m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5ac8aa14b605df8b7656a7f817eb09cc9953c9d6","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-11 06:34:57.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"280ce62b5f293ad7ef6f67f1c1071302682e36f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 09:54:09.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0299e4173644c39be0af2d193425ec9 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c8733044fa34637baa8268d376b58b8 : SUCCESS in 4m 20s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/45744b2f618649d9afbbbe25e5b05391 : SUCCESS in 1h 28m 10s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/a7a043aa3a484f7b91775d4e800c08e7 : FAILURE in 2h 15m 09s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/1327dea9976e4328a7359209088446c5 : SUCCESS in 2h 10m 22s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/80e6ca33e56f4d72a04994771fe6e302 : TIMED_OUT in 3h 06m 05s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/bb0bd8abf40e43038a573ff5f8ab6f8d : SUCCESS in 1h 23m 43s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/0557772768d0472fa2352a5596286303 : SUCCESS in 1h 25m 01s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/2df6a45c2b534a0982ec39e8999ff704 : FAILURE in 1h 48m 12s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/bc0efa023d0a482fa0c5bfd893a67ecb : SUCCESS in 1h 38m 57s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/90e76d1e1b9a4855bf64f3ee80e2c204 : SUCCESS in 1h 14m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3a5eaaaf218320c8f8ac1e8d4b7c3a66423c30be","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-11 12:01:03.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5b9a5319cc813e23fa661a85a01689f3ae254619","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-11-11 13:11:26.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0aaf14061bc13c8268b19ddbb9ddf4ab01142a08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 15:11:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0feb23d573e43e2a29c705ef16dcd04 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ac7fe58070045d7a0268c8bedc12596 : SUCCESS in 3m 04s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/c3fa6d9d3a5e4a9ea5ddca0111eeb386 : SUCCESS in 1h 12m 40s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/47012977f3604af9a8c247b8bbd95aa2 : FAILURE in 1h 55m 51s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/fba87d4ccacf4015b66b7996cbf43299 : FAILURE in 29m 33s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/8508108dbe10450885ad33ec6d060f22 : TIMED_OUT in 3h 04m 48s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/0f9b01599565442db3b486d9b71cef32 : SUCCESS in 1h 36m 17s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/b8bb04b3e9fd409c9a7e6342c5adee98 : SUCCESS in 1h 09m 19s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/073ded6b13464e1d93cd48f3fd859f37 : SUCCESS in 1h 17m 48s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/8b35be3368144addadec56b7541f780f : SUCCESS in 1h 44m 52s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/95675d2464bf4e6fb1eab4a1aab31aa5 : FAILURE in 36m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b2f2d9e550772c9c2c23896d31229542e689bfa1","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-12 15:19:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ee1291221622cf7468eac3f1f188d14241eaf4e8","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-12 15:22:22.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"e254a5a6e747f589eec9e04da1d4d596d19b1e9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 18:26:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/765f2466a3d34b539b6738419f373020 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/535df2e956304fe89f678f5d0300602c : FAILURE in 3m 38s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/a2e2a041045d493b98d344f423581472 : FAILURE in 1h 22m 01s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/5b30374732b2452b8f24394293c69057 : FAILURE in 2h 00m 42s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/3f9493f6249e4d7a8c0fafc2da875d89 : FAILURE in 2h 08m 37s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/8dee4e3b16c242dab58f4947033b6206 : TIMED_OUT in 3h 04m 56s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/3eb6edbbb3174fac92bc4b9d62d2ac6b : FAILURE in 1h 27m 40s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/9cef0303941c45e1a13a50044c4105e1 : FAILURE in 1h 20m 02s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/56cd993a6d4043eb9328a6e28b6aab6b : FAILURE in 1h 30m 01s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/228c8b39a05142a39e9db82870020318 : FAILURE in 1h 42m 40s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/169f3fee971841cca32092e41470a01e : FAILURE in 1h 28m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1112f0f5b2d117e047958b2d10305505e5973358","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-12 21:09:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8a7a8ba648888e8a8833aa0466515f2ed6c5bed8","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-12 21:22:20.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"59b77ed9bdd0704fbc49eae1b75881058bf6cc0f","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-12 21:22:39.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"9b70746e362711b900619e3de316f1461976c1d7","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-12 22:21:08.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6e10cffe6f74425c79160488eb95724499cd5f94","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-12 22:25:10.000000000","message":"Patch Set 6:\n\ncheck experimental","accounts_in_message":[],"_revision_number":6},{"id":"47771b2beb0bde76ba1c1957e1456fae01e48496","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 01:28:22.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da45e976b3fd40889aa40eebf05f590d : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0c1bcc17c3b4989a31d2217510a9543 : SUCCESS in 4m 26s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/68dc1d59f71f4befa3a1b442d37f5703 : POST_FAILURE in 1h 36m 12s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/be98f10c10d14016884698091c71bdb5 : FAILURE in 1h 55m 36s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/46abdaf04cf342c6809ef95af25a953d : FAILURE in 2h 00m 00s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/314b7ee7609b4c3f98f55c1f00ebcf40 : TIMED_OUT in 3h 06m 21s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/44555f46f9f546ab8bb5a204d70784f0 : FAILURE in 2h 06m 13s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/3076a61608e24c788afdec5e68870783 : FAILURE in 1h 49m 26s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/8b89d8e780ba413dbf427a3b1b679a98 : FAILURE in 1h 58m 28s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/f5b6e8ed4402422cae934099de17ddb7 : FAILURE in 2h 22m 00s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/0434bc1587f74d969ea6d1e18c618c59 : FAILURE in 1h 45m 29s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"999083eab012df10397c3bdfb1daee6ca15a214b","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 01:35:54.000000000","message":"Patch Set 6:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/1f2da5165c3c4f72952264f94d18165d : FAILURE in 2h 07m 30s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/c35557f54c514bc2bc75aab466efc197 : FAILURE in 2h 13m 24s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/8afe1d78684344c2a1dc37b1f2b24eed : FAILURE in 2h 14m 54s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/1a39b7440767422783f2d0e684a5f8ff : FAILURE in 2h 18m 19s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/a6f571204fc24a1d837708c70cedb3db : TIMED_OUT in 3h 06m 41s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/c687d4d043c94c3fb566b0d6919e75be : FAILURE in 1h 39m 55s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/fc81008940ca46ffac43bcd0ac003b1b : FAILURE in 2h 33m 21s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/94cb263a21fd420f87bbd716198440f0 : FAILURE in 2h 17m 37s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/09c3ef2ae3eb4ccdbf6cdbef9e346919 : TIMED_OUT in 2h 18m 33s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/1e9f8bcfbfb147209fea801f462280fb : FAILURE in 27m 07s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c9c5c14a2198dc08110f4246983f8393867f1f3f","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-13 09:02:16.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"83d748b3860357b6427a710aaa328afa3da91dca","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-13 09:47:14.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"d2fcb2e6110b405e6f3d5a7010e297189fbb44b3","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-13 09:47:43.000000000","message":"Patch Set 8:\n\ncheck experimental","accounts_in_message":[],"_revision_number":8},{"id":"2b099942a3b6af5777028e8e013e6a56ef860888","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 12:56:53.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ec91a3475e1465e84bf2a58775862e3 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3528239c3e140d0ab33ab3ca190b2f2 : SUCCESS in 3m 03s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/28d5c1d8cf224235be879ad646ed1d1c : FAILURE in 1h 49m 58s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/94535c4ee86c48b7a137657c3838fb77 : FAILURE in 1h 36m 39s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/7d31f584598d4f5782c4d86b0932fec0 : SUCCESS in 1h 30m 30s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/01b4ab3427724f87ad784d18cd35f221 : TIMED_OUT in 3h 04m 05s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/f17c5894ee9142448456c40eaa102d88 : SUCCESS in 1h 35m 05s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/6935947b7920465a81869485368841c5 : SUCCESS in 1h 33m 14s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/75d4c2588d6345a189cefe62b6542af1 : SUCCESS in 1h 52m 14s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/275e5c3820974839a2dd6e2997d0d18c : SUCCESS in 1h 43m 59s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/bfe2717a0a214651b721ddfc41d96fd6 : SUCCESS in 1h 34m 51s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"360b65d9853b9583b1c87dc40c20bb4013008f10","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 12:58:32.000000000","message":"Patch Set 8:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/02b69e143c9a49b9a83263183d4c0e4b : FAILURE in 2h 11m 01s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/3704f635634e45caa1a7b1f20775b477 : FAILURE in 2h 05m 19s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/24297b6413db4df6b64280d7340624bf : FAILURE in 2h 39m 27s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/e08381ce702f44d28684ba9be85ab221 : FAILURE in 1h 45m 25s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/cfda376090f24b37bffc2477d695a98b : TIMED_OUT in 3h 04m 41s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/31ef7c4acbbb4b70886e2b9965c46ee4 : SUCCESS in 1h 16m 54s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/dabec32878284558a6f2455fdafa2864 : FAILURE in 1h 53m 10s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/ebbda3f9940044398d0422c49880c77a : FAILURE in 1h 46m 47s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/14731eb0b1fe4855aff4e871f10e31a0 : TIMED_OUT in 2h 16m 03s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/9ee433fafb2c46d5a97a5896fe545703 : FAILURE in 29m 40s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"6087afa554e98a11d358a8922b796477d45e202e","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-11-13 14:36:51.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"c2db6d9edd5104d9426c2e15b3478336bbe9b7f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 17:54:10.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17a10d4a57c24eeabaa7d92d7af9e4d2 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dedeb3b33da4e18bdbaf13d4a0972db : SUCCESS in 4m 16s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/7599c24000994e498911f27d6653d448 : SUCCESS in 1h 33m 20s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/e6c4e7c275bb40d0ae7832b145d9f4de : FAILURE in 1h 51m 31s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/6849063b599d480a864e85f185e2bf45 : FAILURE in 1h 37m 16s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/0f97d2574e544ebfaa80f2fdf82c1b39 : TIMED_OUT in 3h 05m 48s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/c095598802624407bb6efcfccf4eb145 : SUCCESS in 1h 30m 05s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/05be5ab7818b47cfb5a58261768a76c5 : SUCCESS in 1h 28m 07s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/0cf107d247af4d3ca2fc305de0718028 : SUCCESS in 1h 41m 32s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/755c5f6cd0524bbd843968573c169e48 : FAILURE in 1h 39m 39s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/7e6cce367ba14b1ca028bc5088810b0e : SUCCESS in 1h 30m 46s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"6c33b0410587ee2701ea5bfacd70eef948ca6bb9","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-11-18 08:58:58.000000000","message":"Patch Set 8: Code-Review-1\n\nSeems it is going in the right direction... but failing some tests related to the services/changes","accounts_in_message":[],"_revision_number":8},{"id":"c983c124bba343b7afdec9ea9018bc1526d15ecd","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-18 12:23:13.000000000","message":"Patch Set 8:\n\ncheck experimental","accounts_in_message":[],"_revision_number":8},{"id":"aea9f808dce2bd6133a6901698d4d93b8d3b3128","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 15:34:47.000000000","message":"Patch Set 8:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/e47a3b0c90024444a25ee1f1bca1b912 : FAILURE in 2h 20m 07s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/461efbae93d141e0a0cbe219a6d607bf : FAILURE in 2h 56m 01s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/2f9b9b3b093c4427940492a4fce8316a : FAILURE in 45m 36s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/80553d46f1b541ac934aab2072673576 : FAILURE in 2h 03m 48s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/ae5d5826db584bea883302dae5918b76 : TIMED_OUT in 3h 07m 22s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/a42fd681487d4114a9eb318379e18033 : FAILURE in 1h 45m 52s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/c156e55536e5467888a2811dd1beb09f : FAILURE in 1h 56m 21s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/0c93c257cf1746229ac95bfbb3a94005 : FAILURE in 1h 58m 12s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/075a8bf02eb44b8ab8e61113e6fd14ff : TIMED_OUT in 2h 16m 14s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/ee737eae21a746afbf9afcee99990762 : FAILURE in 26m 36s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"95f7e3eca25bfc7af435a9dca35b503f75016ad4","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-19 01:12:51.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"eaae5f5ea52e9b31de6cd8c7f332620a547249c6","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-19 01:13:32.000000000","message":"Patch Set 9:\n\ncheck experimental","accounts_in_message":[],"_revision_number":9},{"id":"75b0786c4c7f6166a1f9673f3e96426b3f7076f8","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 02:05:40.000000000","message":"Patch Set 9:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/f311587e68664bdda4917ec3ef40c37e : FAILURE in 15m 16s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/7ecd82b2b0c348e4930e2a79512876fd : FAILURE in 12m 52s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/24cb35c23f99462c9906b641c7269f01 : FAILURE in 14m 49s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/2c2d8e0c60154c3dbd8f96fd934dcf5c : FAILURE in 14m 42s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/fc3231809e9e4857b6052d7a6fb9e7cf : FAILURE in 12m 56s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/eec78357b5244192beb1167b8627b97e : FAILURE in 13m 44s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/cbd57d9c49c344048ad62b2c8ad02f15 : FAILURE in 12m 56s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/4b50266e0f044da5a3eb82ee49fac677 : FAILURE in 16m 18s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/5f3eb7696d6e4575a86a93df56733d7c : FAILURE in 27m 54s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/9a9faff08d2c4286b1ffb8b89153ff18 : FAILURE in 18m 09s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"aefb03ba20d317a898ace8e2c5480e7d04372248","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 02:32:56.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da77236cfe384f31bcd970aff4a63da8 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a26307c490a84c33821ec7a919344b79 : SUCCESS in 3m 36s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/21fda89480a0498785f506364a4bc528 : FAILURE in 18m 32s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/b733c1cedfd246f286933f995a722bc0 : FAILURE in 15m 49s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/6d634f676b714defac9eddffe11e0dc3 : FAILURE in 15m 24s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/cf8cb6a3d6174d9b8046be7fab88507a : FAILURE in 15m 38s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/0431c416afff422a8a361563e998438e : FAILURE in 15m 05s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/04fc468b11c3456a9f38e3043a933bfb : SUCCESS in 1h 18m 07s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/5cae17c163514191a2ce2d9cd8c26f93 : SUCCESS in 1h 13m 01s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/9b1aac544a6b4760aed5a493ba3daa5e : FAILURE in 21m 48s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/7f57215405064831ae1acc23e7377b68 : FAILURE in 24m 42s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"0e3a11eba00531a9e041f8ecfe4412a272301374","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-19 06:56:55.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"a6c9d0b6778f3bc341ec52b8ac951c382ab90064","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-19 06:57:02.000000000","message":"Patch Set 9:\n\ncheck experimental","accounts_in_message":[],"_revision_number":9},{"id":"50ecb794627e2fc598578b02fbe77a13dc11ebd6","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 07:32:04.000000000","message":"Patch Set 9:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/89000e6c6a7a49a6b071b5e2aa154000 : FAILURE in 26m 12s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/99ee3aaaf4b248919719df0fab1f1766 : FAILURE in 25m 48s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/6d2fcbc7d4d64a278385ce4943c0ded6 : FAILURE in 14m 41s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/41f467db28804e698e909583ec5da2f2 : FAILURE in 20m 14s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/4e5a6d670cd14d0f93f421697c3f5b1b : FAILURE in 19m 09s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/2ecbc2d3b6004c45818cdb32ce968608 : FAILURE in 19m 17s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/06118b9de42d490fa7edf983af51070c : FAILURE in 19m 46s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/e46d1bd26d044845a5711c593eeb4005 : FAILURE in 22m 03s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/4490a8b03419419885baf952681cfb46 : FAILURE in 30m 25s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/f37ed2d7e691402682d51b17ac18b305 : FAILURE in 18m 59s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"175ec89ecd681cbb109234ced0e2acfb68a2276a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 08:33:02.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a94fd4162204d779504dab1eca608a7 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3d4d86eebf749d8a4ca4d06d4a9b56a : SUCCESS in 3m 11s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/8b5eb973f89f48eeab3dde4600106186 : FAILURE in 20m 35s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/349a5696586c441697c22830320986f8 : FAILURE in 20m 58s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/b46fb647f55c4480a1616d8272e851db : FAILURE in 18m 18s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/42098a4877cf4fc483872e23bbb603ad : FAILURE in 17m 31s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/3a458e6fae0a48958b3f0b13a484932f : FAILURE in 22m 11s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/bcb36014f8c64960a535486bd175550a : SUCCESS in 1h 23m 56s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/cc60293a3065467e820b6d89011633e0 : SUCCESS in 1h 22m 31s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/bfc22f3923864903baa313c935d905d0 : FAILURE in 23m 53s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/1d73638d21da4636b9569ffaea14adc1 : FAILURE in 21m 01s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"d3996a0572d671c6f1fc084c74ea258a337e9aea","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-19 16:41:37.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"02c449c6c9f83213298936c6dacdf2b950bc4894","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-19 16:41:45.000000000","message":"Patch Set 9:\n\ncheck experimental","accounts_in_message":[],"_revision_number":9},{"id":"c68dd076bedf88e18d9b011cb859f20b9d2ccb29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 19:50:11.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6869d716e63440fd8a3af9c10368e672 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10d3bdc503f34f9cad2f6f1845d8baff : SUCCESS in 4m 29s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/d376755a8b6a4415b0da639e88245502 : SUCCESS in 1h 25m 12s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/f480e123c3c940eba8c66f9c9cede77c : FAILURE in 1h 45m 38s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/3abd79b52a1d4b9eb2b281aef1f9f699 : SUCCESS in 1h 24m 06s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/19916d1d653b4395b3fe32b7d01ff088 : TIMED_OUT in 3h 04m 12s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/58b1a1f68fc2425683903c29bb2cdd8c : FAILURE in 2h 00m 30s (non-voting)\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/275730748f4e4e3991862b2757466c16 : SUCCESS in 1h 58m 56s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"ce886053a5a5af4ad4c8dfee96a5a8d7b01cdc76","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 19:52:39.000000000","message":"Patch Set 9:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/8df287008ceb43c8bc13db6b3d464a20 : FAILURE in 1h 50m 40s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/149cc9ccbf8d424b96b6b95cf0680071 : FAILURE in 2h 52m 06s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/7409a2760fdb4071aff4c9923d224dd4 : FAILURE in 1h 49m 10s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/2541b20511cc4658b07d824bb4b563b0 : FAILURE in 1h 49m 37s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/81fe13d5db954c468c812fb54cbd9b17 : TIMED_OUT in 3h 03m 32s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/314121d6a8584c719fc6bfb676afb7a8 : SUCCESS in 1h 32m 06s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/b3416ec1e5a841e88804c8ea09745877 : FAILURE in 1h 48m 54s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/7aa1235f61774c8388a5b940d2e19003 : FAILURE in 1h 34m 43s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/4d8efa80e55241edbc2c96c22fc12239 : TIMED_OUT in 2h 16m 03s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/36e701a796ae4505b1391e7595cf6bd6 : FAILURE in 28m 01s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"d8f14e3ac01b33ba99459b36e73c3298c3306ee2","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-20 09:04:39.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"01ea95c433fc74841fd900570fe49832175bb59d","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-20 09:05:05.000000000","message":"Patch Set 10:\n\ncheck experimental","accounts_in_message":[],"_revision_number":10},{"id":"ef5639b99a78cc0c7ae25c3e80f54c2e3b60f70f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 12:13:24.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/482be403e8f64df1858f4685fc36e372 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cf835dc0bc3420ea18db836ab955c93 : SUCCESS in 3m 05s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/2eed6d0036df4f6583e5cc0571f284d4 : SUCCESS in 1h 19m 49s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/fa2851b0b53348a4a19d2c7fccdd030d : FAILURE in 1h 35m 27s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/e01c1b01b34a46eebd0370a7c44a08fe : SUCCESS in 1h 49m 16s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/aa721bb72a584f5fb04fb65d45536fd8 : TIMED_OUT in 3h 05m 58s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/ce81f1d7544742a6b2b2ebf7f2d22724 : SUCCESS in 1h 34m 58s (non-voting)\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/987768222ecd4b22b41094d21f4f06f5 : SUCCESS in 2h 06m 57s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"8ce996bfb831e2d2c941f0df9c938af4a469d4b9","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 12:16:02.000000000","message":"Patch Set 10:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/fbb24084c93846c18c1299429b8ce103 : FAILURE in 2h 32m 30s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/317173610fff4b178b0a4e984912bc1e : FAILURE in 2h 46m 24s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/3650581a202045b2a9d2965fe36b0472 : FAILURE in 1h 56m 22s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/75e27757d49448edbbb5694fd94ec8b1 : FAILURE in 1h 53m 32s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/636762b7ef394c8986cc345066e61cd4 : TIMED_OUT in 3h 04m 58s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/ca76c18bf7e642d4a723faf50170030d : SUCCESS in 1h 20m 29s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/c3c72d7e4d4f4ee2aac74052a757f148 : FAILURE in 2h 21m 54s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/5af10930b4cc45a8a6022c1c422c6efc : FAILURE in 1h 47m 54s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/f1ad4e3dc4314c1db149db633541a6e0 : TIMED_OUT in 2h 16m 44s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/8ed6199fd68c40259ed61f2a359c98c1 : FAILURE in 28m 47s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"9c4cc6d012ed485b42dd7578b13c7c0c19c0b705","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-20 13:30:21.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"2dfbb133e65411f56bf1ad73807786e227814c5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 16:42:35.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/299f890838d54610849b38f60ec46009 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c99c0aa5f84848ecac32701992e18f9a : SUCCESS in 2m 53s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/56c28ecf52a54bd589d44dddbec1a806 : SUCCESS in 1h 58m 25s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/988cda8d184d475893091b7aa448263a : FAILURE in 1h 54m 31s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/96c3462e760a4bdc8294df7c1e673553 : FAILURE in 2h 13m 36s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/682bbcdd00e24113bf9ac1a40ac78f2f : TIMED_OUT in 3h 04m 18s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/42c781dbea614ca9b6436e1b9b8be721 : FAILURE in 2h 12m 07s (non-voting)\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/7591cb85157843dbabc2b2decd177c37 : SUCCESS in 1h 51m 58s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"c8d5448ca120174616f326837ed5709e28ea80ec","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-21 04:39:56.000000000","message":"Patch Set 11: Workflow-1","accounts_in_message":[],"_revision_number":11},{"id":"6558afee9e80c637d5f990f60c432aaf0b7521e2","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-21 07:20:50.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"4606b7fa738843be40bf68605365dbcb59b9d502","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 10:33:14.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bac632da98054c8fbed514b8a08553ae : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0491af32612d420fa989cd551cc752d1 : SUCCESS in 2m 58s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/95622bb86a9f47f7a15fe1d38849a2bf : SUCCESS in 1h 34m 50s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/bde6049ab82f447ba29ebcf39793bf55 : FAILURE in 1h 40m 18s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/3db07ca69e32417da9c4a41692663180 : SUCCESS in 1h 53m 52s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/e561abf41c584f3fae37f34ef671fcd0 : TIMED_OUT in 3h 05m 37s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/2af4af9089984812bea7daf202681a18 : SUCCESS in 1h 38m 55s (non-voting)\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/32f9bb09dd9c4680b3dbf562503d3002 : FAILURE in 1h 49m 11s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"11c62db702a294b733b87f26c9a1472af5e36b79","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-11-21 11:46:33.000000000","message":"Patch Set 12: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"8336c9c8cf3c1696f51a57e1293a5a2fa5085526","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-11-21 11:47:17.000000000","message":"Patch Set 12:\n\ncheck experimental","accounts_in_message":[],"_revision_number":12},{"id":"68a15db7936be041464203ca8c84a5a83b8d4e68","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-21 13:46:16.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"d9bc9aafce7059621b00ad18d21fe230cce726de","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 14:57:46.000000000","message":"Patch Set 12:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/f74443fdc67c4545938a9d15a97a2270 : FAILURE in 2h 16m 04s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/6452c7b87ede4879b29257ce9a16ebfe : FAILURE in 2h 13m 56s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/156dbf9fbd9b4c7bbccb04c96e49f42f : TIMED_OUT in 3h 04m 05s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/42c792742d004b68ab0941114168b524 : SUCCESS in 1h 45m 21s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/886623a3f0ef447a8e568d6b6b92fe20 : TIMED_OUT in 3h 04m 54s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/47f73c9e4b19441f9cdf764d0f68ad7f : SUCCESS in 1h 14m 35s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/ed326a5403e14bf6931253b29bfb9d23 : FAILURE in 2h 12m 04s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/8c66e260a94e40279e53bed51ca9ef9f : FAILURE in 1h 48m 34s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/868ffd14b93b4141b5a405e3711bfadb : TIMED_OUT in 2h 17m 40s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/2f62c1e4aa5d475886d9eef191031758 : FAILURE in 30m 15s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"09f5d3a40a8db270a133033d7cdcc662089d4379","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-11-22 08:56:33.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"293e56520c9cbbc676e5fc1b365bc3f571a003e3","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-22 11:04:12.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"072d3016f02518d22856538e9fe820bb374ab696","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-22 11:07:49.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"00f907b36407a976262a54ff7bc3b224c364355e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 14:17:01.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c5b76e635b1444eb5524748c6e2510a : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73ce5739570a4b0ea502b27c1de98ff7 : SUCCESS in 2m 40s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/fdd7ceca0acd41d98512a5224cfacaaf : SUCCESS in 1h 17m 10s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/2d5747f3db564a90851321a03f0e9efd : TIMED_OUT in 3h 02m 54s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/d4702b99caf04c648ef3c68934496a83 : SUCCESS in 1h 18m 21s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/d1e4523c9fdc47258522fb72b745f55b : TIMED_OUT in 3h 03m 37s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/f7b229a116b64af497954c061543bfb1 : SUCCESS in 1h 37m 47s (non-voting)\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/60fbc8e0986f42239e0a01de14427e7b : SUCCESS in 1h 53m 34s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"27f6890fbb1e50d572225e35a5e49199b510504c","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-24 07:05:30.000000000","message":"Patch Set 13:\n\ncheck experimental","accounts_in_message":[],"_revision_number":13},{"id":"f6dceb3fa87ffdad30df20aa7f7c2e7653abe2af","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-24 10:13:57.000000000","message":"Patch Set 13:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/ca430eb6f0de4252a543fa64d0bce48c : TIMED_OUT in 3h 03m 38s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/ffdff1a8f68a4f88a5b403253321ca23 : FAILURE in 2h 16m 18s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/a582d651dcae4ec68bde05517866a910 : FAILURE in 2h 11m 59s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/00ede2c030b444feb63763add025f4d1 : SUCCESS in 1h 42m 26s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/5dc3ce1d126f4380b1d626b745d1bb4a : TIMED_OUT in 3h 03m 14s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/cd085f61560945548a6dbba96e5742c8 : SUCCESS in 1h 22m 59s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/47e289f8d4d04bc9b3793d94a50f1c25 : TIMED_OUT in 3h 04m 11s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/c4d39e3e735f4c24bb504d38fbc070a6 : TIMED_OUT in 3h 03m 56s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/ab7b275791174d899dbe65e858efb2d1 : TIMED_OUT in 2h 12m 58s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/7efaf09686f34fb7982a5cb77be10f3b : FAILURE in 30m 50s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"130537dd1697bee1963ba15a206ce20db4639363","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-11-25 09:20:34.000000000","message":"Patch Set 13: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"a6f7cf2e0b49b3c59a6eef80c7ec45eb7a6681a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 12:29:18.000000000","message":"Patch Set 13:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13c472dbc64d4c20bb7144da347411fa : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ae38ed188454fab8e9c6ce78e9fc6ed : SUCCESS in 4m 10s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/34dc8c6590644b48981895ad3f67070d : SUCCESS in 1h 11m 04s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/3468b461c6ff4746ad8a55d70a75cdb2 : TIMED_OUT in 3h 03m 34s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/8998a7434ac640d48d52b38f3836351a : SUCCESS in 1h 35m 16s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/418d2b06e46d463c84108d34ec8e5de5 : TIMED_OUT in 3h 04m 16s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/0dde19d1eb864c969e6eb67797a316a7 : SUCCESS in 1h 41m 44s (non-voting)\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/ef5e78b8496d4cb3bc82f0e09f9123d9 : SUCCESS in 1h 49m 44s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"edba170275a177b51d3995d9bc1af2eabb673cd8","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-25 13:05:24.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"7a43732ce602b8a6a590fa31d6026a866e62fec0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 16:12:24.000000000","message":"Patch Set 13:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44625e7e0be14a02b023ad6950712bb5 : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/723d0cd351de4ca3aa3ba1dcee41985e : SUCCESS in 3m 05s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/ab27f35baa394a4c9713281530b2707e : SUCCESS in 1h 05m 03s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/86288be8633348dc9953bbf1f0d898f9 : TIMED_OUT in 3h 03m 30s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/551ac89592814267b13b08ec08acfd5f : SUCCESS in 1h 44m 26s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/c5bd71fe71ee420ba6f2d03ae979e9cb : TIMED_OUT in 3h 05m 21s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/d924dd0eb2e04aeaa634c4ea0e861863 : SUCCESS in 1h 40m 35s (non-voting)\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/872e5c4441e949528a19a177ac8d1d67 : SUCCESS in 1h 38m 52s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"85096e75eeda4771d21af634706a5cb72c484fc2","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-27 14:27:35.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"c2bd5fbdf824b9e44262c4fd799f573795f5f5dd","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-27 14:27:58.000000000","message":"Patch Set 14:\n\ncheck experimental","accounts_in_message":[],"_revision_number":14},{"id":"745a20e3510922fecde9f83846c66bfa08545594","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 17:39:35.000000000","message":"Patch Set 14:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/97a1453c6c804f62ba78f844f94667bb : FAILURE in 2h 27m 57s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/99f46496e0834831beba7efc3a624ac6 : FAILURE in 1h 59m 29s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/60d1d93e97d44bda814686823c16ee8e : FAILURE in 2h 04m 18s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/203a5be28dbd40029be68964fdf495ce : FAILURE in 1h 40m 14s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/7bee76f38e5644c595d5c73f6ef54af1 : TIMED_OUT in 3h 04m 26s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/75093e2fba354dfa9809c0414d46a280 : SUCCESS in 1h 28m 05s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/2e74c287dddc45a8aff0ff558d2eeac1 : FAILURE in 1h 50m 06s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/470ecd3804714e02a1bd153a03af475a : FAILURE in 1h 53m 06s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/c9c0440b1d044c829b424f1090e5d443 : TIMED_OUT in 2h 15m 28s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/077569b613834c5184350ef1e0061a37 : FAILURE in 26m 30s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"f52791aa11ae60d0b74e510099244236ad99f9e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 18:08:28.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3328b05086de46e3a96556621cf15893 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29aecabe6e2b4d1dad0e6704f9c79322 : SUCCESS in 3m 33s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/86803968945240999f73226c1259a2da : FAILURE in 1h 53m 37s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/0989e9c4d7fc44c884663b26a90749c7 : FAILURE in 2h 41m 41s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/d5d7c3f8b1ff49e8a8bac578b4db2c06 : FAILURE in 43m 21s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/9756675aac6748289a948458c058a357 : TIMED_OUT in 3h 37m 23s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/c6cc430ab0e04be5af22e5b1876aed77 : SUCCESS in 1h 41m 23s (non-voting)\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/70e1847f37d54d07a4d3ec7345be4cbf : SUCCESS in 2h 09m 39s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"f8d9862a2e558b90bb547c4eebf1c34de00094ef","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-11-27 20:36:36.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"68acf7d7b9b1d714a70231a61a121d14d710742a","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-28 03:43:46.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"551596e65a3a5a8d1bd372e2a20a6f2cd75f0953","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 06:48:58.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4edb0321a5f496a8fe54596ba67343a : SUCCESS in 2m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3aa85f55b3dc4b51ac7bddb5e511da6f : SUCCESS in 2m 56s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/68dcf56ca07e40d0a6bbd5897cc9a85a : SUCCESS in 1h 28m 16s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/dfc31c4b35ec4066bf0add2515038ece : SUCCESS in 1h 37m 49s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/ecaf8f24ff8a413dad372e248500dc02 : SUCCESS in 1h 32m 53s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/e94af58ae3f7430d9877bc6d49ebe4b7 : TIMED_OUT in 3h 04m 04s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/351dd01a9dc94ae5af7e8707f8cebccf : SUCCESS in 1h 22m 29s (non-voting)\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/edd5be15034f4afbb40662ea191eee80 : SUCCESS in 1h 47m 46s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"57385809fe03b6c2a93672e94a09ee058a94aa9c","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-11-28 08:26:09.000000000","message":"Patch Set 14:\n\ncheck experimental","accounts_in_message":[],"_revision_number":14},{"id":"eda68ec5e75c0701286920506d933d141531cd63","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 11:33:05.000000000","message":"Patch Set 14:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/61b3c09980204e0188a995198bdf74a1 : FAILURE in 2h 01m 43s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/0fbb3226bb7a4a13aed6ceacdb182b03 : FAILURE in 2h 09m 52s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/5b49016b318649adb7e508c3e67ea199 : FAILURE in 1h 09m 38s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/5c665b3cc34e42c3ba15f4789b9f3521 : SUCCESS in 1h 50m 10s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/2f3e4c1b94c3467b904fe015e21c6cb8 : TIMED_OUT in 3h 06m 13s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/47bd1a40010e4baea7f045b70a70f94d : SUCCESS in 1h 33m 29s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/4052988d30594d8b8a9e0ec56fb45900 : FAILURE in 1h 50m 57s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/8edd6a4ba0b2443b96a2fa405d5c0f64 : SUCCESS in 1h 57m 33s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/4f39c32eb8444200ad4f4645f5868a6c : TIMED_OUT in 2h 13m 18s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/105796d4406a433cb421daf72d731727 : FAILURE in 33m 50s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"a711dafc3055329a0e430ed51e083e6c60242953","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-11-29 08:20:42.000000000","message":"Patch Set 14: Code-Review+2\n\nSeems it is working. Only failed in gate kuryr-kubernetes-tempest-containerized-openshift due to timeout, and in gate kuryr-kubernetes-tempest-containerized-l2, but this last one is broken anyway","accounts_in_message":[],"_revision_number":14},{"id":"e3aa3cb4dcdba5418fb715ae46e699f2e1c3b261","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-12-02 09:22:45.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":14},{"id":"73e566766bbb999dbd039b2467f35c6e3d97e115","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 09:22:57.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"7cdc914b8776065676ed9ccadcfe0819c039a405","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 11:19:04.000000000","message":"Patch Set 14: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cda11d324774748a27dc1f3f8a711e3 : FAILURE in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5987a57819844ba8ccdad99f07b56ea : FAILURE in 4m 03s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/8e91442763144ca08d18bbcb84d9f10d : SUCCESS in 1h 35m 47s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/01f98d2de30a476d8d3235d35f1eab61 : SUCCESS in 1h 19m 46s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"806f9b8a46c2b113c60bd0c10564f8fddd4bc0de","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-02 11:33:11.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"d56fc627d75468b31124a49b21ab2799a299ccee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 14:39:03.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb7e4056e4f44e068c1584b40e7132d6 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24455aabc2924f1b976fa83dacc24196 : SUCCESS in 3m 39s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/e3c22aa87ce54c6ca4cf8b3c889d59cb : SUCCESS in 1h 34m 06s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/50689cf790e14b118982cd4f1adc7196 : FAILURE in 2h 40m 35s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/aa73735255324349973c522e7bcb4996 : SUCCESS in 1h 28m 01s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/8ffe9039291645699fd96392afa13826 : TIMED_OUT in 3h 04m 25s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/f7c91d65e65e4799ac3a30d4f2bd5ecd : SUCCESS in 1h 30m 42s (non-voting)\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/91fb10114c354cf8b5dfb861b3ee7463 : SUCCESS in 2h 04m 39s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"e4b8c9c30a9acd54ca1888b432b5e3e60784ecdf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 14:39:15.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"6a52bbcc5ff9616db78a627332089e22f31a7c4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 16:27:14.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4d8a7ae22f14c019274605ed0b54056 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8efdf6b9d964e8a95842521d7395b4f : SUCCESS in 2m 59s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/280d97d79a344b2887d70a0df228e0a0 : SUCCESS in 1h 33m 12s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/3ada676049504999856427734c5579a8 : SUCCESS in 1h 41m 17s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"b97ae74960036aaaa8f9b446c551a59474582881","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 16:27:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":14},{"id":"aa8ad4eccd9f4f70ae72afc2c9d0a971a190f3d8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 16:28:32.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/157baf085f4f403e82710fbd83fefe57 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"47c0c5478f29c209588fb91944553704d4925bdf","revisions":{"441d3b6cff1570b5181bf5d1293b49809b04403c":{"kind":"REWORK","_number":1,"created":"2019-11-10 11:41:49.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/32/693632/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/32/693632/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/1"}}},"commit":{"parents":[{"commit":"4175a85a1609f093987d24903d7f6d1592ad7c7a","subject":"Testing basic Network policy IPBlock functionality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/4175a85a1609f093987d24903d7f6d1592ad7c7a"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-11-10 11:38:20.000000000","tz":-300},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-11-10 11:41:38.000000000","tz":-300},"subject":"Add a check for ClusterIP IP of LB service","message":"Add a check for ClusterIP IP of LB service\n\nCheck from a pod the connectivity to the internal IP of\nLB services instead of from an external node\n\nChange-Id: I6ce0b16985de6cc7182026975a63f03844d9d9ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/441d3b6cff1570b5181bf5d1293b49809b04403c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/441d3b6cff1570b5181bf5d1293b49809b04403c"}]},"branch":"refs/heads/master"},"dd6d99248fe25e2e27188eb7216d88c5b8a9f63d":{"kind":"REWORK","_number":2,"created":"2019-11-10 14:20:40.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/32/693632/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/32/693632/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/2"}}},"commit":{"parents":[{"commit":"4175a85a1609f093987d24903d7f6d1592ad7c7a","subject":"Testing basic Network policy IPBlock functionality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/4175a85a1609f093987d24903d7f6d1592ad7c7a"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-11-10 11:38:20.000000000","tz":-300},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-11-10 14:19:46.000000000","tz":120},"subject":"Add a check for ClusterIP IP of LB service","message":"Add a check for ClusterIP IP of LB service\n\nCheck from a pod the connectivity to the internal IP of\nLB services instead of from an external node\n\nChange-Id: I6ce0b16985de6cc7182026975a63f03844d9d9ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/dd6d99248fe25e2e27188eb7216d88c5b8a9f63d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/dd6d99248fe25e2e27188eb7216d88c5b8a9f63d"}]},"branch":"refs/heads/master"},"772ced1ad88c7c59ec6a28fbf378938577cb906f":{"kind":"REWORK","_number":3,"created":"2019-11-12 15:19:35.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/32/693632/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/32/693632/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/3"}}},"commit":{"parents":[{"commit":"4175a85a1609f093987d24903d7f6d1592ad7c7a","subject":"Testing basic Network policy IPBlock functionality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/4175a85a1609f093987d24903d7f6d1592ad7c7a"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-11-10 11:38:20.000000000","tz":-300},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-11-12 15:19:33.000000000","tz":0},"subject":"Add a check for ClusterIP IP of LB service","message":"Add a check for ClusterIP IP of LB service\n\nCheck from a pod the connectivity to the internal IP of\nLB services instead of from an external node\n\nChange-Id: I6ce0b16985de6cc7182026975a63f03844d9d9ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/772ced1ad88c7c59ec6a28fbf378938577cb906f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/772ced1ad88c7c59ec6a28fbf378938577cb906f"}]},"branch":"refs/heads/master"},"5c2b72a517c39afae181cf4dc379832a4187239d":{"kind":"REWORK","_number":4,"created":"2019-11-12 21:09:16.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/32/693632/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/32/693632/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/4"}}},"commit":{"parents":[{"commit":"4175a85a1609f093987d24903d7f6d1592ad7c7a","subject":"Testing basic Network policy IPBlock functionality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/4175a85a1609f093987d24903d7f6d1592ad7c7a"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-11-10 11:38:20.000000000","tz":-300},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-11-12 21:09:02.000000000","tz":0},"subject":"Add a check for ClusterIP IP of LB service","message":"Add a check for ClusterIP IP of LB service\n\nCheck from a pod the connectivity to the internal IP of\nLB services instead of from an external node\n\nChange-Id: I6ce0b16985de6cc7182026975a63f03844d9d9ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/5c2b72a517c39afae181cf4dc379832a4187239d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/5c2b72a517c39afae181cf4dc379832a4187239d"}]},"branch":"refs/heads/master"},"441c152b2081ce3adcefc3634efaf0f0750ad741":{"kind":"REWORK","_number":5,"created":"2019-11-12 21:22:20.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/32/693632/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/32/693632/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/5"}}},"commit":{"parents":[{"commit":"4175a85a1609f093987d24903d7f6d1592ad7c7a","subject":"Testing basic Network policy IPBlock functionality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/4175a85a1609f093987d24903d7f6d1592ad7c7a"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-11-10 11:38:20.000000000","tz":-300},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-11-12 21:22:06.000000000","tz":120},"subject":"Add a check for ClusterIP IP of LB service","message":"Add a check for ClusterIP IP of LB service\n\nCheck from a pod the connectivity to the internal IP of\nLB services instead of from an external node\n\nChange-Id: I6ce0b16985de6cc7182026975a63f03844d9d9ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/441c152b2081ce3adcefc3634efaf0f0750ad741"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/441c152b2081ce3adcefc3634efaf0f0750ad741"}]},"branch":"refs/heads/master"},"ba5992b591901cb8edda92947dcb53c712a82f3b":{"kind":"REWORK","_number":6,"created":"2019-11-12 22:21:08.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/32/693632/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/32/693632/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/6"}}},"commit":{"parents":[{"commit":"4175a85a1609f093987d24903d7f6d1592ad7c7a","subject":"Testing basic Network policy IPBlock functionality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/4175a85a1609f093987d24903d7f6d1592ad7c7a"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-11-10 11:38:20.000000000","tz":-300},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-11-12 21:59:55.000000000","tz":120},"subject":"Add a check for ClusterIP IP of LB service","message":"Add a check for ClusterIP IP of LB service\n\nCheck from a pod the connectivity to the internal IP of\nLB services instead of from an external node\n\nChange-Id: I6ce0b16985de6cc7182026975a63f03844d9d9ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ba5992b591901cb8edda92947dcb53c712a82f3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ba5992b591901cb8edda92947dcb53c712a82f3b"}]},"branch":"refs/heads/master"},"48662e4b4859070acb8015d4591a9dcb6a6e7584":{"kind":"REWORK","_number":7,"created":"2019-11-13 09:02:16.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/32/693632/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/32/693632/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/7"}}},"commit":{"parents":[{"commit":"4175a85a1609f093987d24903d7f6d1592ad7c7a","subject":"Testing basic Network policy IPBlock functionality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/4175a85a1609f093987d24903d7f6d1592ad7c7a"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-11-10 11:38:20.000000000","tz":-300},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-11-13 09:01:56.000000000","tz":0},"subject":"Add a check for ClusterIP IP of LB service","message":"Add a check for ClusterIP IP of LB service\n\nCheck from a pod the connectivity to the internal IP of\nLB services instead of from an external node\n\nChange-Id: I6ce0b16985de6cc7182026975a63f03844d9d9ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/48662e4b4859070acb8015d4591a9dcb6a6e7584"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/48662e4b4859070acb8015d4591a9dcb6a6e7584"}]},"branch":"refs/heads/master"},"8a1e28bed3a29f431d89cc016550ea0258340663":{"kind":"REWORK","_number":8,"created":"2019-11-13 09:47:14.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/32/693632/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/32/693632/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/8"}}},"commit":{"parents":[{"commit":"4175a85a1609f093987d24903d7f6d1592ad7c7a","subject":"Testing basic Network policy IPBlock functionality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/4175a85a1609f093987d24903d7f6d1592ad7c7a"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-11-10 11:38:20.000000000","tz":-300},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-11-13 09:47:11.000000000","tz":120},"subject":"Add a check for ClusterIP IP of LB service","message":"Add a check for ClusterIP IP of LB service\n\nCheck from a pod the connectivity to the internal IP of\nLB services instead of from an external node\n\nChange-Id: I6ce0b16985de6cc7182026975a63f03844d9d9ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8a1e28bed3a29f431d89cc016550ea0258340663"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8a1e28bed3a29f431d89cc016550ea0258340663"}]},"branch":"refs/heads/master"},"639dbe671201e325b12eab072e7cce9e92ec4588":{"kind":"REWORK","_number":9,"created":"2019-11-19 01:12:51.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/32/693632/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/32/693632/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/9"}}},"commit":{"parents":[{"commit":"4175a85a1609f093987d24903d7f6d1592ad7c7a","subject":"Testing basic Network policy IPBlock functionality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/4175a85a1609f093987d24903d7f6d1592ad7c7a"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-11-10 11:38:20.000000000","tz":-300},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-11-19 01:12:40.000000000","tz":120},"subject":"Add a check for ClusterIP IP of LB service","message":"Add a check for ClusterIP IP of LB service\n\nCheck from a pod the connectivity to the internal IP of\nLB services instead of from an external node\n\nChange-Id: I6ce0b16985de6cc7182026975a63f03844d9d9ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/639dbe671201e325b12eab072e7cce9e92ec4588"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/639dbe671201e325b12eab072e7cce9e92ec4588"}]},"branch":"refs/heads/master"},"2d617706c548dc90200292ec9bbeee6dae82fa69":{"kind":"REWORK","_number":10,"created":"2019-11-20 09:04:39.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/32/693632/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/32/693632/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/10"}}},"commit":{"parents":[{"commit":"4175a85a1609f093987d24903d7f6d1592ad7c7a","subject":"Testing basic Network policy IPBlock functionality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/4175a85a1609f093987d24903d7f6d1592ad7c7a"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-11-10 11:38:20.000000000","tz":-300},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-11-20 09:04:30.000000000","tz":120},"subject":"Add a check for ClusterIP IP of LB service","message":"Add a check for ClusterIP IP of LB service\n\nCheck from a pod the connectivity to the internal IP of\nLB services instead of from an external node\n\nChange-Id: I6ce0b16985de6cc7182026975a63f03844d9d9ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2d617706c548dc90200292ec9bbeee6dae82fa69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2d617706c548dc90200292ec9bbeee6dae82fa69"}]},"branch":"refs/heads/master"},"d9ecb7fd55b7155672fee679cb261b0a0c03ae31":{"kind":"REWORK","_number":11,"created":"2019-11-20 13:30:21.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/32/693632/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/32/693632/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/11"}}},"commit":{"parents":[{"commit":"4175a85a1609f093987d24903d7f6d1592ad7c7a","subject":"Testing basic Network policy IPBlock functionality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/4175a85a1609f093987d24903d7f6d1592ad7c7a"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-11-10 11:38:20.000000000","tz":-300},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-11-20 13:29:59.000000000","tz":120},"subject":"Add a check for ClusterIP IP of LB service","message":"Add a check for ClusterIP IP of LB service\n\nCheck from a pod the connectivity to the internal IP of\nLB services instead of from an external node\n\nChange-Id: I6ce0b16985de6cc7182026975a63f03844d9d9ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d9ecb7fd55b7155672fee679cb261b0a0c03ae31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d9ecb7fd55b7155672fee679cb261b0a0c03ae31"}]},"branch":"refs/heads/master"},"0906a264a628e57484d2c9e55d20c75e583ce75a":{"kind":"REWORK","_number":12,"created":"2019-11-21 07:20:50.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/32/693632/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/32/693632/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/12"}}},"commit":{"parents":[{"commit":"4175a85a1609f093987d24903d7f6d1592ad7c7a","subject":"Testing basic Network policy IPBlock functionality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/4175a85a1609f093987d24903d7f6d1592ad7c7a"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-11-10 11:38:20.000000000","tz":-300},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-11-21 07:20:34.000000000","tz":120},"subject":"Add a check for ClusterIP IP of LB service","message":"Add a check for ClusterIP IP of LB service\n\nCheck from a pod the connectivity to the internal IP of\nLB services instead of from an external node\n\nChange-Id: I6ce0b16985de6cc7182026975a63f03844d9d9ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/0906a264a628e57484d2c9e55d20c75e583ce75a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/0906a264a628e57484d2c9e55d20c75e583ce75a"}]},"branch":"refs/heads/master"},"44b0f7902ed6810ecdc2925e665bcdedff348df4":{"kind":"REWORK","_number":13,"created":"2019-11-22 11:07:49.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/32/693632/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/32/693632/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/13"}}},"commit":{"parents":[{"commit":"4175a85a1609f093987d24903d7f6d1592ad7c7a","subject":"Testing basic Network policy IPBlock functionality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/4175a85a1609f093987d24903d7f6d1592ad7c7a"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-11-10 11:38:20.000000000","tz":-300},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-11-22 11:03:54.000000000","tz":120},"subject":"Add a check for ClusterIP IP of LB service","message":"Add a check for ClusterIP IP of LB service\n\nCheck from a pod the connectivity to the internal IP of\nLB services instead of from an external node\n\nChange-Id: I6ce0b16985de6cc7182026975a63f03844d9d9ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/44b0f7902ed6810ecdc2925e665bcdedff348df4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/44b0f7902ed6810ecdc2925e665bcdedff348df4"}]},"branch":"refs/heads/master"},"47c0c5478f29c209588fb91944553704d4925bdf":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2019-11-27 14:27:35.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/32/693632/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/32/693632/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/693632/14"}}},"commit":{"parents":[{"commit":"1a7a438a4d6eecee0390d40586b5f6cc4eb35f38","subject":"Merge \"Ensure correct failure message is checked on CRD validation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/1a7a438a4d6eecee0390d40586b5f6cc4eb35f38"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-11-10 11:38:20.000000000","tz":-300},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-11-27 14:27:27.000000000","tz":120},"subject":"Add a check for ClusterIP IP of LB service","message":"Add a check for ClusterIP IP of LB service\n\nCheck from a pod the connectivity to the internal IP of\nLB services instead of from an external node\n\nChange-Id: I6ce0b16985de6cc7182026975a63f03844d9d9ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/47c0c5478f29c209588fb91944553704d4925bdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/47c0c5478f29c209588fb91944553704d4925bdf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
