)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~695679","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~I8a1bfab6e4b6c0e08d8cd2b7e8c6a5efdd87bbd3","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"mark-isolation-test-unstable","hashtags":[],"change_id":"I8a1bfab6e4b6c0e08d8cd2b7e8c6a5efdd87bbd3","subject":"Mark test_namespace_sg_isolation as unstable","status":"MERGED","created":"2019-11-22 13:11:24.000000000","updated":"2019-11-27 15:46:20.000000000","submitted":"2019-11-27 14:33:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695679-1574865233832-ca8dfbed","meta_rev_id":"693ccb8689dc7a4cf7d3ebf8f640bdf1f71edcfe","_number":695679,"virtual_id_number":695679,"owner":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-27 14:33:53.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"all":[{"value":2,"date":"2019-11-27 13:00:43.000000000","_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":1,"date":"2019-11-25 10:21:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-25 09:53:43.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"all":[{"value":1,"date":"2019-11-27 13:00:43.000000000","_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-25 09:53:43.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2019-11-25 10:21:37.000000000","updated_by":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"reviewer":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"state":"REVIEWER"},{"updated":"2019-11-27 13:00:43.000000000","updated_by":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"reviewer":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"state":"REVIEWER"},{"updated":"2019-11-27 14:33:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"729357030162e96556c8f39700920ed17fc2bb55","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-11-22 13:11:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cccef10ae516c826e2eda873eb3524d6934f5a4e","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-11-22 13:11:42.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"b67502364c1e4ee6ee5fad73c4d48b02a3614e53","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-11-22 15:08:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f27f811fc40f00e4eef24a14ee871f5756ba8e1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 18:19:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3795aa4652cc49ca9d0e7aa33f260a02 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3605d1a7f03e4b34918257d8c6f87bd4 : SUCCESS in 5m 43s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/3ddf2bf75bdb491a8301f2edb5f79e26 : SUCCESS in 1h 07m 23s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/8991d95787584b738fd6d2140bf13f14 : TIMED_OUT in 3h 04m 17s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/0d58d684f27549488ac613acbca96cdb : SUCCESS in 1h 24m 21s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/d058e3353b93493fba7400253bef64b8 : TIMED_OUT in 3h 05m 40s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/892a296d807843dbbab511ac31110695 : SUCCESS in 1h 22m 42s (non-voting)\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/13d71c2341084db8b019dcf5ff00f95b : SUCCESS in 1h 40m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d1a3509b9f658da2bed8ab1db5f0334860d5b92b","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-24 07:03:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"22c0ae172129edcb6e8822133422f71009114e41","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-11-25 09:21:13.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f525bac3a494f25de5fc8bf37da3724ad2afd39c","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-11-25 09:23:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b8f21874ecbc176a979cc90de470f248948fa372","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-11-25 09:53:43.000000000","message":"Patch Set 3: Code-Review+2\n\ncheck experimental","accounts_in_message":[],"_revision_number":3},{"id":"00bcc5eda2e8f90c8bbab19207a4e92f1c2e392e","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-25 10:21:37.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"5e8e10321f62adbd8a5980a00b39d52f24b8287e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 12:30:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a13f1333e9124318bcb922e90b9891dd : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b7ee908bbf24a58bffaa2ca68c1e7cf : SUCCESS in 4m 13s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/d98ad6a1edec4bc58e6d57374990e23e : SUCCESS in 1h 00m 09s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/5d17d58017ad44889ac43323c02f25ce : TIMED_OUT in 3h 03m 49s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/f4a55eacbe504811a11575f025ccdfa7 : SUCCESS in 1h 19m 19s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/2a30364176264f8e9afe0686b0a2627c : TIMED_OUT in 3h 04m 39s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/474a9dac95c747d8a6eec540f2ce4f4c : SUCCESS in 1h 21m 54s (non-voting)\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/90e9cc81155044a8adaeacf13edadafb : SUCCESS in 1h 34m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7cf9c8468088d4f3d711bb3417ea6c460b3bedf7","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 13:15:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/8431376ca2d140b998c63a56437d0222 : TIMED_OUT in 3h 06m 05s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/d6d5626e9a514cc6a84e1d93f332ea69 : FAILURE in 1h 47m 00s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/9f20f48edcb44129b6fb4f71a97b8c5b : FAILURE in 2h 15m 28s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/621b26c58b074fbe8897c546f4b9525f : SUCCESS in 1h 40m 42s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/8b9b90036b184019839529d4cb441f7e : TIMED_OUT in 3h 07m 39s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/c0b2af424ce64419ba616347fa31667f : SUCCESS in 1h 25m 53s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/d981b3c8acf744fa9dab6711ed61cd55 : TIMED_OUT in 3h 05m 50s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/d61f9ae0a36a4b6fa215deae812c375b : TIMED_OUT in 3h 05m 33s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/3c95b830226749edabe3e366e86df4d9 : TIMED_OUT in 2h 17m 23s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/86e71a5ad8a341188a341578e8a0d869 : FAILURE in 34m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"892939a733d66eeca5dbccca1e37357deafaa0c4","author":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"date":"2019-11-27 13:00:43.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"250d674e5ea6b049478550d9b47bb16fd77df96e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 13:00:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0e66a3b0a432e92ffc5d0e48a484100de1a56751","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 14:33:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"9cfed785361daa9b29fbcd7469e16c1c7a91da5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 14:33:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ea7594b3f3e42dcb16ddf038315915e : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77cdc12377bf4aa49a5059c3228ecb2d : SUCCESS in 2m 56s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/03644cf5b93045b5986a207fa7e06ec5 : SUCCESS in 1h 06m 20s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/30e11eb4d10a4036b71940d86c40a0e1 : SUCCESS in 1h 27m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"02787efe7639a01b996c213d891bd003cf2d3051","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 14:35:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cbc1b4d9f444c1285e8ebce5f5740d9 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":3},{"id":"962c703d7ea63427612fda74e85a6ed9211f424a","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-11-27 15:46:20.000000000","message":"Patch Set 3:\n\nJust noticed the test name in the commit message is wrong. So just for the record - it\u0027s supposed to be test_namespace_sg_svc_isolation.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8d66ca382aa863a32ec74a01a23fcf9709f48559","revisions":{"8240d3615cc33f2a279d810e331022a3e3db2bd5":{"kind":"REWORK","_number":1,"created":"2019-11-22 13:11:24.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/79/695679/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/79/695679/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/79/695679/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/79/695679/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/79/695679/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/79/695679/1"}}},"commit":{"parents":[{"commit":"1a7a438a4d6eecee0390d40586b5f6cc4eb35f38","subject":"Merge \"Ensure correct failure message is checked on CRD validation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/1a7a438a4d6eecee0390d40586b5f6cc4eb35f38"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2019-11-22 13:10:35.000000000","tz":60},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2019-11-22 13:10:35.000000000","tz":60},"subject":"Mark test_namespace_sg_isolation as unstable","message":"Mark test_namespace_sg_isolation as unstable\n\nWe have issues with this test, most likely due to Neutron bug [1]. This\ncommit marks it as unstable, so it gets skipped on failure.\n\n[1] https://bugzilla.redhat.com/show_bug.cgi?id\u003d1688323\n\nChange-Id: I8a1bfab6e4b6c0e08d8cd2b7e8c6a5efdd87bbd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8240d3615cc33f2a279d810e331022a3e3db2bd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8240d3615cc33f2a279d810e331022a3e3db2bd5"}]},"branch":"refs/heads/master"},"112166b80ad912b2769ff5fc10ab110cee6d772c":{"kind":"REWORK","_number":2,"created":"2019-11-22 15:08:06.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/79/695679/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/79/695679/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/79/695679/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/79/695679/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/79/695679/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/79/695679/2"}}},"commit":{"parents":[{"commit":"1a7a438a4d6eecee0390d40586b5f6cc4eb35f38","subject":"Merge \"Ensure correct failure message is checked on CRD validation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/1a7a438a4d6eecee0390d40586b5f6cc4eb35f38"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2019-11-22 13:10:35.000000000","tz":60},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2019-11-22 15:07:58.000000000","tz":60},"subject":"Mark test_namespace_sg_isolation as unstable","message":"Mark test_namespace_sg_isolation as unstable\n\nWe have issues with this test, most likely due to Neutron bug [1]. This\ncommit marks it as unstable, so it gets skipped on failure.\n\n[1] https://bugzilla.redhat.com/show_bug.cgi?id\u003d1688323\n\nChange-Id: I8a1bfab6e4b6c0e08d8cd2b7e8c6a5efdd87bbd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/112166b80ad912b2769ff5fc10ab110cee6d772c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/112166b80ad912b2769ff5fc10ab110cee6d772c"}]},"branch":"refs/heads/master"},"8d66ca382aa863a32ec74a01a23fcf9709f48559":{"kind":"REWORK","_number":3,"created":"2019-11-25 09:23:25.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/79/695679/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/79/695679/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/79/695679/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/79/695679/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/79/695679/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/79/695679/3"}}},"commit":{"parents":[{"commit":"1a7a438a4d6eecee0390d40586b5f6cc4eb35f38","subject":"Merge \"Ensure correct failure message is checked on CRD validation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/1a7a438a4d6eecee0390d40586b5f6cc4eb35f38"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2019-11-22 13:10:35.000000000","tz":60},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2019-11-25 09:23:16.000000000","tz":60},"subject":"Mark test_namespace_sg_isolation as unstable","message":"Mark test_namespace_sg_isolation as unstable\n\nWe have issues with this test, most likely due to Neutron bug [1]. This\ncommit marks it as unstable, so it gets skipped on failure.\n\n[1] https://bugzilla.redhat.com/show_bug.cgi?id\u003d1688323\n\nChange-Id: I8a1bfab6e4b6c0e08d8cd2b7e8c6a5efdd87bbd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8d66ca382aa863a32ec74a01a23fcf9709f48559"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8d66ca382aa863a32ec74a01a23fcf9709f48559"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
