)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~696274","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~I31a874c5f1d0d134e555a32b09a71e9466daea77","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"namespace-fixes","hashtags":[],"change_id":"I31a874c5f1d0d134e555a32b09a71e9466daea77","subject":"Reduce flakiness of pod creation timeout","status":"MERGED","created":"2019-11-27 10:22:12.000000000","updated":"2019-12-04 12:24:21.000000000","submitted":"2019-12-04 12:22:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696274-1575462166769-dc25f70e","meta_rev_id":"e603e57ce25caec9bad3cc1a7ae7498965077f3b","_number":696274,"virtual_id_number":696274,"owner":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"date":"2019-12-03 13:01:39.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-04 12:22:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2019-12-03 08:41:46.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-28 13:51:04.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":1,"date":"2019-12-03 08:41:46.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-27 13:41:39.000000000","updated_by":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"reviewer":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"state":"REVIEWER"},{"updated":"2019-11-28 08:26:50.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2019-12-03 08:41:46.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2019-12-04 12:22:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f5b2cb090c95a0611e58f79fa82135a6541390bc","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-11-27 10:22:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bcbccb6662a6416401447a3baaca86b0d017e56c","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-11-27 10:23:31.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5a0dabf5158df32bd744cfd59458186ce872372f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 13:31:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94e57e9492034db880fcc7a7b4964784 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e7a5a0d76f3493fb2db06960852b40e : SUCCESS in 3m 19s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/5864188604f74673826aaaba34d030a0 : SUCCESS in 1h 26m 05s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/840ca10be30d45f08f241b204947560c : TIMED_OUT in 3h 04m 44s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/5e775d6e45b646418c665c7ec335637a : SUCCESS in 1h 44m 25s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/5faffd3ee06741cfa432353920138abf : TIMED_OUT in 3h 05m 51s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/1b63c4e04d0c42d2a8d0749a74f01549 : SUCCESS in 1h 41m 44s (non-voting)\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/ef624dc719634e698fee12ef3f83e8be : FAILURE in 2h 13m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"320e6a3b5b0d1b166a3d3b9296ef441d17f5f2df","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-11-27 13:41:39.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ad3df98abcfcb6c4f00fb9b5db492749f0819573","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-11-27 14:08:09.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5c7f1eb0917bac870abfec59c8025061b718cd8d","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-11-27 14:09:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5941c0236f8593480336a626fb31b52c0657e5df","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-11-27 14:17:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"131334e6334cd7c24f273f147cd674bb6905cb0c","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-11-27 14:32:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"476ff5aa709b436f23791eb34671825ad098010e","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-11-27 14:36:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e7e8ded432a6d096ec2f7f6a1dc9c35736d8177a","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-11-27 14:47:05.000000000","message":"Patch Set 2: Code-Review-1\n\nAgree with Luis here, we should try to rely less on checking internal stuff and more on actual results expected from how it\u0027s supposed to work.","accounts_in_message":[],"_revision_number":2},{"id":"ad4ef93369d69c907c193fb7d0e59973c4546c49","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-11-27 14:58:42.000000000","message":"Patch Set 2:\n\n\u003e Agree with Luis here, we should try to rely less on checking\n \u003e internal stuff and more on actual results expected from how it\u0027s\n \u003e supposed to work.\n\nOkay, gonna only increase the timeout then.","accounts_in_message":[],"_revision_number":2},{"id":"e599ef68ffcfbc428043cb2da569210532a81ef7","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-11-27 15:01:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"417c39b682a991c831249fe83f2de0e69fa24c85","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-11-27 15:45:06.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"01dfa297167f602e18586ab9f5c98cac260706b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 18:14:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e5fa82fa6de4244becf4a829b295389 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e73d77144e4a4bb7903139db7b51a6be : SUCCESS in 2m 46s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/f4924923ed9545e5bef9498f2a037b7f : SUCCESS in 1h 23m 44s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/b26a9f8f14464d7890eeb5d78fa45caa : SUCCESS in 1h 24m 32s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/1f53c99f23c441af834b6ba34dd4fea0 : SUCCESS in 1h 18m 34s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/fd82a6c2e7d24deb96fb4cfc04075b78 : TIMED_OUT in 3h 04m 25s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/3565286404104cbdbe9dfd5680f565eb : SUCCESS in 1h 35m 56s (non-voting)\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/ef0191d2df3e442191ddfee34e0a390f : SUCCESS in 1h 30m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d43726945d7cfcda2a9d910be0f4dd68260cc44e","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-11-28 08:26:50.000000000","message":"Patch Set 3: Code-Review+2\n\ncheck experimental","accounts_in_message":[],"_revision_number":3},{"id":"2432d15c04897a6d3c8d4449450a7864f1b52211","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 11:38:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/fc5b5d01857841b6975e18e6e5a5fedf : FAILURE in 1h 59m 59s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/dd43843e0cad42169471ad9813c8e8af : FAILURE in 1h 41m 31s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/6f23b8aa7f8f49558c37b3da9ec7ce82 : SUCCESS in 1h 44m 31s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/daa102d48e5c4bc69f730a27c1cb83e3 : SUCCESS in 1h 12m 50s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/5ccc0ba1e2f44116ac2c5697d1f0c3c3 : TIMED_OUT in 3h 04m 26s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/cde05ebb543b4b959599f46d87f3fabb : SUCCESS in 1h 10m 10s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/2333ca15e0d64b55aa72dfcbcd2d3409 : FAILURE in 1h 47m 45s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/17393e3c596c4934943a4a43010ea05b : SUCCESS in 1h 44m 56s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/d831cc3b05684958bd8dc63a5b7d323a : SUCCESS in 2h 13m 36s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/7683932c08244916a1e76771296609d6 : FAILURE in 28m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5548c23c2c22436d31fadb7a9c1d0d91617b6d52","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-11-28 13:51:04.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"dfa2ec806669dc85ccdd5c3374833989709c21a3","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-11-28 13:52:58.000000000","message":"Patch Set 4:\n\ncheck experimental","accounts_in_message":[],"_revision_number":4},{"id":"b939c7d4ed6c81cc800aeb84c7b03ed7b3d77189","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-11-28 14:02:12.000000000","message":"Patch Set 4:\n\nThe failures seem to be due to lost namespace events on kuryr-controller. What I\u0027m going to prioritize is the 3.11 upgrade of OpenShift - maybe that\u0027s not happening there.","accounts_in_message":[],"_revision_number":4},{"id":"ad93bcfe45322f5af7ec87caae5bc4f2b54cd8d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 17:01:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78bd45ac43384114aa949d2954805686 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d17b097b5ad2477284c301474327c2be : SUCCESS in 3m 28s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/87420990b2174f2693557bec193f5586 : SUCCESS in 1h 10m 38s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/6e79b0f876914a328c79d1c172cf3675 : SUCCESS in 1h 33m 47s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/4ad893493960461fb6a16402bc4a68f5 : SUCCESS in 1h 12m 16s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/26456d8cbde2456790cb32af9f99a180 : TIMED_OUT in 3h 04m 28s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/b6c4763d5bdd443ca93801e3b4ab89f4 : SUCCESS in 1h 22m 13s (non-voting)\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/aa87191c1a4a4b1ebfa65dd88bffebfe : SUCCESS in 2h 12m 34s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"dac8252a046b56055aa521cd1159e3c1f35f71b7","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 17:04:28.000000000","message":"Patch Set 4:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/cff916bcb6f5418dac583315f50c3f99 : FAILURE in 2h 24m 27s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/4c8f4d2007b244838a1aab6a21169e51 : FAILURE in 1h 45m 31s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/924182e8b7524f6f93de90dec2e3584a : SUCCESS in 1h 30m 14s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/79e2cfaaaab14fd68f8e3a91df69541d : FAILURE in 42m 11s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/5822912392994afab392477b0d887a47 : TIMED_OUT in 3h 06m 42s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/a5c6212226704cf4b085ec0ce372a517 : SUCCESS in 1h 22m 08s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/f178a8715d5c4b28945b372d2aeb4b30 : FAILURE in 2h 07m 50s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/0b35ce82307147948b74eb3214b87c41 : SUCCESS in 1h 47m 31s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/4fae06113f694f55a2933aa8a98778d0 : TIMED_OUT in 2h 17m 08s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/bed57ca036654d67982a91eabda014da : FAILURE in 28m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8454d2929a5768df6d97faec1d6fafc1f774194d","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-12-03 08:41:46.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"cc5f89b396e107f3972bb604965edcfbb19b4b65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 08:41:57.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6cbd85e814e1c2860016f21c429914e72ecb8756","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 10:49:29.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3f76c440d7742e99a10cf9f97de31a9 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/801721c609bd41149e09a976b583abbe : SUCCESS in 4m 27s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/a0d28384c708448282e296d0d4220a12 : FAILURE in 1h 06m 33s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/bc5447ef0197462d852051743019fbfb : FAILURE in 1h 12m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"038c4709f598b2a9896cddf6182c43d97cc76a93","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-12-03 10:54:43.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"d72db66e95c26ea042f4e6fd95329b99f3800e02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 12:06:48.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c1ef2b06e2b4db893546723575ac14c : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b8a7cec06e54d17ad527dec7e47f2fe : SUCCESS in 2m 37s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/524e34982ce14415864bcf716926ff0b : FAILURE in 55m 18s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/bd23edfb43e04ad184db03a25f905e37 : FAILURE in 55m 21s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/70a5ef10acf64c919285346d0740968e : FAILURE in 57m 40s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/9463e33ee7e74ed0b5fbe1c20db9a281 : FAILURE in 1h 11m 20s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/42a560d3eae44be3b3a95b2e5ecd2f45 : FAILURE in 27m 28s (non-voting)\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/a0f131edf5d8467fbbf4722c0eaf8789 : FAILURE in 44m 34s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a2739af269ff4554bf1ff1ebef5f790a688562c0","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-12-03 13:01:39.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"f7578bbb61d2b9e387ed894d4fea35bafcb17b45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 14:30:55.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4c2d211ec194073b5c1e13780c77941 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af1b0f739d26441c83c00534445caa57 : SUCCESS in 4m 21s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/a5fa988f8d7e4e4690c1ddaa06ef6217 : FAILURE in 57m 04s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/8b45e02a73c748d095fc7eacbfe127a8 : FAILURE in 56m 29s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/6b856faa03cf4cb39b76edb05f5ff3f2 : FAILURE in 1h 04m 13s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/fbd5f446964a475787f45d011e947695 : FAILURE in 1h 15m 27s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/c5bfd07a32a14806ae6b5f6d6e134760 : FAILURE in 34m 35s (non-voting)\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/434237f10fcb497ebf0eed3ebfeea1cb : FAILURE in 1h 23m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9d24cf5bf30138e719eb07d02de9d552ab8f7f1e","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-12-04 07:59:57.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"3ee1af828b80a2b6e96254439e11667703329b4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 11:10:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a6bb8c7a52640bf8584c2a55454e25c : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e3d9c96d96a413283464a5271d0f18f : SUCCESS in 3m 58s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/e42d231ff3df4345924e5438656157f0 : SUCCESS in 1h 16m 19s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/e70aaaa94b5f4cd99bc7df670feca52f : SUCCESS in 1h 25m 49s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/f8f408bd276f4a248d2e5cd3b7ee4ad8 : SUCCESS in 1h 33m 21s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/e16d07c2dbab4e2293208c5b5bdd556c : TIMED_OUT in 3h 04m 16s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/702e635018c64300826f80308770334a : FAILURE in 28m 42s (non-voting)\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/3215ec84945e440ca1228afbc76c98ac : SUCCESS in 1h 46m 30s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5650e8cd4a024d4bf19d3dba34922e9c53bc03ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 11:10:39.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1a1bfb9a980baa685df4037b24d18e2f79212092","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 12:22:46.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f48df2b04f54a3395af3691128c1a3d : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0e9b927789749b2bdc3f78a5cc814b6 : SUCCESS in 4m 28s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/e20c735ee38f479b949bb4f7632afc5f : SUCCESS in 1h 08m 30s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/8e49880d12024cfd84fddd374dbd2fd2 : SUCCESS in 1h 11m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e920a9a0ccc5342a091de90eeffee7a3d8ab8a33","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 12:22:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"0b98131d7ccbafff9c964783acdf884b68ee6c6f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 12:24:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55c30f110f08456b827c36e951ea1025 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f34269cd224c77621478baca06c94c50fe8bacd0","revisions":{"1285da8528f4fbc716728a5de51c2fe1b744f1fe":{"kind":"REWORK","_number":1,"created":"2019-11-27 10:22:12.000000000","uploader":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"ref":"refs/changes/74/696274/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/74/696274/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/74/696274/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/74/696274/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/74/696274/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/74/696274/1"}}},"commit":{"parents":[{"commit":"1a7a438a4d6eecee0390d40586b5f6cc4eb35f38","subject":"Merge \"Ensure correct failure message is checked on CRD validation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/1a7a438a4d6eecee0390d40586b5f6cc4eb35f38"}]}],"author":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2019-11-27 10:12:39.000000000","tz":60},"committer":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2019-11-27 10:12:39.000000000","tz":60},"subject":"Reduce flakiness of pod creation timeout","message":"Reduce flakiness of pod creation timeout\n\nAs this test recreates the namespace and a pod on it,\nit could take a bit longer for the older KuryrNet CRD\nto be removed and recreated, plus the pool to be populated\ncausing the pod creation to timeout. This commit fixes the\nissue by ensuring the namespace is properly wired by Kuryr,\nto then try the Pod creation, reducing the timeout chances.\n\nChange-Id: I31a874c5f1d0d134e555a32b09a71e9466daea77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/1285da8528f4fbc716728a5de51c2fe1b744f1fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/1285da8528f4fbc716728a5de51c2fe1b744f1fe"}]},"branch":"refs/heads/master"},"f5440329bed9a0ce40bf8df6d996da6b4bd4e2fd":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-11-27 10:23:31.000000000","uploader":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"ref":"refs/changes/74/696274/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/74/696274/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/74/696274/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/74/696274/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/74/696274/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/74/696274/2"}}},"commit":{"parents":[{"commit":"1a7a438a4d6eecee0390d40586b5f6cc4eb35f38","subject":"Merge \"Ensure correct failure message is checked on CRD validation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/1a7a438a4d6eecee0390d40586b5f6cc4eb35f38"}]}],"author":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2019-11-27 10:12:39.000000000","tz":60},"committer":{"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","date":"2019-11-27 10:23:29.000000000","tz":0},"subject":"Reduce flakiness of pod creation timeout","message":"Reduce flakiness of pod creation timeout\n\nAs this test recreates the namespace and a pod on it,\nit could take a bit longer for the older KuryrNet CRD\nto be removed and recreated, plus the pool to be populated\ncausing the pod creation to timeout. This commit reduces the\nchances of timeout by ensuring the namespace is properly wired\nby Kuryr, to then try the Pod creation.\n\nChange-Id: I31a874c5f1d0d134e555a32b09a71e9466daea77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/f5440329bed9a0ce40bf8df6d996da6b4bd4e2fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/f5440329bed9a0ce40bf8df6d996da6b4bd4e2fd"}]},"branch":"refs/heads/master"},"610d1a40a2d1f3ab0974f933e011a1da3c710c1e":{"kind":"REWORK","_number":3,"created":"2019-11-27 15:01:19.000000000","uploader":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"ref":"refs/changes/74/696274/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/74/696274/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/74/696274/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/74/696274/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/74/696274/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/74/696274/3"}}},"commit":{"parents":[{"commit":"1a7a438a4d6eecee0390d40586b5f6cc4eb35f38","subject":"Merge \"Ensure correct failure message is checked on CRD validation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/1a7a438a4d6eecee0390d40586b5f6cc4eb35f38"}]}],"author":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2019-11-27 10:12:39.000000000","tz":60},"committer":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2019-11-27 15:00:02.000000000","tz":60},"subject":"Reduce flakiness of pod creation timeout","message":"Reduce flakiness of pod creation timeout\n\nAs this test recreates the namespace and a pod on it,\nit could take a bit longer for the older KuryrNet CRD\nto be removed and recreated, plus the pool to be populated\ncausing the pod creation to timeout. This commit reduce the\nflakiness by increasing the timeout.\n\nChange-Id: I31a874c5f1d0d134e555a32b09a71e9466daea77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/610d1a40a2d1f3ab0974f933e011a1da3c710c1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/610d1a40a2d1f3ab0974f933e011a1da3c710c1e"}]},"branch":"refs/heads/master"},"f34269cd224c77621478baca06c94c50fe8bacd0":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-11-28 13:51:04.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/74/696274/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/74/696274/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/74/696274/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/74/696274/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/74/696274/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/74/696274/4"}}},"commit":{"parents":[{"commit":"8d66ca382aa863a32ec74a01a23fcf9709f48559","subject":"Mark test_namespace_sg_isolation as unstable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8d66ca382aa863a32ec74a01a23fcf9709f48559"}]}],"author":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2019-11-27 10:12:39.000000000","tz":60},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2019-11-28 13:50:48.000000000","tz":60},"subject":"Reduce flakiness of pod creation timeout","message":"Reduce flakiness of pod creation timeout\n\nAs this test recreates the namespace and a pod on it,\nit could take a bit longer for the older KuryrNet CRD\nto be removed and recreated, plus the pool to be populated\ncausing the pod creation to timeout. This commit reduce the\nflakiness by increasing the timeout.\n\nChange-Id: I31a874c5f1d0d134e555a32b09a71e9466daea77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/f34269cd224c77621478baca06c94c50fe8bacd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/f34269cd224c77621478baca06c94c50fe8bacd0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
