)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~696576","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~Ic08d76f6c3cebdebd7fe13a254c8e22eb186648f","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"namespace-fixes","hashtags":[],"change_id":"Ic08d76f6c3cebdebd7fe13a254c8e22eb186648f","subject":"Refactor test_namespace","status":"ABANDONED","created":"2019-11-28 13:51:04.000000000","updated":"2021-03-02 15:37:17.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ecfdd4373cfcd67b838239d71ceaf7e4a9c62f17","_number":696576,"virtual_id_number":696576,"owner":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-11-28 15:30:32.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-11-28 17:06:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-11-29 08:31:20.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-28 17:06:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-29 08:31:20.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"}],"messages":[{"id":"076d2671886e2aebba98e161ac2620474b9b9605","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-11-28 13:51:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f3b9580a206a9ac1269ee5db85bcfbf9363ad48","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-11-28 13:52:51.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"c06420488a5477f576b1bce6d02a0a4ac2ac8bdc","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-11-28 13:55:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d6246c60b4e35de385286eef5c2c7816715239ae","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-11-28 13:55:45.000000000","message":"Patch Set 2:\n\ncheck experimental","accounts_in_message":[],"_revision_number":2},{"id":"57ca94cb82fa83e4b24b1c6e5112a97f09f2fab1","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-11-28 15:30:16.000000000","message":"Patch Set 2:\n\nI\u0027m checking if this will work at https://review.opendev.org/#/c/696599/.","accounts_in_message":[],"_revision_number":2},{"id":"2d4cf3384b3d1ad06c153927be9295408241c4b2","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-11-28 15:30:32.000000000","message":"Patch Set 2:\n\n\u003e I\u0027m checking if this will work at https://review.opendev.org/#/c/696599/.\n\nDisregard this, wrong commit.","accounts_in_message":[],"_revision_number":2},{"id":"d8803811abc7a491930dec95ba0662a0a66e8008","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 17:06:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/b899c09d01b24afe8664359f2eecb91e : FAILURE in 2h 00m 05s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/50f35061a04d47588e49456c228f71d3 : FAILURE in 1h 47m 49s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/2229598111054e0d931455e88bdf066c : FAILURE in 1h 55m 46s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/46a7c29169da41d5a46811a3eee8aacf : FAILURE in 1h 50m 22s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/87a86f075bea48cebd88907ea41af418 : TIMED_OUT in 3h 04m 17s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/52505e042b0244b2ba1e8a05647421e8 : SUCCESS in 1h 24m 25s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/3be3dfba8e234b6d8db6e11a9dd7b79b : FAILURE in 1h 44m 13s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/81911be286094107ae9b35fe1c231988 : FAILURE in 1h 43m 49s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/79af3c12564440248037459585432664 : TIMED_OUT in 2h 15m 08s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/c26230d9bf7a45218ac41ccad71aac1b : FAILURE in 41m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5761e9afa0eb00176ef642d212bdf5e9fb5392a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 17:06:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/996a41ebc8d64a0f9478e4729c3af93b : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f30313de062476b87f4e7195919a82f : FAILURE in 3m 49s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/e2937918d4b54583a010ce0e1c1375ac : SUCCESS in 1h 15m 46s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/faa3a6ac77494bfa997ad3980de98075 : FAILURE in 1h 49m 03s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/bbe51c753cbb49888f9d859ff35e788c : SUCCESS in 1h 15m 03s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/d0373fe2bf3c479cb77fc7ebddeb4565 : TIMED_OUT in 3h 04m 11s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/030b8e15bfcb4c33b26dd2e7b660a24d : SUCCESS in 1h 36m 32s (non-voting)\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/d0d00c9b8e0744bcbe00919681004ca9 : SUCCESS in 1h 39m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"570e69960d1b6885dd95ed8d21faff1260d34e6f","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-11-29 08:31:20.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ecfdd4373cfcd67b838239d71ceaf7e4a9c62f17","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-03-02 15:37:17.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5502b4a5dd95774eb8771d1134bc5a207ede383c","revisions":{"f0c8f5e2bb201b9a40fc5eb7c903c44e2e093e81":{"kind":"REWORK","_number":1,"created":"2019-11-28 13:51:04.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/76/696576/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/76/696576/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/76/696576/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/76/696576/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/76/696576/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/76/696576/1"}}},"commit":{"parents":[{"commit":"f34269cd224c77621478baca06c94c50fe8bacd0","subject":"Reduce flakiness of pod creation timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/f34269cd224c77621478baca06c94c50fe8bacd0"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2019-11-28 13:43:27.000000000","tz":60},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2019-11-28 13:50:54.000000000","tz":60},"subject":"Refactor test_namespace","message":"Refactor test_namespace\n\nThis commit refactors and fixes some issues with test_namespace. In\nparticular:\n\n* Take care of namespace deletions after test is done.\n* Rewrite waiting code to use test_utils.call_until_true().\n\nChange-Id: Ic08d76f6c3cebdebd7fe13a254c8e22eb186648f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/f0c8f5e2bb201b9a40fc5eb7c903c44e2e093e81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/f0c8f5e2bb201b9a40fc5eb7c903c44e2e093e81"}]},"branch":"refs/heads/master"},"5502b4a5dd95774eb8771d1134bc5a207ede383c":{"kind":"REWORK","_number":2,"created":"2019-11-28 13:55:31.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/76/696576/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/76/696576/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/76/696576/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/76/696576/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/76/696576/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/76/696576/2"}}},"commit":{"parents":[{"commit":"f34269cd224c77621478baca06c94c50fe8bacd0","subject":"Reduce flakiness of pod creation timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/f34269cd224c77621478baca06c94c50fe8bacd0"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2019-11-28 13:43:27.000000000","tz":60},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2019-11-28 13:54:54.000000000","tz":60},"subject":"Refactor test_namespace","message":"Refactor test_namespace\n\nThis commit refactors and fixes some issues with test_namespace. In\nparticular:\n\n* Take care of namespace deletions after test is done.\n* Rewrite waiting code to use test_utils.call_until_true().\n\nChange-Id: Ic08d76f6c3cebdebd7fe13a254c8e22eb186648f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/5502b4a5dd95774eb8771d1134bc5a207ede383c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/5502b4a5dd95774eb8771d1134bc5a207ede383c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
