)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~698221","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~Id1720eaf94b3a278aa126e2ed3cb9bcdda8e631f","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"port_pool","hashtags":[],"change_id":"Id1720eaf94b3a278aa126e2ed3cb9bcdda8e631f","subject":"Adjust initial ports list in port_pool test","status":"MERGED","created":"2019-12-10 11:50:21.000000000","updated":"2020-01-02 14:18:02.000000000","submitted":"2020-01-02 14:16:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":27,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698221-1577974597371-d94f54cc","meta_rev_id":"0917efb86e2b3639c73467f00fc9c8c6ee43a2f0","_number":698221,"virtual_id_number":698221,"owner":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"date":"2019-12-24 10:55:50.000000000","_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-02 14:16:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-24 14:07:57.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-02 12:37:26.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-12 06:54:45.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"},{"updated":"2019-12-16 09:23:23.000000000","updated_by":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"reviewer":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"state":"REVIEWER"},{"updated":"2019-12-17 12:33:59.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2020-01-02 12:37:26.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2020-01-02 14:16:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f3ea6ac0d589fe4c1850cf2eb4fe1239a764063f","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-10 11:50:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4111078411a2aa19d4c36a4f11952dcf6a25945","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-10 11:52:09.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"18bf80a019e3e8d504b1e7bbc9aa14d833a5ecee","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-10 12:01:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cb2b91621815b6d0bebbdba73231ef295acaf0b9","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-10 12:45:21.000000000","message":"Patch Set 2:\n\ncheck experimental","accounts_in_message":[],"_revision_number":2},{"id":"0c1d133466c946b97063a5d333c5c825b18ba860","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-12-10 13:04:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fafb4eb6b3abb14b93c6e117f806bbafd8340698","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-10 14:14:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"33e766fd08a499288e8aad66c3b642b463827d72","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-10 14:14:29.000000000","message":"Patch Set 3:\n\ncheck experimental","accounts_in_message":[],"_revision_number":3},{"id":"d480e4e13881476a33deba0eb44ff93c32e8d2ef","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-10 14:19:57.000000000","message":"Patch Set 3:\n\ncheck experimental","accounts_in_message":[],"_revision_number":3},{"id":"2f9d54d452c2f8fbe89a9cf0c457243b7ab2fe79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 17:24:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a1057bda5184ec782f28ba890646fdd : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f51cd8ae0c544508554240941d6d320 : FAILURE in 2m 44s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/67d16aa75960409993110029f5f9b410 : SUCCESS in 1h 23m 55s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/2b74458e7fa64b0880b3d4812208ae23 : SUCCESS in 1h 51m 46s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/bdf4cf2010904e3cab54dafa6336aac6 : FAILURE in 1h 58m 09s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/6e40c67fbba44d15b4a05b12a240eb30 : TIMED_OUT in 3h 04m 51s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/67bcf17fe51842cea74ab351af7a3136 : SUCCESS in 1h 34m 45s (non-voting)\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/0547997c73b64b5c84736eb0ea7b7333 : SUCCESS in 1h 35m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9450e7303e415f0b7396e6fc2e30f5710a52c64a","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 17:24:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/d8c31a7a184142b79dcc124e4c67f458 : FAILURE in 2h 16m 51s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/3e71d6f34e4243438765b537089ddb05 : SUCCESS in 1h 35m 25s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/e3bdaa8f0a6d40c98bb4bfa2b64df9cd : SUCCESS in 1h 48m 06s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/29c250df402c435d93d8f97970b35097 : SUCCESS in 1h 43m 28s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/03ccb31fd5b74f7b80ba45c91936f5a3 : TIMED_OUT in 3h 05m 11s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/71d836f49ae44b06b160db59e618d5c4 : FAILURE in 1h 41m 41s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/c538b1cf99704dd89e0486ed75a3786f : FAILURE in 1h 57m 31s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/0268fb6b3f034734a167e1d5fab5da91 : SUCCESS in 1h 25m 15s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/fd8c1a7eddc54c7489d3626c6c174e2c : TIMED_OUT in 2h 16m 08s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/4116f76538df47af80461e7c144e1da9 : FAILURE in 35m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b173843284ce8d65fc4766bbe1da0121804b6394","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-11 07:12:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7cc51fcb4a5d83eb75a843556b9cf56546f9544f","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-11 07:21:28.000000000","message":"Patch Set 4:\n\ncheck experimental","accounts_in_message":[],"_revision_number":4},{"id":"37efeab54c740cf207c55773034f8d6c832ebcf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 10:22:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbc70ec4a5634d6caec78e920123b20c : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11804988a9c54769857cce66b9896bda : SUCCESS in 3m 02s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/4d625bc3534f44599d03184f07f38266 : SUCCESS in 1h 13m 36s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/af3113792019430aa5e0c714faed279e : SUCCESS in 1h 22m 08s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/0f72a0498a704e63819e24ea23736b91 : SUCCESS in 1h 53m 23s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/f8789301d3da457da706fe31fade83ca : TIMED_OUT in 3h 04m 28s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/b1bb86f663f3436c9b93e53934f7e357 : SUCCESS in 1h 18m 14s (non-voting)\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/ddf8ae19910647cb96d6dc55edd5629d : SUCCESS in 1h 58m 20s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7b0bc8ac80940220858805fb3f46f111bdf88d55","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 10:31:25.000000000","message":"Patch Set 4:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/17da82a03ff14cb282822f82a954e2d5 : FAILURE in 2h 42m 44s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/a24f8be8f9a74014b6dc8800544a4fdd : SUCCESS in 1h 28m 07s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/2b44755cb4344cd1a1571cac9e2b5bab : FAILURE in 2h 01m 39s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/8b528dd0016b431e821088b6563461e0 : SUCCESS in 1h 20m 12s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/666f2958fba6467d98be27c1de9b3806 : TIMED_OUT in 3h 04m 06s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/f06f2d18d8e244698ce64942732dac3b : SUCCESS in 1h 12m 07s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/4b6cabb7a5c044bf80b9cd2a6451b02e : FAILURE in 54m 53s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/f746a1e49df642dd82d785010ce7c034 : SUCCESS in 1h 39m 50s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/a0b134e1da8f46d7870aa156e33b8aee : TIMED_OUT in 2h 14m 08s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/d2c21a983ebe4542ac30ea978bc460d4 : FAILURE in 30m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e6e2db6182c9f70e56e0142f6e55407a85a2268a","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-12-13 08:42:19.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"8d822809d2bc67ceda5b55632291c3763467fd14","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-12-16 08:29:00.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"934a8f3bf260d23516b8cfe5ba372a7b9a084140","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 08:29:11.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c0004ffb6f480f79a970b6357bf94f1820da216f","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-12-16 08:35:26.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"94d2d9c3924c74dab7ac08f7ef80070482931219","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-16 09:16:42.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7a8b8681d0eef2d1bcfd4400db397f626d5e021e","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-12-16 09:23:23.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"629cc7d5e42cb791d70d4f470a867292ad24b01b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 11:48:23.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5bcf4016d0340bead60049cb173577c : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c5dc5e1d0ca4207bde388b741ae090c : SUCCESS in 3m 41s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/69e3600eb2654f3ebd4f823ec9650068 : SUCCESS in 1h 28m 33s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/2f888555c726430f97261f370891008f : SUCCESS in 1h 45m 43s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/086c2bed93474aef8efecaa872beb401 : SUCCESS in 1h 48m 39s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/01cbc5ab4ab947358380194ba4a87d6b : FAILURE in 40m 02s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/089cf338e9d24857912518af8006abd2 : SUCCESS in 1h 38m 24s (non-voting)\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/739d3a55000c4c8ca8ce1fff4757fe73 : SUCCESS in 1h 41m 09s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"df4f256a0eda39b80ac72825d6d17d63279b6619","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-12-17 12:33:59.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"9375bcab5df5536d798bfc1a72fe9354550f7e4e","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-12-17 12:41:46.000000000","message":"Patch Set 5: Code-Review+2\n\ncheck experimental","accounts_in_message":[],"_revision_number":5},{"id":"ae0f541044949ab00caaafa1ac1f3f316451fab4","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 15:53:14.000000000","message":"Patch Set 5:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/752d53e928e34a4e9f30b94ded48c3dd : FAILURE in 2h 43m 02s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/52d3fe741e19457893390e6b25596693 : SUCCESS in 1h 39m 24s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/5e350074333d44a5bb7633d618af3145 : FAILURE in 2h 18m 12s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/32edc82414d3481caa8f336e499f32d3 : SUCCESS in 1h 42m 19s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/e1de27852e8a4ed38dff8290875d2823 : TIMED_OUT in 3h 05m 41s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/ace3576e669b4b3f8475df452dc1e90d : SUCCESS in 1h 29m 00s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/a501dd135d9a431e89faef8fb85e1055 : FAILURE in 2h 29m 16s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/7d15eb155489456497db93c878e0583a : SUCCESS in 1h 44m 28s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/55ca7635dbb14f5fbc732db89762cf02 : TIMED_OUT in 2h 14m 46s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/39cd973971c3440c8eaf1a0dc9dcaf09 : FAILURE in 39m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c1b550abbd9b1f9c6f962148273a46908e513fb2","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-18 07:18:42.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"33a57ff483d752ef9eb4b0db621f356cad1bc5e2","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-18 07:27:41.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"cef8cd623453c960acaccc994a77c0456700f31f","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-18 07:47:44.000000000","message":"Patch Set 6:\n\ncheck experimental","accounts_in_message":[],"_revision_number":6},{"id":"9afabb8889961147b3b692d26c4e7df5c534a239","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 09:40:51.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a7941f2f5664ff78db411c44f2886c9 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/858d0af7ba8c478a9391a555861041fb : SUCCESS in 5m 08s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/e2236d7f98fc4ea0b7c73db7de1ba4c3 : SUCCESS in 1h 21m 27s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/64746bef075641f6b2afe41ae156b7b1 : SUCCESS in 1h 53m 41s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/f2bbba03308142358dc74f64cee181c8 : SUCCESS in 1h 26m 21s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/f2b12329af3041b399ea83ad0f3eea3e : SUCCESS in 1h 34m 16s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/cc2cc969ccaf43b487fb8252f0814842 : SUCCESS in 1h 24m 34s (non-voting)\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/73b987099fe6474ea3a60752817dee8b : SUCCESS in 2h 05m 41s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"52346a71d98b4bac87ce42d101ec69457f3898ae","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 11:01:02.000000000","message":"Patch Set 6:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/5117189981e542829cc53593e67f31dd : TIMED_OUT in 3h 06m 51s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/ec026572d0c24d17a2bde2ce6106c68f : SUCCESS in 1h 26m 08s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/ce9bac2029f44a9f8aab84f410cd42e3 : FAILURE in 2h 28m 06s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/ae390f82cacf414b9174074803e3536e : FAILURE in 1h 31m 20s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/b833a3b2ccd9444587e8ac3bd948db89 : TIMED_OUT in 3h 05m 44s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/faad2a562ea34c289ae645c0dcedc7fd : SUCCESS in 1h 23m 24s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/ea793bfa61da4651b5c4e1b28c36da62 : FAILURE in 1h 43m 42s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/8bdc48938e4d403c91aa0125bd9cdac5 : SUCCESS in 1h 41m 13s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/c94d136bf1884a68b5f08764c32f80c0 : TIMED_OUT in 2h 20m 45s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/791347e916604ecf98203dff46a81f45 : FAILURE in 26m 29s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a820bd22c5d93112216e7bce073e059e0e6c1a92","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-18 11:45:56.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"198bba6cc8344246d10da9054c61e6d1f29df825","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-19 12:10:12.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"773f435451fc91b6e547b1eb86b115c6475011ea","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-19 12:11:00.000000000","message":"Patch Set 7:\n\ncheck experimental","accounts_in_message":[],"_revision_number":7},{"id":"4b5635f21ee6dcc98502e9e1552dce94d4d8f546","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 15:17:56.000000000","message":"Patch Set 7:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/29b01c073764439291cb3428e5f1d02b : FAILURE in 2h 04m 20s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/9035a0f4319342958d75646e81324db4 : SUCCESS in 1h 16m 54s\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/efdf4cb174f64edfbb37f12c696e4ffc : SUCCESS in 2h 20m 29s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/9031c239038e4feb87883c4fb1e5846b : SUCCESS in 1h 28m 52s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/bd0bde495e394e67aeb294d7f4533823 : TIMED_OUT in 3h 06m 04s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/38d9ca31be524366981ea279a885477b : SUCCESS in 1h 39m 25s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/fa1a8ffb8c3845f6867ee0fa622d9714 : FAILURE in 2h 04m 13s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/198766974177400c956ab3250e28aec8 : SUCCESS in 2h 07m 30s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/9fdf0c5e4516463b833ee17ba3ac8ff4 : TIMED_OUT in 2h 16m 21s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/c44f440e91ee4d40be289dd1c870126f : FAILURE in 38m 45s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"13ac00080e35cb93d427695070cd9adb7f0c705a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 15:20:52.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d7123e6a62e4038a971b4588d496e8a : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb3abaf9e2e34e08b455e6e76738a64e : SUCCESS in 5m 02s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/0c2ec38fef48416c8f8c28868af86b53 : SUCCESS in 1h 17m 02s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/4e0d2e31117f40098ef070bec1b12740 : SUCCESS in 1h 47m 59s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/8417e56b5ef94ffa8c4e31e4b8033087 : TIMED_OUT in 3h 07m 42s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/43cfd6376b6c41368a6a6464d92c457d : SUCCESS in 1h 33m 24s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/b3286bc6dd3242f39c5299b7f2d029ab : SUCCESS in 1h 39m 34s (non-voting)\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/0e9198801c8d41d9914ac54db47b944f : SUCCESS in 1h 43m 32s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"2312abcdb267bf05fb60bc8a71492e37964728ac","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-19 15:29:18.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"89fad99d335aae91599914a73ffea4ec0e65a44e","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-19 15:43:34.000000000","message":"Patch Set 8:\n\ncheck experimental","accounts_in_message":[],"_revision_number":8},{"id":"e9e90af5f0bc55736a4c2e7a53589a9d74903663","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 17:29:07.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/133501a77edc4915864fb7362c147707 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3a1d5217b8a4f6b8e30762b1d543a93 : SUCCESS in 3m 55s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/b0680b5d62ed43e9a390a5b6232ddafa : SUCCESS in 1h 19m 36s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/51fd0880b1684a44a87f04a9e437fc03 : SUCCESS in 1h 45m 54s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/5b4c6b7c32544ba5b8cdf8cab6a77043 : SUCCESS in 1h 28m 59s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/ad877c09f5ba4858999e738ff2b3ecda : SUCCESS in 1h 50m 42s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/0e9538a8f01a4fae983138ff250e5cba : SUCCESS in 1h 33m 35s (non-voting)\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/5dab36423ee64ac4b3326132fe4b166d : SUCCESS in 1h 52m 13s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"8eb4e28a01473e76b5bef7cadaa636516ca1c97c","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 18:54:34.000000000","message":"Patch Set 8:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/818e00e2031f4f42b7b052c9448d127a : FAILURE in 1h 40m 40s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/40f33d02fbb54e05bb90fdd87716921d : SUCCESS in 1h 34m 07s\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/a6adeb9ee4a34aa88712c0bf23ee6dae : SUCCESS in 1h 30m 21s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/e711ff5616a84812b948b4638236cd0b : SUCCESS in 1h 24m 39s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/132fe094c2404171aab47649f3295e4d : TIMED_OUT in 3h 06m 35s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/fa32a0a30456471eb0e305708fa2399f : SUCCESS in 1h 13m 53s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/997b68e7fefc4e5c84806271bf9ea9a9 : FAILURE in 36m 46s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/fe2aa600c89e4c208cd4a4b3ac113937 : SUCCESS in 1h 46m 12s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/72eadaf6d1b24e30a571062ecbb1ae23 : TIMED_OUT in 2h 16m 54s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/2068c09cda1b49f9980d48272099f0bd : FAILURE in 29m 01s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ccfc4512a7ec110cf2219af63622d1f6ae4057d4","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-12-20 08:54:30.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)\n\n+2 as I don\u0027t have any better idea for how to wait for that. Perhaps worth to add those sleeps in the other functions to ensure the same is tested always","accounts_in_message":[],"_revision_number":8},{"id":"05b0a5d0e823fc076af6d4742bde217767ea5648","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-20 09:15:03.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"ea84b3a0e36db057d01aaa064043f6bb8e0ef1ff","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-22 14:33:46.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"71427429b949ad28c4a8734c5b51a612b5ac0205","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-22 14:38:30.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"f8e3d5f5561ad83b47f2eae6292fb4ce778e3fc4","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-22 14:38:46.000000000","message":"Patch Set 10:\n\ncheck experimental","accounts_in_message":[],"_revision_number":10},{"id":"b93c9457bab28976f3b67ebc6f52031908ec61fc","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-22 15:10:58.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"967f12fb36e587d7f3f51f2377d4999d9bf696c7","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-22 15:11:14.000000000","message":"Patch Set 11:\n\ncheck experimental","accounts_in_message":[],"_revision_number":11},{"id":"61175d2d78383d7c3f2633b08039a0abc2d29e0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-22 16:23:12.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/4f4475dba67546ba8772f09ac665208b : SUCCESS in 2m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4aec354836c141e189a10fa37755a53d : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b2b2059a93c41f0a9ddb743922298b3 : SUCCESS in 2m 41s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/e43b17005c954de7929b730ce82b6f2f : FAILURE in 47m 09s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/107326fe3bcf4b3dae67e58062349a3c : FAILURE in 46m 27s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/6c8d3edfed404c43af23960be26a9516 : FAILURE in 45m 33s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/26b0f875e002405780453a4480617d2f : FAILURE in 49m 38s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/d5647f5cad104f1f85ab7a4508fffb8c : FAILURE in 48m 28s (non-voting)\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/32a8a976ef454f6e9d3e92c9c1cc41ef : FAILURE in 1h 07m 07s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"baa96c9e48faa8547c5ee665385c4eeb0e239e0b","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-22 16:31:28.000000000","message":"Patch Set 11:\n\nBuild failed (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/bd2c20c214ff4dd8a9fd8d5c8d61c9ad : FAILURE in 1h 00m 07s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/d35d2ddbe69540b8a33bb57c8fb0a5a3 : FAILURE in 56m 22s\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/5305de674c8f45d6bca97c63c103b597 : FAILURE in 52m 26s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/f8472e8a92b147c2902ba8fbb37a22b1 : FAILURE in 56m 20s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/b02f3f976e894de6a23fc1ef5c206d3f : FAILURE in 52m 40s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/d77cd16ec9e64bda95383d91c88c5458 : FAILURE in 52m 22s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/6251d22fdd0a4533a502b68b8f8104e7 : FAILURE in 51m 01s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/6dd3b077b5774df58a073bd8445ee874 : FAILURE in 49m 00s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/0f50102475f94d4697a03399502f1ff0 : FAILURE in 1h 14m 31s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/af208810817e40b7b1f98eac0bfab776 : FAILURE in 29m 09s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"71c903222910d6e786dc7c557b68fe9b740a825c","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-23 07:12:45.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"261fb17a81c5260f3b7173ba5f1e7780490030dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 08:41:19.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b90e9181b48d4d8a8a20122b0edc4e2c : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea5b925b7e464c8c95fd585eae1577c7 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dd143305d0b4213a36d4f10b8e877b9 : SUCCESS in 4m 01s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/b256137a33c94361983d0cfa0925e036 : FAILURE in 44m 38s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/90ff0947753543679b1c8473a5c15c8b : FAILURE in 45m 56s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/20e28190052d4c3c9f8f614d19324908 : FAILURE in 1h 08m 58s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/8b76d2c147f34066a341e6a5d7c8ce9f : FAILURE in 49m 55s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/5898e4ddf164412dac28f350b00af17c : FAILURE in 56m 27s (non-voting)\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/31cba7183a51486b9fd5ed1cc93c7d51 : FAILURE in 1h 11m 50s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"925e8ae9f15f77f0eb2b08eb8dea8b15249dd01b","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-23 10:16:10.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"444836145d0bbdb4a6d7381f85a95ed87a03030a","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-23 10:18:56.000000000","message":"Patch Set 12:\n\ncheck experimental","accounts_in_message":[],"_revision_number":12},{"id":"6df0965ffe81eda159d189a01740e8f3cc67b9a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 12:06:15.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7bc619ce8507425093aa1057f625a0b7 : SUCCESS in 2m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d02d4a3d730e4d5fa359f545c8874512 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e515d650ce8648f293b71aad4b3c3527 : SUCCESS in 3m 35s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/cdcd902ab9704b929868b1f8ba6175d6 : SUCCESS in 1h 15m 45s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/22ffcc452bb249b39259b902ffc0574d : SUCCESS in 1h 49m 16s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/63663e3a591a4b82b6078b99ff821afb : SUCCESS in 1h 18m 30s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/6e0376fe01674f8ebb795cd2d6668f05 : SUCCESS in 1h 18m 07s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/23c7ecf34f1045e2bd6f9103fe51f0c9 : SUCCESS in 1h 34m 01s (non-voting)\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/6560638f676a421db1c857d29a353df6 : SUCCESS in 1h 35m 07s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"4eb4ea8f3d773db063eea4d04b16b7b6993211a0","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-23 12:48:40.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"6b0fe20d6c314aefe633a783634ed662ff494506","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-12-23 12:54:40.000000000","message":"Patch Set 13: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"aad3eda26cff66a99e92fdc272e27beb464bd91e","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-23 13:39:33.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"1bde02ff5451b48043f918e67ee7474f4df9012c","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-23 14:05:48.000000000","message":"Patch Set 14:\n\n(1 comment)\n\ncheck experimental","accounts_in_message":[],"_revision_number":14},{"id":"72c8c8421e7698fd083e1926928ae1ac22d46944","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-12-23 15:29:35.000000000","message":"Patch Set 14: Code-Review-1\n\n(5 comments)\n\nIf you add a new function to avoid the wait, you should add it in the other parts too","accounts_in_message":[],"_revision_number":14},{"id":"4c5a51b204d2d5f10408f44774bd4ff645e0d670","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 16:11:32.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4d80fcea6ae4bfc8ca19da938557112 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94f1a5ade38b4f4ea1389bfdcb758945 : SUCCESS in 3m 41s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/956a4e0632d444df99e0eb2c333b2b39 : SUCCESS in 1h 13m 48s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/ab50f52b91884cf69a2a44a3403ebeea : SUCCESS in 1h 38m 58s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/f8070bc49c0f40b483d12677bb030183 : FAILURE in 1h 40m 55s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/bb94f2c53328443683c83734542d7da5 : FAILURE in 2h 24m 40s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/fc18416688dc4ed0b06a89c51d299c18 : SUCCESS in 1h 27m 46s (non-voting)\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/68de87bd625a49b590b814972d2be84d : FAILURE in 56m 39s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"f6323281afc51e53c96f18b27acb092c6e71c315","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-23 22:56:17.000000000","message":"Patch Set 14:\n\n(5 comments)","accounts_in_message":[],"_revision_number":14},{"id":"9f1cac4cd6ff775b219e0927b24a04b22d1ba30d","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-23 23:10:07.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"320cd55cecde3d9248168a93baa72d5c276e9b81","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-23 23:16:27.000000000","message":"Patch Set 15:\n\ncheck experimental","accounts_in_message":[],"_revision_number":15},{"id":"c3e3da38eb7665214574c5dfde6cb12d932d2812","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-24 01:04:36.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ce08c65c62e471b884b3fb42c687f7e : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e375c82024e45f4b06284b1973bbf7e : SUCCESS in 3m 08s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/3e47729e8d604f2e850db61ab00c6805 : SUCCESS in 1h 37m 00s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/2bbc6a44a205482f83e608884b3f369b : SUCCESS in 1h 22m 43s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/86bc4ec978e3487ca482037a0d0b31fc : SUCCESS in 1h 22m 47s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/6dbe005a7f184d1ca3eedd7b2a766853 : SUCCESS in 1h 22m 59s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/92e43bf370db4fc297d6ae46783761b9 : SUCCESS in 1h 40m 09s (non-voting)\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/cbb5630f6bc34c0e813ecd36764ed606 : SUCCESS in 1h 53m 03s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"d4caae7d93ed6aca3989d44ac2b14fb52a4844d6","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-24 02:34:12.000000000","message":"Patch Set 15:\n\nBuild failed (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/cee3496377994899917d76f3bc5a33ef : FAILURE in 1h 26m 09s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/2af33a54626e465582068600f9bb8e59 : SUCCESS in 1h 33m 34s\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/ef17cb42863243bf9b1e97459f46856a : FAILURE in 1h 50m 13s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/eb4b1e0e8d4b4e118eb054e4124db787 : SUCCESS in 1h 23m 02s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/d5a4d52d67a04dfbaf0ab6c75b461c6e : TIMED_OUT in 3h 05m 25s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/3e127bae944d4b34b0cb2a538cb28c6f : SUCCESS in 1h 08m 47s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/a8619301bfa5428897a3298a3208a608 : FAILURE in 1h 40m 51s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/b812bb5b9ea045ddad086a7d8fe279f4 : SUCCESS in 1h 22m 13s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/9948716cd4684a41a80d3f5a6938b920 : TIMED_OUT in 2h 15m 59s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/d9c2b1f7b98d46aea02e47521bb134a5 : FAILURE in 27m 03s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"3b17dbb83c10f0b6964aaab5589b48b5cc06a74e","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-12-24 08:35:38.000000000","message":"Patch Set 15: Code-Review-1\n\n(3 comments)\n\njust a couple of nits","accounts_in_message":[],"_revision_number":15},{"id":"100986238dac786c26a05170d0908ef2764874aa","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-24 08:49:35.000000000","message":"Patch Set 14:\n\n(3 comments)","accounts_in_message":[],"_revision_number":14},{"id":"ad3637ad6051c3e7ec6c5e9d6c8b4f27d21b5d3b","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-24 10:08:23.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"73727fed6d3fc1505612868dece3919ce59e0391","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-24 10:55:50.000000000","message":"Patch Set 16:\n\ncheck experimental","accounts_in_message":[],"_revision_number":16},{"id":"f0c671eabbac87a28afe3d4a1eba0d0eb5acee65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-24 11:41:42.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20c4321aabe144cb94102b6db64838d6 : SUCCESS in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4abcb2a537041e58f28ae8785b675de : SUCCESS in 2m 52s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/8b0dbb14a7fc4c718a87877d291e62d3 : SUCCESS in 1h 11m 03s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/ae9c0c9a3b6342a99234957052a24fba : SUCCESS in 1h 14m 44s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/b97eb08374f44bf8bb3e93f767edb36c : SUCCESS in 1h 08m 17s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/007de0fdae0f4e5890f12a90e2eb0807 : SUCCESS in 1h 13m 59s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/f6c1986751c1401396aa48f23fcc53d7 : SUCCESS in 1h 25m 24s (non-voting)\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/1216c0db03b343d6b49c9e4f23437d86 : SUCCESS in 1h 31m 55s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"bcd113068c7dc6ffc79baefb55fc2eacf5b11afd","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-12-24 13:59:38.000000000","message":"Patch Set 16: Code-Review+1\n\nwaiting for the experimental gate. Looks good to me","accounts_in_message":[],"_revision_number":16},{"id":"a42769ef00b7873bdb9e0b1769953766701e786f","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-24 14:02:05.000000000","message":"Patch Set 16:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/027b5d414933437b8f95640d0fadba3e : FAILURE in 1h 41m 29s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/4906f4e1a4344358850932cfee3a0159 : SUCCESS in 1h 38m 46s\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/3b733ecb083f41b78cc83e6c15f33482 : SUCCESS in 1h 49m 41s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/a16323a4ef6c4616aea0012e53256619 : SUCCESS in 1h 31m 11s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/6650b85b199e416cb735579eb6f43656 : TIMED_OUT in 3h 05m 36s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/35ded09a1a254019aaf6ca7fb0b4d1e8 : SUCCESS in 1h 27m 25s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/466c20e9ff524d55a92de8b067740bc4 : FAILURE in 1h 44m 23s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/a2477041d1784c5c9d198ce5ade11070 : SUCCESS in 1h 35m 36s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/f6a724f004ba419abb188b8f00d47f6e : TIMED_OUT in 2h 16m 49s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/1a606b14d5a84a6ebc85a6f2798a9572 : FAILURE in 35m 43s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"b2ce61410532047c6df54fd945e42d4d530ce025","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-12-24 14:07:57.000000000","message":"Patch Set 16: Code-Review+2","accounts_in_message":[],"_revision_number":16},{"id":"dcbbaf40b196ff0869f31dbef4fc6368eb6c4a74","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-01-02 08:21:26.000000000","message":"Patch Set 16:\n\ncheck experimental","accounts_in_message":[],"_revision_number":16},{"id":"707a1c1e28fb47d238d6395ba244fdfc7418a015","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 11:46:43.000000000","message":"Patch Set 16:\n\nBuild failed (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/4e1024d3d43e42c7b08c0583b97796be : FAILURE in 1h 37m 13s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/05b62d0f7bd9419fb6447659b236c0ef : SUCCESS in 1h 16m 46s\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/ccfb0e8bef68469eba59c2e4408410cb : FAILURE in 1h 42m 33s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/7add8bc88cf240e09c99d667ba9ffb67 : FAILURE in 1h 25m 45s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/591753f60b364a3f95fc66c0300a4062 : TIMED_OUT in 3h 05m 15s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/98786d147ff94683bb62c3ee96968abe : SUCCESS in 1h 23m 53s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress https://zuul.opendev.org/t/openstack/build/6d350de288734fd0a9eaa56f60e4ed33 : FAILURE in 2h 28m 21s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/49e6f8abc89741a2a66cb43d6c839d4a : SUCCESS in 1h 34m 21s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/d19a8872701940009073edf079e30273 : TIMED_OUT in 2h 17m 44s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/8e7f88e1b8d5410fa7b4fae3a277626d : FAILURE in 27m 24s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"432c9c8a4c0135cf78ca0972184ebac5fcf9b28c","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-01-02 12:37:26.000000000","message":"Patch Set 16: Workflow+1\n\nAs this already had +2+W from dulek, I\u0027m readding it","accounts_in_message":[],"_revision_number":16},{"id":"2a89f1b8db2ffe7b819f86572d7d58714d2ad2e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 12:37:37.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"d36ec7764980cdc6d229dc0deebf056eb0f56035","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 14:16:37.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ceddbf2de784123a1f315cbefc57d39 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f9b166443964f4eaa4bf670aa33dcc0 : SUCCESS in 3m 17s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/9048ccb3ff5147858c846c4da64d3202 : SUCCESS in 1h 26m 09s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/962fe1f636234de09803cba8b2d490cd : SUCCESS in 1h 27m 22s","accounts_in_message":[],"_revision_number":16},{"id":"dcb76767527c3b4585925d7781392ff28a3ac422","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 14:16:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":16},{"id":"d8401ba97acd0fe255e95843d5f2f6c2a645cf93","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 14:18:02.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07680f80b9cb4697b58f4fca47f29467 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"e701961debfdd58e0999d79921f9d8c9574db43b","revisions":{"a8d7d8ba5002c65cc820871f1a759f74430cdc70":{"kind":"REWORK","_number":1,"created":"2019-12-10 11:50:21.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/21/698221/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/21/698221/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/1"}}},"commit":{"parents":[{"commit":"2ddd1dd65d333911ac33d974116f54e43eaa4ce7","subject":"Merge \"Reduce flakiness of pod creation timeout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2ddd1dd65d333911ac33d974116f54e43eaa4ce7"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-12-10 11:44:07.000000000","tz":-300},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-12-10 11:49:38.000000000","tz":-300},"subject":"Adjust initial ports list in port_pool test","message":"Adjust initial ports list in port_pool test\n\nAdjusting also test_port_pool_min_max_update,\nDue to prepopulation of ports when creating namespace.\n\nChange-Id: Id1720eaf94b3a278aa126e2ed3cb9bcdda8e631f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/a8d7d8ba5002c65cc820871f1a759f74430cdc70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/a8d7d8ba5002c65cc820871f1a759f74430cdc70"}]},"branch":"refs/heads/master"},"9e4b7c55ef54f9901ca6ae0c0f91cddb4b143708":{"kind":"REWORK","_number":2,"created":"2019-12-10 12:01:13.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/21/698221/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/21/698221/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/2"}}},"commit":{"parents":[{"commit":"2ddd1dd65d333911ac33d974116f54e43eaa4ce7","subject":"Merge \"Reduce flakiness of pod creation timeout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2ddd1dd65d333911ac33d974116f54e43eaa4ce7"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-12-10 11:44:07.000000000","tz":-300},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-12-10 12:00:58.000000000","tz":120},"subject":"Adjust initial ports list in port_pool test","message":"Adjust initial ports list in port_pool test\n\nAdjusting also test_port_pool_min_max_update,\nDue to prepopulation of ports when creating namespace.\n\nChange-Id: Id1720eaf94b3a278aa126e2ed3cb9bcdda8e631f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/9e4b7c55ef54f9901ca6ae0c0f91cddb4b143708"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/9e4b7c55ef54f9901ca6ae0c0f91cddb4b143708"}]},"branch":"refs/heads/master"},"b5929f4eb3a09b07c5a32f80ef9f6dd125c0232d":{"kind":"REWORK","_number":3,"created":"2019-12-10 14:14:10.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/21/698221/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/21/698221/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/3"}}},"commit":{"parents":[{"commit":"2ddd1dd65d333911ac33d974116f54e43eaa4ce7","subject":"Merge \"Reduce flakiness of pod creation timeout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2ddd1dd65d333911ac33d974116f54e43eaa4ce7"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-12-10 11:44:07.000000000","tz":-300},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-12-10 14:13:48.000000000","tz":-300},"subject":"Adjust initial ports list in port_pool test","message":"Adjust initial ports list in port_pool test\n\nAdjusting also test_port_pool_min_max_update,\nDue to prepopulation of ports when creating namespace.\n\nChange-Id: Id1720eaf94b3a278aa126e2ed3cb9bcdda8e631f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/b5929f4eb3a09b07c5a32f80ef9f6dd125c0232d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/b5929f4eb3a09b07c5a32f80ef9f6dd125c0232d"}]},"branch":"refs/heads/master"},"d02743f25b1409b6ae1685431bd39bcdcfe71b3f":{"kind":"REWORK","_number":4,"created":"2019-12-11 07:12:47.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/21/698221/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/21/698221/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/4"}}},"commit":{"parents":[{"commit":"2ddd1dd65d333911ac33d974116f54e43eaa4ce7","subject":"Merge \"Reduce flakiness of pod creation timeout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2ddd1dd65d333911ac33d974116f54e43eaa4ce7"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-12-10 11:44:07.000000000","tz":-300},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-12-11 07:12:38.000000000","tz":120},"subject":"Adjust initial ports list in port_pool test","message":"Adjust initial ports list in port_pool test\n\nAdjusting also test_port_pool_min_max_update,\nDue to prepopulation of ports when creating namespace.\n\nChange-Id: Id1720eaf94b3a278aa126e2ed3cb9bcdda8e631f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d02743f25b1409b6ae1685431bd39bcdcfe71b3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d02743f25b1409b6ae1685431bd39bcdcfe71b3f"}]},"branch":"refs/heads/master"},"228866306f84d715706e55858d162dfba5248219":{"kind":"REWORK","_number":5,"created":"2019-12-16 09:16:42.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/21/698221/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/21/698221/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/5"}}},"commit":{"parents":[{"commit":"2ddd1dd65d333911ac33d974116f54e43eaa4ce7","subject":"Merge \"Reduce flakiness of pod creation timeout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2ddd1dd65d333911ac33d974116f54e43eaa4ce7"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-12-10 11:44:07.000000000","tz":-300},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-12-16 09:16:01.000000000","tz":120},"subject":"Adjust initial ports list in port_pool test","message":"Adjust initial ports list in port_pool test\n\nAdjusting also test_port_pool_min_max_update,\nDue to prepopulation of ports when creating namespace.\n\nChange-Id: Id1720eaf94b3a278aa126e2ed3cb9bcdda8e631f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/228866306f84d715706e55858d162dfba5248219"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/228866306f84d715706e55858d162dfba5248219"}]},"branch":"refs/heads/master"},"d3b6cec7e3c366d52fe9a97b50f66e8a96813673":{"kind":"REWORK","_number":6,"created":"2019-12-18 07:27:41.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/21/698221/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/21/698221/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/6"}}},"commit":{"parents":[{"commit":"2ddd1dd65d333911ac33d974116f54e43eaa4ce7","subject":"Merge \"Reduce flakiness of pod creation timeout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2ddd1dd65d333911ac33d974116f54e43eaa4ce7"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-12-10 11:44:07.000000000","tz":-300},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-12-18 07:27:33.000000000","tz":120},"subject":"Adjust initial ports list in port_pool test","message":"Adjust initial ports list in port_pool test\n\nAdjusting also test_port_pool_min_max_update,\nDue to prepopulation of ports when creating namespace.\n\nChange-Id: Id1720eaf94b3a278aa126e2ed3cb9bcdda8e631f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d3b6cec7e3c366d52fe9a97b50f66e8a96813673"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d3b6cec7e3c366d52fe9a97b50f66e8a96813673"}]},"branch":"refs/heads/master"},"136e8280993876556e4c82411ccef2f0fb641384":{"kind":"REWORK","_number":7,"created":"2019-12-19 12:10:12.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/21/698221/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/21/698221/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/7"}}},"commit":{"parents":[{"commit":"2ddd1dd65d333911ac33d974116f54e43eaa4ce7","subject":"Merge \"Reduce flakiness of pod creation timeout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2ddd1dd65d333911ac33d974116f54e43eaa4ce7"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-12-10 11:44:07.000000000","tz":-300},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-12-19 12:09:28.000000000","tz":120},"subject":"Adjust initial ports list in port_pool test","message":"Adjust initial ports list in port_pool test\n\nAdjusting also test_port_pool_min_max_update,\nDue to prepopulation of ports when creating namespace.\n\nChange-Id: Id1720eaf94b3a278aa126e2ed3cb9bcdda8e631f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/136e8280993876556e4c82411ccef2f0fb641384"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/136e8280993876556e4c82411ccef2f0fb641384"}]},"branch":"refs/heads/master"},"982e2fd79186014242d7ff4f641ee6a03e2cd75a":{"kind":"REWORK","_number":8,"created":"2019-12-19 15:29:18.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/21/698221/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/21/698221/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/8"}}},"commit":{"parents":[{"commit":"2ddd1dd65d333911ac33d974116f54e43eaa4ce7","subject":"Merge \"Reduce flakiness of pod creation timeout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2ddd1dd65d333911ac33d974116f54e43eaa4ce7"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-12-10 11:44:07.000000000","tz":-300},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-12-19 15:28:47.000000000","tz":120},"subject":"Adjust initial ports list in port_pool test","message":"Adjust initial ports list in port_pool test\n\nAdjusting also test_port_pool_min_max_update,\nDue to prepopulation of ports when creating namespace.\n\nChange-Id: Id1720eaf94b3a278aa126e2ed3cb9bcdda8e631f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/982e2fd79186014242d7ff4f641ee6a03e2cd75a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/982e2fd79186014242d7ff4f641ee6a03e2cd75a"}]},"branch":"refs/heads/master"},"9ab7cecd9797fdeac0a24259dd5be6359b6b6902":{"kind":"REWORK","_number":9,"created":"2019-12-22 14:33:46.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/21/698221/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/21/698221/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/9"}}},"commit":{"parents":[{"commit":"2ddd1dd65d333911ac33d974116f54e43eaa4ce7","subject":"Merge \"Reduce flakiness of pod creation timeout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2ddd1dd65d333911ac33d974116f54e43eaa4ce7"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-12-10 11:44:07.000000000","tz":-300},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-12-22 14:33:30.000000000","tz":120},"subject":"Adjust initial ports list in port_pool test","message":"Adjust initial ports list in port_pool test\n\nAdjusting also test_port_pool_min_max_update,\nDue to prepopulation of ports when creating namespace.\n\nChange-Id: Id1720eaf94b3a278aa126e2ed3cb9bcdda8e631f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/9ab7cecd9797fdeac0a24259dd5be6359b6b6902"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/9ab7cecd9797fdeac0a24259dd5be6359b6b6902"}]},"branch":"refs/heads/master"},"f8ce0b1591afe1f9c13184acb710fb2ced424986":{"kind":"REWORK","_number":10,"created":"2019-12-22 14:38:30.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/21/698221/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/21/698221/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/10"}}},"commit":{"parents":[{"commit":"2ddd1dd65d333911ac33d974116f54e43eaa4ce7","subject":"Merge \"Reduce flakiness of pod creation timeout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2ddd1dd65d333911ac33d974116f54e43eaa4ce7"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-12-10 11:44:07.000000000","tz":-300},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-12-22 14:38:21.000000000","tz":120},"subject":"Adjust initial ports list in port_pool test","message":"Adjust initial ports list in port_pool test\n\nAdjusting also test_port_pool_min_max_update,\nDue to prepopulation of ports when creating namespace.\n\nChange-Id: Id1720eaf94b3a278aa126e2ed3cb9bcdda8e631f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/f8ce0b1591afe1f9c13184acb710fb2ced424986"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/f8ce0b1591afe1f9c13184acb710fb2ced424986"}]},"branch":"refs/heads/master"},"c44c7dfdf5fec892f851dec29fbc5c80fbd0e1dd":{"kind":"REWORK","_number":11,"created":"2019-12-22 15:10:58.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/21/698221/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/21/698221/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/11"}}},"commit":{"parents":[{"commit":"2ddd1dd65d333911ac33d974116f54e43eaa4ce7","subject":"Merge \"Reduce flakiness of pod creation timeout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2ddd1dd65d333911ac33d974116f54e43eaa4ce7"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-12-10 11:44:07.000000000","tz":-300},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-12-22 15:10:47.000000000","tz":120},"subject":"Adjust initial ports list in port_pool test","message":"Adjust initial ports list in port_pool test\n\nAdjusting also test_port_pool_min_max_update,\nDue to prepopulation of ports when creating namespace.\n\nChange-Id: Id1720eaf94b3a278aa126e2ed3cb9bcdda8e631f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c44c7dfdf5fec892f851dec29fbc5c80fbd0e1dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c44c7dfdf5fec892f851dec29fbc5c80fbd0e1dd"}]},"branch":"refs/heads/master"},"5c5a8ae183b2d9093841a58bf5436d0d429f925a":{"kind":"REWORK","_number":12,"created":"2019-12-23 10:16:10.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/21/698221/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/21/698221/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/12"}}},"commit":{"parents":[{"commit":"2ddd1dd65d333911ac33d974116f54e43eaa4ce7","subject":"Merge \"Reduce flakiness of pod creation timeout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2ddd1dd65d333911ac33d974116f54e43eaa4ce7"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-12-10 11:44:07.000000000","tz":-300},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-12-23 10:16:01.000000000","tz":120},"subject":"Adjust initial ports list in port_pool test","message":"Adjust initial ports list in port_pool test\n\nAdjusting also test_port_pool_min_max_update,\nDue to prepopulation of ports when creating namespace.\n\nChange-Id: Id1720eaf94b3a278aa126e2ed3cb9bcdda8e631f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/5c5a8ae183b2d9093841a58bf5436d0d429f925a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/5c5a8ae183b2d9093841a58bf5436d0d429f925a"}]},"branch":"refs/heads/master"},"40a6459c71bbcc5e1594a8c7c5a0e919ee75fac6":{"kind":"REWORK","_number":13,"created":"2019-12-23 12:48:40.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/21/698221/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/21/698221/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/13"}}},"commit":{"parents":[{"commit":"2ddd1dd65d333911ac33d974116f54e43eaa4ce7","subject":"Merge \"Reduce flakiness of pod creation timeout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2ddd1dd65d333911ac33d974116f54e43eaa4ce7"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-12-10 11:44:07.000000000","tz":-300},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-12-23 12:48:30.000000000","tz":120},"subject":"Adjust initial ports list in port_pool test","message":"Adjust initial ports list in port_pool test\n\nAdjusting also test_port_pool_min_max_update,\nDue to prepopulation of ports when creating namespace.\n\nChange-Id: Id1720eaf94b3a278aa126e2ed3cb9bcdda8e631f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/40a6459c71bbcc5e1594a8c7c5a0e919ee75fac6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/40a6459c71bbcc5e1594a8c7c5a0e919ee75fac6"}]},"branch":"refs/heads/master"},"cfdbce71aacec17f414761cf3439352307e369f9":{"kind":"REWORK","_number":14,"created":"2019-12-23 13:39:33.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/21/698221/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/21/698221/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/14"}}},"commit":{"parents":[{"commit":"2ddd1dd65d333911ac33d974116f54e43eaa4ce7","subject":"Merge \"Reduce flakiness of pod creation timeout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2ddd1dd65d333911ac33d974116f54e43eaa4ce7"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-12-10 11:44:07.000000000","tz":-300},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-12-23 13:38:45.000000000","tz":120},"subject":"Adjust initial ports list in port_pool test","message":"Adjust initial ports list in port_pool test\n\nAdjusting also test_port_pool_min_max_update,\nDue to prepopulation of ports when creating namespace.\n\nChange-Id: Id1720eaf94b3a278aa126e2ed3cb9bcdda8e631f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/cfdbce71aacec17f414761cf3439352307e369f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/cfdbce71aacec17f414761cf3439352307e369f9"}]},"branch":"refs/heads/master"},"0dfa6ed2078d3d6d989b520d9a35a8fd0a5d7c7b":{"kind":"REWORK","_number":15,"created":"2019-12-23 23:10:07.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/21/698221/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/21/698221/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/15"}}},"commit":{"parents":[{"commit":"2ddd1dd65d333911ac33d974116f54e43eaa4ce7","subject":"Merge \"Reduce flakiness of pod creation timeout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2ddd1dd65d333911ac33d974116f54e43eaa4ce7"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-12-10 11:44:07.000000000","tz":-300},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-12-23 23:09:55.000000000","tz":120},"subject":"Adjust initial ports list in port_pool test","message":"Adjust initial ports list in port_pool test\n\nAdjusting also test_port_pool_min_max_update,\nDue to prepopulation of ports when creating namespace.\n\nChange-Id: Id1720eaf94b3a278aa126e2ed3cb9bcdda8e631f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/0dfa6ed2078d3d6d989b520d9a35a8fd0a5d7c7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/0dfa6ed2078d3d6d989b520d9a35a8fd0a5d7c7b"}]},"branch":"refs/heads/master"},"e701961debfdd58e0999d79921f9d8c9574db43b":{"kind":"REWORK","_number":16,"created":"2019-12-24 10:08:23.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/21/698221/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/21/698221/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/698221/16"}}},"commit":{"parents":[{"commit":"2ddd1dd65d333911ac33d974116f54e43eaa4ce7","subject":"Merge \"Reduce flakiness of pod creation timeout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2ddd1dd65d333911ac33d974116f54e43eaa4ce7"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-12-10 11:44:07.000000000","tz":-300},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2019-12-24 10:05:33.000000000","tz":120},"subject":"Adjust initial ports list in port_pool test","message":"Adjust initial ports list in port_pool test\n\nAdjusting also test_port_pool_min_max_update,\nDue to prepopulation of ports when creating namespace.\n\nChange-Id: Id1720eaf94b3a278aa126e2ed3cb9bcdda8e631f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/e701961debfdd58e0999d79921f9d8c9574db43b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/e701961debfdd58e0999d79921f9d8c9574db43b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
