)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~707103","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~Ic2be3cf93bd9d114af907d26198e8011281bfabf","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"ipv6","hashtags":[],"change_id":"Ic2be3cf93bd9d114af907d26198e8011281bfabf","subject":"Basic IPv6 support","status":"MERGED","created":"2020-02-11 10:56:07.000000000","updated":"2020-02-17 18:16:02.000000000","submitted":"2020-02-17 18:14:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707103-1581963265936-0cc33de8","meta_rev_id":"246c4ca658e74247a509166220a5c42e2b4c22cc","_number":707103,"virtual_id_number":707103,"owner":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-17 18:14:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":2,"date":"2020-02-17 12:08:22.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-17 08:37:55.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":1,"date":"2020-02-17 12:08:17.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-17 08:37:55.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2020-02-17 12:08:17.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"},{"updated":"2020-02-17 18:14:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7042bc5bede1f11fb63b5f064f04b32c6bc2c18a","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-02-11 10:56:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"966ce9729415506416b46846c41b1adfb73682b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 12:45:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5968e7307a324a89915adf09cd4c6d08 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33ca23893fd846ae962eb5a14d764ccb : SUCCESS in 5m 09s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/8055f87f6abb4e34aaeecd940a0fa89e : SUCCESS in 1h 11m 40s\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/12fac4ae65fa4abebdc08e06da4192f3 : FAILURE in 35m 39s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/c5eb80c1a6f1466794057c734bcba8de : SUCCESS in 1h 30m 45s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/581939488f18455cb70c8b6325942b03 : SUCCESS in 1h 46m 24s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/b5aa3fddff6a47f196131f0a705f07e9 : SUCCESS in 1h 24m 26s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/1f011a0563044b1f98e43f18a21bb162 : SUCCESS in 1h 45m 01s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/f5c9a474c9d9437ea440229bd95150fd : SUCCESS in 1h 43m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6f97631da6208dd3184bb0de9e7db10a615f0725","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-02-11 15:22:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d5cd1a1439fb82165400e47a33c55aae1b7dd34c","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-02-11 15:40:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5a56665efcc92a43c6e5e8d3cc178eb2f65d4c48","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-02-11 17:17:03.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e845f8a0877d2d191d3bea4a8d85e8978390f671","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-02-11 17:18:19.000000000","message":"Patch Set 4: Workflow-1\n\nWe might want to split IPv6\u0027ness of pod and service subnet.","accounts_in_message":[],"_revision_number":4},{"id":"77b92baaa8bea8193a518c54f6b77e9747d34e2d","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-02-11 17:39:54.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"32911140c553d2c144786e64c6451a1524891b8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 18:54:02.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95fafe4258a84238b75684a09d7cd03f : RETRY_LIMIT in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53483082a99345d786c609acd61cf46d : RETRY_LIMIT in 2m 27s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/2b8ed3c95d674ecc8ed4862f789a35d5 : RETRY_LIMIT in 4m 06s\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/22717c4f1ab8444abc5f776b98c8b620 : RETRY_LIMIT in 4m 24s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/4dfedac8a8114ef8bf6ea42911c81f66 : RETRY_LIMIT in 4m 02s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/e639dbc6c45b47c39bb3e1ed5aa629fa : RETRY_LIMIT in 4m 42s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/91187455bb40445c827f82601eb9ee19 : RETRY_LIMIT in 4m 14s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/17c275f512c849898660f54a7a3b03de : RETRY_LIMIT in 4m 34s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/95fc355c6f4046bb979f5b5991306ef7 : RETRY_LIMIT in 6m 49s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"bdeeab1fd6c154bce2de6cfe0690b840b92e343d","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-02-12 09:08:38.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"0c683f38b8363686da4b3d72a5754205921f934d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 11:27:25.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb8dc3851a0747e3acbe4d6548f7d27d : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65fe4bd254d245ef9ea8425107d72a5b : SUCCESS in 5m 22s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/9e857aa7be144709a5e5b55b374c1edb : SUCCESS in 1h 19m 22s\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/97da4b995b874e70bdb97929c80b75aa : FAILURE in 2h 12m 41s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/59af677f85ba48809e7737f38c7239ce : SUCCESS in 1h 33m 32s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/7e21d32937ec4c5d88b0c2f194c006ed : SUCCESS in 1h 23m 49s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/5132d38469174d14ad65db4ad1032cb5 : SUCCESS in 1h 17m 54s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/0d79e295dbcd4f97a4d10d68b4b9a13b : SUCCESS in 1h 35m 32s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/c781b6b18a27415995f5caa6c4df8dcf : SUCCESS in 1h 41m 11s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cd503696465ec5e5ec4083ba6dcc68cca97fdc65","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-02-14 15:21:44.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)\n\nLooks great, just one question:","accounts_in_message":[],"_revision_number":5},{"id":"f1c43477e99adee239514571a12ede32849354fb","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-02-14 17:07:34.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"38b5f56744b3493a6b080d93f387f0ab7b149e0b","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-02-14 17:08:03.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"08de2d32f5daebf167b0e1d07293ae547c8bd9be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 18:58:29.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d027b8ba40524387bb9577cfc7df2eb9 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4316888e42f14c059c9fd44b92c1f759 : SUCCESS in 5m 10s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/ce1477c0afd74a979243341a61ed7a49 : SUCCESS in 1h 19m 04s\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/d641a4bcb78e44fe81c9dea64d99dbb9 : SUCCESS in 1h 37m 53s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/7af66f5dd87d434e94b3628c34338c6c : SUCCESS in 1h 25m 14s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/cf6e94848e7f45f78a1de8dd6c68c56e : SUCCESS in 1h 18m 23s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/8094c9492b20425db0656b35bfc04a43 : SUCCESS in 1h 16m 48s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/fdad7f13dd21470fbd06771407eeaf8f : SUCCESS in 1h 47m 50s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/26bd9fc09ae642c8818e5cc3718b2f03 : SUCCESS in 1h 40m 31s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4bda74045506ab0ac3a35e6ab3d0e042bc1f90b5","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-02-17 08:37:55.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"4f35e030f273d114776b6ebab950daaa4767f787","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-02-17 12:08:17.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"0ffb0daebc2d1564fd5180ff29ab66fa519e1065","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-02-17 12:08:22.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"e5c482eca257a760cfb2275916de8e69214880d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 12:08:29.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"12d9d1696941c67d4bc41b3be4638c865fa9bc38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 13:48:33.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d608fabb27746fa87bc38ec2d5a85b1 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ab0e686360249f998c4861e95b89cd6 : SUCCESS in 4m 51s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/0c5fc51f34224ecbb884c04a62fd9962 : SUCCESS in 1h 26m 25s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/ce5d6fba5bf44154a286943d4631f2cb : SUCCESS in 1h 25m 56s\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/067207ff01f64f3c8043de8b56231304 : SUCCESS in 1h 19m 59s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/eee365a06f5846c38c7426b155abbd7a : POST_FAILURE in 1h 34m 18s","accounts_in_message":[],"_revision_number":6},{"id":"8b7484b0fe1bebfe1863fdfdca08bbea9627f541","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-02-17 14:25:42.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"1ac67b0f66b1e08d316609d9e888368a03a37ba1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 16:35:14.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18b9bc68d6954cb68fa46ffe43d2ea56 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/289bae57cf3345cfb3592bc5f694f627 : SUCCESS in 6m 52s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/16f1703764234e21b56f8bd198938ffa : SUCCESS in 1h 18m 15s\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/0104375f9c5f42b781f6718f2f168231 : FAILURE in 2h 02m 43s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/731b8bd760404dc9ad32b22337f18ae8 : SUCCESS in 1h 22m 12s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/2e61d4f09d4c408bae1585218a961038 : SUCCESS in 1h 25m 45s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/f12e96e00ed1440cb255750fa4e0e34e : SUCCESS in 1h 49m 12s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/70ac49d30c7548099fcbaa93afd75d90 : SUCCESS in 1h 35m 54s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/22395ab7b7554b339b5fa1d6ad65cc14 : SUCCESS in 1h 37m 24s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e051494d1fe133999f682c5c115e846afba0acf1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 16:35:26.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"34d0ea2f12b4b9b4c163977c0ed18177ec033561","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 18:14:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"30558ade125b85d3dcc2dd02465bb1feae476b20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 18:14:25.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2240979f6b0349c2918c483ba92c862b : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/771f65e4f4f74a04bba6789d0e362bc8 : SUCCESS in 4m 47s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/42303a462853400990caf1503413bc37 : SUCCESS in 1h 28m 19s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/94e3cccacc9045fea0b351eb579d7b28 : SUCCESS in 1h 33m 25s\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/38e9007d409c47b6b1d778a497a7372b : SUCCESS in 1h 19m 54s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/fc5730df492c4f488c351b00f1baf9cd : SUCCESS in 1h 24m 39s","accounts_in_message":[],"_revision_number":6},{"id":"b9170d81eee43545f976147173a836740faaddb1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 18:16:02.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3200fe621874b6c816bd469b1dee1e2 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"681ff7a0cc2295a9d71923ac7a7da999c959a0fc","revisions":{"7bbd2480a6b6f4cd2fca5d9404812ff65d38e323":{"kind":"REWORK","_number":1,"created":"2020-02-11 10:56:07.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/03/707103/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/03/707103/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/03/707103/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/03/707103/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/03/707103/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/03/707103/1"}}},"commit":{"parents":[{"commit":"c440c298c4bbe7f179cc0e643a2f9f8b1d9a7f6d","subject":"Fix the decorator argument for failing test.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c440c298c4bbe7f179cc0e643a2f9f8b1d9a7f6d"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-02-11 10:55:49.000000000","tz":60},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-02-11 10:55:49.000000000","tz":60},"subject":"WIP: IPv6 support","message":"WIP: IPv6 support\n\nChange-Id: Ic2be3cf93bd9d114af907d26198e8011281bfabf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/7bbd2480a6b6f4cd2fca5d9404812ff65d38e323"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/7bbd2480a6b6f4cd2fca5d9404812ff65d38e323"}]},"branch":"refs/heads/master"},"c6458344be66aaccf44a59956f9b70403cedd354":{"kind":"REWORK","_number":2,"created":"2020-02-11 15:22:18.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/03/707103/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/03/707103/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/03/707103/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/03/707103/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/03/707103/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/03/707103/2"}}},"commit":{"parents":[{"commit":"c440c298c4bbe7f179cc0e643a2f9f8b1d9a7f6d","subject":"Fix the decorator argument for failing test.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c440c298c4bbe7f179cc0e643a2f9f8b1d9a7f6d"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-02-11 10:55:49.000000000","tz":60},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-02-11 15:22:08.000000000","tz":60},"subject":"WIP: IPv6 support","message":"WIP: IPv6 support\n\nImplements: blueprint kuryr-ipv6-support\nChange-Id: Ic2be3cf93bd9d114af907d26198e8011281bfabf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c6458344be66aaccf44a59956f9b70403cedd354"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c6458344be66aaccf44a59956f9b70403cedd354"}]},"branch":"refs/heads/master"},"ecbd4d7fa545ef954b6efc9df14472954480f3a0":{"kind":"REWORK","_number":3,"created":"2020-02-11 15:40:46.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/03/707103/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/03/707103/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/03/707103/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/03/707103/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/03/707103/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/03/707103/3"}}},"commit":{"parents":[{"commit":"c440c298c4bbe7f179cc0e643a2f9f8b1d9a7f6d","subject":"Fix the decorator argument for failing test.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c440c298c4bbe7f179cc0e643a2f9f8b1d9a7f6d"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-02-11 10:55:49.000000000","tz":60},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-02-11 15:40:35.000000000","tz":60},"subject":"WIP: IPv6 support","message":"WIP: IPv6 support\n\nImplements: blueprint kuryr-ipv6-support\nChange-Id: Ic2be3cf93bd9d114af907d26198e8011281bfabf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ecbd4d7fa545ef954b6efc9df14472954480f3a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ecbd4d7fa545ef954b6efc9df14472954480f3a0"}]},"branch":"refs/heads/master"},"89a23493feabff7b8b14ea418ae9816ee7220aa1":{"kind":"REWORK","_number":4,"created":"2020-02-11 17:17:03.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/03/707103/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/03/707103/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/03/707103/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/03/707103/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/03/707103/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/03/707103/4"}}},"commit":{"parents":[{"commit":"c440c298c4bbe7f179cc0e643a2f9f8b1d9a7f6d","subject":"Fix the decorator argument for failing test.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c440c298c4bbe7f179cc0e643a2f9f8b1d9a7f6d"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-02-11 10:55:49.000000000","tz":60},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-02-11 17:12:45.000000000","tz":60},"subject":"Basic IPv6 support","message":"Basic IPv6 support\n\nThis commit fixes stuff related to the fact that most apps require IPv6\naddresses to be in brackets. Also now tests that use floating ips are\nskipped if IPv6 is configured. FIPs are not supported on IPv6 and we\nneed to figure out other way to test them.\n\nImplements: blueprint kuryr-ipv6-support\nChange-Id: Ic2be3cf93bd9d114af907d26198e8011281bfabf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/89a23493feabff7b8b14ea418ae9816ee7220aa1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/89a23493feabff7b8b14ea418ae9816ee7220aa1"}]},"branch":"refs/heads/master"},"30872ebdd679c7e9470fc124a07fced6845bff80":{"kind":"REWORK","_number":5,"created":"2020-02-11 17:39:54.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/03/707103/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/03/707103/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/03/707103/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/03/707103/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/03/707103/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/03/707103/5"}}},"commit":{"parents":[{"commit":"c440c298c4bbe7f179cc0e643a2f9f8b1d9a7f6d","subject":"Fix the decorator argument for failing test.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c440c298c4bbe7f179cc0e643a2f9f8b1d9a7f6d"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-02-11 10:55:49.000000000","tz":60},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-02-11 17:39:44.000000000","tz":60},"subject":"Basic IPv6 support","message":"Basic IPv6 support\n\nThis commit fixes stuff related to the fact that most apps require IPv6\naddresses to be in brackets. Also now tests that use floating ips are\nskipped if IPv6 is configured. FIPs are not supported on IPv6 and we\nneed to figure out other way to test them.\n\nImplements: blueprint kuryr-ipv6-support\nChange-Id: Ic2be3cf93bd9d114af907d26198e8011281bfabf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/30872ebdd679c7e9470fc124a07fced6845bff80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/30872ebdd679c7e9470fc124a07fced6845bff80"}]},"branch":"refs/heads/master"},"681ff7a0cc2295a9d71923ac7a7da999c959a0fc":{"kind":"REWORK","_number":6,"created":"2020-02-14 17:07:34.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/03/707103/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/03/707103/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/03/707103/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/03/707103/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/03/707103/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/03/707103/6"}}},"commit":{"parents":[{"commit":"c440c298c4bbe7f179cc0e643a2f9f8b1d9a7f6d","subject":"Fix the decorator argument for failing test.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c440c298c4bbe7f179cc0e643a2f9f8b1d9a7f6d"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-02-11 10:55:49.000000000","tz":60},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-02-14 17:07:26.000000000","tz":60},"subject":"Basic IPv6 support","message":"Basic IPv6 support\n\nThis commit fixes stuff related to the fact that most apps require IPv6\naddresses to be in brackets. Also now tests that use floating ips are\nskipped if IPv6 is configured. FIPs are not supported on IPv6 and we\nneed to figure out other way to test them.\n\nImplements: blueprint kuryr-ipv6-support\nChange-Id: Ic2be3cf93bd9d114af907d26198e8011281bfabf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/681ff7a0cc2295a9d71923ac7a7da999c959a0fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/681ff7a0cc2295a9d71923ac7a7da999c959a0fc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
