)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~712982","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~Ibaf2b70dfb8222dfafb9e0a451674473377074c0","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"retry-exec","hashtags":[],"change_id":"Ibaf2b70dfb8222dfafb9e0a451674473377074c0","subject":"Retry exec_command_in_pod","status":"MERGED","created":"2020-03-13 15:42:31.000000000","updated":"2020-03-18 10:29:51.000000000","submitted":"2020-03-18 10:28:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"712982-1584527290818-74ea99eb","meta_rev_id":"58dd0a86b42779a4950155662a2242206a305929","_number":712982,"virtual_id_number":712982,"owner":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-18 10:28:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":2,"date":"2020-03-18 08:58:31.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-17 08:29:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":2,"date":"2020-03-18 07:23:28.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":1,"date":"2020-03-18 08:58:31.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-15 06:22:09.000000000","updated_by":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"reviewer":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"state":"REVIEWER"},{"updated":"2020-03-17 08:29:08.000000000","updated_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"},{"updated":"2020-03-18 07:23:28.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2020-03-18 08:58:31.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"},{"updated":"2020-03-18 10:28:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"652b6ccc8efd94376930a5d7ef565f6669528a01","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-03-13 15:42:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd64b8076878d71ede100e2d41821d86543f0754","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 17:29:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/100d9cde1d364a469ab4d7e08ef71232 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fec240eb230748c1b3221d0fe403de58 : SUCCESS in 3m 58s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/a98fb76abd184af294813e672925f610 : SUCCESS in 1h 16m 19s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/499d125a17e543c5881b5edd72b35a53 : SUCCESS in 1h 36m 05s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/25816e2eeb4b4354b4dde24e69843e6f : SUCCESS in 1h 30m 26s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/d22e6732dbb447d4ad47651e7acd2299 : FAILURE in 1h 43m 23s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/c12fd72b68b742c18190197c8be950f7 : SUCCESS in 1h 17m 09s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/d2eac96cbe6a432f87dfdfc6d6831a55 : SUCCESS in 1h 33m 24s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/027683abe85642a4bab60745f56fc24f : SUCCESS in 1h 14m 13s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/31711df0c0fc4a178feb86a5b2b616e7 : SUCCESS in 1h 11m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9fc3547e0b868ec0b45fa7430578c5cbdfcfdc79","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-03-16 08:51:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5af7191e4876e3a8d3b0971e62e6c7e0abc22fd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 11:15:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4eded7ad542a45719fc0742e05eb0991 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d79287029628498d8a686d633f92c399 : SUCCESS in 5m 00s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/83b19d2bff494796a4408b91b320b196 : SUCCESS in 1h 22m 35s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/36a2e7655ead469dbacc6e3585702cb8 : SUCCESS in 2h 14m 22s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/09d2c7c54561419ab5acaeca3b02634f : SUCCESS in 1h 26m 50s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/448843771f3b4aa6ae31f72163b872e2 : SUCCESS in 1h 11m 49s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/ac075fd7d69d4d83880c20d5a657d287 : SUCCESS in 1h 36m 35s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/b1d334ec81e24b24922a353e136cb8c2 : SUCCESS in 1h 38m 55s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/70810a6618dd4b35892fda4f9a3f2f89 : SUCCESS in 1h 09m 45s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/6f2a1fd787fd4dedac76577cfe3cd358 : SUCCESS in 1h 18m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b3fddd0afcbc5d1e188d096cc978c25286bd619e","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2020-03-17 06:56:10.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ac67429936001ad90ca52ca45e8ce6458bebc116","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-03-17 08:20:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"110b7ddcdbbfaac58cd649ddb2f457119fdc25f9","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-03-17 08:20:40.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f1b90d8e62e686ea2e9f61e795c9feefe6e8a1e0","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2020-03-17 08:29:08.000000000","message":"Patch Set 2: Code-Review+1\n\nNicely done.","accounts_in_message":[],"_revision_number":2},{"id":"8053dc9f2ef743f3d643dd483a54de8cbeb545d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 10:02:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/787de43959ea44eb8750eea28575c94d : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83fad0051dc241e2a682b1c98916f1dd : SUCCESS in 3m 34s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/7cfef7973def49cfb98b7bdb8ec28259 : SUCCESS in 1h 29m 13s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/f2cb4027239f4321b9ba47327b1054c3 : SUCCESS in 1h 23m 29s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/f8cf00c6f758488fa386a6f7ed59b4dc : SUCCESS in 1h 33m 30s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/9ecdd62562a5416592aede571bd110ae : SUCCESS in 1h 28m 14s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/da20a9f5e880460bb7952d0b73652467 : SUCCESS in 1h 18m 57s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/2788925c37a44cbc80705bc0e123593e : SUCCESS in 1h 36m 06s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/36d7b49624b7445f947b0edec26dd0f9 : SUCCESS in 1h 11m 09s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/c461e8bde7704b48831c7c6a8d824da7 : SUCCESS in 1h 11m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"13013929ba3b496981f0284a9e04f0e600e8eddc","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-03-18 07:23:28.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0f7d03c4f6e23a3ae215b1529350bc97a93e215a","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-03-18 08:58:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"283b1d0083fcff8e4b536e80c3664047b4634195","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 08:58:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e16e43a5eb3edfc51863a168ce5632180fd5c302","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 10:28:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78beb70a00394fb7a4d055e21dfb5337 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bebae113b814c01b334119c9ee153a3 : SUCCESS in 3m 38s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/fa3662f4bf634047b7e367e5fec9ab9e : SUCCESS in 1h 10m 18s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/8e8fad4a2cdb4585aa2572484e6c1ae5 : SUCCESS in 1h 13m 48s\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/c79b8e5cb6bc48d397fdacf583760347 : SUCCESS in 1h 23m 55s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/73959c963fc44d28974c0c5ae8c363d4 : SUCCESS in 1h 23m 19s","accounts_in_message":[],"_revision_number":2},{"id":"e14b17b13b93ce11557f0f8cbdbc3c05b999e6e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 10:28:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"fc1a85951f6cd26810529402cfbff4e5b7b7b6d2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 10:29:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/108206c73e654d35aa8d80a548861e35 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bf2afa0c54128479cfb5c7bc44cedd906f083bbf","revisions":{"1988c7feacb15b4bdc1950ddf9b7e99f87a22640":{"kind":"REWORK","_number":1,"created":"2020-03-13 15:42:31.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/82/712982/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/82/712982/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/712982/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/712982/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/712982/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/712982/1"}}},"commit":{"parents":[{"commit":"6cd840bff2816350b2d4738672341f3d1b9f9318","subject":"Merge \"Check ingress and egress IPBlock rules for NP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/6cd840bff2816350b2d4738672341f3d1b9f9318"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-03-13 15:39:26.000000000","tz":60},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-03-13 15:42:01.000000000","tz":60},"subject":"Retry exec_command_in_pod","message":"Retry exec_command_in_pod\n\nFrom time to time we get some etcd issues on the gate and when that\nhappens Kubernetes API likes to greet us with a 500 error. This commit\nmakes sure that exec_command_in_pod retries requests like that so that in\nsuch case we don\u0027t fail a test.\n\nChange-Id: Ibaf2b70dfb8222dfafb9e0a451674473377074c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/1988c7feacb15b4bdc1950ddf9b7e99f87a22640"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/1988c7feacb15b4bdc1950ddf9b7e99f87a22640"}]},"branch":"refs/heads/master"},"bf2afa0c54128479cfb5c7bc44cedd906f083bbf":{"kind":"REWORK","_number":2,"created":"2020-03-17 08:20:39.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/82/712982/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/82/712982/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/712982/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/712982/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/712982/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/82/712982/2"}}},"commit":{"parents":[{"commit":"6cd840bff2816350b2d4738672341f3d1b9f9318","subject":"Merge \"Check ingress and egress IPBlock rules for NP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/6cd840bff2816350b2d4738672341f3d1b9f9318"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-03-13 15:39:26.000000000","tz":60},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-03-17 08:20:29.000000000","tz":60},"subject":"Retry exec_command_in_pod","message":"Retry exec_command_in_pod\n\nFrom time to time we get some etcd issues on the gate and when that\nhappens Kubernetes API likes to greet us with a 500 error. This commit\nmakes sure that exec_command_in_pod retries requests like that so that in\nsuch case we don\u0027t fail a test.\n\nChange-Id: Ibaf2b70dfb8222dfafb9e0a451674473377074c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/bf2afa0c54128479cfb5c7bc44cedd906f083bbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/bf2afa0c54128479cfb5c7bc44cedd906f083bbf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
