)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~721610","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~I94eecb307f72b50c938369c64cd473b224c6b19a","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"namespace","hashtags":[],"change_id":"I94eecb307f72b50c938369c64cd473b224c6b19a","subject":"Add retry for check ports in test_port_pool","status":"MERGED","created":"2020-04-21 13:19:08.000000000","updated":"2020-05-05 10:22:03.000000000","submitted":"2020-05-05 10:20:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721610-1588674040561-2682b9c6","meta_rev_id":"53503edf3dea058fe7b388c368efb21a19c6db80","_number":721610,"virtual_id_number":721610,"owner":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"date":"2020-05-04 12:01:01.000000000","_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-05 10:20:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2020-05-05 08:11:00.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":1,"date":"2020-05-05 08:10:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-05 06:34:35.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":1,"date":"2020-05-05 08:11:00.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"date":"2020-04-22 06:51:51.000000000","_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"date":"2020-04-22 04:04:33.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-22 04:04:33.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"},{"updated":"2020-04-22 06:51:51.000000000","updated_by":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"reviewer":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"state":"REVIEWER"},{"updated":"2020-05-05 06:34:35.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2020-05-05 08:10:53.000000000","updated_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"},{"updated":"2020-05-05 08:11:00.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2020-05-05 10:20:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9c6365f99bfc33922f3df9b12e3b910dfb54dde7","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-04-21 13:19:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"625297ac140fc3378f1747809b4ad2b026c14fda","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-04-21 13:21:21.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"67ad4ee98e268f1e59223cf65625f59203a4cb4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 16:27:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba6867fd9c914e9fb65a2fbd43df1e5c : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54df7bd1cef446ed8c1098a1d1cbb9d6 : SUCCESS in 5m 21s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/8a4276ca477d47e498b4dc05ab0d1bd0 : SUCCESS in 1h 37m 41s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/11a03467999b40779bc90f12ac57da0f : SUCCESS in 2h 06m 41s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/9b2c1f40816c4f86a66b228ff7bf30f4 : FAILURE in 2h 45m 51s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/766897b8947c4457a149c873626a5697 : SUCCESS in 2h 01m 53s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/9328a7023dfe41e3ba87c7bfb891347d : FAILURE in 2h 27m 14s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/83d6d9bf9cfe46bc8c8a76e133ddcf55 : SUCCESS in 2h 59m 21s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/cc13155a480f4f10b18fef62b54748cd : SUCCESS in 1h 34m 02s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/37b9f6fac6334e66a67caa934abf4c8e : SUCCESS in 1h 32m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0b6e4ebff735fdb038556c4ec3c9fddf06bbc66d","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 16:33:09.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/18422f86d5994d3e924745283694b528 : SUCCESS in 1h 37m 06s\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/12e471643c4f4dadb170c49b4feb8af5 : FAILURE in 2h 36m 58s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/3227ddd032024c238bb1239b87aab951 : FAILURE in 1h 52m 56s\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/aa5c010ce8fc4f398a95d65618f64899 : TIMED_OUT in 3h 08m 08s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/565b5a81d8ea408eb52b25284b8545c4 : TIMED_OUT in 3h 05m 44s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/eb7cda073d4446f3b3e1a1fd0c6378f7 : FAILURE in 2h 18m 57s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/8b7497c81ca647509ff09e6840ca1f79 : FAILURE in 2h 14m 54s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/8f1b18226e754127849c7409e909618f : TIMED_OUT in 2h 15m 19s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/c028ac62f5f5443298bb9a8c9bbb7a33 : FAILURE in 27m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"102d8b7d48ea7e67e15c44eb5027d251a60f7971","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-04-22 04:03:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"98446afd7690ceb882e233a0cc8cc0364eafec13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 06:28:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bae4c1502f546c8b22e1a8fe2f7da23 : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b176836bbf334103b50c3373646e6dcd : SUCCESS in 3m 11s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/e1264813ba114053943da56125b29691 : SUCCESS in 1h 15m 51s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/a366a2df8767467d8a9463fb6aaf553b : SUCCESS in 1h 48m 20s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/08196b3a85204b3eaf47f6a0d78adbf0 : SUCCESS in 1h 15m 47s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/d1402886286048939085318acb47b9c0 : SUCCESS in 1h 39m 40s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/dd08382940c2496b81ba5e41c7a2b7a9 : SUCCESS in 1h 52m 16s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/03a96117ce444d5abecef1e8c26a742e : FAILURE in 2h 19m 48s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/f3357541a99044e18fbffd32266d8a0b : SUCCESS in 1h 30m 12s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/b8ae0f99676e4ac5af8c00044737fbf1 : SUCCESS in 1h 33m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8d213c992078286929c97521ed791dc628b87b4a","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2020-04-22 06:51:51.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"71c64601669bcb18cc3e1fedb88ee0aea96e350b","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 10:02:55.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/c6d339a0a70348678a620af0dff48b0b : SUCCESS in 1h 41m 21s\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/7822271c8ab346bc9b9c99c8673140c4 : FAILURE in 2h 38m 58s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/e3a73e99f6154402ac578a7a8a872635 : FAILURE in 2h 00m 07s\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/55e699a7488f476382a805807af60fc9 : TIMED_OUT in 3h 04m 07s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/d21c6cf31f2a48b2a44077aecac029fd : SUCCESS in 1h 23m 49s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/7fc46957916649a590200a9acca362ad : FAILURE in 2h 31m 19s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/d151f38d69734cd38b4fe9f726e1fe0e : FAILURE in 2h 14m 51s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/dd947a3058e9414188103b8b97ef47c0 : TIMED_OUT in 2h 14m 00s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/5505909af94440a49bae4a426afaa7b5 : FAILURE in 22m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b0515db4b1ed776deef8932afe8efdf8983fda65","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-04-23 07:03:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bf92c67545c5baa710076ffc7494ac42cd118289","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-04-24 07:18:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e4cd0afa2328dc20514b08e5dedfb1c19f67ff33","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-04-24 07:18:27.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"9ba762c7d22afcd1c64e005c66dc34ea43149cbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 10:12:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d221dc93a3f4253bc147dd4aa0f9dd8 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faffb2d5f948445293ecac3f467a2c39 : SUCCESS in 4m 03s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/699cbdddff124158aec90fbc127fd473 : SUCCESS in 1h 12m 10s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/99af8c0985b345739d3e54d8e0b00399 : FAILURE in 1h 18m 07s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/deb1fe837f98498db4908be9883553db : FAILURE in 1h 09m 07s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/294a6e5da27647ef984bd48236c7dacd : FAILURE in 1h 23m 30s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/20f3e5eab2c84401a436de9f22f4e1d8 : FAILURE in 2h 50m 45s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/0303b11351554678a756b012e636b922 : FAILURE in 1h 49m 48s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/10e1d535f42243f8a6c7220a040d82b9 : FAILURE in 1h 19m 46s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/afd341c8e2f24570b4e8d156fc4dfd86 : FAILURE in 1h 19m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"57248e46870bc1e6c34efa9204bd961b56d44f3b","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 10:28:27.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/b787d92d153e4e3ba13fa7238755c57c : FAILURE in 1h 34m 28s\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/10c46edaced7408abe106e7061a99a1a : TIMED_OUT in 3h 05m 32s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/262b0c95d5654a19b48d6d2f61e05d1b : TIMED_OUT in 3h 04m 17s\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/d67b5b4cd6434265ab5606625ff81d80 : TIMED_OUT in 3h 03m 42s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/c745623fe99d4b37bc27189d8894acb8 : SUCCESS in 1h 12m 18s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/53002d03ba864839abea347c03e04b1c : FAILURE in 2h 18m 08s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/3763ace0a8aa408e969835f0dbeeffbb : FAILURE in 2h 28m 29s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/9743f4dd6a2c40428d58be3d10074af8 : TIMED_OUT in 2h 14m 34s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/6e7b220a5b414e478109be3758d3b38a : FAILURE in 25m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3de9fb18a9b60099d356b65f0540e2b66dbbdd11","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-04-27 07:16:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7f93609272f0ee9d2c43657c8b603efae815c8fe","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-04-27 15:08:25.000000000","message":"Patch Set 1:\n\ncheck experimental\n\nId36a00b94f9df837c49ced7155f8117fe9625fc2 is merged, this should do better in pools-namespace tests now.","accounts_in_message":[],"_revision_number":1},{"id":"d1848047e65903d090d68cbcdd8baee56b423adb","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-04-27 15:08:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"26c63e91c322b0298bcb09f45e898bc7554dfe60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 17:43:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2525aa06ab7141e9b4398564441b9106 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69e688f8fa95474eab2c421d6ceed589 : SUCCESS in 3m 25s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/c4a98cb1f25d414b8358daedd8a469ec : SUCCESS in 1h 23m 45s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/47d1051accdb4300ac0ea80d01543be0 : FAILURE in 1h 02m 00s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/85334275d94d4583921ebe182c3ef3f0 : SUCCESS in 1h 45m 01s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/54657197c3db485d999f2d111c9deff7 : FAILURE in 55m 00s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/3dd36307127747a8ae1989ad02b35770 : SUCCESS in 2h 08m 40s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/98abd62e616b45129db1f90109b0e856 : SUCCESS in 2h 18m 53s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/77f98b82af00458f9d321baeb228284e : SUCCESS in 1h 41m 43s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/a3815d9d2bcd4483a430a941bf481e84 : FAILURE in 1h 37m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9f3421ca776d2ea7ba3fe7d37cb61a22ea6d1f9c","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-04-28 07:31:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c096d1f1ad710c7b2ee4282ad5fbf5d5e1100b1f","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-04-28 07:31:58.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"007ab643d2a7ea8c49474101559b396e123eabde","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 09:14:43.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/da40cb3b9e374e7c912560464ca0b271 : EXCEPTION\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/82573f8f05974549b286ce626ea6d637 : EXCEPTION (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/65f2cec156154d7a878c29e6bae9488d : EXCEPTION\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/a68bb64409f64f6f882c87e410463aca : EXCEPTION (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/46a8c987df2c4a0e8c72be74578544d4 : FAILURE in 54m 06s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/4a2a7c2703454e2faf63a750de5dd4ac : EXCEPTION (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/47098f40835d44dd9db09d526427ecc7 : EXCEPTION (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/8a7d2ba4dd6b4bd6b87d084cb3114e57 : EXCEPTION (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/172ab1b95057417ebba90aa371200d70 : FAILURE in 33m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cc3b13232457aa65b31391343f21e9ac09bf1b42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 09:20:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d63fdf393c31468da0602d64376faba6 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e632efcc46bd4ac8a46363d540bc517f : SUCCESS in 6m 06s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/e1d9f8c1a27a4a42a0aa4961191af217 : SUCCESS in 1h 10m 01s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/dd4302e713b54865964fde0fcb1d46bc : EXCEPTION\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/7be746ab766f43af9d516c1efc849a8f : EXCEPTION (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/0326bb0842c54bd1b709b1261f173508 : EXCEPTION\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/bf512f88f9e241f3a9662f41424ab5f4 : EXCEPTION\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/3e803e62748346848f4355f182680055 : EXCEPTION (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/44cc78f4836c4c8cbe7cbf8c96fae3e8 : EXCEPTION (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/aa89d3307a11437497e0b15cde56ad3a : EXCEPTION (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"50f77ed5b1633baa41f56949d9db62e40c92be7b","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-04-29 07:35:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4e94fc6b9356105e9a9d9d36972391a0cffcf163","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 10:05:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d71c48d096274f069ff2fdfc13025b5c : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/477da6d14bf54cbcb4f554dbddb07435 : SUCCESS in 5m 25s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/f607a4675a794c59a06000d13709e529 : SUCCESS in 1h 06m 50s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/97ed961fe0884d228ae4627aaadc7170 : SUCCESS in 1h 47m 22s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/092e689c6133491d99fe974583ede854 : SUCCESS in 1h 19m 36s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/6ed1fa1721aa44f6b29b888ef4b0f6d0 : FAILURE in 1h 29m 15s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/1c3fbe2fc6914740a8b13ed469f581df : SUCCESS in 1h 36m 26s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/6e22688b96ad4857b4db1ddd39212fea : SUCCESS in 2h 18m 41s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/8131c044f741412a83887ca75821d96c : SUCCESS in 1h 32m 33s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/121afb7b4d494637a40833985b955771 : FAILURE in 1h 26m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a68df43496aaebbab6d92e88f20898aa7948aa75","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-04-30 06:23:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"518347a563e21c66f0ddaef1c35f573c1f76242c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 08:57:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12887c88bfbc4f40b69d7bff568e3362 : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/900d6e438b5c4333baa5aa512c333ca0 : SUCCESS in 5m 59s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/55a4d0bdf1854aa7abe4bc04e7f3feca : SUCCESS in 1h 06m 38s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/d05749300ead4c618b183f8b3f985a0b : SUCCESS in 1h 58m 36s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/93058f605f6c432b945be2d798476574 : SUCCESS in 1h 29m 32s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/12c6b977830141c19473ab38af5fb6a2 : FAILURE in 1h 28m 35s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/c888a778992e4080b592d764fd2c2a13 : SUCCESS in 1h 59m 40s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/25f1f18b73d146f48491ae7fa835e0e4 : SUCCESS in 2h 25m 41s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/2700e671b0eb4788a00ce99f05224e30 : SUCCESS in 1h 22m 28s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/dfd4f4ee7b304d048479dabddc060787 : FAILURE in 1h 26m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"674f3238b2b22ab3571efd1805e4cd19dc8c5778","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-05-03 04:08:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"73084ad97ac551e9354ad0d7f704d0eb45a3f7a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 06:22:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4071afdbc7844509b1b6cdf77f2f6c4a : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67f25b7e75a6406da9a2921791af70c2 : SUCCESS in 3m 33s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/969d7e3007ab49ed8102337ccdee223b : SUCCESS in 1h 35m 13s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/84ba92e3c07c4eac895d06fa2514db0d : SUCCESS in 1h 31m 29s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/756c672c2bde43ee9a0977f3c9ddbc13 : SUCCESS in 1h 11m 17s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/f353f05f169548b99feba7bdaf63d14b : SUCCESS in 1h 32m 28s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/7df2f5169a5147b4af044ae03aab0cce : SUCCESS in 1h 50m 05s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/9b132a8e24b34cc3b2adb484450a5c67 : SUCCESS in 2h 12m 36s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/adb08f6fd37342cb86aa0979bbef2f29 : SUCCESS in 1h 22m 47s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/c24e0613652a49099a6539700efbdfc1 : SUCCESS in 1h 23m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eeaccff21209b7a3dad10c893bdfa58d5320a1f1","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-05-03 07:12:48.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"7545ed7b2c5e4bcb70ff33f3513486075cff23a7","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 10:19:43.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/11ea5e9b65354dd292824f6db9eff2ab : SUCCESS in 1h 36m 46s\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/0b7f93890e90487f91525319cc634fea : FAILURE in 3h 05m 40s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/169cf7b8e4124a2db1b1a1f69c9a7ba1 : FAILURE in 2h 52m 02s\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/db0fd5bd8bb5485a8087da12f655b5e4 : TIMED_OUT in 3h 05m 15s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/e46288599296467682bfd88111f9a46b : SUCCESS in 2h 01m 33s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/ee0dae93bfe44f609f93e7b48c4b9d35 : FAILURE in 2h 19m 57s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/5fef310586a34eebafccc1258945885f : FAILURE in 2h 27m 53s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/555a48d2f93e457ea7d210dc9988682a : TIMED_OUT in 2h 13m 27s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/def62a01be074310a4a9c72a3c91d2ab : FAILURE in 31m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6d9c2525a755a20f04257bb7a11793c390234aa6","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-05-04 07:56:46.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"fea67889d07070f15f0caa26b013dda2e26effd5","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 10:22:47.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/7274f14537d447c0b5da483dacb86aa4 : FAILURE in 7m 21s\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/4bce8716dd2349019dfb2472173a486d : FAILURE in 8m 53s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/7f05bf3ac38d42449c715bb37f9764c6 : FAILURE in 7m 40s\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/836e721f580c453d82070bc7f69db2b8 : FAILURE in 7m 07s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/9e5a07a90c4e490da15a948584c0e421 : FAILURE in 9m 00s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/b72e15b3b45f48f7b0d0f8a76b7b8512 : FAILURE in 7m 41s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/569fa15c9497498484ebefe5e6608f74 : FAILURE in 7m 23s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/a79cce59adfc445d92d1a6ca807caba7 : TIMED_OUT in 2h 13m 39s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/7ab37765f31e4e83b157f27da00fa9aa : FAILURE in 7m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7a0569d10631fb47f19db25b1a2c894419c14964","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-05-04 12:01:01.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"224dccdc10af711aa2e9c8decfb8373910929477","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 15:11:03.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/360ff03eba4e482f9b37bbeed363d662 : SUCCESS in 1h 55m 10s\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/fb46bcebdaaa463fb205c34c3f6b4589 : FAILURE in 2h 42m 50s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/3c6fffebc6424d10b8175547f293f725 : FAILURE in 2h 16m 26s\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/3a8a6f6b87494ce4bb0633e68fdad1a6 : TIMED_OUT in 3h 04m 58s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/3803fa3cf6d541d18a03af151d1b83e0 : SUCCESS in 1h 30m 38s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/ac7c2757338645a1a31d14237b7adf5e : FAILURE in 2h 08m 43s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/da2693c9a3164d2c9d7d16253728c6df : FAILURE in 2h 20m 11s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/70ddc21973b9460e97933d616ab15c4a : TIMED_OUT in 2h 15m 28s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/e6a0ce8ef6624e74a15d5277c6946641 : FAILURE in 25m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9b27561bb81757ac68cf563d58ac0752c3f5548c","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-05-05 06:34:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"779821f484dae57c864b83288cd45ed4d631e9be","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2020-05-05 08:10:53.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0dbc4fee47818dcc13a64a19e0827aeeb84fea3f","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-05-05 08:11:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5f45d9bf03e92323901cd4f9f006382b23907436","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 08:11:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"57f3e70164c42bd0be2757bbbea076e9ce538a14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 10:20:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97ba7e7ccc324b5ebbdf3bb6f4564dd4 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca5d58b30075437b8ce7c926f2ba514c : SUCCESS in 5m 21s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/d511468c73464df0a62d13c0cadadd9d : SUCCESS in 1h 28m 59s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/e79382c5c26447ccac6cf2a4449161ad : SUCCESS in 1h 36m 40s\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/e5a546aae37e44c3bac1d13a3782b2e4 : SUCCESS in 2h 08m 06s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/9572f1e559fe47acbd33d9d7e5f698b1 : SUCCESS in 2h 07m 33s","accounts_in_message":[],"_revision_number":1},{"id":"5f569dedd20d5d56343528c19c610a0c6c57cef1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 10:20:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2cf98d842de8ebd734f3e845a859e8656156f0e5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 10:22:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88ac3ec81ec641a486ec33b06e6e272f : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"50dd6acbab7b2aea3cec1f1828905ef6f832dd99","revisions":{"50dd6acbab7b2aea3cec1f1828905ef6f832dd99":{"kind":"REWORK","_number":1,"created":"2020-04-21 13:19:08.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/10/721610/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/10/721610/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/10/721610/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/10/721610/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/10/721610/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/10/721610/1"}}},"commit":{"parents":[{"commit":"9db2bf03ed1c9c0a7049bf662e91fa85a48364ac","subject":"Merge \"[ussuri][goal] Drop python 2.7 support and testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/9db2bf03ed1c9c0a7049bf662e91fa85a48364ac"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-04-21 12:39:02.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-04-21 13:18:33.000000000","tz":180},"subject":"Add retry for check ports in test_port_pool","message":"Add retry for check ports in test_port_pool\n\nWe want to make sure we give enough time for the ports to be created\n\nChange-Id: I94eecb307f72b50c938369c64cd473b224c6b19a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/50dd6acbab7b2aea3cec1f1828905ef6f832dd99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/50dd6acbab7b2aea3cec1f1828905ef6f832dd99"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
