)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~725694","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~Iffc01ce846db232ee4f1cd4c3603bc17dd8e7cfd","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"stable-testing-tempest-plugins","hashtags":[],"change_id":"Iffc01ce846db232ee4f1cd4c3603bc17dd8e7cfd","subject":"Add stable branch jobs on the plugins master gate","status":"MERGED","created":"2020-05-05 22:16:07.000000000","updated":"2020-05-12 09:14:04.000000000","submitted":"2020-05-12 09:10:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725694-1589274628661-d1a6a7d1","meta_rev_id":"fd19cc7fd3140cb73346d5b3390cb7bddda48bf9","_number":725694,"virtual_id_number":725694,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-12 09:10:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2020-05-08 14:31:08.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-11 07:26:28.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-11 07:26:28.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-08 14:31:08.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2020-05-11 07:26:28.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2020-05-12 09:10:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ec6d4ee3bc5497a9a441ae692535d23b0ac4a62a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-05 22:16:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb9308ea7f945cab6112f5a2832fba7c95c99bfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 00:30:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd9c62e8feb24bd7a2a9ec43254dd4aa : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da66729ee543484a9a1d0771a9663858 : SUCCESS in 3m 40s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/75d53c7c9bda4b2799866cbf2359cde7 : SUCCESS in 1h 05m 30s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/5157fd07213f4607a49f609ac465cca5 : SUCCESS in 1h 49m 26s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/fd5f06a886ff4e21818c86f98888dd36 : SUCCESS in 1h 16m 08s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/e975941ce5ff489ab04aabe50c2a4c86 : FAILURE in 1h 31m 18s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/6d4e3177efdc45c7bbb4a3a641d39277 : SUCCESS in 1h 59m 15s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/3ba0b4f0066e436daa68245186cfab1b : SUCCESS in 2h 09m 28s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/14f577c42f7e4471923fcb69903b9610 : SUCCESS in 1h 26m 01s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/c919b7946f114904b926518d0d6e6f68 : FAILURE in 1h 32m 18s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/e1e0315c7a074542b3d709b085f1d060 : SUCCESS in 1h 11m 24s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/97fa6210fabe422f92aab2d79a320107 : SUCCESS in 1h 27m 14s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/b1f32e4e7dda443d9932d1c778e56349 : FAILURE in 47m 20s","accounts_in_message":[],"_revision_number":1},{"id":"a8f01215b5b40732783a14cccb1f41b5f3f1dd45","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-05-06 07:19:04.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c12b5c24efeebae93618f9243224b2b88d0c82c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 09:39:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bac9c7ae94174e628a95a4dbd9d530e6 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/027ac219d55c45fb8e13fe6a870582a6 : SUCCESS in 3m 43s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/b808acdd13c44b1296ee957683fbc9c1 : SUCCESS in 1h 20m 54s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/b70a1022bbc746a1aee4f7d4b7fb1c77 : SUCCESS in 2h 04m 40s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/1b1a3a69665b4387a329004f2dd43f0f : SUCCESS in 1h 21m 03s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/92a34b2b9d3d4b45b57ff60d9dd8d5c7 : FAILURE in 1h 32m 20s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/9fdcbe33faa84623ada33a10595e93d6 : SUCCESS in 1h 54m 24s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/28f1a2ebb4524d01a76d9f793e285edf : SUCCESS in 2h 12m 53s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/d3d4a33978854e69b0d931a199a74394 : SUCCESS in 1h 33m 39s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/6a689997ac03483d9700f2fb817c1b83 : FAILURE in 1h 34m 58s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/82e3b9867f89454ab1a75924d4b04c12 : SUCCESS in 1h 12m 50s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/407f8700170b4911897de051cdad0b62 : FAILURE in 1h 50m 08s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/2cdcdee0dca44608bd4b4c8e0f2e5895 : FAILURE in 59m 23s","accounts_in_message":[],"_revision_number":1},{"id":"9b3ccd8e3b5b38fb8bb207c07152f202505d3abc","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-05-07 14:11:49.000000000","message":"Patch Set 1: Code-Review-1\n\nIt seems like in stein it still tries to install on Python 2. Is that expected? It won\u0027t work unless we downgrade hacking to version supporting Python 2, which we should not really need.","accounts_in_message":[],"_revision_number":1},{"id":"c85f178ef48973ac72a2ae923a3d24cfbd7fe69d","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-05-07 14:29:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"da8210120bd73463ff57b2fa6f85bcc0740b023e","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-05-07 14:35:52.000000000","message":"Patch Set 2:\n\nThis will most likely fail on OVN gates, as they need to wait for https://review.opendev.org/#/c/726161/, but this should work now. It\u0027s more of a hack though, that I downgraded hacking.","accounts_in_message":[],"_revision_number":2},{"id":"f8c9484f6a6c8e15269b1d3cd87da2a0f41a9367","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 17:21:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d5cb2a6c8c9445ba87f7902ed661815f : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/596e6370ea8e427796447446dcca0753 : SUCCESS in 8m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1127210cdf8d4c1c8b91dec00e283782 : SUCCESS in 7m 52s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/d77a7e174ce84d6491367359ed0cfb9a : SUCCESS in 1h 32m 12s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/c363f3429e7f43e58bbf52b16d087395 : SUCCESS in 1h 52m 02s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/2c273e223a7741de851f2f3b42d8767b : SUCCESS in 1h 37m 43s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/934a4cd15e284d2a97fdb7be418e621e : FAILURE in 1h 59m 31s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/588b0ffe03644c15a4ebce02823238bd : FAILURE in 2h 09m 21s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/f3c64d31b7a1487b8ed11081e432d607 : SUCCESS in 2h 45m 32s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/ce20687f832344258cae3b877c68ae83 : SUCCESS in 1h 24m 41s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/9fda70adaf6345398070053c217c955f : FAILURE in 1h 37m 50s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/c6e0f12b66a24d29a0d1c5a6816e32d0 : SUCCESS in 1h 20m 37s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/c59e8dd6db34400194cfd754cfffc4ec : SUCCESS in 1h 29m 02s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/f6b80712efda48fd8c8d9ffe392d8e10 : SUCCESS in 1h 26m 14s","accounts_in_message":[],"_revision_number":2},{"id":"41f08d2c43f91dbe4a14890f1845eadd01e48284","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-07 20:44:06.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"4267a6c7d7408975b6e7b1689fb7603717530855","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 23:35:04.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/508e5577417d4e159db126db419adc1c : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9208e2fd63ee47cf99ad97692fcc7629 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2cb68c7b17f4915a70069286a9b4215 : SUCCESS in 4m 02s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/813b7bc3de7b47788cd882d60537eb40 : FAILURE in 1h 16m 10s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/4d84a3d6e99b4e439942e56ee9fc02af : SUCCESS in 1h 50m 34s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/1dd2dd9341c74937816131fcbd47fa53 : SUCCESS in 1h 42m 18s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/d4b917e07d3c4487a6989e7986d542f5 : FAILURE in 1h 36m 28s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/8adec45de9d4468e94e187d1b1f1e438 : SUCCESS in 1h 57m 00s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/5a4b97a524a74fe3833c28645fc25a07 : SUCCESS in 2h 49m 11s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/f2f83e18f1b9416cb08a19fc1a58c3a9 : SUCCESS in 2h 15m 46s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/95be0eb29e8b4f02b5397096f24b64dc : FAILURE in 2h 19m 53s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/0dd9e53c13024f9498e6a76c64075498 : SUCCESS in 1h 27m 30s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/405d179a602a4ea4bd00d62fe6b8c54a : FAILURE in 2h 13m 21s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/2ac3fac5aa0d48948c35abe4deb90315 : FAILURE in 2h 22m 35s","accounts_in_message":[],"_revision_number":3},{"id":"7f952dcd2ffe2e240f6d603fabfdb4ac0551785c","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-05-08 11:51:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e83a72c80c9b6397a8aceaa9ea7563ea4a4812ee","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-05-08 11:52:11.000000000","message":"Patch Set 4: -Code-Review\n\nAlright, let\u0027s see how it\u0027ll play now.","accounts_in_message":[],"_revision_number":4},{"id":"339639cc84d1543f427a910a6e8b2ef39346562b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 14:25:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b4c040a70ff4ac99030af5ee49448ba : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f10970735db749c59af5a8dcd4ed686f : SUCCESS in 6m 02s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/b2e636918cf646e49982c760f6329903 : SUCCESS in 1h 11m 22s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/9b2fcd88f74843919d6d6ce3886b335a : SUCCESS in 1h 54m 19s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/2a1a0f14aae545c7a2750db1dc93f85c : SUCCESS in 1h 22m 23s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/e4ee5edb6bde493ca68ae3bdababe3d2 : SUCCESS in 1h 35m 57s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/92a390e8b79a4bfa89221d3f13b7afd4 : SUCCESS in 1h 47m 24s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/9e9d469840284ad9b736fc3e84cd315a : SUCCESS in 2h 25m 48s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/d76851352bca4923ba8f95bbf2bd9c7a : SUCCESS in 1h 37m 40s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/7206d9845ab341a2a0a4aa67a785c8ad : SUCCESS in 1h 38m 39s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/0f6fc16391bb4d94a9c1f23e3b6d35dc : SUCCESS in 1h 26m 27s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/941c3a3ad0fb4e568b92982997468c9a : SUCCESS in 1h 27m 02s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/7360e741308f4c42922081516a6afbc2 : SUCCESS in 2h 00m 53s","accounts_in_message":[],"_revision_number":4},{"id":"118baf198130454c047b261a23592f3a2a9fa5a9","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-05-08 14:31:08.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"b73eeeaab6079bb789db4a1301b78c1a042fbd6f","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-05-11 07:26:28.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b9dd4fdcc78f5ea4dcc155b088ec26f34b4faaaf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 07:06:33.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ba0728c5aa0fe9f02fa65ec900cf307b025a86cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 09:10:28.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d416bfc2b0f240f391c2c320a6b7a5e2 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f74a447ba9dc409fb9f9c53633dbd598 : SUCCESS in 4m 12s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/f6f9c249f46c4aafbdbef4930c8a81e1 : SUCCESS in 1h 07m 37s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/0889fff019d842e2a2ba5e38158ba5a5 : SUCCESS in 1h 36m 30s\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/ae5536e57f1a4be5b1f99aca5e8b7d2b : SUCCESS in 1h 58m 46s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/d8bb4363957d4f309bd7bdaedb5ebe60 : SUCCESS in 1h 46m 31s","accounts_in_message":[],"_revision_number":4},{"id":"3102cb2e6256bcc581d5b7575bac1640bb57eb0f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 09:10:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"3c04cf201c29e4c2b5cf06fdd71ead1acc599994","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 09:14:04.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7419a9249fe049bdb7c9405009e00d0b : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d4919d80cf29291e2e58d20b36f0bd9f9d783558","revisions":{"faf7f155945364a8d2f06f81da1ef0f84c54ddd7":{"kind":"REWORK","_number":1,"created":"2020-05-05 22:16:07.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/94/725694/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/94/725694/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/94/725694/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/94/725694/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/94/725694/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/94/725694/1"}}},"commit":{"parents":[{"commit":"1c6df3b03b4114e6edc62e4de4906bb419e51495","subject":"Merge \"Update docs building, py27 cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/1c6df3b03b4114e6edc62e4de4906bb419e51495"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-05 21:57:21.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-05 22:14:44.000000000","tz":-300},"subject":"Add stable branch jobs on the plugins master gate","message":"Add stable branch jobs on the plugins master gate\n\nSupported stable branches in maintenance state\n(stein, train, ussuri) use the Tempest and plugins master\nversion for their testing. To make sure master version (new\ntests or any changes) is compatible with stable branch testing,\nwe need to add the stable branches job on the plugins master gate.\nTempest and many other plugins are already testing the stable\nbranches in same way[1].\n\nUssuri onwards, Tempest or plugins are py3 only, so we are\nrunning only python3 version of stable jobs\n\nTempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\n [1] https://opendev.org/openstack/tempest/src/commit/e8f1876aa6772077f85f380677b30251c2454505/.zuul.yaml#L646\n\nChange-Id: Iffc01ce846db232ee4f1cd4c3603bc17dd8e7cfd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/faf7f155945364a8d2f06f81da1ef0f84c54ddd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/faf7f155945364a8d2f06f81da1ef0f84c54ddd7"}]},"branch":"refs/heads/master"},"1e5aec8c4181f40335acd63b815dddf4aff92f7e":{"kind":"REWORK","_number":2,"created":"2020-05-07 14:29:57.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/94/725694/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/94/725694/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/94/725694/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/94/725694/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/94/725694/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/94/725694/2"}}},"commit":{"parents":[{"commit":"1c6df3b03b4114e6edc62e4de4906bb419e51495","subject":"Merge \"Update docs building, py27 cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/1c6df3b03b4114e6edc62e4de4906bb419e51495"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-05 21:57:21.000000000","tz":-300},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-05-07 14:29:45.000000000","tz":120},"subject":"Add stable branch jobs on the plugins master gate","message":"Add stable branch jobs on the plugins master gate\n\nSupported stable branches in maintenance state\n(stein, train, ussuri) use the Tempest and plugins master\nversion for their testing. To make sure master version (new\ntests or any changes) is compatible with stable branch testing,\nwe need to add the stable branches job on the plugins master gate.\nTempest and many other plugins are already testing the stable\nbranches in same way[1].\n\nUssuri onwards, Tempest or plugins are py3 only, so we are\nrunning only python3 version of stable jobs\n\nTempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\n [1] https://opendev.org/openstack/tempest/src/commit/e8f1876aa6772077f85f380677b30251c2454505/.zuul.yaml#L646\n\nChange-Id: Iffc01ce846db232ee4f1cd4c3603bc17dd8e7cfd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/1e5aec8c4181f40335acd63b815dddf4aff92f7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/1e5aec8c4181f40335acd63b815dddf4aff92f7e"}]},"branch":"refs/heads/master"},"e3b24714854cc77e983350654634e616d0a80b0b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-05-07 20:44:06.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/94/725694/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/94/725694/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/94/725694/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/94/725694/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/94/725694/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/94/725694/3"}}},"commit":{"parents":[{"commit":"6a8eeea27744ed88feeb4f39b5142b2d984e7334","subject":"Install plugin system-wide if INSTALL_TEMPEST is set","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/6a8eeea27744ed88feeb4f39b5142b2d984e7334"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-05 21:57:21.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-07 20:44:06.000000000","tz":0},"subject":"Add stable branch jobs on the plugins master gate","message":"Add stable branch jobs on the plugins master gate\n\nSupported stable branches in maintenance state\n(stein, train, ussuri) use the Tempest and plugins master\nversion for their testing. To make sure master version (new\ntests or any changes) is compatible with stable branch testing,\nwe need to add the stable branches job on the plugins master gate.\nTempest and many other plugins are already testing the stable\nbranches in same way[1].\n\nUssuri onwards, Tempest or plugins are py3 only, so we are\nrunning only python3 version of stable jobs\n\nTempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\n [1] https://opendev.org/openstack/tempest/src/commit/e8f1876aa6772077f85f380677b30251c2454505/.zuul.yaml#L646\n\nChange-Id: Iffc01ce846db232ee4f1cd4c3603bc17dd8e7cfd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/e3b24714854cc77e983350654634e616d0a80b0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/e3b24714854cc77e983350654634e616d0a80b0b"}]},"branch":"refs/heads/master"},"d4919d80cf29291e2e58d20b36f0bd9f9d783558":{"kind":"REWORK","_number":4,"created":"2020-05-08 11:51:29.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/94/725694/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/94/725694/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/94/725694/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/94/725694/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/94/725694/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/94/725694/4"}}},"commit":{"parents":[{"commit":"2bbb08d7e4b8e904c00b3b4bb7203a58f3704d1a","subject":"Install plugin system-wide if INSTALL_TEMPEST is set","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2bbb08d7e4b8e904c00b3b4bb7203a58f3704d1a"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-05 21:57:21.000000000","tz":-300},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-05-08 11:51:08.000000000","tz":120},"subject":"Add stable branch jobs on the plugins master gate","message":"Add stable branch jobs on the plugins master gate\n\nSupported stable branches in maintenance state\n(stein, train, ussuri) use the Tempest and plugins master\nversion for their testing. To make sure master version (new\ntests or any changes) is compatible with stable branch testing,\nwe need to add the stable branches job on the plugins master gate.\nTempest and many other plugins are already testing the stable\nbranches in same way[1].\n\nUssuri onwards, Tempest or plugins are py3 only, so we are\nrunning only python3 version of stable jobs\n\nTempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\n [1] https://opendev.org/openstack/tempest/src/commit/e8f1876aa6772077f85f380677b30251c2454505/.zuul.yaml#L646\n\nChange-Id: Iffc01ce846db232ee4f1cd4c3603bc17dd8e7cfd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d4919d80cf29291e2e58d20b36f0bd9f9d783558"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d4919d80cf29291e2e58d20b36f0bd9f9d783558"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
