)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~726204","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~I9e9b2af6404d6cfd3e43e149fd89b8825fa2cb35","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"np-ipv6","hashtags":[],"change_id":"I9e9b2af6404d6cfd3e43e149fd89b8825fa2cb35","subject":"Adapt Network Policy tempest tests for IPv6.","status":"MERGED","created":"2020-05-07 16:49:26.000000000","updated":"2020-05-12 15:51:47.000000000","submitted":"2020-05-12 15:50:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726204-1589298612823-fc221c0e","meta_rev_id":"949a156e9dad9ea707adc4dbaa25d65a60128b5f","_number":726204,"virtual_id_number":726204,"owner":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"date":"2020-05-08 04:22:58.000000000","_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-12 15:50:12.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2020-05-08 12:51:28.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":2,"date":"2020-05-08 07:32:54.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-08 07:04:45.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":1,"date":"2020-05-08 12:51:28.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-08 07:04:45.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2020-05-08 07:32:54.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"},{"updated":"2020-05-08 12:51:28.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2020-05-12 15:50:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0447d605df860bc81b20970cf1af6b40eb452536","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2020-05-07 16:49:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e89867375c0c34bcb64e154671d14866e2e289c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 19:02:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/021177c7ef9d49b88836a4564e09ce5c : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f07a00747d341a8ac45614cd791e572 : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/288052131b6f4d3784369353cf52e8d2 : SUCCESS in 6m 54s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/7f35af5fb38149c7a0afd552216c2669 : SUCCESS in 1h 06m 17s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/fc2f9b978adf45df9e89b39751d17f35 : SUCCESS in 1h 36m 21s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/1392f35041a146b1b487e3628959b8a0 : SUCCESS in 1h 20m 42s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/963e90779e214122821ec207284e17ca : FAILURE in 1h 32m 17s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/37c58f8875a543b89a92b1a20b52cf13 : SUCCESS in 1h 50m 15s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/be76496fe3f24582abc5371fb2600e85 : SUCCESS in 2h 10m 46s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/88aee3e257324473a4d5c93f7a08072b : SUCCESS in 1h 54m 59s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/aacffae9b5374484b0e429e39eb99b74 : FAILURE in 1h 44m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f6916ca084f9e2a7c95b67362dcdf60df0915ac0","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2020-05-08 04:22:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2b2aca534c9b7bc5bdd6bcecc026138dfd7eca62","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-05-08 07:04:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ed6fd5f75834c44e0e3e3e8e91c596fe262beb36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 07:29:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/518c5a26d956477ca574fe18982e0be7 : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31bc37a98a7f4724b0a1ca819d007def : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a56879f54097499c8050b166c4aea239 : SUCCESS in 3m 28s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/572e05994d3e438480fcc162a961a178 : SUCCESS in 1h 27m 49s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/e9f085ad470a4daa8217ae5bc9cff905 : SUCCESS in 1h 46m 17s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/343e08d56a8940a58524378733c9ee9b : SUCCESS in 1h 11m 48s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/b01576637a5b4f82ad3f2efdcf425f9e : SUCCESS in 2h 22m 28s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/2a5b98ae42bd49c9b3044981b6aa4a0d : SUCCESS in 1h 35m 58s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/b9597b6148a34562884fb8b5095ffd08 : SUCCESS in 3h 05m 28s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/402302d0ea88439a8828f2a2a1128d92 : SUCCESS in 1h 20m 44s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/b8680ec96f6440de9bd8c77ccb03942a : SUCCESS in 1h 25m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a075690f28accc0578b704b43c2866c6cd2650a1","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-05-08 07:32:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ef473bd947061d79760d9d043d37349e76e50472","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-05-08 12:51:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8dc2c5d5a6e2e6cbeae54f98bb7b115cdcf5ebf5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 12:51:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d0a25547d4ac3a3bbb454475151fd7b48feec34a","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-05-08 12:51:53.000000000","message":"Patch Set 1:\n\ncheck experimental\n\nWon\u0027t hurt to take a look at those too.","accounts_in_message":[],"_revision_number":1},{"id":"0866ed736d344b9afa479895d78cd50c13a63285","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 15:05:48.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/936f295aa4e84addba5b74ddd76c5735 : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ff0682a42274c6eae125d364e409639 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b029345e31a24509bc91de734d25a10c : SUCCESS in 3m 53s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/8f244c3a26b24189899b937576cc8e2a : SUCCESS in 1h 07m 25s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/3be0d5ac2d854fc485a78d62af6d5a0c : FAILURE in 49m 52s\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/9da16722ebfb486fb9d47541912c4663 : FAILURE in 1h 53m 39s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/930929c273954fd88aa1ef336a739b37 : FAILURE in 2h 11m 21s","accounts_in_message":[],"_revision_number":1},{"id":"4c70f9b24ebbc2c7da64fe16f62468293eeeff2d","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-05-11 07:26:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9bca605228a7f58c93c64982b52592991b6fabe5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 10:08:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0e5eb26023a04addaca2574352a147e1 : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be1a56b5300e4da6b6fce3ac7f379a13 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea618e6e815e4bf09676b744fb71b88c : SUCCESS in 4m 40s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/0fb2ada2423541b28f041145a87d91d0 : SUCCESS in 1h 14m 50s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/9fbaf388aec04a608743dfac88306677 : SUCCESS in 1h 45m 36s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/e808acbfd6154c6e9910edba32f126c9 : SUCCESS in 1h 16m 29s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/63ae87e0de594683b28ded6c84fd635c : FAILURE in 37m 06s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/e903a1767f764c35a228f902e7a0e03d : SUCCESS in 1h 44m 08s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/73e27fbcb8a34f0696494705402c882a : SUCCESS in 2h 37m 46s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/404fc3d0839b42e3bfcc841bbe239b1e : FAILURE in 1h 42m 38s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/a0fc88eabb144d4ebb940b59b868792c : SUCCESS in 1h 22m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eb2df3f831e6a4951ea978a3d7c0e6b2cd7e14f2","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-05-11 16:07:17.000000000","message":"Patch Set 1:\n\nrecheck\n\nWe have amphora tests broken most likely. :(","accounts_in_message":[],"_revision_number":1},{"id":"5faa3421f42679a00583dbeafcc076683423abbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 18:52:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/32e19ec0f5194d748518cda95d005bb1 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37e50417c05e451599699e7249f261fa : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e70535e80514332834e384e22504396 : SUCCESS in 5m 48s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/44e11b4a5dec4f79ac3df8519a307307 : SUCCESS in 1h 15m 11s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/a6fc931021654db4bfdb4b474bcd0472 : SUCCESS in 2h 08m 30s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/04e32ab74a424ab584ebb4ce3b3459be : FAILURE in 1h 33m 53s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/1c57f52730824aba9e5d199cddd3d02e : SUCCESS in 1h 47m 36s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/88c87be479c04f24b802afe0f2614f53 : SUCCESS in 1h 47m 15s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/698f4d7a7f3540d2bdbf5b5d6eded2c4 : SUCCESS in 2h 39m 02s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/d4f3caf0f8534bbb821112fdf90c54e3 : SUCCESS in 1h 25m 05s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/337fd463178a4393a90efab348323cc1 : SUCCESS in 1h 23m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dd1db651c1ad72282f9c74fef3c8c52d310026eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 18:52:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ea3df102374a5e129d90c1229c1abe4e67be2b5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 20:51:20.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/fe68cbf76ba444c48a7be9f4646ab4d7 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/610334f314ce485f8b9c7189f3466ed7 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c19efac707ef4bcfa70676771977b503 : SUCCESS in 4m 16s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/aacc51e0037e4e4ebaa5883ac127599b : SUCCESS in 1h 08m 52s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/ab933caffa6b4fb8aaf2ec2f2f091440 : FAILURE in 29m 58s\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/9a27a0debda84c2591c0f8e55b81804a : SUCCESS in 1h 56m 28s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/17aac1a99bc24b95b4622e96396c3983 : FAILURE in 1h 46m 41s","accounts_in_message":[],"_revision_number":1},{"id":"a084296b94b40971670e95a8f7d26d714d24d70e","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-05-12 07:07:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"22b437d8d3b0983e3fd5a42e0d32d30ba93ecb87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 10:34:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/4bb134128f714a8ebddce0a342d17fde : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b622ba1ca7c84cc8bbdf72bd0ef80488 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb78fa497d524a00aab79eb8fdb440eb : SUCCESS in 6m 14s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/a1a6da0ff92345ddb734f9ab162cd668 : SUCCESS in 1h 29m 44s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/12556b1e21e649418d25c8c076d292d5 : SUCCESS in 1h 43m 56s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/bbed3faba9a04494b9eff721d645ee65 : SUCCESS in 1h 13m 35s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/cec4a3423f3e47bd9107b121581244fc : SUCCESS in 1h 58m 10s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/54d88e60c23745d991c62a846a3c116c : SUCCESS in 1h 54m 13s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/a9e7083868754b2990f392e2f0bc3053 : SUCCESS in 2h 06m 53s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/42d5dbe5bd2f4c4aa9d373d6960a71ee : SUCCESS in 1h 29m 00s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/860e61708c0b461ca1475c55d3e014ac : SUCCESS in 1h 36m 45s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/471c501099a84524b286b54e66d0db63 : SUCCESS in 1h 17m 04s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/cebd7dc6992446cca7f6b717fc1fe9fa : SUCCESS in 1h 15m 39s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/819567a777194a0188dc605a4c4cb3af : FAILURE in 57m 31s","accounts_in_message":[],"_revision_number":1},{"id":"1a997e6dfc760ff6bbceb7b46002eddd7c31b554","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-05-12 11:20:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"714973b736112fd98c9cabb7ca931f1c6645afe6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 13:44:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/72e135db33f14e6d8912e9a31a47e584 : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c870fbfd6b994dcc93fe51a4df43c923 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4da193a3941f4e78ae3858b4b1f238a2 : SUCCESS in 3m 42s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/0cc42f0e78284aeaac56bab3134f2214 : SUCCESS in 1h 17m 58s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/7d5d1ef5f80b4bd590d1ec6ae3947ca1 : SUCCESS in 2h 04m 36s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/2ce271a0e74a4d07884d939328a956fc : SUCCESS in 1h 58m 38s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/cabbd4da94db47e0a532b04fc897066c : SUCCESS in 1h 42m 00s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/5d42e1b65cc44c73b55b15ec98b4cb5f : SUCCESS in 1h 53m 26s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/3c3e8ede4c4c4effae517779fd60decd : SUCCESS in 2h 16m 17s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/13b30bc82e8e4409b39f0689527d4da3 : SUCCESS in 1h 24m 48s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/fc0159b8a67a4d149e5b46a20df121fa : SUCCESS in 1h 32m 05s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/2ad78534d8f54fb0aca29ca2211125e5 : SUCCESS in 1h 15m 09s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/04a92814d01b44c881453b20630a48f9 : SUCCESS in 1h 17m 32s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/c13b83487b974927a7eecc6f52c23c3e : SUCCESS in 1h 21m 49s","accounts_in_message":[],"_revision_number":1},{"id":"5d3dfe8b296fbee0e9309f061d009bc2a8dc233f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 13:44:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dc98ce32cd5270720bc2aa0bab2112c6488b1f99","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 15:50:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"127dfeda7d62d764aa2e0f5f8343ec4bafdbd536","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 15:50:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0f7a3339b519432c97bb4f71dd329bf4 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45898c88969b4af3bd6a81b8afbd0638 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e28ce6b73e34f17a64c78f0abaa14e3 : SUCCESS in 4m 25s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/19a82dfb6b0b446a845e29963b9c3df1 : SUCCESS in 1h 18m 47s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/a0273509001e422e873671ebaf8a5a1d : SUCCESS in 1h 47m 10s\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/3e48e5b0f37340dab76a248b8c9e27eb : SUCCESS in 1h 50m 06s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/c41aa1aa54c740439a1379a57bb239cb : SUCCESS in 1h 59m 05s","accounts_in_message":[],"_revision_number":1},{"id":"d93719df8fc756ecb9a51763210254bbf4ee0761","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 15:51:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6220a3efc3847e0a7c39ce45505480e : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5569dabadc1f5466f79e78488ad34e013613b32a","revisions":{"5569dabadc1f5466f79e78488ad34e013613b32a":{"kind":"REWORK","_number":1,"created":"2020-05-07 16:49:26.000000000","uploader":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"ref":"refs/changes/04/726204/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/04/726204/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/04/726204/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/04/726204/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/04/726204/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/04/726204/1"}}},"commit":{"parents":[{"commit":"9db2bf03ed1c9c0a7049bf662e91fa85a48364ac","subject":"Merge \"[ussuri][goal] Drop python 2.7 support and testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/9db2bf03ed1c9c0a7049bf662e91fa85a48364ac"}]}],"author":{"name":"gryf","email":"gryf73@gmail.com","date":"2020-05-07 16:12:55.000000000","tz":120},"committer":{"name":"gryf","email":"gryf73@gmail.com","date":"2020-05-07 16:12:55.000000000","tz":120},"subject":"Adapt Network Policy tempest tests for IPv6.","message":"Adapt Network Policy tempest tests for IPv6.\n\nAdditionally, there was narrowed down amount of security group rules for\ntest_ipblock_network_policy_allow_except test, as it only requires pod\nto pod connection, so that instead of creating rules for entire wide\nrange (like 0.0.0.0/0 or ::/0) but only for network, which the pods are\nin.\n\nChange-Id: I9e9b2af6404d6cfd3e43e149fd89b8825fa2cb35\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/5569dabadc1f5466f79e78488ad34e013613b32a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/5569dabadc1f5466f79e78488ad34e013613b32a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
