)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~726241","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~I9a96acfdbc8bfab1e1d56bcd2be5c465c45015be","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"stable-testing-tempest-plugins","hashtags":[],"change_id":"I9a96acfdbc8bfab1e1d56bcd2be5c465c45015be","subject":"Install plugin system-wide if INSTALL_TEMPEST is set","status":"MERGED","created":"2020-05-07 20:43:47.000000000","updated":"2020-05-12 08:50:22.000000000","submitted":"2020-05-12 08:48:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726241-1589273315199-ad9425fa","meta_rev_id":"4ab438b0d3c84fd21a9ae65a72e75db94ba2622b","_number":726241,"virtual_id_number":726241,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-12 08:48:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2020-05-08 14:31:20.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-11 07:27:19.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":1,"date":"2020-05-12 07:06:17.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-11 07:27:19.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2020-05-12 07:06:17.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2020-05-12 08:48:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"32e05a72c780bc354a8106bf14bf733fbe07c92f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-07 20:43:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee72d5a8bd223f13d262a7b388101fefe9216169","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 23:16:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/854b1ca0533d4535b57c68e2115e336c : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1c9b1b733714543a65fe2486253c7f6 : SUCCESS in 4m 57s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/cb0ac2a7576f45959501da4c5b5efb52 : SUCCESS in 1h 33m 54s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/665522de44464475aaf2430b8223987c : SUCCESS in 1h 45m 33s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/d76cf3b4fd4a4e9093953bfad44cc46e : SUCCESS in 1h 18m 26s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/81d5d8d1d1364c0f94a6ceaf453f2575 : FAILURE in 1h 52m 05s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/656bce5883404ebbb94b22fae6fd817d : SUCCESS in 1h 59m 31s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/fcab7b62a0c241e785d9e0e447b442ec : SUCCESS in 2h 30m 37s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/111bf07c152d41c88b4396e815411520 : SUCCESS in 1h 26m 09s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/c938fbc70cd549cca20bf8612c8109f1 : FAILURE in 1h 54m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6fef217854a5d0835c5815da59738a1f4d45131c","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-05-08 11:48:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a381951c66e3df992bed2003f4814299da0a3add","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-05-08 11:51:29.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"3c219b66696b0a49454e1c9f287def646019ea58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 14:12:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84cb9d8d73a046d49b05a7f709e0df20 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e1122e09d4642da8cd75864fecc5b66 : SUCCESS in 4m 15s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/98a3e6be93f64dbdac4eb725c0420a07 : SUCCESS in 1h 15m 52s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/04cadd9cde4e42bca6dc06926d74f354 : SUCCESS in 1h 41m 02s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/2630689631684be799c57398e1333029 : SUCCESS in 1h 33m 14s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/1db6709e17a54519aab462e280ae395e : SUCCESS in 1h 39m 58s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/30a468d6f473481f99954d85b2614392 : FAILURE in 2h 10m 06s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/b31c6aba0ef642099abcb4d81ef6c03e : SUCCESS in 2h 17m 01s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/dfbadcdf798a4fcf94b2055186162f15 : FAILURE in 1h 54m 13s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/acc61068471d4877bdbabc8d15f5c883 : SUCCESS in 1h 40m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2d6fbb057edce9e51cc4101b657fe19e00ed7698","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-05-08 14:31:05.000000000","message":"Patch Set 3:\n\nrecheck\n\nI bet those are not related.","accounts_in_message":[],"_revision_number":3},{"id":"b1c1f91380cf9b7a4021fb424f55cb00a9fb40fb","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-05-08 14:31:20.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"7619d22a140a6547a0a97bece12a0df91c7d7b8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 16:43:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2153d75570b74858aab512d2e0df7e22 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea2ada40511e469183fe53f362381db8 : SUCCESS in 4m 03s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/3f053583ae4646c6ac48eb1f2de1bd2b : SUCCESS in 1h 16m 02s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/ec2dd6f1335747d280f27b40bdf10726 : SUCCESS in 1h 31m 20s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/04bac836a70c46df9f7e537a56b3e879 : FAILURE in 33m 05s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/41bfe96e23d547058b7ab95aaa02f07d : SUCCESS in 1h 50m 22s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/5f5e35516bcc4416b1c3ed7daf74516a : SUCCESS in 1h 51m 12s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/f11009dfb5434fd6b232629c609474b9 : SUCCESS in 2h 09m 49s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/7280d96bde634a5f92856c5930ba1d8f : SUCCESS in 1h 27m 24s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/c832d25e651048b69a8878d1be223dd2 : FAILURE in 49m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9cceabc8c218e0fd66dddda6055e692d439960da","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-05-11 07:27:19.000000000","message":"Patch Set 3: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"78daa2d406c0777d6f771dd3d7c835e993864bf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 10:34:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/634e943a468f4a9091e5f95a9d4221db : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a898b6e812e40a388ed2c085dd6c23f : SUCCESS in 4m 16s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/74891a548843468bbbe8b540bbb6ae70 : FAILURE in 1h 24m 38s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/04a6d31bc7b74083afba0216e64eeb82 : TIMED_OUT in 3h 04m 42s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/10925ad5d983494c9144192df838bcb3 : FAILURE in 2h 32m 10s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/ddbfaeaa76464c4ba1b2d6b0ed474f91 : SUCCESS in 1h 54m 15s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/5f93e1ba6a914e10a6ff8618a12d274a : FAILURE in 2h 25m 18s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/8d152b20b47747dc82e03fe5d19ada77 : TIMED_OUT in 3h 04m 50s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/a02ede2b98d84233bcedaa1402c73c17 : FAILURE in 2h 07m 22s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/ab00d77ae9564a9c9bc79ff51fc7da24 : SUCCESS in 1h 50m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ebc5d8d4a33ef007320a58a74ed1fb4f09f999cd","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-05-11 15:41:42.000000000","message":"Patch Set 3:\n\nrecheck\n\nWeird?","accounts_in_message":[],"_revision_number":3},{"id":"4813f851189d7ea115a97c16e5faf6867eaa2615","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-05-11 15:42:06.000000000","message":"Patch Set 3:\n\nOh, probably Amphora\u0027s broken. :(","accounts_in_message":[],"_revision_number":3},{"id":"5e8951a6268b79e3f8d69167d895f8de7966e88d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 18:20:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46cd839e37a642c1bf7b5d992fa75ebe : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09f8511ced034268b51ccad651972ab5 : SUCCESS in 6m 49s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/5ebfdac27cea480f9c8225e423c18657 : SUCCESS in 1h 18m 01s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/6e1ef36c0cc04e04b702cf3f342c3033 : SUCCESS in 1h 52m 44s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/d59c811c24ff41e4856afce7c2b3f329 : FAILURE in 1h 16m 15s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/bab73b89238240b3aa2ca431791c2282 : SUCCESS in 1h 48m 53s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/09f334fc2b234151a82dbc1e9958ddaa : SUCCESS in 1h 46m 38s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/8c958eea8f504eb1b438e0554b998406 : SUCCESS in 2h 31m 10s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/6bb630e89afc41dd8cbef3b831889cca : SUCCESS in 1h 35m 01s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/ae83c4321bd64da0ba499ab11fd3ebd1 : SUCCESS in 1h 35m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5132fc9317b9a1feb85fae2ecd63a3f93628d185","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-05-12 07:06:17.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a4505ce9fd23695b866f75dfb78f87faf6e03e35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 07:06:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"597c320d777e9574ef3c796a86cb6c273f511e2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 08:48:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/120d28a33189422a98fddec2844aceb3 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bed4827117214cc2a2b44ba5263e84d6 : SUCCESS in 5m 23s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/8824ce748a21494fa64e468e4cd18e9d : SUCCESS in 1h 10m 07s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/8898d1bbe89d463ebc785f377093f5df : SUCCESS in 1h 33m 51s\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/b49496472fa04a3a908fc49d1cfb1dbe : SUCCESS in 1h 38m 18s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/9644392d2f2843e89df19083b4607f59 : SUCCESS in 1h 40m 15s","accounts_in_message":[],"_revision_number":3},{"id":"b8ff88656a280d1ddb2224340e2a13d0acb139ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 08:48:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"2198b3f25e4d180eb2448f2084f1b6f631c6ddde","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 08:50:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b3a02b443044c80a541f9b24cbcd2a3 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2bbb08d7e4b8e904c00b3b4bb7203a58f3704d1a","revisions":{"6a8eeea27744ed88feeb4f39b5142b2d984e7334":{"kind":"REWORK","_number":1,"created":"2020-05-07 20:43:47.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/41/726241/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/41/726241/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/41/726241/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/41/726241/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/41/726241/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/41/726241/1"}}},"commit":{"parents":[{"commit":"9d79a0380723835287ed107f43c46305aa845b7f","subject":"Merge \"Add retry for check ports in test_port_pool\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/9d79a0380723835287ed107f43c46305aa845b7f"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-07 20:40:58.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-07 20:40:58.000000000","tz":-300},"subject":"Install plugin system-wide if INSTALL_TEMPEST is set","message":"Install plugin system-wide if INSTALL_TEMPEST is set\n\nSimilarly to main tempest installation check, we should only install\nplugin on master branches (where INSTALL_TEMPEST is True), and not on\nstable branches\n\nChange-Id: I9a96acfdbc8bfab1e1d56bcd2be5c465c45015be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/6a8eeea27744ed88feeb4f39b5142b2d984e7334"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/6a8eeea27744ed88feeb4f39b5142b2d984e7334"}]},"branch":"refs/heads/master"},"a9c01b340b1db68ea00b9444ba50cb99c8fdbadf":{"kind":"REWORK","_number":2,"created":"2020-05-08 11:48:30.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/41/726241/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/41/726241/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/41/726241/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/41/726241/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/41/726241/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/41/726241/2"}}},"commit":{"parents":[{"commit":"9d79a0380723835287ed107f43c46305aa845b7f","subject":"Merge \"Add retry for check ports in test_port_pool\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/9d79a0380723835287ed107f43c46305aa845b7f"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-07 20:40:58.000000000","tz":-300},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-05-08 11:47:24.000000000","tz":120},"subject":"Install plugin system-wide if INSTALL_TEMPEST is set","message":"Install plugin system-wide if INSTALL_TEMPEST is set\n\nSimilarly to main tempest installation check, we should only install\nplugin on master branches (where INSTALL_TEMPEST is True), and not on\nstable branches\n\nChange-Id: I9a96acfdbc8bfab1e1d56bcd2be5c465c45015be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/a9c01b340b1db68ea00b9444ba50cb99c8fdbadf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/a9c01b340b1db68ea00b9444ba50cb99c8fdbadf"}]},"branch":"refs/heads/master"},"2bbb08d7e4b8e904c00b3b4bb7203a58f3704d1a":{"kind":"NO_CHANGE","_number":3,"created":"2020-05-08 11:51:29.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/41/726241/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/41/726241/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/41/726241/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/41/726241/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/41/726241/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/41/726241/3"}}},"commit":{"parents":[{"commit":"9d79a0380723835287ed107f43c46305aa845b7f","subject":"Merge \"Add retry for check ports in test_port_pool\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/9d79a0380723835287ed107f43c46305aa845b7f"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-07 20:40:58.000000000","tz":-300},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-05-08 11:51:08.000000000","tz":120},"subject":"Install plugin system-wide if INSTALL_TEMPEST is set","message":"Install plugin system-wide if INSTALL_TEMPEST is set\n\nSimilarly to main tempest installation check, we should only install\nplugin on master branches (where INSTALL_TEMPEST is True), and not on\nstable branches\n\nChange-Id: I9a96acfdbc8bfab1e1d56bcd2be5c465c45015be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2bbb08d7e4b8e904c00b3b4bb7203a58f3704d1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2bbb08d7e4b8e904c00b3b4bb7203a58f3704d1a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
