)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~728693","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~I20f23ef805320e63fad99c49de0c8fdfd9c9936d","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"namespace","hashtags":[],"change_id":"I20f23ef805320e63fad99c49de0c8fdfd9c9936d","subject":"Raise timeout exception kuryrnetwork events","status":"MERGED","created":"2020-05-17 14:32:59.000000000","updated":"2020-05-27 09:13:15.000000000","submitted":"2020-05-27 09:10:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"728693-1590570649917-ba137570","meta_rev_id":"e5ad29f8a1301bff49515e2a51f0cd6107e8fb0d","_number":728693,"virtual_id_number":728693,"owner":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-05-26 12:34:25.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"date":"2020-05-25 12:55:26.000000000","_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-27 09:10:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":2,"date":"2020-05-26 08:20:44.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":2,"date":"2020-05-26 07:31:39.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":1,"date":"2020-05-27 07:04:11.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-18 05:56:01.000000000","updated_by":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"reviewer":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"state":"REVIEWER"},{"updated":"2020-05-19 06:35:27.000000000","updated_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"},{"updated":"2020-05-26 08:20:44.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"},{"updated":"2020-05-26 12:34:25.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2020-05-27 07:04:11.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2020-05-27 09:10:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"656a0e9401b14f48154826a3644f3688d0904dc3","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-05-17 14:32:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"24b902d290b6d2997b74973e6a0654899482b51d","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-05-17 14:33:19.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"96be1de299e07e9bbc54309255f755b60dad1086","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 16:37:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20c325c4b99b489595aaea11d482798e : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84dd7ba5d4774c44b8d13c1907adac2f : SUCCESS in 3m 21s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/d08a78639d3a42678493bf62fec0989b : SUCCESS in 1h 01m 53s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/5c7e3e263a78402290dfe19c4bfe476f : SUCCESS in 1h 42m 06s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/f80a733a5f9f4963bbf649df4f39f4b1 : SUCCESS in 1h 24m 21s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/1823c332878c40ceb327f8fec9a845e4 : SUCCESS in 1h 55m 01s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/6743af52aab5462782c71ed67008cca8 : SUCCESS in 1h 34m 27s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/bde86c298e1c4cabb41e9427141662bf : FAILURE in 2h 03m 48s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/f7bddc3bdd51415794e82cf81d3f1b44 : SUCCESS in 1h 26m 52s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/e8f7fc451db04c25ba0fd98460074295 : SUCCESS in 1h 39m 20s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/8259acf6b665418cb0b7a2ca5b59d2e9 : SUCCESS in 1h 11m 57s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/05c2f8826bc14fef8cd0f91888ec57b1 : SUCCESS in 1h 34m 36s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/931c3b927b8b47fca4ed686e2cb2c271 : SUCCESS in 1h 43m 12s","accounts_in_message":[],"_revision_number":1},{"id":"13bc3d79fde51d9698076202f05e4763e6fd1c30","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 16:52:06.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-l2 https://zuul.opendev.org/t/openstack/build/0e3b35a6a4564926a2a76e3a58ac2a0b : SUCCESS in 1h 35m 12s\n- kuryr-kubernetes-tempest-containerized-openshift https://zuul.opendev.org/t/openstack/build/79e8ce4c30124a51ae7436ce3d50922e : FAILURE in 1h 21m 48s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace https://zuul.opendev.org/t/openstack/build/cd13b34226e24b2fabbdf20ffeaa36ba : FAILURE in 1h 54m 35s\n- kuryr-kubernetes-tempest-containerized-openshift-serial https://zuul.opendev.org/t/openstack/build/bae81f465808417b8bd7ccd8bece0c4d : FAILURE in 2h 06m 25s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/a513f9961c72440898b1e980739785a9 : SUCCESS in 1h 32m 24s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/35cb198caa734085bf4c6866714155c0 : FAILURE in 59m 59s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif https://zuul.opendev.org/t/openstack/build/c84ddd78d4e443ceb54b440989b6ccf8 : FAILURE in 1h 12m 11s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/fdeceab6dc7c4479841cc0d438f28e08 : TIMED_OUT in 2h 13m 43s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio https://zuul.opendev.org/t/openstack/build/40f32082e01e43669ed7388edc670695 : FAILURE in 26m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0984ae827d671ecfc12784d45a62607ed7961770","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2020-05-18 05:56:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e03b1b1ae2ef87079f3b95ad0e68ecf614bd6d5a","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-05-18 10:49:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7dbd3096b9500441ed50ccc1170ad17953b1a579","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 13:18:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8837950e8fc4ba1bc5a9d3f5bf21e01 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b9be7544da4435e84a8790f1088af57 : SUCCESS in 3m 25s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/23dbe3cdb1914fab96f118ac770142c3 : SUCCESS in 1h 15m 21s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/a1c4b26275de40ed9f6bbcce546351f6 : SUCCESS in 1h 43m 33s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/51467d2f478d433b95b5a4c474d5c03e : SUCCESS in 1h 21m 55s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/960662e37aff49b6a0ace9f32aad087a : SUCCESS in 1h 38m 55s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/cfbb590d397b43fe95e826aeb13c59ec : FAILURE in 1h 53m 12s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/f7e06d1aedcb4384815217cc4b3423a6 : SUCCESS in 2h 21m 51s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/701be1e790794f409a36186511604266 : SUCCESS in 1h 46m 23s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/69dccaaab70f4e13a14669cdb6e66d13 : SUCCESS in 1h 40m 38s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/56166b5a13064f1d9dc40293219fbd88 : SUCCESS in 1h 24m 14s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/0dce15e14fc74fe4b8ee636d040b29f5 : SUCCESS in 1h 48m 22s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/f4cbbcc2625f4cbb858a8cf351414759 : SUCCESS in 1h 39m 18s","accounts_in_message":[],"_revision_number":2},{"id":"389dabfd6d9049566098ed70ab47026812d6c612","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2020-05-19 06:35:27.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ca6d53e07b992843e9b2b11319c565be9ab01cd8","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-05-19 07:05:36.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"0cc54fdba70e017c4a9abc57789f8db095393b77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 09:22:28.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fedeca3be36644288c369d6b102a28d8 : SUCCESS in 7m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9486c9461e24696a73d8590a2afc830 : SUCCESS in 6m 15s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/8d7c1e64ebc14ce2b1fde745a46be917 : SUCCESS in 1h 15m 54s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/80630e47953a4cf3a5b941472e1d4285 : SUCCESS in 1h 50m 19s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/f81281b738b04161bed67d28ce193990 : SUCCESS in 1h 14m 36s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/1023fbd0d2324e9182a96b6035455d5c : SUCCESS in 1h 37m 01s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/8ed154dc345c48fea828f1b7ac9ed449 : FAILURE in 1h 45m 45s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/7581277aa5e349b39a3362751d46cd69 : SUCCESS in 2h 38m 27s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/84c261444edf451bb09fe6f75f063bbc : SUCCESS in 1h 35m 32s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/6f27988c38e04bdc9f3cf4b17893d32c : SUCCESS in 1h 30m 43s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/36bdaec23723432fb6b1786b220a7d21 : SUCCESS in 1h 32m 49s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/44df872ec5e44aea954566e4e81bd005 : FAILURE in 2h 02m 17s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/5950e37132204365b7cf4b43897bf74d : SUCCESS in 1h 36m 28s","accounts_in_message":[],"_revision_number":2},{"id":"a06b06d4ed502f91c55b6877f9535b5367d5ef70","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-05-19 10:05:38.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"bc4c38a70903261fe1cd7d73101771d65773bb4d","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-05-19 10:08:54.000000000","message":"Patch Set 2: Code-Review+1\n\nOkay, that doesn\u0027t look like timeout.","accounts_in_message":[],"_revision_number":2},{"id":"4029f2996d32b96cc771d3b07bd72910f1d8da5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 12:49:35.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27920b118adc4727a29f8b303d2434eb : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2a4eaafe4d44abd94b8af69e79831b4 : SUCCESS in 4m 08s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/45f2c86841c04750b23e4191bed54ebf : SUCCESS in 1h 46m 17s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/79a06d56749e46e5b9119e1f4c4e1a5e : SUCCESS in 1h 47m 24s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/21662681ff2e48d8985162c39f1f5035 : SUCCESS in 1h 58m 43s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/bec54ede6fd04bd097a769948dc4ec36 : SUCCESS in 2h 04m 39s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/6770031249264b56823754b36e63707b : FAILURE in 2h 31m 09s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/e7adff2143c743ffa808ecf2c7dca49c : SUCCESS in 2h 42m 06s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/8acb4d6f8d9d4ee8b57246bd46c645fd : FAILURE in 2h 10m 32s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/ce6569ba818249fcb132495b878be0be : SUCCESS in 2h 00m 35s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/ee9e5f994a064e79a0afac7319ac650c : SUCCESS in 1h 36m 49s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/72958f56a4d84981b39fbd595bfe5b6b : FAILURE in 2h 16m 14s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/b9bb387baf564dd08d2d47fbf329d879 : FAILURE in 2h 02m 02s","accounts_in_message":[],"_revision_number":2},{"id":"b10a78530f2ca872c41c7d17baa852c184403ea2","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-05-21 15:26:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b237152842ee8ff2984084a55642c509b4e847c4","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-05-21 15:37:54.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fa1949be9afa5272c3384f74baff03ef0c84242f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 17:46:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3ac4d89a36e4e88a8577dd321f70395 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4db6aa85eba64248ad8313efaae178e6 : SUCCESS in 6m 33s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/6c644680ad334ccba934362801a8d655 : FAILURE in 1h 18m 20s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/a907f05672094df7989e86a283610b74 : FAILURE in 1h 45m 49s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/063a91e1cf3249bfacd9473c49e23d18 : FAILURE in 1h 31m 46s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/f72d0c474c26494bbf75800a225b12c1 : FAILURE in 1h 47m 43s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/8dd570c86fb74caf8f633b8a10521c40 : FAILURE in 1h 44m 56s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/22cfd3878d1e4a79acee308cd151efa5 : FAILURE in 2h 17m 15s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/483223ae84234bce90b53293dd92ce2d : FAILURE in 1h 32m 02s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/e990d6fbd5ed4190bc81466423cdcfce : FAILURE in 1h 22m 26s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/11aa3e7531bd4aabbd0a6c636f2e5c8e : FAILURE in 1h 09m 00s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/574849d7da0d46eab8737572bdb413f4 : FAILURE in 1h 30m 54s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/2689144366b445429fe5e95dee237876 : FAILURE in 1h 37m 11s","accounts_in_message":[],"_revision_number":3},{"id":"eac34cb24f0aa2ef30fa4c9149ae98953489a006","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-05-22 08:19:59.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1af9a7896174058a151c4f187bad6b9fe839327a","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-05-22 11:04:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5a8a582e82e7cf423df73952eed056159a0aa30c","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-05-22 11:04:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2c41634a234a7fcff533ef31ffc7093ce1daa6af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 13:03:57.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f81568498fc4c3898f4b8c6bcc33012 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a0e5eecf116404895eca69502e92a92 : SUCCESS in 3m 35s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/35091f69892342909093cb29cbd732b3 : FAILURE in 1h 12m 34s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/10e40d0c371048e6b5d8b7b01e277959 : FAILURE in 1h 22m 51s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/5235579d062e4fcf9796a26cfa0c652a : POST_FAILURE in 1h 27m 31s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/89f7773d798f4eb0b686b84051389c01 : FAILURE in 1h 37m 59s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/b1677083c97d4d7eb53d1776e06cd35d : FAILURE in 1h 53m 24s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/0f221333d45344dcbad17767b56ffad7 : FAILURE in 1h 38m 24s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/b8c4575ef02c4772b6c40fa740059974 : FAILURE in 1h 23m 54s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/8be56344e5664b72b03dfa306d1db4e5 : FAILURE in 1h 35m 00s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/1c0d1e55db7a4df986ecb86a2dfb5c6c : FAILURE in 57m 38s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/b04867af41284aa8999ef8ddfe6b1aba : FAILURE in 1h 13m 25s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/be56635827404776ad0c6328dcb46042 : FAILURE in 1h 31m 09s","accounts_in_message":[],"_revision_number":4},{"id":"7698f4a8fa112ca2891fb19bd47cadfb6f45cb12","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-05-22 13:09:38.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"342d10fd7f4b2fc9b17466b04d1cee9d15f19bb2","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-05-24 06:57:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"67f972345991bb03b1397e7173baf46c15952a08","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-05-24 06:58:15.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"096671a4a8ddadbf51ec684730457107234870c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 09:39:31.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/734347c0e4e74c3187c082aef242acf7 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/657f5b7afc3542a68e3cd679c6f5a56b : SUCCESS in 4m 05s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/0822987a19bc410fb1599bf7e4b12371 : SUCCESS in 1h 16m 29s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/78c4e74d1e8b442988abcfe084aaa937 : SUCCESS in 1h 39m 53s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/e9cdcf9eaf994a9b893177f719059d81 : FAILURE in 1h 28m 38s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/d806d258b2d44edcb34309cf13dde2c7 : SUCCESS in 2h 00m 24s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/52a9aa5c991c42ae8dd59af172e918de : FAILURE in 1h 49m 07s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/46291991f9ff4566a6b499127716115e : SUCCESS in 2h 25m 59s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/b3dde6325bbb418ab57a6fc44637a02e : SUCCESS in 1h 32m 54s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/8748b4298adf4600b8f80d856fb0fc44 : SUCCESS in 1h 46m 19s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/41a28b5756de4b6d8aaddd35040e44ab : SUCCESS in 1h 14m 30s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/b763edf672f14b7f9f4e7d70c1158b10 : SUCCESS in 1h 25m 15s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/7839ddf79ebd430cae07c477d9ef72ea : FAILURE in 1h 42m 52s","accounts_in_message":[],"_revision_number":5},{"id":"3804d1cfcbbbe6121ec5e08a1d8118f3bd27a226","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-05-25 07:46:03.000000000","message":"Patch Set 5: Code-Review-1\n\nseems 600 is not enough for the loadbalancer to get created... but eventually the annotations are made. We can try raising that to 1200 instead","accounts_in_message":[],"_revision_number":5},{"id":"c77bbbe65b43b69b03e8d1e75e58f4feb80c189c","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-05-25 09:51:22.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1147271758b1da5c33e06d62035736c3a18d3d05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 12:12:39.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eacc622664b643d7b84fa0e8b8b8c76b : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f9a3043efcc46b9bf801b3bb4b331e5 : SUCCESS in 4m 26s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/253df369d7ea43df9374425b0557a792 : SUCCESS in 1h 10m 08s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/9fc83890da1848c9ba33f5288e9fb628 : SUCCESS in 1h 44m 46s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/f0de1d34e7e0431888b03da17a969daf : FAILURE in 1h 36m 39s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/7cafcc11223f4f3a9306622bce64fb60 : FAILURE in 1h 53m 50s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/7f88c16ae73044eaafb2183edcb0145d : SUCCESS in 1h 53m 20s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/36ec2b143821415e80ebf75eeb44c0d3 : SUCCESS in 2h 14m 54s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/8ca53f2ac17d437b989be3f6d6dd558f : SUCCESS in 1h 37m 29s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/2cded4903d65467db7eccd141f86dd4b : SUCCESS in 1h 47m 41s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/b5d5a2bc0fc7401a9f93a9748977485c : SUCCESS in 1h 20m 08s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/2b4b0ce13b7a48a18eeeac6481ad3b2b : SUCCESS in 1h 24m 58s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/92d7e091c80a4bc4bc6e51417a604a22 : SUCCESS in 1h 41m 28s","accounts_in_message":[],"_revision_number":6},{"id":"78fc39156cafde52338617101149af0c65b36324","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-05-25 12:55:26.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"43822a5959f43f2549676f878036cfcb900f43fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 15:13:54.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c1c3f7f2ee64f149c4aad6188cd62d0 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/512d5533e9f14d7cb06cc8b4b1661ede : SUCCESS in 3m 30s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/ee0e5054f11646f38c185cc9f9629b82 : SUCCESS in 1h 27m 02s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/0f1a82b38643463c80e6e971bd364aee : SUCCESS in 1h 45m 10s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/ad16c85037ee442d91eb2dfb44e1722e : SUCCESS in 1h 31m 13s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/991dc479f313446b8b5162f7bf3d9adc : SUCCESS in 1h 53m 39s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/a373eb29d3164f43b4f2b6e71cbb8db8 : SUCCESS in 1h 59m 52s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/2d192639fc684470a4455b1bfa27bb1e : SUCCESS in 2h 09m 53s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/ad0c7225140a436ab26f691a74d362c9 : SUCCESS in 1h 27m 33s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/2ed3699f54664f88bf7b49cfe4cc8753 : SUCCESS in 1h 42m 59s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/6d631cabe6cd4fab96f45214d19d3ec5 : SUCCESS in 1h 24m 28s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/bd7be24b6d904b028ead0bce4e8817b7 : SUCCESS in 1h 28m 37s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/fceede57a725492b87f25fb993d59a05 : FAILURE in 2h 16m 12s","accounts_in_message":[],"_revision_number":6},{"id":"44b357c180f99f7822cdbe6ed2c4dfcb3caddcb4","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-05-26 07:31:39.000000000","message":"Patch Set 6: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"0f50272d6059f440b766a9a42cb52a2978decbd4","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-05-26 08:20:44.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"3201c4efea8538c8c8cabffc12f0256790a02c54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 10:28:00.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97e14b57dc834d3a88e3d330eb066d96 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcaa4ca3bc984a91a70502996e7202a9 : SUCCESS in 4m 13s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/bba5bc61071646abafffe08d2e70d5f4 : SUCCESS in 1h 30m 29s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/0020b6050cea49608b18b1b976c5f019 : FAILURE in 2h 12m 31s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/a76d49af84e04f7f9791d1f2bf36c84c : SUCCESS in 1h 49m 52s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/a3550e1e92ff49cc9cfa93dad016debb : FAILURE in 2h 22m 57s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/648aa6d920e94eb58c17149eb51f4a2e : FAILURE in 2h 16m 21s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/b2158205c6db4776b78ed1fe9976406c : SUCCESS in 2h 54m 11s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/e46f4433b92f407e81667847f24d5124 : FAILURE in 1h 52m 12s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/df33d0d24d87462ab269fc6012bf9fb8 : SUCCESS in 1h 40m 12s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/26c68adc81ed469d8c53d633ed0f15f1 : SUCCESS in 1h 51m 50s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/d4973e69105f40e7b6218180a441a859 : FAILURE in 1h 41m 50s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/a6bbae1d862d4844bbdfc152d1f60734 : FAILURE in 2h 04m 24s","accounts_in_message":[],"_revision_number":6},{"id":"4fa50b42853dd4dc42fefef33a05cf1a8271e4dc","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-05-26 12:34:25.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"a777e1aafdd3a188df35a31ba0eaa9df91b1fb4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 15:10:50.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a98ac3a0c4f47dd96da8bfba5d2ec3a : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a10a2365f17458ca8f37570161909fd : SUCCESS in 6m 45s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/c78c0a26cc954485b07fc01f1c92258b : SUCCESS in 1h 21m 31s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/da50ae90443e44e496e7b5cc08b3c977 : SUCCESS in 1h 39m 43s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/1067d74524a2448688105a49fea7e7aa : SUCCESS in 1h 22m 38s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/04c775af0cf1424dae1a950a0631a676 : SUCCESS in 1h 39m 23s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/cac1a242b50d4858b598110e0d089d68 : SUCCESS in 1h 48m 11s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/cdc6b21d7e954e9b9086eb4ead20c7f0 : SUCCESS in 2h 25m 52s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/9b809a1124aa495586c9e0cc5e72d20f : FAILURE in 1h 46m 00s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/24495041562646328f7b28eaf6149692 : SUCCESS in 1h 47m 34s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/6c87e07a49424478986d6e04753d232a : SUCCESS in 1h 25m 33s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/f63f9f559b134441ad9eab15c282ed68 : SUCCESS in 1h 31m 40s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/65cdcf83e59f441db069671811be6e32 : SUCCESS in 1h 37m 34s","accounts_in_message":[],"_revision_number":6},{"id":"6a2f56d0bdd200b8629b2433e850538e3a55c1ff","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-05-27 07:04:11.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"f71906f401cc06c970c0f3a12cf70e3212b87e64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 07:04:22.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"28f622e839e83e8c267fd6e378feea6817507db1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 09:10:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"286783dedfa8a49f7afaba72bf62b2b14dfef7f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 09:10:49.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b185c7b94254d96bf4ca8201533a5ac : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6c072c3102044b6b1f8639a3c0fdd58 : SUCCESS in 3m 23s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/357983be23b3429c8b9501b63e71116e : SUCCESS in 1h 35m 10s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/d27d2bb1d8a543c38fb9613c49ac902c : SUCCESS in 2h 00m 51s\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/09de956bf55241d68282a86554d53d61 : SUCCESS in 1h 46m 13s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/ab1695ecd5f14f39adaa4677bd1df845 : SUCCESS in 1h 40m 11s","accounts_in_message":[],"_revision_number":6},{"id":"2f570253c99470554f55c9197308f23ca887e261","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 09:13:15.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a3bf260074942899de25097dfe8ebe9 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"c281c48a534e5c0ac1e1715284279694a692ab96","revisions":{"e5d0f15c14a994b18d982abd636713a19f5da828":{"kind":"REWORK","_number":1,"created":"2020-05-17 14:32:59.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/93/728693/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/93/728693/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/93/728693/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/93/728693/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/93/728693/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/93/728693/1"}}},"commit":{"parents":[{"commit":"f3e5484218fbf357f317a947625da97a51e46d65","subject":"Fix hacking min version to 3.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/f3e5484218fbf357f317a947625da97a51e46d65"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-05-17 14:28:51.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-05-17 14:28:51.000000000","tz":180},"subject":"Raise timeout exception kuryrnetwork events","message":"Raise timeout exception kuryrnetwork events\n\nRaise a timeout exception when kuryrnetwork CRD creation times out\n\nChange-Id: I20f23ef805320e63fad99c49de0c8fdfd9c9936d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/e5d0f15c14a994b18d982abd636713a19f5da828"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/e5d0f15c14a994b18d982abd636713a19f5da828"}]},"branch":"refs/heads/master"},"d5a9e30bf0dffeafced2c15e9513fa85ce86170c":{"kind":"REWORK","_number":2,"created":"2020-05-18 10:49:40.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/93/728693/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/93/728693/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/93/728693/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/93/728693/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/93/728693/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/93/728693/2"}}},"commit":{"parents":[{"commit":"f3e5484218fbf357f317a947625da97a51e46d65","subject":"Fix hacking min version to 3.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/f3e5484218fbf357f317a947625da97a51e46d65"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-05-17 14:28:51.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-05-18 10:49:29.000000000","tz":180},"subject":"Raise timeout exception kuryrnetwork events","message":"Raise timeout exception kuryrnetwork events\n\nRaise a timeout exception when kuryrnetwork CRD creation times out\n\nChange-Id: I20f23ef805320e63fad99c49de0c8fdfd9c9936d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d5a9e30bf0dffeafced2c15e9513fa85ce86170c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d5a9e30bf0dffeafced2c15e9513fa85ce86170c"}]},"branch":"refs/heads/master"},"467fbff08378488210982f411b40a61b0d4949e7":{"kind":"REWORK","_number":3,"created":"2020-05-21 15:26:47.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/93/728693/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/93/728693/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/93/728693/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/93/728693/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/93/728693/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/93/728693/3"}}},"commit":{"parents":[{"commit":"50a569cd300ab727e14f0945a8aaea2b9f634a69","subject":"Merge \"Fix a typo in base\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/50a569cd300ab727e14f0945a8aaea2b9f634a69"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-05-17 14:28:51.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-05-21 15:26:33.000000000","tz":180},"subject":"Raise timeout exception kuryrnetwork events","message":"Raise timeout exception kuryrnetwork events\n\nRaise a timeout exception when kuryrnetwork CRD creation times out\n\nChange-Id: I20f23ef805320e63fad99c49de0c8fdfd9c9936d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/467fbff08378488210982f411b40a61b0d4949e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/467fbff08378488210982f411b40a61b0d4949e7"}]},"branch":"refs/heads/master"},"17966efe99d89f6c00ca99693afd67a7d35c92cf":{"kind":"REWORK","_number":4,"created":"2020-05-22 11:04:51.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/93/728693/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/93/728693/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/93/728693/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/93/728693/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/93/728693/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/93/728693/4"}}},"commit":{"parents":[{"commit":"50a569cd300ab727e14f0945a8aaea2b9f634a69","subject":"Merge \"Fix a typo in base\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/50a569cd300ab727e14f0945a8aaea2b9f634a69"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-05-17 14:28:51.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-05-22 11:04:41.000000000","tz":180},"subject":"Raise timeout exception kuryrnetwork events","message":"Raise timeout exception kuryrnetwork events\n\nRaise a timeout exception when kuryrnetwork CRD creation times out\n\nChange-Id: I20f23ef805320e63fad99c49de0c8fdfd9c9936d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/17966efe99d89f6c00ca99693afd67a7d35c92cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/17966efe99d89f6c00ca99693afd67a7d35c92cf"}]},"branch":"refs/heads/master"},"f2718679b692a579b52985384782c27c01d88cbe":{"kind":"REWORK","_number":5,"created":"2020-05-24 06:58:15.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/93/728693/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/93/728693/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/93/728693/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/93/728693/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/93/728693/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/93/728693/5"}}},"commit":{"parents":[{"commit":"50a569cd300ab727e14f0945a8aaea2b9f634a69","subject":"Merge \"Fix a typo in base\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/50a569cd300ab727e14f0945a8aaea2b9f634a69"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-05-17 14:28:51.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-05-24 06:57:58.000000000","tz":180},"subject":"Raise timeout exception kuryrnetwork events","message":"Raise timeout exception kuryrnetwork events\n\nRaise a timeout exception when kuryrnetwork CRD creation times out\n\nChange-Id: I20f23ef805320e63fad99c49de0c8fdfd9c9936d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/f2718679b692a579b52985384782c27c01d88cbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/f2718679b692a579b52985384782c27c01d88cbe"}]},"branch":"refs/heads/master"},"c281c48a534e5c0ac1e1715284279694a692ab96":{"kind":"REWORK","_number":6,"created":"2020-05-25 09:51:22.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/93/728693/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/93/728693/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/93/728693/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/93/728693/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/93/728693/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/93/728693/6"}}},"commit":{"parents":[{"commit":"50a569cd300ab727e14f0945a8aaea2b9f634a69","subject":"Merge \"Fix a typo in base\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/50a569cd300ab727e14f0945a8aaea2b9f634a69"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-05-17 14:28:51.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-05-25 09:51:12.000000000","tz":180},"subject":"Raise timeout exception kuryrnetwork events","message":"Raise timeout exception kuryrnetwork events\n\nRaise a timeout exception when kuryrnetwork CRD creation times out\n\nChange-Id: I20f23ef805320e63fad99c49de0c8fdfd9c9936d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c281c48a534e5c0ac1e1715284279694a692ab96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c281c48a534e5c0ac1e1715284279694a692ab96"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
