)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~737260","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~I8f5a141704c0a7c2095134c4afce2433e83f9415","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"fix-stable-gates","hashtags":[],"change_id":"I8f5a141704c0a7c2095134c4afce2433e83f9415","subject":"Workaround missing DIB on stable jobs","status":"MERGED","created":"2020-06-22 11:29:45.000000000","updated":"2020-06-24 20:06:57.000000000","submitted":"2020-06-24 20:00:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737260-1593028810317-e9b972b9","meta_rev_id":"b587005edb2e869a068f3afa64e62117ddb7a74e","_number":737260,"virtual_id_number":737260,"owner":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":0,"date":"2020-06-24 13:05:48.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-24 20:00:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"all":[{"value":0,"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":2,"date":"2020-06-22 12:25:57.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":1,"date":"2020-06-22 12:25:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-22 14:03:15.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"all":[{"value":0,"date":"2020-06-22 13:31:40.000000000","_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":1,"date":"2020-06-22 14:12:11.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-22 12:25:54.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-06-22 13:31:40.000000000","updated_by":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"reviewer":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"state":"REVIEWER"},{"updated":"2020-06-22 14:03:15.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2020-06-22 14:12:11.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"},{"updated":"2020-06-24 20:00:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fed3b082c14364abfaacc2330afd04ce5cfd33a9","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-06-22 11:29:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe596433c32df4a1a0517213ed39709620a7e740","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-22 12:25:54.000000000","message":"Patch Set 1: Code-Review+1\n\nWaiting for CI result.\n\n(just a comment to keep a track of related discussion c.f Carlos\u0027s comment on https://review.opendev.org/#/c/734364/).","accounts_in_message":[],"_revision_number":1},{"id":"763827de83be1790a5342e5ac811a51795a8cb68","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-06-22 12:25:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1c1632483a87cd47dae268d93fe92d96ea808d02","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-06-22 14:03:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f246a1d59538cc036faec50b4cc676f22655ccbc","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-06-22 14:12:11.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d4b3a56a9251e032a32ec882de6f21883b0a88ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 14:46:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7caafc6ea5514fb3bd9eca9ae4c77e5c : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdb31f343307482db4ee1bc14196f24c : SUCCESS in 4m 13s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/76c672c8854c44668baa806df571224c : SUCCESS in 1h 20m 49s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/e0517ecbd3e14fdabd2376c80f960a92 : SUCCESS in 1h 53m 35s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/d5e96c653f354cc29642f57028581301 : SUCCESS in 1h 25m 30s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/0b53ad8530954abfb57b70653a2f7caa : FAILURE in 2h 26m 45s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/7e19d5ed9cb6459482536616009335d4 : SUCCESS in 1h 35m 46s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/583a7af4f8974911a913210b854cbfd4 : SUCCESS in 3h 02m 29s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/1b0c81b7e36e4635853ea2880ea42ab6 : FAILURE in 1h 06m 34s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/84f92f20800148169517538f8964f5a9 : FAILURE in 1h 08m 43s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/376d88e16f854d67af356a962c834ab8 : SUCCESS in 1h 05m 03s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/70ca9c5273594bf09e23ba4393f5f7d6 : SUCCESS in 1h 37m 12s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/f62e96fae8c54f7ea69fd5d70dd3b2d5 : SUCCESS in 2h 02m 39s","accounts_in_message":[],"_revision_number":1},{"id":"a5ef1daf0a7a05efeba77f4da69c70fd3a87bf04","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-06-22 14:47:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b2b9b0023f6755cc493993689b7838ef37fc49a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 16:54:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e56a4c95573a4be29cd3b367a1b7b40d : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5e6e89602284f4cae2dc9b29709cbe4 : SUCCESS in 5m 31s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/9f46be6e937349b5a14897455c24f20d : SUCCESS in 1h 22m 34s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/34cabb27db604637a5303ac3b26a1dfb : SUCCESS in 2h 02m 13s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/bc03500027be418aa0c7a3ee9959abdf : SUCCESS in 1h 20m 47s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/bddb96967be84eee8eda001bab38cf20 : SUCCESS in 1h 36m 58s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/99657e2d63dd4961922a9884a7ee6668 : SUCCESS in 2h 01m 16s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/1190baf335ab46ffbf8d8f113f32fabf : SUCCESS in 2h 03m 04s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/83db88db9a3f4e629a0c630d5aa0b1dd : SUCCESS in 1h 22m 59s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/88dc428d58674bcc9d86bbaed9ab1de8 : SUCCESS in 1h 26m 57s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/3b27e789ee7c4b59bbae6485f9b09871 : SUCCESS in 1h 23m 00s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/5febabc0860642628f5ec242a2dde1f8 : SUCCESS in 1h 16m 20s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/2ab6bcfa93a943d1bc72beaba32567ed : SUCCESS in 1h 38m 20s","accounts_in_message":[],"_revision_number":1},{"id":"61b0efccd0f513c8646a0a8a08fd3116898d90a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 16:55:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e29f67f9d5d90681f5e8c18093af76bbe197efa4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 19:09:27.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fcd275ce24c4b13974501d841cd38ce : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ffe61f8b2ba41cfbf4a1d2addc301a5 : SUCCESS in 3m 39s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/0633395e8b944d1699f8e1864c92318d : SUCCESS in 1h 22m 25s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/2c14f03f94fe45af8e0b604e4d9d5663 : FAILURE in 1h 56m 00s\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/7d741d7b95dd485ab0bd79f3c61e2689 : SUCCESS in 2h 01m 50s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/e77ee6b9f6ee48dc86cfc3d570d5c382 : SUCCESS in 2h 11m 47s","accounts_in_message":[],"_revision_number":1},{"id":"d76f9d62b2943115673f04f75bd36e416a1c7d03","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-06-23 07:10:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2880c801aebd33ae022333d1152d4f8f6f7b838e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 09:55:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2abac498f3f44544b213ceea018d295d : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8be225d4575c4914b24d16f174488c5d : SUCCESS in 5m 30s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/0666118e443948d19e1515912488c971 : SUCCESS in 1h 24m 05s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/32613d549f564d048ba2d7da320e3134 : SUCCESS in 1h 53m 21s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/872d96e85f2a4021bd993643d003152f : FAILURE in 1h 46m 02s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/780d7b36045e42739d9587ea28c2f3b5 : SUCCESS in 1h 48m 41s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/5002dbc7b91642c6be1e58a7486af826 : SUCCESS in 1h 09m 14s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/be2e2657dc2d471fa000583a9f3a5d70 : SUCCESS in 2h 14m 04s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/4fb09cdc01cc4e0b9d86c171b43c0bd1 : SUCCESS in 1h 28m 39s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/8c4bdab01cdd4077aaf88807e1373ae2 : SUCCESS in 1h 25m 40s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/84d8dd4ee4ea4b28b8f3f7d4b7cb7e4f : SUCCESS in 1h 18m 22s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/40e2d70ed5224e49a06ea3c7d32216dd : SUCCESS in 1h 42m 58s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/2e17669ae5fd4d6ca471a3f4a39a8f09 : FAILURE in 2h 42m 36s","accounts_in_message":[],"_revision_number":1},{"id":"abdfd9df2cec7a20a668aef27295973b8cb550d3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-23 10:15:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1b3d9ebeb8790acd024fedb00ebcb6d2246ddbec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 12:53:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e52bf4ec8dd441ea48740231a35c980 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fe5408df77b412a8a0bd36d5a1bfcbe : SUCCESS in 4m 44s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/96ad9326f719492ab0211d74b4361ed1 : SUCCESS in 1h 29m 53s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/471b812b62c4472ea6b446686945ffac : SUCCESS in 2h 18m 03s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/656b0dc2603744e5b9e0a7b951b8b4d6 : FAILURE in 1h 53m 20s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/b08f6b069de84f49878c51fd9b1e5e49 : POST_FAILURE in 2h 11m 45s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/ca6bc7f65d38411e9a58e36c81599255 : FAILURE in 2h 36m 22s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/956d19f9c6d443c09aa95b44dcac2001 : POST_FAILURE in 2h 15m 15s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/14599aa17a7f47e9aef583cb8844d5a0 : SUCCESS in 1h 29m 15s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/04fc2b1258f2428b96d003111a6911b3 : SUCCESS in 1h 29m 56s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/767fb775a5354674885afb76e338cf95 : SUCCESS in 1h 07m 57s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/366329bcf0104e5fac4e8601bc813cc5 : FAILURE in 1h 40m 15s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/a6df1397036c4c1eaefcd7f68946c41c : SUCCESS in 1h 43m 09s","accounts_in_message":[],"_revision_number":1},{"id":"42cc9fd97c00a6355db9277bf902fce2becf17de","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-06-23 13:44:47.000000000","message":"Patch Set 1:\n\nrecheck\n\nOh come on, what the heck is wrong with this now?!","accounts_in_message":[],"_revision_number":1},{"id":"b994e643920be3948c04ed2c33adb31558b481f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 16:22:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c73a37e96020409495d2975b76764bf6 : POST_FAILURE in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a30b11d2d204f518131f5f6c1e7f081 : SUCCESS in 6m 05s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/99b9980743534ba0b34cd8f0d49f7565 : SUCCESS in 1h 32m 57s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/c08178a49a8f4f7890b634542418ae29 : FAILURE in 2h 22m 57s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/a77a54b9f72e4f259424a3f5b87f65a1 : SUCCESS in 1h 30m 30s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/ceed7ce8e5dc4e5eb04cdefeb0988092 : SUCCESS in 1h 07m 28s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/35b65883a50942bf86cb15048a95ed46 : SUCCESS in 1h 28m 00s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/9f69dfb349344d459d2a6eb51ed5c8d7 : SUCCESS in 2h 29m 22s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/3bd212153009474bb879cf366502e05d : POST_FAILURE in 1h 37m 19s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/b01e254d5ff84a8984c05c278457e146 : SUCCESS in 1h 37m 48s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/0133cca109204354a15cfe9ce2011fb8 : SUCCESS in 1h 15m 28s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/eadc817d58df42d1ac5a6a72a8d52fc8 : POST_FAILURE in 2h 15m 31s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/ac77dea0a66949b28dfc686614d79f62 : SUCCESS in 1h 32m 58s","accounts_in_message":[],"_revision_number":1},{"id":"aeef1741768366d63a5540ff77a80c3191ec166b","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-06-24 07:05:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fea42a35fc352755adfbf5725817cc50977114f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 09:31:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aef167efa8bf4882982826b961e0ca1d : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52eeb2a3d7aa4a8f835936375ed5a485 : SUCCESS in 4m 28s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/ecb4b553e7f5422f99567c1dc08c0d0f : SUCCESS in 1h 17m 44s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/18e2ed1edd6e457fab1b80dab6d97b63 : SUCCESS in 1h 49m 07s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/203d2c3205584561aeadc0afc6ef9d6c : SUCCESS in 1h 26m 19s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/6914d978b9604c2f9937eda7ba74fd80 : FAILURE in 1h 03m 16s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/4de2843bb54c4a00bcf95dd82c21ccc2 : SUCCESS in 2h 11m 19s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/dc61749a116c4b99a02993c4978d4bd3 : SUCCESS in 2h 19m 54s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/ce810966a5874d2c8729a332463ca028 : SUCCESS in 1h 36m 55s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/3f5e56d129c64609b56d1da387b61ce0 : SUCCESS in 1h 34m 09s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/8bd56b5b4f7e4ed09225e943985e15dd : SUCCESS in 1h 22m 50s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/ec61aeecfdb3497487c62491253a302b : SUCCESS in 1h 38m 12s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/925fcd126b364f1aa97178a6b679b706 : SUCCESS in 1h 44m 01s","accounts_in_message":[],"_revision_number":1},{"id":"c8aa3a6287c118bee243fc6fb35c3d86471ee400","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-06-24 09:36:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"03d7b417153b1abb678483af1eb2747970f3e0fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 12:06:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13c45fc14d0c42539f98730093ad00d8 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dd0a6eeeaae4bb4a0637e382f6d2c0a : SUCCESS in 4m 06s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/686fc044e4b74cd7b8feed4b34d7b8c0 : SUCCESS in 1h 08m 54s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/2d99603d24184b1999d0d7105c5842b5 : FAILURE in 1h 30m 38s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/36001e10bf5249be9a81ce921282f6df : SUCCESS in 1h 16m 57s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/a5766a0f818d431d8c8f6be55ac8a98f : SUCCESS in 1h 42m 38s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/bee6a43d093c44ca80463a0a5d450c79 : SUCCESS in 1h 48m 16s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/64acaab11b7e44d78e1a193fbf5e455f : SUCCESS in 2h 22m 50s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/23333f7e3aa3482faadd8689e5d53abd : SUCCESS in 1h 36m 12s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/1c53ca131db84e46a16e8540d6c2bbcc : SUCCESS in 1h 45m 48s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/6d07a7b83af343f69d2149483d0e96cd : SUCCESS in 1h 36m 09s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/a07247a672834202be67c4728b236ea2 : SUCCESS in 1h 24m 11s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/c8e0eed88c8249eeb7cf8c12163d07a6 : SUCCESS in 1h 23m 03s","accounts_in_message":[],"_revision_number":1},{"id":"8a8403217b2ec506b631582d20b5c0c2a884022c","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-06-24 13:05:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a324c7aae93b5212b13ec206a3763fdff81277e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 15:18:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae8bb6ed31be459ba5bf896ad3691f8e : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af5c67ca4ca24e67a5083a3b1093ada1 : SUCCESS in 3m 59s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/cac69d7896b14f72aaa336ad1629f6ff : SUCCESS in 1h 32m 59s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/e4e71b6e8b5d424e8263d72b23d1526d : SUCCESS in 2h 05m 26s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/37b497d406e14703bc27a4fb86b78dd7 : SUCCESS in 1h 18m 18s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/56ae2f7f5aa44fa3842d11bb17415398 : SUCCESS in 1h 53m 59s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/91d5714ec266422da74dc5eb4850e574 : SUCCESS in 1h 44m 17s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/0b007c5b62514353b3d1748711b9bb26 : SUCCESS in 1h 27m 02s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/f75f5444bb9647de978da339105f7515 : SUCCESS in 1h 40m 21s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/695a2cc23bd4463ebc2c43a8a771b04f : SUCCESS in 1h 36m 58s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/e71afc3495724fbebb18df98b07a8871 : SUCCESS in 1h 17m 46s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/f75575d9e2f74ad89d418ce5ba4eeca1 : SUCCESS in 1h 26m 42s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/2e9bc770579641a9bf9282968ad1ceea : FAILURE in 50m 24s","accounts_in_message":[],"_revision_number":1},{"id":"6ee1a4e04049dbf8871ccd2382853256c01ad429","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-24 15:19:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"18e26ba6d6b5c822bd99869f9116963811ebaf38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 18:08:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f90aca40da844735969dab21bdbf7070 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c9fe7d6c3b7447b96be4e56ca5b2de2 : SUCCESS in 5m 54s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/295766bab1fd41018a857ce2d92c2ee3 : SUCCESS in 1h 11m 27s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/0efbf34740634079b1d657d52c5b64e0 : SUCCESS in 1h 38m 34s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/842589aa805145c392555027f14ca3b1 : SUCCESS in 1h 05m 58s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/6cbe1485d2bc4e50a1a84c4fffa570d0 : SUCCESS in 1h 29m 19s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/eb013d0b9c334125bc6dfbb00d3a3d97 : SUCCESS in 1h 54m 12s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/28fe409e005141408de0bf4d8f08c2e9 : SUCCESS in 2h 38m 47s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/f5edb2e626814825af3b9dc5507d0a67 : SUCCESS in 1h 26m 05s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/dfc666176ada4862beeeeb4cb3b98e3a : SUCCESS in 59m 50s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/49c6964813004ba088442b3d8b94362e : SUCCESS in 1h 06m 26s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/b1c1fe8285c843adb4bb122b32308759 : SUCCESS in 1h 20m 44s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/bfb8147255f44409866115223246e1e3 : SUCCESS in 1h 42m 09s","accounts_in_message":[],"_revision_number":1},{"id":"97874780b52dd3ed1373449f53c61541a28b9488","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 18:09:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c88dff94bf81de24dd88c56a1c1fb01f366cf277","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 20:00:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e71c8e7faf3d4c34b335952b03c1fc9e : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/710e1e1219564201826cf49b48fb212c : SUCCESS in 4m 08s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/aa9a74763770415d8de110249c916845 : SUCCESS in 1h 09m 54s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/b3a6cc9d8e6f4bf9b5430cf76d8f1249 : SUCCESS in 1h 37m 23s\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/c7a8de4d5a514b6681c2cdd9046ba7f3 : SUCCESS in 1h 44m 45s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/7958417f2f9f4b1aa1ab0a3ac73c3e08 : SUCCESS in 1h 36m 17s","accounts_in_message":[],"_revision_number":1},{"id":"39d60c6f5d5854609629c15a01a37d9f82d30f9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 20:00:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ee77dbbcb29480bf1ea41fa11b519d411ab342e3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 20:06:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b395188bc0754529b44c2b1207aa9d83 : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a916c0f2eaa5e5ef9a1d37daca26bf45f917e4e8","revisions":{"a916c0f2eaa5e5ef9a1d37daca26bf45f917e4e8":{"kind":"REWORK","_number":1,"created":"2020-06-22 11:29:45.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/60/737260/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/60/737260/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/60/737260/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/60/737260/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/60/737260/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/60/737260/1"}}},"commit":{"parents":[{"commit":"488016d88c2e91ff0bc619671738fefc48f285df","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/488016d88c2e91ff0bc619671738fefc48f285df"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-06-22 11:27:51.000000000","tz":120},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-06-22 11:27:51.000000000","tz":120},"subject":"Workaround missing DIB on stable jobs","message":"Workaround missing DIB on stable jobs\n\nThis should fix Train and Stein jobs failing due to missing DIB.\nProbably we can remove this once fixes on Octavia side [1][2] are\nmerged and ported to Train.\n\n[1] https://review.opendev.org/#/c/736965/\n[2] https://review.opendev.org/#/c/736824/\n\nChange-Id: I8f5a141704c0a7c2095134c4afce2433e83f9415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/a916c0f2eaa5e5ef9a1d37daca26bf45f917e4e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/a916c0f2eaa5e5ef9a1d37daca26bf45f917e4e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
