)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~739341","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~I7f51b0189003d4489a929b6c5ceebb9bd5fbd159","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"udp","hashtags":[],"change_id":"I7f51b0189003d4489a929b6c5ceebb9bd5fbd159","subject":"Skip service_udp_ping until Bug 1886909 is fixed","status":"MERGED","created":"2020-07-05 07:32:07.000000000","updated":"2020-07-13 17:35:06.000000000","submitted":"2020-07-13 17:33:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739341-1594661631101-858d08f5","meta_rev_id":"e6aebc8fd1e3f8b99fb8637a756b0387e9666029","_number":739341,"virtual_id_number":739341,"owner":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-13 17:33:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2020-07-10 14:00:43.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":1,"date":"2020-07-09 06:39:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":2,"date":"2020-07-09 07:07:18.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-09 07:22:12.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":1,"date":"2020-07-13 10:24:00.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-09 06:39:18.000000000","updated_by":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"reviewer":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"state":"REVIEWER"},{"updated":"2020-07-09 07:22:12.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2020-07-10 14:00:43.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2020-07-13 10:24:00.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"},{"updated":"2020-07-13 17:33:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"70de420170d08ff5f51c047d2d62d089c0468dff","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-07-05 07:32:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d383fb1142e260394c9c0f2944f644884ccc5a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-05 09:53:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fb05aeaae91440f964d40bbbec0121f : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f30552e8f43341baac27b0a08778ae1d : SUCCESS in 3m 26s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/5143d8a7c7b74752a53b9fb1b2980806 : SUCCESS in 1h 05m 32s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/bf9306e9060a4f3584bab415fe463039 : SUCCESS in 1h 43m 44s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/c52b544d04e94992b6969368b62eb5ef : SUCCESS in 1h 11m 45s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/828a875983a4401e95d53c59de83216f : SUCCESS in 1h 05m 55s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/6a10fd01e9764bff9a3cc582e8870618 : FAILURE in 34m 47s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/095a488dba3b42feb991bd032b541870 : SUCCESS in 2h 15m 54s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/c0822ad4a0884708ae2cfed1e7a64c4f : SUCCESS in 1h 23m 49s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/cd38588f50ca4d7e8c1e63b5f65460ae : SUCCESS in 1h 25m 04s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/230380c0c73d4cafb5ae1a75642a0378 : SUCCESS in 1h 07m 48s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/b9fcb9507d7d41d98c585144480865d5 : SUCCESS in 1h 15m 59s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/0df08808646d44a181d46be4c4d050b1 : SUCCESS in 1h 27m 46s","accounts_in_message":[],"_revision_number":1},{"id":"6f4b269d4a1fb6149ce4e668fb5c15362cf5d1e6","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-07-06 06:59:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dde767014c784f5c7fd57bcea93ef60ea528da93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 09:23:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da00a390784b49f6a02d88c6c828c602 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80587d3265454979a4c29a9ad81fcc8a : SUCCESS in 6m 30s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/eb75c69bd90844bca8e89caba6cc632f : SUCCESS in 1h 26m 13s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/7d6ea62de0ce4360a27e820233915cb5 : SUCCESS in 1h 50m 42s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/4515945f6cb64d1abfe71e62e27065ac : SUCCESS in 1h 04m 07s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/4ffb1b2ace74429bb41452a4cfc06d7b : SUCCESS in 1h 26m 49s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/738d31c679bf4609b921e42dddb3c666 : FAILURE in 50m 19s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/32bc3bb455014a8481dc332005cfb05e : SUCCESS in 2h 12m 49s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/25d1899e44434cbe968a0ff09ed2e57d : SUCCESS in 1h 24m 54s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/a4d8551c9dc24379a62dbec42ee4ad39 : SUCCESS in 1h 31m 39s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/21d1971b6330428091772efc8e5fcfa2 : SUCCESS in 1h 15m 07s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/9a23050fb9de42c6ac8ce6e2a7099a6c : SUCCESS in 1h 13m 28s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/00519ee76d974f799213daf0b6ca04a1 : FAILURE in 2h 01m 13s","accounts_in_message":[],"_revision_number":1},{"id":"385fb6603302b7221db23002be646a725823a7aa","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-07-07 14:12:00.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3aee043209ef77ff7efd639ddc6177e7cda7f622","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-07-08 10:33:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"559f8000b05bc06815c59db49f665e5cb31dbaec","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-07-08 11:09:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"786e55e3ee13bf4160d8950e9b9d4289c48df56d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 12:59:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/830bc6db5e534bcfb3a968a3d40c80e3 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fea0c8ee0bd4425eaf5749166b116f1e : SUCCESS in 4m 11s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/27de4e59457e42daa1b00d2510210c69 : SUCCESS in 52m 46s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/e2b9f3652bd84079b4286e0bf5f4d97d : SUCCESS in 1h 38m 48s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/00fa8b763d4c4ed1b7973b6acb1d956d : SUCCESS in 1h 14m 03s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/109b7a481a1547389b1e9d337e914bdc : SUCCESS in 1h 32m 47s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/2eeef5ae2aa64c65bf43227b0f3167eb : SUCCESS in 1h 44m 39s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/01c27b8ffe6245edaa71d7f71c97d111 : SUCCESS in 2h 18m 03s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/d986b42ef98a42ec8840a53d66d56946 : SUCCESS in 1h 48m 37s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/d38b06bbd9a34e25b9293ddcd9d6dde9 : SUCCESS in 1h 31m 54s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/9dc9fca31dbd4a50bbbe3e79ab988fed : SUCCESS in 1h 21m 05s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/7b75615e531a4cc58c85c84854a1ee58 : SUCCESS in 1h 18m 23s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/010a51e0df2c4946a31b098c5db03647 : SUCCESS in 1h 44m 02s","accounts_in_message":[],"_revision_number":1},{"id":"2640e5ae54b934fff851ed4dd04f6b02b10e86c2","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-07-09 06:23:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c81b8cb22fb05a7210630e460ee148c1229cc4a","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-07-09 06:23:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"eaac2c1a14aed5fa6dc9049e08c039d39176adc9","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2020-07-09 06:39:18.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"770dd80865f350e753eeb2b4f67e6a8e2ba98572","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-07-09 07:07:18.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"85b3ef0288029ac5cea59c6872cfa6035d8277cb","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-07-09 07:22:12.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b9c7a126a85bad7c963bc69fe89838ab574d8d8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 09:15:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a26b0583138146f1ac49096adc12e518 : SUCCESS in 7m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8edb94f926a4fef9a327e12aa541280 : SUCCESS in 5m 05s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/5ed6582000224a248136d5e68e177158 : SUCCESS in 1h 28m 35s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/4329d97fb0ca49769ecb32b509da10df : SUCCESS in 1h 51m 23s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/731bc3dc113c4ae09f43e1ee3e3526b2 : FAILURE in 47m 16s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/83f4c19ad25f4aa6b8f0228451a3ea95 : SUCCESS in 1h 51m 53s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/fac2a6ce81e24ddb8c2c152b79dfa116 : POST_FAILURE in 1h 51m 17s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/292192044962423c875ac6a1b44e1151 : SUCCESS in 2h 37m 52s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/eeeadd8c83cb4e1e8b80790948e4dc7b : SUCCESS in 1h 22m 00s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/4557a1dff21846aeacbc2139fab19bf3 : FAILURE in 37m 54s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/e2f99edfd19e408487c164027ee140fd : SUCCESS in 58m 00s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/0ea155331de84183bf295eac102c837e : SUCCESS in 1h 02m 39s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/ebf1d9dca65b42f6aa466ff534e29191 : SUCCESS in 1h 33m 59s","accounts_in_message":[],"_revision_number":2},{"id":"b0c2a35e3f9bd47574f625d15384abbc038ac7eb","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-07-09 09:17:11.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f6a54ab1b42ea71cece8b4d4bac3fe7e379712ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 12:03:10.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e7eacaa17e94f12bf89e837f9e2be5f : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce91070a2fe748b28ec56682f9745835 : SUCCESS in 5m 44s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/75c0144ab35f4f3da2cdcd072fabce6d : SUCCESS in 1h 07m 38s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/2c5dff3ff46f475bab931cbda075f90a : FAILURE in 2h 43m 34s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/420e4a82bd23403ebb8af32e42ca8446 : SUCCESS in 1h 20m 50s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/f11f5273177f481daae68796d0c5673b : SUCCESS in 1h 52m 48s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/e76549fa7e0b4a6fa9694b396c5fe848 : SUCCESS in 1h 40m 23s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/e5f11dd2e2f648bf8ca03f46bb047335 : SUCCESS in 2h 19m 34s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/a4f3b6c2836549a0bb158d9ef3798a94 : SUCCESS in 1h 45m 58s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/74e8d7fb5acf4aa5ac0d3e41b32b6785 : SUCCESS in 1h 35m 49s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/c0653fc5106742f8bac177b31476f8a6 : SUCCESS in 1h 11m 12s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/8aae46732d1e42d1adbb572614114297 : SUCCESS in 1h 37m 26s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/05c4c7a5a346480c970bce338be78bd0 : SUCCESS in 1h 33m 02s","accounts_in_message":[],"_revision_number":2},{"id":"d71fd178c69a4abd985ced916ac90883d5d2d1ca","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-07-10 14:00:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e6aae973aa7dc5dff505a6064de6fb6926c128dc","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-07-10 14:00:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d21fd6addc9aca17d888fafd4a435b399e3fd8e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 16:26:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14d2ab2889bd4a59abed56613cb8ae92 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37f344a0e1374a8f9945187cd7c206b1 : SUCCESS in 9m 13s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/1bbd2fe2b3f04fc1b25ef40bad201a16 : SUCCESS in 1h 27m 20s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/9ba770ca71c14ef4aa4fe8f33cbd2c29 : SUCCESS in 1h 41m 03s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/401e57772b914f948429bba8be11d701 : SUCCESS in 1h 13m 26s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/0742d1acd3574599a2774861cd82e33b : SUCCESS in 1h 45m 02s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/3c4e6e09eb744c6e9a299124341ca517 : SUCCESS in 1h 57m 39s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/16efde8d8a5648e4a5929cf2400b60cf : SUCCESS in 2h 18m 48s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/681d27ad068949a882ef5e0e50ede91a : SUCCESS in 1h 33m 21s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/28ed6a70fa5649fc8963bef2bbb1af95 : SUCCESS in 1h 34m 12s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/ef4a62e8704a42e9adae83d33c86e099 : SUCCESS in 1h 02m 58s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/0a8cc6021fb2438089d00a049a7100e0 : SUCCESS in 1h 29m 27s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/ce821e412a844813ba4e71a3450b39ca : SUCCESS in 1h 35m 01s","accounts_in_message":[],"_revision_number":2},{"id":"c24c6642463f9fc382b916cf1963d1800665756f","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-07-13 10:24:00.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1192cb07f36c2c60075674ff831609f6912b9522","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 10:24:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f1ecec8d4b3473ebee3cd10e1fa68cb969b28bbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 12:50:24.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b9b5275d2f34585bf157f580b9a6d71 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f9a62ac161144bebb04772d6f75285e : SUCCESS in 5m 09s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/300f425f69f64ec88ede7c7e401b14d9 : FAILURE in 1h 58m 58s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/60cf66904aec4d3b889e622511361a0f : SUCCESS in 2h 24m 43s\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/d7549fa377dd4701afea2f5f46958479 : SUCCESS in 1h 47m 05s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/49a2b7bcb1284bfd94579ed7e57e7419 : SUCCESS in 1h 42m 19s","accounts_in_message":[],"_revision_number":2},{"id":"e65bbf46a91d327ce58c7c66be72d4de246dd82e","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-07-13 13:00:22.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d8fbbf6d5fe7dbf646a31efbd53b47662aa0bab0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 15:36:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcb0d91845ed4860b4432c82d0ca8c3b : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/019da8efe3954617a8e6058cb99b7131 : SUCCESS in 4m 59s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/7b7bd18dac14428e8f3449685785b7b5 : SUCCESS in 1h 19m 54s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/931ad1537e944e489e4477fae8e185a1 : SUCCESS in 1h 36m 13s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/3ff16d8edd6c467b93ffcec5b4c0da01 : SUCCESS in 1h 19m 16s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/542ca6a9e3714438ae434ee41f68282c : SUCCESS in 1h 56m 23s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/0fbd0106aea1491eb30398cc55e02c46 : SUCCESS in 1h 50m 31s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/dc9eb77fc8a94edd9b9901c26d8b7369 : SUCCESS in 2h 25m 34s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/2d30aa60781c42cf85b46ccbbe3b54f4 : SUCCESS in 1h 24m 18s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/7aa57f2799e54095adaf47fe6877e044 : SUCCESS in 1h 27m 33s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/94bb15efa3f04b23bf5f7c52b73a57da : SUCCESS in 1h 09m 28s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/8ba03b1481964d18b406cce989a4d8c6 : SUCCESS in 1h 28m 42s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/beb2ed87a7a14f2581f7513d0e4b20c1 : SUCCESS in 1h 00m 30s","accounts_in_message":[],"_revision_number":2},{"id":"ecbaa95d5be65689da0c1be99ef2a37b0c3dc4e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 15:36:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"947b9cc96bb6bc83b3a27dc3739eda1365726763","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 17:33:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eec2a63c58f14d8090849f287f2cf807 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6814a78bc46e4894a0fb14f29cc81255 : SUCCESS in 7m 55s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/03d5d7811a294da6bde704609abc3ad9 : SUCCESS in 1h 10m 14s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/5219a579e5de4edd9e7df668466e2cd8 : SUCCESS in 1h 51m 26s\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/82e6227db6ee41f199fde214da425704 : SUCCESS in 1h 45m 48s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/615f5353c75947c0b6ebaf3ca3dee8f1 : SUCCESS in 1h 31m 42s","accounts_in_message":[],"_revision_number":2},{"id":"56110e38121a4c51c48cded642148ba2a8dd28b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 17:33:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"738efa9db41e93f1130f160ecd48496702304b39","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 17:35:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6b374519dfa43b0960f674a7421e62a : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6760376f170412748b966da0121dc1f353628c18","revisions":{"16473c16440491ad7de73baf5036d57e514154ea":{"kind":"REWORK","_number":1,"created":"2020-07-05 07:32:07.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/41/739341/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/41/739341/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/41/739341/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/41/739341/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/41/739341/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/41/739341/1"}}},"commit":{"parents":[{"commit":"8c866223b8b872849da1b6c860b335afeb3b9600","subject":"Merge \"Retrieve KuryrNets CR when supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8c866223b8b872849da1b6c860b335afeb3b9600"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-07-05 07:30:46.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-07-05 07:30:46.000000000","tz":180},"subject":"Skip service_udp_ping until BZ1846189 is fixed","message":"Skip service_udp_ping until BZ1846189 is fixed\n\nChange-Id: I7f51b0189003d4489a929b6c5ceebb9bd5fbd159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/16473c16440491ad7de73baf5036d57e514154ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/16473c16440491ad7de73baf5036d57e514154ea"}]},"branch":"refs/heads/master"},"6760376f170412748b966da0121dc1f353628c18":{"kind":"REWORK","_number":2,"created":"2020-07-09 06:23:23.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/41/739341/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/41/739341/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/41/739341/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/41/739341/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/41/739341/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/41/739341/2"}}},"commit":{"parents":[{"commit":"8c866223b8b872849da1b6c860b335afeb3b9600","subject":"Merge \"Retrieve KuryrNets CR when supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8c866223b8b872849da1b6c860b335afeb3b9600"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-07-05 07:30:46.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-07-09 06:22:47.000000000","tz":180},"subject":"Skip service_udp_ping until Bug 1886909 is fixed","message":"Skip service_udp_ping until Bug 1886909 is fixed\n\nLaunchpad: https://bugs.launchpad.net/networking-ovn/+bug/1886909\nBZ: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1846189\n\nChange-Id: I7f51b0189003d4489a929b6c5ceebb9bd5fbd159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/6760376f170412748b966da0121dc1f353628c18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/6760376f170412748b966da0121dc1f353628c18"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
