)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~742209","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~Id303620090df194297e16384d3522c3e3bb2ca58","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"np-crd","hashtags":[],"change_id":"Id303620090df194297e16384d3522c3e3bb2ca58","subject":"Adapt NP tests to new CRD format","status":"MERGED","created":"2020-07-21 14:30:07.000000000","updated":"2020-07-30 21:45:46.000000000","submitted":"2020-07-30 21:41:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742209-1596145315762-c9d102ff","meta_rev_id":"f3a196c2f21ec6002edee52f6c246db6540eab85","_number":742209,"virtual_id_number":742209,"owner":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-30 21:41:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2020-07-30 15:34:38.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":2,"date":"2020-07-30 08:26:33.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":1,"date":"2020-07-30 11:38:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":1,"date":"2020-07-30 15:34:38.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-22 07:14:33.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2020-07-30 08:26:33.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"},{"updated":"2020-07-30 11:38:47.000000000","updated_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"},{"updated":"2020-07-30 21:41:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b4d97cff387876944bb9f80a836b616a5f455c70","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-07-21 14:30:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"548477b224c17454475eeeacc8a08e86143bbe02","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-07-21 14:33:32.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"87b31da8935cec614c509148514e10edacc58940","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 17:15:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25116c2561ae476a9d65815bebd22c8e : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1cf5e2ae0e848b4a3f355a060303fcc : SUCCESS in 5m 01s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/bbe37b123d874ce082be291decaa60f5 : SUCCESS in 1h 21m 58s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/072a1fed9375413993ece50e29c2e01e : SUCCESS in 1h 48m 09s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/18f4748956ff421c9fde9fc85d927fde : SUCCESS in 1h 22m 26s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/1df4c0d9661a4875800a6fcdf2410188 : SUCCESS in 1h 52m 03s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/61d17fd1625e40378e07527a99752bc3 : SUCCESS in 1h 40m 16s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/10c7ee4c650c464e93aa21818fb71319 : SUCCESS in 2h 33m 54s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/6781ff58615a44d98965af77b5a4fdc3 : SUCCESS in 1h 39m 36s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/af02a59808b04d18a2ce7f6eac3c9d5e : SUCCESS in 1h 24m 52s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/70d77501199a42bebd05a65fffbad97c : SUCCESS in 1h 45m 39s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/6993f2111e4e4021ac2b3ba307f1fa1d : SUCCESS in 2h 26m 15s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/e75369dad1504486aefbf802fe8db452 : SUCCESS in 1h 12m 00s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/b8aebb03e2c14d5a97bf526c83357aad : SUCCESS in 1h 26m 57s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/4b5e52e146c94e629bd75f216477953a : SUCCESS in 1h 36m 18s","accounts_in_message":[],"_revision_number":2},{"id":"cb99ac39c614c38accd29d34da9e7b3c3c11e13b","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-07-22 07:18:15.000000000","message":"Patch Set 2:\n\nWe can see this works just fine on https://review.opendev.org/#/c/720454/ too.","accounts_in_message":[],"_revision_number":2},{"id":"477be55e8203d7680ffdbc038a12f66cc9d61834","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-07-27 10:55:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cf89e4ab9b330a52d639259d40c0e28a6ba40c99","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-07-27 10:58:31.000000000","message":"Patch Set 3:\n\nI updated the tests to correctly do sleeps on each iteration. That should solve the issue of tempest logs buffer being saturated.","accounts_in_message":[],"_revision_number":3},{"id":"afbd7b5a82bc4e1e169cc6bfe32feaf2cf1b66b6","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2020-07-27 13:09:05.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"42454cea10a735e5885f6d3a0661e48fc913244b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 13:26:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb966b824d1d4dcc9a15017eef8c43c5 : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3b8982c4ecb447586b413f640dec4ba : SUCCESS in 6m 15s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/6ff2c2ab81994865a5e686256aa2f551 : SUCCESS in 1h 26m 57s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/59e404dfdcac45339fe8ec169660d12e : SUCCESS in 1h 43m 08s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/15732d1fbc014012bf90efb93deafa36 : SUCCESS in 1h 18m 28s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/d65669af7d2b40c89683fe189e2fced1 : SUCCESS in 1h 37m 58s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/81b4839289384e75bdc462fe81b8f870 : SUCCESS in 1h 39m 26s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/1c8105f29b9249ab85ec09760017e94a : SUCCESS in 1h 42m 38s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/f3ef46f911f341d9926a936b8cae996e : SUCCESS in 1h 21m 59s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/5595596f44d24be185b241e8096723a1 : SUCCESS in 1h 28m 47s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/d9f6229b472343648d4b29e16f631892 : SUCCESS in 1h 48m 58s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/16d58c91016846b6acc002a651b76ddb : POST_FAILURE in 2h 27m 56s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/196c1a54a0e245918e4d789ec268908c : SUCCESS in 1h 13m 47s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/f080dd8eb9c040e6ad7284d5c1a34ce0 : SUCCESS in 1h 28m 20s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/ee16e0c42204461ea786c985c0887f75 : SUCCESS in 1h 52m 50s","accounts_in_message":[],"_revision_number":3},{"id":"5855ad404ba34d7fc6780c3b46edc92bd3de25c9","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-07-30 07:47:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a34b1f76748426aed16c2df890b611cc44a8b388","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-07-30 08:09:20.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)\n\nGreat working on supporting both version! Just a nit:","accounts_in_message":[],"_revision_number":4},{"id":"702cced9360712a9f09f14cbd3293e15a14cfdb1","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-07-30 08:25:51.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"878b374383c93e31db665bebccfa418ece2d0783","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-07-30 08:26:33.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"21427c34fe1aecb4ffc33053628831d3b78e52c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 11:08:27.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d60bf92a6904cf99ae0a68e38b44922 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cf1efd242c34574a6cec52a741befe4 : SUCCESS in 3m 06s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/2a6f9f8b267f435c834f239dc1a7de24 : SUCCESS in 1h 21m 19s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/12406d50854f49dcb30891b9edca46f2 : SUCCESS in 1h 25m 37s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/96a77aa825e74de28357083c92985321 : SUCCESS in 1h 34m 02s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/e277e86dcc3149b6b6df5bd082777051 : SUCCESS in 1h 34m 13s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/06a767ca327e43e9a85ac72b8e350aea : SUCCESS in 2h 00m 04s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/b85d5f18ee62473aae80c71901c8d659 : SUCCESS in 2h 23m 53s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/6a042f372f71452b87834a7c53b3c47e : SUCCESS in 1h 29m 09s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/3edfdf6340d04925bbcbcc7bd9b00483 : SUCCESS in 56m 36s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/bcf42b0cd4da42d1b4d952e2203898e4 : SUCCESS in 1h 39m 05s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/1b0bab71289e4c8c94f66c00d96228b0 : SUCCESS in 2h 32m 05s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/d7ddb8bc8f914b7babeedf884a92a0b4 : SUCCESS in 1h 05m 12s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/42bfa806cd0244b3a94edf4a6a6522d9 : SUCCESS in 55m 12s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/a05cd66210ca4a1f818b603e7b69a17e : FAILURE in 1h 17m 42s","accounts_in_message":[],"_revision_number":5},{"id":"b21477343eecdf705cd46d785d6d3eecec461e0b","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2020-07-30 11:38:47.000000000","message":"Patch Set 5: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"9b3549ae03bbaaa8e5e740c4c272304e45529148","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-07-30 15:34:38.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"dee82e639c9da2e0a28d9ff9dc219a4aa4ff2699","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 17:32:07.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/345b167beef942719677bad94838d273 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7820c630be94c45b7f4629ef8a129bd : SUCCESS in 3m 28s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/f2704b535b254c3abded3ff052132a7b : SUCCESS in 1h 04m 41s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/7e118e3419af4a8c905b06cbd8da62e5 : SUCCESS in 1h 44m 04s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/7a7d69e05d3d4e61be2baa87b12f8bcf : SUCCESS in 53m 41s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/e6c1068ed23049deaae431020b6bfad2 : FAILURE in 2h 10m 09s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/a3721185364a409197cef4a4d1739517 : FAILURE in 2h 02m 57s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/709a584016214719939965a6c950dd20 : SUCCESS in 2h 27m 15s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/f60dad53dca44d0b8a1a8c85c5e123fc : SUCCESS in 1h 30m 43s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/2ad259af4b154c248482246c5447d1a4 : SUCCESS in 1h 06m 02s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/120f90ddbe664c8ebddd8e2fae9e6ad7 : SUCCESS in 1h 40m 10s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/5e77c52b04494c83b21b1862b86fa9e1 : SUCCESS in 2h 39m 25s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/7423f30739d24d979a04503846c46d69 : SUCCESS in 1h 15m 13s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/760ecd11184a432fbe31206d2d75be1f : SUCCESS in 1h 15m 00s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/ebffb6ca58ee4908966f4ac9ff09bcc9 : FAILURE in 41m 00s","accounts_in_message":[],"_revision_number":5},{"id":"e55b92723a42ca1811457804a25c8695a3756b6e","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-07-30 18:00:03.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"25ed65431041efeb87a1e938d6eb1f2b6ce8181a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 20:11:00.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f3fa80a89ec44659f2d2b48229d5f84 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9198efa3c564e469158850d8e3586db : SUCCESS in 2m 53s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/d3e521f7b8974d9badae228b5aa9bf1a : SUCCESS in 1h 10m 47s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/c838e02d08324b5fa0473e349fa4f6fa : SUCCESS in 56m 35s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/8657f30c8a314107a5d0b4dbc0fd541b : SUCCESS in 1h 07m 52s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/af31f51deb1b4a21bff5c0c3968a752c : SUCCESS in 1h 06m 06s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/57af6f9021384ed89ce770a6faf3c7b1 : SUCCESS in 1h 24m 22s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/aa684b19b8d245d4a0e502cd1ac86e43 : SUCCESS in 1h 24m 51s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/1b2245d197a1468895d7987469627732 : SUCCESS in 52m 08s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/dff0d854528c4e3c8a4e6c8a3e4284a7 : SUCCESS in 1h 08m 57s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/c4c766a2a2da4785bd43c5cbf67eab95 : FAILURE in 1h 13m 43s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/2a69243249b046fb9d7e6876f4eab4ce : SUCCESS in 2h 06m 08s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/bc68d454686547b49c0144dbfd4ec0a8 : SUCCESS in 1h 13m 07s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/fae173e1262a4012878cf0ac595a5a0f : SUCCESS in 1h 28m 24s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/8e95bec9c89c43f1b6e941b78d5a1317 : SUCCESS in 1h 18m 33s","accounts_in_message":[],"_revision_number":5},{"id":"f6489a46b9ff5a73fb5d4dc5fee8124d41811625","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 20:11:11.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"d09b78f69c9e1b9a1063fabdcba4c696f03eed03","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 21:41:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"bfe9f3842be22a24f627596e48154e1741aa8e61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 21:41:55.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01d6ee0d3cd74126803e022e6401343a : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9f528158cbc40fe8afb309a628a5fd8 : SUCCESS in 3m 06s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/8401c135b24a4533a13f386c0e0a97f1 : SUCCESS in 1h 02m 13s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/6888c49b10fc477bb177119fa29f95fb : SUCCESS in 1h 22m 28s\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/358822844827443c922fc5b1f5af9b8c : SUCCESS in 1h 12m 57s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/d0be2cf921064cc69db412be182d9f3d : SUCCESS in 1h 07m 39s","accounts_in_message":[],"_revision_number":5},{"id":"926c8d96c08cd7480a7b5bfb5b31070cda003818","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 21:45:46.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb6a87cc8ed24b19b5235e5ce93a1d73 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"5dc54def33d179b6c0b72a5c538409256b8ba303","revisions":{"b5c3a64a57d33d1684590a9477b2326f6903c385":{"kind":"REWORK","_number":1,"created":"2020-07-21 14:30:07.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/09/742209/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/09/742209/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/742209/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/742209/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/742209/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/742209/1"}}},"commit":{"parents":[{"commit":"8c866223b8b872849da1b6c860b335afeb3b9600","subject":"Merge \"Retrieve KuryrNets CR when supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8c866223b8b872849da1b6c860b335afeb3b9600"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-07-21 11:55:06.000000000","tz":120},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-07-21 14:15:22.000000000","tz":120},"subject":"Adapt NP tests to new CRD format","message":"Adapt NP tests to new CRD format\n\nChange-Id: Id303620090df194297e16384d3522c3e3bb2ca58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/b5c3a64a57d33d1684590a9477b2326f6903c385"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/b5c3a64a57d33d1684590a9477b2326f6903c385"}]},"branch":"refs/heads/master"},"ce763838c8ecdf4915b5514bbe6f900b805129f5":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-07-21 14:33:32.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/09/742209/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/09/742209/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/742209/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/742209/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/742209/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/742209/2"}}},"commit":{"parents":[{"commit":"8c866223b8b872849da1b6c860b335afeb3b9600","subject":"Merge \"Retrieve KuryrNets CR when supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8c866223b8b872849da1b6c860b335afeb3b9600"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-07-21 11:55:06.000000000","tz":120},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-07-21 14:32:34.000000000","tz":120},"subject":"Adapt NP tests to new CRD format","message":"Adapt NP tests to new CRD format\n\nThis commit adapts network policy tests to new KuryrNetworkPolicy CRD\nformat. This includes splitting TestNetworkPolicy class into two\nversions and a bit of refactoring to allow that.\n\nChange-Id: Id303620090df194297e16384d3522c3e3bb2ca58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ce763838c8ecdf4915b5514bbe6f900b805129f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ce763838c8ecdf4915b5514bbe6f900b805129f5"}]},"branch":"refs/heads/master"},"27a64b6a640fcd3796ec0b31b0eb7a751bfd3a20":{"kind":"REWORK","_number":3,"created":"2020-07-27 10:55:06.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/09/742209/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/09/742209/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/742209/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/742209/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/742209/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/742209/3"}}},"commit":{"parents":[{"commit":"8c866223b8b872849da1b6c860b335afeb3b9600","subject":"Merge \"Retrieve KuryrNets CR when supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8c866223b8b872849da1b6c860b335afeb3b9600"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-07-21 11:55:06.000000000","tz":120},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-07-27 10:54:57.000000000","tz":120},"subject":"Adapt NP tests to new CRD format","message":"Adapt NP tests to new CRD format\n\nThis commit adapts network policy tests to new KuryrNetworkPolicy CRD\nformat. This includes splitting TestNetworkPolicy class into two\nversions and a bit of refactoring to allow that.\n\nChange-Id: Id303620090df194297e16384d3522c3e3bb2ca58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/27a64b6a640fcd3796ec0b31b0eb7a751bfd3a20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/27a64b6a640fcd3796ec0b31b0eb7a751bfd3a20"}]},"branch":"refs/heads/master"},"f43ad36209aa8b53abcff27a9c5e56eb950025af":{"kind":"REWORK","_number":4,"created":"2020-07-30 07:47:47.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/09/742209/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/09/742209/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/742209/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/742209/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/742209/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/742209/4"}}},"commit":{"parents":[{"commit":"8456f2fe16d0239095d06143733c8a3ae8f43eac","subject":"Fix test in order to work with new kuryrport CRD.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8456f2fe16d0239095d06143733c8a3ae8f43eac"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-07-21 11:55:06.000000000","tz":120},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-07-30 07:46:40.000000000","tz":120},"subject":"Adapt NP tests to new CRD format","message":"Adapt NP tests to new CRD format\n\nThis commit adapts network policy tests to new KuryrNetworkPolicy CRD\nformat. This includes splitting TestNetworkPolicy class into two\nversions and a bit of refactoring to allow that.\n\nChange-Id: Id303620090df194297e16384d3522c3e3bb2ca58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/f43ad36209aa8b53abcff27a9c5e56eb950025af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/f43ad36209aa8b53abcff27a9c5e56eb950025af"}]},"branch":"refs/heads/master"},"5dc54def33d179b6c0b72a5c538409256b8ba303":{"kind":"REWORK","_number":5,"created":"2020-07-30 08:25:51.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/09/742209/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/09/742209/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/742209/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/742209/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/742209/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/742209/5"}}},"commit":{"parents":[{"commit":"8456f2fe16d0239095d06143733c8a3ae8f43eac","subject":"Fix test in order to work with new kuryrport CRD.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8456f2fe16d0239095d06143733c8a3ae8f43eac"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-07-21 11:55:06.000000000","tz":120},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-07-30 08:25:42.000000000","tz":120},"subject":"Adapt NP tests to new CRD format","message":"Adapt NP tests to new CRD format\n\nThis commit adapts network policy tests to new KuryrNetworkPolicy CRD\nformat. This includes splitting TestNetworkPolicy class into two\nversions and a bit of refactoring to allow that.\n\nChange-Id: Id303620090df194297e16384d3522c3e3bb2ca58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/5dc54def33d179b6c0b72a5c538409256b8ba303"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/5dc54def33d179b6c0b72a5c538409256b8ba303"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
