)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~744115","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~I2cb66e25534e44b79f660b10498086aa88ad805c","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"kuryrport-scheme","hashtags":[],"change_id":"I2cb66e25534e44b79f660b10498086aa88ad805c","subject":"Support for vifs being in \u0027spec\u0027 and \u0027status\u0027 objects.","status":"MERGED","created":"2020-07-31 10:14:34.000000000","updated":"2020-08-18 11:32:40.000000000","submitted":"2020-08-18 11:31:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744115-1597750278461-4ababe2a","meta_rev_id":"bcb80698f9864251cdb98682c6a7100fb729ca6a","_number":744115,"virtual_id_number":744115,"owner":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":30963,"name":"Sarka Scavnicka","display_name":"sscavnic","email":"scavnicka.sarka@gmail.com","username":"sarka_scavnicka"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"date":"2020-08-18 04:57:29.000000000","_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-18 11:31:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2020-08-17 06:47:31.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":1,"date":"2020-08-17 06:47:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30963,"name":"Sarka Scavnicka","display_name":"sscavnic","email":"scavnicka.sarka@gmail.com","username":"sarka_scavnicka"},{"value":2,"date":"2020-08-17 06:47:31.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-17 06:47:31.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":30963,"name":"Sarka Scavnicka","display_name":"sscavnic","email":"scavnicka.sarka@gmail.com","username":"sarka_scavnicka"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-17 07:30:39.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"_account_id":30963,"name":"Sarka Scavnicka","display_name":"sscavnic","email":"scavnicka.sarka@gmail.com","username":"sarka_scavnicka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-31 10:18:32.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2020-07-31 11:02:41.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"},{"updated":"2020-07-31 11:08:48.000000000","updated_by":{"_account_id":30963,"name":"Sarka Scavnicka","display_name":"sscavnic","email":"scavnicka.sarka@gmail.com","username":"sarka_scavnicka"},"reviewer":{"_account_id":30963,"name":"Sarka Scavnicka","display_name":"sscavnic","email":"scavnicka.sarka@gmail.com","username":"sarka_scavnicka"},"state":"REVIEWER"},{"updated":"2020-08-17 07:30:39.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2020-08-18 11:31:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"271ee76914a4b2108b608a9b6aa43a719f01fbc1","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2020-07-31 10:14:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f58b47749722bb4efa4a3d101935310d6ddf53af","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-07-31 10:18:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c1fd46ae20ebc45ab737860b8e61c729fa095cc0","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-07-31 11:02:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"162aef5362c1b1aa562273a8844ed35e019d6375","author":{"_account_id":30963,"name":"Sarka Scavnicka","display_name":"sscavnic","email":"scavnicka.sarka@gmail.com","username":"sarka_scavnicka"},"date":"2020-07-31 11:08:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fe585d15587586e3fd85fbdc247b0db0d0ac242f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 12:42:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dd969aa728044928df0ee49009c3007 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5dd6f79c53c41649e4aea91e649f60e : SUCCESS in 3m 51s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/2d35788e2eb94007b83a52322c44bbf8 : SUCCESS in 1h 06m 35s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/1d61df551ee6423d8a7c1130d3f8b77d : SUCCESS in 1h 22m 28s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/092b6ffc6d014e57980548876ab52d88 : SUCCESS in 1h 15m 32s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/08237b82a0ab46c483ec043f8f184938 : SUCCESS in 1h 19m 54s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/252aeabaf9454954a0b3affc60a3ca2b : SUCCESS in 1h 55m 48s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/8c4bdb21561a4f5bb07df02718a8902c : SUCCESS in 1h 33m 59s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/40afb3ea5aeb4934926121df04b6b17c : SUCCESS in 42m 00s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/6ef03cdeec784bc09c29a5589ef611cc : SUCCESS in 58m 15s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/186cc413896b44879e79530947dbb264 : SUCCESS in 1h 15m 30s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/ea7e4653c20e4a4794214108eae13bf7 : SUCCESS in 2h 22m 26s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/cf1acd8cba264fd48e0b8ba8addac1a3 : SUCCESS in 1h 27m 41s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/64fe3ec0e47b484d9226b9775e8d0969 : FAILURE in 1h 29m 47s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/717eabfc661c4fec8170c627407f7d78 : SUCCESS in 1h 15m 03s","accounts_in_message":[],"_revision_number":1},{"id":"153c3666bab671a52549c60d5f3bb98673fe1d39","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2020-07-31 13:11:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"da127390962ee0a72fb79f55b7f2fa36cb91268b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 15:19:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee49946fd77f4ebaabee2507817970ce : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8a8a58a1d4748a5a9eb5edb335fd15d : SUCCESS in 2m 58s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/c9c036733ca941bd9319f104c12e838c : SUCCESS in 1h 14m 37s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/17d55353928d46bb9015b7414870200f : SUCCESS in 1h 19m 31s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/367d860fdbca4d11ac3688f08aceeedb : SUCCESS in 1h 06m 02s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/a0f608244524409fba1cc17ffbd23daf : SUCCESS in 1h 18m 52s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/435bd625d0b749d7b066cb50ff7ecd97 : SUCCESS in 1h 33m 11s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/9ac3ab1af3c94838a7be420a97c2745c : FAILURE in 1h 34m 46s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/ad290b0daf354f1ba5d20a5da4a02f72 : SUCCESS in 54m 45s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/916e838c63a54d8fb0183235b87af964 : SUCCESS in 58m 45s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/ff3e1f30ac6342afa75e11becdb72d91 : SUCCESS in 48m 04s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/7344f12034b3475d9fad91cee8b1b353 : SUCCESS in 2h 02m 25s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/fab4950f5f6a45af86b29dbd63a6cafe : SUCCESS in 1h 17m 01s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/faf85908902346e694c610614d7d43ce : SUCCESS in 1h 40m 56s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/ff5792ef8fca49cda2135b12d001ae2f : SUCCESS in 1h 19m 40s","accounts_in_message":[],"_revision_number":1},{"id":"41d05d6f6f15283de048aa118ba0a7689520a44c","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-08-10 09:00:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"874500c158fc426e6e370987f0e5962be02d98ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 09:00:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"70a7fb2773999c8d2594a9c750b7479ebb7c732b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 10:48:53.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e3d002d80a8468b87485e215720cc03 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b34b88f2d6148b3a3cefce52478e192 : SUCCESS in 3m 59s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/1170a196cc724d3ea0f733187f941ef5 : SUCCESS in 1h 06m 44s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/b9a242a131b14fd39bb37c1c69af5960 : SUCCESS in 1h 16m 24s\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/aa83109b9c6949c9b14db9fb776be0bf : SUCCESS in 1h 04m 25s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/8bd3ac1717234ceab809567d89f25865 : FAILURE in 1h 41m 06s","accounts_in_message":[],"_revision_number":1},{"id":"63b62dc0bf4582bc502233f58c3949df005d5b48","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-08-10 10:58:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"37ef443028fe479420cf7b753c431c9ce371aaf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 13:26:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2f4070c63f24173aa864ca3d5448249 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6f70cf59b4f44818492080c01a3dcfb : SUCCESS in 3m 24s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/a7df62b9d8ea437cb47236f661c951ab : SUCCESS in 1h 10m 46s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/42e3dfef78a24566a7c3c5fe19d4fc0c : SUCCESS in 1h 25m 18s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/6c35f2e4fd5d494e88b8b5b3335566e8 : FAILURE in 2h 19m 58s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/d3abb0874d994a11822f00b3e0433262 : SUCCESS in 1h 22m 10s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/b497351f633a42dc83543f33891669c0 : SUCCESS in 1h 47m 44s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/52f56ad6e2e4466fa465b0826fef6a83 : SUCCESS in 2h 03m 31s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/956b4534a9eb40af82ceeb47e458cae1 : SUCCESS in 1h 08m 03s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/9184dcda935f4397803194bd24374f25 : SUCCESS in 1h 09m 23s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/d6bf2bf8261c42a7bcf12acd7cc427f3 : SUCCESS in 1h 13m 15s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/92d4aa45183a4d26b1dcb780e4dbbc6d : SUCCESS in 2h 22m 52s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/05ff06b04e3a4e07adc8307de54b855f : SUCCESS in 1h 22m 40s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/3b3b9463d24b4b68925d8e92ad0acda7 : SUCCESS in 1h 36m 23s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/cc499fc2782744be8ebdbcceb859491b : FAILURE in 1h 12m 16s","accounts_in_message":[],"_revision_number":1},{"id":"513a7ad970dd04da4e26896f98fe82a9ca10ec62","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-08-10 22:33:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"079b873836323adfd77fc8ed77fdff73e24b5e3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 00:40:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab1e651b05c94cf798d9754ed0c4c2da : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b8e7c82ce234a8ba96ad21662597df4 : SUCCESS in 3m 16s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/163603b1579e43ccb4fc22ed94f06da3 : SUCCESS in 1h 20m 18s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/df12aef983f34a1ca9f0ca368d6b0b46 : SUCCESS in 1h 20m 04s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/b248d3c1d9e249d29c7534480c4501f6 : FAILURE in 1h 32m 33s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/ec967c676f6c428ca6959ccfbea23cca : SUCCESS in 1h 35m 32s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/4df7226179404c198a016a1d75e1da16 : SUCCESS in 2h 06m 09s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/46438bacfa8e46569b4fe1de24912fd9 : SUCCESS in 1h 30m 59s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/36009029334648a096d57b170f83cdee : SUCCESS in 1h 01m 16s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/5b85fbf73273414e86ea83f0d6593624 : SUCCESS in 1h 02m 58s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/79da2477d8b141a6a09d7a689adffbef : SUCCESS in 1h 30m 21s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/1f620172e728487b9b76ae263c80a4ff : POST_FAILURE in 2h 00m 05s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/bfc8756f04994ccba0e41ac405261d6b : SUCCESS in 1h 14m 51s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/7132190d25274a25a4f87d85778ca2df : SUCCESS in 1h 12m 47s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/ff4cb4f620904a308e2a9fe101349432 : SUCCESS in 1h 56m 02s","accounts_in_message":[],"_revision_number":1},{"id":"bf57bdf492bd2430c5a06f40d594b52d7894512b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 00:40:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b84c4dd250bc9988e829662d967ee0e3dc8d83d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 02:29:33.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b52f78d1a1954cd4bd84cb1d15b85990 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70f24ee7d0d845e88a32a4cd70b559b4 : SUCCESS in 3m 24s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/b2ca691283b948338e2d648ae34625d0 : SUCCESS in 1h 09m 28s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/4fa50f569c3b4c14971378f691c27a66 : SUCCESS in 1h 23m 56s\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/9121ec621b6e464a8cceae7c7cef3300 : SUCCESS in 1h 24m 47s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/e19a415b9ea54651ad81e44ae5cfb03a : FAILURE in 1h 25m 32s","accounts_in_message":[],"_revision_number":1},{"id":"8d97fa4b9c9a89986e8fa337bb79efe0f185e1c1","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-08-11 06:50:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"014202f0e2f9bd189f9824961c2122f02b195a87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 08:46:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f405416f6e246afbe9b7b4dc3e719e8 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a0552ffd4f84ed29476c2a7cc0654b4 : SUCCESS in 3m 13s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/c20e8c085eb946669447949cfb5540d3 : SUCCESS in 1h 23m 17s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/66f966fefdc8445ba329851c972a9eba : SUCCESS in 1h 24m 34s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/01e5c2d83c6a4108bbccb877459d4c30 : SUCCESS in 53m 08s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/cbdd8cca538e4500b1b5ece0e515c244 : SUCCESS in 55m 08s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/d1e1f2a0e1704b52900e80c810436173 : FAILURE in 1h 12m 22s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/88975ee8f63244d8a4dabc411b6f32d4 : SUCCESS in 1h 27m 05s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/34b5faeb40f44a5e850f8ad2da29318e : SUCCESS in 1h 04m 20s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/8992f9d0402048128f0c1243ac14a501 : SUCCESS in 57m 10s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/81e8d92331724d2aa2aae365ac291c44 : SUCCESS in 1h 05m 52s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/a8e830283ab342ca95114bad9ca34db9 : SUCCESS in 1h 48m 10s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/5006e6464ec94a2baf1954e17231d596 : SUCCESS in 47m 32s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/37715cc6dbdd4672b455f5c51166b4d0 : SUCCESS in 1h 21m 17s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/bbe371fae7df40aba00410589f3d2653 : SUCCESS in 1h 20m 08s","accounts_in_message":[],"_revision_number":1},{"id":"554d374bbbf913f91fcc9f0a8029f2ba0a08cbaa","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-08-11 12:37:34.000000000","message":"Patch Set 1:\n\nrecheck\n\nThat CI issue will be handled in another patch, it\u0027s related to NP code.","accounts_in_message":[],"_revision_number":1},{"id":"5c81d9f5af8b6e856ff93462ca602f90e57458d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 15:00:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b6d9e68da234f819fc4b4256197407f : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78f5d58e571a4fd0907a3e3e4064ffba : SUCCESS in 4m 42s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/83d4e15f4b04469dbe750c807694e9e6 : SUCCESS in 1h 48m 01s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/a318e331734b473a93fac05135e64b7f : SUCCESS in 1h 37m 41s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/effd794ea98342e89da3fc30dea6d9a3 : SUCCESS in 1h 07m 59s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/65615aecb7314d0fbb9e376c4062b1df : SUCCESS in 1h 32m 52s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/c6be56c93a2a4c2590573bdf110b7148 : FAILURE in 1h 24m 37s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/789a910e98d74af597ed6d4cc14c1f94 : SUCCESS in 2h 13m 58s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/f89ba013fe1246c893770db22736284a : SUCCESS in 1h 05m 11s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/9fa35cdf823344ddba822b1f0c9f7637 : SUCCESS in 1h 15m 48s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/3f634d2c53ba4a6dad928f710991b578 : SUCCESS in 1h 19m 23s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/5b848f9a483044c9b11d240ef4c34c7e : SUCCESS in 2h 06m 15s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/c615d9f60b9f4ba9b2589a29a1e085f1 : FAILURE in 1h 24m 28s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/65d0c57c77474b8c8e1fd77319223d70 : SUCCESS in 1h 22m 54s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/de3c905bd80d4253b0de2ea86d52d36f : SUCCESS in 1h 41m 07s","accounts_in_message":[],"_revision_number":1},{"id":"7cd4a1e21d5b62fc5aad61f61231421682805d02","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-08-12 15:42:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8920daf1ee59205057689be3a2e7c75207fee55b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 17:55:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf17bd9cf24d4ee789cc8ea1cccd3247 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cf19008159748999d6c7b44d1899779 : SUCCESS in 6m 07s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/aae5f9cf04de4688937b54b26220803c : FAILURE in 28m 11s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/df9424eb29064250899a262cb1fe65ba : SUCCESS in 1h 27m 24s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/55679ee23c2b4da9bc87b30b89b1a0ed : SUCCESS in 1h 23m 56s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/eac8c66bb3a94eb78e981d7e06c55ab9 : SUCCESS in 1h 27m 50s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/b0ae6593a18948d2abd68e6aba007e2f : SUCCESS in 1h 30m 09s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/eb81fdb2a46343819b99defbdbe37289 : SUCCESS in 1h 37m 38s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/2a65375ad84e45548653d34db2bd5447 : SUCCESS in 1h 13m 13s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/3b2adc722aaa454386a18b00bcc07880 : SUCCESS in 1h 15m 50s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/43d30023f60a437ea99690827e0cd5f6 : SUCCESS in 1h 23m 03s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/2ea8647236c741878c203394e2afe783 : SUCCESS in 2h 04m 53s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/0dc0d672cd2a4898b16c974c3bf019a0 : SUCCESS in 1h 10m 28s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/85dc4b4570b64da0a071bb67a1030511 : SUCCESS in 1h 13m 56s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/99cb753dd37e402eaf288faec0fde6a1 : SUCCESS in 1h 20m 59s","accounts_in_message":[],"_revision_number":1},{"id":"2b7fb14b71b59835b499626d183abd8bad086199","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-08-13 07:07:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1215c384f6d32ae80e4a2f00189ce087d83eba7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 09:35:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5a7d5b25c614a7e92bb7496207989cc : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56626ed1d5554ffb8d89bc45291c6728 : SUCCESS in 5m 27s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/5a2318ecae8c4eadbd1698ed5823398a : SUCCESS in 1h 41m 55s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/440d608638a84913908904c8ae5e73c8 : SUCCESS in 1h 14m 19s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/6eb0628299d64dbdae3c812294729a3d : SUCCESS in 1h 24m 52s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/0005984629c34a1da7961db2d6cd17c7 : FAILURE in 43m 02s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/0cef61534ea6450f9689100ef913b302 : SUCCESS in 1h 20m 41s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/45cff39aa09a49a69887b90f97eef448 : SUCCESS in 1h 37m 23s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/96013af7cbf64de5b3f44120f09908c7 : FAILURE in 1h 07m 07s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/925440cf09154ef5a1a581cb82b5f47e : SUCCESS in 1h 29m 09s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/9ff84292236646468ce6e9ac24c80839 : SUCCESS in 1h 24m 48s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/a442a64121094542bbff5f7d5af092ea : SUCCESS in 2h 11m 31s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/a77673dcfc824afdb18a6e93d0b5e834 : SUCCESS in 1h 23m 37s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/4e6c6d81464941da9d0708343e0265a3 : SUCCESS in 1h 45m 21s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/cebb131a154e4a18a18cf63fe7dc9441 : SUCCESS in 2h 04m 40s","accounts_in_message":[],"_revision_number":1},{"id":"75fdf413988241a707d435e9a6120112fdd8c2fb","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-08-13 11:29:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"320da776663988f2e58a24fa966343a630beed03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 13:42:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e41dfa4441244df9c79741afe171114 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67439df363ef4b2fa7319555fc095600 : SUCCESS in 3m 16s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/5ed483426e8443548c4cc17b855a76cf : SUCCESS in 1h 36m 09s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/b9d16c6da91446bba19cf2afd66f1be0 : SUCCESS in 1h 22m 57s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/523e70c3358d44b78c0f2174e3983b94 : SUCCESS in 1h 30m 22s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/58991c4bf6e74dfe82ceb9500b813415 : SUCCESS in 1h 35m 27s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/9b2ecca8a56744eabdb7fee3f15eabf0 : SUCCESS in 1h 21m 50s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/bba734de05374cfcbf2b9bb14fd63dc8 : SUCCESS in 1h 42m 43s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/f0e8d7e1ce1148db8f70753428e137aa : SUCCESS in 1h 08m 22s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/2541ee39fc0c4ad4933d3cabfd0c8b91 : SUCCESS in 1h 09m 11s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/a0df7a4059764b1cb2526640a2e3e170 : SUCCESS in 1h 21m 27s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/afbde79940e842e0bb720e4303357017 : POST_FAILURE in 2h 08m 18s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/e131c42af6444655870d6a24db6fa13c : SUCCESS in 1h 01m 34s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/7fd2725638ee460999640f7a1172d3c6 : SUCCESS in 1h 18m 48s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/40a35f241de44c4a92d85211cfc799d9 : SUCCESS in 1h 25m 01s","accounts_in_message":[],"_revision_number":1},{"id":"be918aaf9f7c873d2aa5b737c9ffe5c236ff0bf0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 13:42:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"629dd16b048e1e65507d451003edb6fbfa96744d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 15:34:38.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3c9082b83e24428bb1cdd96b224de65 : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afdbf39834d842b591252a7ed53b2155 : SUCCESS in 4m 35s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/6b214ca1591d457ba0dd2e3cf567df76 : SUCCESS in 1h 15m 58s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/8b300044e85742368c200903405a2d4d : SUCCESS in 1h 44m 20s\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/8e5f6562a1214e6399189864dc768438 : SUCCESS in 1h 48m 35s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/4ab890586e8a4d77ba80d696c5e7dcf8 : FAILURE in 1h 35m 05s","accounts_in_message":[],"_revision_number":1},{"id":"5c23ae7df997772ad02240c60ff2eb7b5e0c7417","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-08-15 08:24:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"33f9119225a35dc159896d51150473401ec59d13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-15 10:19:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2f4ba386f5f4e8db926a62c8abe3beb : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1813354595843ef8bc875798c6c7e66 : SUCCESS in 3m 05s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/d592a73745464279b055232620c8c5f8 : SUCCESS in 59m 08s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/57c62605725343d0beefd3b952b3080e : FAILURE in 16m 19s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/04b0e93ea313432babf21b45ee3b019c : FAILURE in 15m 48s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/6d5dd0e3cc86452e88c181877acd7098 : FAILURE in 16m 03s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/5bfcd4f1aea94d558214324cba5abe7a : FAILURE in 17m 16s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/4604bf033c054f1f89505d6f59821d17 : FAILURE in 22m 10s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/48d02f05bb2040b9821e02a4aaf77306 : FAILURE in 15m 18s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/4080258b2dc643c4bee9d1c05efe710b : FAILURE in 16m 56s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/8e30fa1ee92a4a4785cb409a422f2840 : FAILURE in 16m 15s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/cc2ab19b3dd94484bf57bc27f6c6be22 : POST_FAILURE in 22m 57s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/722b9db5fb5d4f97956f0661c34781e5 : SUCCESS in 1h 10m 13s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/bc1e1589ef4c4923af364977614b32bb : SUCCESS in 1h 29m 05s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/78a11e3eb26644b282ca0acc08101de5 : FAILURE in 1h 43m 16s","accounts_in_message":[],"_revision_number":1},{"id":"e9bf2a19d069721dd1cbd40396c30c20035b9a09","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-08-17 06:47:31.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"7533da7a26c07bd379af206dee0df32e8a02a958","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-08-17 07:30:39.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"15be8e137769f6207120f61964be74dfb1bc4e22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 08:47:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31cfbdbc33cf4558bd9acdf621147f1f : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dc8a057cda147e5987dfd11a42efd4b : FAILURE in 3m 54s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/aaf2ceb998924d86bd804045541c5067 : SUCCESS in 1h 24m 40s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/2c6ee3020b86421caf94a2eed660891a : SUCCESS in 1h 21m 16s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/e5ebfd2f5e71463880c32732571c6a4c : SUCCESS in 1h 18m 19s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/7604d56c261649bbb5cb70250cc1ecc3 : SUCCESS in 1h 12m 11s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/787659765e10479992888d1a39929359 : SUCCESS in 1h 21m 06s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/1ac17c1d3a664f9ba67f02ed85364ca7 : SUCCESS in 1h 26m 30s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/7450cddecd95411483b53842db190c0d : SUCCESS in 1h 15m 30s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/4e3e3f08850f456dbbedad47a07f0730 : SUCCESS in 1h 22m 22s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/b2d1e238022f4dcb8c4d0a287e09d058 : FAILURE in 57m 56s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/3c7315ff646e47469c8bc61af0f8959a : SUCCESS in 1h 52m 24s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/5fb62fa88f0345448901c8a9d58ef52b : FAILURE in 19m 21s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/060eebb197094afcbb67f18479cacffd : SUCCESS in 1h 25m 08s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/337a8b8c2ea04de78bff167b189119f7 : FAILURE in 20m 36s","accounts_in_message":[],"_revision_number":2},{"id":"60290c59ad512140ec7f0cdf9271436e78d0b159","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2020-08-17 11:00:15.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e55929f1fb593ed011ba669d0919b6f9bfa1f306","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 13:34:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35c96d31674848fc9b417642a688df3f : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe4b5393bed144dfb49e30db07e99b0e : SUCCESS in 3m 20s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/bfcad2e0d995459fa4ed2225e3152951 : SUCCESS in 1h 29m 38s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/0d6d1e14757a44908474ebc52574cc53 : SUCCESS in 1h 26m 54s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/72844ad88ced42699897b30925efce91 : SUCCESS in 1h 40m 42s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/46bee35c26964cbfb6d7e77d5b6cfdbe : SUCCESS in 1h 40m 42s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/b3b533964aef4ec28baa6e791e6481b7 : SUCCESS in 1h 26m 53s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/4680c6fea45746ca8abc7ea74c000b8b : SUCCESS in 1h 31m 14s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/e32a6c886fd24065b75e31512362f097 : SUCCESS in 52m 37s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/1230cf480e6c426f9b49999f503d8a02 : FAILURE in 1h 42m 54s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/c462c519b3784da38d2a4a17a58ee10f : FAILURE in 41m 54s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/eb866b1b36934f79bd00b929f701b292 : SUCCESS in 2h 31m 46s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/916fd05013cd481d9d0b0e47ade7472d : SUCCESS in 1h 04m 38s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/b5629763346e443e926778b33c89b165 : SUCCESS in 1h 43m 37s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/cec3951d956143219ccf61461dfd0d71 : SUCCESS in 1h 19m 35s","accounts_in_message":[],"_revision_number":2},{"id":"c1c9c68a412c164902d1be4cffb68df53c764f98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 13:34:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8a62e6340bde77e8aa3e759c955666f8645025ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 15:25:12.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/415bd438da9f495aac01174f2874f223 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9ea973ee8304f5cae6095993963229d : SUCCESS in 12m 45s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/c306a68f561b4e3c9844e52c18731560 : SUCCESS in 1h 13m 33s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/606d597b604a4f59ad79255d6115011f : SUCCESS in 1h 41m 00s\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/c809ab4fd3c8432b886b761dd9fc5622 : FAILURE in 1h 32m 06s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/190ebbb5acc44501ab1e75a6178e408b : FAILURE in 1h 25m 11s","accounts_in_message":[],"_revision_number":2},{"id":"19776d535d58d1db4fb9fbda75c5aa01c8d6ac41","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2020-08-17 15:33:29.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5019a75e52ccba457471f09cfa2070b2c1ecbd2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 18:08:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a50024004bd4b60a7f44da2d5b10ca1 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7ceaf25efe64f19b740c440255d916f : SUCCESS in 5m 32s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/a291ac32eec54c5eb47bcbd5611fe831 : SUCCESS in 1h 06m 22s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/35fe96f9549a4ba38aed8fd5c86a1c7d : SUCCESS in 1h 10m 24s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/6ab90ff03884440289539803c12d2630 : FAILURE in 18m 31s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/68924c0e45084558a7d5a7cd667e2072 : SUCCESS in 1h 18m 30s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/5a28320c5ef748fc8158287427526942 : FAILURE in 17m 40s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/8d291c47d7114b34a8fd65e8a75935a4 : SUCCESS in 1h 20m 00s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/a4c89c94c3534779ba44ac17e063ea37 : SUCCESS in 1h 07m 21s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/265b55b2833f41b6ac25bd9389c1e4b5 : SUCCESS in 1h 14m 54s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/26efbeaaf6a74c77b7db26fcbf68653b : SUCCESS in 1h 26m 09s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/8d003b18ca2844d5a56f19829da0381a : SUCCESS in 2h 16m 53s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/7a727c828ec84ba2b1dcd71b00866721 : SUCCESS in 1h 17m 35s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/4df3dd492c6549cc83b130a4e126677b : FAILURE in 1h 54m 10s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/2713ade409ce46a49e65097b96cc8a87 : SUCCESS in 1h 59m 36s","accounts_in_message":[],"_revision_number":2},{"id":"e0a1c55151ee3bf77643ec095c67790e19a06751","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2020-08-18 04:57:29.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7c51ce4eaec01df2e54d59ed849fe6c6c0ee94bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 07:30:02.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81f87ff887464765af00ddb0f1e81fd7 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7583d9b52734c45ad1af869f3db3139 : SUCCESS in 3m 55s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/76e9e3c1f75e4310bbcfaa2df87ebce3 : SUCCESS in 1h 30m 07s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/621eafdce72944be80be80f7d1c46fe6 : SUCCESS in 1h 53m 10s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/2adc758a43a3462f809ca1683c31ef68 : SUCCESS in 58m 26s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/58d84f04fc67482f9fdc18881fccc4c6 : FAILURE in 2h 06m 52s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/c1ce26985cc5400aaa5a11037837cfe4 : SUCCESS in 1h 26m 52s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/9fd49451a3f74a0a9cf0eeb38463d02b : SUCCESS in 2h 21m 08s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/c4b988762dec4fb582a2a0f794f517dc : FAILURE in 1h 55m 33s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/1b310a2ac8d34b44ba1c34b3d68219df : FAILURE in 2h 00m 46s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/63b91bea556f49d1b027f62949529bbb : SUCCESS in 1h 55m 05s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/09da8d7883be4b619d7725ea1a2e5461 : POST_FAILURE in 2h 30m 50s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/44846e14d0344a33a2aff8c35154559f : SUCCESS in 2h 01m 25s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/12574aee31c5437e888a566e8f081a70 : FAILURE in 1h 33m 30s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/3db03db61b70415eb9a8c4ba8324f0e3 : FAILURE in 1h 52m 24s","accounts_in_message":[],"_revision_number":2},{"id":"b06f00ef122740b750ffd4e7610de4ae09395399","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-08-18 07:42:02.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d93c81711630818b9e3ee6444be27f27c574ec36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 10:02:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c35bbdc6997d41de84a0dc24165a9799 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4efd698deb2b435198d3d8af28088426 : SUCCESS in 4m 18s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/a2bcc77187c042aaa35daf2871b2fbc1 : SUCCESS in 1h 19m 48s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/ee806e7ad7be452994a8a4ec2bed33a4 : SUCCESS in 1h 02m 54s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/9192f36ef9e94985a9714ec746844db0 : SUCCESS in 1h 24m 42s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/0b564ae0fadc438ca42fbdcec2f3080e : SUCCESS in 1h 20m 32s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/b2b361fafe274e9e89910dd611675881 : SUCCESS in 1h 05m 27s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/ff810d95ecb3483cb3a6c55a925a1e0a : SUCCESS in 1h 26m 06s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/40cc49becb944338a0eb03b42416395e : SUCCESS in 1h 16m 17s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/f1df055719ab4567a76303a530b554b1 : SUCCESS in 1h 15m 20s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/5a50dc9d5c1348609c669eb3809972b8 : SUCCESS in 1h 17m 37s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/160e519a661849afa490e709dc849627 : SUCCESS in 2h 00m 36s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/12c7700ee78d4876a384eb8bfb0768a6 : SUCCESS in 1h 08m 31s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/fc50d3792c12488b8cadb934e8f3f006 : SUCCESS in 1h 10m 18s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/63bb706d9b744ac7963476f533e069af : SUCCESS in 1h 17m 33s","accounts_in_message":[],"_revision_number":2},{"id":"9b7cedf6e0092bb8775d33be0f6ec5c249e7d030","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 10:03:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c4256e7e79745ff51bdd92c5eadd09271f5e50ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 11:31:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2dc74833e9f0fd5de5229c8811d86a60f387057e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 11:31:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ce835bb3f394f4bb480a65c2bac7b94 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c06e8d851024aa1913f736e9f4290b7 : SUCCESS in 3m 13s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/f3e773c5d17f40fe9d106973f7b10d9c : SUCCESS in 1h 04m 58s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/880db26839d34c59bc731e762399a88e : SUCCESS in 1h 21m 52s\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/e5fc4c228284447a97443f6a6bb6fc91 : SUCCESS in 1h 22m 27s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/c7030fe703464fd485166fd6b783f8c3 : SUCCESS in 57m 10s","accounts_in_message":[],"_revision_number":2},{"id":"5d1b1911d01a4f8427855663903dd001471cd892","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 11:32:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d4e5258369d4515a30f22fa4ef76a8b : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"53f01c482c9fc0440a7ae92b235ae7fcf43857d1","revisions":{"183a5ae81f31b1ab6cf35072d00511d0b8a26281":{"kind":"REWORK","_number":1,"created":"2020-07-31 10:14:34.000000000","uploader":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"ref":"refs/changes/15/744115/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/15/744115/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/15/744115/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/15/744115/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/15/744115/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/15/744115/1"}}},"commit":{"parents":[{"commit":"2b388b8a938d185339830131ce6efb9f4d5aa7c1","subject":"Support KuryrLoadbalancer CRD","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2b388b8a938d185339830131ce6efb9f4d5aa7c1"}]}],"author":{"name":"Roman Dobosz","email":"gryf73@gmail.com","date":"2020-07-31 10:10:35.000000000","tz":120},"committer":{"name":"Roman Dobosz","email":"gryf73@gmail.com","date":"2020-07-31 10:10:35.000000000","tz":120},"subject":"Support for vifs being in \u0027spec\u0027 and \u0027status\u0027 objects.","message":"Support for vifs being in \u0027spec\u0027 and \u0027status\u0027 objects.\n\nIn order to move vif information from \u0027spec\u0027 to \u0027status\u0027, we need\nto support both for the transition time.\n\nChange-Id: I2cb66e25534e44b79f660b10498086aa88ad805c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/183a5ae81f31b1ab6cf35072d00511d0b8a26281"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/183a5ae81f31b1ab6cf35072d00511d0b8a26281"}]},"branch":"refs/heads/master"},"53f01c482c9fc0440a7ae92b235ae7fcf43857d1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-08-17 06:47:31.000000000","uploader":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"ref":"refs/changes/15/744115/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/15/744115/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/15/744115/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/15/744115/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/15/744115/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/15/744115/2"}}},"commit":{"parents":[{"commit":"2fe1791ff2e6283c7f8d92e10fcaffa104457a95","subject":"Remove skip of test_service_udp_ping","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2fe1791ff2e6283c7f8d92e10fcaffa104457a95"}]}],"author":{"name":"Roman Dobosz","email":"gryf73@gmail.com","date":"2020-07-31 10:10:35.000000000","tz":120},"committer":{"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","date":"2020-08-17 06:47:31.000000000","tz":0},"subject":"Support for vifs being in \u0027spec\u0027 and \u0027status\u0027 objects.","message":"Support for vifs being in \u0027spec\u0027 and \u0027status\u0027 objects.\n\nIn order to move vif information from \u0027spec\u0027 to \u0027status\u0027, we need\nto support both for the transition time.\n\nChange-Id: I2cb66e25534e44b79f660b10498086aa88ad805c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/53f01c482c9fc0440a7ae92b235ae7fcf43857d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/53f01c482c9fc0440a7ae92b235ae7fcf43857d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
