)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~744432","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~Ia28cf8b9c27560f4672a3bc46fd8a12eae790a51","project":"openstack/kuryr-tempest-plugin","branch":"master","hashtags":[],"change_id":"Ia28cf8b9c27560f4672a3bc46fd8a12eae790a51","subject":"Remove skip of test_service_udp_ping","status":"MERGED","created":"2020-08-03 09:44:05.000000000","updated":"2020-08-11 11:19:39.000000000","submitted":"2020-08-11 11:18:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744432-1597144685342-88ec3396","meta_rev_id":"4ba1af2affda795add1082189efa102116eb403f","_number":744432,"virtual_id_number":744432,"owner":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"date":"2020-08-11 05:19:24.000000000","_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-11 11:18:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":2,"date":"2020-08-06 11:50:49.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-10 06:52:41.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-10 06:52:41.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-03 12:59:46.000000000","updated_by":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"reviewer":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"state":"REVIEWER"},{"updated":"2020-08-04 12:58:29.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2020-08-06 11:50:49.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"},{"updated":"2020-08-10 06:52:41.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2020-08-11 11:18:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8557aff8ebc2124567839b4baa48f30175db995e","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-08-03 09:44:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf5338504af90d0de49fb16e304db236ea54b5b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 12:06:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c8b101fffa34e3c85b40636b417a849 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bdb3775f1564aed9be0ed68a6f4d66c : SUCCESS in 3m 10s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/a4fc386ea49140d28f0ee7f0d00eb465 : SUCCESS in 1h 07m 36s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/5d967a5a95114e97897ca8f897e3fa99 : SUCCESS in 1h 17m 43s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/2891ad8c95944e08b451b16d7199236e : SUCCESS in 1h 05m 13s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/3535807530a94b2b968bceebeab38c08 : SUCCESS in 1h 16m 56s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/bb778ffc0d4f4df6a00c42398f0c13c4 : SUCCESS in 1h 22m 04s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/308d362486864b6fa38abf8440cdd62e : SUCCESS in 1h 59m 57s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/698e270687a24fab9f1ea1551c9f76d3 : SUCCESS in 1h 07m 07s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/9963b14486564882b71b0b7dc5ef2969 : SUCCESS in 1h 11m 23s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/db381fe7b9e74bca818e4df7da3e041b : SUCCESS in 1h 11m 58s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/2677b7f908c84bdbb58d97faa7d322b5 : POST_FAILURE in 2h 06m 53s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/e3ffc7fa95d240ba88683e4ccd0332ba : SUCCESS in 52m 27s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/17812fe4576445ca8d37f4ee349d9e19 : SUCCESS in 1h 12m 12s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/f0994616369548c48dfc29c7ea4323cb : SUCCESS in 1h 25m 54s","accounts_in_message":[],"_revision_number":1},{"id":"06e0bc6c26b72a2be0f601922def7f07c9f34911","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-08-03 12:58:31.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5b11a89f52ad718d5f4a3dbcbae02f058d218e9d","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-08-03 14:48:33.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5f545488f56364184e316d8090e2cc2790d7d8cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 15:12:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e46483e23f944e42b4574df1ee681eb3 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b57cd52adfc4748bb72d7bd3f66e90d : SUCCESS in 3m 43s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/ed1af0eaec4a4ae49741fa95e7f92b1f : SUCCESS in 59m 28s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/33c76727433940df9e13d034098f31c7 : SUCCESS in 1h 17m 49s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/810fc240b0fc4c17a51a9f60f699d8b6 : SUCCESS in 1h 23m 23s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/03855a2b7b3447b586ae9b58e5408cfe : SUCCESS in 1h 04m 33s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/761a3c0722394b5c953b900a6c0ed50e : FAILURE in 1h 19m 38s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/54b41c2b342043a1999b73b060f78b43 : SUCCESS in 1h 13m 02s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/b7193b243e2447b1999843cf7fbb51e8 : SUCCESS in 1h 28m 31s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/f999d3ea88f443e689b928998f9a4876 : SUCCESS in 1h 09m 06s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/3aed9f0d2f2e4acc9f6b81ae6497dc68 : SUCCESS in 1h 37m 27s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/16fefeab01f94cba9c83904647c9d1c1 : POST_FAILURE in 2h 00m 21s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/33db63e503c54d24855e15b3e9780dd2 : SUCCESS in 1h 19m 05s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/070dbd1dff274adcb4a09a2d42a873d6 : SUCCESS in 1h 23m 51s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/497e62f3f9a541198c79ab7c0fbfa264 : SUCCESS in 1h 57m 47s","accounts_in_message":[],"_revision_number":2},{"id":"65b3da6657f27a60fc8184d1a199b1e7fb2718c8","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-08-04 11:36:42.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"526fe1a62071cc27585d658f31e07fdb3f1e9842","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-08-04 12:58:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nSeems like containerized-network-policy gate is still unstable, probably as a result of one of CRD patches.","accounts_in_message":[],"_revision_number":2},{"id":"92338b4f1dd0d6a6445d56a5f76fa7762c0dd52c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 14:00:43.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfd8197458204fa88ff57f2678ae64f5 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0318b7a2e3f64c5e9f57b2b7f988b42e : SUCCESS in 3m 33s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/bcbc29c84b83438c8b8033cdf6428e98 : SUCCESS in 1h 10m 18s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/27d2cc20e05143ec8be7ce609c694ec6 : SUCCESS in 1h 05m 17s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/054bd3dd467446998db7fa69c6d1a690 : SUCCESS in 1h 46m 13s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/a29b4e48bd9c42b6b3608bb135ac7c19 : SUCCESS in 1h 06m 56s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/5436bb446b1247dba0a592686e943397 : FAILURE in 1h 15m 24s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/9fefbbbb60354cbda40eea923c91905e : SUCCESS in 1h 25m 08s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/68477fc0409549508a6b9f87b2a32ce6 : SUCCESS in 1h 03m 24s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/13dec98bd97747c4a9f50c8cb27a80c7 : SUCCESS in 1h 06m 17s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/a0b78f24d37343ea9511c914feaf18bf : SUCCESS in 1h 15m 15s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/50e02fb14e12456699569a2d3a0ae9c4 : SUCCESS in 2h 17m 25s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/152284b40a984815895102ee4cfd5b85 : SUCCESS in 1h 10m 42s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/4bddd72667914b89b312888636a35ef6 : SUCCESS in 1h 28m 05s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/c1a07528066445eaa347982f8e132314 : SUCCESS in 1h 47m 57s","accounts_in_message":[],"_revision_number":2},{"id":"4d79c5e209603e611e0cb8043ab7f9d8d0baba80","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-08-04 14:46:56.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3666a0ed51418c47c5a7d82d67d0da4f526c5eea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 16:57:39.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fe9df14b1f9403187c7dd0a26028717 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12c797f9ecb3438e877e20960975aa9b : SUCCESS in 6m 01s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/33a407f552bf4bc18719022dfab158df : SUCCESS in 1h 08m 34s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/19274410768a43d499efecdc8e176a92 : SUCCESS in 1h 46m 35s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/4dc435032ab14c34af92a952d5ea5c74 : SUCCESS in 1h 35m 08s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/4041a8d0e2764282bbe835896a2e920a : SUCCESS in 1h 42m 36s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/44b9d14a161d44dbbd02978462331eb9 : FAILURE in 1h 35m 14s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/9bf7c42d25f0450b905611f468703913 : SUCCESS in 1h 32m 29s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/d25e45190f6047cd9a5503550d8e22e6 : SUCCESS in 1h 13m 28s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/4a9585d78c6548ff85a62654382d07ac : SUCCESS in 1h 10m 54s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/1885b0e85a1247ab85d1e3dc3b875136 : FAILURE in 1h 20m 12s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/ce92e34671544933bca21f91bca3c1d0 : SUCCESS in 2h 07m 35s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/9e1360a2a07a4345ae2eb165b447425c : SUCCESS in 1h 00m 31s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/406832f0c2204113a419d1cef11e3c51 : SUCCESS in 1h 20m 38s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/2fee5e63ef364cdb8e2630b4686add94 : SUCCESS in 1h 47m 48s","accounts_in_message":[],"_revision_number":2},{"id":"d4cf4eda434668d1377bbae0f8d956d8b99ec8ab","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-08-04 17:10:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"af215db33611749e1b7b5b667a4f36aac036b26a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 19:17:06.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b113c1a7d39445ca7fb536aebc15866 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93639a76fd0446e2b3a742131ce4dd0c : SUCCESS in 3m 47s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/f36b5a985f574188b95d416e08b813ed : SUCCESS in 1h 17m 25s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/469f28adf7874ac5a06fc34341157234 : FAILURE in 1h 45m 00s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/35f8f6eebf9b46c0900939c71621ec06 : SUCCESS in 1h 24m 37s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/697a0e410ba14f0fa5d7c62d460276f6 : SUCCESS in 1h 28m 41s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/4f6e06fa8e2c4e4782fe59e9d4214fb6 : FAILURE in 1h 38m 02s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/e43fc825d0b346bf99d26e350b954885 : SUCCESS in 1h 49m 17s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/71bfa34dbe434b02a6c819bff5a85b3d : SUCCESS in 1h 17m 43s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/64acfd0705004641a92b2d2b8e82d17b : SUCCESS in 1h 15m 59s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/328b3a535b50427da76d0cf2fdeab4d2 : SUCCESS in 1h 11m 01s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/33a98b01253648859d45442117b32e1d : POST_FAILURE in 2h 00m 32s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/7f917fb703d742d2ba606813dd7430c4 : SUCCESS in 1h 23m 09s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/2fa7aaf5030540efa176b8e9d41c039d : SUCCESS in 1h 34m 01s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/eaa4b795bc0b4518af65a4d9724c616c : SUCCESS in 1h 05m 14s","accounts_in_message":[],"_revision_number":2},{"id":"7a83a5ef4e3273ef4169ed696714c008a5273be1","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-08-04 19:31:55.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a162c41058cdc02f58cff90ae8bd5eda9fdee883","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 21:35:54.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7518c7cddb6e42c984e042116a6f2dac : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2771752e33324221a2a405b48531c549 : SUCCESS in 3m 09s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/0e2966a2b8334c668118da09db30a8b4 : SUCCESS in 47m 07s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/4fcdc377d4424a6bbcdc032a962361b5 : SUCCESS in 1h 22m 33s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/ece295f6b5f440cdb8f61ea178cb7dcb : SUCCESS in 1h 18m 39s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/e952f7b42da64421824742461089c30d : FAILURE in 1h 31m 46s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/c711ecda869943a38ee4b4221a66edad : FAILURE in 1h 12m 46s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/9094828fc6f64ca88112593f0a3e4c64 : SUCCESS in 1h 31m 24s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/9725b6855eea4363a4ba44df6e0ffb90 : SUCCESS in 1h 10m 50s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/d990137253674deeb5b08b8425eac077 : SUCCESS in 1h 08m 02s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/c7298623b8ad4531887f75ff88f6d6ef : SUCCESS in 1h 34m 49s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/91dd3e73942148879ecef34f71b0a1cf : SUCCESS in 2h 00m 15s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/e918dc30bee9461e81dd3a197e38761c : SUCCESS in 59m 30s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/af958ef949a14a87bc77a9d5e1940ebc : SUCCESS in 1h 09m 05s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/c698f3adb18e48708c8fbc9d7c746b17 : SUCCESS in 1h 23m 56s","accounts_in_message":[],"_revision_number":2},{"id":"241eb6fdbf2772623c017ddab8ed0353f5e3919e","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-08-05 10:37:26.000000000","message":"Patch Set 2:\n\nhttps://review.opendev.org/#/c/744880/2 should help with test failures. I hope!","accounts_in_message":[],"_revision_number":2},{"id":"ebb2d44f49f070a1c90b9b0c18a3820600baf26c","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-08-05 10:38:02.000000000","message":"Patch Set 2: -Workflow\n\nHm, but I see the test still failed the OVN gate: https://2cc2ebcbe02fad10dd3c-ba27e931a99f05bd6f205438b3cd6a3a.ssl.cf5.rackcdn.com/744432/2/check/kuryr-kubernetes-tempest-containerized-ovn/e952f7b/testr_results.html.","accounts_in_message":[],"_revision_number":2},{"id":"a2164e37772ac37ca16cf5d4a8fb33155cc4fea7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 13:10:15.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/482054ee050040ca9e664c0520613772 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2918a12dbb6f43b98546359f4c144793 : SUCCESS in 4m 32s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/fc638c495bd2405d81294a69ab6d5c01 : SUCCESS in 1h 14m 55s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/1607079fda2b466b898ca7e8988ed48e : SUCCESS in 1h 13m 51s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/4099320c1c7e4c3a851f0a13230a5a23 : SUCCESS in 1h 36m 09s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/5206d50c87ed4ba0b561a12da0c6075f : SUCCESS in 1h 39m 36s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/55c70dc7564f46ffae32c5d2f457d6fe : SUCCESS in 54m 54s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/6746a238f111428985df348a1c6e736c : SUCCESS in 1h 16m 13s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/947764333ca7493d835e9c78506d05f9 : SUCCESS in 1h 10m 33s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/c7dc57a00edd4295978a6264d2fcec02 : SUCCESS in 1h 33m 35s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/c214a72f4da44aaa83b3cf8884e0fbc5 : FAILURE in 1h 15m 01s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/ec7cfd00c3d34da896bcdec4344f4c6a : SUCCESS in 2h 27m 27s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/927f0afdbeb244ef92028e15266abc6e : SUCCESS in 1h 32m 25s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/610402942df545bc95163af163dd68a8 : SUCCESS in 1h 38m 56s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/83c9094d9c0b40dea231ad3fac7da46c : FAILURE in 1h 39m 21s","accounts_in_message":[],"_revision_number":2},{"id":"2843a384360b4b24067da0df49650eb579bbf75b","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-08-05 14:58:19.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"0fed2f770e4c5e1c6a2691beb3818af617e80ad6","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-08-05 14:58:50.000000000","message":"Patch Set 3:\n\nLet\u0027s see if it will help.","accounts_in_message":[],"_revision_number":3},{"id":"d4068d2e4c6b334afabbc674d9025aa8030acf00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 20:15:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a81ab4c8cea34eaba4d76096d3f8d3b4 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/237070ae7bb746aaa7230baedf84e1a0 : SUCCESS in 4m 05s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/898eaea86a4046ba9e51405f38a955fe : SUCCESS in 1h 15m 27s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/a2bdcb121dbd4a1a98cfad6205f4cb32 : SUCCESS in 1h 04m 16s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/7243528e466c4d9ba60de644bc6f0d31 : SUCCESS in 1h 10m 42s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/a61c2f8ef66c4620aabd9ff4c5719fb7 : SUCCESS in 1h 13m 57s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/95798dd6a2e1420580056982cb9873e0 : SUCCESS in 1h 23m 35s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/804bfdcf20634b07b38a77f157d9532c : SUCCESS in 1h 59m 55s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/af054e9c5f1c4fc0a0f9f9af9431d9af : SUCCESS in 1h 24m 32s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/013d6be46ecb420b9ee23677b86b441f : SUCCESS in 1h 12m 26s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/0e854bb9af184ff288222cdf0fd7cefd : FAILURE in 1h 15m 46s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/99380973929443169729ad959ff53ad5 : SUCCESS in 2h 10m 43s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/9d7881fa169448ea8684697900802bb2 : SUCCESS in 1h 41m 15s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/30c560da45a8445198636a2491588f34 : SUCCESS in 1h 35m 20s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/581535eea8fb420ea67e5cc06b87e478 : SUCCESS in 1h 24m 04s","accounts_in_message":[],"_revision_number":3},{"id":"33f568622f692f8c9baa1572f396fb91af4c85e6","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-08-06 11:50:49.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f8f8c723d38d5136744afa40285a9319673867c1","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-08-10 06:52:41.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5ede01d016886b1ea393cb01301be464370235da","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-08-10 22:34:08.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"bcf17f67ed69f322846972582d40514fe4a82fd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 00:22:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adf28d86515b4a97819b51d24ccaab06 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe5f1c4c61c949f4bbc94ecf609d2986 : SUCCESS in 3m 27s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/9cc65806297440c7a063383cc716b1b6 : SUCCESS in 55m 21s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/5b7aa423776e415e9d99ddfbe9f87e4e : SUCCESS in 51m 46s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/d063bd3582af422ca76c143eb245db48 : SUCCESS in 1h 00m 53s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/eacb1b8b08d34f31bfb210d484711335 : SUCCESS in 1h 07m 23s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/84bfb320e8794c609efcc22c7208ced2 : SUCCESS in 58m 39s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/cd0031bb5ddc41099f8b0e9f2f446d81 : SUCCESS in 1h 14m 00s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/b4c0f9a5cabf422f817328576957b3de : SUCCESS in 49m 26s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/9af4c1f79e0b4389ac671f1ab91c5159 : SUCCESS in 47m 30s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/fe821c84c1214d1d9b013b6d2b5181a0 : SUCCESS in 50m 26s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/353c395b24164012a98b93f3118324ec : POST_FAILURE in 1h 46m 47s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/d64219787395471e9ea21d6584631178 : SUCCESS in 46m 50s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/5e5c471ef5e8419290f082e6bd23c7a7 : SUCCESS in 1h 00m 09s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/6a2d3067077d44f282fa8b5e8ca08101 : SUCCESS in 1h 09m 29s","accounts_in_message":[],"_revision_number":3},{"id":"3da2e43795b1acd71dfb79ccfd4d06c3c8cb62a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 00:23:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"37343cc55e430202a1f64059638c2198db63ea04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 01:52:59.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e861a8588c494304a7c5411cdaad8a24 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/746caef4b4a54002830ace76f90806ea : SUCCESS in 3m 20s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/8f0c7f5888c8412c9ad1c9121881d7e3 : SUCCESS in 1h 06m 43s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/7f5dd45c651e48058f2a55ef326309b6 : SUCCESS in 1h 28m 31s\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/284de59f230440b58c7d4e4d4408e57f : SUCCESS in 53m 21s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/5b0455346ea14f35a8f5e81a964c305d : FAILURE in 35m 18s","accounts_in_message":[],"_revision_number":3},{"id":"fd75bf9df7633fceebc43650f2678b46e8f33d96","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-08-11 05:19:24.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"1d4190189ea35576d25c671b31c53d7344b32a46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 07:45:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/accba5138fa74ec9a112abc61104a740 : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c803b29fe504d6aa8d3ae61ad8742f9 : SUCCESS in 2m 49s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/fea99091d7ca43ea9fe78d4b71d886ff : SUCCESS in 1h 13m 19s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/05f9397d5fa2438383fbecfda1e5613a : SUCCESS in 1h 15m 45s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/8a11be574cf8499a95549503845cbb50 : SUCCESS in 1h 08m 23s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/09ce7e52e88c4270b70fa8dab28bf382 : SUCCESS in 1h 24m 53s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/33d549c6e8754eac98bc79e79cd87c72 : FAILURE in 2h 12m 48s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/c1078be10f174257ba720528101870f0 : SUCCESS in 1h 49m 11s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/9ff9e805af1e4c69817db84873c1c50c : SUCCESS in 58m 47s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/30ec613e481b4b0b813bee8266fdc4bb : SUCCESS in 1h 02m 40s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/34dc5ada039546448f3cae63562a5236 : SUCCESS in 1h 30m 17s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/679616a5491544258c3599e391149611 : SUCCESS in 1h 58m 50s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/68002c6b1d9a48329c8d2a22525db989 : SUCCESS in 1h 08m 03s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/827f09da6c9544069b4cab9d3f216112 : SUCCESS in 1h 21m 22s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/8163671e226a4fe8913932761c0aeef4 : SUCCESS in 1h 47m 51s","accounts_in_message":[],"_revision_number":3},{"id":"e5d20e8f9cf507623bb50a24b7ed5b1e3163532b","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-08-11 07:47:29.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"54379ad1b2f24433b953504b028dfa9e21bdcd96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 09:41:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f29251702df4d468b5aad66b630a488 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3b657a61626429a8dfbcc7d09230d70 : SUCCESS in 3m 13s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/5f779e5cdc044c88bf77661e2e84a3c6 : SUCCESS in 58m 36s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/3c82931e009a469f9a63982d439f4495 : SUCCESS in 1h 20m 02s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/8bc01f3b7518497db122dcb8e0509077 : SUCCESS in 59m 40s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/26315b0d67704e48a392ac4c26dd69a3 : SUCCESS in 1h 24m 33s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/f6ab8c64973249dd80c917ec3e93a9de : SUCCESS in 1h 30m 55s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/c3789ee98df945d1b2c3f2a835c4f67c : SUCCESS in 1h 27m 19s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/a94fb2a491d5403b8201010d642e1ec1 : SUCCESS in 1h 23m 13s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/b85c44741e7849369163249ddb3fe380 : SUCCESS in 1h 15m 23s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/0c2f77df526641b7b18f8f4e4df52b2d : FAILURE in 1h 03m 20s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/4c21f06d57114c568397e41b8f237e9c : POST_FAILURE in 1h 42m 23s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/20208fca8f1d49aba529b43f8f1c949e : SUCCESS in 1h 19m 37s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/fc6d88084ea54719856a228fe8fb2103 : SUCCESS in 1h 21m 44s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/c60e93f5b0164fd5b3d1cba64c2df3bd : SUCCESS in 1h 33m 07s","accounts_in_message":[],"_revision_number":3},{"id":"1a669d62ffabfa4c9458defbb30ba5d443403cd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 09:41:31.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c98d3759db7b69fdf9557a896c8a1975935900fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 11:18:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"5e05c0e830eb122b8060b0dae58770f4099b3e31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 11:18:05.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e74fbcb70d0243f3bd5519caa64c5046 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f23f4e8d068437685eb5fdc1f270c41 : SUCCESS in 3m 52s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/0859dd84fa4c4c6194cbda1376a7ee5a : SUCCESS in 1h 04m 44s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/108eacf17c454ff991f136765da3125b : SUCCESS in 1h 27m 48s\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/cfa48263783a4b85907317a948fb8fd4 : SUCCESS in 1h 00m 10s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/a74b899c652e4896ba03e406aa90dc74 : SUCCESS in 52m 17s","accounts_in_message":[],"_revision_number":3},{"id":"4ba1af2affda795add1082189efa102116eb403f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 11:19:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f6172d27f79455bbdb71ca70d3cacf7 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2fe1791ff2e6283c7f8d92e10fcaffa104457a95","revisions":{"0049dee445d3ee44ee81ee2c2d4bb5a741f68d0c":{"kind":"REWORK","_number":1,"created":"2020-08-03 09:44:05.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/32/744432/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/32/744432/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/744432/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/744432/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/744432/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/744432/1"}}},"commit":{"parents":[{"commit":"2b388b8a938d185339830131ce6efb9f4d5aa7c1","subject":"Support KuryrLoadbalancer CRD","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2b388b8a938d185339830131ce6efb9f4d5aa7c1"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-08-03 09:41:02.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-08-03 09:41:02.000000000","tz":180},"subject":"Remove skip of test_service_udp_ping because","message":"Remove skip of test_service_udp_ping because\n\nIt as skipped because of 1886909 which should be fixed\n\nChange-Id: Ia28cf8b9c27560f4672a3bc46fd8a12eae790a51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/0049dee445d3ee44ee81ee2c2d4bb5a741f68d0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/0049dee445d3ee44ee81ee2c2d4bb5a741f68d0c"}]},"branch":"refs/heads/master"},"fd28f3959e7abfd18640f937cab4adcf80938017":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-08-03 12:58:31.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/32/744432/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/32/744432/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/744432/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/744432/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/744432/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/744432/2"}}},"commit":{"parents":[{"commit":"2b388b8a938d185339830131ce6efb9f4d5aa7c1","subject":"Support KuryrLoadbalancer CRD","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2b388b8a938d185339830131ce6efb9f4d5aa7c1"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-08-03 09:41:02.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-08-03 12:57:53.000000000","tz":0},"subject":"Remove skip of test_service_udp_ping because","message":"Remove skip of test_service_udp_ping because\n\nIt was skipped because of 1886909 which should be fixed\n\nChange-Id: Ia28cf8b9c27560f4672a3bc46fd8a12eae790a51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/fd28f3959e7abfd18640f937cab4adcf80938017"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/fd28f3959e7abfd18640f937cab4adcf80938017"}]},"branch":"refs/heads/master"},"2fe1791ff2e6283c7f8d92e10fcaffa104457a95":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-08-05 14:58:19.000000000","uploader":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"ref":"refs/changes/32/744432/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/32/744432/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/744432/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/744432/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/744432/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/32/744432/3"}}},"commit":{"parents":[{"commit":"2b388b8a938d185339830131ce6efb9f4d5aa7c1","subject":"Support KuryrLoadbalancer CRD","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2b388b8a938d185339830131ce6efb9f4d5aa7c1"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-08-03 09:41:02.000000000","tz":180},"committer":{"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","date":"2020-08-05 14:58:15.000000000","tz":0},"subject":"Remove skip of test_service_udp_ping","message":"Remove skip of test_service_udp_ping\n\nIt was skipped because of 1886909 which should be fixed\n\nDepends-On: https://review.opendev.org/#/c/742384/\nChange-Id: Ia28cf8b9c27560f4672a3bc46fd8a12eae790a51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2fe1791ff2e6283c7f8d92e10fcaffa104457a95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2fe1791ff2e6283c7f8d92e10fcaffa104457a95"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
