)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~745601","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~I0c8247566422686f3ce2727c8347ef8b621dad70","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"set-kubeconfig-path","hashtags":[],"change_id":"I0c8247566422686f3ce2727c8347ef8b621dad70","subject":"DNM: Set kubeconfig path","status":"ABANDONED","created":"2020-08-10 23:10:53.000000000","updated":"2020-08-11 12:55:22.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1f4a13501d43116abc8c4926f299891fb479c319","_number":745601,"virtual_id_number":745601,"owner":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-08-11 06:12:15.000000000","_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-08-11 01:15:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-11 01:15:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-11 06:12:15.000000000","updated_by":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"reviewer":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"state":"REVIEWER"}],"messages":[{"id":"d60e4d335c28c9248878af9f4b9a9c2e7d75231e","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-08-10 23:10:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a43d7584eb629fd02cba5612df5062fe95b44e99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 01:15:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/464be42c342148e49af55b9cc2cad897 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bba2842fd7d14d87afeebbe5051addf0 : SUCCESS in 4m 02s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/e8667f5e6e3b4c83ae0cf975c02f21fa : SUCCESS in 58m 52s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/fa16e12620e74717b8cc432d69eb5d3e : SUCCESS in 1h 06m 48s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/0ccc93515aba499e9199e4ff3c894c27 : SUCCESS in 1h 03m 35s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/5ed84f8932da453da1a7a8f95c854b83 : SUCCESS in 1h 12m 38s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/12a88e49b59f4dfdae37ae976c000a42 : SUCCESS in 1h 05m 39s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/1e611dc81f424630bf31aab70c1654e0 : SUCCESS in 1h 38m 08s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/cb172e15c71d48dfab21db7614c0e5f0 : SUCCESS in 1h 02m 08s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/3765fbd3192c4977bfec9dd5c4df0ca3 : SUCCESS in 59m 07s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/744e42955678456e9dab5bb4a2cbae85 : SUCCESS in 1h 00m 46s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/1e1f4c5d857a46e1b4db69d07b840340 : POST_FAILURE in 2h 02m 15s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/bb4a7e98fc1a4e2793fcbd7ffc0ed7a4 : SUCCESS in 1h 17m 28s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/5076a8c3b0e64a179dee760835747990 : SUCCESS in 1h 20m 58s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/7d30466230b34eb79d9d11bd3ab71625 : SUCCESS in 56m 32s","accounts_in_message":[],"_revision_number":1},{"id":"9fed98e70f7c97cf6afee7f5a3eb5ba0cd19aa22","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-08-11 06:12:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aae5a70cd32dbc1df16a772bdcceeefd2bb9187e","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-08-11 12:55:22.000000000","message":"Abandoned\n\nThis change is not needed anymore.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c9edf4916b3c6f07007fb99e35204ee6a0f60ac6","revisions":{"c9edf4916b3c6f07007fb99e35204ee6a0f60ac6":{"kind":"REWORK","_number":1,"created":"2020-08-10 23:10:53.000000000","uploader":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"ref":"refs/changes/01/745601/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/01/745601/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/01/745601/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/01/745601/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/01/745601/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/01/745601/1"}}},"commit":{"parents":[{"commit":"2b388b8a938d185339830131ce6efb9f4d5aa7c1","subject":"Support KuryrLoadbalancer CRD","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2b388b8a938d185339830131ce6efb9f4d5aa7c1"}]}],"author":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2020-08-10 23:09:38.000000000","tz":0},"committer":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2020-08-10 23:10:34.000000000","tz":0},"subject":"DNM: Set kubeconfig path","message":"DNM: Set kubeconfig path\n\nLoad kubeconfig from specfic path.\n\nChange-Id: I0c8247566422686f3ce2727c8347ef8b621dad70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c9edf4916b3c6f07007fb99e35204ee6a0f60ac6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c9edf4916b3c6f07007fb99e35204ee6a0f60ac6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
