)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~754956","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~Ibf8113d9667c02a1b6d080aa50e91fde0809fd75","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"np","hashtags":[],"change_id":"Ibf8113d9667c02a1b6d080aa50e91fde0809fd75","subject":"Add test_network_policy_add_remove_pod","status":"MERGED","created":"2020-09-29 09:55:27.000000000","updated":"2020-11-03 11:07:02.000000000","submitted":"2020-11-03 11:05:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":84,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754956-1604401545288-3f776333","meta_rev_id":"088e3c7b69de395f4fbdb4fbe8a632372e624740","_number":754956,"virtual_id_number":754956,"owner":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"date":"2020-11-03 04:13:47.000000000","_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-03 11:05:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2020-11-02 15:08:37.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":2,"date":"2020-11-02 15:25:25.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-03 06:34:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":1,"date":"2020-11-02 16:02:52.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-29 13:10:24.000000000","updated_by":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"reviewer":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"state":"REVIEWER"},{"updated":"2020-10-29 07:42:34.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2020-11-02 15:25:25.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"},{"updated":"2020-11-02 16:02:52.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2020-11-03 06:34:13.000000000","updated_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"},{"updated":"2020-11-03 11:05:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1c5c80211a3fe09d5f7a768765c845c5ea5586cf","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-09-29 09:55:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"501dcc57218e67abd4906c46e8ddf9b8db655320","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 11:59:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eac033753ec04a8ea5c9e349a8461595 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eed137d5dbbe43ffb5d8906e9e511fcc : SUCCESS in 3m 08s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/c23edccee0374dc49b160d2f2dce7834 : SUCCESS in 1h 04m 15s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/1236c37a5e934f07a59b7a8c738380d4 : SUCCESS in 1h 05m 47s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/3d9a2e5ca49d4284ab2976f08c8cf95c : SUCCESS in 1h 09m 18s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/85562374cb0e4242bbe04784280e9294 : SUCCESS in 1h 21m 15s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/4eca3d861a1d411b806edea88d294615 : SUCCESS in 1h 30m 25s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/73d4311899d9494f887327db39e8893d : SUCCESS in 1h 19m 51s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/436ae34c775d41ba8b69a5f5a10d5ff7 : SUCCESS in 58m 20s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/bd85230346774c009c013ce87d648c08 : SUCCESS in 58m 35s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/7340c993501a429c9be383b64c6072e8 : SUCCESS in 1h 03m 23s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/00b6da6f81494c2e8d9d80b51c333626 : SUCCESS in 1h 58m 15s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/fd61bc4233864c288c4e1caaae5ecabf : SUCCESS in 1h 05m 03s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/e5e1bd23a52f4ee5b3a246e7ed3ae9bf : SUCCESS in 1h 03m 43s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/ccf22a12d22047898aff1ccd629af90d : SUCCESS in 1h 18m 19s","accounts_in_message":[],"_revision_number":1},{"id":"3e49df287b98b3d39bebfb67c8054cadce2cb375","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-09-29 12:33:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1f0799579bfd9025594f8e250386573d7b3a013a","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2020-09-29 13:10:24.000000000","message":"Patch Set 2: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3f7af73022097a704f8d0e6955b31fdbf439b64c","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2020-09-29 13:11:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b791bce78ab787c6299811224aae60676fd8dfef","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-09-29 13:34:02.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a8ec7697d3a6a01dbdcec663b1571d30ab5f6354","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-09-29 13:34:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8869dd00aab23708a431013f83e5bcbb13f38316","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 15:59:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/750ca6a88fd64cdca2538760d7323060 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f46215c843874a56a279a7534e8f8ab4 : SUCCESS in 5m 47s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/dd27f9cbc3104169bd2f3d62432792c2 : SUCCESS in 1h 10m 32s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/8cae58a9552d4ff49ed5b60cc78c900a : SUCCESS in 1h 24m 21s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/8fa038a316bd46ec9510c22ceb9dbf6e : SUCCESS in 1h 19m 37s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/cde5c92b57ae4c368fbeeb68fc83f657 : SUCCESS in 1h 23m 09s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/8c37e1c89aff4858950eef916a64b723 : FAILURE in 1h 32m 38s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/7a214fd800b24d60984d13355bf6ae90 : SUCCESS in 1h 28m 22s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/971187d2277a425f96eef678647f84c6 : SUCCESS in 1h 09m 08s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/196909c18f314fdd919a1b165d426a0a : SUCCESS in 1h 07m 48s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/672cde04b9f949b69c06f0a0e92e08b4 : SUCCESS in 1h 11m 01s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/23d0914bd7ea458e84b9dcad11e821c7 : POST_FAILURE in 2h 18m 25s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/9562c043e75443de99b31072153ee2ac : SUCCESS in 1h 27m 55s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/d689fac10ba3454693a0b9d964aaeb83 : SUCCESS in 1h 21m 57s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/bba6e64c0e7341df867819f5852266d3 : SUCCESS in 1h 24m 47s","accounts_in_message":[],"_revision_number":3},{"id":"4aba9b308de1bef0c44d318cdc4cf318c2462eed","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-09-30 09:00:25.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"57958d0796b0eeedb9b83c11dae504394552dc79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 11:22:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47bc3aef1897443da8fdfb36becf3fdb : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e0dd8ebe4a34f59b78c4c28c19315dc : SUCCESS in 3m 55s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/535d74761101496eb2de111d7c75cfa7 : SUCCESS in 1h 09m 27s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/2b6814944af3495ea87eea4bc34d1502 : SUCCESS in 1h 14m 24s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/bbf91492df8345aca502f574a0295c86 : SUCCESS in 1h 02m 53s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/d8a14bbd46384606bc12147346b086a6 : SUCCESS in 2h 06m 37s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/49358bd171424dc5925eb1b0a0ffb71a : SUCCESS in 1h 24m 12s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/97c901c734dd48eab72a414b5db73e3a : SUCCESS in 1h 54m 03s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/14dd56072e7d4ac79e0115ea79d01fdd : SUCCESS in 58m 48s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/a01ff5200d3a4b25a2e0844e629a0e42 : SUCCESS in 1h 09m 15s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/9da6479771a14e6fb25c62d8bc76a75a : SUCCESS in 1h 13m 24s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/061f479e61b14ed5b0a7e795c1f3cd0d : POST_FAILURE in 1h 54m 01s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/ed2c57e5e57f4c78b08f7d3e9d9d1c6f : SUCCESS in 1h 06m 49s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/ffe28423b7654b4e87a480224add340d : SUCCESS in 1h 04m 35s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/7b670d7592cc40ddb4087f03835cc2ed : SUCCESS in 1h 12m 44s","accounts_in_message":[],"_revision_number":3},{"id":"8a924b83c5d10d0e3985c4e62d4b10bb2a3c690d","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2020-10-01 06:23:40.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nOne inconsistency in log, other than that its looking good!","accounts_in_message":[],"_revision_number":3},{"id":"0d675dd8e3dc374b21e76c51973871e0919f4b95","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-01 06:39:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4ec62522fe361b60ce1e88ba14382dcce486be9c","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-01 06:39:57.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"927dabf80b60ca61afc76e40343d57a1fe8a7132","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 08:47:02.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff18a2a0787b459694fa3e2e121d839e : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/598c7710d87f4b1a865c3e910c6d66a5 : SUCCESS in 4m 26s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/35419e7f87b64fc489fc7f25d6352b7b : SUCCESS in 1h 07m 07s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/b35bd4f4c8514bf49c0e90eb2e2f777e : SUCCESS in 1h 13m 19s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/cfa1b681de144c73954d452397ba7811 : SUCCESS in 1h 04m 46s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/87cb33b90859421586ea33bcf99ba0ff : FAILURE in 1h 39m 25s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/c24dbb0f72cb426aa2bc029fe1c4576b : SUCCESS in 1h 27m 25s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/aac4f22d923f49cab7a70b327cf21564 : SUCCESS in 1h 28m 28s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/ca1e751752544dd9846dd2e9ceb92148 : SUCCESS in 1h 07m 36s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/9e2025e21cae4060a5de600a0f78ae26 : SUCCESS in 1h 05m 10s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/a12c6089e45642f68da49c0ceed01c04 : SUCCESS in 1h 12m 24s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/fe9bad0510cf4c489ffb0e8c0f7447a9 : POST_FAILURE in 2h 04m 01s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/8309b320411c4bce8c51b24b9061de62 : FAILURE in 1h 04m 20s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/81496ddbc32e4d369ec8d6bf44b8e48e : SUCCESS in 1h 25m 54s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/930303c0a3e743a399b3fbbe4729dfe8 : SUCCESS in 1h 28m 22s","accounts_in_message":[],"_revision_number":4},{"id":"d5f33448c48abac5ed433889ffe3097997469ff5","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-01 08:47:35.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"ccd0b81925728378ad4bde6e254bd6eb16e6a16e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 10:48:46.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa127ba865af4bcd92cb4e9b2d2e948c : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cff47bb1e8043c5afc216e368a633be : SUCCESS in 3m 31s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/1ecbfb5c16ab41408be57c8f4f5ca009 : SUCCESS in 1h 16m 49s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/771b62ce94084ee79d5a8b889820a1a5 : SUCCESS in 1h 06m 23s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/815d6bd783b341a0aef7bfeb1eeb9cb7 : SUCCESS in 1h 09m 12s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/cacb4b566d95462cb7251d0325bbe017 : FAILURE in 1h 23m 59s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/9406c592322447f39305edc0e01a7f5e : SUCCESS in 1h 29m 47s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/17c594c2e57949e591fdd7a304875f35 : SUCCESS in 1h 26m 02s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/0c49e69e70e5425a810268d5d08b7d15 : SUCCESS in 1h 05m 17s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/85c8ea3405c146b5b5a6c1f5d8d370c1 : SUCCESS in 51m 25s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/51dacaf20b54461ea9cfd5f9da15009a : SUCCESS in 1h 25m 55s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/c800d87278674c71a41356ad3587bba6 : POST_FAILURE in 1h 57m 11s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/e2345065c96340c09e051d510f38f76c : SUCCESS in 1h 15m 44s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/de9aa833dd20431fb6cebb0314226fbe : SUCCESS in 1h 28m 46s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/457e8e34b6d24934bb3bd9ba71386cfc : SUCCESS in 1h 43m 06s","accounts_in_message":[],"_revision_number":4},{"id":"e4ab94e271a00e159c6769594bb81be6ac21c95a","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2020-10-02 09:03:26.000000000","message":"Patch Set 4: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"086587b403f207a630b5268b52d268192f738737","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 11:10:56.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bb686ce7b9c4357948ab87f6dbb827c : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5160ff05fe24d8ca374e4b23e063cf9 : SUCCESS in 3m 00s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/533fe73c8cd6460889a5d34eec8c4928 : SUCCESS in 1h 11m 26s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/e84fd2fdf23846e9bcdb627629e65e3d : SUCCESS in 1h 03m 13s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/a60fc389bc954999becfee67d9f3b551 : SUCCESS in 1h 47m 11s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/cfc9f77b29ac4194bd17f88232875026 : SUCCESS in 1h 14m 59s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/798ae0fda1d4480b9eabcf93c39ac8f7 : SUCCESS in 1h 23m 59s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/be92c04db282418d991509f1064f7e08 : SUCCESS in 1h 49m 30s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/458bc7b1bff2434aa0e5e05d0abbff19 : SUCCESS in 58m 55s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/8a295b1ee08849cd9f0e4458a4b53c0f : SUCCESS in 1h 14m 01s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/79f2656797ce4baaa67e29c18af34387 : SUCCESS in 1h 23m 04s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/02437f7607dc489b8b47cd6d09054bbe : SUCCESS in 2h 01m 06s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/3e6b2bbcff52453f9a766506a8f82e8b : SUCCESS in 1h 07m 40s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/469f9d2059984eda94c6bb9869238955 : SUCCESS in 1h 23m 15s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/582227bea4e54d5d85258be4f455d282 : FAILURE in 1h 37m 09s","accounts_in_message":[],"_revision_number":4},{"id":"016a40eec088d12e67680fa430da65689101a870","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-10-02 14:53:40.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)\n\nThanks for the new test, Itzik. Here goes a few suggestions:","accounts_in_message":[],"_revision_number":4},{"id":"b35c9eb70ded2d4bd2f6f652d0f2e2aaf0facb98","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-04 09:41:34.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"eb89a725e5fe94594ae867495b9b377dc91820ef","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-04 09:41:42.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"9a097ffb26ecbac477f0e9b7e756f2bbf3a5431c","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-04 10:44:12.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5af69853b7d61080c24a049049b7171867875ed5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-04 11:48:54.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e68bdb8b035444bfa1b59f7d68035169 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d55184861b53452a91d41a1a36fdc0e9 : SUCCESS in 2m 56s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/a00d9972778d4ed7abbaedd964929381 : SUCCESS in 58m 52s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/799f954da3cf41e68a8ad6488bd41b95 : SUCCESS in 1h 25m 09s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/ca33aefb36d04b788d7f6ccad5d20bcb : SUCCESS in 1h 31m 37s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/c839147e5ce2459b80a1c65182a0deec : SUCCESS in 1h 11m 11s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/4cab939f47f240bbb48a4e96b3f4930b : FAILURE in 1h 37m 22s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/2ff8f81581464fce9b339162e427b9de : SUCCESS in 1h 36m 53s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/e4e3e30868c94fceb449b980d93d982b : SUCCESS in 1h 08m 14s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/830a06486492408fb0203074a12d82d7 : SUCCESS in 1h 09m 40s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/0ec561dd9d6842eb8809a458755082f5 : FAILURE in 1h 14m 04s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/c0ee12976419471aa6a6f1bb38e12e75 : SUCCESS in 2h 01m 44s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/f8b847c78d284e9f8a58646be1848553 : SUCCESS in 1h 00m 31s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/11ab8a8135a0484da4770e2731b62ff3 : SUCCESS in 1h 14m 37s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/03ee315d04de45158cbc2ef96dd635af : SUCCESS in 1h 18m 18s","accounts_in_message":[],"_revision_number":5},{"id":"d4a50ffcea1edebe426e8c8bfc653fee6becfebd","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-04 16:04:33.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0d3a54b34072799ddcdd16c11461b75ec0ed66a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-04 18:11:53.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f8174ad75ac4a02b0eff0585e1e3781 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a344e0bfba94cfe96755a87eb732199 : SUCCESS in 2m 53s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/a1f18049108849699b4012187670038f : SUCCESS in 1h 07m 31s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/1e231c300b214aa293d5f1537ac53029 : SUCCESS in 1h 16m 58s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/bc7aec86a5144b418d4b62e3a4b7e47c : SUCCESS in 1h 06m 07s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/588fde25aeb240bd8cc56f749cf9894a : SUCCESS in 1h 12m 35s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/8dfcd0349d124b2cb6668bab736fade7 : SUCCESS in 1h 16m 01s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/33e4ee2b17c3481e9af42798fddb9b58 : SUCCESS in 1h 22m 08s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/e891e35c3af247d1892dee23aa41a3f3 : SUCCESS in 53m 36s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/b8e6bbc974994c16bf2cae3a45fc4142 : SUCCESS in 1h 02m 42s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/ed324024c9624f9a85ecbefb75347ab9 : SUCCESS in 1h 22m 54s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/a0d57e96a47240ea8b8be7b37a582690 : SUCCESS in 2h 02m 44s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/0e5400262d144688889836ae3d32fd5d : SUCCESS in 58m 02s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/54dcd720aaf74241b9ce02d0691a004b : SUCCESS in 1h 12m 06s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/e108d42f95044d59aab3959e0257acf4 : SUCCESS in 1h 16m 19s","accounts_in_message":[],"_revision_number":6},{"id":"13380d14166092124ec248c59ea0509f9014e8cc","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2020-10-05 07:33:13.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"6aedc57468296f49d1c6ae5bd55e161b92714ae2","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-10-05 07:49:26.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)\n\njust one nit. In general looks good, but I don\u0027t know if we should try to add these type of test to the k8s NP suite instead (or at least in addition to)","accounts_in_message":[],"_revision_number":6},{"id":"9c61eacb0e6c48f6f47aea495ea1d9aa44b74c87","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-05 09:15:31.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"59afd62e3556f156477f438ef908bdabc522bc06","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-05 09:15:43.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"63797600f5a17bd16f85289ea9f753e5b8baa1bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 11:36:31.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/537de0239bc845abb95a2168957a6fec : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/015889297903488f9d6fb1f0eafc2a3d : SUCCESS in 3m 03s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/0ddce6fc805d46b482ef4d84c293fc3e : SUCCESS in 1h 16m 32s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/3cedfca60a3d4933b9e1c5f3a4c0e628 : SUCCESS in 59m 27s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/dcf10fc9c3734f488ee4e8b0a5af09c6 : SUCCESS in 1h 16m 51s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/a2fe0cc911954ee5a2cff0e56949834a : FAILURE in 1h 26m 51s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/602318f98756405fbbd1bfa711c3562a : FAILURE in 1h 33m 07s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/8d1e5d1ed8da4e629f44bec75e553e51 : SUCCESS in 1h 23m 14s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/6bfe63117a59429f90795da49cfb0ea9 : SUCCESS in 1h 43m 12s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/067de1a0af6049e38e93598849ebcc14 : SUCCESS in 49m 00s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/7894f16de8424028b84fd7a39112621a : FAILURE in 1h 01m 14s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/d507b0bdb5454b9bbfe8173c0891bff6 : SUCCESS in 2h 11m 29s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/867784359c014d1da4a2a3a0379801eb : SUCCESS in 1h 01m 34s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/ed8e446256ef49e2bbef48bc43336b38 : SUCCESS in 1h 22m 16s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/d091b526fbfd447b8ed67f136af2c2fe : SUCCESS in 1h 20m 39s","accounts_in_message":[],"_revision_number":7},{"id":"436ae4ac36b358bef6c81f278b1e623da229e696","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-06 08:42:09.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"0f1d2480dd491563e883892da04a02c7b006119b","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2020-10-06 08:46:54.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"bc3aa9df5c8614db2802a8ef0bf2f30af114e29f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 10:30:56.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d60142eecf524e87aed9737d99c85a0a : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/451cf12178fe4eceb34a8e43cfaa16a9 : SUCCESS in 5m 09s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/f72c83eb15be496e85572514c80147fb : SUCCESS in 1h 03m 08s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/bab0a4f85df64c80ad1f66e21f321541 : SUCCESS in 1h 14m 59s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/5478665183d74870a8fa7533d9258915 : SUCCESS in 1h 09m 47s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/dfd75dcaf08e4c7ab6d099050c961897 : FAILURE in 31m 18s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/aba724fce9a74d6d910da24265c255c7 : SUCCESS in 1h 31m 16s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/7616a9f6a0d94c66837050a6261ba5df : SUCCESS in 1h 30m 51s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/6552b303a96e412791836822273a559b : SUCCESS in 1h 06m 14s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/6a2ceeddc082494da765c8892d25e39f : FAILURE in 26m 23s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/97353c1ee6624270875a13de62b9ad4e : FAILURE in 30m 35s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/175ceee074494436b071b7c1baf5414c : POST_FAILURE in 34m 15s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/89f0ae66f4e74a7c80b4714dea75cf6d : SUCCESS in 1h 01m 15s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/8743771bdadc499c8254ab950e230a31 : SUCCESS in 1h 16m 28s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/299c34412c794c81b50c5d52bc36836c : SUCCESS in 1h 32m 55s","accounts_in_message":[],"_revision_number":7},{"id":"c3ff944b71b0c621eac79699b11c3fb8a0b06643","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-10-06 10:47:52.000000000","message":"Patch Set 7: Code-Review-1\n\n(11 comments)\n\nThank you for the update. Here goes a few more suggestions/questions:","accounts_in_message":[],"_revision_number":7},{"id":"247b8d04087d314946b19729d90e8c60c9cff7fd","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-07 08:49:20.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b5bcc3b7299dae66067479aceb7f9897c3277b2e","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-07 10:28:14.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"434fad88751ac59aead3685492c0f315485ccbc2","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-07 10:28:52.000000000","message":"Patch Set 7:\n\n(11 comments)","accounts_in_message":[],"_revision_number":7},{"id":"b7884a551f0b52193a0cae39ad33d132cfc9f1ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 12:04:41.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f194d68e73d4508aa590ed30e93ede8 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c14584599774d97aa50a842e66ef9a5 : SUCCESS in 4m 10s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/e6bf1b401d344a979cf5a41685c33c43 : SUCCESS in 1h 11m 16s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/4b73e4ac293e4979bfe23f350a2b97ce : SUCCESS in 1h 15m 49s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/92f270ffc3df40ebb12cae11bd68bd4b : SUCCESS in 1h 14m 46s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/3929d83f1d36429faf7a16920422d654 : FAILURE in 34m 17s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/e71a7f02323943b6ab5d6e0f883b3686 : SUCCESS in 1h 26m 56s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/5676cbee53254481baed4d401ec2840d : SUCCESS in 1h 27m 58s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/3c5bf142102e40fd8e88053c8376b63c : SUCCESS in 1h 06m 47s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/1271943182c9439896da944440f5c05a : FAILURE in 34m 32s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/1a82b568b68c4e9ba9235fc9c74af3e5 : FAILURE in 37m 35s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/1875f6a03a4b42f1976cc2e5ccfe1bb9 : POST_FAILURE in 35m 32s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/83c140e0a15b4efeab4bf587ddff4d80 : SUCCESS in 1h 04m 30s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/bd3f82cbc78f4a81a42e0dd618208a12 : SUCCESS in 1h 18m 55s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/d3d10eebeb4d4a56a80bdc7c25fc2718 : FAILURE in 1h 28m 05s","accounts_in_message":[],"_revision_number":9},{"id":"4d18131e308aac75327b6704ab857874ebba806e","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-07 12:17:37.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"d6e1a30d59b4d313c9ae921a97b870f8c504f47d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 13:42:47.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccb1aac8a8964492aa7dabd81d968889 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc0b282fde744fb0b28a3e261c8792ac : SUCCESS in 3m 02s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/3d0ee011f7504e09ad706b61e94339e0 : SUCCESS in 1h 01m 41s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/3acf6517f0e84852964d81a6c84626b2 : SUCCESS in 1h 12m 53s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/a900021ad6164420809964c467675292 : SUCCESS in 1h 06m 31s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/68b042aab7f3460399f7f82ae54ab36c : FAILURE in 26m 19s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/7ebd3ff41b5745f6b291c1b946b41805 : POST_FAILURE in 1h 21m 34s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/d48a27b1815042e287716916560590e4 : SUCCESS in 1h 16m 18s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/5e01dc19b4f649afb2f761048ed7c2f2 : SUCCESS in 1h 05m 41s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/1a13c088ebc74553b465d05cfb0ea202 : FAILURE in 32m 12s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/c807064d7ffc459e91d4844736d13941 : FAILURE in 36m 38s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/e51d929cba404d83a447802da7700c02 : POST_FAILURE in 42m 36s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/8951d0662e4c4d67aa207e3d719bacb7 : SUCCESS in 59m 34s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/7478074358d54bf4940b564ea041b185 : SUCCESS in 1h 10m 31s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/f7f0741c66f34a74ac52a21f79c7bef8 : SUCCESS in 1h 16m 23s","accounts_in_message":[],"_revision_number":9},{"id":"f4e088f0d3d74ec9e1bf18776b6b135241f208e6","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-08 03:07:00.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"e6fdeb8152698a504e881136735129c970af0dbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 04:36:35.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c0a2d35517d43789c274bc6a6fd7e01 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fa1f932252c4e57bed55670e8eafe16 : SUCCESS in 3m 06s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/22d209a5ea1a4be0a913195c1671c4a1 : SUCCESS in 1h 17m 02s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/eaa4f4f1b38649ee87c2a503bb5f380d : SUCCESS in 1h 04m 22s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/36e693c05b514762ac16dceee14d4844 : SUCCESS in 1h 12m 38s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/972355e907a94268881818054bea1994 : FAILURE in 29m 26s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/ee770e61a05f46feb521245acfc60d1a : SUCCESS in 1h 13m 47s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/e7b5bb67ed494a3a9f9823948d6d7fd3 : SUCCESS in 1h 28m 20s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/105413ab62d84fa08047ea0ace62867d : SUCCESS in 1h 02m 25s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/2aba01e72ecf47bd9a591c30a9a4a797 : FAILURE in 49m 27s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/7a67242fdc2c47e588c9eba44fbf5d5f : FAILURE in 36m 00s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/9ecd9986fa5b4abb9b410caed5ac2828 : POST_FAILURE in 38m 49s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/b7bbbe2d0a534b6b9c405eb4b5808352 : SUCCESS in 1h 08m 27s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/30d95a72b2054a1ba26f8d49e20214d4 : SUCCESS in 1h 20m 13s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/df86cf589fa54f2883e8e34514347b9a : SUCCESS in 1h 17m 54s","accounts_in_message":[],"_revision_number":9},{"id":"423f0e7f390bca1053961ca2c2422a03c299064e","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-08 04:39:27.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"c2a7dce4ac15b86b851a998154ee87cbe376f459","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 06:18:10.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac362169494e465fbf3331b840daa9a2 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ac6003b693146918683a57297fdd24e : SUCCESS in 3m 52s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/386df88a80c7430aaf9a05b9e31d79b9 : SUCCESS in 1h 01m 37s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/b21e81a2c09246b781d987ece92f13d5 : SUCCESS in 1h 15m 02s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/8acf9adece934dd3b58525ff081a2fb9 : SUCCESS in 59m 49s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/fdae335f2fd3469c805b633a153499cc : FAILURE in 29m 35s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/dcf53a3d0da845f088a398b59c8018e6 : SUCCESS in 1h 34m 15s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/ad862175d0584b45af19e4677f3e98fb : SUCCESS in 1h 26m 50s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/52b0b80e4f0142eca3a1960e50ae6ffb : SUCCESS in 59m 22s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/fe26106e7e2443d38b8c010122c32741 : FAILURE in 33m 07s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/22ab51a042a24510bf974695e5589695 : FAILURE in 30m 52s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/5275650b23304414adbc6b3f437ac5c6 : POST_FAILURE in 35m 36s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/df22956531214d648772a233b0429422 : SUCCESS in 1h 12m 05s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/34b1f642b87a4ca8a72f50bf7caa4b08 : SUCCESS in 1h 26m 03s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/4f22c2a665fa4ae785bba973ff2f3d91 : SUCCESS in 1h 20m 33s","accounts_in_message":[],"_revision_number":9},{"id":"04520ed8c6f2a8704ff7ed90f34beba6a6495d83","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-08 06:36:36.000000000","message":"Uploaded patch set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"f6251e9161cb1a00131f5110fcee3612ff513d67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 08:52:32.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4183ead1dae48b792ab5b6c2e475a82 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89f1c061633d437380e08a9a6f1043a1 : SUCCESS in 4m 49s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/6ce8af40770f4cb7b51a4712bbf709e7 : SUCCESS in 59m 21s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/4d0ce5e2686f4229bbab6e7e25f7f706 : SUCCESS in 1h 06m 57s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/de0680a0ab24451f98fa5a4bf6668c59 : FAILURE in 1h 06m 49s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/d5821aff7118479abe050c81c361edcf : SUCCESS in 1h 13m 52s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/90d59f5620d342eb83ecc3aace0c6994 : SUCCESS in 1h 12m 45s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/8a152958edb3410d96c6cf701a5c067b : SUCCESS in 1h 27m 17s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/9ef7392fa9144044af5d8e2d10e879a6 : SUCCESS in 1h 04m 40s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/94cfbad149b048f9b4b654084d1b77dd : SUCCESS in 1h 14m 36s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/223bf7b05b5e479891a8ae12bf4495d9 : SUCCESS in 1h 25m 13s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/bca33c2d0df54504919a666b55116e2b : POST_FAILURE in 2h 13m 24s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/ad29f63cb0ab412db94074f944d6ebae : SUCCESS in 1h 08m 01s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/1efd7c90afa0476ba54574a203f737a8 : SUCCESS in 1h 06m 36s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/f7d5c25c79f549899d7d64b96dafc65d : SUCCESS in 1h 23m 29s","accounts_in_message":[],"_revision_number":10},{"id":"f4577b25fdb00e97a6fe51b5b24b6cb803708e31","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-10-08 15:28:29.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)\n\nSome additional thoughts...","accounts_in_message":[],"_revision_number":10},{"id":"f47f9c771892d51b989fd6efbac11c3b04f5dfc8","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-10-08 15:31:42.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"f1e6c1923073011326bad7dc2516d43ba1850d10","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-11 02:01:00.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"cb7a61b26a8604d37c4d2181b01404a5748452ae","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-11 02:01:33.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"b36e5ac7aaa6ce9aa83a384e7769592bee0f2fe8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-11 04:11:53.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8486d9ba04a44d64aee1430b0a66a956 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2051843c0ffc4fe392a37efb96b0cca7 : SUCCESS in 2m 49s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/191b5e2184ee44668b7dd3aeec3248d8 : SUCCESS in 1h 05m 39s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/e704be56c6134ea38c55c2ca029911f8 : SUCCESS in 1h 25m 06s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/e1fa7330612c4f9da8bb6ccb8d44b49b : SUCCESS in 1h 07m 26s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/cee23bed0d6a4d75bf05bb1ecc257ce4 : SUCCESS in 1h 25m 25s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/9168d56c6cea4cf5ab91cc0b8d76bf68 : SUCCESS in 1h 30m 26s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/292d4b8495224052bcf32dec54fcf9d8 : SUCCESS in 1h 28m 05s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/9a7ea443cc1f4443bb1dc72a31bdbf03 : SUCCESS in 1h 04m 33s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/20e1c347f5fa4fa9a7daf1a6d7fe9860 : SUCCESS in 1h 07m 48s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/7be3d329ec5547878aa5952119abeebb : SUCCESS in 1h 08m 46s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/2cfed4404b86444a91cb3230f186e29e : POST_FAILURE in 2h 05m 11s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/8a515eec9086421fb7db07b212705548 : SUCCESS in 1h 16m 47s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/a0a9159ab46c4d8b97d2c0ace6732a6f : SUCCESS in 1h 20m 03s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/35a0a25abb6140e2afb6bbe2bfff4d99 : SUCCESS in 1h 24m 24s","accounts_in_message":[],"_revision_number":11},{"id":"b45fe0c872ab1c737f25f5c82b000b27dec4ec0d","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-13 06:46:07.000000000","message":"Patch Set 11: Workflow-1","accounts_in_message":[],"_revision_number":11},{"id":"3e6709ab5fd738775487858f6e982456578f4e40","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-14 14:59:40.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"ab6f561e008e66841c6bb6bcd0b6216cef8cccaa","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-14 14:59:57.000000000","message":"Patch Set 12: Workflow-1","accounts_in_message":[],"_revision_number":12},{"id":"d60af9a055f53847a314f5e56fbd5b5846267645","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-14 15:23:50.000000000","message":"Patch Set 12: -Workflow","accounts_in_message":[],"_revision_number":12},{"id":"c569142eaf9c8b626ed2030c06cdc3ff0f526e4a","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-14 16:30:30.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"c79a28683564b6688b9dbeb78289b74b5a515504","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-14 17:54:42.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"42a97f9472f3ae670f0f7fe1696a543eb571611c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 20:39:35.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03295663e1e14744b736dea03a9ef773 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1be444511c84fb39858867891511f38 : SUCCESS in 3m 46s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/313b9ba08edc49e693e3cb7a92d32251 : SUCCESS in 1h 10m 22s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/6559d5b2f4b24703ac76fecce5286736 : SUCCESS in 1h 21m 27s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/a052c8281a664a638421d929086c50f5 : SUCCESS in 1h 16m 12s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/38711fb92c544c6ebbacf22fb616bff0 : FAILURE in 1h 16m 26s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/32e3adcf49f748a483038ecdcc4c1cb5 : FAILURE in 1h 23m 22s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/17f26f60723d4fefb9806a923950c6d2 : SUCCESS in 2h 38m 34s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/8d6086a67eb44d8fb183efb720ec0a83 : SUCCESS in 1h 16m 18s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/39e40633201b43b380ac7b533790d927 : SUCCESS in 1h 10m 28s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/cccbbc70fbc84604ac5f106de72d05ee : FAILURE in 1h 31m 46s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/77ba3878deb645be89720488d8a92e68 : POST_FAILURE in 2h 00m 37s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/04ebe873fd964d569a73754d99cba277 : SUCCESS in 1h 02m 28s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/af29e7c9d51b48e89dcf995ee1a36318 : SUCCESS in 1h 32m 47s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/7983321a05b74ffdb18f5ba308661c9c : SUCCESS in 1h 27m 19s","accounts_in_message":[],"_revision_number":14},{"id":"b626e6f9d101c56e494095e9e8fde45b3d684c95","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-15 05:47:06.000000000","message":"Patch Set 14: Workflow-1","accounts_in_message":[],"_revision_number":14},{"id":"baa973f6bac02f2954f9d3ae6a602792e3d58ea6","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-15 09:05:11.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"2753eeca4a831611448a41e1a292b16a65595085","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 11:12:32.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3260e3279b348268a2f486368114796 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e73c327031649e5866ebc1710dbb490 : SUCCESS in 3m 43s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/68a26311906546c2b4975b899290747c : SUCCESS in 1h 07m 44s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/63a056b16c464f588b29d1f606f4e8ab : SUCCESS in 1h 06m 19s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/0fae82452a914f61bedc31614878f752 : SUCCESS in 1h 25m 12s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/fdb239620e7b45609e3369e71c4a6b89 : SUCCESS in 1h 26m 15s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/f08e6548b5764989bbf55a246b2fe36e : SUCCESS in 1h 30m 38s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/7a2bd6fd2cc048cda2e196279d5d145a : SUCCESS in 1h 34m 20s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/0280764a60df4f828c89238c05746e2f : SUCCESS in 1h 00m 27s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/659e3eb4ce9a444d99396737f1ba51fe : SUCCESS in 59m 49s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/4f065e465b4b4afc93273d408226671a : SUCCESS in 1h 14m 53s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/fd4a4bdfa74a4da2aaac5219d2737f56 : POST_FAILURE in 1h 56m 06s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/c04e598bae5346c3869641b48baf28bb : FAILURE in 1h 10m 21s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/c5c21ac4689c422f95ca34ffc4e0dedc : SUCCESS in 1h 34m 04s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/d2277fa44e3b4b15b19481261ea51374 : SUCCESS in 1h 23m 02s","accounts_in_message":[],"_revision_number":15},{"id":"85d4b848c0716e1a67eb0eaee561b6b47b9c731f","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-15 11:18:24.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"11543737baef6d12b7e49bd792bb064b01134a1b","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-15 13:09:08.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"2658a4e76d8a86710e97927869092886be1bfdd4","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-15 14:35:50.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"2670cfe4eebd84974eea37af8f8334ebdcb7fdef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 17:13:09.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa04fa257ce04d58b61d33ffe12403cf : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0278e1aeb4024fe681f32df69fbc6af4 : SUCCESS in 4m 19s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/7130a0882eb643c4a928848a14becd0c : SUCCESS in 1h 06m 57s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/1578ce8deb15404788475d1f34fce4e4 : SUCCESS in 1h 16m 30s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/c585b6c261454fc0986cbf7664a1112b : SUCCESS in 1h 08m 10s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/6390245dbf1b4c5da3b8d81d3cb90434 : SUCCESS in 1h 29m 07s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/d586f802546c477a8e61206b1521c4ea : SUCCESS in 1h 23m 21s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/7676abf8c9c3461d871d71d407858590 : SUCCESS in 1h 39m 43s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/fdd44f9231754be1b856a872dd8198fb : SUCCESS in 1h 06m 10s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/ec6604e6675b4f8480f5e4d71edd0bbe : SUCCESS in 1h 16m 02s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/1fa484c667024ae8ac85df5fc9c58788 : SUCCESS in 1h 25m 30s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/d15ea7b59d6d4dfbb710571aa4a69f36 : POST_FAILURE in 2h 07m 10s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/b72219757b154e3291af663bbf615c8e : SUCCESS in 1h 22m 11s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/e8300c7d9d7247b68accdd2ddf0b817c : FAILURE in 25m 46s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/ca20482882bc4abdbba6f26285685929 : FAILURE in 25m 00s","accounts_in_message":[],"_revision_number":17},{"id":"4201f881e7ca28aa3c95993a6ca1cec9e35272a8","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-16 06:27:37.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"f1bfa39d108aa736840cd67c57d5e1e07205f2a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 08:44:58.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/152394042fb44353918ed8b6fa812eef : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5deb53caade142b291cf8cc020891dcd : SUCCESS in 4m 31s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/82ddf1d333c24ee8ad4235cce1ba3fec : SUCCESS in 1h 06m 21s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/36dc1464fc1c4092b22c56b868312815 : FAILURE in 7m 22s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/1d9a224733af4ce3bb2682d0ca0e511e : SUCCESS in 1h 21m 00s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/49c3e0d416f2477cab27b9b5424ec8f8 : SUCCESS in 1h 17m 17s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/5daa974cc764469a92e7fadfcaecd6bd : SUCCESS in 1h 20m 25s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/346f29f2902d4acd86f23f22798273e4 : SUCCESS in 1h 53m 40s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/91381375063845288f4d71cf997946c2 : SUCCESS in 1h 01m 08s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/6d52f8635af3416b8d20f56da364d604 : FAILURE in 1h 00m 21s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/7308dff00996489990f3df1b43af3639 : FAILURE in 1h 25m 03s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/e452306531b4492286597cf40f74abdc : POST_FAILURE in 2h 13m 59s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/3ea8065675cb4feca41d5bbfd65bcb75 : POST_FAILURE in 1h 06m 37s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/a27fd9a7bdc1483aa0f1b385e7fea630 : SUCCESS in 1h 23m 03s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/7426201cf3cd49bb8a1694754efcf1bd : SUCCESS in 1h 27m 51s","accounts_in_message":[],"_revision_number":17},{"id":"51dd25e38b8d982092c3fc0005e3dbcb3e3fe668","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-16 11:55:03.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"c819a700f80ef3e08c235f1f3bd3a1bff5935beb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 13:55:45.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06eb4cc4b9ac49469cd27f872c896899 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b71f877195164b71b5d7398f62f211ed : SUCCESS in 4m 04s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/73358ecff29c4f8d88e2115977b8308b : SUCCESS in 58m 36s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/89a1d28720544dad945ada39a2784887 : SUCCESS in 1h 12m 00s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/ac93eed1b42e4c4883b5fbfe4f8bc612 : SUCCESS in 1h 21m 13s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/826632f3fd67484790d92e3030685270 : SUCCESS in 1h 13m 31s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/f1135157031340d890aebdfd555d0616 : SUCCESS in 1h 30m 11s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/257a0ac7207c493f810089521568abf9 : SUCCESS in 1h 38m 10s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/1e6120865bd54db5815671f585d8fca0 : SUCCESS in 1h 08m 30s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/10428e6c6e804e02af3b4b2f68bcdce8 : SUCCESS in 1h 12m 03s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/a9dc725db10848ec82ece2fcf7fd5a96 : SUCCESS in 1h 08m 38s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/6aa6593ef17b471880d5f4f7b5f60f00 : POST_FAILURE in 1h 58m 46s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/471e4c3e91c847998a2c9dc0349586c3 : SUCCESS in 1h 07m 07s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/09dcb57a9a4a4057bcd60341c401d538 : FAILURE in 1h 33m 21s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/ef28fd1329f44b508ffcb79608db7af7 : FAILURE in 1h 38m 18s","accounts_in_message":[],"_revision_number":17},{"id":"1250815e84ed4bd049a5cca31265987c5a3dd4c8","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-17 21:42:30.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"8bb7ba406ef080d21ddbba492682201d036bda15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-17 23:40:26.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb0c49c8d9eb45feb6911e609ccda700 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4cc40ff82e84d7991cd67f61865b595 : SUCCESS in 3m 53s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/85125df51cdf4d478f2a1a18680ed40d : SUCCESS in 52m 11s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/ce8c3e488e674393b0ee08da2b0a8b9b : SUCCESS in 59m 18s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/3e98d066a21c4c4597794ade0d27fb12 : FAILURE in 6m 29s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/2a9cd2386b084bf4a82a202a7153840d : SUCCESS in 1h 13m 55s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/2bfc27bd14ef4f35b90ec28df4f37aed : SUCCESS in 1h 10m 55s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/a9b6ec4b523641ea95ceb25821a3b277 : SUCCESS in 1h 15m 33s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/890609b52f0245d08e28da588eeabd70 : SUCCESS in 58m 31s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/722ef6d4abc34aef9f846e352f135eb7 : SUCCESS in 1h 01m 15s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/0606d9d51bc5455da493b73483a888d6 : FAILURE in 1h 13m 32s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/8984c6dfd8e14eabac317963f6ea48c4 : POST_FAILURE in 1h 56m 38s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/f60aadb19b29403c8dda5b1b90dcc828 : SUCCESS in 51m 18s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/81dfd96f66ad475a8e186f86bce1fcf0 : SUCCESS in 1h 09m 03s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/2451ceb60ef9462c8f55afc7e7766a2f : SUCCESS in 1h 21m 48s","accounts_in_message":[],"_revision_number":17},{"id":"06565a6c2ac1700b8b2faf55e5302dbc58d53d85","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-10-21 07:40:49.000000000","message":"Patch Set 17: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"e48e3ef26aa0163bc10deac169076e5f62b61646","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-21 11:16:00.000000000","message":"Patch Set 17:\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"5fc4403ea9bd78df45204e200b8134a73ec7c549","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-21 12:03:54.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"6ff48535c74c52dfa9b8d07afeb59a42cd88106e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 14:06:54.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a44b995affcc4aae92cacf5adc2c5902 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cacd691ee7994712b92cee0e49742998 : SUCCESS in 4m 22s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/83f44d57d82d44a8befe19415d4412bb : SUCCESS in 1h 14m 02s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/dbaa9ec044714f8a897f6b6f20ae3a27 : SUCCESS in 1h 08m 22s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/aab274be2a9c41b9a6b380ece3787cc9 : SUCCESS in 1h 14m 06s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/56ff1b2d7ba94e7c81f57ad99bcc52d3 : SUCCESS in 1h 14m 43s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/a6ffe1cce5c24bd0ad5052caaa78c3e9 : SUCCESS in 1h 29m 43s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/1228b43469294e8db1739a7acafdc6ae : SUCCESS in 1h 28m 46s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/017895da02884033b77ce8a315d7f1fe : SUCCESS in 1h 09m 58s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/53c86f0048304248a2bf4d6adc811e3d : SUCCESS in 1h 02m 37s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/8eed71e36aeb4cdd95768db5b46621b2 : FAILURE in 1h 46m 39s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/c6ea794f0c5e4d58803f780408bae924 : POST_FAILURE in 1h 58m 23s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/c77395d9754d432291a272a670a3546e : SUCCESS in 1h 23m 58s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/65bb86102c274f96a94f5f2424860406 : FAILURE in 1h 23m 57s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/d0f2f76b98144d6d83d851e4cf536b4b : FAILURE in 1h 09m 38s","accounts_in_message":[],"_revision_number":18},{"id":"5a4854857a978b4b4ac3bbad67676aa73025ae93","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-21 14:47:33.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"8158f2d542aca29abf45a95f6f7e24083cb1c38a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 17:06:23.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c2d3509dcd0457990856ef5622167e6 : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1123f5d9fe714130bc6e757bdae7648a : SUCCESS in 5m 01s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/d9714cc06c2d40d4b7cc30348740243b : SUCCESS in 55m 05s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/0108661b6e3e426a988e4d4f21a2b4df : SUCCESS in 1h 14m 10s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/55b9239076dd4fdc819f030081234ecf : SUCCESS in 1h 22m 00s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/54d7dc9062484065bad90f42b355009a : SUCCESS in 1h 16m 11s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/d289ee3b5bf84bedafb7734763d98ce6 : SUCCESS in 1h 28m 39s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/b2e46df16c894fc686352fd0d04f6995 : SUCCESS in 1h 41m 19s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/767de7b6ff2c471e9bdd86fa749cdca6 : SUCCESS in 1h 09m 48s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/ee2cada32b5a476993e4e9d04d57420a : SUCCESS in 1h 17m 00s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/95864d42671d42c68cf2eccee9b82ec7 : SUCCESS in 1h 16m 09s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/646a99b38f9d4ab6b45e6c3dc0172733 : POST_FAILURE in 2h 11m 31s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/86b82faa4ed247a2b9330e2b53ac6bef : SUCCESS in 1h 15m 30s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/c2c49b4588ff489a8b801fbc660f12fe : SUCCESS in 1h 38m 40s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/79748d8e49a34c8ab4a4e1c68a4e2d5c : SUCCESS in 1h 41m 19s","accounts_in_message":[],"_revision_number":18},{"id":"b20a3fa7cda366cb03336ce001ece4da566c45ad","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-10-22 07:18:03.000000000","message":"Patch Set 18: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":18},{"id":"81dbc10b480f423f9f69e290eaf99c9f1460e40b","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-10-22 07:25:44.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"954cadb0b96a2dbe8d845ff7cfa66c352e466a5d","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-22 09:43:54.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"5fa7d5ebdcc18b0c42c9d13b0c7bb2dca4b4e2e5","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-22 10:00:01.000000000","message":"Patch Set 18:\n\n(2 comments)","accounts_in_message":[],"_revision_number":18},{"id":"2a5ae1d4f16c43164005256025ea57849011a341","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-22 10:00:55.000000000","message":"Patch Set 19:\n\nFor some reason I have overridden the changes in patchset 16 so adding them here","accounts_in_message":[],"_revision_number":19},{"id":"74484416e30d10c21f2a01291aee62dd2dbd08c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 12:15:29.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7066b449ef71491b85f740d0166d44cf : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80b69fa824c742338b025f7a76a96898 : SUCCESS in 5m 07s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/65cd3196476e445ab3b711607469b03d : SUCCESS in 1h 20m 58s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/b74d978d9cc547399e74eaecda46a1fa : SUCCESS in 1h 22m 39s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/ea246ed356e948338fc3765589a3cc7d : SUCCESS in 1h 05m 38s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/aa357724a94a43bfb998796048566668 : SUCCESS in 1h 22m 48s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/6adcf472beff4afa9a0b2db3d37cbe2f : SUCCESS in 1h 28m 13s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/749be8953b6c488694d678e4c9e76fd8 : SUCCESS in 1h 18m 20s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/367bf8a3cf7a4d9c8789c25697830661 : SUCCESS in 1h 09m 34s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/19202b6b3ff74e1394d2b35d044a47fd : SUCCESS in 1h 07m 44s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/15b0e4005ce64a07aa259e7abb655b58 : SUCCESS in 1h 11m 31s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/8ef1f07601264ca08fd5586825f88404 : POST_FAILURE in 1h 56m 38s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/1f3307f127c24b73930f07d1a233fa8e : SUCCESS in 1h 15m 30s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/93a715e7337149ffa17159a5f4ba27d6 : SUCCESS in 1h 19m 22s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/e8756d17644d42f6a78c4ef3747ca5d8 : SUCCESS in 1h 50m 26s","accounts_in_message":[],"_revision_number":19},{"id":"ddf810e6ba008e7a0292fffa0779cb925d05c2d2","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-10-26 19:31:50.000000000","message":"Patch Set 19: Code-Review-1\n\n(2 comments)\n\nHere goes a few thoughts:","accounts_in_message":[],"_revision_number":19},{"id":"e7170303181b05a416f4e07ad79d76b2d033c8e1","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-27 12:57:39.000000000","message":"Patch Set 19:\n\n(2 comments)","accounts_in_message":[],"_revision_number":19},{"id":"523996bd1b3cf49d0cbac75e3dc99aa53be18941","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-27 12:58:45.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"8333d9d0118ecd38abefc6db57711bc1f18967c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 15:33:52.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc256417877b46fb8c6fb26155ae2a1f : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6788292a828477da32becc2f1dd0d8c : SUCCESS in 3m 48s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/7a102eb837d244b3b95862569c1d6673 : SUCCESS in 1h 02m 20s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/0ab1600274a842b2a060250be8256a32 : SUCCESS in 1h 07m 30s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/07e804698e7b4d6193ec488a46f93c89 : SUCCESS in 2h 19m 15s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/72566d681d4d44c199b31f290894c4aa : SUCCESS in 1h 13m 27s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/ee064faf827d4325b3d67270b9f5209d : SUCCESS in 2h 25m 19s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/1a4a8942b4364322a7fdfd10a64760db : SUCCESS in 1h 16m 25s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/6daddc2381424353b0a6d9af2f3a0baa : SUCCESS in 1h 06m 42s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/35f05b8108254db0bb18c584a28471b6 : SUCCESS in 1h 09m 39s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/6244c8274ecb4d0a94a28e4ef838f0a0 : FAILURE in 1h 33m 11s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/6481914f83224feeb00041a0ce1a3033 : POST_FAILURE in 1h 56m 47s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/d57c88d9fa8d4c2292c208b368b03cab : SUCCESS in 1h 03m 04s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/e3ad5d0120ab48449e07d1095d56fd84 : SUCCESS in 1h 09m 05s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/ee6e42d56e8a40e896284426b95fe9bf : SUCCESS in 1h 22m 20s","accounts_in_message":[],"_revision_number":20},{"id":"05bf55b0e918dabe0ab153ed584e2b15844b6409","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-10-27 17:10:06.000000000","message":"Patch Set 20: Code-Review-1\n\n(5 comments)\n\nSome minor remarks and one crucial - the helper methods have to be defined on the base class if the test is in the base class. Otherwise you\u0027ll break the OldNetworkPolicyScenario.","accounts_in_message":[],"_revision_number":20},{"id":"f24f565fa72a331d26bb12e325b8000f8f9ab3b9","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-28 05:26:38.000000000","message":"Patch Set 20:\n\n(4 comments)","accounts_in_message":[],"_revision_number":20},{"id":"2ea9c4c36bc19331140a77e688e59a4543a78343","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-28 05:27:30.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"4a85508b9beabf71b6eb4ba314b785c3dd683955","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 07:41:06.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1a5d7af766f468091189c5bf7fe0110 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d9aedc2be6f41fd9ef9859e676ae124 : SUCCESS in 2m 56s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/c6fd64f2c73e4c129b2aa09dadc414c8 : FAILURE in 1h 41m 45s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/629b883f8cb042f88b5fc9f86d1c7108 : SUCCESS in 1h 27m 44s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/5abb7c476ad742debfbf0a9a0db9cab7 : SUCCESS in 1h 04m 52s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/524b357e5fe14dca971998ccb43775a0 : FAILURE in 1h 09m 05s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/12e55cbd9d7e4bab85c5d8beac5065b6 : SUCCESS in 1h 35m 15s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/ab0837d4ba5c4ed9988c0636dd0c831b : SUCCESS in 1h 32m 30s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/cceccbe9f1974fddaccadf28f7dfaca3 : SUCCESS in 1h 04m 33s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/ff27c4bf8c624aa1b5483d2f3465b494 : SUCCESS in 1h 21m 26s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/ad1be12f345b43b8bb66f37542adf6a8 : SUCCESS in 1h 19m 08s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/ab38ec82ff954c3e954f13e8a4663bde : POST_FAILURE in 2h 00m 54s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/fec345520e1f4b28bbd1c1ca655aede9 : SUCCESS in 1h 18m 31s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/382b99c0c2574058aca1d1a08ae63024 : SUCCESS in 1h 24m 58s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/470c93e8130f4862b737ec3648744862 : SUCCESS in 1h 27m 23s","accounts_in_message":[],"_revision_number":21},{"id":"24e71974743e3124ced8ee7b09b2b8de01713544","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-10-28 08:17:59.000000000","message":"Patch Set 21:\n\nrecheck","accounts_in_message":[],"_revision_number":21},{"id":"702a8e925039693207aa1ea81aa42803167c7f75","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-10-28 08:19:44.000000000","message":"Patch Set 21: Code-Review+2","accounts_in_message":[],"_revision_number":21},{"id":"2f2e5e52b3740f5bfcc91dff14823b7175ccffa3","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-28 08:48:05.000000000","message":"Patch Set 21:\n\nrecheck","accounts_in_message":[],"_revision_number":21},{"id":"227ef02b5758cd66038d4f1c097286b3cd5fe029","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 11:30:04.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e11b08d056594f8d9bacbe8218ff6bf2 : SUCCESS in 10m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5e065b6804a457983d67a28d2f01c53 : SUCCESS in 9m 42s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/e576507df50c41ee859f6728bd81e3d7 : SUCCESS in 2h 02m 24s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/9d5e8d813f3549618b3947a6d445eaab : SUCCESS in 2h 08m 16s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/0c28f759b808427b963c7762702cbf40 : FAILURE in 1h 59m 24s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/eacf61c393bf4cbcb22f03fac8034597 : SUCCESS in 1h 18m 11s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/ab42dd6dcacd4acda62ebd174e2e52e0 : SUCCESS in 1h 22m 13s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/fca16e01113d477787277367bcd8b844 : SUCCESS in 1h 37m 33s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/27741e5688554baa9f0221df5fc4d1ed : SUCCESS in 1h 02m 38s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/d7148cbbf4cb4cce94ca37092dbb6598 : SUCCESS in 1h 18m 54s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/e1ddda3574b240899d62603197a518d4 : FAILURE in 1h 17m 46s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/cdec9c76feea474185d0724a6896189c : POST_FAILURE in 1h 57m 32s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/7ddcb000dcfa4489bd8bde3a18973f16 : SUCCESS in 54m 30s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/430f5cdf4fd744aabecb61b84f1a5f9c : SUCCESS in 1h 30m 22s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/aeb365ec58074358b02f2a2c44c0b8c0 : SUCCESS in 3h 00m 39s","accounts_in_message":[],"_revision_number":21},{"id":"898450f30fa3b254d77bc63a13fdc139a2201ec2","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-10-29 07:42:34.000000000","message":"Patch Set 21: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"999edb4240b742d376f6aa832db1f10952ccead9","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-29 07:55:30.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"ddf901f1a72bcc83629eeb498cbde77238f62028","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-29 07:55:49.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"b207b035359a7a17ed20b572561fcc5674d709e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 10:13:21.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b5870aff49841eb813229f3a9d41210 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ffaac6481c1432b88f9382616866427 : SUCCESS in 4m 11s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/a3530f1f0d9e42c9b57046dece17b21f : SUCCESS in 1h 13m 41s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/c691459ad5484a40be1de6d3f48242b9 : SUCCESS in 1h 07m 21s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/5bb2cda8c5ca426dac5b5aaf12665745 : SUCCESS in 1h 10m 51s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/36f2e8792f0f419b97ebeba7d1375287 : SUCCESS in 1h 04m 56s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/d23cad25e68b44abac513470c6f09a85 : SUCCESS in 1h 49m 50s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/42b254150a914a7f89a47e67d9116312 : SUCCESS in 2h 05m 26s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/1136aeb585a44e4ab6b319d4a16d37ee : SUCCESS in 1h 25m 26s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/538a0a1df8bb496dac77ac504f91bb54 : SUCCESS in 1h 19m 50s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/fb970c0c13c04a3e96a08e2a06e64569 : SUCCESS in 1h 22m 25s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/17abfb83ec0f4875a389d7ed649b7034 : POST_FAILURE in 2h 09m 59s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/09629221645b41539c20312123cc4098 : SUCCESS in 1h 06m 02s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/c6471d6111494c64883b54931ea4b4ef : SUCCESS in 1h 40m 27s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/b6e70419d8514974aa819ffec3f42a75 : SUCCESS in 1h 27m 21s","accounts_in_message":[],"_revision_number":22},{"id":"07467d9d171856e1133789227301308f575c82f4","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-10-29 14:00:49.000000000","message":"Patch Set 22: Workflow-1\n\nWe have an issue in d/s. Need to investigate","accounts_in_message":[],"_revision_number":22},{"id":"213f90961dfb9a958bd56a513432820169ab2df9","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-10-29 14:34:52.000000000","message":"Patch Set 22: Code-Review+2","accounts_in_message":[],"_revision_number":22},{"id":"9e5ab45c526a244587f967d1b82c312b5e4f6cd0","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-11-02 11:12:52.000000000","message":"Patch Set 22: Code-Review-1\n\n(5 comments)\n\nThanks, Itzik. The test scenario looks great, I just have a few suggestions:","accounts_in_message":[],"_revision_number":22},{"id":"104f37fbcc215b794aa9dcb6cc8534d33af931a6","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-11-02 12:33:54.000000000","message":"Patch Set 22:\n\n(4 comments)","accounts_in_message":[],"_revision_number":22},{"id":"d0d23ca2eaeebd02331f4e09c4f2103d8acccb11","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-11-02 12:34:21.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"4a6634a053611c58092e933b80376d5e4ea2ea33","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-11-02 13:01:59.000000000","message":"Patch Set 23: Workflow-1","accounts_in_message":[],"_revision_number":23},{"id":"4cc8d4f34050e66fb1ba0f7c654a57a81c0fd67f","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-11-02 13:50:33.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"4129e44234c2e25c949eea7b430c5230fc82543c","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-11-02 15:08:37.000000000","message":"Patch Set 24: Code-Review+2","accounts_in_message":[],"_revision_number":24},{"id":"aceb3123ae543025d5943a0eea8a743d53fb093e","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-11-02 15:25:25.000000000","message":"Patch Set 24: Code-Review+2","accounts_in_message":[],"_revision_number":24},{"id":"0b7339df647a9b069ab7a031d03cb4c924524a77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 15:53:13.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ccd89ee06df4f73ba882748810646b4 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcd10981bef2415ab881ba4d2bc89bae : SUCCESS in 3m 04s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/b19a11c837434a98ad6c0da9db3f00db : SUCCESS in 1h 01m 12s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/50680ebc52144f0184635761f23805e1 : SUCCESS in 1h 15m 59s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/977a5d99c97a4d55a0fb11c20a62b531 : SUCCESS in 1h 27m 57s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/cb0f64afdb564c2e8c417331c97456ef : SUCCESS in 1h 13m 31s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/a3ab53d6ae9a4b268fa82753124fb4cb : SUCCESS in 1h 37m 57s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/45a557d809a84bc58431d213fc5d80e3 : SUCCESS in 1h 23m 13s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/885e4c72ece94adebe3c8fd37a8f5c9b : SUCCESS in 1h 01m 28s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/c5dee33cbab147e6b94bd5e3742bfaf3 : SUCCESS in 1h 04m 59s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/e790439bdad140c4bcb79b1e3e372dd8 : SUCCESS in 1h 19m 06s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/287e5e725f384aef829c2a0cbc629605 : POST_FAILURE in 1h 58m 56s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/ef4df52438bc4636bbd4528ad823c419 : SUCCESS in 1h 03m 33s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/50a9febd185f47a8b9b0b80f7c4f6e07 : SUCCESS in 1h 17m 59s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/bd6a6ef413d34df292f483c0e277e858 : SUCCESS in 1h 36m 16s","accounts_in_message":[],"_revision_number":24},{"id":"e49a77ec61a23c0365b4eb57c486ff5d12e8960c","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-11-02 16:02:52.000000000","message":"Patch Set 24: Workflow+1","accounts_in_message":[],"_revision_number":24},{"id":"a37dd9172848b50e365f65cf8a543261ce2cfae5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 16:03:04.000000000","message":"Patch Set 24: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":24},{"id":"dc0016b036da9d06e4676f444df506250fe2a0ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 17:49:29.000000000","message":"Patch Set 24: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/919984654b5045da85857a88858f0d3f : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9bf500d5e764fa3bc4a0c9fa5804112 : SUCCESS in 3m 14s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/92919e4a8a024d2f92d8c4b6bc2b2ee9 : SUCCESS in 1h 11m 50s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/1da1e81701e64377b89d465258525f15 : SUCCESS in 1h 17m 08s\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/3c1f9a9e49ba40c594897a401acd01a4 : FAILURE in 1h 15m 12s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/53f83a5343bf46638c13f230219e832d : FAILURE in 1h 43m 37s","accounts_in_message":[],"_revision_number":24},{"id":"78d14b5417a439a9def775174a6b5f0c23880650","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-11-02 18:41:01.000000000","message":"Patch Set 24:\n\nrecheck","accounts_in_message":[],"_revision_number":24},{"id":"24915229283ed1f824eb69c6a5d31673f3e4889e","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-11-02 19:36:58.000000000","message":"Patch Set 24:\n\nrecheck","accounts_in_message":[],"_revision_number":24},{"id":"8e388835ee2a3c543c5e3ffff896a432ca16720a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 21:16:20.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bce4a2a7c38c4d00ae64cfcf755478b4 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80ae2298a31d4800b39b0db0ec376dd3 : SUCCESS in 3m 20s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/79eb4e540aec43619633df2daada2e92 : SUCCESS in 56m 45s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/4a0fc8b125c145d096e8534ccd61c6c3 : SUCCESS in 1h 13m 21s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/fed5fffa0e73465daedba2f5ed4fa075 : SUCCESS in 1h 07m 56s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/7dd32b8a4bf84da28a731b174b0d1f17 : FAILURE in 1h 02m 24s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/891422c5bdb44c6992c65c1972c54bc8 : FAILURE in 2h 30m 34s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/8ee274dd3b9846d1a8af3cba90959b14 : SUCCESS in 1h 23m 06s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/3b8d9099bb2d47eaa0c0bb9f4dcec8bf : SUCCESS in 1h 00m 19s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/7753d4a82dd6477a82b9c78791ad9e14 : SUCCESS in 1h 07m 50s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/6b598942e36a42bba81eb3a45063555c : SUCCESS in 1h 21m 53s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/b25e7a4fc8404f998d4668d527537266 : POST_FAILURE in 2h 00m 07s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/35cbe2e50f454e5c90bb79ae88b97e5f : SUCCESS in 54m 18s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/62ba6066f89647258a55227ba59c709b : SUCCESS in 1h 15m 16s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/8d002715b8d34a5e9cc7382df60d46ae : SUCCESS in 1h 24m 31s","accounts_in_message":[],"_revision_number":24},{"id":"7a4c4f8b771b82a6adf58989eb7c8b57c01131ea","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-11-03 04:13:47.000000000","message":"Patch Set 24:\n\nrecheck","accounts_in_message":[],"_revision_number":24},{"id":"b176e2bd0b560d9f6013793553ebe5c013d9c72c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 06:13:50.000000000","message":"Patch Set 24:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b32670018f084aa5be4cf990ae269be7 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c324f4a0417461ba08b5d0d9cef8d2c : SUCCESS in 3m 34s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/8d827981eb5f4a26bec5f7065f586428 : SUCCESS in 56m 05s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/41483545b6c94da7a31ae9c870994439 : SUCCESS in 1h 11m 55s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/6c295cf7436b4f159d18188717444c01 : SUCCESS in 1h 24m 09s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/3bb7c68d489d4e4ab7370e51d1ecd751 : SUCCESS in 1h 09m 55s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/18ceb517f4e6413e8dba27173bbf7a13 : SUCCESS in 1h 25m 01s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/610cf802f2be41db9f8bcf91d664c37c : SUCCESS in 1h 40m 51s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/1d3bcb71534548ce9e7200398c117858 : FAILURE in 1h 02m 00s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/9ceba5cf075f4f658541a581539c9bb9 : SUCCESS in 58m 10s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/93ecdb09c04e4f79b25b56deb78cdee4 : FAILURE in 1h 31m 39s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/3c710535d1bf4c279a6a40049dade5bd : POST_FAILURE in 1h 55m 58s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/c3a0dd23931049f0965a4ccc28c7c055 : SUCCESS in 54m 40s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/91397c70a4574be8a4e89977addb730b : SUCCESS in 1h 14m 48s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/9434eb17836d4b6b85c17a99e1c63b38 : FAILURE in 1h 28m 38s","accounts_in_message":[],"_revision_number":24},{"id":"2049e4dcf1738634f69ca67152cdaff9dc027267","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2020-11-03 06:34:13.000000000","message":"Patch Set 24: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":24},{"id":"a16ea36fdc01195db4b34fc1d975b41582f6bbd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 09:38:25.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1af6bec9c86547ba963388983d2bfbe9 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/434106c9c8024c52aa2e5fdb13d25f10 : SUCCESS in 3m 31s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/e02e6acd78354b98bcb7e08ba5e9ceeb : SUCCESS in 58m 02s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/b160b4bf3e7e47b2bdc053c890a00416 : SUCCESS in 1h 34m 13s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/83a8c0ab0111411aa408e37dd447d136 : SUCCESS in 1h 27m 05s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/2e42e29b515043d4b770f27be0f69ed7 : SUCCESS in 1h 08m 45s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/922179131fdd49f58a6ed15ead5613f1 : SUCCESS in 1h 29m 24s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/0003906824ed45a2b22d44037e9674b7 : SUCCESS in 1h 42m 48s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/819719d8badd4026baba9dac4054e95c : SUCCESS in 1h 09m 55s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/95360bf4411149968b4d33ea29c9e4e3 : SUCCESS in 1h 06m 30s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/431ba380e0614c27a368c9bb52b218af : SUCCESS in 1h 09m 08s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/08a27377e1a84c788cd4ff4e5a2b1ef0 : POST_FAILURE in 1h 57m 37s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/1a88a0846d5e4fe082f640e866392a4c : SUCCESS in 1h 06m 56s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/97e26e14e5e0413e91a7c3de681e52c1 : SUCCESS in 1h 22m 57s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/42c9d9c3f2964746840557cfa0ed38b8 : SUCCESS in 1h 18m 54s","accounts_in_message":[],"_revision_number":24},{"id":"6d723909f81ff7d2205acda51b24aed93588d363","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 09:38:37.000000000","message":"Patch Set 24: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":24},{"id":"5377437f8f3d98b7a7535d19d6cbeb24ad6e3459","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 11:05:44.000000000","message":"Patch Set 24: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ead12ff7f89640f9bd5efdc4c42f1a6e : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/937daca9819d488c82338c694f5658b2 : SUCCESS in 4m 49s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/4c6b0b804458454bbb4230451c7b507f : SUCCESS in 1h 08m 17s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/c264d336a9d24c6dac9d3dd6666a275a : SUCCESS in 1h 07m 36s\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/1b520ef66368404796a37727008e477e : SUCCESS in 1h 10m 58s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/52cfe87d37844d8db1b4f77ca2dcad5b : SUCCESS in 1h 18m 53s","accounts_in_message":[],"_revision_number":24},{"id":"ad0a77905e41d2f5347eb5f1f6d62ccdca462781","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 11:05:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":24},{"id":"e56289481193f8bda7fb5c397213d0f4ad24b5e5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 11:07:02.000000000","message":"Patch Set 24:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed23c203781e4e51baeafc29f94b021a : SUCCESS in 52s","accounts_in_message":[],"_revision_number":24}],"current_revision_number":24,"current_revision":"28fbe115ebd97760b128d4c9efd21d7d426922d3","revisions":{"27b2ed53b8326b00c8603f1672a4360bc52b68fc":{"kind":"REWORK","_number":1,"created":"2020-09-29 09:55:27.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/56/754956/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/56/754956/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/1"}}},"commit":{"parents":[{"commit":"2fe1791ff2e6283c7f8d92e10fcaffa104457a95","subject":"Remove skip of test_service_udp_ping","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2fe1791ff2e6283c7f8d92e10fcaffa104457a95"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-09-29 09:48:08.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-09-29 09:55:08.000000000","tz":180},"subject":"[WIP] Add test_network_policy_add_remove_pod","message":"[WIP] Add test_network_policy_add_remove_pod\n\n- Creating a service\n- Creating a network policy\n- Deleting one of the service pods and creating a new one\n- Check connectivity to the service\n\nChange-Id: Ibf8113d9667c02a1b6d080aa50e91fde0809fd75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/27b2ed53b8326b00c8603f1672a4360bc52b68fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/27b2ed53b8326b00c8603f1672a4360bc52b68fc"}]},"branch":"refs/heads/master"},"826cbf77df04d7836fab608d432e1b2b4d9668f3":{"kind":"REWORK","_number":2,"created":"2020-09-29 12:33:00.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/56/754956/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/56/754956/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/2"}}},"commit":{"parents":[{"commit":"2fe1791ff2e6283c7f8d92e10fcaffa104457a95","subject":"Remove skip of test_service_udp_ping","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2fe1791ff2e6283c7f8d92e10fcaffa104457a95"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-09-29 09:48:08.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-09-29 12:32:48.000000000","tz":180},"subject":"Add test_network_policy_add_remove_pod","message":"Add test_network_policy_add_remove_pod\n\n- Creating a service\n- Creating a network policy\n- Deleting one of the service pods and creating a new one\n- Check connectivity to the service\n\nChange-Id: Ibf8113d9667c02a1b6d080aa50e91fde0809fd75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/826cbf77df04d7836fab608d432e1b2b4d9668f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/826cbf77df04d7836fab608d432e1b2b4d9668f3"}]},"branch":"refs/heads/master"},"a8a21c0d4ff1efb1fe725ba927abff1704c6e853":{"kind":"REWORK","_number":3,"created":"2020-09-29 13:34:17.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/56/754956/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/56/754956/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/3"}}},"commit":{"parents":[{"commit":"2fe1791ff2e6283c7f8d92e10fcaffa104457a95","subject":"Remove skip of test_service_udp_ping","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2fe1791ff2e6283c7f8d92e10fcaffa104457a95"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-09-29 09:48:08.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-09-29 13:29:59.000000000","tz":180},"subject":"Add test_network_policy_add_remove_pod","message":"Add test_network_policy_add_remove_pod\n\n- Create a service and check connectivity to the service\n- Create a network policy with a podSelector with labels of the pods in the\n  service as well as for the tester pod and check connectivity to the service\n- Deleting one of the service pods and creating a new one\n- Check connectivity to the service and verify only one pod answering\n- Create a new pod with the same label as the one we deleted\n- Check connectivity to the service, expecting two pods answering\n\nChange-Id: Ibf8113d9667c02a1b6d080aa50e91fde0809fd75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/a8a21c0d4ff1efb1fe725ba927abff1704c6e853"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/a8a21c0d4ff1efb1fe725ba927abff1704c6e853"}]},"branch":"refs/heads/master"},"bb8768b279f95fd495955ea3c5a6af2d366d34e5":{"kind":"REWORK","_number":4,"created":"2020-10-01 06:39:57.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/56/754956/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/56/754956/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/4"}}},"commit":{"parents":[{"commit":"2fe1791ff2e6283c7f8d92e10fcaffa104457a95","subject":"Remove skip of test_service_udp_ping","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2fe1791ff2e6283c7f8d92e10fcaffa104457a95"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-09-29 09:48:08.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-10-01 06:39:29.000000000","tz":180},"subject":"Add test_network_policy_add_remove_pod","message":"Add test_network_policy_add_remove_pod\n\n- Create a service and check connectivity to the service\n- Create a network policy with a podSelector with labels of the pods in the\n  service as well as for the tester pod and check connectivity to the service\n- Deleting one of the service pods and creating a new one\n- Check connectivity to the service and verify only one pod answering\n- Create a new pod with the same label as the one we deleted\n- Check connectivity to the service, expecting two pods answering\n\nChange-Id: Ibf8113d9667c02a1b6d080aa50e91fde0809fd75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/bb8768b279f95fd495955ea3c5a6af2d366d34e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/bb8768b279f95fd495955ea3c5a6af2d366d34e5"}]},"branch":"refs/heads/master"},"365029c4e09b8446eb78e09d21da425ab7104ba2":{"kind":"REWORK","_number":5,"created":"2020-10-04 09:41:42.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/56/754956/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/56/754956/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/5"}}},"commit":{"parents":[{"commit":"2fe1791ff2e6283c7f8d92e10fcaffa104457a95","subject":"Remove skip of test_service_udp_ping","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2fe1791ff2e6283c7f8d92e10fcaffa104457a95"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-09-29 09:48:08.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-10-04 09:41:21.000000000","tz":180},"subject":"Add test_network_policy_add_remove_pod","message":"Add test_network_policy_add_remove_pod\n\n- Create a service and check connectivity to the service\n- Create a network policy with a podSelector with labels of the pods in the\n  service as well as for the tester pod and check connectivity to the service\n- Deleting one of the service pods and creating a new one\n- Check connectivity to the service and verify only one pod answering\n- Create a new pod with the same label as the one we deleted\n- Check connectivity to the service, expecting two pods answering\n\nChange-Id: Ibf8113d9667c02a1b6d080aa50e91fde0809fd75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/365029c4e09b8446eb78e09d21da425ab7104ba2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/365029c4e09b8446eb78e09d21da425ab7104ba2"}]},"branch":"refs/heads/master"},"31ec1c8927a55371ed4b5bfaac8cc12a685bb983":{"kind":"REWORK","_number":6,"created":"2020-10-04 16:04:33.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/56/754956/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/56/754956/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/6"}}},"commit":{"parents":[{"commit":"2fe1791ff2e6283c7f8d92e10fcaffa104457a95","subject":"Remove skip of test_service_udp_ping","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2fe1791ff2e6283c7f8d92e10fcaffa104457a95"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-09-29 09:48:08.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-10-04 16:04:20.000000000","tz":180},"subject":"Add test_network_policy_add_remove_pod","message":"Add test_network_policy_add_remove_pod\n\n- Create a service and check connectivity to the service\n- Create a network policy with a podSelector with labels of the pods in the\n  service as well as for the tester pod and check connectivity to the service\n- Deleting one of the service pods and creating a new one\n- Check connectivity to the service and verify only one pod answering\n- Create a new pod with the same label as the one we deleted\n- Check connectivity to the service, expecting two pods answering\n\nChange-Id: Ibf8113d9667c02a1b6d080aa50e91fde0809fd75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/31ec1c8927a55371ed4b5bfaac8cc12a685bb983"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/31ec1c8927a55371ed4b5bfaac8cc12a685bb983"}]},"branch":"refs/heads/master"},"52eb3c55bea858781580050ae3ef6983cffd677f":{"kind":"REWORK","_number":7,"created":"2020-10-05 09:15:43.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/56/754956/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/56/754956/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/7"}}},"commit":{"parents":[{"commit":"2fe1791ff2e6283c7f8d92e10fcaffa104457a95","subject":"Remove skip of test_service_udp_ping","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2fe1791ff2e6283c7f8d92e10fcaffa104457a95"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-09-29 09:48:08.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-10-05 09:14:21.000000000","tz":180},"subject":"Add test_network_policy_add_remove_pod","message":"Add test_network_policy_add_remove_pod\n\n- Create a service and check connectivity to the service\n- Create a network policy with a podSelector with labels of the pods in the\n  service as well as for the tester pod and check connectivity to the service\n- Deleting one of the service pods and creating a new one\n- Check connectivity to the service and verify only one pod answering\n- Create a new pod with the same label as the one we deleted\n- Check connectivity to the service, expecting two pods answering\n\nChange-Id: Ibf8113d9667c02a1b6d080aa50e91fde0809fd75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/52eb3c55bea858781580050ae3ef6983cffd677f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/52eb3c55bea858781580050ae3ef6983cffd677f"}]},"branch":"refs/heads/master"},"119ded429fa30506107aa5b95d446d6497ad92c8":{"kind":"REWORK","_number":8,"created":"2020-10-07 08:49:20.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/56/754956/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/56/754956/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/8"}}},"commit":{"parents":[{"commit":"2fe1791ff2e6283c7f8d92e10fcaffa104457a95","subject":"Remove skip of test_service_udp_ping","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2fe1791ff2e6283c7f8d92e10fcaffa104457a95"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-09-29 09:48:08.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-10-07 08:48:54.000000000","tz":180},"subject":"Add test_network_policy_add_remove_pod","message":"Add test_network_policy_add_remove_pod\n\n- Create a service and check connectivity to the service\n- Create a network policy with a podSelector with labels of the pods in the\n  service as well as for the tester pod and check connectivity to the service\n- Deleting one of the service pods and creating a new one\n- Check connectivity to the service and verify only one pod answering\n- Create a new pod with the same label as the one we deleted\n- Check connectivity to the service, expecting two pods answering\n\nChange-Id: Ibf8113d9667c02a1b6d080aa50e91fde0809fd75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/119ded429fa30506107aa5b95d446d6497ad92c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/119ded429fa30506107aa5b95d446d6497ad92c8"}]},"branch":"refs/heads/master"},"c6f2a324118c4e778e75054a6d53648f1f092ffb":{"kind":"REWORK","_number":9,"created":"2020-10-07 10:28:14.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/56/754956/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/56/754956/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/9"}}},"commit":{"parents":[{"commit":"2fe1791ff2e6283c7f8d92e10fcaffa104457a95","subject":"Remove skip of test_service_udp_ping","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2fe1791ff2e6283c7f8d92e10fcaffa104457a95"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-09-29 09:48:08.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-10-07 10:28:00.000000000","tz":180},"subject":"Add test_network_policy_add_remove_pod","message":"Add test_network_policy_add_remove_pod\n\n- Create a service and check connectivity to the service\n- Create a network policy with a podSelector with labels of the pods in the\n  service as well as for the tester pod and check connectivity to the service\n- Deleting one of the service pods and creating a new one\n- Check connectivity to the service and verify only one pod answering\n- Create a new pod with the same label as the one we deleted\n- Check connectivity to the service, expecting two pods answering\n\nChange-Id: Ibf8113d9667c02a1b6d080aa50e91fde0809fd75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c6f2a324118c4e778e75054a6d53648f1f092ffb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c6f2a324118c4e778e75054a6d53648f1f092ffb"}]},"branch":"refs/heads/master"},"ca82b0f2e05504fdd32bf863e7bca16f04b137b5":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2020-10-08 06:36:36.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/56/754956/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/56/754956/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/10"}}},"commit":{"parents":[{"commit":"2fe1791ff2e6283c7f8d92e10fcaffa104457a95","subject":"Remove skip of test_service_udp_ping","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2fe1791ff2e6283c7f8d92e10fcaffa104457a95"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-09-29 09:48:08.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-10-08 06:36:21.000000000","tz":180},"subject":"Add test_network_policy_add_remove_pod","message":"Add test_network_policy_add_remove_pod\n\n- Create a service and check connectivity to the service\n- Create a network policy with a podSelector with labels of the pods in the\n  service as well as for the tester pod and check connectivity to the service\n- Deleting one of the service pods and creating a new one\n- Check connectivity to the service and verify only one pod answering\n- Create a new pod with the same label as the one we deleted\n- Check connectivity to the service, expecting two pods answering\n\nChange-Id: Ibf8113d9667c02a1b6d080aa50e91fde0809fd75\nDepends-On: If23b311ed07578b3fbe85f46aa4a314e6a05b7f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ca82b0f2e05504fdd32bf863e7bca16f04b137b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ca82b0f2e05504fdd32bf863e7bca16f04b137b5"}]},"branch":"refs/heads/master"},"cb2046abd051841f6aef8c62dc40115d1c23f75e":{"kind":"REWORK","_number":11,"created":"2020-10-11 02:01:00.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/56/754956/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/56/754956/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/11"}}},"commit":{"parents":[{"commit":"2fe1791ff2e6283c7f8d92e10fcaffa104457a95","subject":"Remove skip of test_service_udp_ping","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2fe1791ff2e6283c7f8d92e10fcaffa104457a95"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-09-29 09:48:08.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-10-11 01:59:22.000000000","tz":180},"subject":"Add test_network_policy_add_remove_pod","message":"Add test_network_policy_add_remove_pod\n\n- Create a service and check connectivity to the service\n- Create a network policy with a podSelector with labels of the pods in the\n  service as well as for the tester pod and check connectivity to the service\n- Deleting one of the service pods and creating a new one\n- Check connectivity to the service and verify only one pod answering\n- Create a new pod with the same label as the one we deleted\n- Check connectivity to the service, expecting two pods answering\n\nAlso adds an option to create_network_policy to apply a rule to allow\ntraffic only from pods with specific labels\n\nChange-Id: Ibf8113d9667c02a1b6d080aa50e91fde0809fd75\nDepends-On: If23b311ed07578b3fbe85f46aa4a314e6a05b7f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/cb2046abd051841f6aef8c62dc40115d1c23f75e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/cb2046abd051841f6aef8c62dc40115d1c23f75e"}]},"branch":"refs/heads/master"},"9607acf30952664b84e4f672c78ebd9aaa6845c4":{"kind":"REWORK","_number":12,"created":"2020-10-14 14:59:40.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/56/754956/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/56/754956/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/12"}}},"commit":{"parents":[{"commit":"2fe1791ff2e6283c7f8d92e10fcaffa104457a95","subject":"Remove skip of test_service_udp_ping","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2fe1791ff2e6283c7f8d92e10fcaffa104457a95"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-09-29 09:48:08.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-10-14 14:58:47.000000000","tz":-240},"subject":"Add test_network_policy_add_remove_pod","message":"Add test_network_policy_add_remove_pod\n\n- Create a service and check connectivity to the service\n- Create a network policy with a podSelector with labels of the pods in the\n  service as well as for the tester pod and check connectivity to the service\n- Deleting one of the service pods and creating a new one\n- Check connectivity to the service and verify only one pod answering\n- Create a new pod with the same label as the one we deleted\n- Check connectivity to the service, expecting two pods answering\n\nAlso adds an option to create_network_policy to apply a rule to allow\ntraffic only from pods with specific labels\n\nChange-Id: Ibf8113d9667c02a1b6d080aa50e91fde0809fd75\nDepends-On: If23b311ed07578b3fbe85f46aa4a314e6a05b7f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/9607acf30952664b84e4f672c78ebd9aaa6845c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/9607acf30952664b84e4f672c78ebd9aaa6845c4"}]},"branch":"refs/heads/master"},"2b2328142dbf2cf32c2a511d389eb85c9b265612":{"kind":"REWORK","_number":13,"created":"2020-10-14 16:30:30.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/56/754956/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/56/754956/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/13"}}},"commit":{"parents":[{"commit":"2fe1791ff2e6283c7f8d92e10fcaffa104457a95","subject":"Remove skip of test_service_udp_ping","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2fe1791ff2e6283c7f8d92e10fcaffa104457a95"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-09-29 09:48:08.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-10-14 16:30:19.000000000","tz":-240},"subject":"Add test_network_policy_add_remove_pod","message":"Add test_network_policy_add_remove_pod\n\n- Create a service and check connectivity to the service\n- Create a network policy with a podSelector with labels of the pods in the\n  service as well as for the tester pod and check connectivity to the service\n- Deleting one of the service pods and creating a new one\n- Check connectivity to the service and verify only one pod answering\n- Create a new pod with the same label as the one we deleted\n- Check connectivity to the service, expecting two pods answering\n\nAlso adds an option to create_network_policy to apply a rule to allow\ntraffic only from pods with specific labels\n\nChange-Id: Ibf8113d9667c02a1b6d080aa50e91fde0809fd75\nDepends-On: If23b311ed07578b3fbe85f46aa4a314e6a05b7f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2b2328142dbf2cf32c2a511d389eb85c9b265612"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2b2328142dbf2cf32c2a511d389eb85c9b265612"}]},"branch":"refs/heads/master"},"e50e6afe9773709424970826cfc54c601d40bd55":{"kind":"REWORK","_number":14,"created":"2020-10-14 17:54:42.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/56/754956/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/56/754956/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/14"}}},"commit":{"parents":[{"commit":"2fe1791ff2e6283c7f8d92e10fcaffa104457a95","subject":"Remove skip of test_service_udp_ping","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2fe1791ff2e6283c7f8d92e10fcaffa104457a95"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-09-29 09:48:08.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-10-14 17:54:32.000000000","tz":-240},"subject":"Add test_network_policy_add_remove_pod","message":"Add test_network_policy_add_remove_pod\n\n- Create a service and check connectivity to the service\n- Create a network policy with a podSelector with labels of the pods in the\n  service as well as for the tester pod and check connectivity to the service\n- Deleting one of the service pods and creating a new one\n- Check connectivity to the service and verify only one pod answering\n- Create a new pod with the same label as the one we deleted\n- Check connectivity to the service, expecting two pods answering\n\nAlso adds an option to create_network_policy to apply a rule to allow\ntraffic only from pods with specific labels\n\nChange-Id: Ibf8113d9667c02a1b6d080aa50e91fde0809fd75\nDepends-On: If23b311ed07578b3fbe85f46aa4a314e6a05b7f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/e50e6afe9773709424970826cfc54c601d40bd55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/e50e6afe9773709424970826cfc54c601d40bd55"}]},"branch":"refs/heads/master"},"8d6e6c9f81575b3410cbbc1c4da3a6b5f10b18d7":{"kind":"REWORK","_number":15,"created":"2020-10-15 09:05:11.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/56/754956/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/56/754956/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/15"}}},"commit":{"parents":[{"commit":"2fe1791ff2e6283c7f8d92e10fcaffa104457a95","subject":"Remove skip of test_service_udp_ping","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2fe1791ff2e6283c7f8d92e10fcaffa104457a95"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-09-29 09:48:08.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-10-15 09:04:59.000000000","tz":-240},"subject":"Add test_network_policy_add_remove_pod","message":"Add test_network_policy_add_remove_pod\n\n- Create a service and check connectivity to the service\n- Create a network policy with a podSelector with labels of the pods in the\n  service as well as for the tester pod and check connectivity to the service\n- Deleting one of the service pods and creating a new one\n- Check connectivity to the service and verify only one pod answering\n- Create a new pod with the same label as the one we deleted\n- Check connectivity to the service, expecting two pods answering\n\nAlso adds an option to create_network_policy to apply a rule to allow\ntraffic only from pods with specific labels\n\nChange-Id: Ibf8113d9667c02a1b6d080aa50e91fde0809fd75\nDepends-On: If23b311ed07578b3fbe85f46aa4a314e6a05b7f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8d6e6c9f81575b3410cbbc1c4da3a6b5f10b18d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8d6e6c9f81575b3410cbbc1c4da3a6b5f10b18d7"}]},"branch":"refs/heads/master"},"edfeab32fd591d8d7c9bb17aa7f901b218ba0934":{"kind":"REWORK","_number":16,"created":"2020-10-15 13:09:08.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/56/754956/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/56/754956/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/16"}}},"commit":{"parents":[{"commit":"2fe1791ff2e6283c7f8d92e10fcaffa104457a95","subject":"Remove skip of test_service_udp_ping","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2fe1791ff2e6283c7f8d92e10fcaffa104457a95"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-09-29 09:48:08.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-10-15 13:08:56.000000000","tz":-240},"subject":"Add test_network_policy_add_remove_pod","message":"Add test_network_policy_add_remove_pod\n\n- Create a service and check connectivity to the service\n- Create a network policy with a podSelector with labels of the pods in the\n  service as well as for the tester pod and check connectivity to the service\n- Deleting one of the service pods and creating a new one\n- Check connectivity to the service and verify only one pod answering\n- Create a new pod with the same label as the one we deleted\n- Check connectivity to the service, expecting two pods answering\n\nAlso adds an option to create_network_policy to apply a rule to allow\ntraffic only from pods with specific labels\n\nChange-Id: Ibf8113d9667c02a1b6d080aa50e91fde0809fd75\nDepends-On: If23b311ed07578b3fbe85f46aa4a314e6a05b7f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/edfeab32fd591d8d7c9bb17aa7f901b218ba0934"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/edfeab32fd591d8d7c9bb17aa7f901b218ba0934"}]},"branch":"refs/heads/master"},"c135330b973e5c2302e6fc6429efd696a2c53b6e":{"kind":"REWORK","_number":17,"created":"2020-10-15 14:35:50.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/56/754956/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/56/754956/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/17"}}},"commit":{"parents":[{"commit":"2fe1791ff2e6283c7f8d92e10fcaffa104457a95","subject":"Remove skip of test_service_udp_ping","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2fe1791ff2e6283c7f8d92e10fcaffa104457a95"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-09-29 09:48:08.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-10-15 14:33:07.000000000","tz":180},"subject":"Add test_network_policy_add_remove_pod","message":"Add test_network_policy_add_remove_pod\n\n- Create a service and check connectivity to the service\n- Create a network policy with a podSelector with labels of the pods in the\n  service as well as for the tester pod and check connectivity to the service, add rules to allow ingress/egress traffic only to those pods with the lables mentioned\n- Check that the expected SG rules exist\n- Deleting one of the service pods and creating a new one\n- Check connectivity to the service and verify only one pod answering\n- Create a new pod with the same label as the one we deleted\n- Check that only the expecte SG rules for the existing pods exist and\n  not including the deleted pod\n- Check connectivity to the service, expecting two pods answering\n\nAlso adds an option to create_network_policy to apply a rule to allow\ntraffic only from pods with specific labels\n\nChange-Id: Ibf8113d9667c02a1b6d080aa50e91fde0809fd75\nDepends-On: If23b311ed07578b3fbe85f46aa4a314e6a05b7f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c135330b973e5c2302e6fc6429efd696a2c53b6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c135330b973e5c2302e6fc6429efd696a2c53b6e"}]},"branch":"refs/heads/master"},"156ff86beda75198553ff241d577ecad42bd0be8":{"kind":"REWORK","_number":18,"created":"2020-10-21 12:03:54.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/56/754956/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/56/754956/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/18"}}},"commit":{"parents":[{"commit":"2fe1791ff2e6283c7f8d92e10fcaffa104457a95","subject":"Remove skip of test_service_udp_ping","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2fe1791ff2e6283c7f8d92e10fcaffa104457a95"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-09-29 09:48:08.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-10-21 11:57:21.000000000","tz":-240},"subject":"Add test_network_policy_add_remove_pod","message":"Add test_network_policy_add_remove_pod\n\n- Create a service and check connectivity to the service\n- Create a network policy with a podSelector with labels of the pods in the\n  service as well as for the tester pod and check connectivity to the service, add rules to allow ingress/egress traffic only to those pods with the lables mentioned\n- Check that the expected SG rules exist\n- Deleting one of the service pods and creating a new one\n- Check connectivity to the service and verify only one pod answering\n- Create a new pod with the same label as the one we deleted\n- Check that only the expecte SG rules for the existing pods exist and\n  not including the deleted pod\n- Check connectivity to the service, expecting two pods answering\n- Also check that we don\u0027t have connectivity from a pod that not\nincluded in the NP\n\nAlso adds an option to create_network_policy to apply a rule to allow\ntraffic only from pods with specific labels\n\nChange-Id: Ibf8113d9667c02a1b6d080aa50e91fde0809fd75\nDepends-On: If23b311ed07578b3fbe85f46aa4a314e6a05b7f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/156ff86beda75198553ff241d577ecad42bd0be8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/156ff86beda75198553ff241d577ecad42bd0be8"}]},"branch":"refs/heads/master"},"b93bb77956ce80ca4df6ed57484166cff2fa75e9":{"kind":"REWORK","_number":19,"created":"2020-10-22 09:43:54.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/56/754956/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/56/754956/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/19"}}},"commit":{"parents":[{"commit":"d8644b598ae2bd506a8e37f205dd51b9183fe608","subject":"Bump py37 to py38 in tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d8644b598ae2bd506a8e37f205dd51b9183fe608"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-09-29 09:48:08.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-10-22 09:43:18.000000000","tz":-240},"subject":"Add test_network_policy_add_remove_pod","message":"Add test_network_policy_add_remove_pod\n\n- Create a service and check connectivity to the service\n- Create a network policy with a podSelector with labels of the pods in the\n  service as well as for the tester pod and check connectivity to the service\n- Deleting one of the service pods and creating a new one\n- Check connectivity to the service and verify only one pod answering\n- Create a new pod with the same label as the one we deleted\n- Check connectivity to the service, expecting two pods answering\n\nAlso adds an option to create_network_policy to apply a rule to allow\ntraffic only from pods with specific labels\n\nChange-Id: Ibf8113d9667c02a1b6d080aa50e91fde0809fd75\nDepends-On: If23b311ed07578b3fbe85f46aa4a314e6a05b7f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/b93bb77956ce80ca4df6ed57484166cff2fa75e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/b93bb77956ce80ca4df6ed57484166cff2fa75e9"}]},"branch":"refs/heads/master"},"d4f0b55b24f2ac695a003ae17e2957f7137e70ae":{"kind":"REWORK","_number":20,"created":"2020-10-27 12:58:45.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/56/754956/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/56/754956/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/20"}}},"commit":{"parents":[{"commit":"d8644b598ae2bd506a8e37f205dd51b9183fe608","subject":"Bump py37 to py38 in tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d8644b598ae2bd506a8e37f205dd51b9183fe608"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-09-29 09:48:08.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-10-27 12:58:08.000000000","tz":-240},"subject":"Add test_network_policy_add_remove_pod","message":"Add test_network_policy_add_remove_pod\n\n- Create a service and check connectivity to the service\n- Create network policies with a podSelector with labels of the pods in the\n  service as well as for the tester pod and check connectivity to the service\n- Deleting one of the service pods and creating a new one\n- Check connectivity to the service and verify only one pod answering\n- Create a new pod with the same label as the one we deleted\n- Check connectivity to the service, expecting two pods answering\n\nAlso adds an option to create_network_policy to apply a rule to allow\ntraffic only from pods with specific labels\n\nChange-Id: Ibf8113d9667c02a1b6d080aa50e91fde0809fd75\nDepends-On: If23b311ed07578b3fbe85f46aa4a314e6a05b7f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d4f0b55b24f2ac695a003ae17e2957f7137e70ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d4f0b55b24f2ac695a003ae17e2957f7137e70ae"}]},"branch":"refs/heads/master"},"f5b2118320261c113fc37ba1771e369cb63b3607":{"kind":"REWORK","_number":21,"created":"2020-10-28 05:27:30.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/56/754956/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/56/754956/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/21"}}},"commit":{"parents":[{"commit":"d07fefba073fa63363e04744861f5432a22e6afe","subject":"Merge \"Remove releasenotes test from envlist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d07fefba073fa63363e04744861f5432a22e6afe"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-09-29 09:48:08.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-10-28 05:27:20.000000000","tz":-240},"subject":"Add test_network_policy_add_remove_pod","message":"Add test_network_policy_add_remove_pod\n\n- Create a service and check connectivity to the service\n- Create network policies with a podSelector with labels of the pods in the\n  service as well as for the tester pod and check connectivity to the service\n- Deleting one of the service pods and creating a new one\n- Check connectivity to the service and verify only one pod answering\n- Create a new pod with the same label as the one we deleted\n- Check connectivity to the service, expecting two pods answering\n\nAlso adds an option to create_network_policy to apply a rule to allow\ntraffic only from pods with specific labels\n\nChange-Id: Ibf8113d9667c02a1b6d080aa50e91fde0809fd75\nDepends-On: If23b311ed07578b3fbe85f46aa4a314e6a05b7f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/f5b2118320261c113fc37ba1771e369cb63b3607"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/f5b2118320261c113fc37ba1771e369cb63b3607"}]},"branch":"refs/heads/master"},"cd37fdffe3ca6a3ffadf135ffda4e51326351e5c":{"kind":"REWORK","_number":22,"created":"2020-10-29 07:55:49.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/56/754956/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/56/754956/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/22"}}},"commit":{"parents":[{"commit":"d07fefba073fa63363e04744861f5432a22e6afe","subject":"Merge \"Remove releasenotes test from envlist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d07fefba073fa63363e04744861f5432a22e6afe"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-09-29 09:48:08.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-10-29 07:55:39.000000000","tz":-240},"subject":"Add test_network_policy_add_remove_pod","message":"Add test_network_policy_add_remove_pod\n\n- Create a service and check connectivity to the service\n- Create network policies with a podSelector with labels of the pods in the\n  service as well as for the tester pod and check connectivity to the service\n- Deleting one of the service pods and creating a new one\n- Check connectivity to the service and verify only one pod answering\n- Create a new pod with the same label as the one we deleted\n- Check connectivity to the service, expecting two pods answering\n\nAlso adds an option to create_network_policy to apply a rule to allow\ntraffic only from pods with specific labels\n\nChange-Id: Ibf8113d9667c02a1b6d080aa50e91fde0809fd75\nDepends-On: If23b311ed07578b3fbe85f46aa4a314e6a05b7f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/cd37fdffe3ca6a3ffadf135ffda4e51326351e5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/cd37fdffe3ca6a3ffadf135ffda4e51326351e5c"}]},"branch":"refs/heads/master"},"45cdaf15530f7d3b9c872de43367638efe43a11b":{"kind":"REWORK","_number":23,"created":"2020-11-02 12:34:21.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/56/754956/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/56/754956/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/23"}}},"commit":{"parents":[{"commit":"d07fefba073fa63363e04744861f5432a22e6afe","subject":"Merge \"Remove releasenotes test from envlist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d07fefba073fa63363e04744861f5432a22e6afe"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-09-29 09:48:08.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-11-02 12:34:10.000000000","tz":-300},"subject":"Add test_network_policy_add_remove_pod","message":"Add test_network_policy_add_remove_pod\n\n- Create a service and check connectivity to the service\n- Create network policies with a podSelector with labels of the pods in the\n  service as well as for the tester pod and check connectivity to the service\n- Deleting one of the service pods and creating a new one\n- Check connectivity to the service and verify only one pod answering\n- Create a new pod with the same label as the one we deleted\n- Check connectivity to the service, expecting two pods answering\n\nAlso adds an option to create_network_policy to apply a rule to allow\ntraffic only from pods with specific labels\n\nChange-Id: Ibf8113d9667c02a1b6d080aa50e91fde0809fd75\nDepends-On: If23b311ed07578b3fbe85f46aa4a314e6a05b7f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/45cdaf15530f7d3b9c872de43367638efe43a11b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/45cdaf15530f7d3b9c872de43367638efe43a11b"}]},"branch":"refs/heads/master"},"28fbe115ebd97760b128d4c9efd21d7d426922d3":{"kind":"REWORK","_number":24,"created":"2020-11-02 13:50:33.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/56/754956/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/56/754956/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/56/754956/24"}}},"commit":{"parents":[{"commit":"d07fefba073fa63363e04744861f5432a22e6afe","subject":"Merge \"Remove releasenotes test from envlist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d07fefba073fa63363e04744861f5432a22e6afe"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-09-29 09:48:08.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2020-11-02 13:50:24.000000000","tz":-300},"subject":"Add test_network_policy_add_remove_pod","message":"Add test_network_policy_add_remove_pod\n\n- Create a service and check connectivity to the service\n- Create network policies with a podSelector with labels of the pods in the\n  service as well as for the tester pod and check connectivity to the service\n- Deleting one of the service pods and creating a new one\n- Check connectivity to the service and verify only one pod answering\n- Create a new pod with the same label as the one we deleted\n- Check connectivity to the service, expecting two pods answering\n\nAlso adds an option to create_network_policy to apply a rule to allow\ntraffic only from pods with specific labels\n\nChange-Id: Ibf8113d9667c02a1b6d080aa50e91fde0809fd75\nDepends-On: If23b311ed07578b3fbe85f46aa4a314e6a05b7f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/28fbe115ebd97760b128d4c9efd21d7d426922d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/28fbe115ebd97760b128d4c9efd21d7d426922d3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
