)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~755819","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~I26a22962da008b272ffcd72622f32924df3f32fe","project":"openstack/kuryr-tempest-plugin","branch":"master","hashtags":[],"change_id":"I26a22962da008b272ffcd72622f32924df3f32fe","subject":"Fix tox command failed due to .stestr.conf missing","status":"NEW","created":"2020-10-02 15:03:29.000000000","updated":"2020-10-09 13:00:56.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"20d994d90c20561bdd90b32bf5ba6db93046ebb2","_number":755819,"virtual_id_number":755819,"owner":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-06 13:18:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":-1,"date":"2020-10-09 13:00:56.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-06 13:18:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-09 13:00:56.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"}],"messages":[{"id":"54a0d2c1ec2c8f3b6d993ba910ae8edb0314c57f","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-10-02 15:03:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a6bdc7422e22dc60a95137e062e5e6eb31166e7","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-10-02 15:18:43.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"df3e3039bfead98e141aca99ae29a617efed084f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 17:29:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f1ce10948b44083b750b59c4d160ab3 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee295cfb630241279f9d6c68c9897863 : SUCCESS in 3m 46s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/18ae47da7cfb4e36aff909a2c65d6ec0 : FAILURE in 55m 41s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/95c60ac24da94c6585843dac53bf64a2 : SUCCESS in 1h 20m 38s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/3099d030fa1a4c1094d8d76f0ebb8abb : SUCCESS in 1h 14m 13s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/b8dd0721686c40edab446e6c2262d40a : FAILURE in 1h 18m 24s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/669e9ea65f03464d983adb633284cb89 : SUCCESS in 1h 18m 45s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/a51155a26573469c865b2896ec92582e : SUCCESS in 1h 22m 05s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/1a639b40decb4d57b946161c7aed04f2 : SUCCESS in 1h 03m 31s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/f043b7d254ce42878a7df71acd2a1472 : SUCCESS in 1h 07m 18s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/663094b96848446986274f343739891c : SUCCESS in 1h 07m 15s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/061091077f0e48f49882a86aaee4626f : POST_FAILURE in 2h 04m 16s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/8e3f03c6ab2b4c84afc34c24a41eb2ff : SUCCESS in 1h 22m 48s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/92c24ea7d6a44bbea228e3702af60a30 : FAILURE in 1h 43m 15s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/1d48abe0a0d1435d9ba1da6ac7bb37bb : FAILURE in 1h 44m 54s","accounts_in_message":[],"_revision_number":2},{"id":"77a872a3dbbe2eae939300b977db519087ea0f30","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-10-06 11:25:22.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"8c246ae71bd459c8b1473af2019c45d3376f1f92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 13:18:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13b201a9b4aa4780966c3c9981a1d280 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2e3c08f0b324e50ac73275ce096ac3b : SUCCESS in 3m 44s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/663ae566ec4d4a908125e6e4fa8ab2ae : SUCCESS in 1h 08m 51s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/766445979f5e4cf7bb269f155cc2db45 : SUCCESS in 1h 08m 17s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/726fad27ee3545608453a46c73724206 : FAILURE in 1h 09m 18s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/dab83707e7884be58100aab111c44394 : FAILURE in 31m 36s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/277586b6e9304d029e160c9e4bac78aa : SUCCESS in 1h 12m 39s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/ac34a2ddf3f74d3788d9f5ef273f4b46 : SUCCESS in 1h 42m 55s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/865e32aeb175495e838dc0f527774267 : SUCCESS in 1h 04m 19s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/df8d483668cd4445ab65d94d26690f14 : FAILURE in 1h 15m 36s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/0f8e398ce4ee4ec8a121e776a5c19bf6 : FAILURE in 1h 16m 31s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/51fd24484ced4dd8855dc0caddf09dc9 : POST_FAILURE in 46m 45s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/4bf7c52d9b114f4aae41d2c69f0984c8 : SUCCESS in 1h 04m 48s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/79c91dfe5195471283d9ec5579a26c48 : SUCCESS in 1h 15m 25s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/624ba99d6dd14d4d82c81a2e5e048d5f : SUCCESS in 1h 20m 38s","accounts_in_message":[],"_revision_number":3},{"id":"20d994d90c20561bdd90b32bf5ba6db93046ebb2","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-10-09 13:00:56.000000000","message":"Patch Set 3: Code-Review-1\n\nInstead of this I\u0027d rather remove py38 and releasenotes envs from the main list. E.g. cinder-tempest-plugin seems to have tox.ini that makes sense: https://github.com/openstack/cinder-tempest-plugin/blob/master/tox.ini.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"727032206bf9c1ebaa1e3e37622b62c2c6bf02df","revisions":{"e831f4f526d789a29dc8496f50477d3fc787db7c":{"kind":"REWORK","_number":1,"created":"2020-10-02 15:03:29.000000000","uploader":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"ref":"refs/changes/19/755819/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/19/755819/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/19/755819/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/19/755819/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/19/755819/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/19/755819/1"}}},"commit":{"parents":[{"commit":"1020b0f9065cccc02165a5fa053662380c88ba29","subject":"Merge \"Fail get_controller_pod_names when no pods found\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/1020b0f9065cccc02165a5fa053662380c88ba29"}]}],"author":{"name":"zhoulinhui","email":"df.some@foxmail.com","date":"2020-10-02 14:56:55.000000000","tz":480},"committer":{"name":"zhoulinhui","email":"df.some@foxmail.com","date":"2020-10-02 15:02:47.000000000","tz":480},"subject":"Fix tox command failed due to .stestr.conf missing","message":"Fix tox command failed due to .stestr.conf missing\n\nWhen run tox -e py37 on localhost, it will failed\nbecause of the .stestr.conf missing, it raise error message:\n\"No .stestr.conf found, please create one.\"\nThis PS to fixed it.\n\nChange-Id: I26a22962da008b272ffcd72622f32924df3f32fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/e831f4f526d789a29dc8496f50477d3fc787db7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/e831f4f526d789a29dc8496f50477d3fc787db7c"}]},"branch":"refs/heads/master"},"97a6a9bd7b0f5b0526a291a7d4cb755a8c178400":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-10-02 15:18:43.000000000","uploader":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"ref":"refs/changes/19/755819/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/19/755819/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/19/755819/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/19/755819/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/19/755819/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/19/755819/2"}}},"commit":{"parents":[{"commit":"1020b0f9065cccc02165a5fa053662380c88ba29","subject":"Merge \"Fail get_controller_pod_names when no pods found\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/1020b0f9065cccc02165a5fa053662380c88ba29"}]}],"author":{"name":"zhoulinhui","email":"df.some@foxmail.com","date":"2020-10-02 14:56:55.000000000","tz":480},"committer":{"name":"zhoulinhui","email":"df.some@foxmail.com","date":"2020-10-02 15:18:20.000000000","tz":480},"subject":"Fix tox command failed due to .stestr.conf missing","message":"Fix tox command failed due to .stestr.conf missing\n\nWhen run tox command on localhost, it will failed\nbecause of the .stestr.conf missing, it raise error message:\n\"No .stestr.conf found, please create one.\"\n\nChange-Id: I26a22962da008b272ffcd72622f32924df3f32fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/97a6a9bd7b0f5b0526a291a7d4cb755a8c178400"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/97a6a9bd7b0f5b0526a291a7d4cb755a8c178400"}]},"branch":"refs/heads/master"},"727032206bf9c1ebaa1e3e37622b62c2c6bf02df":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-10-06 11:25:22.000000000","uploader":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"ref":"refs/changes/19/755819/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/19/755819/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/19/755819/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/19/755819/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/19/755819/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/19/755819/3"}}},"commit":{"parents":[{"commit":"d8644b598ae2bd506a8e37f205dd51b9183fe608","subject":"Bump py37 to py38 in tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d8644b598ae2bd506a8e37f205dd51b9183fe608"}]}],"author":{"name":"zhoulinhui","email":"df.some@foxmail.com","date":"2020-10-02 14:56:55.000000000","tz":480},"committer":{"name":"zhoulinhui","email":"df.some@foxmail.com","date":"2020-10-06 11:25:22.000000000","tz":0},"subject":"Fix tox command failed due to .stestr.conf missing","message":"Fix tox command failed due to .stestr.conf missing\n\nWhen run tox command on localhost, it will failed\nbecause of the .stestr.conf missing, it raise error message:\n\"No .stestr.conf found, please create one.\"\n\nChange-Id: I26a22962da008b272ffcd72622f32924df3f32fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/727032206bf9c1ebaa1e3e37622b62c2c6bf02df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/727032206bf9c1ebaa1e3e37622b62c2c6bf02df"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
