)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~765306","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~I3ad7cc10a307cb479dd335704a72fcef110e8bc3","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"remove-useless","hashtags":[],"change_id":"I3ad7cc10a307cb479dd335704a72fcef110e8bc3","subject":"Remove uselsess files and tests","status":"MERGED","created":"2020-12-03 09:42:17.000000000","updated":"2021-01-15 13:08:40.000000000","submitted":"2021-01-15 13:06:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"765306-remove-useless","meta_rev_id":"ee98d54c9cbf65234baf1fb4ee192d3768218688","_number":765306,"virtual_id_number":765306,"owner":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-15 13:06:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"recommended":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"all":[{"value":1,"date":"2020-12-03 10:45:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":2,"date":"2020-12-11 13:42:14.000000000","_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-04 08:00:30.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"all":[{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":1,"date":"2020-12-11 13:42:14.000000000","_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-03 10:45:34.000000000","updated_by":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"reviewer":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"state":"REVIEWER"},{"updated":"2020-12-03 11:27:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-04 08:00:30.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2020-12-11 13:42:14.000000000","updated_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"}],"messages":[{"id":"49d4053a8ce06f03a80669ba9722466854c3f4db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-12-03 09:42:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c0bf422540c1bd9baa7840f485117ce266c0035","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-12-03 10:45:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ed818fab06f94753d4e22ecd5ac40400031a86ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 11:27:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7d87384dcc9471eac2289ac8343b2f9 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd5c8847e9644b66b300ec152eeb4103 : SUCCESS in 3m 14s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/74ef9826948d4010bf5419bd68a7b287 : SUCCESS in 1h 12m 46s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/c65ca186868b456bac5311a023b6ef92 : SUCCESS in 1h 26m 58s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/4934be684da1489899ad48bd882daab5 : FAILURE in 50m 02s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/61101c8d4e2a4ea88ac4827932667c5d : FAILURE in 27m 17s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/1897065dfef34a9fa9356d8b3c8dd531 : FAILURE in 26m 30s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/cce5015a2396462eb5aee4c8b1dc227f : SUCCESS in 1h 34m 42s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/456805c77044404ebc2dec1e529c9da4 : SUCCESS in 1h 31m 50s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/2ae42859aa0744a59a57da4a68a5c808 : SUCCESS in 1h 05m 23s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/09a60961030643be995a75771d0d509c : FAILURE in 31m 00s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/8fa885d1e4c8439b899d1b5887ea8361 : POST_FAILURE in 1h 11m 11s (non-voting)\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/de7458dc4d4e48a7a3170dae28cb1ec1 : FAILURE in 32m 01s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/52f67a74d3ae485b9c8fbbf95710af20 : SUCCESS in 1h 04m 56s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/b94a81ea1d184509ac7f9ce91095d5f8 : SUCCESS in 1h 07m 10s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/81185140e16444209767145799c346d4 : SUCCESS in 1h 09m 46s","accounts_in_message":[],"_revision_number":1},{"id":"8a4feeafc4802f68fad546bda1f39c908be537c0","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-12-03 12:06:37.000000000","message":"Patch Set 1:\n\nDocker problems, unrelated.","accounts_in_message":[],"_revision_number":1},{"id":"394bad5671f8a04d867647c35f06a180e0ab85a1","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-12-04 08:00:30.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7df51d3ca234d998cb29dd1daea9307a18f94198","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 10:16:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96c8136291384b16b48d7ea95396f32d : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf90850deec64117b4c78706b8cca269 : SUCCESS in 3m 01s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/9fd209131f6e47a9a9113601bb54a52f : SUCCESS in 1h 11m 21s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/7897a008421b4e01bf290d052537957d : SUCCESS in 1h 27m 19s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/cab700fca7b9404c93488096319f7e9e : FAILURE in 1h 10m 00s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/5d3d5d9cd78b4e63b4f790e7b4186c4c : SUCCESS in 1h 26m 31s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/f2e783bf37844370b3895051c3ab03d3 : SUCCESS in 1h 23m 39s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/37616af0b71b45dc8bb9c7b714e49b86 : SUCCESS in 1h 26m 51s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/83bf9e1be30945cd9176d8905ee7645e : SUCCESS in 1h 05m 42s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/39d3ff0ca266471bbc38e28277c16c0e : SUCCESS in 1h 12m 40s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/77e6538ec6804ae7861ab1c55036a541 : FAILURE in 1h 23m 23s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/ebf015adf36c48f78ca6ac7b3ce822b3 : POST_FAILURE in 2h 11m 51s (non-voting)\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/4294fbfb1e084d8fb62f0b4547737aaf : SUCCESS in 1h 03m 13s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/0f5fa7c201d74960b70d15725933f56a : SUCCESS in 1h 22m 41s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/f5ffa80924a5434bbb67ac8a2aacf556 : SUCCESS in 1h 31m 07s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/cb8030af7f8d44b987b0853b56f28ddd : SUCCESS in 1h 38m 07s","accounts_in_message":[],"_revision_number":1},{"id":"21928bc4694b8d0ed8dda327d0eb50c01c17c92f","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2020-12-11 13:42:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"06d6f529a94e140a3e73ab7644dc8ebd23d9c77d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 13:42:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4d48b71e4d90b02e3195d009d97b29ae5123a57a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 15:51:05.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f6d8a5e4a184a5e9a81fde89a102823 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/545843325dc949428743b240cf550eee : SUCCESS in 3m 58s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/6de1d00f4e244c8481d01e7d2abe5472 : SUCCESS in 1h 04m 32s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/a46065a1d8644b1dbb4d3962b8afcdb2 : SUCCESS in 1h 16m 53s\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/7084e586cfda4c608a84435e2cb69929 : SUCCESS in 1h 43m 17s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/0b42f1a5d755491ba793338ad98b75dd : FAILURE in 2h 02m 08s","accounts_in_message":[],"_revision_number":1},{"id":"bd6d94e9ed210b58641da749b397eb4c0060e718","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-12-11 15:59:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0d61f590899a25ac7ce1cbd03af5fd1a704f32ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 18:08:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93c4d4442c604de980383d8e07eed436 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbf466d527fe4d69a5cf0b8177554c4f : SUCCESS in 3m 14s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/a8159b11831342b5ad9d5db86f4c85be : SUCCESS in 1h 11m 50s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/96e51ba32e4d4774b7803f8946e9712f : SUCCESS in 1h 14m 19s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/8b5177211bc64a3287b6329551454d33 : SUCCESS in 1h 17m 46s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/b70e0d33217e4a55b6e1f28b70f96737 : SUCCESS in 1h 27m 45s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/773e24188765493086ae4248a408d768 : SUCCESS in 1h 39m 39s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/d3335f901e82401ebcb0725ef7613b0e : SUCCESS in 1h 30m 09s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/aad7328e541a40fd83b8436f862b49c3 : SUCCESS in 1h 00m 30s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/26866df457bc4d40bdffa92d2bfb050f : SUCCESS in 1h 10m 33s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/d19f9a79639743b0937c0efc1cff6804 : SUCCESS in 1h 21m 09s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/55a3c53ab9c7494da2d9bc40cd7b6fad : POST_FAILURE in 2h 03m 10s (non-voting)\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/1c8c5709d3564673bb4085ea55879b7b : SUCCESS in 55m 34s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/d0be27d564bf4f5788d1175f29d1f6ca : FAILURE in 55m 18s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/4082048e8be54e4b8345abf7f723a469 : FAILURE in 29m 35s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/3b40b3a60ce940b09635eccafecaaa33 : FAILURE in 27m 53s","accounts_in_message":[],"_revision_number":1},{"id":"2d6f966255d264a4f61607e9a95e4763f26723b3","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-12-13 09:25:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"40d3c38165a69a80a8f228fe9614ad5b406c28b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-13 11:17:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/255641c5d72d4e1b98236af17d8dd1da : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e32e55b46c0d49b68935811dcc51b0a4 : SUCCESS in 3m 45s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/37eb612810f64a1db1197682131b9818 : SUCCESS in 1h 02m 43s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/ed1701d3b8c740928b3a2d4da50e468f : SUCCESS in 1h 05m 30s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/143d708d12be4c8c88d6a700d6a99eae : SUCCESS in 1h 20m 36s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/e2eea28ea65b46a3b381117dddd51e1f : SUCCESS in 1h 16m 25s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/d94b032dd14344729a285708deb86347 : SUCCESS in 1h 38m 28s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/c2d31624fbc24f24b4665582a49a4856 : SUCCESS in 1h 29m 28s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/ae091101a8d246329e8d31dbcdf5ba18 : SUCCESS in 1h 07m 57s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/25d6806496554aaa92186080824701c7 : SUCCESS in 1h 01m 46s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/6838342433154d58aa81d3362a9d86a9 : SUCCESS in 1h 20m 31s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/2d0228bd2fe74e6aa23d5ce30b774006 : POST_FAILURE in 1h 50m 57s (non-voting)\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/9c4dfe35566a48f68907f44f28630685 : SUCCESS in 1h 00m 03s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/a484618345b449e885e4110d23b9f971 : SUCCESS in 1h 15m 11s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/0560c53adce4427fa457f92ea308b833 : FAILURE in 29m 51s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/0e1f37514d9548b9b1b798dc55f9f98e : FAILURE in 26m 08s","accounts_in_message":[],"_revision_number":1},{"id":"2b34348d0d83e26ef231e7512d12aa21ec09d97c","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-12-17 07:59:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e9eefbb9b59ccb46091f799ee580ce8d8ca698c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 10:32:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1784783051e14fecbd6997045b273963 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47c1dfe1e8c449e986d43e175d44e115 : SUCCESS in 3m 57s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/3a238c20bc634e33b740f1acb1d39e18 : SUCCESS in 1h 00m 25s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/beff8b47efbf41e9824983aadef06990 : SUCCESS in 1h 05m 18s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/418b4dc43dcf424091554eba1937ffc1 : SUCCESS in 1h 41m 06s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/5de58f82665144d1ab0017526fb05b4e : SUCCESS in 1h 11m 03s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/d8f10413604748b1861cb735261951ce : SUCCESS in 1h 22m 42s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/f16eed1851a14265884b333f86c5a838 : SUCCESS in 1h 26m 20s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/758333d34f6043dc9c7b17338f05c274 : SUCCESS in 1h 19m 35s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/bc54885b08794a49870346a74fc86234 : SUCCESS in 1h 14m 25s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/9e3de8a4ba224024b45900f863035236 : SUCCESS in 1h 29m 28s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/7aa0aa02c0f74a6b97f7866696c15cf9 : POST_FAILURE in 2h 18m 03s (non-voting)\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/0e0be1b9b09f436796d3ea3bd981763e : SUCCESS in 58m 28s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/ce39d1b1b6924e4a86a2b2a661f22255 : SUCCESS in 1h 16m 43s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/88f0608a01324c51a928ce6c272457d3 : FAILURE in 1h 23m 21s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/22d170d1785248cda8c2bd454d340995 : FAILURE in 39m 02s","accounts_in_message":[],"_revision_number":1},{"id":"4873001274bbead2f042a1bd81b67a1066ec0ec7","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-12-22 07:32:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1c3c8326e393205f3df0dd858ac68c4851e72589","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 10:39:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a62e317dfe14200add8cc1c96993eb4 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cae441ce8a784b2f99c4864f040bba9e : SUCCESS in 3m 34s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/266a28dddf244177871fdd1a709b33b1 : SUCCESS in 1h 12m 28s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/8076b98ad3c04943b8d8279aaa3f2ff0 : TIMED_OUT in 3h 05m 09s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/be6d323d880e45328f514454d8c118af : SUCCESS in 1h 12m 27s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/4a7442c46fd24849a2d0b7238265fd40 : FAILURE in 1h 57m 17s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/b407e495ddab4d969116b5a50bc4f4f4 : SUCCESS in 1h 30m 40s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/0b8a7deba3534047b7078dfeb2c55c3a : TIMED_OUT in 3h 05m 10s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/83a1bea3d57b441c9bae1ad320602403 : SUCCESS in 1h 06m 58s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/bf92865c16fb44f093f837bc59986478 : SUCCESS in 1h 03m 48s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/33b25b58476341acbf1f8d8d612f4e97 : SUCCESS in 1h 09m 48s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/b5f7bac3075b4f70a00dfa3300386dc5 : POST_FAILURE in 1h 53m 03s (non-voting)\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/6186fa9e58914ea295d17ac2f2e57d43 : SUCCESS in 1h 03m 12s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/5fc29b70345645c3aff7fdfd1fc40f4a : SUCCESS in 1h 00m 23s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/a3d0a7ea577d49c1947302cf61c2064c : FAILURE in 32m 51s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/9b6a8ea5375b4850b02c941e9af1846e : FAILURE in 26m 49s","accounts_in_message":[],"_revision_number":1},{"id":"e520b51d8e5c683ddc3acfad55a907b9f63a55ed","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2021-01-07 08:03:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"806082f24ed9d15a037f3e68c06ea2cf8687b09a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 11:21:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51da6fdd2f414b829e04bff0a36bb85d : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/764bea4f51f744819f2d03878ee8f8c6 : SUCCESS in 3m 14s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/9123a7edae434c94960d64efadfb3667 : SUCCESS in 1h 19m 22s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/6d9889dbaa4647e388a30a1e32fceb28 : SUCCESS in 1h 34m 30s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/a3689dd1ad4c4249a825c49ce706f662 : SUCCESS in 1h 16m 59s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/fa38d2321cca4f11a019a22d162e3b18 : FAILURE in 2h 21m 15s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/167bd35dac5b4a9db24cb4b0dd07c920 : SUCCESS in 1h 33m 51s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/d53c55fc86c5432597d6a0849d4ef630 : SUCCESS in 1h 30m 07s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/481d879c128d468cbba958892d6a3cd1 : SUCCESS in 2h 08m 53s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/608ede5a8fb54100816ae62b2e49be10 : SUCCESS in 1h 14m 23s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/d7cec3cc97d445b4969fcce37f244001 : FAILURE in 2h 19m 20s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/3a52ad9aabc449e2ad5823b5db5d0b93 : POST_FAILURE in 3h 16m 02s (non-voting)\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/5dc3968e39954f369904479bd36c5ae3 : SUCCESS in 1h 06m 22s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/8867f11e48bb4485b014d14415b53dc9 : SUCCESS in 1h 12m 31s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/7b27e998240947ebb8a43a16392eebe0 : FAILURE in 26m 23s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/85acd5728d6c411d9275279b7c52d49b : FAILURE in 32m 00s","accounts_in_message":[],"_revision_number":1},{"id":"12c0cb553863af81f4ec70a0c2a5e6a9d62a881b","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-01-14 12:23:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bc9d1197bc07b3303ef465fb1f90351984cbeae1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 14:30:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4763747904464b52aca6b73fb07faf22 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d48e67b6d52e49e0a92a830bdb66f609 : SUCCESS in 4m 26s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/61180db0b43b44dd8b49f6a275d1a6b0 : SUCCESS in 1h 13m 52s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/10684067b0b74e6b8a56a820e45afd22 : SUCCESS in 1h 21m 21s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/5a230bab61c4498eb4a3db5f887d36f4 : SUCCESS in 1h 15m 03s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/24f8620d34234e0ea166c527140002c8 : SUCCESS in 1h 17m 52s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/d04d5d2a6fbd4a6686dd7ea7ba80f7b0 : SUCCESS in 1h 53m 51s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/0930802d55a24929b0e9ba054e98e181 : SUCCESS in 1h 39m 46s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/2eb15183482846f1860884e6f71a25c5 : SUCCESS in 1h 07m 40s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/daa92934bc654fa7b189da4183464e97 : SUCCESS in 1h 39m 23s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/6955ec25ad0e40b7954ae0402551d0df : SUCCESS in 1h 17m 31s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/b29fe31cfd114a188ff500d04d8b3dc3 : POST_FAILURE in 2h 00m 25s (non-voting)\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/d8a3fc40229a49078e6d3e6004fdfbbf : FAILURE in 1h 05m 26s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/4868c707295a4091990803b261fee528 : SUCCESS in 1h 04m 25s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/9f529866d7304a72aae8e73776ac2313 : SUCCESS in 1h 32m 19s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/e31ca6931dbd4ea98bbcf0615454ecf7 : FAILURE in 42m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3b1672a560df278c0b6e726dec9e4365321f89ef","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-01-14 14:56:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"032db613a5bdec3d86861487c2e966ab1b223698","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 17:18:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/942ced559d3a40d594fdbd1a2cc60b73 : SUCCESS in 9m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a70a93f56604d1c9d5bf39105a9be2f : SUCCESS in 8m 58s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/b86c8fd2656443189c950e1e0e4ae69c : SUCCESS in 1h 50m 00s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/e42f616e831f4aa4ac73871039db5026 : SUCCESS in 1h 46m 29s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/61af893b64ec4b6e9550a92657d9df06 : SUCCESS in 1h 08m 41s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/635e7fa13a7548a98079ff6e83a033ec : SUCCESS in 1h 20m 56s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/58d8e8cda5124479b7c56942c8d9b90a : SUCCESS in 2h 00m 55s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/f23f4aa00135434ebb69b59e0bd0e5da : SUCCESS in 2h 15m 37s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/18a4729551d64476b6587f14471abcc0 : SUCCESS in 1h 56m 20s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/59a4e95d3e6e443eb7070372315ebeb4 : SUCCESS in 1h 56m 25s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/e8b16cb3555f4c4db540f3a779bb53c6 : SUCCESS in 1h 19m 45s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/4c9d95f5592c4bb5adcbd3a37d6dbec2 : POST_FAILURE in 1h 56m 55s (non-voting)\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/e6d0dedada3a4fe58f7b97cb3e9b4eac : SUCCESS in 55m 59s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/1f9c916470624998ad1140d25f573e87 : SUCCESS in 1h 05m 44s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/bf941b051e2b4c509496e22e494e32b4 : SUCCESS in 1h 05m 35s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/11cd07a823b7495cb224145560edb0f8 : FAILURE in 27m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"df21ea7fb3155f0e87842b6321008227a8c2bb89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 17:18:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6814099cab5ea7eadf1f72835537064c83f359bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 19:18:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91745fdef6714031b62b49a4db7eccc3 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81013e3fcddf4eb8a1e0391c34eb46f3 : SUCCESS in 4m 35s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/860b6d8de2d64e6e9ec0d9e6bd71cbb5 : SUCCESS in 1h 04m 42s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/1e29c6b77b4a4096bb5bd3966fb79aa5 : SUCCESS in 1h 13m 19s\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/80df55a4b53a458f8510b1196eb6382a : SUCCESS in 1h 53m 36s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/c09129892a9d4875b94775b356a45634 : FAILURE in 1h 24m 05s","accounts_in_message":[],"_revision_number":1},{"id":"452b4df47a16a4c849368b8f67e3d247a4a352da","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2021-01-15 05:26:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3b5d1064f4f8fb0d1aca6f04c05ab476cf0a8a5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 08:34:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95dd169554fc47ca8fe06c9c97e06e23 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d1176d58c3b4b50bd344b0645637707 : SUCCESS in 4m 10s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/08fe9117aca2403ebc8e0db69eb9d30c : SUCCESS in 1h 03m 34s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/b88fcc5d88344a04bca951b4a3de2f8d : SUCCESS in 1h 12m 15s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/481c9c1e0b2743199128f0180f276a42 : SUCCESS in 1h 32m 38s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/58b9455fc701493a9dac09ac0b09431c : SUCCESS in 1h 23m 30s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/6b94a44bcf7a46d481795749970e859a : SUCCESS in 1h 26m 58s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/84042df6e9d34b80b5746c7a1a1c00aa : SUCCESS in 1h 22m 18s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/77fbd451672f40e287beb7ca168811c7 : SUCCESS in 1h 05m 58s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/b7d1d383ba4845c99c3640fb8c7e5d8e : SUCCESS in 1h 02m 27s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/afaf4c23eb044f399b6d81bd985c925e : SUCCESS in 1h 21m 34s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/4e60363c9f0540009bd1000129fbbdde : POST_FAILURE in 3h 04m 26s (non-voting)\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/ae18c8d326ca47ffb4c519c3cce50e4a : FAILURE in 1h 37m 48s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/476d653d7e894e779a7f3ad9a72bd6f8 : SUCCESS in 1h 25m 02s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/3d576db4305a4a9c864ca92f6ed23bb7 : SUCCESS in 1h 06m 39s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/ee9bb77479a348b0906d69529575edd4 : FAILURE in 29m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"05cdf4f7e467af9f6d5ad816762d8067e1a0f793","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-01-15 09:01:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"804538e414e00edc9b6de7c5d6e2e00968069315","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 11:17:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38d35bdb013e4c99be3438417bd80e24 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd777f4de9b5452d830f2a2394b83ade : SUCCESS in 3m 44s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/6e87cec75c88476a9a5cc14211803a65 : SUCCESS in 1h 13m 27s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/6bb1ba98eda041caac9b09bcf71a7d6b : SUCCESS in 59m 09s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/df87153637f540d0b31d0dda09a47ad2 : SUCCESS in 1h 13m 56s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/a686ea1be6354b49afcc17d62214d2d5 : SUCCESS in 1h 10m 39s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/663a1903cc4b4d2dac1e4e7f51558c58 : SUCCESS in 1h 38m 10s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/f4f37aee5da64320bab76b89830a6672 : SUCCESS in 1h 36m 54s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/46c43349cc9e473d9ab8aece09f3f3aa : SUCCESS in 1h 08m 40s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/f5c24338d3d24eaa9a36860739029cc8 : SUCCESS in 1h 43m 29s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/56c45857f31245ba9e1aee0976b2ed78 : SUCCESS in 1h 24m 01s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/f209f27ec6924b4e8fe1643eb66f45f3 : POST_FAILURE in 2h 06m 25s (non-voting)\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/a9d529193cb94e7dabaf1cdce0937e35 : SUCCESS in 1h 21m 18s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/387069a6b1aa41b683e9bb52e557dc2f : SUCCESS in 1h 09m 32s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/61fabbd2085040f1a6acd55df19036f3 : SUCCESS in 1h 19m 55s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/e4eb30f87366436aaa43983e5f1a4a29 : FAILURE in 44m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7ab8be77e8c029ae87e9657456d46b6ffc3758f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 11:17:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6a0103c0c5d3b676e74389ee51c59863a4ba5c5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 13:06:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80749febebe84e359f586dac612ad70c : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9734eacd5201444ba28d6ea23646750b : SUCCESS in 3m 27s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/08356a5cf4c84ed3a92dacdcaea4a027 : SUCCESS in 1h 11m 21s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/1e12a2d4fae44b3e87bd003e7e10e4bd : SUCCESS in 1h 15m 46s\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/c6d8699d10e847808cda01ec844c6f98 : SUCCESS in 1h 44m 31s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/8ea77269760c449aaeea4254dc5bc0ef : SUCCESS in 1h 25m 07s","accounts_in_message":[],"_revision_number":1},{"id":"952cc671925e85443da215e8fbc9196671274140","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 13:06:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ee98d54c9cbf65234baf1fb4ee192d3768218688","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 13:08:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/981f69cb01684ad188f8a3981a08cbae : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"202409c062069bbbfbbbb4ea5fe62d12af0efffd","revisions":{"202409c062069bbbfbbbb4ea5fe62d12af0efffd":{"kind":"REWORK","_number":1,"created":"2020-12-03 09:42:17.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/06/765306/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/06/765306/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/06/765306/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/06/765306/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/06/765306/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/06/765306/1"}}},"commit":{"parents":[{"commit":"fa37cdfbda7b52f37482b4f000d4e5b7af33fe1e","subject":"Merge \"Create Tempest case to ensure Services without Selectors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/fa37cdfbda7b52f37482b4f000d4e5b7af33fe1e"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-12-03 09:39:27.000000000","tz":60},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-12-03 09:39:27.000000000","tz":60},"subject":"Remove uselsess files and tests","message":"Remove uselsess files and tests\n\nRemoves some useless code:\n\n* Services directory is not used at all.\n* BaseAdminKuryrTest is not used as a base of any tests besides…\n* …test_create_list_pod which does not test Kuryr at all as it\u0027s not\n  checking if the pod got IP. Also it duplicates what other tests do.\n\nChange-Id: I3ad7cc10a307cb479dd335704a72fcef110e8bc3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/202409c062069bbbfbbbb4ea5fe62d12af0efffd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/202409c062069bbbfbbbb4ea5fe62d12af0efffd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
