)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~769735","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~I0cc13b0cecee2868273bebaeff41f99c60a19af5","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"make-stein-nonvoting","hashtags":[],"change_id":"I0cc13b0cecee2868273bebaeff41f99c60a19af5","subject":"Make Stein gate non-voting","status":"MERGED","created":"2021-01-07 14:29:57.000000000","updated":"2021-01-08 10:21:16.000000000","submitted":"2021-01-08 10:16:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"769735-make-stein-nonvoting","meta_rev_id":"f73aeb56035c5cecb6bb0a0ed3feb19a1ecb9223","_number":769735,"virtual_id_number":769735,"owner":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-08 10:16:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"all":[{"value":2,"date":"2021-01-07 14:36:30.000000000","_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-08 08:35:46.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"all":[{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-08 08:35:46.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"}],"reviewers":{"REVIEWER":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-07 14:36:30.000000000","updated_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"},{"updated":"2021-01-07 14:37:59.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2021-01-07 14:37:59.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2021-01-07 16:59:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3a3cf0869719dd269240e94247e6a290cca75941","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2021-01-07 14:29:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"14184ef41d5b2d22bcc4dda8dd3bac841e7c68c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2021-01-07 14:33:48.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3670df4e0891f968ab9ebde68f70517c5afc6af2","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2021-01-07 14:36:30.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7e6cf36d27f5b3c23eeb196184c368e7e7cc5b5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 16:59:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21526a25d3b3452884bcd81f6bd3afe7 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d532908ed7154d50bd328c1b4e909e09 : SUCCESS in 4m 20s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/c7ae57d9493f46afb594431779a0111c : SUCCESS in 1h 07m 08s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/a0ac2d9554554a3bac079fec9d097885 : SUCCESS in 1h 24m 56s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/f000f73f30114ab989f54bcdd5ff445d : SUCCESS in 1h 34m 54s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/4659752848354b65b45a98ab42148295 : SUCCESS in 1h 07m 52s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/877fb7abd4034a9aa25760581435a368 : SUCCESS in 1h 29m 11s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/7f1a96c650124cc283659e1531c05c03 : SUCCESS in 1h 27m 25s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/328b8d1959da45e39515a4f0c9c3cd9c : SUCCESS in 59m 53s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/6b979b5d030d44dfa43a9b26f0bfe5e0 : SUCCESS in 1h 08m 24s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/d1959394992243d6a9f63bbffbf9a1af : FAILURE in 1h 13m 53s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/da9537453f514226acf73d49e82e1bd3 : POST_FAILURE in 2h 12m 01s (non-voting)\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/50b0e3bed0d2479ea0a8cecc5a398fc4 : SUCCESS in 1h 10m 04s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/689d1e2758dd4671b8f1c7e665a7c31b : SUCCESS in 58m 22s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/17b64dcc5a5642d7a67af2154deffc58 : SUCCESS in 1h 25m 21s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/bf9a82f32b844f06b7c517109aacbcfe : FAILURE in 24m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b7eb0e35dc7803b9994b8fa0e2ab49565e779865","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2021-01-08 08:35:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b9e939d9ff3b97f9d679dc723f5c31fa6fc6ce8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 08:36:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"889c5ba676a77240b3602486a0193ec3acdf1e1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 10:16:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77f8bb560df04f409da4005d058b0171 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b799b24e11fd4a719be1977eb9fabac3 : SUCCESS in 3m 00s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/01897556f5c448a38f53c198f11b7ce4 : SUCCESS in 1h 04m 39s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/bb1d5289ac664b038af7fd4a11c62da7 : SUCCESS in 1h 12m 17s\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/cd7d4f0c784840cdbee5319805ad13dc : SUCCESS in 1h 34m 43s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/06e14769e7294d15aca48bf6811ec0aa : SUCCESS in 1h 21m 41s","accounts_in_message":[],"_revision_number":2},{"id":"8186cf93e305037c7205f093d2f973eec2e430d8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 10:16:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f73aeb56035c5cecb6bb0a0ed3feb19a1ecb9223","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 10:21:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caf7b82e317c4567ae433515931815ca : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a64609151864687daa2dbb042bfaeaa97bccf4c2","revisions":{"b136d9171fed1c6d5e1ea758fe2e9346196e4294":{"kind":"REWORK","_number":1,"created":"2021-01-07 14:29:57.000000000","uploader":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"ref":"refs/changes/35/769735/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/35/769735/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/35/769735/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/35/769735/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/35/769735/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/35/769735/1"}}},"commit":{"parents":[{"commit":"fa37cdfbda7b52f37482b4f000d4e5b7af33fe1e","subject":"Merge \"Create Tempest case to ensure Services without Selectors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/fa37cdfbda7b52f37482b4f000d4e5b7af33fe1e"}]}],"author":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2021-01-07 14:27:25.000000000","tz":-180},"committer":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2021-01-07 14:27:25.000000000","tz":-180},"subject":"Make Stein gate non-voting","message":"Make Stein gate non-voting\n\nWith the addition of Victoria gate, we\u0027re currently supporting\n4 other release jobs. It makes sense to make Stein job at least\nnon-voting.\n\nChange-Id: I0cc13b0cecee2868273bebaeff41f99c60a19af5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/b136d9171fed1c6d5e1ea758fe2e9346196e4294"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/b136d9171fed1c6d5e1ea758fe2e9346196e4294"}]},"branch":"refs/heads/master"},"a64609151864687daa2dbb042bfaeaa97bccf4c2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-01-07 14:33:48.000000000","uploader":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"ref":"refs/changes/35/769735/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/35/769735/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/35/769735/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/35/769735/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/35/769735/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/35/769735/2"}}},"commit":{"parents":[{"commit":"fa37cdfbda7b52f37482b4f000d4e5b7af33fe1e","subject":"Merge \"Create Tempest case to ensure Services without Selectors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/fa37cdfbda7b52f37482b4f000d4e5b7af33fe1e"}]}],"author":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2021-01-07 14:27:25.000000000","tz":-180},"committer":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2021-01-07 14:32:48.000000000","tz":-180},"subject":"Make Stein gate non-voting","message":"Make Stein gate non-voting\n\nWith the addition of Victoria gate, we\u0027re currently supporting\n3 other release jobs. It makes sense to make Stein job at least\nnon-voting.\n\nChange-Id: I0cc13b0cecee2868273bebaeff41f99c60a19af5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/a64609151864687daa2dbb042bfaeaa97bccf4c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/a64609151864687daa2dbb042bfaeaa97bccf4c2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
