)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~774205","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~I171a2ddd58626717e1c35ebf4134f11be1942c7e","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"readme","hashtags":[],"change_id":"I171a2ddd58626717e1c35ebf4134f11be1942c7e","subject":"Updates Kuryr tempest README","status":"MERGED","created":"2021-02-05 08:38:22.000000000","updated":"2021-02-10 21:54:31.000000000","submitted":"2021-02-10 21:52:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"774205-readme","meta_rev_id":"5886d53945123b19337a9aa814157fd4ad8c1245","_number":774205,"virtual_id_number":774205,"owner":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-10 21:52:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2021-02-09 16:16:31.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":2,"date":"2021-02-05 15:28:59.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":2,"date":"2021-02-08 07:51:28.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":1,"date":"2021-02-08 11:57:59.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}],"CC":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-05 08:39:26.000000000","updated_by":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"},{"updated":"2021-02-05 08:39:26.000000000","updated_by":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"},{"updated":"2021-02-05 08:43:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-05 15:29:38.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2021-02-05 15:29:38.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2021-02-10 21:34:26.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"CC"}],"messages":[{"id":"2302fbe02b466aee79b5fbc8e240c8a7f6905407","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"date":"2021-02-05 08:38:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e4b0ebf6675eb9bb274e75a34cf43792025e653d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 08:43:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/407e00c47e2a45a1a823162e6c71d7c2 : SUCCESS in 1m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53b3a4c12bbc43148b08e9cd7e0f77fb : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc6e2637c52745869b754c6528c39b7d : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":1},{"id":"45113ce2d48028cf1874f8919a4c7084f8cb66fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"date":"2021-02-05 09:56:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6fcbd26d3010b00c8631c7997bd97d92a581d48c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 10:06:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/176d10dd3fdb4cf9855891380b998f3c : SUCCESS in 1m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72956ad45acb4448bb895c5136abc9fd : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3a73120050341229eefc364b89fa154 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":2},{"id":"e0ddb5215bb9a9fca7bf45bb2036bf237928f060","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"date":"2021-02-05 10:25:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"38d91de2919ba21ff9fad5cbab2a0bd1268c39ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 10:31:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ad2cfaa5b8464c1cb92c03e291ef005d : SUCCESS in 1m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2db2e40bbf9a44b8b0bb841532339fc6 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9db3bf191b1c478080de26305d337ee3 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":3},{"id":"e39e7acb2ed2dead0982208338e5b049784387c3","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2021-02-05 10:35:33.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nLooks good, only one remark.","accounts_in_message":[],"_revision_number":3},{"id":"3d6c27130b79d428b4e7708e9f695f3f51ad9da7","author":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"date":"2021-02-05 10:45:53.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3: Code-Review-1\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e Looks good, only one remark.\n\nOh, okay, thank you ;)","accounts_in_message":[],"_revision_number":3},{"id":"9f5ddadf27614dc9876c00d0d2a731953c090de6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"date":"2021-02-05 10:48:21.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b52ae326469484f0d1fd1a3e66234710b21424a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 10:52:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e2c264b9172f4341b5ed5bccc4f82150 : SUCCESS in 2m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee643920d509418fb61e89a8fbd51d65 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72f315d57af24a25a84f33bd7e299565 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":4},{"id":"4e16a10b8a49fa39242ce719b750eaa45d6fcab1","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2021-02-05 15:28:59.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"d2d4a1c109f77081f9090f1cfaf749db7b166a38","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2021-02-08 07:51:28.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"faddd70907c12621f4f3cc8a515e41d6b3b80472","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2021-02-08 11:57:59.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"e66e56eeb660d3f1bd6bc87fe5e34ea8632c5b18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 11:58:13.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"3101f610e168b6c1135cbf26864398c1497d4a68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 12:07:00.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2d4aa171d3584464bf78ae68fd2b45da : FAILURE in 2m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee42c9a9063c467d8933c32d06a2bce7 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b30e60ce1b4449bc8c23249cd88c230e : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":4},{"id":"4f4d0a82b1f2726477c97eaed382b0ddd0ca9f10","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2021-02-08 12:34:25.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"0618e60bbac2844f89a0f5c4c5922e0a119be6e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 12:42:50.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8e6c69fc913449c0bc9042a3686b4753 : FAILURE in 2m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/867109c967004e3aade0f2e4f79e5877 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bae21339fba74f3bbb9d0f05686a64ba : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":4},{"id":"29320689fcb91524befa96310a51a3d7bafce783","author":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"date":"2021-02-08 12:52:32.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"f1c3e0b50e542a80bd9e8f5add143720d7ccab03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 13:01:43.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ce594a8ef8bb49f3bc3031b744ee7b54 : FAILURE in 2m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9baba5d4015a4c24a0912833325e54c4 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8682ee9f5a0a48c1b32572fdef251ff6 : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":4},{"id":"16c15d4919a13020b537bcb532ea4349cab8b9a4","author":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"date":"2021-02-08 13:29:28.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"dae32ec68f422b807df6226bb8383d6882e97a9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 13:43:19.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/46ccb4ccb84c4876abe58739dc47919f : FAILURE in 2m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/613e08854cbb4c508d6dc6488c4c81e5 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87d7b7865f7544818e6d5bacf6f3df54 : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":4},{"id":"57666e4fb187a18d6c3b756b5ace7cf7ef10641d","author":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"date":"2021-02-08 20:48:45.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"0620ce9a4ba9fc9471f9da17579890a3fb4d450e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 20:52:54.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a5d4b658d5b149abb0412c49020b4e5b : FAILURE in 1m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac97ef64aeba423095265ab6527c0740 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceeba9ee1d9e402183ce6ddddb90d609 : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":4},{"id":"28533f0d40e11686078797f18e1dae2ca62b49f7","author":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"date":"2021-02-09 15:36:36.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"a384385aaa38ac7dbb4c3dc553926f457c26c069","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 15:56:15.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fef6bf324a194717b01e913af0a6aa5a : FAILURE in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3edadbbaceb44b0d8b58c89b2f16b9ca : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/117b1b8e81d24583a2f4189bf00e3f72 : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":4},{"id":"0afb82393d5a28f27b528388240345792d4987dd","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-02-09 16:16:08.000000000","message":"Patch Set 4:\n\nIIUC patch [1] should fix this.\n\n[1] https://review.opendev.org/c/openstack/requirements/+/774593","accounts_in_message":[],"_revision_number":4},{"id":"a3a8b71527d997cfee4f3b0a4db451c9af5854e5","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-02-09 16:16:31.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"1754d58c4871d025ec4631e59b2ab2ee2c0db065","author":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"date":"2021-02-09 16:44:30.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 4:\n\u003e \n\u003e IIUC patch [1] should fix this.\n\u003e \n\u003e [1] https://review.opendev.org/c/openstack/requirements/+/774593\n\nOh okay, thank you.","accounts_in_message":[],"_revision_number":4},{"id":"5a43ccf7991fa2bea10e1c1502451b9de87e88ae","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-02-10 17:01:39.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"afa30d50a85f5da7f75fc0e1b3b709629133f684","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 17:23:44.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e64485629adf4362b4d0a7a361c2b440 : FAILURE in 2m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f42d48fcf2114bdeac596cfc663a6a9f : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/514c21f7e4844573af5c53f4d0c86474 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":4},{"id":"7501e13992a5b52044e10d51b7dbd38385143d40","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-02-10 21:34:26.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"2ce44e2250662c019ed7da439d1587c8ad61175d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 21:43:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a488b1c24c594f08805136f44d9986ee : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42e028d15a6d4f589cab806c7bd66595 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29935fd24e5048afa110ec9546e565d7 : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":4},{"id":"1340b73e2cc41ca9d71a18e840fa37b09dd97625","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 21:43:57.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"def3fcd04fc954b52ede3cbe0399aa51c6b4b909","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 21:52:42.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a316be5a8e464c8593528bc1b1bc17cf : SUCCESS in 2m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85e2a0db12d44632ae43427d3a81f34e : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d327bebbed2341da94d5a4c025b8c4a2 : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":4},{"id":"60b77653ea1f29280c79460d6c3639e934087d33","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 21:52:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"5886d53945123b19337a9aa814157fd4ad8c1245","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 21:54:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37b2f05630024207a6d08c5fecce17e1 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e38e4323063abe3810d00188e4d9402cbee05320","revisions":{"4a5b9d932aba62eb63f54c9cd08d7701c6b54a77":{"kind":"REWORK","_number":1,"created":"2021-02-05 08:38:22.000000000","uploader":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"ref":"refs/changes/05/774205/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/05/774205/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/05/774205/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/05/774205/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/05/774205/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/05/774205/1"}}},"commit":{"parents":[{"commit":"a9114ee09ae88ec58bee1597a5767b9b3020e2a8","subject":"Merge \"Adds Kuryr-sctp-demo application\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/a9114ee09ae88ec58bee1597a5767b9b3020e2a8"}]}],"author":{"name":"Tabitha","email":"fifedolapo02@gmail.com","date":"2021-02-05 08:36:01.000000000","tz":60},"committer":{"name":"Tabitha","email":"fifedolapo02@gmail.com","date":"2021-02-05 08:36:01.000000000","tz":60},"subject":"Updates Kuryr tempest README","message":"Updates Kuryr tempest README\n\nUpdates the steps and commands for running the kuryr-tempest-plugin\ntests.\n\nChange-Id: I171a2ddd58626717e1c35ebf4134f11be1942c7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/4a5b9d932aba62eb63f54c9cd08d7701c6b54a77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/4a5b9d932aba62eb63f54c9cd08d7701c6b54a77"}]},"branch":"refs/heads/master"},"f841a61cd21983406741655ed8c6264bced43e47":{"kind":"REWORK","_number":2,"created":"2021-02-05 09:56:46.000000000","uploader":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"ref":"refs/changes/05/774205/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/05/774205/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/05/774205/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/05/774205/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/05/774205/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/05/774205/2"}}},"commit":{"parents":[{"commit":"a9114ee09ae88ec58bee1597a5767b9b3020e2a8","subject":"Merge \"Adds Kuryr-sctp-demo application\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/a9114ee09ae88ec58bee1597a5767b9b3020e2a8"}]}],"author":{"name":"Tabitha","email":"fifedolapo02@gmail.com","date":"2021-02-05 08:36:01.000000000","tz":60},"committer":{"name":"Tabitha","email":"fifedolapo02@gmail.com","date":"2021-02-05 09:55:40.000000000","tz":60},"subject":"Updates Kuryr tempest README","message":"Updates Kuryr tempest README\n\nUpdates the steps and commands for running the kuryr-tempest-plugin\ntests.\n\nChange-Id: I171a2ddd58626717e1c35ebf4134f11be1942c7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/f841a61cd21983406741655ed8c6264bced43e47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/f841a61cd21983406741655ed8c6264bced43e47"}]},"branch":"refs/heads/master"},"f74c0030c57d617d6d4a2dd668ca69e3b87822e1":{"kind":"REWORK","_number":3,"created":"2021-02-05 10:25:29.000000000","uploader":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"ref":"refs/changes/05/774205/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/05/774205/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/05/774205/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/05/774205/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/05/774205/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/05/774205/3"}}},"commit":{"parents":[{"commit":"a9114ee09ae88ec58bee1597a5767b9b3020e2a8","subject":"Merge \"Adds Kuryr-sctp-demo application\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/a9114ee09ae88ec58bee1597a5767b9b3020e2a8"}]}],"author":{"name":"Tabitha","email":"fifedolapo02@gmail.com","date":"2021-02-05 08:36:01.000000000","tz":60},"committer":{"name":"Tabitha","email":"fifedolapo02@gmail.com","date":"2021-02-05 10:24:40.000000000","tz":60},"subject":"Updates Kuryr tempest README","message":"Updates Kuryr tempest README\n\nUpdates the steps and commands for running the kuryr-tempest-plugin\ntests.\n\nChange-Id: I171a2ddd58626717e1c35ebf4134f11be1942c7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/f74c0030c57d617d6d4a2dd668ca69e3b87822e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/f74c0030c57d617d6d4a2dd668ca69e3b87822e1"}]},"branch":"refs/heads/master"},"e38e4323063abe3810d00188e4d9402cbee05320":{"kind":"REWORK","_number":4,"created":"2021-02-05 10:48:21.000000000","uploader":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"ref":"refs/changes/05/774205/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/05/774205/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/05/774205/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/05/774205/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/05/774205/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/05/774205/4"}}},"commit":{"parents":[{"commit":"a9114ee09ae88ec58bee1597a5767b9b3020e2a8","subject":"Merge \"Adds Kuryr-sctp-demo application\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/a9114ee09ae88ec58bee1597a5767b9b3020e2a8"}]}],"author":{"name":"Tabitha","email":"fifedolapo02@gmail.com","date":"2021-02-05 08:36:01.000000000","tz":60},"committer":{"name":"Tabitha","email":"fifedolapo02@gmail.com","date":"2021-02-05 10:47:44.000000000","tz":60},"subject":"Updates Kuryr tempest README","message":"Updates Kuryr tempest README\n\nUpdates the steps and commands for running the kuryr-tempest-plugin\ntests.\n\nChange-Id: I171a2ddd58626717e1c35ebf4134f11be1942c7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/e38e4323063abe3810d00188e4d9402cbee05320"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/e38e4323063abe3810d00188e4d9402cbee05320"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
