)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~774545","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~Ic63a2807238d3d879bb172d74b05bada2399f07f","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"bp/sctp-support","hashtags":[],"change_id":"Ic63a2807238d3d879bb172d74b05bada2399f07f","subject":"Adds connectivity test for SCTP service","status":"MERGED","created":"2021-02-08 21:51:27.000000000","updated":"2021-02-23 13:35:09.000000000","submitted":"2021-02-23 13:33:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"774545-bp/sctp-support","meta_rev_id":"d6e711c7a0d6d58990c0c003f0c58ed3e79a0889","_number":774545,"virtual_id_number":774545,"owner":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-23 13:33:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":2,"date":"2021-02-22 17:34:12.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":2,"date":"2021-02-23 11:46:24.000000000","_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":1,"date":"2021-02-23 11:46:24.000000000","_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-09 00:04:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-09 07:57:03.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2021-02-09 09:27:27.000000000","updated_by":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"},{"updated":"2021-02-09 09:27:27.000000000","updated_by":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2021-02-09 09:27:27.000000000","updated_by":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"}],"messages":[{"id":"de439d6f48cc19f925a873450774c864b6447267","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"date":"2021-02-08 21:51:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff412c49b341aa5ddba3d8be1f6643c1a891859d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 00:04:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7afbd7d8b1aa46de8f516bc69409e350 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f78982db3404999a401493292160f0f : SUCCESS in 3m 46s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/3e94565591ed4fe7b1982b08cad9ffc2 : SUCCESS in 1h 01m 24s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/99d506caa41d4fa9a9f4ef175bbc24d1 : SUCCESS in 1h 09m 12s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/b3546132dd9b4bd093b0369f3e322051 : SUCCESS in 1h 07m 55s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/13ce52a21cd44184ab012074cb961a20 : SUCCESS in 1h 05m 44s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/b3b12f5d08024b01a6c1535e2e1a6811 : SUCCESS in 1h 08m 41s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/2b12531d22424f8ba49ca4b0bf7266b3 : SUCCESS in 1h 22m 30s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/e05b381e28fd42919488d8245bc562de : SUCCESS in 58m 40s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/58f349545e8f46afa9893bd54af07ce1 : SUCCESS in 57m 24s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/aea54e95f4c749d683dde97da6ff2349 : SUCCESS in 1h 09m 23s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/c3b4b34efc074d688deab3b3c7b9dcb8 : POST_FAILURE in 2h 09m 34s (non-voting)\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/4c6b34b956b04cc2bb9eb55ddee5d67d : SUCCESS in 51m 33s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/8ff6279970384189b7cf639a1769def6 : SUCCESS in 58m 06s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/d35d91c0e0b14355b170d2cb2861c671 : FAILURE in 35m 13s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/d6970ec1d6a941a5b5057a2f977a98f5 : FAILURE in 27m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2ee56b7b3f5f052a6a3f7aee82d80c4e2b968bc2","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2021-02-09 07:57:03.000000000","message":"Patch Set 1: Code-Review-1\n\nwould be nice to temporary enable the SCTP test to check if it works as expected. Other than that, it looks great","accounts_in_message":[],"_revision_number":1},{"id":"44b37767a41f8c5586b072b8a8ea717cac2769ed","author":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"date":"2021-02-09 09:15:26.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e would be nice to temporary enable the SCTP test to check if it works as expected. Other than that, it looks great\n\nWill do, thanks.","accounts_in_message":[],"_revision_number":1},{"id":"66d2e11ffd7c740f3c5b67326274e5d4d78e218b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"date":"2021-02-09 09:25:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e7a8c12f8478cfdcb4e168f53a2d4410055190c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 12:18:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f0475a1fa8a45788cf3f6a35cbf1df8 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1476d96e9f34fe4829e58c11e96ff87 : SUCCESS in 3m 22s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/e78ecea8af374eeb9430de9f9e2609c9 : FAILURE in 1h 01m 25s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/37ece9c4d13a40699b69f546bb0d9d96 : FAILURE in 1h 24m 40s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/5f0d104830224756a69ffea61035402e : FAILURE in 1h 36m 45s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/cf74545e78f74b018daf338631339129 : FAILURE in 1h 16m 06s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/50f43e6f725e4da8a811f6c8322ae638 : FAILURE in 1h 31m 19s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/eb097d7e8cfc4aa3ace7f64be1af1dfd : FAILURE in 1h 43m 44s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/ad291ae204b9443a8a2f265881e3e04a : FAILURE in 1h 19m 21s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/a282afa58adc433f8bcb17acc8df14c6 : FAILURE in 1h 17m 55s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/14d686d42ed44614aa1de4f5e70fa9e9 : FAILURE in 1h 16m 14s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/41e1d400082f4ac7be3744f1aa4dadc9 : SUCCESS in 2h 32m 45s (non-voting)\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/9b79833112574d5f97f2588ed81b92b0 : FAILURE in 1h 00m 09s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/e59e1e756a384b5c8e669b248adaa85b : FAILURE in 1h 02m 24s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/3ab35e3afb2c422d8c33778738018831 : FAILURE in 42m 47s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/169665f6519a463d92a7b0639577ac03 : FAILURE in 32m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3ff5470283e9c86f6dae3ab1808cd6585eb18d3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"date":"2021-02-09 15:31:14.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"c98e85cfd7a87bac553b16e10fd15b4e8c060aa6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 18:17:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56887b068a434abda4432c071d757dd5 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ab2418b181e425487c5b305df403f23 : SUCCESS in 3m 27s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/c10f51025cd64c66927a5a335c4ef94d : FAILURE in 1h 05m 31s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/9915f8ae5d2946729d87c5f1558e28b6 : FAILURE in 1h 14m 26s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/59c1a5cd701044228a4bc1174855d8e2 : FAILURE in 1h 10m 28s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/41a5ae6216c24fc9a75ee1fa18f9f92d : FAILURE in 1h 44m 58s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/3cee4645883341a9b6b50c721aeba54b : FAILURE in 1h 42m 28s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/08bf4d93a93d41a1ba2eb26a0dd97550 : FAILURE in 1h 38m 26s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/3fc34836254a410c8a0f10b3a6b01fef : FAILURE in 1h 21m 08s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/4a2521a81faf452d90b21ed19cb1f5ac : FAILURE in 1h 44m 47s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/e7dcaf9ae9654c099109f97ccc4b290e : FAILURE in 1h 20m 52s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/cf5a5eaeae264777b5f65a6436ddea47 : POST_FAILURE in 2h 26m 27s (non-voting)\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/da6a69c019ac4ef39797a5b6efd7e631 : FAILURE in 58m 54s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/4f2013a49cf04cf5829f9d5635e08b6f : FAILURE in 1h 01m 30s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/f8cb09bfe7dc45e887fc866a57e353cd : FAILURE in 50m 02s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/ae747fac4a8d4048aa22d8348c8fb5ee : FAILURE in 26m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8970c86c59a5753acc00dec805ce692e7232cb34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"date":"2021-02-10 12:23:22.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c1534d7f41ad6c28e6ce5ac9a0df6a426d9bff2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 14:47:02.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff0c3a3ebaee4f0396e274f80ce1d590 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbd6f2d3967d41cbba4d698c2fb00066 : SUCCESS in 4m 00s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/063388ca34744a6c93c572d9ba920d42 : FAILURE in 1h 19m 42s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/f0590075c0014ccebb54fd8422bc4f2e : FAILURE in 1h 22m 27s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/bddfd8e64a554b2c8689d368ddea3123 : FAILURE in 1h 24m 49s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/3502912c0c6d4804a08bb54aacb32f56 : FAILURE in 1h 07m 48s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/6435923e9ef34fcf8d33d8c7d364d903 : FAILURE in 1h 18m 57s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/5daaf4e74fe744709d53f743d6686e3b : FAILURE in 1h 21m 57s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/9aba073644be4ebabcf27db32bc80ea3 : FAILURE in 1h 17m 08s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/fe68ee4eb0254bf5bbdaa808435f0bac : FAILURE in 1h 10m 47s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/65cc8be11cbd4c85891d816d214bcd27 : FAILURE in 1h 26m 00s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/5661ad74636945ec887856bac291c405 : POST_FAILURE in 2h 17m 35s (non-voting)\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/a4abf8353e864354a11e75d4e02f80f0 : SUCCESS in 56m 55s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/36b9638055944666baac2f2f0241ccaa : SUCCESS in 1h 04m 21s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/58257b033bf14d309bb3f5248d733d81 : FAILURE in 29m 52s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/3f6170c3bf074995b9b037f32c4c95bb : FAILURE in 26m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b61810a5da0d2cea37123c4d4eda6e2e4a3eff23","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2021-02-10 19:33:33.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"d5a5fc998fd2715c9355fa0c8eacb84e4f7c4ff0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 22:04:05.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9108d07f7a414e24ad9d999f991926d2 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06e435b851514ff89660b9a9ac9cecaf : SUCCESS in 5m 49s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/a3fc91d86e344fc89c36457a615cc265 : FAILURE in 1h 26m 10s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/aabafa492d234beb9913dc439fd34215 : FAILURE in 1h 27m 35s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/8d7202f4f9c245809d87437362f38892 : FAILURE in 1h 19m 32s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/b7a0a91427484ae1a7e8068bb7a372f9 : FAILURE in 1h 28m 32s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/c8d7d1c1475247979a689ca75a39b424 : FAILURE in 1h 15m 41s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/0c0cc8f46d964d78910ed0b07935d320 : FAILURE in 1h 46m 23s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/ddb3975ce8e547e79dc05490a2c0da40 : FAILURE in 1h 19m 05s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/335f3464610c4d22b1b7de600ee31296 : FAILURE in 1h 18m 22s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/6d5298b38d724e898328bdff5e3877d1 : SUCCESS in 1h 20m 37s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/aa3b4c3cfe3141568d2aee2464ebeebd : SUCCESS in 2h 20m 59s (non-voting)\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/6dfb92ec665b4e2c871c01154668b88f : SUCCESS in 1h 05m 27s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/2412b4faee5b45f1bec48ad480abbc7f : SUCCESS in 59m 45s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/8d4cb34b268c43338a945ff148663cb2 : FAILURE in 31m 33s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/64d0a00f81b54f28a4bcce59cbb0dad2 : FAILURE in 26m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"848e50e00ca22de192b164029e37b2a1ea5d63e5","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2021-02-11 15:01:02.000000000","message":"Patch Set 4: Code-Review-1\n\nGreat that one of the OVN gates is passing ;)\n\nI see the kuryr-kubernetes-tempest-containerized-ovn-ipv6 is failing. Can you check it? Thanks.","accounts_in_message":[],"_revision_number":4},{"id":"38b109ef318bca29b551dbf7bb1c3f8a3d837148","author":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"date":"2021-02-12 08:15:10.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 4: Code-Review-1\n\u003e \n\u003e Great that one of the OVN gates is passing ;)\n\u003e \n\u003e I see the kuryr-kubernetes-tempest-containerized-ovn-ipv6 is failing. Can you check it? Thanks.\n\nI\u0027m working on it, thanks","accounts_in_message":[],"_revision_number":4},{"id":"94e7377752b6de89a308ca9391d31caff56bcf46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"date":"2021-02-12 14:36:59.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"373a660386c3c1b653b2a067950b19d4be5f19ac","author":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"date":"2021-02-12 14:39:16.000000000","message":"Patch Set 5:\n\n\u003e Patch Set 4: Code-Review-1\n\u003e \n\u003e Great that one of the OVN gates is passing ;)\n\u003e \n\u003e I see the kuryr-kubernetes-tempest-containerized-ovn-ipv6 is failing. Can you check it? Thanks.\n\nSo the kuryr/sctp-demo application was only allowing connections from the IPv4 address family, I have updated it to include IPv6, that should fix the problem.","accounts_in_message":[],"_revision_number":5},{"id":"64d87d7bc3f00c8a743be40fd73a4757815f8049","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-02-12 14:42:59.000000000","message":"Patch Set 5:\n\n\u003e Patch Set 5:\n\u003e \n\u003e \u003e Patch Set 4: Code-Review-1\n\u003e \u003e \n\u003e \u003e Great that one of the OVN gates is passing ;)\n\u003e \u003e \n\u003e \u003e I see the kuryr-kubernetes-tempest-containerized-ovn-ipv6 is failing. Can you check it? Thanks.\n\u003e \n\u003e So the kuryr/sctp-demo application was only allowing connections from the IPv4 address family, I have updated it to include IPv6, that should fix the problem.\n\nNice, turns out having IPv6 gates pays off.","accounts_in_message":[],"_revision_number":5},{"id":"bc3abf63d1bf3fe0472006810407e4cfb2980c2d","author":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"date":"2021-02-12 14:45:52.000000000","message":"Patch Set 5:\n\n\u003e Patch Set 5:\n\u003e \n\u003e \u003e Patch Set 5:\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 4: Code-Review-1\n\u003e \u003e \u003e \n\u003e \u003e \u003e Great that one of the OVN gates is passing ;)\n\u003e \u003e \u003e \n\u003e \u003e \u003e I see the kuryr-kubernetes-tempest-containerized-ovn-ipv6 is failing. Can you check it? Thanks.\n\u003e \u003e \n\u003e \u003e So the kuryr/sctp-demo application was only allowing connections from the IPv4 address family, I have updated it to include IPv6, that should fix the problem.\n\u003e \n\u003e Nice, turns out having IPv6 gates pays off.\n\nright ;)","accounts_in_message":[],"_revision_number":5},{"id":"ba8e323d0bd8ea8b1b59d1d719fb01bd860ce195","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 15:16:07.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"d14046712e7aadaf8c8c4fd1ac88c3d81fec9ef2","author":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"date":"2021-02-15 09:18:13.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"37ba884cccee397065b072e309f2952953251b13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 11:46:41.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/363878418f504ccf96f55a013a1e828e : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fac21c3eb15e4f7098489478554c31f4 : SUCCESS in 3m 48s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/11f6ff827fd744cabdb8eda69f50f7af : FAILURE in 1h 35m 02s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/b3be3a4c39464fb28e87625a00de1a72 : SUCCESS in 1h 16m 55s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/661040038a894d0fb3eb65ed5379d414 : SUCCESS in 1h 30m 53s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/9e266df041c54e1b843a82ef4e6d0520 : FAILURE in 1h 36m 44s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/8cf62c13b2834222893f9dc4f361df2d : SUCCESS in 1h 32m 29s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/609778ee33294d0ab39ff8422bf649e9 : FAILURE in 1h 27m 32s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/13e75f388f3d48a5a991abec9f53061a : SUCCESS in 1h 05m 40s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/50a3dfacb0dd430bb85ecb5e8d686e40 : FAILURE in 1h 17m 19s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/9bc033165ac04be888e8ff47c1368bd1 : SUCCESS in 1h 19m 31s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/60f1f9b71562435d86032a0902ad4f89 : SUCCESS in 2h 19m 56s (non-voting)\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/d27fc77e61a9468dbe22294dbb09ccc5 : SUCCESS in 53m 49s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/c65f73e0ec1449df871da925ddfb71ef : SUCCESS in 1h 01m 59s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/a95ae2caa51843e2a96df388dfbf1c8f : FAILURE in 36m 12s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/628e11501f7641cc8cea9b6f1d7868f5 : FAILURE in 35m 23s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d76e01c872572f2284ab904aa01048bd172d157b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"date":"2021-02-15 12:21:39.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d24710f3d0c6b56d24b69c4560190d1c66465bbf","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-02-15 13:19:20.000000000","message":"Patch Set 6:\n\nHey, that\u0027s starting to look green! So the failures on kuryr-kubernetes-tempest-train and stein are due to some issues with Amphora build, we can ignore that for now (we\u0027ll probably just make the train gate non-voting to tackle this).\n\nRegarding the rest I don\u0027t really see a clear pattern here and there\u0027s nothing too obvious in the logs. This\u0027ll require some closer debugging to get right.","accounts_in_message":[],"_revision_number":6},{"id":"45c770e7ff78304983de1fdc32a9c57dc5f5fc20","author":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"date":"2021-02-15 13:37:23.000000000","message":"Patch Set 6:\n\n(1 comment)\n\n\u003e Patch Set 6:\n\u003e \n\u003e Hey, that\u0027s starting to look green! So the failures on kuryr-kubernetes-tempest-train and stein are due to some issues with Amphora build, we can ignore that for now (we\u0027ll probably just make the train gate non-voting to tackle this).\n\u003e \n\u003e Regarding the rest I don\u0027t really see a clear pattern here and there\u0027s nothing too obvious in the logs. This\u0027ll require some closer debugging to get right.\n\n\nProgress ;)... For those other tests, I\u0027ve removed the suspect in the last patch, let\u0027s see how it goes.","accounts_in_message":[],"_revision_number":6},{"id":"82df3ce191ce59855afabd994a99c2a2b4669d44","author":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"date":"2021-02-15 13:46:19.000000000","message":"Patch Set 6:\n\n\u003e Patch Set 6:\n\u003e \n\u003e Hey, that\u0027s starting to look green! So the failures on kuryr-kubernetes-tempest-train and stein are due to some issues with Amphora build, we can ignore that for now (we\u0027ll probably just make the train gate non-voting to tackle this).\n\u003e \n\u003e Regarding the rest I don\u0027t really see a clear pattern here and there\u0027s nothing too obvious in the logs. This\u0027ll require some closer debugging to get right.\n\nOh okay, thank you, I was wondering what was going on with the tempest-train and stein.","accounts_in_message":[],"_revision_number":6},{"id":"05a54ad1b51af2ff9142654cde3e54f89ed8fd18","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-02-15 14:35:04.000000000","message":"Patch Set 6:\n\n\u003e Patch Set 6:\n\u003e \n\u003e \u003e Patch Set 6:\n\u003e \u003e \n\u003e \u003e Hey, that\u0027s starting to look green! So the failures on kuryr-kubernetes-tempest-train and stein are due to some issues with Amphora build, we can ignore that for now (we\u0027ll probably just make the train gate non-voting to tackle this).\n\u003e \u003e \n\u003e \u003e Regarding the rest I don\u0027t really see a clear pattern here and there\u0027s nothing too obvious in the logs. This\u0027ll require some closer debugging to get right.\n\u003e \n\u003e Oh okay, thank you, I was wondering what was going on with the tempest-train and stein.\n\nI proposed [1] then.\n\n[1] https://review.opendev.org/c/openstack/kuryr-tempest-plugin/+/775659","accounts_in_message":[],"_revision_number":6},{"id":"48b65eacbaad42bf7b8c5c000606d09452b2ab3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 14:49:30.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91a2bb79c5f24a1788041383d47da054 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43f75f98acf54d17aec1d638d248013f : SUCCESS in 3m 46s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/cfc352f7792248bf89c85a35e41117da : SUCCESS in 1h 16m 02s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/c6125f1c2a574cbaab93fd099425183c : SUCCESS in 1h 27m 26s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/2c5e24cb46a24cdc8dcdc5963677bc3c : SUCCESS in 1h 12m 51s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/65ac99afe97642ed90b65f4f10e9df1d : SUCCESS in 1h 36m 11s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/f6afc362b7f345319957c828b471ffd7 : SUCCESS in 1h 28m 17s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/db640e4d46c84e91a69742ab600f0b7b : SUCCESS in 1h 22m 30s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/60c1d40a23734f9e97dc5370ccbc4c23 : SUCCESS in 1h 04m 29s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/e27cee34542b4b8183f16ce5dc56409b : SUCCESS in 1h 25m 20s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/23a3e517901f486eafaaf828f54c4f02 : SUCCESS in 1h 35m 16s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/9f66aa4246844a52aa8224254cb82fd9 : SUCCESS in 2h 13m 02s (non-voting)\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/b0c392dac78646e9b3f3de8993b4ffd4 : SUCCESS in 59m 39s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/d295015ab93546ba8edb7238ee6e2fba : SUCCESS in 57m 02s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/b330708e646043b4b9fee20e671b2388 : FAILURE in 35m 12s\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/c926f2c531aa4c8e88a40e361726cf25 : FAILURE in 30m 32s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b9d61efd85f3b038572fe97506a6edde3b28968b","author":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"date":"2021-02-15 14:55:28.000000000","message":"Patch Set 6:\n\nfinally \\-/","accounts_in_message":[],"_revision_number":6},{"id":"047db274b8d7c545534f39022b807dfa817d5942","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-02-15 16:04:42.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)\n\nNice! Now we just need all the Octavia patches to merge. 😄","accounts_in_message":[],"_revision_number":6},{"id":"987db7f17a824b8668fab945a2afe3c25b089a98","author":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"date":"2021-02-15 16:34:24.000000000","message":"Patch Set 6:\n\n\u003e Patch Set 6: Code-Review+2\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e Nice! Now we just need all the Octavia patches to merge. 😄\n\nRight, thank you ;)","accounts_in_message":[],"_revision_number":6},{"id":"a52f611aab97ba5f9ccaed7476e6c96e79ed0823","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2021-02-15 22:11:07.000000000","message":"Patch Set 6: Code-Review+2\n\nLooks good!","accounts_in_message":[],"_revision_number":6},{"id":"72bb3b19e3273c1f8f31bdda46f2ccf7866c3cef","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2021-02-16 07:56:34.000000000","message":"Patch Set 6: Code-Review+2\n\nLooks good. If another respin is needed, please change the sctp_client to not use the camelCase","accounts_in_message":[],"_revision_number":6},{"id":"88376b4fcabbbdfd91fce7f27030b2e238fee702","author":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"date":"2021-02-16 08:37:08.000000000","message":"Patch Set 6:\n\n\u003e Patch Set 6: Code-Review+2\n\u003e \n\u003e Looks good. If another respin is needed, please change the sctp_client to not use the camelCase\n\nWill do, thanks.","accounts_in_message":[],"_revision_number":6},{"id":"f6483eac5e35bf0ff18be200a3604ef15bcc232c","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-02-17 16:11:11.000000000","message":"Patch Set 6:\n\nrecheck\n\nJust to get green here, patch making train job non-voting is merged.","accounts_in_message":[],"_revision_number":6},{"id":"3aefc203a6ace5f8941668beb0dbd2779e65d95a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 18:45:01.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca4a61cc95e14c0897b87d21f7de0a33 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc4f0d21e626488a8619f60f38ac006c : SUCCESS in 4m 27s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/08261d4dbeb249a98e34c5e7c28376d3 : SUCCESS in 1h 23m 07s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/f3576fd455cf49f0bac586dcf4bf1733 : SUCCESS in 1h 30m 57s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/eb10e8d26cd749cb82cc8a9c2c8e20a3 : SUCCESS in 1h 11m 02s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/d625ef5d75df4a4eb2439294042981d5 : SUCCESS in 1h 51m 31s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/f9127f56618c40beb00cad24ff41ad1d : SUCCESS in 1h 41m 09s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/465507a79e5d486f8ef35d2c4d63ebe2 : SUCCESS in 1h 37m 37s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/4041a3a3bd3e49a0845b9a1f887865d6 : SUCCESS in 1h 25m 21s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/42e2f8ecd7f3493c89bb43646f9ab22a : SUCCESS in 1h 30m 15s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/0620bbf484c24fef97cd2dc11621f419 : SUCCESS in 1h 32m 23s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/bfdb5f2a1dd14378b2619bbf4e38b3af : POST_FAILURE in 2h 24m 28s (non-voting)\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/468b883175c84d7d878cd3b0441a56c9 : POST_FAILURE in 57m 49s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/16b61ca875e946849a7bd84bd8cb9a16 : SUCCESS in 1h 07m 34s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/d7a9e312dd004288b3a68f75b874c9cf : FAILURE in 37m 19s (non-voting)\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/0d298f86edf34903a17abd2f9d77f372 : FAILURE in 27m 22s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e429f19a9aed1d826ab5a19f36339aa82dcfc162","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2021-02-17 21:11:41.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"499f4b5a9f7499aa465e41347ad9c2d23b14b74a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 23:28:55.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54dda83d0e9a4401945ae212319d1f49 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e09ce8d74e0d469e8987295cfd68e4df : SUCCESS in 4m 02s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/f0c4c772cff74e7d8c3cac765a4b681a : SUCCESS in 1h 32m 18s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/f5aaa89e05394e4b8df74cb140acea29 : SUCCESS in 1h 40m 19s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/501b72dd11a4488e9816605130c1b322 : SUCCESS in 1h 13m 46s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/39ddd5c29c6643f38cfbc0dfdfe312c8 : SUCCESS in 1h 18m 57s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/9b70cb7a675d4556a2c3fa04b66c224e : SUCCESS in 1h 56m 41s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/9ad545e8c45141ee94ad27a392c86fcf : SUCCESS in 1h 40m 16s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/cf843463c753409fa670e316435b98ed : SUCCESS in 1h 23m 32s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/30217a8c2965428691e24f6e288922ef : SUCCESS in 1h 12m 55s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/0421d48cf12c49f2afbd60d922d9f8b4 : SUCCESS in 1h 38m 50s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/749f436837c04b128e46b9b556cd5ce7 : SUCCESS in 2h 15m 16s (non-voting)\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/b44c722054714f20a3edd67af1ec2ebb : SUCCESS in 1h 01m 44s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/de6238ddf97a4f5b8910c5c807e060b6 : SUCCESS in 59m 35s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/ea73631d484642008cb8dc843d054fb7 : FAILURE in 43m 14s (non-voting)\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/8d328e6d1e51478ab1a1ce213a594d44 : FAILURE in 57m 28s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c374cf74b6188676645a9d1e16ba9484c1a8ce0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"date":"2021-02-18 15:34:56.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"f7b3df05d5a29aa6de11160625d7cc630f705909","author":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"date":"2021-02-18 15:39:17.000000000","message":"Patch Set 7:\n\nJust to confirm that the Amphora tarball is up to date and also remove the camel case in the sctp_client.","accounts_in_message":[],"_revision_number":7},{"id":"1e2ee0008eb46409a8403a172d157e6f531a8a48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 17:56:15.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfc3b7bc13b3442a8b2039bf4d6eb4f6 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de97cc159fcc40e2a4c8499ee9bf16b8 : SUCCESS in 4m 36s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/185181c85478444d815aad13eee79bfb : SUCCESS in 1h 17m 35s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/cb2eddbdbae041879f32460f7da41259 : SUCCESS in 1h 16m 04s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/d907720332db4e39bf69e62bb00b01fa : SUCCESS in 1h 29m 59s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/03928e2b978d4c5d9645a990dcc67e94 : SUCCESS in 1h 13m 05s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/7f54fa4dd95b48d7b37e6af4226bfea4 : SUCCESS in 1h 31m 15s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/8d188e5ed7f9490382ece218c4cda367 : SUCCESS in 1h 31m 21s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/9ded20a3959e4cabb0a1bf84ea6890b2 : SUCCESS in 1h 02m 13s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/f3d459b0ac6144d5af6b6c68da9ebc42 : SUCCESS in 1h 18m 54s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/e25decb4e88e42dab34accc8076b5341 : FAILURE in 1h 30m 03s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/1bffbc04e0c44b66a92120cca2869cfe : SUCCESS in 2h 12m 14s (non-voting)\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/670cf6e7ffbf417f8d2207a27c9ffd58 : FAILURE in 35m 11s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/cf158bae1e8942da84187f155c19cb4b : SUCCESS in 57m 18s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/6ff61617d5ff4bb9a91d20d8f4c4a41e : FAILURE in 35m 38s (non-voting)\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/b12427bc88b64688a0d315ba81c8151b : FAILURE in 25m 18s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b81cbef98ac9122682222b451f13c046ca218b9a","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2021-02-18 19:59:40.000000000","message":"Patch Set 7: Code-Review+2\n\nrecheck.","accounts_in_message":[],"_revision_number":7},{"id":"3cb5212041b6328ad8181a94cdab0578bca14fde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 22:35:35.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29e280e1a22b43ab855f2951244e4461 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2695421bf6b14fd8bdaff0afc2d7b826 : SUCCESS in 3m 51s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/852f4d76ae9c4ebcaf5c27ceb3d1f235 : SUCCESS in 53m 53s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/20555ef4d84f438aaf70cb638c8ed8f6 : SUCCESS in 1h 25m 43s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/a38b321321664e77883fc159f6215cd3 : SUCCESS in 1h 12m 54s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/901e83d188a94f2ab50973258c8127f7 : SUCCESS in 1h 15m 31s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/d46b0d7e13a245c7be77f149807c45af : SUCCESS in 1h 47m 45s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/f2cf51be08b543919faf7889ee1572e2 : SUCCESS in 1h 47m 47s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/24e9864bac094b40b60e7a1ee034f959 : SUCCESS in 1h 12m 03s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/b014c50c87dc452997e04037ba4beac8 : SUCCESS in 1h 12m 23s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/f5754554302e4944b60145fa6da5af22 : SUCCESS in 1h 23m 21s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/0131c46950f14d73aaa007bae24e3ae7 : POST_FAILURE in 2h 31m 54s (non-voting)\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/1d851cbbff9f4bc0a4e00e1d7a7f669d : SUCCESS in 1h 08m 58s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/00c722677a924b7691b8e00b9757ead6 : SUCCESS in 1h 02m 39s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/73cb362de7364cc8aefafb53e9aba8c2 : FAILURE in 36m 02s (non-voting)\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/70cfa80cc79047699459748af405869d : FAILURE in 32m 58s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b990373d5e6953be0f9c5ad89184a2b0b8e5c58d","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2021-02-19 07:29:09.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"8916092ea48e1045d41d430122cfc355097b5480","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2021-02-19 18:25:07.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"163d4943932bdfe14c55af0de2e2a214f91ac61a","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2021-02-22 14:50:15.000000000","message":"Patch Set 7:\n\nTabitha, could you remove the Depends-on from this patch so we can merge it?\n\nIt\u0027s fine to merge it as we already confirmed it works as expected.","accounts_in_message":[],"_revision_number":7},{"id":"8f7e88404f5bf2580662dff40874d726b426db65","author":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"date":"2021-02-22 17:07:34.000000000","message":"Patch Set 7:\n\n\u003e Patch Set 7:\n\u003e \n\u003e Tabitha, could you remove the Depends-on from this patch so we can merge it?\n\u003e \n\u003e It\u0027s fine to merge it as we already confirmed it works as expected.\n\nAlright, I will, thanks.","accounts_in_message":[],"_revision_number":7},{"id":"d25cab4563de443abe1b31ff8a563e6eea98a283","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"date":"2021-02-22 17:14:07.000000000","message":"Uploaded patch set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"92664121217a749fa52b6da080f76d6a9a58830b","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2021-02-22 17:34:12.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"5548d05454ef1de9d524ea0f2398e012c1b950bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 19:26:19.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ddf21cf4f0a4db9a4313fca15d5e102 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c7e389549fc47af9197abc7a6c6add7 : SUCCESS in 3m 25s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/2d6d4a89568c45b3bda2d87b005bd065 : SUCCESS in 1h 14m 33s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/c53b86c25c064aaa9888909a00a828b3 : SUCCESS in 1h 29m 04s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/b66ff708a0e2462381fd5e8e834c9798 : SUCCESS in 1h 17m 17s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/53dda83fb69642ec8aa90886739558cd : SUCCESS in 1h 18m 36s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/04ee0e2799c846b4976c098ab939ba42 : SUCCESS in 1h 35m 03s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/904fef6456ae47cbbcba0b4658521ecd : SUCCESS in 1h 32m 44s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/5643bc9e64e24672ba55488c571e2334 : SUCCESS in 1h 10m 27s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/2f22e9a1b5e744b682406826f7a63c9e : SUCCESS in 1h 22m 22s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/74ef153cdcb1495094af6eb1dccd0629 : SUCCESS in 1h 14m 39s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/f1449dd9b2cf4f15af67cd239483605b : SUCCESS in 2h 08m 34s (non-voting)\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/b6a532b36bc54220bc4ad53134cd0c53 : SUCCESS in 48m 20s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/7e7794cf0cd64d1c8955b6906b614034 : SUCCESS in 1h 28m 47s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/ca812cbd21a44781af612dd12589a27c : SUCCESS in 1h 13m 57s (non-voting)\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/f7f9c61912104416a95b65fbb3514bc6 : FAILURE in 34m 13s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"1895a12cb47e85dc19c67a5b11444bee4360cc8c","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2021-02-23 11:46:24.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"33c213e57321306ecaf7773cd53c5b8302a3209e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 11:48:49.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"7abb738bd48a5ebe681448ec9c6c78217aa7cc56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 13:33:58.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/792c1bab7c134cb0b3266368538f05df : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab7a4124c3734e87af10e881d24f3e73 : SUCCESS in 4m 47s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/f9b4bf9f81e6405fbb33ad161083894f : SUCCESS in 1h 00m 42s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/44498b3d7b7548e6901433c820d85542 : SUCCESS in 1h 18m 21s\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/97dbd702e705464b91adb59aca603f39 : SUCCESS in 1h 25m 51s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/488679af713e4c1b8f878040f8986871 : SUCCESS in 1h 38m 30s","accounts_in_message":[],"_revision_number":8},{"id":"e862b3877f4dc8f4666e48ad131861fba103a5c9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 13:33:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"d6e711c7a0d6d58990c0c003f0c58ed3e79a0889","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 13:35:09.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06ab8300fe5e4af3b9a55955d46e3f39 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"34fe966f340cda0926a008a555e770ae53ad337f","revisions":{"ae7ddd61a1c7140e02bc453703a092fa0c6e835c":{"kind":"REWORK","_number":1,"created":"2021-02-08 21:51:27.000000000","uploader":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"ref":"refs/changes/45/774545/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/45/774545/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/45/774545/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/45/774545/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/45/774545/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/45/774545/1"}}},"commit":{"parents":[{"commit":"a9114ee09ae88ec58bee1597a5767b9b3020e2a8","subject":"Merge \"Adds Kuryr-sctp-demo application\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/a9114ee09ae88ec58bee1597a5767b9b3020e2a8"}]}],"author":{"name":"Tabitha","email":"fifedolapo02@gmail.com","date":"2021-02-08 21:48:39.000000000","tz":60},"committer":{"name":"Tabitha","email":"fifedolapo02@gmail.com","date":"2021-02-08 21:51:23.000000000","tz":60},"subject":"Adds connectivity test for SCTP service","message":"Adds connectivity test for SCTP service\n\nThis patch adds a new test to check SCTP connectivity from pods.\nThe \u0027create_setup_for_service_test\u0027 and \u0027assert_backend_amount_from_pod\u0027\nfunctions were also extended to support SCTP services.\n\nDepends-On: https://review.opendev.org/c/openstack/ovn-octavia-provider/+/754429\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/753247\nPartially-Implements: blueprint sctp-support\nChange-Id: Ic63a2807238d3d879bb172d74b05bada2399f07f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ae7ddd61a1c7140e02bc453703a092fa0c6e835c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ae7ddd61a1c7140e02bc453703a092fa0c6e835c"}]},"branch":"refs/heads/master"},"0451266672aeaeb812730b7bd761d36aa4457e1b":{"kind":"REWORK","_number":2,"created":"2021-02-09 09:25:04.000000000","uploader":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"ref":"refs/changes/45/774545/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/45/774545/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/45/774545/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/45/774545/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/45/774545/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/45/774545/2"}}},"commit":{"parents":[{"commit":"a9114ee09ae88ec58bee1597a5767b9b3020e2a8","subject":"Merge \"Adds Kuryr-sctp-demo application\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/a9114ee09ae88ec58bee1597a5767b9b3020e2a8"}]}],"author":{"name":"Tabitha","email":"fifedolapo02@gmail.com","date":"2021-02-08 21:48:39.000000000","tz":60},"committer":{"name":"Tabitha","email":"fifedolapo02@gmail.com","date":"2021-02-09 09:23:12.000000000","tz":60},"subject":"Adds connectivity test for SCTP service","message":"Adds connectivity test for SCTP service\n\nThis patch adds a new test to check SCTP connectivity from pods.\nThe \u0027create_setup_for_service_test\u0027 and \u0027assert_backend_amount_from_pod\u0027\nfunctions were also extended to support SCTP services.\n\nDepends-On: https://review.opendev.org/c/openstack/ovn-octavia-provider/+/754429\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/753247\nPartially-Implements: blueprint sctp-support\nChange-Id: Ic63a2807238d3d879bb172d74b05bada2399f07f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/0451266672aeaeb812730b7bd761d36aa4457e1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/0451266672aeaeb812730b7bd761d36aa4457e1b"}]},"branch":"refs/heads/master"},"43ca75db7503667931df75916c57f2cca52f3377":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-02-09 15:31:14.000000000","uploader":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"ref":"refs/changes/45/774545/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/45/774545/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/45/774545/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/45/774545/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/45/774545/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/45/774545/3"}}},"commit":{"parents":[{"commit":"a9114ee09ae88ec58bee1597a5767b9b3020e2a8","subject":"Merge \"Adds Kuryr-sctp-demo application\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/a9114ee09ae88ec58bee1597a5767b9b3020e2a8"}]}],"author":{"name":"Tabitha","email":"fifedolapo02@gmail.com","date":"2021-02-08 21:48:39.000000000","tz":60},"committer":{"name":"Tabitha","email":"fifedolapo02@gmail.com","date":"2021-02-09 15:17:06.000000000","tz":60},"subject":"Adds connectivity test for SCTP service","message":"Adds connectivity test for SCTP service\n\nThis patch adds a new test to check SCTP connectivity from pods.\nThe \u0027create_setup_for_service_test\u0027 and \u0027assert_backend_amount_from_pod\u0027\nfunctions were also extended to support SCTP services.\n\nDepends-On: I30997ae6cc6b8ec724f0e9dcfdfe49356b320ff4\nDepends-On: Ica57a5c77af61bffba812b35608bfc2a90011ba8\nPartially-Implements: blueprint sctp-support\nChange-Id: Ic63a2807238d3d879bb172d74b05bada2399f07f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/43ca75db7503667931df75916c57f2cca52f3377"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/43ca75db7503667931df75916c57f2cca52f3377"}]},"branch":"refs/heads/master"},"9f455c10c79fb867fc7c35db4b859aa6921ae92f":{"kind":"REWORK","_number":4,"created":"2021-02-10 12:23:22.000000000","uploader":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"ref":"refs/changes/45/774545/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/45/774545/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/45/774545/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/45/774545/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/45/774545/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/45/774545/4"}}},"commit":{"parents":[{"commit":"a9114ee09ae88ec58bee1597a5767b9b3020e2a8","subject":"Merge \"Adds Kuryr-sctp-demo application\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/a9114ee09ae88ec58bee1597a5767b9b3020e2a8"}]}],"author":{"name":"Tabitha","email":"fifedolapo02@gmail.com","date":"2021-02-08 21:48:39.000000000","tz":60},"committer":{"name":"Tabitha","email":"fifedolapo02@gmail.com","date":"2021-02-10 12:22:36.000000000","tz":60},"subject":"Adds connectivity test for SCTP service","message":"Adds connectivity test for SCTP service\n\nThis patch adds a new test to check SCTP connectivity from pods.\nThe \u0027create_setup_for_service_test\u0027 and \u0027assert_backend_amount_from_pod\u0027\nfunctions were also extended to support SCTP services.\n\nDepends-On: https://review.opendev.org/c/openstack/kuryr-kubernetes/+/774844\nPartially-Implements: blueprint sctp-support\nChange-Id: Ic63a2807238d3d879bb172d74b05bada2399f07f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/9f455c10c79fb867fc7c35db4b859aa6921ae92f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/9f455c10c79fb867fc7c35db4b859aa6921ae92f"}]},"branch":"refs/heads/master"},"fd8cf738c659cd59f7c80050d6e03db7b16d59b3":{"kind":"REWORK","_number":5,"created":"2021-02-12 14:36:59.000000000","uploader":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"ref":"refs/changes/45/774545/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/45/774545/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/45/774545/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/45/774545/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/45/774545/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/45/774545/5"}}},"commit":{"parents":[{"commit":"a9114ee09ae88ec58bee1597a5767b9b3020e2a8","subject":"Merge \"Adds Kuryr-sctp-demo application\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/a9114ee09ae88ec58bee1597a5767b9b3020e2a8"}]}],"author":{"name":"Tabitha","email":"fifedolapo02@gmail.com","date":"2021-02-08 21:48:39.000000000","tz":60},"committer":{"name":"Tabitha","email":"fifedolapo02@gmail.com","date":"2021-02-12 13:54:25.000000000","tz":60},"subject":"Adds connectivity test for SCTP service","message":"Adds connectivity test for SCTP service\n\nThis patch adds a new test to check SCTP connectivity from pods.\nThe \u0027create_setup_for_service_test\u0027 and \u0027assert_backend_amount_from_pod\u0027\nfunctions were also extended to support SCTP services. The kuryr-sctp\napplication was also extended to support connection with IPv6 address family.\n\nDepends-On: https://review.opendev.org/c/openstack/kuryr-kubernetes/+/774844\nPartially-Implements: blueprint sctp-support\nChange-Id: Ic63a2807238d3d879bb172d74b05bada2399f07f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/fd8cf738c659cd59f7c80050d6e03db7b16d59b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/fd8cf738c659cd59f7c80050d6e03db7b16d59b3"}]},"branch":"refs/heads/master"},"553ac2852b91f32bd7992b68947ac5a249eacd2b":{"kind":"REWORK","_number":6,"created":"2021-02-15 12:21:39.000000000","uploader":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"ref":"refs/changes/45/774545/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/45/774545/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/45/774545/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/45/774545/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/45/774545/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/45/774545/6"}}},"commit":{"parents":[{"commit":"a9114ee09ae88ec58bee1597a5767b9b3020e2a8","subject":"Merge \"Adds Kuryr-sctp-demo application\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/a9114ee09ae88ec58bee1597a5767b9b3020e2a8"}]}],"author":{"name":"Tabitha","email":"fifedolapo02@gmail.com","date":"2021-02-08 21:48:39.000000000","tz":60},"committer":{"name":"Tabitha","email":"fifedolapo02@gmail.com","date":"2021-02-15 12:20:44.000000000","tz":60},"subject":"Adds connectivity test for SCTP service","message":"Adds connectivity test for SCTP service\n\nThis patch adds a new test to check SCTP connectivity from pods.\nThe \u0027create_setup_for_service_test\u0027 and \u0027assert_backend_amount_from_pod\u0027\nfunctions were also extended to support SCTP services. The kuryr-sctp\napplication was also extended to support connection with IPv6 address family.\n\nDepends-On: https://review.opendev.org/c/openstack/kuryr-kubernetes/+/774844\nPartially-Implements: blueprint sctp-support\nChange-Id: Ic63a2807238d3d879bb172d74b05bada2399f07f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/553ac2852b91f32bd7992b68947ac5a249eacd2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/553ac2852b91f32bd7992b68947ac5a249eacd2b"}]},"branch":"refs/heads/master"},"9ce17c4844e3139155fc40a0368011a065b26189":{"kind":"REWORK","_number":7,"created":"2021-02-18 15:34:56.000000000","uploader":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"ref":"refs/changes/45/774545/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/45/774545/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/45/774545/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/45/774545/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/45/774545/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/45/774545/7"}}},"commit":{"parents":[{"commit":"a9114ee09ae88ec58bee1597a5767b9b3020e2a8","subject":"Merge \"Adds Kuryr-sctp-demo application\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/a9114ee09ae88ec58bee1597a5767b9b3020e2a8"}]}],"author":{"name":"Tabitha","email":"fifedolapo02@gmail.com","date":"2021-02-08 21:48:39.000000000","tz":60},"committer":{"name":"Tabitha","email":"fifedolapo02@gmail.com","date":"2021-02-18 15:34:07.000000000","tz":60},"subject":"Adds connectivity test for SCTP service","message":"Adds connectivity test for SCTP service\n\nThis patch adds a new test to check SCTP connectivity from pods.\nThe \u0027create_setup_for_service_test\u0027 and \u0027assert_backend_amount_from_pod\u0027\nfunctions were also extended to support SCTP services. The kuryr-sctp\napplication was also extended to support connection with IPv6 address family.\n\nDepends-On: https://review.opendev.org/c/openstack/kuryr-kubernetes/+/774844\nPartially-Implements: blueprint sctp-support\nChange-Id: Ic63a2807238d3d879bb172d74b05bada2399f07f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/9ce17c4844e3139155fc40a0368011a065b26189"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/9ce17c4844e3139155fc40a0368011a065b26189"}]},"branch":"refs/heads/master"},"34fe966f340cda0926a008a555e770ae53ad337f":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2021-02-22 17:14:07.000000000","uploader":{"_account_id":32533,"name":"Tabitha Fasoyin","email":"fifedolapo02@gmail.com","username":"tabbie-fash"},"ref":"refs/changes/45/774545/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/45/774545/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/45/774545/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/45/774545/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/45/774545/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/45/774545/8"}}},"commit":{"parents":[{"commit":"a9114ee09ae88ec58bee1597a5767b9b3020e2a8","subject":"Merge \"Adds Kuryr-sctp-demo application\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/a9114ee09ae88ec58bee1597a5767b9b3020e2a8"}]}],"author":{"name":"Tabitha","email":"fifedolapo02@gmail.com","date":"2021-02-08 21:48:39.000000000","tz":60},"committer":{"name":"Tabitha","email":"fifedolapo02@gmail.com","date":"2021-02-22 17:13:00.000000000","tz":60},"subject":"Adds connectivity test for SCTP service","message":"Adds connectivity test for SCTP service\n\nThis patch adds a new test to check SCTP connectivity from pods.\nThe \u0027create_setup_for_service_test\u0027 and \u0027assert_backend_amount_from_pod\u0027\nfunctions were also extended to support SCTP services. The kuryr-sctp\napplication was also extended to support connection with IPv6 address family.\n\nPartially-Implements: blueprint sctp-support\nChange-Id: Ic63a2807238d3d879bb172d74b05bada2399f07f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/34fe966f340cda0926a008a555e770ae53ad337f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/34fe966f340cda0926a008a555e770ae53ad337f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
