)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~776487","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~I56a9e8aad87e2281ccae22ed7f66538990e740a3","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"port-ool","hashtags":[],"change_id":"I56a9e8aad87e2281ccae22ed7f66538990e740a3","subject":"Increase the timeout for controller to be ready","status":"MERGED","created":"2021-02-18 16:03:26.000000000","updated":"2021-02-18 21:36:47.000000000","submitted":"2021-02-18 21:35:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"776487-port-ool","meta_rev_id":"fa04d67ce56cb768cf8e63011f0fff65cc837e06","_number":776487,"virtual_id_number":776487,"owner":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-18 21:35:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2021-02-18 16:20:40.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":2,"date":"2021-02-18 19:56:26.000000000","_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":1,"date":"2021-02-18 19:56:26.000000000","_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-18 16:20:40.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2021-02-18 18:40:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-18 19:56:26.000000000","updated_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"}],"messages":[{"id":"70fae8dc2137ecf0478b87f66d8071f175be3f45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2021-02-18 16:03:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"383d81170b099c3b6928b71a93570b282da7620b","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-02-18 16:20:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"025fc2020c9e4c1f30beed18b2ee9ddc6d8fa2dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 18:40:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c25b991cde774b9aa642e996c64c6357 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf9a6aa7a3dd4b0ea24459917ec3e3ff : SUCCESS in 3m 57s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/32521679dd0d4854861d5fc553bb907a : SUCCESS in 58m 37s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/a90f6a41734b43d88625d14fba930e6c : SUCCESS in 1h 05m 01s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/8796116e38584453a4f649a1bf0650f3 : SUCCESS in 1h 22m 57s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/33d6c3caad984caa97b5743a0728451e : SUCCESS in 1h 42m 32s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/b7c6232bd54a4cc19f7c2bd0ea828b7f : SUCCESS in 1h 29m 44s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/1cf7afef8e1d4d01813cf90fac1e550b : SUCCESS in 1h 24m 17s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/0e7f70391b454fecaacc9d4c361f501e : SUCCESS in 1h 15m 00s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/751c846eacb145a1a0fb5587a9c6c6ff : SUCCESS in 1h 12m 42s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/889513e7a28d46d6b80765e8ecd94193 : SUCCESS in 1h 35m 54s (non-voting)\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/b911419601ee493eac4d8dab2f2ea7c4 : SUCCESS in 2h 28m 49s (non-voting)\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/fec3ccdb2ea54e48a825676625d554fd : SUCCESS in 1h 01m 36s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/86b8aecc11fa41acaf8af7cc797d41d6 : SUCCESS in 53m 44s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/fdfe4960004c4f13a833af0e0ef9cbd7 : FAILURE in 48m 57s (non-voting)\n- kuryr-kubernetes-tempest-stein https://zuul.opendev.org/t/openstack/build/473725858e59439c80911f7a2e55292d : FAILURE in 32m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4e1e00fd8b08bb981b8e8c4c051df30be26ff1c5","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2021-02-18 19:56:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cd838986fa2796e6ad4ec3a2ff2ce33d7f51c01b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 19:57:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5c833eb7aa15a79dc053a70bbcdeb0b08f84339a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 21:35:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe65b6c74ad6413099191c9dafb697b9 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ce73c9edd8246488a95295aa9c15e2d : SUCCESS in 3m 30s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/441a25fd2dea4c0aaf7f77d72c9ec547 : SUCCESS in 1h 03m 10s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/8df678585b5540a5b1eb9e2973a5f8b1 : SUCCESS in 1h 27m 46s\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/8ad8216b9c364ea196cb41fdf6fb88e6 : SUCCESS in 1h 13m 08s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/e82fc533d14949a3a1acf0beba981819 : SUCCESS in 1h 30m 31s","accounts_in_message":[],"_revision_number":1},{"id":"21943d2e27d66beeb1dc52116a17b77261229afa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 21:35:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fa04d67ce56cb768cf8e63011f0fff65cc837e06","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 21:36:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3f450f4e67a48e5b77d854c21f336b8 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8f241906dd58b8467b0e3f533d261106d9b0ad60","revisions":{"8f241906dd58b8467b0e3f533d261106d9b0ad60":{"kind":"REWORK","_number":1,"created":"2021-02-18 16:03:26.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/87/776487/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/87/776487/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/87/776487/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/87/776487/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/87/776487/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/87/776487/1"}}},"commit":{"parents":[{"commit":"0412a574143abb059ab8236d521a43913c354020","subject":"Make stable/train job non-voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/0412a574143abb059ab8236d521a43913c354020"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2021-02-18 16:02:10.000000000","tz":120},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2021-02-18 16:02:10.000000000","tz":120},"subject":"Increase the timeout for controller to be ready","message":"Increase the timeout for controller to be ready\n\ntest_port_pool fails and it seems that\u0027s because it takes time for the\nKuryr controller to be ready\n\nChange-Id: I56a9e8aad87e2281ccae22ed7f66538990e740a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8f241906dd58b8467b0e3f533d261106d9b0ad60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8f241906dd58b8467b0e3f533d261106d9b0ad60"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
