)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~788839","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~Icc37ad3818d40f17b51827f8b0d19ca0c7c5bade","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"wallaby-stable-jobs","hashtags":[],"change_id":"Icc37ad3818d40f17b51827f8b0d19ca0c7c5bade","subject":"Add stable/wallaby jobs on master gate","status":"MERGED","created":"2021-04-29 23:32:25.000000000","updated":"2021-04-30 11:03:55.000000000","submitted":"2021-04-30 11:01:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"788839-wallaby-stable-jobs","meta_rev_id":"4bbd30da50ddbd59fdffa37a1d12190dfa990b8a","_number":788839,"virtual_id_number":788839,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-30 11:01:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2021-04-30 09:33:11.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":2,"date":"2021-04-30 08:23:03.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":1,"date":"2021-04-30 09:33:11.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-30 02:14:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-30 08:23:03.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"},{"updated":"2021-04-30 09:33:11.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"}],"messages":[{"id":"cdb41726ca1bc4a482d45066250d7d71700c8a8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-04-29 23:32:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c5f66bf5ae82b75108ac6776f75c3da19dc487e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 02:14:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b2f5f49eedb42cfb4a6de7b9bd9afce : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30c4f6500ccd41b197a47cbb0392cde3 : SUCCESS in 3m 23s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/8d79f1b6391c4e7db40cb8a89c5f673c : SUCCESS in 1h 27m 21s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/fc3be077bae94445a8a57a9fce4d72f3 : SUCCESS in 1h 16m 28s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/770b822892214c24a03329dc07f32afd : SUCCESS in 1h 32m 51s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/5359c62064874216b2e9b441955a6d78 : SUCCESS in 1h 22m 10s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/b450f9f795be4ebf9ac4803206e26b8c : SUCCESS in 1h 46m 53s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/85d5a1ede9f7435a9a2697ae0e2655bb : SUCCESS in 1h 19m 08s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/a422c961db5c4859bec5f571c3a9acca : SUCCESS in 1h 07m 55s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/8c4b71112e0a41909e72429f1779556f : SUCCESS in 1h 28m 30s\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/323cd9bbc0954fe084646b648a69eaf9 : SUCCESS in 1h 21m 54s\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/19f3d1d3f97d46779364ef1878e6740f : POST_FAILURE in 2h 35m 18s (non-voting)\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/f28a16420c3c40fab9ca65d62fb87cc1 : SUCCESS in 1h 08m 32s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/7e3ebaf1ecb043c48a039a3a70a1f4d8 : SUCCESS in 1h 18m 17s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/1b27f3d6144a4053bd3401a414887078 : SUCCESS in 1h 22m 41s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/e115960282f74b979772f3af868fb069 : SUCCESS in 1h 14m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ab5ac291e14c4636b9dbee35841f3ecb2f255860","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2021-04-30 08:23:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a09c4bd2f45fe970a61c5c02460f41907f80c92b","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-04-30 09:33:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8f5b045359f5915c26bf50dedc819bbf0f9deccd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 09:33:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"319786c5f3c7d2e5413afb00d859bc14e4884a65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 11:01:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98b3a21aee5c4c0eb7fe4cc941003b51 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac7806c838644a1db5db11dee3cf0a6b : SUCCESS in 4m 15s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/80b53f28e3884bfeb49a3acc4095d2c5 : SUCCESS in 50m 28s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/9576038eaf6a42adbf86e481a71dea0a : SUCCESS in 1h 24m 44s\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/0b55c5611db84b9a9eeea3539515b982 : SUCCESS in 1h 25m 39s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/eb08d1825caa46c7859fe0a619f759d4 : SUCCESS in 1h 17m 28s","accounts_in_message":[],"_revision_number":1},{"id":"565275414b703f17fdbd9d7b75364e5661b1ab51","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 11:01:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4bbd30da50ddbd59fdffa37a1d12190dfa990b8a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 11:03:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c0c471041704e63b0e959e98d4b6677 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b47422f56ec2f15d144a054d1bc1412b27ee7eb8","revisions":{"b47422f56ec2f15d144a054d1bc1412b27ee7eb8":{"kind":"REWORK","_number":1,"created":"2021-04-29 23:32:25.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/39/788839/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/39/788839/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/39/788839/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/39/788839/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/39/788839/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/39/788839/1"}}},"commit":{"parents":[{"commit":"bb32e5064472facc2156b99094d9f2b92ca67e32","subject":"Adds tests for making listener timeouts configurable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/bb32e5064472facc2156b99094d9f2b92ca67e32"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-04-29 23:31:15.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-04-29 23:32:23.000000000","tz":-300},"subject":"Add stable/wallaby jobs on master gate","message":"Add stable/wallaby jobs on master gate\n\nWe have stable/wallaby released so we should add\ntheir job on master gate to keep branchless tempest\nplugins compatible to stable branch.\n\nThis also removes the stable/stein job as that is in EM\nstate now.\n\nRef: Tempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\nChange-Id: Icc37ad3818d40f17b51827f8b0d19ca0c7c5bade\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/b47422f56ec2f15d144a054d1bc1412b27ee7eb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/b47422f56ec2f15d144a054d1bc1412b27ee7eb8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
