)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~788977","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~I45e7c6b9a114283cf469472cbe867b343fe4c738","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"test-hairpin","hashtags":[],"change_id":"I45e7c6b9a114283cf469472cbe867b343fe4c738","subject":"Add service hairpin traffic test","status":"MERGED","created":"2021-04-30 15:29:53.000000000","updated":"2021-05-13 08:58:04.000000000","submitted":"2021-05-13 08:56:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"788977-test-hairpin","meta_rev_id":"15dcfe661fa451e92b8d8d86f40be3ba87de0453","_number":788977,"virtual_id_number":788977,"owner":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-13 08:56:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"all":[{"value":2,"date":"2021-05-12 10:10:25.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":2,"date":"2021-05-13 07:22:01.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"all":[{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":1,"date":"2021-05-13 07:22:01.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-30 18:00:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-04 07:04:03.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"CC"},{"updated":"2021-05-05 07:07:07.000000000","updated_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"},{"updated":"2021-05-07 07:10:29.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2021-05-11 10:29:46.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"}],"messages":[{"id":"78fda82e5f357914f0dc90f4533f53e17d0c1188","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-04-30 15:29:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab3bdab640a6ccf35f2a9a6e0ef77b2b993205e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 18:00:53.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f60928d9ace40f89e630414ec69e0db : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c11f9bd981043439f1cb97b1bff3e0c : FAILURE in 3m 32s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/d111385979a647cd98b41b15780ea6d6 : SUCCESS in 55m 37s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/4586cf2bb2774b4caee6d16cbd61593a : SUCCESS in 1h 21m 51s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/d6298791bf984fe282968d198ef48aa9 : SUCCESS in 1h 18m 14s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/ffb8384b16d54f31b8b57e1080aeb369 : SUCCESS in 1h 36m 31s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/194523a9c15040ecb9d65614ccf646c1 : SUCCESS in 1h 25m 53s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/13df0f78de8543ca9685e0c6b5063979 : SUCCESS in 1h 29m 57s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/2c7ca77a419d46efb0838c30d77c5334 : SUCCESS in 1h 01m 47s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/62a0a8c10b1a417facb46ae550f9b61e : SUCCESS in 1h 23m 28s\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/ec68494948bb49c8bc4ca662ffe5ad58 : FAILURE in 1h 47m 56s\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/d95a4e5c8b534c6e9ee9b14b991359b8 : POST_FAILURE in 2h 28m 42s (non-voting)\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/bff250845c624e9aa601f945aa590023 : SUCCESS in 1h 16m 10s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/3c4c81342e1d4684a32aab40c8e0dfe6 : SUCCESS in 1h 01m 44s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/1917cc17fbff4312bd511f7311840380 : SUCCESS in 1h 18m 06s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/504f120454fa46b6a23ca354e1433902 : SUCCESS in 1h 10m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"65e9aee6503f44c5bf3097db9246895e14c2836a","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2021-05-04 07:04:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3612c8fb0feeb4b4ca34ed184d1dcde244ebea87","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2021-05-05 07:07:07.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nBesides docstring Luis mentioned, it looks okay-ish.","accounts_in_message":[],"_revision_number":1},{"id":"5f5ca6d210b6ca315bd46ec13c4b4bd61e46f3bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-05-05 14:23:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"34a77db350a008160ca965d7750c9d2365a58411","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-05-05 14:27:12.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"6bef03d9e41aaf7680ecdfdf17781d6d41abee31","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-05-05 14:27:42.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nLet\u0027s try it with Depends-On on the fix.","accounts_in_message":[],"_revision_number":3},{"id":"35a1244284522a9e73952423bf0c2941b7c663dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 14:36:59.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"16cab267369cd2ea5462fed70745236e559df705","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-05-05 14:38:36.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"42560212871257883c849e1e3c93e9d4043f021c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 17:05:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d452b75aa604affb8f86978e4f6f5ed : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/679ef25ac7a645868be6511ae4d12cf4 : SUCCESS in 3m 55s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/5d985ade867d4179851c6f1d0883819e : SUCCESS in 1h 05m 55s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/4a39912495814052bddd4c9d73146c31 : SUCCESS in 1h 26m 55s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/eef02b0c73e2494eacbdb606b14ec3aa : SUCCESS in 1h 19m 24s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/64e29d3199c240eb826a619bc282af0d : SUCCESS in 1h 23m 43s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/cdfa2f04d45540dd888ed3ec6c43b483 : SUCCESS in 1h 31m 46s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/9ee27263848d4c37827aad6c6932786a : SUCCESS in 1h 26m 30s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/9f48e1e2eab14221b35f1779ecedd3f6 : SUCCESS in 58m 27s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/a147ed1b0348492cba5c14c5f934d50e : SUCCESS in 1h 04m 36s\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/0bfe3d4c8377438ab0ed0d6c903f56f7 : SUCCESS in 1h 23m 22s\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/09d2f37f57c843e68f0607f962f4e9fc : SUCCESS in 2h 24m 51s (non-voting)\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/da9fabfdc131487cb2b6796d97f8dfb5 : SUCCESS in 1h 16m 53s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/d174464a298d45cfbca9143d1513b35b : SUCCESS in 55m 59s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/1e16a08390db43dba977ae050edb4c56 : SUCCESS in 58m 25s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/fa1527a581104385ba3672f381c0eebc : SUCCESS in 1h 25m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c22f0b1a21411bd2efe48e5ab67b2a4fe23ff972","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2021-05-07 07:10:29.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a0c03d14a5c489cfcae5f70c06610ac3483ec913","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-05-07 11:33:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7b2286e8c093d06a56e90bede70dcd4863186caf","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-05-07 11:41:22.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3d9fc8c193e0b9991a4c5cb8d982a25b109862d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 14:19:18.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cfc6b6e07d844ccafcb2aa4c828871a : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/443935f1f03446e592eca465b17c63da : SUCCESS in 3m 25s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/4fc5d0d5c2634494b999aa692ac06d44 : SUCCESS in 1h 25m 07s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/d73ff95beb4740399992fae51d4b0918 : SUCCESS in 1h 10m 40s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/89f10b14601e4aad81a08e90c63ab9fb : SUCCESS in 1h 30m 07s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/d471e27124e74390b95a90e116f0be01 : FAILURE in 1h 58m 08s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/53de4b48dcc641b58fa4091d434cb28a : SUCCESS in 1h 33m 48s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/45ec14eaab344e85897a6e80995fdfa4 : SUCCESS in 1h 57m 58s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/6f42a6e6ea854ddc89c80c6e6372cdbd : SUCCESS in 50m 44s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/d84e3ef49fc14d50b6cb762937be01cd : SUCCESS in 1h 10m 40s\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/85525dc7e82145f08c6f3e0561433280 : SUCCESS in 1h 16m 29s\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/d4c55565e05d448b9cfbecf9dfc004a9 : POST_FAILURE in 2h 44m 14s (non-voting)\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/48eb22c1113f4e00be5c350e9d24d294 : SUCCESS in 1h 00m 53s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/99665e5bd6e0456d8287dd60af369715 : SUCCESS in 53m 53s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/d4513557b47647838235275bcd9399f4 : SUCCESS in 1h 18m 55s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/f981329a66a64e648c3177ea41eebfff : SUCCESS in 1h 03m 12s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9e3b7307525e024fc4017383e237b852dc3bb3c1","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-05-07 14:31:33.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"f3ac68e73a8eceeae2890de6c470dfe1c3defbfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 17:03:26.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b29960fe220470c9bcb1472e3867fc5 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a42775af7c174a519d54f198dc5df574 : SUCCESS in 7m 19s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/d06716b0de9346dca195169ae136a55d : SUCCESS in 1h 09m 59s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/513dec66f4a14a6dbe5483b3a8ec90de : SUCCESS in 1h 04m 57s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/bd289ee55f4d4a67bf95e8de5d163406 : SUCCESS in 1h 02m 40s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/0ce9c17c71b1488f8cc2d25cfe72948c : SUCCESS in 1h 16m 38s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/b3466e9e0ce54aaf8be5d0581365ffdc : SUCCESS in 1h 39m 31s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/b30386ad423c44339d0b0678b7ca98b8 : SUCCESS in 1h 24m 40s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/d040e4230064416192f77a5f9ac88746 : SUCCESS in 56m 52s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/a740302c788646df98167909762e0918 : SUCCESS in 1h 06m 04s\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/a2d581992ed445eba327fc9751bb253c : FAILURE in 1h 23m 37s\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/74f3b60706f74360ab4bc98b39b4cb4d : SUCCESS in 2h 25m 36s (non-voting)\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/7a72b8b1bec041a198e6fa84f77b7975 : SUCCESS in 57m 50s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/6548b61c777048c799c822947af54a73 : SUCCESS in 56m 02s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/08b58bac0e594440958cdcff5686e640 : SUCCESS in 1h 03m 43s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/693c4f21e9644ae19948ad9c047e4b78 : SUCCESS in 1h 19m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"90d9f465c13e9d38268381a8c3dfdb62a84821d0","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-05-10 11:12:23.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"e95509ae9c4e5bf44456e1c20caa139690e75f99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 13:56:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba4c7978ce3c43549ce9b8c9af7189fc : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6df5c3a73034f278ffbeec34d296663 : SUCCESS in 4m 12s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/3e9994e6eb4849feb05eb63687bb05fc : SUCCESS in 1h 02m 26s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/b3985c60607641bb82db2c472ee76cf5 : SUCCESS in 1h 17m 08s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/35ce192c5d9a4e1fa36c7b072d64f861 : SUCCESS in 1h 22m 06s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/19359e5f62584429a0daef99d46983b5 : SUCCESS in 1h 23m 07s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/648c9eb06299420ba534456ed0a910a7 : SUCCESS in 1h 29m 01s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/7eb0917a96524d00ab5506de84ce15f0 : SUCCESS in 1h 25m 08s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/501243755d954a2faba3caccf6bc2e41 : SUCCESS in 1h 01m 39s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/88009afd4c7541f3b33f38a6c3255d18 : SUCCESS in 1h 03m 09s\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/25cffdcdcbb643d98e76be32fd54211c : SUCCESS in 1h 19m 12s\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/d0a301920dc04706968e0a672f236e50 : POST_FAILURE in 2h 43m 03s (non-voting)\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/915babef68054554bf9196da85f0ff48 : SUCCESS in 1h 05m 11s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/9963ab5e440740f8bd68fc71da6d7bfb : SUCCESS in 1h 21m 57s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/ff58e1cebf784e72b60b63e97afa2c9a : SUCCESS in 58m 49s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/0a19ab43f5d3470ca754418616105fa9 : SUCCESS in 1h 05m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b1b6c287631869f8c94dbd98f13791369ecebbb6","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2021-05-11 07:30:19.000000000","message":"Patch Set 4: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"88fa083e81bd263e9cd4f6660c1d312202b334a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 09:56:29.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f305c08fbf7f4b20978811ad5ce64b88 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f55d88b48d8c4c4fa2fe68875e9adb33 : SUCCESS in 3m 25s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/8d967b9549a444fa9ad44aa0140d8255 : SUCCESS in 1h 02m 58s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/ff3fcb4c50c04ef3b543782815f6ff08 : SUCCESS in 1h 15m 06s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/2e7314c5e7ec4cb2b34792245b18e224 : SUCCESS in 1h 18m 09s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/6fbcad989bbc460fbd56499f02a8e72a : SUCCESS in 1h 21m 53s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/9bdfc013c7c147a18791d0c289dbfa36 : SUCCESS in 1h 49m 41s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/cccdcb7dbbab484ba7fb9afe1e75f1cd : SUCCESS in 1h 41m 32s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/cba796f3141a4cfba39479a7c63f9328 : SUCCESS in 1h 07m 59s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/b69f3ce27a6a4fa0927d4319673e247d : SUCCESS in 52m 19s\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/0ba53b827f0848e488465be154a3fff5 : SUCCESS in 1h 24m 26s\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/90db1653b6aa4f8d8856750cef5bbe52 : SUCCESS in 2h 24m 24s (non-voting)\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/6d8d653906fc4167bcc6b4188481501c : SUCCESS in 1h 07m 51s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/ed495f1e576a49acb3f6f9dd930bafcc : SUCCESS in 1h 07m 31s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/287a71254f8c4ad0946881bc52509f86 : SUCCESS in 1h 11m 15s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/7d9af9d2e8a9410c84544429e440a94b : SUCCESS in 1h 28m 25s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ff177087e622f6ac9c846fd6e7b09c9330ba738b","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2021-05-11 10:29:46.000000000","message":"Patch Set 4: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"371c65d1e97395f80d9a34c1b48d83235c2b8e19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-05-11 12:12:35.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"bae88d405499b4c31aaa431e0903cbc3da58b347","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 14:32:12.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d0f8ccf76ff448297c966ee743048cb : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93a5babb3f2e455f825d6c9027c5805a : SUCCESS in 3m 49s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/510be005c41045b6b5e5590b931f08ac : SUCCESS in 1h 02m 18s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/f889bbfd6bc3461d82aa111d065a40f8 : SUCCESS in 1h 04m 30s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/b34c7dd5a533462681c1776aaefdcec6 : SUCCESS in 1h 16m 44s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/6419296589b24c339e46574d91672709 : SUCCESS in 1h 20m 17s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/8cfb40ccc3264792b2d2e34ed1234ddc : SUCCESS in 1h 25m 46s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/3d546aaaac994a88a1958cab9ec514af : SUCCESS in 1h 29m 33s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/b80ceb0653964b09a387958ec03cc831 : SUCCESS in 1h 04m 51s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/576694b7761745fba85f61bcb334f955 : SUCCESS in 1h 11m 35s\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/4bec82133a0b433ab4112f38f6f61819 : FAILURE in 1h 23m 45s\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/7330a6e51020471586dcece01b8c227c : POST_FAILURE in 2h 18m 01s (non-voting)\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/292f8fed0b9e4c37aa3b920138a797bb : SUCCESS in 58m 30s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/6c2763d8821245eb820a038f377e3769 : SUCCESS in 49m 37s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/c7f464f1c3b44109aac36dfc055ad003 : SUCCESS in 57m 26s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/9f622f41e35d453392388670a3c66585 : SUCCESS in 1h 18m 24s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ce4d4065687b063a3c1c7f504e267568b0778ec3","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-05-11 14:43:24.000000000","message":"Patch Set 5:\n\n(2 comments)\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"cbf2ad3e420563d841bb31321743b8270f179dd6","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2021-05-12 10:10:25.000000000","message":"Patch Set 5: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"4d1d1a9c0d47fa12782b500ab106949eea292a13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 12:43:45.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a975384688134fa99b60cceb64f984eb : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7779d718e99041ca9a8cc542d812d8b5 : SUCCESS in 3m 25s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/9dded34381dd456e8f94764e8df15dc1 : SUCCESS in 1h 16m 58s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/22a99e78965a4de9a596865e93242e82 : SUCCESS in 1h 17m 40s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/9390d9d6304749b99b605f4efe78f921 : SUCCESS in 1h 11m 36s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/30b6afef7d2745a2b35c23eac3dcd5c8 : POST_FAILURE in 1h 31m 58s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/71fd113afd6f4d01844f21d177580f09 : SUCCESS in 2h 03m 45s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/301fcfd21211420a92f83fb2957b2130 : POST_FAILURE in 1h 50m 36s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/a76ed14de8c64d9ba6150f3783159e7c : SUCCESS in 1h 20m 17s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/bc16dceaa9454b30b9549993d1b53ab1 : SUCCESS in 1h 02m 32s\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/91a988dc5f464367a4601bbfb1ccefe5 : SUCCESS in 1h 17m 04s\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/d5ec4242793d48e9afeb485c13e72166 : POST_FAILURE in 2h 31m 35s (non-voting)\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/4632bad7a66944c3b3f592cadcca2e08 : SUCCESS in 1h 16m 18s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/27a925b11ef44993a4ca5fa9241267ca : SUCCESS in 53m 01s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/21051713c16b4a20b2afea81b8d03e1b : SUCCESS in 1h 00m 29s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/1ae95d4dbd22497b88498188595333d0 : SUCCESS in 1h 21m 35s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2cd9642d32b1b516166bdba39ec30bfb20047eaf","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-05-12 13:04:10.000000000","message":"Patch Set 5:\n\nrecheck\n\nNow it\u0027s Zuul\u0027s fault?","accounts_in_message":[],"_revision_number":5},{"id":"8699f2a6d3797c49e493da08f6f56b0224af13d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 15:42:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7ec390ae8dc46f68350f1f90642d4e7 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26e654cba1d84a6fad2810b4127e76f0 : SUCCESS in 3m 52s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/18cd49f2eadd4e42ba460c3bf1632eb7 : SUCCESS in 58m 22s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/acde6ea6f7d140ac96a391380cc7e980 : SUCCESS in 1h 19m 27s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/7b0d657161df4ff087cd16d6bf8a4765 : SUCCESS in 1h 14m 41s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/23311df09e7044c1be2ea473d15dadfa : SUCCESS in 1h 19m 45s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/81493b7a9027400ba1a3d6bb155cde8d : SUCCESS in 1h 46m 11s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/13ba8129f6524e4784877bf64d00cd37 : SUCCESS in 1h 38m 17s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/f1d4084772c34098a57868d1770df68f : SUCCESS in 52m 26s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/13e674afa2f741e885850ae4b75fc90b : SUCCESS in 1h 13m 18s\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/8e3154826aec4fc5a22a19c5b257366b : SUCCESS in 1h 46m 18s\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/4b8c44182eec442cbbe0cf57019f0931 : SUCCESS in 2h 32m 15s (non-voting)\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/99ef8eae928e4cad9b59496526bb87a4 : SUCCESS in 56m 28s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/dbd688ec53ae44fcb99d5c15031d1243 : SUCCESS in 1h 05m 37s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/658d21cdac8d43b98a2fb79b00150964 : SUCCESS in 1h 06m 02s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/8fa0ebe926fd4a13b33922f3962f984e : SUCCESS in 1h 18m 36s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f19c659e4dd9656764a99abf0673fd100dcd56a0","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2021-05-13 07:22:01.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"8287744afdf1d4b4d941ba1bada8c92994829dfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 07:22:30.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"2f56d64b508473c52ec992f985629bec376f975a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 08:56:43.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed9a1d225d6e4b39ae6be3164bbde80e : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efaf7f92ef8b45bcb495578e9c40a6cf : SUCCESS in 3m 08s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/301986ea22bf471e8155efa9d80ca16c : SUCCESS in 1h 01m 35s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/a06133ad90fb4857b43f8f86b7c7e6fc : SUCCESS in 1h 11m 14s\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/f05f746764fd46349f247af5a8e24f7f : SUCCESS in 1h 21m 07s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/b9520c1f840b48cd831a5a6b499d520e : SUCCESS in 1h 30m 38s","accounts_in_message":[],"_revision_number":5},{"id":"f4eec0d92acdc3adf1f54efe074bb768f59ecb6e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 08:56:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"15dcfe661fa451e92b8d8d86f40be3ba87de0453","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 08:58:04.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a276486efd84390bf694b22ec1b1527 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"e3a65991a781d11ba0c01f57e4d975f980bb6e0a","revisions":{"ccf56fe889955b2f82558b0d5d3510dce0038d20":{"kind":"REWORK","_number":1,"created":"2021-04-30 15:29:53.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/77/788977/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/77/788977/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/77/788977/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/77/788977/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/77/788977/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/77/788977/1"}}},"commit":{"parents":[{"commit":"bb32e5064472facc2156b99094d9f2b92ca67e32","subject":"Adds tests for making listener timeouts configurable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/bb32e5064472facc2156b99094d9f2b92ca67e32"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2021-04-30 10:22:08.000000000","tz":120},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2021-04-30 15:22:42.000000000","tz":120},"subject":"Add service hairpin traffic test","message":"Add service hairpin traffic test\n\nThis adds a simple test checking traffic from service pods, through the\nservice and back to that service. This is a regression test related to\nthe problem with OVN changing the source IP for such traffic.\n\nChange-Id: I45e7c6b9a114283cf469472cbe867b343fe4c738\nRelated-Bug: 1923452\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ccf56fe889955b2f82558b0d5d3510dce0038d20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ccf56fe889955b2f82558b0d5d3510dce0038d20"}]},"branch":"refs/heads/master"},"fa3423428f23c4c4bea1c61e9ea00511a75349c8":{"kind":"REWORK","_number":2,"created":"2021-05-05 14:23:55.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/77/788977/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/77/788977/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/77/788977/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/77/788977/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/77/788977/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/77/788977/2"}}},"commit":{"parents":[{"commit":"bb32e5064472facc2156b99094d9f2b92ca67e32","subject":"Adds tests for making listener timeouts configurable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/bb32e5064472facc2156b99094d9f2b92ca67e32"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2021-04-30 10:22:08.000000000","tz":120},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2021-05-05 14:23:24.000000000","tz":120},"subject":"Add service hairpin traffic test","message":"Add service hairpin traffic test\n\nThis adds a simple test checking traffic from service pods, through the\nservice and back to that service. This is a regression test related to\nthe problem with OVN changing the source IP for such traffic.\n\nChange-Id: I45e7c6b9a114283cf469472cbe867b343fe4c738\nRelated-Bug: 1923452\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/fa3423428f23c4c4bea1c61e9ea00511a75349c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/fa3423428f23c4c4bea1c61e9ea00511a75349c8"}]},"branch":"refs/heads/master"},"4bd61b65cc39befe23616c552ccfbc0f1d68a87e":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-05-05 14:27:12.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/77/788977/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/77/788977/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/77/788977/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/77/788977/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/77/788977/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/77/788977/3"}}},"commit":{"parents":[{"commit":"bb32e5064472facc2156b99094d9f2b92ca67e32","subject":"Adds tests for making listener timeouts configurable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/bb32e5064472facc2156b99094d9f2b92ca67e32"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2021-04-30 10:22:08.000000000","tz":120},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2021-05-05 14:26:44.000000000","tz":120},"subject":"Add service hairpin traffic test","message":"Add service hairpin traffic test\n\nThis adds a simple test checking traffic from service pods, through the\nservice and back to that service. This is a regression test related to\nthe problem with OVN changing the source IP for such traffic.\n\nDepends-On: I7e0458c4071e4a43ab4d158429e05c67cd897a3c\nChange-Id: I45e7c6b9a114283cf469472cbe867b343fe4c738\nRelated-Bug: 1923452\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/4bd61b65cc39befe23616c552ccfbc0f1d68a87e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/4bd61b65cc39befe23616c552ccfbc0f1d68a87e"}]},"branch":"refs/heads/master"},"2d2daf7814e43c7211a603240e2667e1404cdb8d":{"kind":"REWORK","_number":4,"created":"2021-05-07 11:33:51.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/77/788977/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/77/788977/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/77/788977/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/77/788977/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/77/788977/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/77/788977/4"}}},"commit":{"parents":[{"commit":"bb32e5064472facc2156b99094d9f2b92ca67e32","subject":"Adds tests for making listener timeouts configurable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/bb32e5064472facc2156b99094d9f2b92ca67e32"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2021-04-30 10:22:08.000000000","tz":120},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2021-05-07 11:33:42.000000000","tz":120},"subject":"Add service hairpin traffic test","message":"Add service hairpin traffic test\n\nThis adds a simple test checking traffic from service pods, through the\nservice and back to that service. This is a regression test related to\nthe problem with OVN changing the source IP for such traffic.\n\nDepends-On: I7e0458c4071e4a43ab4d158429e05c67cd897a3c\nChange-Id: I45e7c6b9a114283cf469472cbe867b343fe4c738\nRelated-Bug: 1923452\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2d2daf7814e43c7211a603240e2667e1404cdb8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2d2daf7814e43c7211a603240e2667e1404cdb8d"}]},"branch":"refs/heads/master"},"e3a65991a781d11ba0c01f57e4d975f980bb6e0a":{"kind":"REWORK","_number":5,"created":"2021-05-11 12:12:35.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/77/788977/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/77/788977/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/77/788977/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/77/788977/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/77/788977/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/77/788977/5"}}},"commit":{"parents":[{"commit":"bb32e5064472facc2156b99094d9f2b92ca67e32","subject":"Adds tests for making listener timeouts configurable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/bb32e5064472facc2156b99094d9f2b92ca67e32"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2021-04-30 10:22:08.000000000","tz":120},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2021-05-11 12:12:26.000000000","tz":120},"subject":"Add service hairpin traffic test","message":"Add service hairpin traffic test\n\nThis adds a simple test checking traffic from service pods, through the\nservice and back to that service. This is a regression test related to\nthe problem with OVN changing the source IP for such traffic.\n\nDepends-On: I7e0458c4071e4a43ab4d158429e05c67cd897a3c\nChange-Id: I45e7c6b9a114283cf469472cbe867b343fe4c738\nRelated-Bug: 1923452\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/e3a65991a781d11ba0c01f57e4d975f980bb6e0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/e3a65991a781d11ba0c01f57e4d975f980bb6e0a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
