)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~791387","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~Iccb0174a7f7dd10affcfe2f08bfd0b4ea9ca26a3","project":"openstack/kuryr-tempest-plugin","branch":"master","hashtags":[],"change_id":"Iccb0174a7f7dd10affcfe2f08bfd0b4ea9ca26a3","subject":"setup.cfg: Replace dashes with underscores","status":"MERGED","created":"2021-05-14 07:24:17.000000000","updated":"2021-05-14 11:56:09.000000000","submitted":"2021-05-14 11:54:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791387","meta_rev_id":"83de64feaaee36c3e2a43e73ff426588a6a0ba30","_number":791387,"virtual_id_number":791387,"owner":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-14 11:54:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2021-05-14 07:53:05.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":2,"date":"2021-05-14 07:25:54.000000000","_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":1,"date":"2021-05-14 07:53:05.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-14 07:25:54.000000000","updated_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"},{"updated":"2021-05-14 07:53:05.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2021-05-14 10:07:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6159b2e6431d3d8f778cc21120f8ca23d4d0a518","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"date":"2021-05-14 07:24:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7008a039c1f4210a8966e844968e563ffcd7bb59","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2021-05-14 07:25:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c2ce92ad6f1d837d35709ac017978359b5e3af44","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-05-14 07:53:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"31a8af2f38fe3abcb97352cc9a8f73a909b14827","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 10:07:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d5a8fa60c8514f25a740161e1fc8d1d2 : SUCCESS in 2m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9b1744aa10a40f8be3cdf8cdd7577a5 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd59a461d949408fab4ed9b565d6006e : SUCCESS in 3m 38s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/495dffb7f4534b669bdfcb973987ac6d : SUCCESS in 1h 05m 58s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/b5be166cf2a94bcdb7482e2001915922 : SUCCESS in 1h 00m 41s\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/c9472cbb74494a07b759a3406283c265 : SUCCESS in 1h 23m 48s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/987b8e021ce14a699933f7c22e91aaf2 : SUCCESS in 1h 51m 13s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/500f13bc9e9e475980ab2c6135b557b3 : SUCCESS in 2h 10m 09s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/b154158f0645472b9ee585981bc167dc : SUCCESS in 1h 43m 44s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ipv6 https://zuul.opendev.org/t/openstack/build/d0d940484a8141d980cad78433defb63 : SUCCESS in 1h 15m 59s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/655ae33555ab4411bab75874cd0762e9 : SUCCESS in 1h 35m 51s\n- kuryr-kubernetes-tempest-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/eaf74c47d935473c9676ee0c8a3f11a2 : SUCCESS in 1h 44m 27s\n- kuryr-kubernetes-e2e-np-containerized-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/80d6910fc2b04c37b072091524069c5b : SUCCESS in 2h 42m 46s (non-voting)\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/7945fafaea8c40178dc10a8d65b0c268 : SUCCESS in 1h 22m 23s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/73f8b30de3014e579909ba1197590708 : SUCCESS in 57m 25s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/1b4110b8bbfa4ee99505719781196ce6 : SUCCESS in 1h 33m 01s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/22e3da909be74891a4f40b7fc5a6078a : SUCCESS in 1h 19m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d016c85a7cd9c313cc35c3284f788dfebbc1a399","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 10:08:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1b1b6a704a92b7b404d0244f30e1932ee296f218","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 11:54:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/997347eeffcb452fae23dd5210a2d3d1 : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f3bfb07bdc3492e94d844d51939a2f9 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/647ad496a3a24dae8f24b949c262da9e : SUCCESS in 4m 00s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/ead7c76bbde14b0cac35e85d707e8ac1 : SUCCESS in 1h 15m 37s\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/4f02a6c210744d25b0fc4b45eb6c55b9 : SUCCESS in 1h 26m 48s\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/26b936d1d74a44f4898ed1d996d21908 : SUCCESS in 1h 31m 47s\n- kuryr-kubernetes-tempest-containerized-network-policy https://zuul.opendev.org/t/openstack/build/b3cf123c87714b09b995f260f8894258 : SUCCESS in 1h 43m 08s","accounts_in_message":[],"_revision_number":1},{"id":"9fc0440a0f4ea85a968988f7e9c2649d6e61ebb1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 11:54:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"83de64feaaee36c3e2a43e73ff426588a6a0ba30","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 11:56:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dfbd9ad4d214a60837aba7a55549d0f : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8e5addef560eae211c039f1d0d6b43501299cbce","revisions":{"8e5addef560eae211c039f1d0d6b43501299cbce":{"kind":"REWORK","_number":1,"created":"2021-05-14 07:24:17.000000000","uploader":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"ref":"refs/changes/87/791387/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/87/791387/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/87/791387/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/87/791387/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/87/791387/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/87/791387/1"}}},"commit":{"parents":[{"commit":"6179af3d4ba16ba688989c714d76d2af21fd856a","subject":"Merge \"Add service hairpin traffic test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/6179af3d4ba16ba688989c714d76d2af21fd856a"}]}],"author":{"name":"yangyawei","email":"yangyawei@inspur.com","date":"2021-05-14 07:23:01.000000000","tz":480},"committer":{"name":"yangyawei","email":"yangyawei@inspur.com","date":"2021-05-14 07:24:14.000000000","tz":480},"subject":"setup.cfg: Replace dashes with underscores","message":"setup.cfg: Replace dashes with underscores\n\nSetuptools v54.1.0 introduces a warning that the use of dash-separated\noptions in \u0027setup.cfg\u0027 will not be supported in a future version [1].\nGet ahead of the issue by replacing the dashes with underscores. Without\nthis, we see \u0027UserWarning\u0027 messages like the following on new enough\nversions of setuptools:\n\n  UserWarning: Usage of dash-separated \u0027description-file\u0027 will not be\n  supported in future versions. Please use the underscore name\n  \u0027description_file\u0027 instead\n\n[1] https://github.com/pypa/setuptools/commit/a2e9ae4cb\n\nChange-Id: Iccb0174a7f7dd10affcfe2f08bfd0b4ea9ca26a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8e5addef560eae211c039f1d0d6b43501299cbce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8e5addef560eae211c039f1d0d6b43501299cbce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
