)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~805221","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~I2347e49eaac018c630e9f437efe4db9af2e7fa23","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"service","hashtags":[],"change_id":"I2347e49eaac018c630e9f437efe4db9af2e7fa23","subject":"Check service longevity","status":"MERGED","created":"2021-08-19 13:31:16.000000000","updated":"2021-09-01 12:18:57.000000000","submitted":"2021-09-01 12:17:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"805221-service","meta_rev_id":"2a3a351a1f4a060deeda20de4b9c15cf54e37f25","_number":805221,"virtual_id_number":805221,"owner":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-01 12:17:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2021-08-27 16:24:14.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":2,"date":"2021-09-01 10:25:14.000000000","_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-08-31 06:43:25.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":1,"date":"2021-09-01 10:25:14.000000000","_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}],"reviewers":{"REVIEWER":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-19 14:02:21.000000000","updated_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"CC"},{"updated":"2021-08-19 16:10:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-26 06:45:24.000000000","updated_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"},{"updated":"2021-08-26 06:49:03.000000000","updated_by":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2021-08-26 06:49:03.000000000","updated_by":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"},{"updated":"2021-08-31 06:43:25.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"}],"messages":[{"id":"04886e599add690488c5bed146eeea0fdaa95713","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2021-08-19 13:31:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eeacb73cfcf94cc5649e6d12f62a1165183329e0","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2021-08-19 14:02:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8576178baf6d66e49ce5a58953ccc721741c52c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 16:10:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7da511b0ac4a493fbec6e9d959a0eb8c : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4a2c2fbd90a4f2c9510cfbdb0d21f62 : SUCCESS in 4m 37s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/c7e2999f5cc64fc79952b6129cd231fb : SUCCESS in 1h 10m 22s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/94e1ca95b0604871a3d14f4feb9e7daa : SUCCESS in 1h 31m 49s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/230cb8a750aa4f25a6fa8f7d982eb6b8 : FAILURE in 1h 14m 30s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/e36c794e2856451eb679fed403d087f9 : SUCCESS in 1h 24m 45s\n- kuryr-kubernetes-tempest-network-policy https://zuul.opendev.org/t/openstack/build/a563301bc0d948bb82a905ed363c370f : SUCCESS in 1h 29m 51s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/92dfea43fe014b0eab57af3555bd3642 : SUCCESS in 1h 03m 38s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovn https://zuul.opendev.org/t/openstack/build/5fbf65abc73d4459a47f73a66b4bb4aa : SUCCESS in 1h 11m 46s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/4a5d2a426adf4dd9972733ffabb7ca5c : SUCCESS in 1h 09m 51s (non-voting)\n- kuryr-kubernetes-tempest-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/9fb914157f2c4d2c9c473d63941694c1 : SUCCESS in 1h 05m 49s\n- kuryr-kubernetes-tempest-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/b4030626b5db403f8209eadcdc68d331 : FAILURE in 40m 44s\n- kuryr-kubernetes-e2e-np-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/7da51bf99c63438b9b02489450dfb03a : POST_FAILURE in 2h 33m 52s (non-voting)\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/d918c73b6ef04316a9ed415c808c5bec : SUCCESS in 1h 21m 23s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/97d80de96a604ea6bced67808b6c20ae : SUCCESS in 1h 18m 34s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/e7fa66f05db5460c80d353f227c5c3f9 : SUCCESS in 1h 05m 12s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/f89e2ba96a59437cab000fb7ba2eaa1a : SUCCESS in 1h 13m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9632f1e8f3f5717608d630b7a42d93e39953acc5","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2021-08-22 05:55:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"649654c66952992338bbc0c905c3d71edbc2dd49","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2021-08-22 05:56:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"331affdff8b594cfc9c40fca5499c7241a4f31d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-22 08:23:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5322840b3d51494da08be4bdb3f596b4 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb816b72630449ffb533e7ec92b1511a : SUCCESS in 3m 55s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/72b53508cb9940fa97e4166e69b5b3e7 : SUCCESS in 1h 21m 11s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/e5be7a9a13e24f8fbc7d47ffff7b6bcb : SUCCESS in 1h 23m 12s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/66685b615f4641d097c5e0ae1808ed6e : SUCCESS in 1h 20m 24s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/f381c47cd7804a05a4e318d5b19f4597 : FAILURE in 1h 23m 04s\n- kuryr-kubernetes-tempest-network-policy https://zuul.opendev.org/t/openstack/build/d7108a181213470fa4af334466491a2b : SUCCESS in 1h 45m 14s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/084061266d1043b7bb49ba3662bc3256 : SUCCESS in 57m 53s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovn https://zuul.opendev.org/t/openstack/build/22521e1001e545fe9953d6be0a3742f8 : SUCCESS in 1h 00m 44s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/165f43a78cce4fe5b6b3e6c6d9eed9b0 : SUCCESS in 56m 58s (non-voting)\n- kuryr-kubernetes-tempest-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/8b07c809cd8c453f9c542dc1f3c711c4 : SUCCESS in 1h 02m 35s\n- kuryr-kubernetes-tempest-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/979e04939a364a03af30ebe1a7601de7 : SUCCESS in 1h 13m 46s\n- kuryr-kubernetes-e2e-np-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/f780a52cefa84d43ada6538fe9b1c724 : POST_FAILURE in 2h 25m 09s (non-voting)\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/c53741d0122d467da662d3f00b05e594 : SUCCESS in 1h 05m 06s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/db887964fe8343448f3575fd904413b9 : SUCCESS in 53m 33s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/6a67aba738a54a268f6cb867af1856cf : SUCCESS in 1h 17m 51s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/41727fcbe52f4057a7dcc0ef32d973d6 : SUCCESS in 1h 08m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c530c08b857eef212629deea5b476dc70cc0d9b3","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2021-08-26 06:45:24.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d571667d9657d3ff9ad6995d5049c392cf7736c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 09:31:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7ddcf2a156445dbab0cfa59b82360ed : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bd2fc43787c46dca5f0fd356566f3be : SUCCESS in 5m 52s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/ccea078d37874f88818ca733276364e0 : SUCCESS in 1h 16m 15s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/f4283211e05e4c5ca4249e4e0fa8c19c : SUCCESS in 1h 25m 52s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/93601863a5f8499d8dbf24347aee75e6 : SUCCESS in 1h 34m 08s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/5fd668d86adf4c558e2c8d2fb821e64f : SUCCESS in 1h 33m 31s\n- kuryr-kubernetes-tempest-network-policy https://zuul.opendev.org/t/openstack/build/914a1c6e8298478cb8d124177dff6611 : SUCCESS in 1h 46m 07s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/c90c1a7092584f03b0dc3eb24da2e54b : SUCCESS in 1h 11m 48s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovn https://zuul.opendev.org/t/openstack/build/bd414833a1784fa2900f56bed334c0c1 : SUCCESS in 1h 14m 21s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/784d6c3495e1417b979f6c44e3b5fcc9 : SUCCESS in 1h 12m 32s (non-voting)\n- kuryr-kubernetes-tempest-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/b1a2ba3fd6b04441ba6a331034deff7c : SUCCESS in 1h 19m 15s\n- kuryr-kubernetes-tempest-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/a18291a17a094c15a7eae06823bf10e3 : SUCCESS in 1h 37m 41s\n- kuryr-kubernetes-e2e-np-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/3ef4492201754f079c833098e96f0648 : SUCCESS in 2h 40m 03s (non-voting)\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/5398f207acc148b6b61cbe7f8db7b721 : SUCCESS in 1h 08m 59s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/bfc892233ee2411ab4ef2a597f889e20 : SUCCESS in 1h 03m 21s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/adc4b7d6a6634e1d8387109ea44aec80 : SUCCESS in 1h 12m 55s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/a55ed69bbe744aa7aad595c01f92b1ab : SUCCESS in 1h 50m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1fb409031bcbc6af9a5627632756995b95a52c6b","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-08-27 16:24:14.000000000","message":"Patch Set 1: Code-Review+2\n\nI\u0027m not against this, looks good.","accounts_in_message":[],"_revision_number":1},{"id":"bc9c1ccff8f6c9a402dd7368804bdce67bb9ac65","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2021-08-31 06:43:25.000000000","message":"Patch Set 1: Code-Review+2\n\nI suppose this is to catch the problem with OVN loadbalancers loosing the FIPs associated to them, right?","accounts_in_message":[],"_revision_number":1},{"id":"ab8a02a768385cf191060bec99964aba3030e43b","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2021-08-31 06:54:04.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review+2\n\u003e \n\u003e I suppose this is to catch the problem with OVN loadbalancers loosing the FIPs associated to them, right?\n\n@Luis To test BZ1879389","accounts_in_message":[],"_revision_number":1},{"id":"2ff04a07fd62b38d35726bcc9c1685d8011c9087","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2021-08-31 07:18:02.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1: Code-Review+2\n\u003e \u003e \n\u003e \u003e I suppose this is to catch the problem with OVN loadbalancers loosing the FIPs associated to them, right?\n\u003e \n\u003e @Luis To test BZ1879389\n\nYep, the one I meant, thanks!","accounts_in_message":[],"_revision_number":1},{"id":"a6376fba97ce690819bc3d42ebe9c62e6a9c015b","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2021-09-01 10:25:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLooks good, thank you!","accounts_in_message":[],"_revision_number":1},{"id":"d6c6f208c4f972cd011d2c9024129ff454cdda55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 10:25:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f81ff81946f3c12c46018a5669d6de34d9c333cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 12:17:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56468d3866ae4c559849bed00e058dc4 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bf4e1fb82c14a01b9f755fd72f38e4d : SUCCESS in 5m 13s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/2b41c9c6f01c4d4da80df3724be90bf5 : SUCCESS in 1h 08m 19s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/ccfec5832a414cf5bf6a5c2b421dd9ce : SUCCESS in 1h 19m 45s\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/c73525feb7af4ca7b1d0907bd4a7e63b : SUCCESS in 1h 37m 50s\n- kuryr-kubernetes-tempest-network-policy https://zuul.opendev.org/t/openstack/build/c77a3975a57145ef9444ddcf9f1e0a1a : SUCCESS in 1h 51m 02s","accounts_in_message":[],"_revision_number":1},{"id":"616fe65b1c42e56172c20ab1b53c78fbd0954300","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 12:17:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2a3a351a1f4a060deeda20de4b9c15cf54e37f25","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 12:18:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c94c1042bd00470bb81d1ff19324149d : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"02778b3625e3b7ba96ea9612c7602f251c9abd4d","revisions":{"02778b3625e3b7ba96ea9612c7602f251c9abd4d":{"kind":"REWORK","_number":1,"created":"2021-08-19 13:31:16.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/21/805221/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/21/805221/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/805221/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/805221/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/805221/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/21/805221/1"}}},"commit":{"parents":[{"commit":"8e5addef560eae211c039f1d0d6b43501299cbce","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8e5addef560eae211c039f1d0d6b43501299cbce"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2021-08-19 13:29:33.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2021-08-19 13:30:02.000000000","tz":180},"subject":"Check service longevity","message":"Check service longevity\n\nCheck a service connectivity for couple of minutes\n\nChange-Id: I2347e49eaac018c630e9f437efe4db9af2e7fa23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/02778b3625e3b7ba96ea9612c7602f251c9abd4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/02778b3625e3b7ba96ea9612c7602f251c9abd4d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
