)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~813404","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~Id0879f128ca8c057ea6ec540fec0e76d656e7b13","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"fix-ns-handling-tempest","attention_set":{},"removed_from_attention_set":{"4727":{"account":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"last_update":"2021-11-05 08:41:32.000000000","reason":"Change was submitted"},"27032":{"account":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"last_update":"2021-11-05 08:41:32.000000000","reason":"Change was submitted"},"13692":{"account":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"last_update":"2021-11-05 06:42:11.000000000","reason":"Roman Dobosz replied on the change"}},"hashtags":[],"change_id":"Id0879f128ca8c057ea6ec540fec0e76d656e7b13","subject":"Fix Namespace handling","status":"MERGED","created":"2021-10-11 10:12:23.000000000","updated":"2021-11-05 08:43:06.000000000","submitted":"2021-11-05 08:41:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"813404-fix-ns-handling-tempest","meta_rev_id":"c4b571dc92bf7aac4f89fec47574827f6ec6dd56","_number":813404,"virtual_id_number":813404,"owner":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-05 08:41:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2021-10-21 15:29:46.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-11-05 06:42:11.000000000","_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-11-05 06:42:11.000000000","_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"}],"reviewers":{"REVIEWER":[{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-11 12:53:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-11 13:45:16.000000000","updated_by":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"reviewer":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"state":"REVIEWER"},{"updated":"2021-10-21 15:29:46.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2021-11-03 23:56:30.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"}],"messages":[{"id":"3b291995fd769a62491b289103aee453b408087a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2021-10-11 10:12:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e73cfc6d8c4c35aba64b71b608cbe62ae4b7093","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2021-10-11 10:14:40.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"46f69531f18713c390a48ad042486894b17c577b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 12:53:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d05aed411224779a0e1559488159e38 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5669908b225f4119a20e9d95396eb271 : SUCCESS in 3m 42s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/d81cd600b8a447a797ca883c17135fcd : SUCCESS in 1h 23m 53s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/1e34498f53f8430782c434efc0608fb9 : FAILURE in 2h 06m 27s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/f945c11cdcd8416d9a1913bca207e9a8 : SUCCESS in 1h 20m 50s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/d758ec260dbd49209b4d2d2b408c7a1d : SUCCESS in 1h 39m 20s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/d3f6e0518ced4bc79fe500e0832bb8e5 : SUCCESS in 1h 25m 33s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/38dcc89e9b534012827c738ce779ed2a : SUCCESS in 1h 20m 23s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/485e39759807486d9ee1d1e75905949f : SUCCESS in 1h 55m 30s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/a215ec4d71f14b46a03569772ae0ca8a : SUCCESS in 1h 59m 36s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/2a265265ef864631ac6ac6e063ca8c2d : SUCCESS in 1h 55m 15s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/585d74d00dcf45968c547afd9384589f : SUCCESS in 2h 03m 44s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/14ae8ceff60f4607afe3fc4931b61267 : SUCCESS in 2h 37m 02s (non-voting)\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/25da60357ea74ef8bdd63a9dbca75ed0 : SUCCESS in 1h 12m 45s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/482b3a9d820442dda66ddecda59efc80 : SUCCESS in 1h 41m 07s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/b234411ce2064194af822901fdd94f54 : SUCCESS in 1h 06m 57s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/4faed172b0c44762b269e96d7d79c2a8 : FAILURE in 1h 24m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"779395ec445fb9e248c276853f5c0457f1ff2f94","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2021-10-12 17:28:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e91928fbb414969d9fc990b63dcd909cf1feb473","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 20:02:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c0b22e69a78463690a204eadf6bd07c : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/419531ede11348d092a5b8e1615376ef : SUCCESS in 3m 41s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/cb632f52a0df43ccb6f20f621e1f0ecb : SUCCESS in 1h 36m 45s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/091c290746e14cef875d26a10e376bc8 : FAILURE in 1h 25m 32s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/7501d07e170e454abaa9ca4954f3c6b9 : SUCCESS in 1h 32m 49s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/310e12848ca8496089f9bb92707e4b93 : SUCCESS in 1h 25m 28s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/3c18c157766b43529e17de10fc478682 : SUCCESS in 1h 14m 51s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/44ee7ec237a64443a80718f07c973f49 : FAILURE in 31m 06s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/6e6af0aee4e9479aa862ee61a664068a : FAILURE in 1h 24m 39s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/c4620836b8b24ad6941e50aaa84395d9 : SUCCESS in 2h 06m 35s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/c0e04b40b9354ec89e30f5cf6440d1dc : SUCCESS in 2h 04m 45s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/58d70c5093954b2499adb66eb788217f : FAILURE in 1h 44m 17s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/c71e733e09db436fbd360edd0135f36f : POST_FAILURE in 2h 31m 49s (non-voting)\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/c6d40d39c3de424eb8767914b6874b93 : SUCCESS in 1h 22m 58s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/9ff9992f644e49368528c26bf22cacbe : SUCCESS in 1h 09m 50s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/b3965f6968134a7c83f921f749125373 : SUCCESS in 1h 12m 58s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/a73137fb9c1343818998306942ea2ee3 : FAILURE in 1h 15m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"97a94c73a56a69f00ac6a0e0e8a59c1b8e4dfbd0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2021-10-14 13:02:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fedac7113b5d38f66f5a952d602d0bb7618b4625","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2021-10-14 13:27:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"01870ba3c5da8a91367d8c760b61ad92b5d3b44c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 16:05:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/638de5cdff604efa9bce30386c3c70cb : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f5de6d3c17949029130475e03b13e10 : SUCCESS in 4m 57s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/ec0c12e658f8479ba293c52920ef911a : SUCCESS in 1h 51m 24s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/b422c8b7139f422ca6f84376999401d4 : FAILURE in 2h 11m 54s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/a90dda0b268840fc96cd37d59a4e7ae8 : FAILURE in 1h 35m 00s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/78d3eefce2474b8fb22860d61a50b444 : SUCCESS in 1h 26m 41s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/89df833746f24eebbe8da6bc0670374e : SUCCESS in 1h 23m 26s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/6110bf919a114ef8985837f5885699fd : SUCCESS in 1h 00m 23s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/67a9df0a02ed47258cdf6856d77c2e6e : SUCCESS in 1h 51m 14s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/11be93ab15d0459381708f26ede3feaa : SUCCESS in 1h 31m 43s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/4ee7b85ac0ea48259f5811de5ac7e8cc : FAILURE in 2h 15m 45s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/4a70d43db5dc4d92aeb9db9eb20d75ca : FAILURE in 2h 01m 17s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/8d85e29e140a4857815194a16948a85e : SUCCESS in 2h 35m 01s (non-voting)\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/5449b44bb1ba4c5eb6f7758f818900e2 : SUCCESS in 1h 19m 05s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/f176e55644bf4bd88cfa0ae314bcf24c : SUCCESS in 1h 10m 22s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/aa67a8f937564ed88bd06532e805abd4 : SUCCESS in 1h 14m 06s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/0e6d9435f8684c74b5d39478438e48e7 : RETRY_LIMIT in 5m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ed776d809fd98ae07f3094932b4457b01a2f5e0d","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2021-10-18 10:49:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"861f29eafc3bbb53f5fe4bcae37243b8d1286045","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 13:57:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5daf7eda7cc24265a51c76555685a6e8 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bb1760c27c84ff285f59f5c52248628 : SUCCESS in 3m 54s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/3c64af0397ca471699c3bbf9bd71410c : SUCCESS in 1h 41m 17s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/17c9fa279665410aa3743de500068606 : SUCCESS in 2h 11m 22s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/745640faf4fa495cab32864b38baa390 : SUCCESS in 1h 43m 24s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/9485102478134c4ba7ead8f43b7394ad : SUCCESS in 1h 40m 39s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/637f3f3e69354a69964a6eccdc8466d2 : SUCCESS in 1h 33m 28s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/4d1cb5586101418e880b321d9b864c2c : SUCCESS in 1h 15m 13s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/a0b479e838a44865afae31236f33754e : SUCCESS in 1h 34m 23s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/d4bcf638563d44d6aae92a742e9b2dd8 : SUCCESS in 1h 50m 14s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/1d72daa91e6144899140465eae32e02e : SUCCESS in 2h 09m 00s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/bcd7ba7dbf774f30ad3f9c46642c0640 : SUCCESS in 2h 08m 12s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/eb10b476d60b44ddabe39b01ecdbba52 : POST_FAILURE in 2h 55m 28s (non-voting)\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/ab1947d9881d4dc08424b67cdc307783 : SUCCESS in 1h 38m 36s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/a74ebb09712a495db891a9e7d1c2aa38 : SUCCESS in 1h 04m 19s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/3edd818f70e74c3eb7e03a2b53ceeef4 : SUCCESS in 54m 55s\n- kuryr-kubernetes-tempest-train https://zuul.opendev.org/t/openstack/build/fa82abb71b1049f3bfe53c3087803a6b : FAILURE in 1h 22m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e87ee6c2475ec6d758f0e36ec77ed973e5b7f513","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-10-21 15:29:46.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a33186fca2423b683fede684a5a7001622483230","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2021-10-26 08:28:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ef14d4bfecb18b3e71a159c060d5b408b05f828c","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2021-11-03 23:56:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f5f14078faeff296c646be7d3ee3db61eb3d2a03","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2021-11-05 06:42:11.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"637ecd1cf42b0a0950c5022e92dd6caddf925f4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 06:42:57.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b35184b0b1ae09ece75c92a465a2d76be51e33b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 08:41:32.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54a60521d7204eae8cac8f95c0a63f52 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cb20357ebeb4d488aaa960079308588 : SUCCESS in 3m 36s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/a38d3c7103ef4accb4f8a9cd5d047422 : SUCCESS in 1h 54m 18s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/eacc07e8d36740cb90b9778b6818c790 : SUCCESS in 1h 43m 36s","accounts_in_message":[],"_revision_number":3},{"id":"a95d9580264a47a6a3ea1d632d2e5d9ae42b88a7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 08:41:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"c4b571dc92bf7aac4f89fec47574827f6ec6dd56","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 08:43:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbfa53f1a25e467a85cd87db37606881 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"774fc6bae0209edb366adbddd16aa2f205d389cb","revisions":{"f0e32460d7df94280f766b24dd9c33e0650a21a5":{"kind":"REWORK","_number":1,"created":"2021-10-11 10:12:23.000000000","uploader":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"ref":"refs/changes/04/813404/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/04/813404/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/04/813404/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/04/813404/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/04/813404/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/04/813404/1"}}},"commit":{"parents":[{"commit":"ddef1710fb33463c2898d4011ed44d6441f55b78","subject":"Add Kuryr-tempest-plugin test for LoadBalancer Reconciliation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ddef1710fb33463c2898d4011ed44d6441f55b78"}]}],"author":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2021-10-11 10:06:35.000000000","tz":0},"committer":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2021-10-11 10:12:01.000000000","tz":0},"subject":"Fix Namespace handling","message":"Fix Namespace handling\n\nWith the handling of Namespaces now being\ndependent on Pods creation, tempests tests\nneed to get updated to account for that.\n\nChange-Id: Id0879f128ca8c057ea6ec540fec0e76d656e7b13\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/f0e32460d7df94280f766b24dd9c33e0650a21a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/f0e32460d7df94280f766b24dd9c33e0650a21a5"}]},"branch":"refs/heads/master"},"b5fa361955e9eba9334cccf91fa1425c32efab63":{"kind":"REWORK","_number":2,"created":"2021-10-14 13:02:14.000000000","uploader":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"ref":"refs/changes/04/813404/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/04/813404/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/04/813404/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/04/813404/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/04/813404/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/04/813404/2"}}},"commit":{"parents":[{"commit":"ddef1710fb33463c2898d4011ed44d6441f55b78","subject":"Add Kuryr-tempest-plugin test for LoadBalancer Reconciliation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ddef1710fb33463c2898d4011ed44d6441f55b78"}]}],"author":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2021-10-11 10:06:35.000000000","tz":0},"committer":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2021-10-14 13:02:51.000000000","tz":0},"subject":"Fix Namespace handling","message":"Fix Namespace handling\n\nWith the handling of Namespaces now being\ndependent on Pods creation, tempests tests\nneed to get updated to account for that.\n\nChange-Id: Id0879f128ca8c057ea6ec540fec0e76d656e7b13\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/b5fa361955e9eba9334cccf91fa1425c32efab63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/b5fa361955e9eba9334cccf91fa1425c32efab63"}]},"branch":"refs/heads/master"},"774fc6bae0209edb366adbddd16aa2f205d389cb":{"kind":"REWORK","_number":3,"created":"2021-10-14 13:27:08.000000000","uploader":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"ref":"refs/changes/04/813404/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/04/813404/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/04/813404/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/04/813404/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/04/813404/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/04/813404/3"}}},"commit":{"parents":[{"commit":"ddef1710fb33463c2898d4011ed44d6441f55b78","subject":"Add Kuryr-tempest-plugin test for LoadBalancer Reconciliation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ddef1710fb33463c2898d4011ed44d6441f55b78"}]}],"author":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2021-10-11 10:06:35.000000000","tz":0},"committer":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2021-10-14 13:25:50.000000000","tz":0},"subject":"Fix Namespace handling","message":"Fix Namespace handling\n\nWith the handling of Namespaces now being\ndependent on Pods creation, tempests tests\nneed to get updated to account for that.\n\nChange-Id: Id0879f128ca8c057ea6ec540fec0e76d656e7b13\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/774fc6bae0209edb366adbddd16aa2f205d389cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/774fc6bae0209edb366adbddd16aa2f205d389cb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
