)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~814025","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~Ie5b54ca4c1213709c47d01fc5986f15a898ee625","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"update-jobs","attention_set":{},"removed_from_attention_set":{"11600":{"account":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"last_update":"2021-10-19 15:46:00.000000000","reason":"Michał Dulko replied on the change"},"27032":{"account":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"last_update":"2021-10-20 08:55:52.000000000","reason":"Change was submitted"},"13692":{"account":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"last_update":"2021-10-20 06:55:29.000000000","reason":"Roman Dobosz replied on the change"}},"hashtags":[],"change_id":"Ie5b54ca4c1213709c47d01fc5986f15a898ee625","subject":"Add xena jobs on master gate","status":"MERGED","created":"2021-10-14 14:59:35.000000000","updated":"2021-10-20 09:01:39.000000000","submitted":"2021-10-20 08:55:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"814025-update-jobs","meta_rev_id":"aec1bf1aeba52a7e828c7a16fe4bbcb1bac772d4","_number":814025,"virtual_id_number":814025,"owner":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-20 08:55:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2021-10-19 15:46:00.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":2,"date":"2021-10-20 06:55:29.000000000","_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":1,"date":"2021-10-20 06:55:29.000000000","_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-14 18:08:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-17 12:45:00.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"},{"updated":"2021-10-17 12:45:00.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"}],"messages":[{"id":"276dc2e34cc4d44917db4409f273d46dfdc6d539","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2021-10-14 14:59:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"28cc7d6d55a9eefa75fd80bfc14fceabebd736ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 18:08:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c995457430c84d2b90be895038dbc2af : SUCCESS in 8m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0058d2a2c6a24f86aaff7184331fd393 : SUCCESS in 4m 12s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/ce2a8598cab747709a007918547d9151 : SUCCESS in 1h 30m 07s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/754f3f3b7b7d45fcb5b332bb3d1b2965 : FAILURE in 2h 01m 26s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/f97665e83c6e4b8294a123788862230c : SUCCESS in 1h 40m 45s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/4d4b3122c5034bb49e444d35de408b12 : SUCCESS in 1h 44m 50s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/19f5284fe8874882b08119bbca961bdb : SUCCESS in 1h 26m 38s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/df554db9f0e14dbe841ee6a7562b327a : SUCCESS in 1h 19m 02s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/537222c2aaf84d4e9edaf1ddcc87d9e7 : FAILURE in 1h 39m 09s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/96725f8a328649b58d3ae4901fee461c : FAILURE in 47m 15s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/cecec6b11e554b338cb37e09b755473d : FAILURE in 2h 35m 58s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/8f5e1ab380fe4100b6dbdce7b9f47f97 : FAILURE in 2h 07m 23s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/ec33e15724aa4021ac14fd0934690ab3 : POST_FAILURE in 3h 02m 47s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/49cecfbcd8ea4df69e20af1e9c699503 : SUCCESS in 1h 36m 40s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/16a4178384e04453931ef522a2a92702 : SUCCESS in 1h 30m 14s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/4f6c2a6368fa43d8ac488dc801c9177b : SUCCESS in 1h 23m 47s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/7b3aab4db4de4b7dae940ec30353dd5a : SUCCESS in 1h 35m 06s","accounts_in_message":[],"_revision_number":1},{"id":"df33a0dc5e05770e69dd8c86390580ea2d27a5c0","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2021-10-17 12:44:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7f96ad0172d406c5acb515d0045ce4c397eb3945","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-17 15:36:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6015a59a98044e23bbee742220b3d56d : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f2c5fb828b9413aae59e4f1b2338aa5 : SUCCESS in 3m 43s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/954c600f3f1841a4bd9ccb40460007df : SUCCESS in 2h 04m 50s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/e829d2eadcfd4131946cbdce5bb35512 : SUCCESS in 2h 39m 19s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/85217525eb3a4f3483358fb34efd2635 : SUCCESS in 2h 13m 41s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/5e42504193004de1bdcd4a0d9414ee64 : SUCCESS in 1h 48m 27s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/01f920d559674c1c93c7521927147c0d : SUCCESS in 1h 00m 38s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/819b70cd37b44d3c98c36c010d188f83 : SUCCESS in 1h 20m 07s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/a236a9c5d03f458cb2d040ee69afb96b : SUCCESS in 2h 10m 37s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/eeb3491f48974ccf88b0cddafcc86451 : SUCCESS in 2h 08m 45s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/ae497f5fbf914717a338c74ce0dcfa86 : FAILURE in 2h 05m 33s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/1eb6fb88aa1f496289f6653c0e7c28be : SUCCESS in 1h 55m 00s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/97f465c131b84c42afa38b71e40cb140 : SUCCESS in 2h 47m 26s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/afbf066f0ddc4d2e8fb3bba68dd8a1ce : SUCCESS in 2h 18m 52s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/5c7ebbcf43a946a48b27038a62457b00 : SUCCESS in 1h 32m 04s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/d484839623494a79a3d0c8e05701837f : SUCCESS in 1h 39m 45s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/77bf15aa607c47958dbf487033479b74 : SUCCESS in 48m 56s","accounts_in_message":[],"_revision_number":1},{"id":"a9c52f85e0c6f43c9491077432645eb368c52048","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-10-19 15:46:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"98bce270c4dd9cf1c8154896c0a83115eb5ad12e","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2021-10-20 06:55:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"04039e418710dd655ceb359efec9e06a627d1109","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 06:56:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bf35c3a148f835a1e6ad4f3eda177109a0850020","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 08:55:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f99c726df4046c1bced96e2b61ab15a : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07c47a34933540b9a824a4a138ac8165 : SUCCESS in 5m 33s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/1ad19edbd03f4bb5a75675aa35a00ad9 : SUCCESS in 1h 39m 33s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/4102202989084b97972637cc11613eb5 : SUCCESS in 1h 35m 38s\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/4e2f21dd50a84a9083441b4192167281 : SUCCESS in 1h 58m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"80224b0d4d5f92989ca78ed6a9deb2b8591625bc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 08:55:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"aec1bf1aeba52a7e828c7a16fe4bbcb1bac772d4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 09:01:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58cc3bea02d34c6fb0621cbec1d9fdc3 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"31abfc8b3f141d9e73908a5899e265e62da896c7","revisions":{"31abfc8b3f141d9e73908a5899e265e62da896c7":{"kind":"REWORK","_number":1,"created":"2021-10-14 14:59:35.000000000","uploader":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"ref":"refs/changes/25/814025/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/25/814025/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/25/814025/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/25/814025/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/25/814025/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/25/814025/1"}}},"commit":{"parents":[{"commit":"ddef1710fb33463c2898d4011ed44d6441f55b78","subject":"Add Kuryr-tempest-plugin test for LoadBalancer Reconciliation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ddef1710fb33463c2898d4011ed44d6441f55b78"}]}],"author":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2021-10-14 14:57:12.000000000","tz":0},"committer":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2021-10-14 14:57:12.000000000","tz":0},"subject":"Add xena jobs on master gate","message":"Add xena jobs on master gate\n\nWe have stable/xena released so this commit\nadds a gate for it, it also removes\nTrain jobs which are on EM state.\n\nChange-Id: Ie5b54ca4c1213709c47d01fc5986f15a898ee625\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/31abfc8b3f141d9e73908a5899e265e62da896c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/31abfc8b3f141d9e73908a5899e265e62da896c7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
