)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~814780","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~I5d01029bae9c782ab57d38dff8c6aea1b2128305","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"lb-wo-selector","attention_set":{},"removed_from_attention_set":{"11600":{"account":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"last_update":"2021-11-05 09:40:34.000000000","reason":"Michał Dulko replied on the change"},"4727":{"account":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"last_update":"2021-11-05 16:05:34.000000000","reason":"Change was submitted"},"27032":{"account":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"last_update":"2021-11-05 16:05:34.000000000","reason":"Change was submitted"},"13692":{"account":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"last_update":"2021-11-05 06:42:33.000000000","reason":"Roman Dobosz replied on the change"}},"hashtags":[],"change_id":"I5d01029bae9c782ab57d38dff8c6aea1b2128305","subject":"Update test_service_without_selector test","status":"MERGED","created":"2021-10-20 13:29:10.000000000","updated":"2021-11-05 16:07:17.000000000","submitted":"2021-11-05 16:05:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"814780-lb-wo-selector","meta_rev_id":"c4e05477e9769c0e177c8c084c3ccee1f40d133e","_number":814780,"virtual_id_number":814780,"owner":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-05 16:05:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2021-11-05 14:32:03.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-11-05 06:42:33.000000000","_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":1,"date":"2021-11-05 14:32:03.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-20 13:36:15.000000000","updated_by":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2021-10-20 13:36:15.000000000","updated_by":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"},{"updated":"2021-10-20 13:36:15.000000000","updated_by":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"},{"updated":"2021-10-20 16:31:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"182f6b075ae19c8bf59b3146b4a01e4f73adf10f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2021-10-20 13:29:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c561c8f623164f00bd2bbf0b7ed6711e515d734c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 16:31:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/829cb6a5677a4a4e9ab1ee130e6b5b92 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fd96c6c01544a4bbd4ec6b10af8d52f : SUCCESS in 5m 53s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/f039691da19d407fbbd2f83d16e02203 : SUCCESS in 1h 34m 08s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/1c6f27dc460c4273b3b860a79f1de930 : SUCCESS in 1h 38m 23s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/33139f084ee1400480ffce4057e8ff2d : SUCCESS in 1h 22m 12s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/816ebf2be3a74b58aa28a1a784f4e23c : SUCCESS in 1h 27m 27s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/99e5b7a83fca4b03a959f0db2bfa271e : SUCCESS in 1h 20m 42s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/961fb51b89374e8eafa0d90a205dc05a : FAILURE in 1h 13m 43s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/e9b8c16d462644f3bda1f867c8d6bcc8 : SUCCESS in 1h 59m 19s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/ce05c8282a37495f97916a1206a2fd91 : SUCCESS in 1h 54m 44s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/c049d190c6b949b79dacc3ab76a8fa80 : SUCCESS in 2h 42m 37s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/2c9029a306c74809a7830ded773b5034 : FAILURE in 2h 11m 07s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/8bf1384af4ec4626b177a6b48e1b6bd0 : POST_FAILURE in 2h 55m 45s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/179c44f99a134d80b618ad7e396a1307 : SUCCESS in 1h 34m 35s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/338134af281f4b9e8868d4fafa8dd3fb : FAILURE in 1h 41m 23s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/ea841adb45a74c5ab9a79d7b80c094ca : SUCCESS in 1h 17m 17s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/6ee2a63a8e634e1e985a9d7facc2f409 : SUCCESS in 1h 22m 26s","accounts_in_message":[],"_revision_number":1},{"id":"48149e8eb08afb4ff7fbb8b1ae3bfd440e79fec5","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2021-10-21 09:12:16.000000000","message":"Patch Set 1: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b53648640b452eb4fe0f0e653c71866b8eaef5a8","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2021-10-21 09:12:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d5343b2694bfd6f1eb04c329379298feb3f06bf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 12:09:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20838256c4c441a5b7ad75409e137205 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fd244e4016f40aba3b3b88dee8ed3a3 : SUCCESS in 5m 44s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/02b35c58556e4502a85d44261c80f2ea : SUCCESS in 1h 43m 23s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/40c8f03fb65347188083fd95c1ff73fd : FAILURE in 2h 24m 18s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/5ee037ac524a42ee90e6eed74ac3a151 : FAILURE in 1h 29m 31s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/aec8a920ce444925ac101dde1207c045 : SUCCESS in 1h 52m 32s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/c3ab6c81500e4ab8b272bdee0e64ad56 : SUCCESS in 1h 27m 19s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/388c005fc127405b958d1e71b46d9101 : SUCCESS in 1h 13m 52s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/45d6a0eaa98d4d049e098a7c35889be0 : SUCCESS in 2h 04m 01s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/a599e751bad54707a3ef3647e887c192 : SUCCESS in 2h 02m 20s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/abebb11b684d4b2ab6547cf7a9cc7d1f : FAILURE in 2h 37m 46s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/53bf7605bf094da08b3d0a9548699d9a : FAILURE in 2h 26m 04s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/427a9b26aec3404abffa41f90bba2de4 : POST_FAILURE in 2h 51m 52s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/fca649e7a6404e9a9c21727a44e04f21 : SUCCESS in 1h 49m 02s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/2a783b840fa4429db5c1efc990183d0a : FAILURE in 1h 39m 18s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/5d7bc22305cc40aa9a1969fc11a06d99 : SUCCESS in 1h 26m 28s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/c69b7de760984310867f130fd610dcfa : SUCCESS in 1h 29m 29s","accounts_in_message":[],"_revision_number":1},{"id":"d7465ae8dbe656080fa3eb43adbebef01eae25da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2021-10-21 13:52:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"93176fe8c94d3f4d74ef23c48756ba150cee080c","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-10-21 15:16:06.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fddc3a8b4ddb907d819588b3eb8e6ef5271a7a21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 16:54:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3f861a4a98e413aa24baeecd1d535c5 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72f8ceaef8864a69a94710c143a94abe : SUCCESS in 3m 44s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/ac9826ddcf834b35b50f829572479513 : SUCCESS in 1h 34m 52s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/98c540863fe24c8db6e9f0679754d7f6 : FAILURE in 2h 30m 34s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/39dac885bc3e4f4387331fd34cda1a35 : SUCCESS in 1h 38m 01s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/978ef21196984c2785385087878c562c : SUCCESS in 1h 39m 57s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/86aa2b76a92a4ec5a04283a846c29e2f : SUCCESS in 1h 31m 51s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/b8f2b59a6360469dab2329772da7bf67 : SUCCESS in 1h 13m 34s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/a33638a0f5414e3bbb53fc53cd96e3f6 : SUCCESS in 1h 45m 20s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/521996ebb7ab4d7a9760e06bd8bb7dc3 : SUCCESS in 2h 01m 24s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/b943133b761e4624aaa8e9d3c96c5d27 : FAILURE in 2h 43m 25s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/1080cd42174142db8bf8db8c2f407117 : SUCCESS in 2h 00m 35s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/df01ac02b4bb46d9ae282a035b070b6c : POST_FAILURE in 2h 55m 42s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/66df3a45081b453da08ab2a3dbb504ed : SUCCESS in 1h 48m 12s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/c560673b0c5446569c25225869783308 : FAILURE in 1h 33m 37s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/c9505aab6f5d4bbcb6a565ed196acbf8 : SUCCESS in 1h 13m 33s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/eed36a94b7924e55a5c0637064414f83 : SUCCESS in 1h 21m 46s","accounts_in_message":[],"_revision_number":2},{"id":"800f0d66beafb5b3e91d0aea2df3ecee670eaa4c","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2021-10-22 03:25:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"34771889ccf8ddbac67c1392b3dfc8eef53638ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 06:13:17.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9346cb34c904954895f745216e224dc : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3140138ee49467c9e4cc48fbbc7b390 : SUCCESS in 3m 47s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/848ec903d4f645f7adc4bbceeb054ad1 : SUCCESS in 1h 30m 10s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/487f3ba46fbb49c18ef5c16e8e526312 : SUCCESS in 1h 46m 31s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/28604701ae0040a5b3fbc69019605748 : SUCCESS in 1h 17m 53s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/aa27168f077141eebe52dcbd8ddce95b : SUCCESS in 1h 38m 53s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/2c970d5a91e24d8fa6c1ea711ef85778 : SUCCESS in 1h 14m 22s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/ec201feb338845d0af78fc436e83f6aa : FAILURE in 1h 04m 37s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/eff403ac1d2844108d13850f2bdb2344 : SUCCESS in 1h 35m 05s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/4379ab706f354b6a8fb828c1a4413aa7 : SUCCESS in 1h 56m 43s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/58cf42794108447b86487203b03edaff : FAILURE in 1h 32m 56s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/c9e07ba528fe46fe98601b4c3f9ecf96 : FAILURE in 2h 42m 10s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/2cafc60b50d54d80b6064caa9b528910 : SUCCESS in 2h 26m 51s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/d8b3f6a5e55b4b7e9d2afcc3fc5ceda1 : SUCCESS in 1h 28m 17s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/6d55d354249442898a3db74f9b76a035 : FAILURE in 1h 29m 16s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/2df7d739acb24427bda4acd721de2c70 : SUCCESS in 1h 06m 54s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/e7005ae09cce44e1bb6e249da167f334 : SUCCESS in 1h 07m 29s","accounts_in_message":[],"_revision_number":2},{"id":"177c31a26dfb783e1d6edad23b84d68b05131670","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2021-10-25 09:17:21.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"cfd47d5aa8491bb3e6451e41b92bb8331173df99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 12:12:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8704c8de986449b6ae6642ba9a0bae88 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9f8ffd3f48148c5a9ebde05fdbf935d : SUCCESS in 4m 27s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/8b94847340704aea8761f1e4ebeb5820 : SUCCESS in 1h 43m 24s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/0edda22729634728abca8052a260e76a : FAILURE in 2h 04m 33s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/d6cb3124ab79434f873b5721385cff66 : SUCCESS in 1h 37m 50s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/1f6999665bef468c9eb59bf74a91abb4 : SUCCESS in 1h 49m 22s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/aa9ca97270ce4126b0c7222c296637b0 : SUCCESS in 1h 24m 41s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/fb6e144de30b46f89f001cfc5e561396 : SUCCESS in 1h 14m 55s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/d842fb9b9a004021bc8ab3a3b9e54bce : SUCCESS in 1h 55m 39s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/bb4beb715af144d0b31ddcc6e17ccc6d : FAILURE in 1h 41m 36s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/29d6f61565f74e4899905eb78b6373ad : FAILURE in 2h 51m 48s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/441d50a29adc455992a89acb9212a493 : FAILURE in 2h 26m 42s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/e21c7da906c34311a60d427d5d2e6279 : POST_FAILURE in 2h 30m 00s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/46ae3d434a8c4de1bdb9b703351b6903 : SUCCESS in 1h 30m 09s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/30eb6558ed434ea18979cf7d93ee5995 : SUCCESS in 1h 12m 23s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/0d437b5802684b0ebae1f49b0b7c8173 : SUCCESS in 1h 02m 47s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/3d8f62765b794627907482d311392092 : SUCCESS in 1h 29m 29s","accounts_in_message":[],"_revision_number":3},{"id":"09d92dd91065369b2b223e731a9fabf63600a0e9","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2021-10-25 12:57:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c775102ad16240dfe603a797524b87bdfe4b4407","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 16:05:12.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/621ca8df2bdd4a1eac96fc34dd74889c : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a2d0964075e42c2b12a730fed7adf6e : SUCCESS in 4m 40s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/ba2de81474b54584a8e01ed9739efb6c : SUCCESS in 2h 05m 15s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/f80f9cc3d51f43d6a5e91166d055a40a : FAILURE in 2h 17m 09s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/214ae844450b44f6a346a93f71eecfcc : SUCCESS in 1h 36m 40s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/0cbfc36edf7c482f97f95430293330ae : SUCCESS in 1h 54m 30s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/9c5278f1fb2a4bf8b59bff5708b4a75e : SUCCESS in 1h 23m 34s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/3c32301a4175478496774d5d48a8f627 : SUCCESS in 1h 25m 46s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/6ccc0f3f4990481e9ff72b7597887ad8 : FAILURE in 1h 57m 11s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/050ca70b6510401d8712ac71dc86b73a : FAILURE in 1h 55m 37s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/3dd2bd7b57a342aa85fe2fde0acd139c : FAILURE in 2h 17m 54s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/6617df3d9b2849848b862b1f96fea29e : FAILURE in 2h 44m 57s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/cec87fe67eee42f4b075a8c09f10c9d8 : POST_FAILURE in 3h 02m 58s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/bca3f17f051c4a11a31648a797019ca8 : SUCCESS in 1h 36m 22s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/764a8117842b4a55899c5bb9b943bb94 : SUCCESS in 1h 04m 05s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/bbc8fc8564b648dbb6b9b3e960a5531f : SUCCESS in 1h 16m 52s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/ae801938ec404015b3c3a038e174b8d9 : SUCCESS in 1h 16m 09s","accounts_in_message":[],"_revision_number":3},{"id":"1d1c5dde77bac46046eb3974a6d2c697a357be1c","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2021-10-25 16:44:24.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dc2480acc53bb46a431b967714e26b58865540ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 19:28:48.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4919feec5d5244789d20695d6e5349e6 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/579f21289e1343999e62b9e886279dee : SUCCESS in 4m 47s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/1bbf8f6ea58047fdbad1db871cee2f84 : SUCCESS in 1h 28m 39s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/28c3a0e79aa74c988959124fcfb722ca : FAILURE in 2h 24m 52s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/52a17eaa8c68474caf5b1d7eac190446 : SUCCESS in 1h 21m 33s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/dd979f84680c41e0a35ccaa3f5a66730 : SUCCESS in 1h 50m 56s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/509972da526540c59f23c153ff22fdb3 : SUCCESS in 1h 14m 16s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/9ac5488d75164a8e88694c683b8962b9 : SUCCESS in 1h 25m 33s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/8e0b42da52a74e8dbe2459a70d5b1b4e : FAILURE in 1h 57m 56s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/ebf1cfe98a9b441080be61814f009f32 : SUCCESS in 2h 03m 01s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/1d73864366a44cbd93b289483e50edbe : FAILURE in 2h 41m 09s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/b7bcfa50b1cc44078c7b7d3d7ac370cb : FAILURE in 2h 05m 29s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/9a5321b0796e404f8b37b843b0aa710f : SUCCESS in 2h 25m 07s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/48487bc0807e4a798dc6c851209f1fe9 : SUCCESS in 1h 25m 53s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/4f7099f61b6e4bdabacc16d636391129 : SUCCESS in 1h 10m 25s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/116b87679fdf4c1ab73a735b439ade1a : SUCCESS in 1h 18m 42s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/1a990bc430034469a6142befe4599379 : SUCCESS in 1h 06m 38s","accounts_in_message":[],"_revision_number":3},{"id":"7ae0a3cced1bf92c11e8db9ce4ee8088d7aea5ed","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2021-10-26 08:11:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0842db9fc606cacdb853dc21528163c8d1df117d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 10:59:06.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a170835bc9b46be8f8baa8afa402074 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a03599cd70b4cfaade11fbca88bdb39 : SUCCESS in 4m 26s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/63971e538ad043418404ec17b678ad2b : SUCCESS in 2h 19m 02s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/f4fec2d2083444acbcaae0156f96f1a9 : FAILURE in 2h 11m 45s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/b1da49ebb00447968e73d9f9557a38c6 : SUCCESS in 1h 30m 12s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/58a7d4966be145e49d75100a6c27e3ff : SUCCESS in 1h 39m 11s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/6d8c4f1e163f409c9f50f67a4bf2ca78 : SUCCESS in 1h 24m 47s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/187321a3daa54317a8300106970b9c35 : SUCCESS in 1h 21m 25s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/f295cbf756b14765be62e232ef42b541 : FAILURE in 2h 22m 26s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/ea00085ca6e94f74abd04584674534e7 : FAILURE in 2h 14m 59s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/00bdb4421d254f398e4ed1bd445689c9 : FAILURE in 2h 14m 32s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/1588d8ebd1054d309d6a91898f2a4219 : FAILURE in 44m 13s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/6b5e47a86daf4190a33d258a2f1d1491 : SUCCESS in 2h 45m 06s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/40bd71d53eb04f43bd2cf1949a9ca9e1 : SUCCESS in 1h 47m 40s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/410bdc25908146d3943d94e30717b125 : SUCCESS in 1h 30m 55s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/f1414ff0229c4ddd85878e4df6527ce7 : SUCCESS in 1h 40m 51s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/37218984085f4392ba3bee9e0dd61e2d : SUCCESS in 1h 13m 51s","accounts_in_message":[],"_revision_number":3},{"id":"6969b0689eb7d950700548b526dc41e0e694c65b","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-10-26 11:30:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"06123b181565911f4ffaf45f422977011e7dd665","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 14:17:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da200967aa7b4ea49eafaa928725bd9d : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/813eb2dc2ddf4b54b6291e422afd9216 : SUCCESS in 4m 25s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/91b7e0380cff454987dba3e34d53a77a : SUCCESS in 1h 56m 48s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/3432cc37e91c4c179ac71dcb19ec0ec8 : SUCCESS in 2h 02m 28s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/691378a7b33d4ff68dba3075ff7df6e4 : SUCCESS in 1h 24m 39s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/14f7427cadfb4f8db2162ed4d3631397 : SUCCESS in 1h 40m 52s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/2527624d81b34a92b995e76fc52a19be : SUCCESS in 1h 24m 19s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/e96c35c17b714347989f351af4a8da15 : SUCCESS in 1h 19m 27s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/deddf3e6618441da9c2ef3a3dddfc3b3 : FAILURE in 2h 05m 24s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/fd10e84880f247b7a72f906075968ad3 : SUCCESS in 1h 47m 45s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/fb43233a17164307819760b7db0bda2b : FAILURE in 2h 22m 14s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/581d4b192d7c46dbbaaf43c55854d154 : FAILURE in 2h 39m 29s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/48d1b3a1e49b477a856ba0a17d7bb349 : POST_FAILURE in 2h 43m 41s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/01c1520d294148dab0521df003d9932d : SUCCESS in 1h 35m 37s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/224c814b27384f3a9c9ce9284d2703de : SUCCESS in 1h 15m 20s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/10465a4a1dd445a08bb290560be9f95a : SUCCESS in 1h 21m 40s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/7bf6045b28ae4e73a7fe5b16ee0cb7b0 : SUCCESS in 1h 15m 05s","accounts_in_message":[],"_revision_number":3},{"id":"204aaa22c9e2bd57c745b57c68721899c1dc021c","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-10-26 14:37:33.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"999dab2a1be66cd1a49ef11f476d6d9785687e3b","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2021-10-27 07:33:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d2c2f15c63a353401338f09a02ee77556f03bdba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 10:18:04.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4db4e96b76a642e2957c0d297af2bbdf : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46173f56ff6140e0bcbb56ed9a933528 : SUCCESS in 3m 56s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/86b65911042b46ec8c813ae7d1f01dca : SUCCESS in 2h 07m 13s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/c7c54ecdd04c48319ff13d5d085fd663 : FAILURE in 1h 59m 56s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/3dc546da20b4465e98f0e6b6f413a825 : SUCCESS in 1h 48m 13s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/de120f76762d4a9dbd6d383607b2ec37 : SUCCESS in 1h 27m 31s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/a69f852bb1234254bc8424911dc92aa9 : SUCCESS in 1h 29m 59s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/6641d467336049ac8bcae75982b3e042 : FAILURE in 1h 35m 37s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/a428d79245e04d96915d5410df1bbce7 : FAILURE in 1h 53m 20s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/4aa71112d2ac4e019db074b4071effed : FAILURE in 2h 29m 53s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/76c95eb82e984f989e9bcdd5caad3d04 : FAILURE in 2h 38m 22s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/3233971f79c04ca1bc3b56499d220aac : FAILURE in 2h 01m 58s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/d17e1d98c7ee4dd5b23342e3aa5b9024 : SUCCESS in 2h 39m 48s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/70149d7c47114934931009a21a02b3c9 : SUCCESS in 1h 49m 06s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/0a5d41c8c04244b79a41a78f446867b2 : SUCCESS in 1h 08m 31s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/352cfece82e64f27a12e253329452447 : SUCCESS in 1h 27m 51s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/5c8f6ae41e214c6ca929d1f42649e7ab : SUCCESS in 1h 18m 33s","accounts_in_message":[],"_revision_number":3},{"id":"98b2fd77f10c84dca1c33f90f536048f4d4a53ac","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2021-10-28 13:22:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"39a5eeed0a6de11576b3fdada6bcd96bfa936d76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 16:43:54.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a564d30bfa348e38a6d4f92e1c0d41b : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ccc5dddb8b34261a251176edad497cf : SUCCESS in 4m 47s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/677a0c97d3dd493fb3a73c28bc8578d6 : SUCCESS in 1h 36m 21s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/69169d4b471e46f3aae325ece31df589 : FAILURE in 1h 56m 29s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/8225f1af3737420db2f2a4146322ebc8 : SUCCESS in 1h 22m 35s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/abe8e78b2d5a42f7bb3647105dffa21a : TIMED_OUT in 3h 05m 43s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/84b2def03a254f5e90821fc46bdcbe3f : SUCCESS in 3h 00m 01s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/6e1f7df7251240889b736fa63abda76f : SUCCESS in 3h 05m 27s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/cf915125a5cf426b9176c3dfe6abdf5d : TIMED_OUT in 3h 04m 45s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/0f402e0d6eee488a907d43e7f56f739c : FAILURE in 1h 52m 22s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/32492afe8db94302ad727a33aea37a04 : FAILURE in 2h 38m 07s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/c80ffab0f0f74b359975dbc9b32b4774 : FAILURE in 2h 30m 13s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/0e8c1451f1274842b00494ecd1061db5 : POST_FAILURE in 2h 44m 44s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/57266596757b464596dce9a07e600caa : SUCCESS in 1h 34m 26s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/6c8a1472a9fb4b2cbd1b9314bfc15abb : SUCCESS in 1h 12m 50s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/13dbb53cc0be4928bb72f1855296f6c1 : SUCCESS in 1h 14m 38s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/031fcde65a484aa6939811ef7f339074 : SUCCESS in 1h 10m 39s","accounts_in_message":[],"_revision_number":3},{"id":"27b1cc356b384a381ef2a301c69738d7f22ffcb8","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2021-10-29 07:11:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"00927c6aeebd67dca6016af325529947c9526eb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 10:34:33.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87f8428ab6194de28dd030d7b825153e : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c698b45272e477ea37edd96e9063ed4 : SUCCESS in 3m 56s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/15782351ef384267bcc845f8e714c89e : SUCCESS in 1h 57m 26s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/87755c8cb33d4fb292621c218c49367a : FAILURE in 2h 22m 46s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/10b7c667445f46dc8d129cd47e355210 : SUCCESS in 1h 23m 11s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/ecaf387d8ef5491cbc023c6819127a39 : SUCCESS in 1h 49m 04s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/e4b4f8ac09d540db9e5f641ccc947962 : SUCCESS in 1h 38m 57s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/59d3fe3bc7584f05ad4f6d98b0ca0cda : FAILURE in 52m 37s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/574e125a159445e5a4f5111e91ca5676 : FAILURE in 2h 00m 42s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/ee462a7ac03b4f2aa164bfdde086cef8 : FAILURE in 2h 17m 04s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/880ed679a9454920ab04a37d8fb0f872 : FAILURE in 3h 03m 51s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/ffc55c2f21d444c09a452ef14c35fc37 : FAILURE in 2h 33m 16s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/da4a286ff3e14a4991e4a4f3e5dfa3a1 : SUCCESS in 3h 21m 04s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/e62d29cadf144d24b5821a0a6550682d : SUCCESS in 1h 36m 47s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/145d60f3b6af4a29a9bfed4718071bdd : SUCCESS in 1h 10m 32s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/f633249beb14469b95fcf3e44c679c13 : SUCCESS in 1h 06m 00s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/c4ab65b72cee4db483bc4fea5e63bd4a : SUCCESS in 1h 20m 10s","accounts_in_message":[],"_revision_number":3},{"id":"741849126322550a749679b260c3b1eacb68546c","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2021-10-29 11:14:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"07e0dae8f3a44a8c8c5544fe7f70a3c34cf3b2a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 14:12:07.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/597eca1f0934418fb4c79699f62a8edd : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e93b841aab5a4b548c466f4d2eebe1db : SUCCESS in 3m 54s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/8292db0e711b4d28a3b5940d5e8c3a21 : SUCCESS in 1h 52m 58s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/d6c67885524d4d2da89f8edca9f20edf : FAILURE in 2h 39m 10s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/1f45db929130495a954d9d9ae2472905 : SUCCESS in 1h 23m 42s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/a6531c7e27534cb0b7329ba777a0dd22 : FAILURE in 2h 12m 37s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/8a0be3594b3d44d59125da13f95b1edf : SUCCESS in 1h 37m 52s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/f2a19010a2f64a5592d0a62c8e206e0c : SUCCESS in 1h 35m 06s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/df2a86c7d52f4f6386e321aeeda1cf24 : FAILURE in 1h 46m 54s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/b4c0604b509e4212a77b58440298e03d : FAILURE in 1h 44m 32s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/79e6906a3cd64ddd86da390963b41391 : FAILURE in 2h 54m 24s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/e2d0cc9230e648929e32a0f00f1a6ba8 : FAILURE in 2h 18m 27s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/945123597e4d4217ac4998d768ae78cd : SUCCESS in 2h 43m 40s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/140e4983769743dcb67b51ac01fb7df6 : SUCCESS in 1h 49m 07s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/3b7a031f796442eeade5f51e8a2e2d45 : SUCCESS in 1h 16m 54s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/e6361d4603564fd697469b1c7da55ebf : SUCCESS in 1h 39m 54s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/e8cc131e37294cf9839c706b2108b6d7 : SUCCESS in 1h 02m 23s","accounts_in_message":[],"_revision_number":3},{"id":"3ff6e4e7ccc51b6043802ad23c371053e63716a7","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2021-10-30 20:00:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ba0cc35174bc73122ac5907cf014e28722d38e15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-30 23:16:31.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed77c21b792d459d9ceac4d94decf414 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b16a2fde907404c92a0050979677dcd : SUCCESS in 3m 31s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/307302c46e614aeea5b68bc2db007cbf : SUCCESS in 1h 57m 10s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/c5ea3dfb3802468cb211181a2de0532e : FAILURE in 2h 24m 13s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/f263053cdf5544f78791ab5bbec6642c : SUCCESS in 1h 35m 32s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/98ae7f197f6049b4940a8fae3d27207e : SUCCESS in 1h 40m 08s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/b8b0bc60840d431d8117ff9a1b3befff : SUCCESS in 1h 36m 25s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/6ca4944851354d2c8e5d6075005ac411 : SUCCESS in 1h 18m 06s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/d6544f2a0f6b4162ac55bed7fa5fa192 : FAILURE in 2h 17m 02s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/40c7155baedc4c449b3b6c835e44d612 : FAILURE in 2h 12m 40s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/5022ddb74577400a8c73aa6df49f7916 : FAILURE in 2h 36m 36s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/1ce8ace5ec094780aed20f06a6dd0a0b : FAILURE in 2h 15m 50s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/a888cf1791d74b7e919c033a50dda81a : POST_FAILURE in 3h 09m 36s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/38bb0213623749aaa91cdf75c45f7e80 : SUCCESS in 1h 48m 16s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/2e6bb27952e44b5e8a78c4c47dc2e86d : SUCCESS in 1h 25m 25s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/b411407eb19448b38c3665528ad0a9ed : SUCCESS in 1h 08m 44s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/5fb8094dbfca42b2bf7e19356a139a6e : SUCCESS in 1h 13m 38s","accounts_in_message":[],"_revision_number":3},{"id":"c615faf5c0e1f558f7c2a68edf5f3167d80bd939","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2021-10-31 06:24:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"78edfa664d3ec40a007a4f5874f361b38b1d87dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-31 09:33:54.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8faf2b884ade4cdc9f760c697a3fb451 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f989d7167ae48b5bb5d20f1600b064a : SUCCESS in 3m 34s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/d1cf7b092e044d39b54343a13c8e8306 : SUCCESS in 1h 26m 42s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/1e78eed27d2a42e6b8724b45036dc7eb : FAILURE in 2h 12m 22s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/645e910fd9fa487b99ffc8d07270a26b : SUCCESS in 1h 22m 50s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/fbde2aa649e34e12b86bb051b0c647a6 : SUCCESS in 1h 46m 57s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/7cc1595ebe31406f9750d7337312a1a0 : SUCCESS in 1h 24m 46s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/235d6a9d0a4542609ed9a1085f37e4c7 : SUCCESS in 1h 15m 26s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/b39dab5d1ccd4dd4abd52ab6cc825625 : FAILURE in 2h 02m 21s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/845a3485d3d54b5c9ecd22875050801b : SUCCESS in 2h 00m 12s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/547cefca7f3049f5a74346c974b95d70 : FAILURE in 3h 04m 28s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/0538c77628cd43339c5b73cf58f38abd : FAILURE in 2h 41m 38s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/5fc16d7f7e2c485fad2d9a8f807e313f : POST_FAILURE in 3h 01m 37s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/de02cfd2919d4b15b6ee56feeaf73852 : SUCCESS in 1h 29m 53s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/848b6b5a311843c785868428f1ac0148 : SUCCESS in 1h 29m 22s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/fe5b2a5e955f46a3ab2bdb275d3eb089 : SUCCESS in 58m 11s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/8063c86aea634311ad8ab0be472a59a8 : SUCCESS in 1h 13m 22s","accounts_in_message":[],"_revision_number":3},{"id":"ff81afa9c34c17e5cc27584891dfa40465cee6bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2021-11-03 03:06:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3c1ed0b07299c551ac695fbbd029136536881d0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2021-11-03 03:07:38.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"4e4263678d152713461d79ae698429fb0a01dc55","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2021-11-03 03:38:15.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bc45c28d43c12a76a22e9cf1f99df79cf182c417","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 06:40:26.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fbf3ba69e3748089c87bd1f35397d6a : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c062807d3b64a23b441453f36a8ed44 : SUCCESS in 3m 49s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/9b6c9b21fc074077831151235a9c04c1 : FAILURE in 1h 53m 44s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/ba5e7fc0befa42348f1ab4b87894077a : FAILURE in 1h 52m 02s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/88c0ae3416624517b63847708132b25f : FAILURE in 1h 45m 00s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/c96cf84cae56476a8e19b0ac36c5fa2e : FAILURE in 1h 58m 08s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/58fe2e3c9d4842a89128d58ccecb8ff1 : SUCCESS in 1h 30m 45s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/c3047f45e7144c1b8359b49a804e580c : SUCCESS in 1h 16m 37s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/28a2899fe6d94bfc8fc5590f2f310d4d : FAILURE in 2h 01m 40s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/fd2b923808b3411b9b5fdf6a92ce9217 : FAILURE in 1h 41m 13s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/ab43d4c05b634ffc81e15d514fb929b8 : FAILURE in 2h 26m 36s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/459762fa14d94ec18ee248b7284f617c : FAILURE in 2h 59m 50s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/1d3363f97b6f40108adab38376a7effc : SUCCESS in 2h 33m 10s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/93433dec05f143ae840f5b315927eaf1 : FAILURE in 1h 50m 58s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/cde15d8258e24ddfac9b2ef04b1c5c90 : FAILURE in 1h 31m 21s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/bdd345d93988489c903b833a6d5c7d98 : SUCCESS in 1h 22m 37s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/6e6775c758294c1cb88968962f129478 : SUCCESS in 1h 26m 58s","accounts_in_message":[],"_revision_number":5},{"id":"0ae502fdc23b07608df778f3dab2ef99547528ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2021-11-03 06:48:28.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a6bde3b39915f1b15c068af2dde918ed0efee514","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 08:44:10.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f872c897bfb4568a3a937fa03480622 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19d4bc99cd2049e895e745ba27a473e2 : SUCCESS in 4m 48s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/df113cdfef604fe69e97fb48f1f71dc5 : SUCCESS in 1h 54m 24s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/932203c1ae364e649aaaa189e84b296a : SUCCESS in 1h 38m 33s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/bcde8df99ea24f5b9e0066fd709e1b9e : SUCCESS in 1h 00m 23s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/55352231160e46b8ab678a380173979c : SUCCESS in 1h 05m 02s","accounts_in_message":[],"_revision_number":6},{"id":"6744a268da31cadea099720b76b6172b6e316689","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-11-03 12:43:55.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"201c81b4ed6cbe20cb7fdde760941171bce06d1d","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-11-03 12:44:04.000000000","message":"Patch Set 6: -Code-Review","accounts_in_message":[],"_revision_number":6},{"id":"5285ad98e090338d88b75c3154662c812df3901b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-11-03 12:45:53.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"5a31ae0479173735c426bc02dbfc53281847912f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-11-03 12:55:11.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"78607ddc8fe71ec1d84d1f4c6afcf510fbeab615","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 15:54:15.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ddfb6ec47194f79bbb6874eefe35a6e : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65e6af3f86134a969e1bdaa02bdf63dd : SUCCESS in 5m 07s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/97d059bda861441395c8829d802abac9 : SUCCESS in 2h 23m 53s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/fc2ebf8205744cfdbc993b93a82faeca : FAILURE in 2h 43m 45s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/a16cae4333bc43bba70749a584f9f7c6 : SUCCESS in 1h 48m 51s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/64bad9933cd84fbd8038069eb22874b3 : SUCCESS in 1h 33m 40s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/12a0925901454c5a825dafdad0425aeb : SUCCESS in 1h 25m 25s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/fe06a90515b44f45a5ba8b3fb2cfb3f3 : SUCCESS in 1h 30m 11s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/a41c52e43cd5445390df38b96a93e884 : FAILURE in 1h 52m 28s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/a22f3d788d634b07b1e683a697da607a : FAILURE in 2h 10m 08s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/cfa71cc0d1a04e2f943ea26df6e3b549 : FAILURE in 2h 47m 10s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/621eb6ca8a474a8c8741c5495edaf141 : SUCCESS in 2h 45m 58s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/d72bf5b8083148f4b748b03d0aaa5f11 : SUCCESS in 2h 56m 46s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/7a72015741404beca414aadbe79611d1 : FAILURE in 2h 08m 25s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/b24eb6986f4c4890a41a7064c2162237 : SUCCESS in 1h 31m 28s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/8a9b21e8a9164b95ad1fcefe3c8a1bba : SUCCESS in 1h 55m 42s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/e2f2d7e379674754a73e9111493e5ff3 : SUCCESS in 1h 24m 10s","accounts_in_message":[],"_revision_number":8},{"id":"099aefe321e89a809882160475fb4abaecc93363","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2021-11-04 00:00:08.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"b5283dcfb318ce122fc49e0e20e123a2e646ae8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 03:03:41.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/868c3be3ccea405788c6b01822fbf58b : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b3149354cfa4a3a927c71d58800ba0c : SUCCESS in 3m 50s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/e2f63f9c223147ba944436aa2afd92c2 : SUCCESS in 2h 32m 53s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/23c3e2afcab14e4dbd55d312705379b1 : FAILURE in 2h 26m 17s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/47889bad0962448582d944b607ba4dbb : SUCCESS in 1h 53m 09s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/9743e76537954d97b40fca86f5f29528 : SUCCESS in 1h 40m 21s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/6a421b440892472fa5d2e5e411348f01 : SUCCESS in 1h 46m 48s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/505c2f7b37bf4c29a3c4c98509dab45e : SUCCESS in 1h 41m 39s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/d3e58eb86ec840c2a4f5621c0c23f6b0 : SUCCESS in 2h 34m 06s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/11834df7eaf7421c9a373037cd04d28f : FAILURE in 59m 17s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/ad22c0e077b44333aa2f18bab87012ed : FAILURE in 2h 57m 39s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/991bf97215ba4826a97edfff0be2475b : FAILURE in 2h 49m 45s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/65e87f2584b74337b8531785bb652534 : POST_FAILURE in 1h 00m 05s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/5028239392f9481f94dcc12196946015 : SUCCESS in 2h 14m 24s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/9905656356b94c14b56bdedd051ca809 : SUCCESS in 1h 28m 39s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/d51405afde4949d19f9ae070ff6eb553 : SUCCESS in 1h 19m 55s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/cf2f1448284a4e14ab6add8d2cb2b553 : SUCCESS in 1h 03m 53s","accounts_in_message":[],"_revision_number":9},{"id":"2a12a2f5d61b15570437e9e03403391b8f24981f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2021-11-04 04:08:48.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"8118e20d0ad707e27468be3c79f59bbfdf983c37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 05:49:26.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/949bc8118070442ea847219682cab25a : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/663a3aef49584dddb76db553068e64c1 : SUCCESS in 3m 50s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/a658b407b310404ab91d8c705a78a3b8 : SUCCESS in 1h 39m 03s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/1123aa4a433945bf8c9ab002e8b64e47 : SUCCESS in 1h 23m 41s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/4abcaf8d258a424a8f4900e4ece45132 : SUCCESS in 1h 17m 06s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/d2cc5cb445bd43f5b72068195c20fcb5 : SUCCESS in 1h 23m 14s","accounts_in_message":[],"_revision_number":10},{"id":"53c3a895da77382c0a8a5460e64973c85a9f5fc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2021-11-04 14:22:03.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"354f857bf38a0c5336fe734c8dbd6941f06dd83d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 16:28:02.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d28e5a00ebda40fbb7648a9ec8ff63d9 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7afc763005ce4c9c9bfd327354f023c9 : SUCCESS in 4m 51s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/2f08151d3b29441ebaa894fcdb81b997 : SUCCESS in 2h 01m 17s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/5c1a04cb357842eea9b8078690df1b93 : SUCCESS in 1h 21m 01s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/c43610cd913b4b1ea3eaeec17ce85272 : SUCCESS in 1h 50m 18s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/b592f9b937bd486b8c02c82f4d184895 : SUCCESS in 1h 41m 28s","accounts_in_message":[],"_revision_number":11},{"id":"67f6c41dacb70d55a63abd39cb1641ee5e0c2f78","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2021-11-05 06:42:33.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"e9e3ebc316dac69a1d02810c6a3419cde3fe1da7","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-11-05 09:40:34.000000000","message":"Patch Set 11: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"5b9f05b3a05ece3772c301141a1196d8717a1fd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 12:37:03.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5574f0fb21845fe884e4ef500811790 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e4c8311b7e14287913c78d50b89cbd3 : SUCCESS in 4m 08s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/72c9b0f852fa454c91bf4cf50403e50e : SUCCESS in 1h 34m 12s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/871d5751b84541a09493b006d9196a87 : SUCCESS in 2h 22m 03s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/1364cf6085ce4a7a8f4a55707ccf45b2 : SUCCESS in 1h 33m 34s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/749e60ee972a4af1ab55cbd2d983db12 : SUCCESS in 1h 50m 20s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/3362e58b06124323aa0aafe1f83d74c0 : SUCCESS in 1h 33m 41s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/25e4767cc4874d1caea891975c3ff165 : SUCCESS in 1h 18m 35s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/077cfcec4a244ff29c460644a22116ed : SUCCESS in 2h 17m 00s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/71d186616cb34443883315392c74c2e6 : SUCCESS in 1h 44m 01s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/c858af81f79548b1a79fb1340a788527 : SUCCESS in 2h 44m 10s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/5cb01d49688a46dba2317d38c4ba3833 : SUCCESS in 2h 44m 25s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/1e1681dff96a436086fa8d2fd13f2227 : POST_FAILURE in 2h 51m 36s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/dc60974ecda8435da2867b7fa10ebbae : SUCCESS in 1h 32m 20s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/30f0584607a94bceb653494562cadb43 : SUCCESS in 1h 23m 56s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/23b2da12b3d14bba82ee09da1d26ebec : SUCCESS in 1h 03m 05s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/8066400bce6d416583c3f7ae8b5fb346 : SUCCESS in 1h 20m 04s","accounts_in_message":[],"_revision_number":11},{"id":"d1f94b9ea161ffc1d9c14740f6118252249789d6","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-11-05 14:32:03.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"5d726b394d4f4fae7dc0d2d130731b2dadacd1bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 14:32:44.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"855c0cbf3f84f7d6f9071bdefe664919d18adc99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 16:05:34.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcc2e481ed6248e58f88d3e2137f2535 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06007378f6564c19838320424e5183de : SUCCESS in 4m 03s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/f2b51e4ce0fb4f9dba6d80c94679d928 : SUCCESS in 1h 32m 01s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/450aa05d5aa944dfa9859636daa29070 : SUCCESS in 1h 26m 55s","accounts_in_message":[],"_revision_number":11},{"id":"27ca778b066e530fd0f55fb31bb38b92692dbb69","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 16:05:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"c4e05477e9769c0e177c8c084c3ccee1f40d133e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 16:07:17.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3228dd98f6e04145bb28a2886ebe004f : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"4bb12bd975a17dccfef0a7647c4a9bf3eae7134d","revisions":{"ad2a1a5860ded5eac1b1c36c4b91f894f70677cf":{"kind":"REWORK","_number":1,"created":"2021-10-20 13:29:10.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/80/814780/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/80/814780/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/1"}}},"commit":{"parents":[{"commit":"e00ba7738c36db7087e0eb62500618367203483f","subject":"Merge \"Add xena jobs on master gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/e00ba7738c36db7087e0eb62500618367203483f"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2021-10-20 13:26:35.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2021-10-20 13:27:57.000000000","tz":180},"subject":"Update test_service_without_selector test","message":"Update test_service_without_selector test\n\nCheck that loadbalancer\u0027s memebers are delete when deleting the endpoint\n\nChange-Id: I5d01029bae9c782ab57d38dff8c6aea1b2128305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ad2a1a5860ded5eac1b1c36c4b91f894f70677cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ad2a1a5860ded5eac1b1c36c4b91f894f70677cf"}]},"branch":"refs/heads/master"},"bfb47f77ea98ca4de202651a8b24bfa7706fda30":{"kind":"REWORK","_number":2,"created":"2021-10-21 13:52:47.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/80/814780/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/80/814780/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/2"}}},"commit":{"parents":[{"commit":"e00ba7738c36db7087e0eb62500618367203483f","subject":"Merge \"Add xena jobs on master gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/e00ba7738c36db7087e0eb62500618367203483f"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2021-10-20 13:26:35.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2021-10-21 13:52:38.000000000","tz":180},"subject":"Update test_service_without_selector test","message":"Update test_service_without_selector test\n\nCheck that loadbalancer\u0027s members are delete when deleting the endpoint\n\nChange-Id: I5d01029bae9c782ab57d38dff8c6aea1b2128305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/bfb47f77ea98ca4de202651a8b24bfa7706fda30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/bfb47f77ea98ca4de202651a8b24bfa7706fda30"}]},"branch":"refs/heads/master"},"3222df9d0d6f7a295c6fe5a761c2b0bb58d791f0":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-10-25 09:17:21.000000000","uploader":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"ref":"refs/changes/80/814780/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/80/814780/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/3"}}},"commit":{"parents":[{"commit":"e00ba7738c36db7087e0eb62500618367203483f","subject":"Merge \"Add xena jobs on master gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/e00ba7738c36db7087e0eb62500618367203483f"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2021-10-20 13:26:35.000000000","tz":180},"committer":{"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","date":"2021-10-25 09:17:21.000000000","tz":0},"subject":"Update test_service_without_selector test","message":"Update test_service_without_selector test\n\nCheck that loadbalancer\u0027s members are delete when deleting the endpoint\n\nDepends-On: https://review.opendev.org/c/openstack/kuryr-kubernetes/+/815233\nChange-Id: I5d01029bae9c782ab57d38dff8c6aea1b2128305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/3222df9d0d6f7a295c6fe5a761c2b0bb58d791f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/3222df9d0d6f7a295c6fe5a761c2b0bb58d791f0"}]},"branch":"refs/heads/master","description":"Edit commit message"},"550ef591842bf1afbf03baaf384f1e55ca8a3850":{"kind":"REWORK","_number":4,"created":"2021-11-03 03:06:18.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/80/814780/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/80/814780/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/4"}}},"commit":{"parents":[{"commit":"89b66e60085ebd7c3492952a3940d88b7c3fb9b7","subject":"Ensure lb is ACTIVE before connectivity check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/89b66e60085ebd7c3492952a3940d88b7c3fb9b7"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2021-10-20 13:26:35.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2021-11-03 03:06:09.000000000","tz":120},"subject":"Update test_service_without_selector test","message":"Update test_service_without_selector test\n\nCheck that loadbalancer\u0027s members are delete when deleting the endpoint\n\nChange-Id: I5d01029bae9c782ab57d38dff8c6aea1b2128305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/550ef591842bf1afbf03baaf384f1e55ca8a3850"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/550ef591842bf1afbf03baaf384f1e55ca8a3850"}]},"branch":"refs/heads/master"},"2c5dc086beb450ed6597ce2f2a81e965a906a8f2":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-11-03 03:07:38.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/80/814780/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/80/814780/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/5"}}},"commit":{"parents":[{"commit":"d411b2dc22ef4eac1a2b2b33adf2cc9f9b7deb7c","subject":"Enable kuryr tempests tests again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d411b2dc22ef4eac1a2b2b33adf2cc9f9b7deb7c"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2021-10-20 13:26:35.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2021-11-03 03:07:38.000000000","tz":0},"subject":"Update test_service_without_selector test","message":"Update test_service_without_selector test\n\nCheck that loadbalancer\u0027s members are delete when deleting the endpoint\n\nChange-Id: I5d01029bae9c782ab57d38dff8c6aea1b2128305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2c5dc086beb450ed6597ce2f2a81e965a906a8f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2c5dc086beb450ed6597ce2f2a81e965a906a8f2"}]},"branch":"refs/heads/master","description":"Rebase"},"5e22c037b729f2788ee1c29f4b228597ebc489e7":{"kind":"REWORK","_number":6,"created":"2021-11-03 06:48:28.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/80/814780/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/80/814780/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/6"}}},"commit":{"parents":[{"commit":"89b66e60085ebd7c3492952a3940d88b7c3fb9b7","subject":"Ensure lb is ACTIVE before connectivity check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/89b66e60085ebd7c3492952a3940d88b7c3fb9b7"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2021-10-20 13:26:35.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2021-11-03 06:48:18.000000000","tz":120},"subject":"Update test_service_without_selector test","message":"Update test_service_without_selector test\n\nCheck that loadbalancer\u0027s members are delete when deleting the endpoint\n\nChange-Id: I5d01029bae9c782ab57d38dff8c6aea1b2128305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/5e22c037b729f2788ee1c29f4b228597ebc489e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/5e22c037b729f2788ee1c29f4b228597ebc489e7"}]},"branch":"refs/heads/master"},"ea1a6f3b6d30ebe3fd7cdccf2c6a0bfbfb0dd2d4":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-11-03 12:45:53.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/80/814780/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/80/814780/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/7"}}},"commit":{"parents":[{"commit":"d411b2dc22ef4eac1a2b2b33adf2cc9f9b7deb7c","subject":"Enable kuryr tempests tests again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d411b2dc22ef4eac1a2b2b33adf2cc9f9b7deb7c"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2021-10-20 13:26:35.000000000","tz":180},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2021-11-03 12:45:53.000000000","tz":0},"subject":"Update test_service_without_selector test","message":"Update test_service_without_selector test\n\nCheck that loadbalancer\u0027s members are delete when deleting the endpoint\n\nChange-Id: I5d01029bae9c782ab57d38dff8c6aea1b2128305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ea1a6f3b6d30ebe3fd7cdccf2c6a0bfbfb0dd2d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ea1a6f3b6d30ebe3fd7cdccf2c6a0bfbfb0dd2d4"}]},"branch":"refs/heads/master","description":"Rebase"},"55dce96b7fa7b9d26fe77e4aa8041c76a8392152":{"kind":"NO_CHANGE","_number":8,"created":"2021-11-03 12:55:11.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/80/814780/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/80/814780/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/8"}}},"commit":{"parents":[{"commit":"7b7bad31291589b3e94703d2794fb2a0cbc1366e","subject":"Enable kuryr tempests tests again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/7b7bad31291589b3e94703d2794fb2a0cbc1366e"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2021-10-20 13:26:35.000000000","tz":180},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2021-11-03 12:55:11.000000000","tz":0},"subject":"Update test_service_without_selector test","message":"Update test_service_without_selector test\n\nCheck that loadbalancer\u0027s members are delete when deleting the endpoint\n\nChange-Id: I5d01029bae9c782ab57d38dff8c6aea1b2128305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/55dce96b7fa7b9d26fe77e4aa8041c76a8392152"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/55dce96b7fa7b9d26fe77e4aa8041c76a8392152"}]},"branch":"refs/heads/master","description":"Rebase"},"b32da978e8c168781c05b2ee295767fdd5dc2a94":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-11-04 00:00:08.000000000","uploader":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"ref":"refs/changes/80/814780/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/80/814780/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/9"}}},"commit":{"parents":[{"commit":"0605e39aa0a8a802ff821b28d76155d30dac67bc","subject":"Enable kuryr tempests tests again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/0605e39aa0a8a802ff821b28d76155d30dac67bc"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2021-10-20 13:26:35.000000000","tz":180},"committer":{"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","date":"2021-11-04 00:00:08.000000000","tz":0},"subject":"Update test_service_without_selector test","message":"Update test_service_without_selector test\n\nCheck that loadbalancer\u0027s members are delete when deleting the endpoint\n\nChange-Id: I5d01029bae9c782ab57d38dff8c6aea1b2128305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/b32da978e8c168781c05b2ee295767fdd5dc2a94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/b32da978e8c168781c05b2ee295767fdd5dc2a94"}]},"branch":"refs/heads/master","description":"Rebase"},"f3a9d093dd47df3a88b4abac2d71ca694fd32524":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2021-11-04 04:08:48.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/80/814780/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/80/814780/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/10"}}},"commit":{"parents":[{"commit":"5264e4c207b07fa95a19afca4af800df55ed72d1","subject":"Ensure connectivity is checked after members are configured","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/5264e4c207b07fa95a19afca4af800df55ed72d1"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2021-10-20 13:26:35.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2021-11-04 04:08:48.000000000","tz":0},"subject":"Update test_service_without_selector test","message":"Update test_service_without_selector test\n\nCheck that loadbalancer\u0027s members are delete when deleting the endpoint\n\nChange-Id: I5d01029bae9c782ab57d38dff8c6aea1b2128305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/f3a9d093dd47df3a88b4abac2d71ca694fd32524"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/f3a9d093dd47df3a88b4abac2d71ca694fd32524"}]},"branch":"refs/heads/master","description":"Rebase"},"4bb12bd975a17dccfef0a7647c4a9bf3eae7134d":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2021-11-04 14:22:03.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/80/814780/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/80/814780/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/80/814780/11"}}},"commit":{"parents":[{"commit":"15eeb47ca8fdae439abb94a86c6fe7825ba7ee09","subject":"Increase timeout waiting for namespace deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/15eeb47ca8fdae439abb94a86c6fe7825ba7ee09"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2021-10-20 13:26:35.000000000","tz":180},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2021-11-04 14:22:03.000000000","tz":0},"subject":"Update test_service_without_selector test","message":"Update test_service_without_selector test\n\nCheck that loadbalancer\u0027s members are delete when deleting the endpoint\n\nChange-Id: I5d01029bae9c782ab57d38dff8c6aea1b2128305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/4bb12bd975a17dccfef0a7647c4a9bf3eae7134d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/4bb12bd975a17dccfef0a7647c4a9bf3eae7134d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
