)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~822298","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~I4c38c09faba99c8ca56abc0a32bc421011f9b2dd","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"fix-v1endpointport","attention_set":{},"removed_from_attention_set":{"11600":{"account":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"last_update":"2021-12-20 16:37:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4c38c09faba99c8ca56abc0a32bc421011f9b2dd","subject":"Fallback from V1EndpointPort to CoreV1EndpointPort","status":"MERGED","created":"2021-12-20 09:53:13.000000000","updated":"2021-12-20 16:38:59.000000000","submitted":"2021-12-20 16:37:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822298-fix-v1endpointport","meta_rev_id":"86cd1129e1801bc52105c60c4e520629926338fe","_number":822298,"virtual_id_number":822298,"owner":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-20 16:37:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"all":[{"value":2,"date":"2021-12-20 10:09:11.000000000","_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-12-20 14:05:14.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"all":[{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-12-20 14:05:14.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-20 10:09:11.000000000","updated_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"},{"updated":"2021-12-20 12:59:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-20 14:05:14.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"}],"messages":[{"id":"07f1b62f8f65f7259fa5068fb49957183defafd4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-12-20 09:53:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55091fe773726d42afc9c53fa6ddc3897ebad7c8","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2021-12-20 10:09:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"83af45d6f216620b27fbd0c80b3c66d66c29895c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 12:59:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec336a33c9b44170b00043292113d3e4 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee8aa092176742b898cfb4dd24e228b4 : SUCCESS in 4m 31s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/a40596a4e2124c829426947bac9de1a9 : SUCCESS in 2h 35m 15s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/fdac3ac416ea457b9b476f5ff86bf84a : SUCCESS in 1h 51m 03s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/d58a321ac02c4b8b9f69bdc640486fef : SUCCESS in 2h 26m 35s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/b50d4ccc19504cec82a6a151d2f57da8 : SUCCESS in 1h 58m 35s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/7e7dde171ec94374851e3fce3924dc8b : SUCCESS in 2h 02m 50s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/8d2268f4e8bd4e0f99a15227d756d4cd : SUCCESS in 1h 24m 02s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/2110beb4c9bc41888acb23da4d006e57 : SUCCESS in 1h 18m 47s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/ac3754086914404b8e06503ccc42cf63 : SUCCESS in 1h 58m 36s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/621aac50b4194b3884c93a506f734202 : SUCCESS in 1h 36m 20s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/55fd90f3760a4656b1691fbd57dff49d : SUCCESS in 2h 56m 26s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/9465055947824c99bbb90c856de0d2da : FAILURE in 3h 03m 20s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/13ff2439301743b887f5b0fddb6522b4 : POST_FAILURE in 1h 33m 39s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/56698e79e15a4ad2976fae4e6e741e06 : SUCCESS in 1h 45m 31s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/f68bdddab02f4d9cb04e030e6822b8e4 : SUCCESS in 1h 27m 35s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/ad04dc8e3f57475ab7d20a7b89dc959b : SUCCESS in 1h 11m 39s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/a268a66174ab4b3fbaf86b7343783af8 : SUCCESS in 1h 22m 19s","accounts_in_message":[],"_revision_number":1},{"id":"16764d18d34194603e00e2de037954a1022f3f1e","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2021-12-20 14:05:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"30c4167195515e2da6caeafb5846ba487ac7dd85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 14:06:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cba4c7ad572ea506445d7a9799d0d71570e6cb0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 16:37:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ddf1c4a1a8e4d57a13a0566036793af : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aae3e2354a6e469d8a6d47d7b852eab9 : SUCCESS in 4m 25s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/33baf79cd58e47c4b50cbb40aeeb5875 : SUCCESS in 2h 25m 52s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/464f040713014bd98e122589266b30ae : SUCCESS in 2h 13m 19s","accounts_in_message":[],"_revision_number":1},{"id":"bcfbe72a3119c52e1e7892c6ba3dcfc21b80fa4c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 16:37:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"86cd1129e1801bc52105c60c4e520629926338fe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 16:38:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/276931f6748c40079b9047e549e9b200 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"34378686c705359146dc81d116aabecf91a625a7","revisions":{"34378686c705359146dc81d116aabecf91a625a7":{"kind":"REWORK","_number":1,"created":"2021-12-20 09:53:13.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/98/822298/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/98/822298/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/98/822298/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/98/822298/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/98/822298/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/98/822298/1"}}},"commit":{"parents":[{"commit":"e1fdcc0a5bdfa81a0f49b3018748392593b4f031","subject":"Update scale deployment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/e1fdcc0a5bdfa81a0f49b3018748392593b4f031"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2021-12-20 09:45:06.000000000","tz":60},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2021-12-20 09:52:31.000000000","tz":60},"subject":"Fallback from V1EndpointPort to CoreV1EndpointPort","message":"Fallback from V1EndpointPort to CoreV1EndpointPort\n\nFor some unknown reason kubernetes\u003d\u003d21.7.0 renamed V1EndpointPort to\nCoreV1EndpointPort while other V1Endpoint* haven\u0027t got the prefix. This\nseems like a bug, but in order to deal with it, this patch adds a\nfallback to the new name.\n\nChange-Id: I4c38c09faba99c8ca56abc0a32bc421011f9b2dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/34378686c705359146dc81d116aabecf91a625a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/34378686c705359146dc81d116aabecf91a625a7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
