)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~822409","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~Id19f85625901b35900f9441aace8fe8938cb1d68","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"services","attention_set":{},"removed_from_attention_set":{"11600":{"account":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"last_update":"2022-01-19 09:30:57.000000000","reason":"Michał Dulko replied on the change"},"4727":{"account":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"last_update":"2022-01-24 11:06:37.000000000","reason":"Change was submitted"},"27032":{"account":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"last_update":"2022-01-24 11:06:37.000000000","reason":"Change was submitted"},"13692":{"account":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"last_update":"2022-01-24 09:29:58.000000000","reason":"Roman Dobosz replied on the change"},"27086":{"account":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"last_update":"2022-01-24 11:06:37.000000000","reason":"Change was submitted"},"23567":{"account":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"last_update":"2022-01-24 11:06:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id19f85625901b35900f9441aace8fe8938cb1d68","subject":"Add test_service_with_not_ready_endpoints test","status":"MERGED","created":"2021-12-21 08:57:50.000000000","updated":"2022-01-24 11:09:40.000000000","submitted":"2022-01-24 11:06:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":22,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"822409-services","meta_rev_id":"0e4b108c4e7c5d28070995823bbffafbeb7288c3","_number":822409,"virtual_id_number":822409,"owner":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-24 11:06:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2022-01-19 09:30:57.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-24 09:29:58.000000000","_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-24 09:29:58.000000000","_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"reviewers":{"REVIEWER":[{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-21 10:04:45.000000000","updated_by":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"reviewer":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"state":"REVIEWER"},{"updated":"2021-12-21 10:04:45.000000000","updated_by":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"},{"updated":"2021-12-21 10:04:45.000000000","updated_by":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2021-12-21 12:02:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-03 11:11:11.000000000","updated_by":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"},{"updated":"2022-01-03 11:11:11.000000000","updated_by":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"}],"messages":[{"id":"0ca0752c09dc603e82bc667ebec5c34908815c8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"date":"2021-12-21 08:57:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7c1c4f06b5a035f715ca7a6b4d2f9f8e4104316","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"date":"2021-12-21 08:59:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"db0fda401024085922cbcfb43aae98e493925e16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 12:02:43.000000000","message":"Patch Set 2: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7e9e82e0042458e9cb609851ae0e729 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f648d9811d3f44c78fbf35775e3a6e96 : FAILURE in 4m 06s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/33e4f1c83dc846eeb38eb56c78693649 : FAILURE in 2h 24m 14s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/c6df456514ac449f9ee41f07ded226b0 : FAILURE in 1h 53m 07s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/bb43f923d0ea4aadb4aed14cde9147fa : FAILURE in 1h 49m 43s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/19c675550ad54088b921a2e16ca4325e : SUCCESS in 1h 49m 17s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/8a8d20906bd24ff4b848a82382d430f8 : FAILURE in 2h 01m 57s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/fd49727db576463a80b5320e52d41f63 : SUCCESS in 1h 13m 49s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/cbf7dc3da8234ea495b03e90262b77de : SUCCESS in 1h 11m 07s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/20588ad623a64e9299ecdf2a3073ac9b : FAILURE in 2h 06m 02s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/24549f01f53946f68a52cb9a78c8a037 : FAILURE in 1h 57m 03s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/1d9a001e40504797a5e90371077be3a8 : FAILURE in 2h 57m 58s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/0c8e736430394ee4bbab29b2a4675fba : FAILURE in 3h 00m 37s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/8d2bab3a0dea4d1e9ae1fe58c75cd7a0 : POST_FAILURE in 52m 05s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/3e780f7cf35a47fe941b3a8ffb47e618 : SUCCESS in 1h 20m 57s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/faee2979256a44f4b618a3ad237e5ee4 : SUCCESS in 1h 28m 26s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/193dce6500a64ebca2c481f000190d2a : SUCCESS in 1h 02m 20s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/d9c312dda4b048b2b37298fb541387be : SUCCESS in 1h 01m 53s","accounts_in_message":[],"_revision_number":2},{"id":"9cbe7587497115c47a5e20c4baa8e39f9d495aa7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"date":"2021-12-21 15:54:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"923b149bd25045bcff27772b453808ad06b3b954","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 19:28:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3a8ae7a77d94951ad35cf63a17a347b : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6df85c8a4334460486b80b72a68857cc : SUCCESS in 4m 09s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/fbde61fc2de54acca3162e6ea76cf44c : FAILURE in 2h 14m 30s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/2fa9b329853d4c35b795f9c604fb0ca4 : SUCCESS in 1h 56m 51s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/ac8d908714e74d519dab21a06d619a1b : FAILURE in 1h 35m 54s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/5c6c14bac4524db1a2e324402d55f187 : SUCCESS in 2h 09m 39s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/5c0fef691ec04931b5241bc1796fba8a : FAILURE in 2h 28m 02s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/34f4529651db42ac976d538b8c7bf341 : SUCCESS in 1h 31m 33s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/cb606c38ad1145ee94a4e9de726c75f2 : SUCCESS in 1h 38m 52s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/fbc48c7959d840598065694368da16b7 : FAILURE in 1h 52m 37s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/17a3d781865b43fd9468b0270b2b2d94 : SUCCESS in 2h 03m 08s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/60f2d898311d4d1587fd31b89ab5f651 : TIMED_OUT in 3h 06m 42s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/6f28362e89d34b30a56aa128fae29142 : FAILURE in 2h 53m 59s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/1a9b8c149bcb426d8d6f8edd08106c70 : POST_FAILURE in 54m 10s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/4d3feeb626bf479a8e3e490212dc0449 : SUCCESS in 1h 53m 52s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/11ee497554d24f368a39ca4655c22cce : SUCCESS in 1h 12m 27s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/7f2fa2cfb26a4309ab8b754222ab9bd5 : SUCCESS in 1h 16m 59s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/2e5c44016f9e447abf915c419db14038 : SUCCESS in 1h 04m 10s","accounts_in_message":[],"_revision_number":3},{"id":"8878c836e16d901763c320489e4391bb13ca7910","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"date":"2021-12-21 22:49:36.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1eb1b8b4016d4d6bd39da44175d02ed96a43b652","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 02:01:51.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11b6b1395587479e9bde771287213007 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0542bfb3d242453fa81de33846df15cb : SUCCESS in 3m 28s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/fd341f399d7045969439ff896910fc8e : SUCCESS in 1h 47m 10s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/395a555c1ff04e049939c3217ceaf5cf : SUCCESS in 2h 01m 29s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/ba32f7ae59124d63a5c815590ba6da76 : FAILURE in 1h 58m 40s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/ae7c4ae0fe8c4779a49fbeeb060aaa47 : SUCCESS in 1h 46m 15s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/bbdeff6d61454b96a1409a6a46ceb552 : SUCCESS in 2h 21m 45s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/b0443bf648a84289a7a53e47ffc5312e : SUCCESS in 1h 45m 15s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/0b826ff6b90e41ddbb8869c46e97cdd0 : SUCCESS in 1h 04m 37s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/4880e120fd334b7b9118c164f8149db9 : SUCCESS in 1h 34m 27s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/601528efc8214992bd04f61328276136 : SUCCESS in 2h 17m 13s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/faf4ba1f885b4f1b9197cd3ad66a39f0 : TIMED_OUT in 3h 05m 25s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/e2e0d62a9d7d4853a0730e6b0ca45bb6 : SUCCESS in 3h 01m 02s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/c1a5272654b241e0b86b8af81405b125 : POST_FAILURE in 49m 21s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/6dcb7f768076467898968b994ccefe88 : SUCCESS in 1h 35m 35s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/ac9fc1ed12db4bb58edaf1e729bc1a90 : SUCCESS in 1h 22m 58s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/4381210c5ebb4fdeb8dee179ac13127d : SUCCESS in 1h 18m 28s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/1b24888b1f7449ff86ab1255a707a1ad : SUCCESS in 1h 02m 32s","accounts_in_message":[],"_revision_number":4},{"id":"ba9f354c37fc62a00c599f69aa1bbddc69234318","author":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"date":"2021-12-22 08:30:12.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cca7bd4f83faed529dca9259917beb45696752fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"date":"2021-12-22 10:27:18.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"fe0e633e3ad3c8db343d757be330f26ac8048b52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"date":"2021-12-22 10:59:05.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"7fe76a7289ae817f5305bb63f69605401deaf811","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 14:08:54.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/643f903cbe544c5cb5145d47aa44ba2e : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15ca6bb9cf964bcfbe34a3a73fef6076 : SUCCESS in 3m 42s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/829fe75cc4ea4424958c4980b65927c7 : FAILURE in 2h 14m 58s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/ff0f2ea855eb4c10a555c151174a6bc3 : SUCCESS in 2h 09m 45s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/a450550f805b47fc81a4bcdaae11db37 : FAILURE in 2h 07m 17s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/d60dcffa270549daaf5796e678cd9cd0 : SUCCESS in 1h 55m 52s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/1caad2e2847c448db70dc0adc006c0fb : FAILURE in 2h 59m 58s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/48da11663ec147dd90f0a819aabb17b6 : SUCCESS in 1h 11m 28s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/8a2891daa0d746e9ab0588739ad3867c : SUCCESS in 1h 16m 41s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/76be7cdaa8664048aac86375d93fb954 : SUCCESS in 1h 24m 15s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/577b35f7b8b34e0caa8b8be03257bf20 : FAILURE in 1h 52m 21s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/b3f4f227dad04653a54801b314ff1f20 : FAILURE in 2h 27m 02s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/2fcb46b0965f4136a4fec5c6a851b312 : FAILURE in 3h 05m 08s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/b4e2bac7b86a4ce6ae3b8dcfd0b79f73 : POST_FAILURE in 43m 57s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/a14bc453cb6b4f40bf825a8cfad9f33a : SUCCESS in 1h 21m 02s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/668bd46591d846859ad926d0e476377c : SUCCESS in 1h 16m 17s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/5256d1b8fad94be783705ce1b31645fe : SUCCESS in 1h 05m 32s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/4b805fd462cc4b5785a08680a303235d : SUCCESS in 1h 11m 12s","accounts_in_message":[],"_revision_number":6},{"id":"39f43f48ad623f0d68902c55cdb7bb19da6f625e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"date":"2021-12-22 22:54:08.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"3e5253bd380f5a58d97aad774d93cdc3b6cdb329","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 02:04:23.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b643a030289d49038fd03a959e341408 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f67e8c2ee75b4f99aa0a946bf45d7b1f : SUCCESS in 4m 03s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/dfe01901b86c45ae8835da72a6f8f5a1 : FAILURE in 1h 51m 05s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/c65f3c7b08de485da460ad36f280db83 : SUCCESS in 2h 12m 31s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/89382a5aaabb4046a5a85537394632a2 : SUCCESS in 1h 46m 37s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/c6734d1104db4b348e02c81777d7c6c5 : FAILURE in 2h 31m 35s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/79fa9ea16fe342c697d35ca4aa0fb61d : SUCCESS in 1h 54m 34s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/fedceb3ee1a7476c9562b98963016447 : TIMED_OUT in 3h 06m 06s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/4ab4b354975c46d4b6487cd23e358f48 : SUCCESS in 1h 06m 43s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/30691d9af40b405c9f22b7bc7bbf5ee0 : FAILURE in 1h 45m 42s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/cbf1a6fc0f584a90b3114294b985209f : FAILURE in 1h 38m 46s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/099ebaa00c274af1955cba19a23bb6fb : SUCCESS in 2h 47m 44s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/acfd0688b4d6400daaf4477a39ec5660 : SUCCESS in 2h 42m 16s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/2903cb936cdd46a4b545a71b164eb649 : POST_FAILURE in 54m 07s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/baf2e60b8aa4454db813eed08fd1e307 : SUCCESS in 1h 48m 54s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/0721d4f591054e629125152accd0f162 : SUCCESS in 1h 13m 57s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/9d3e4200ea7b4d6287a320cfadae455e : SUCCESS in 53m 55s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/adf70b1877174c01b03a29f87d87b9dc : SUCCESS in 1h 08m 25s","accounts_in_message":[],"_revision_number":7},{"id":"142858d5cad3478784d10b7826ffa97b4ea64c58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"date":"2021-12-23 15:11:24.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"7acc2fa72eeffd5fe9f5d98dd0654e0dc4684d5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 17:47:39.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/295f023b89074685a8150eccf7c55f9b : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ddc1774df9c4fb99413041cb6286562 : SUCCESS in 3m 41s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/07847dac1ff64142afe376d2305b0658 : SUCCESS in 1h 47m 43s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/4f3a2f310acb41c18983e0857e908cd6 : SUCCESS in 1h 37m 10s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/eed774f2c68b492b8ada5c49d7696859 : SUCCESS in 1h 41m 43s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/6f203b0761b642718bdd1fcb3699fafe : SUCCESS in 1h 23m 59s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/199913db355b49c3ab4529c60b6227cb : SUCCESS in 1h 42m 40s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/dcd83f5ac4904e32a8b30cf9437b50b3 : SUCCESS in 1h 20m 09s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/bcdb8747964045fca14898854f639a56 : SUCCESS in 1h 02m 27s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/e46623eaa36d4c0a8c27853888ee4601 : SUCCESS in 1h 35m 38s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/88a7dc42e1d8494ea197099aebc91071 : SUCCESS in 1h 37m 48s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/265980d9d0044c5fb7efc553ea7351ec : FAILURE in 2h 33m 24s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/99e284bcb214477f878b28c36da4cb88 : SUCCESS in 2h 05m 44s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/84ed1de7076b489aba6526bd9d87dafd : POST_FAILURE in 47m 09s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/88008929365e4f69a2f614cd1a82a1ed : SUCCESS in 1h 43m 35s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/a002c4fcafe141729e5d2fde9d538bd1 : SUCCESS in 1h 24m 53s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/1888e0fd5e72468ea4534c8fa4678104 : SUCCESS in 1h 18m 12s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/4f2dc1a671fd4a208afcf0ea111a5150 : SUCCESS in 1h 15m 56s","accounts_in_message":[],"_revision_number":8},{"id":"bc445cb329a75105adae1bd90e45ce5ff50ff6fb","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-12-24 10:26:11.000000000","message":"Patch Set 8: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"d4d7b94f02af6cc4dc0853bf0ea58b5be4672c50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"date":"2022-01-03 11:09:13.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"0be0d23b517d1b0939dfce78113f616fc9e7a006","author":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"date":"2022-01-03 11:10:32.000000000","message":"Patch Set 9:\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"fcf6027d1336a92b6e7b412413dcdb6c6d96c361","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 13:51:40.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5681a7391c55411592d3630d54868065 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efa7d4aa8082432185d8aea8e6c7b0dc : SUCCESS in 3m 58s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/00ba7550992d42a591dfe6e88ee2e626 : SUCCESS in 1h 28m 52s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/f021caafa72247f2bcf9bc6d59a7f5c7 : SUCCESS in 1h 31m 22s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/3fac5a138d6f4b2c8aba8d465536de46 : SUCCESS in 1h 41m 39s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/28de347debe54a02be14b11d594c9ad3 : SUCCESS in 1h 40m 15s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/4d85a43140464becb8cc6fb53cf34270 : FAILURE in 1h 43m 20s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/d100bfe5254a4d80acea050028c956f1 : SUCCESS in 1h 16m 49s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/68472b8cd90840fca4e3970cd35e1f46 : SUCCESS in 1h 10m 44s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/4b6ba3d2e81b42eca10c875306eb0b33 : SUCCESS in 1h 49m 25s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/181a7b6280c6450381901db7a2f99d6d : SUCCESS in 1h 24m 35s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/fe0b5df1db5341a4b59e896a37dfb6af : SUCCESS in 1h 56m 20s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/4fcff9f997934f3aa9dd67b2c2779f7a : FAILURE in 2h 38m 26s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/d971f34e25b64d0fab273fdb34f6441d : POST_FAILURE in 39m 28s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/da6d5f1f160240fcad29bc089397fb1f : FAILURE in 1h 35m 42s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/2171efdaa07d440d896af9a51d4c063c : SUCCESS in 1h 14m 18s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/b36a8b49e0f24502a04dfad91ea57e0b : SUCCESS in 1h 15m 44s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/6f46ec020f134543bde8b376d7c6426d : SUCCESS in 51m 45s","accounts_in_message":[],"_revision_number":9},{"id":"abdff7de427302be7d1cb8bf6ef8a500e8d413ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"date":"2022-01-03 15:43:37.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"121b208dbd9fcc636c11675d0c24e0df74720a47","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2022-01-03 16:01:13.000000000","message":"Patch Set 10: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"de3a8807333877c15522d61a8b8369fd0a232abf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 18:15:06.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7beb910403354ddf8d02af005eb55c61 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36e5f96b84a1444f8e49744c002d87ab : SUCCESS in 3m 28s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/306b623622a146c6a55cdf5053db2339 : SUCCESS in 1h 37m 32s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/3a7878adb8e04b2192d77126cb7743ec : SUCCESS in 1h 39m 13s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/d61383ba97a04855a56ab466f44f5bf5 : SUCCESS in 1h 56m 14s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/be15ef389c21423ab0e3387ca5948538 : SUCCESS in 1h 22m 19s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/a406faee1a264f3db5ae91b7aceb5bcb : SUCCESS in 1h 51m 03s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/a4cc78867c1149a3830748b20b6ee818 : SUCCESS in 1h 45m 13s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/b07b6b853c3846a08c42048b6d3e95b8 : SUCCESS in 1h 20m 22s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/c0b0d91c0700449fa3d81ac4d017b7b9 : FAILURE in 1h 42m 47s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/5a31badd3fd94dfc98aaaa9a9f979eb9 : SUCCESS in 1h 28m 01s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/b7444757f58d4cdebdd5c64b9215f75f : FAILURE in 1h 49m 16s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/2600b655532c464795ff2dc273e9b3e1 : SUCCESS in 2h 29m 17s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/012a9fbd95aa45689c0977a2dcae4b02 : POST_FAILURE in 48m 07s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/3532c47c8f0944f184ac0e4a7f741854 : SUCCESS in 1h 05m 35s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/481b187189fd4c83b6cc6c6db9032856 : SUCCESS in 1h 00m 19s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/ff7023d8b08443adaf772a9f391720ee : SUCCESS in 1h 35m 59s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/f897bb16cac3434780a21c33bc040147 : SUCCESS in 1h 04m 54s","accounts_in_message":[],"_revision_number":10},{"id":"491fd940fd3ecdc559bcd89231133c934865dfbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"date":"2022-01-05 13:09:17.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"ca352e47048ee6a45aaec0e4492b9e9a3ca33be5","author":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"date":"2022-01-05 13:11:31.000000000","message":"Patch Set 11:\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"02229b12f3d56b20f0c8958d6de40f980c0778e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 15:27:41.000000000","message":"Patch Set 11: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47eb7cd9041d4b859cf4c39a8e9a6ead : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/701dde08f0804358aabdebb46032bfe4 : FAILURE in 5m 07s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/2f9c73191d3a40f78fed06ffa66a50a9 : FAILURE in 1h 41m 40s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/184e3edb662649f3beb843218a42936d : FAILURE in 1h 50m 05s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/a432846d89ba40ff893e87a7fc4d72c2 : FAILURE in 2h 05m 40s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/42daa2afb40041d1ac071ec062add26d : SUCCESS in 1h 39m 39s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/bec373e88c514ac6b1a3b8f72fe463b0 : FAILURE in 1h 49m 07s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/cdda1987b44b45d7a011bcc1db710a59 : SUCCESS in 1h 23m 29s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/1688ba2ea82c41b6b4860946a2245ac6 : SUCCESS in 1h 17m 18s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/0a66380e2c464d11ba07f00756f9868a : FAILURE in 1h 37m 25s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/73e4d7c29031412399545d6a548cd574 : FAILURE in 1h 31m 27s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/a0e3be0a57684b298f657666b41f08c2 : FAILURE in 2h 11m 33s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/57bed6c1e0d545a880152a87047f4fed : FAILURE in 1h 46m 57s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/99f21d4162b34604b9b1ae29f3ddc1f5 : POST_FAILURE in 1h 00m 03s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/558f8ca02782494da6dff4ed45bc6d1a : FAILURE in 1h 45m 51s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/265712e02f5f45a48a5c2d1bc4355e9b : SUCCESS in 1h 18m 36s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/95c290f7fe3746719f28b94e954d888d : SUCCESS in 1h 11m 14s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/bd81425b0eb14ccf9cf3d2477eaede8e : SUCCESS in 1h 12m 11s","accounts_in_message":[],"_revision_number":11},{"id":"3ca2dfd3bbb0f9d9b4b7ae41d9f025ab2c54b38f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"date":"2022-01-10 15:37:27.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"d14f12517f3f0ccdd6d9a64a00ffbed48144233e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"date":"2022-01-10 15:54:03.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"0b1c2349d2b2680d5be5d318069f261cb44eaa63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 17:57:31.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60c31afec28545e8910331057aeb9044 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f37de54d54794a4a83a8e3bd7dfa2058 : SUCCESS in 5m 25s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/3f464c2daa854aa0add88288e8f22f32 : SUCCESS in 1h 33m 50s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/37a7d4c214f64b9fb7041a39b7bb6fb2 : FAILURE in 1h 32m 14s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/5ff2e409a72c45378e80592aef7e1689 : FAILURE in 1h 52m 10s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/bb48c17017e74562b355bdb6253f9ecc : SUCCESS in 1h 57m 07s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/1148451d5db248c79e64317b26c376d6 : SUCCESS in 1h 44m 50s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/9ce7941c12584f2296b709bf72ccbc61 : SUCCESS in 1h 26m 37s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/9a167096f0054f06b9b096c6631197c7 : SUCCESS in 1h 18m 31s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/140ea2df3142431ca68b3e69fc0007e8 : SUCCESS in 1h 32m 07s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/866d258e6efc4b8d8e9076f9908ff51e : SUCCESS in 1h 47m 49s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/c274964aae4f4128a98d1032e64c3dad : FAILURE in 1h 27m 55s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/60a8333e99c9427c8f5eab34aadebe4c : SUCCESS in 1h 49m 26s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/b4dc50492536444892d55ad347d17b13 : POST_FAILURE in 55m 31s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/bcf2b6b728ff4aaaa13b034d4c4141d7 : SUCCESS in 1h 57m 09s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/3e65c2bc0a1a4c65b13c8398bc103201 : SUCCESS in 1h 24m 18s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/1621b479f92040248beca603f050759b : SUCCESS in 1h 12m 03s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/79df9a78ad404848be7cbd243a2bc6f2 : SUCCESS in 1h 18m 01s","accounts_in_message":[],"_revision_number":13},{"id":"bf78b54795714f358dd58b19e0b8257129e3cd3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"date":"2022-01-11 17:37:49.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"2a0c4b85c9feb8f3f6c1d869478e81a804d91139","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 19:44:21.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d8533531da8493cafe3b137013c883c : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d08d33e1ba74258828dc9007db3d2ac : SUCCESS in 3m 40s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/91a01da1a8134563b0a7c0b40d6fd523 : SUCCESS in 1h 43m 05s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/1b8d1ddecf2c42cab45705410b5766ec : SUCCESS in 1h 54m 57s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/59156a77651b429a9271697becc02adb : SUCCESS in 1h 51m 08s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/d9f77e320c61433ebcadc78745d672ca : SUCCESS in 1h 37m 02s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/c5fed6b054314b42b2b78c6874338fe6 : SUCCESS in 1h 29m 48s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/1e4cb23de8cc4c49824c272238001d4b : FAILURE in 32m 44s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/4d008bc080554510a08b6bed8c8e9bb4 : SUCCESS in 1h 12m 19s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/e762c340fba8474287be724511fdb537 : SUCCESS in 1h 26m 12s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/b163755c9b8945508041442d694189da : SUCCESS in 1h 32m 21s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/fc69675a809542b9803907a925750e36 : FAILURE in 2h 00m 00s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/d6e4ef03d7d24128b40f9bf4a8dd9c88 : SUCCESS in 1h 43m 56s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/aa715477a9574b87a1c7b275855aa392 : POST_FAILURE in 56m 44s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/a2855912e8d84d1085c2fd6c9115e09c : FAILURE in 1h 28m 50s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/d22307a3420a48d78c6225637e1a8bc7 : SUCCESS in 1h 08m 22s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/83185bce8e3f424e8dd57d7f56982f44 : SUCCESS in 1h 12m 10s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/1b338be203134e4e8711f66415620350 : SUCCESS in 1h 31m 54s","accounts_in_message":[],"_revision_number":14},{"id":"aaa9a56f606f15cf84b24806df2f85d0345bf832","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"date":"2022-01-11 21:23:53.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"79a2815d5e8acd4e2b5722b5630cd4b348ee3d1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 00:03:03.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa3a355409124076a1363e6cf1e88d39 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76bd61ab54454158b9fc2e239333b5ad : SUCCESS in 4m 31s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/7bb241f8a23e43dc8452feaf9174cf95 : SUCCESS in 1h 59m 37s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/3f8d98ac0c4844f8a05c60c8e981ffac : SUCCESS in 1h 25m 37s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/310f919681514cd08800b7a3ee5dd218 : FAILURE in 2h 10m 26s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/b8cc09ddcf834ae3b6be452526e30b8c : SUCCESS in 1h 26m 41s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/001c76d2437c451f9f2cff8ddbc0fd70 : SUCCESS in 1h 38m 18s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/88339a4ebaa54c919446fbe04f6c9e1e : SUCCESS in 1h 21m 40s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/58443686336148128b1f6ecae4a867fe : SUCCESS in 2h 02m 56s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/ea379b5f1b984e59b2ef6e9e660eb49b : SUCCESS in 1h 33m 33s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/b7c328ff9d4243149e486bc93fd208ca : SUCCESS in 1h 51m 22s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/d1244209580f4ac9b4d74b3524c518de : SUCCESS in 2h 35m 50s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/296eed7a38c0492eaf92e3b1b7e7a8cc : SUCCESS in 2h 17m 06s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/0cc443f1e7f74511b1b0bd4bc9b072bf : POST_FAILURE in 56m 05s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/b1a695bdf88e425aa443bfebf0f01d6b : SUCCESS in 1h 39m 25s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/ab7e9947a5a64a94bab997edc37b9b0f : SUCCESS in 1h 10m 01s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/98c39c585d6b4b34b2024c6309eb541c : SUCCESS in 1h 12m 17s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/7159314c1a5242f8953b30b2ee9646a1 : SUCCESS in 1h 24m 00s","accounts_in_message":[],"_revision_number":15},{"id":"031e778a1f98a20b2e54b301e412016e08858418","author":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"date":"2022-01-12 06:40:04.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"f4dcf1fd21a9b6d903925fcf924576f4356bf032","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 09:36:07.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2c8fb4b107640dcbdd6b0aa598439e4 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/135c971f98ad48b7b30ed3d3f8dedd12 : SUCCESS in 3m 23s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/ce39131c027541dd9217737f7ee6ef2e : SUCCESS in 2h 13m 21s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/cfc273be04354190ad70190aea4f46f2 : SUCCESS in 1h 52m 59s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/5ee1090ac3ea467493c7016ec4e37ed9 : SUCCESS in 1h 51m 13s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/a2673f83ef724684bffe1c27bc7003ad : SUCCESS in 1h 29m 40s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/56f4ed3ab08a4e91bbc90e8dfec8edd5 : SUCCESS in 1h 40m 43s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/12e0dbafed2246ff90d5cbbfe9b40b82 : SUCCESS in 1h 12m 36s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/6c7cc68186f9449d80899c53d308e01f : SUCCESS in 1h 07m 17s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/4f9e33f799e24edca0999d7408d5df87 : FAILURE in 1h 19m 04s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/64d90b3fe4024abb855b7c6cad677ba7 : SUCCESS in 1h 31m 27s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/3e9896dd9b124102b52ae715aa10a758 : SUCCESS in 1h 43m 03s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/34a2030bbafe4c6cae8d46e46103aadc : SUCCESS in 2h 51m 41s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/03818924c423419abf08380c875e907a : POST_FAILURE in 56m 57s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/0e669440b0064d3ca13930ab5496d646 : FAILURE in 1h 56m 47s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/bbc0e905ac11491eb9a0bb9c2ae47e9a : SUCCESS in 1h 07m 00s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/2680c107f458478f9563f38809ce4782 : SUCCESS in 1h 21m 10s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/f331ceda6dc24df29f17d186291da0c8 : SUCCESS in 1h 09m 48s","accounts_in_message":[],"_revision_number":15},{"id":"b8c4e4ef1dedc5311c478abdfea25732a402fc4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"date":"2022-01-12 14:55:11.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"1d12ed9e4e5aba14ba835c84e048891748c47d8e","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-01-12 17:20:30.000000000","message":"Patch Set 16:\n\n(2 comments)","accounts_in_message":[],"_revision_number":16},{"id":"7177c0a14ff25d2565c09b7e06762e7d86713da8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 17:47:39.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/061f024d19324c53939054e9ab7d6b2f : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/048f8508c91347b391455ef2c7868cbd : SUCCESS in 5m 02s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/6bf77787e3e148f78bdbc35345a1edc1 : SUCCESS in 2h 11m 59s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/74e3fb576c8e49959961e6db7692a15c : SUCCESS in 1h 50m 40s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/f1bbee9216f44f5a945d8220e7b3950b : FAILURE in 2h 40m 40s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/17cf43e1a0df496c81c4a6e9c3f88544 : SUCCESS in 1h 30m 40s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/a00f89b1fb3f41c88fb606a3398f520c : SUCCESS in 1h 42m 41s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/986b946cdd254447bcfa8d66a5147c72 : SUCCESS in 1h 26m 39s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/5a7d8809bd4046bf9a4d02db761d0a6b : SUCCESS in 1h 19m 09s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/d6b595e5e19f44b682daa270c106a811 : SUCCESS in 1h 44m 46s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/711c6263f37b4ab59a0736c6daf8446e : SUCCESS in 1h 39m 05s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/e5eb3ec47de445838ecfa0ac878bc8a9 : FAILURE in 1h 43m 33s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/a1392a6e3aae43a6bf3d1305aa257591 : FAILURE in 1h 38m 59s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/ccccf27fb28f47e5bd49751c5cb34477 : POST_FAILURE in 54m 52s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/e720ea640eda499c9792c683de368179 : SUCCESS in 1h 46m 26s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/14e68a480c2249f486d579ff6d85295b : SUCCESS in 1h 23m 40s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/4daaebcef65b468fb7b73e1debc7c3c1 : SUCCESS in 1h 18m 05s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/4e4e8cfb737147168982d676891b4cdc : SUCCESS in 1h 20m 20s","accounts_in_message":[],"_revision_number":16},{"id":"4569bd56581526c13fd9a96cf98275174ffb20cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"date":"2022-01-13 12:52:00.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"07beb1cbcd3ca63dc43fdb7947e276c6aa5dee97","author":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"date":"2022-01-13 12:53:31.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"405b59f99bd91d88f03a82f108aeefda99525fbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 15:13:03.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8550759ca1b7490880aafbf37d59e8d8 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45390197ad7e4c0498c11960d3ec5dba : SUCCESS in 4m 00s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/e54b62908586467d8e3074c5a8e88e76 : FAILURE in 1h 24m 49s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/25aea4cba81b4a559650615f990ca3bf : SUCCESS in 1h 49m 59s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/b146625ddda24f6c9b3a6cffb0099f1a : FAILURE in 1h 54m 16s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/377178725f49445fbcea031c589d073e : SUCCESS in 1h 46m 07s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/316e69bc2b0940c1827f4601e1b577a2 : SUCCESS in 1h 32m 11s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/07b620bad9f145d9a75cd4b49ab21fcc : SUCCESS in 1h 25m 17s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/37b4438fe7c14c8c864c245f335f3e21 : SUCCESS in 1h 15m 49s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/62eef7d1c59d410488058b63a79508a3 : SUCCESS in 2h 09m 24s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/09506643a5304db680f54afa604b6ff4 : SUCCESS in 1h 30m 58s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/8d1d6169ac4641c59414f07ab9294e49 : SUCCESS in 1h 29m 22s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/348e1fe3061e400faef855aaa916f394 : FAILURE in 1h 24m 15s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/4a515b3c9eed414996acee4df0bad555 : POST_FAILURE in 58m 09s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/b385878334c1465e963680b025526f15 : SUCCESS in 1h 19m 25s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/d10b7f0abb70442f81839db67772ef8a : SUCCESS in 1h 23m 48s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/3fb15c8f416243a1a2b1a164ddef7aa1 : SUCCESS in 1h 01m 00s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/681639e04ecb4516914da3d20c268eb2 : SUCCESS in 1h 11m 14s","accounts_in_message":[],"_revision_number":17},{"id":"901c6895de63d36aac71f384664dbe5c47352ac8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"date":"2022-01-13 18:13:34.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"f561e2107eba91111693f55ab7023ce422d4e2c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 20:18:26.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b90a643a8a39433380f682bbf878b59c : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80db29b46852447d94eac523883f6fad : SUCCESS in 4m 44s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/4421fbd999b54e83a95206b871e66ad4 : FAILURE in 1h 23m 21s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/5f2320a5ad4243cbba77077108e86418 : SUCCESS in 1h 30m 16s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/3c91b0513a654d759716719c07841aed : SUCCESS in 1h 52m 50s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/3318dd314bfa403d934b6c268e61ee91 : SUCCESS in 1h 16m 27s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/6219743635274259b8bb8abce5febc9f : SUCCESS in 1h 44m 32s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/9f7c0d365fd548d59f09594510315249 : SUCCESS in 1h 22m 02s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/63e4e62d1dda45ffa969375780683d42 : SUCCESS in 1h 14m 31s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/44543b7d73fc44399ab354f2ccf531ba : SUCCESS in 1h 52m 44s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/07129ed5f300479bbc35360404edeafa : SUCCESS in 1h 21m 59s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/e3a2a69870eb4117a5cd1bd35948db61 : FAILURE in 1h 20m 48s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/9014602ad17e47e2bd85de51ef7d6d6a : SUCCESS in 1h 35m 27s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/411d13a8491447a4a38aff0e696a93a7 : POST_FAILURE in 52m 22s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/43927e320cf04e02a13a1fc32c452a93 : SUCCESS in 1h 40m 00s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/3c266d5d864e48cbb3a68526540ed9a9 : SUCCESS in 1h 12m 36s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/6ed3f40e9da146deb463fa2579a2af72 : SUCCESS in 1h 12m 24s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/857c597543fb4eebb08cfaa045dcdb59 : SUCCESS in 1h 12m 51s","accounts_in_message":[],"_revision_number":18},{"id":"cb8bc183f6b6804616820d302d562ac7f83b0b52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"date":"2022-01-14 18:24:17.000000000","message":"Patch Set 19: Patch Set 18 was rebased","accounts_in_message":[],"_revision_number":19},{"id":"48df08688f7de66170ed5d5b6f674ea0d2816c87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 21:21:38.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8fbf4b17429414f908ce55052129f61 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a3bd6a4c7cf41e6b675120263f5e407 : SUCCESS in 4m 52s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/953f5179dc654bf7bfb93877b0ff9233 : SUCCESS in 1h 20m 00s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/3a2deebb208846fc86e90f6dea36b12f : SUCCESS in 1h 37m 50s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/f8df78dc2eae4e239056988d82ff752c : FAILURE in 1h 29m 48s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/e0ea73c87820434183d94079404240d3 : SUCCESS in 1h 26m 48s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/b32a1c97abf8415887e6260ddb73b74f : SUCCESS in 1h 18m 22s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/ee51e29dea09451ca3a85a894cc77464 : SUCCESS in 1h 07m 40s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/9b22a1b3245145b7877f5488f17b8d8e : SUCCESS in 1h 17m 04s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/db8d2f1a42b740c0862470399c2925b2 : SUCCESS in 1h 37m 47s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/5209b160a53345ef90c4a2359c648c99 : SUCCESS in 1h 47m 24s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/acb03f0e417d426ca3ceb9ca0c3d267b : FAILURE in 1h 23m 13s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/fe209e0c353343d29152be7dcc63ebad : SUCCESS in 1h 43m 09s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/a5f2c0b1407b4e62bf9e3a4a57e636c3 : POST_FAILURE in 2h 52m 31s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/8c4eb6e2546a4ea8a0a83d7c727ee231 : FAILURE in 1h 44m 50s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/50c36dce9714476b888fbb375383d279 : SUCCESS in 1h 04m 12s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/13d2b206081f4f4b8b1b9ce01b73e767 : SUCCESS in 54m 40s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/9b5ec0c6cde84a029f2bde76689e8b8e : SUCCESS in 1h 20m 04s","accounts_in_message":[],"_revision_number":19},{"id":"427ad464669d00a6c4cbcda9229b21dfc8302c04","author":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"date":"2022-01-15 10:01:33.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"7a2fcf973599c0fa64a06203dae3ef944f3bb470","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-15 12:38:25.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c709869d20de44948dc1f93f3a972cc8 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/216bc562d6614a2bb27c7cf3b4e0e8bc : SUCCESS in 3m 01s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/477d8387584440e5a5119da7ad2e9b0e : SUCCESS in 1h 15m 54s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/395824600c9640eb884b615e2502390c : SUCCESS in 1h 26m 21s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/bf37c604ddb746fdb78e295473a09070 : FAILURE in 1h 35m 43s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/b46e34c74be74703b2ce9f34e57ef811 : SUCCESS in 1h 11m 37s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/792f7080d2d041b58c6fb8da471672fd : SUCCESS in 1h 19m 54s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/487c8bd3068943abb086364cdaec8dfc : SUCCESS in 1h 18m 16s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/0fcfcd559ba442a190dbe572ace5384c : SUCCESS in 1h 04m 11s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/9a5b7d89781142c1962cc3f67f22f203 : SUCCESS in 1h 16m 33s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/3ce4a71a3e5a47f6971e1f0733581a97 : SUCCESS in 1h 25m 29s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/f32a910e41334697a2625d67114d61bd : FAILURE in 1h 25m 02s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/e7d5af226d074d42a78497dd8ccc183c : FAILURE in 1h 40m 27s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/ef4f7d0285324b08a156c3a092467e6d : POST_FAILURE in 2h 35m 31s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/b3eba783f4664e8bbd8113b0ceba8648 : SUCCESS in 1h 19m 53s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/19bf84fab1064100a886166ed86f08fc : SUCCESS in 1h 13m 06s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/7a5853e17c5845f38ff3d55f8114f2ec : SUCCESS in 1h 02m 22s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/f5a8a2512bd24a33a32e093d09d89787 : SUCCESS in 50m 49s","accounts_in_message":[],"_revision_number":19},{"id":"471dd7fa4dcbb7cbbd7ba0eee2f5a4f943f16319","author":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"date":"2022-01-15 19:14:02.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"11a13c9347db7b92ed22dc49d1b0175eeea8d6db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-15 21:34:27.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f71ea350dbf04c29b31837971e5f5c08 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20e5ab03378c4e65b2bf985ee4a7c7ad : SUCCESS in 3m 25s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/0a1a74668d674411841a41fe5537d709 : SUCCESS in 1h 43m 48s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/1d36d4261a4c4370ac7c405103f57dc4 : SUCCESS in 1h 18m 23s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/1cce287c68f5426ea2afe183179e2366 : FAILURE in 1h 59m 40s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/f5ce67fbe25b4ccd87288fbbf03589b7 : SUCCESS in 1h 14m 37s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/479a283def21433c831545ef4aceae32 : SUCCESS in 1h 31m 49s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/a09903c240a84c19980c34a4758b0caf : SUCCESS in 1h 29m 57s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/0aef902c1e22437aa72a139ac774a91b : SUCCESS in 1h 05m 54s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/42f57bd931cf42f38e385eec1c4f17ba : SUCCESS in 1h 54m 39s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/dc6e1f0e63974cd5a3b7dd5b8e1d8220 : FAILURE in 1h 17m 23s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/87edcc81b75b48448b95517003bc045f : FAILURE in 2h 15m 13s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/d0f894a8193f481dbc8fb6cde655f322 : SUCCESS in 1h 48m 53s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/db27c84a9a17472d8fc863cf07a9cf1f : POST_FAILURE in 2h 17m 55s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/f880322af2f241a9b84332dffb80524c : SUCCESS in 1h 33m 18s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/e8986cc811f142fdbbdaf70adc65f0c0 : SUCCESS in 1h 02m 48s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/40ec77a556cf4925b8d217322bebbe34 : SUCCESS in 1h 05m 54s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/566b1c76a47147fcb50f632acd263c7c : SUCCESS in 1h 06m 40s","accounts_in_message":[],"_revision_number":19},{"id":"97fe07c27b90c960a1c436b55e9c898aec5e972e","author":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"date":"2022-01-17 08:14:35.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"7ce18f739ed006ca2bc38ef2d9dc7a14b62f95b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 10:56:40.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8109ed6279c4628b3eb18f582a0efc8 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a3c8a851d4f49b48835ec0ebec5af8f : SUCCESS in 3m 31s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/9d57b4ff816944a5ba756090a8f8cd7d : SUCCESS in 1h 39m 40s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/e2359679f66c4183a8c0a34d152c9f96 : SUCCESS in 1h 41m 26s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/9b54ab0f76b44d15b028c3fda144bd6f : SUCCESS in 1h 59m 53s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/16e9b136b5b6456fa282f197843b9450 : SUCCESS in 1h 27m 20s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/ed044b37a0e843e2bac3f8428c07cfd3 : SUCCESS in 1h 36m 09s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/9994a602bdf44dff808247c56d4ad60b : SUCCESS in 1h 26m 30s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/c0400c1924ea47cfb28068a5b0ae39a6 : SUCCESS in 1h 17m 36s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/13fd3b93fb0b46d3b3e6b5be7513f82c : SUCCESS in 1h 33m 32s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/33f66acd374b4f10bc30304cbda23c39 : FAILURE in 2h 06m 21s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/815d268f9dfa43489ef3f267bf7ccc32 : SUCCESS in 1h 17m 13s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/d9fc1f713a6c418c80d98acfd7da5135 : SUCCESS in 1h 27m 33s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/2a0935fd3ea945079f5cfc0b2ac3ce4f : POST_FAILURE in 2h 39m 04s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/3ee24ace43d846bf8cfdb61cf7780213 : SUCCESS in 1h 33m 16s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/773e33aaf6454423b9a34abb85c83ca1 : SUCCESS in 1h 14m 15s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/cb214742584d48d0859a652333aec6c9 : SUCCESS in 58m 34s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/d08f3dcfb0414cac80cd0a8e2ab66d71 : SUCCESS in 1h 01m 45s","accounts_in_message":[],"_revision_number":19},{"id":"82d35c65e3e687b17a13146e7faffec963645326","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-01-19 09:30:57.000000000","message":"Patch Set 19: Code-Review+2","accounts_in_message":[],"_revision_number":19},{"id":"89b1c1dcd8e1c7248b1da8de9c596a4547bdde31","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2022-01-24 09:29:58.000000000","message":"Patch Set 19: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":19},{"id":"5c0062710710e852e34db9d1dfd185a957b27258","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 09:30:41.000000000","message":"Patch Set 19: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":19},{"id":"ecbedd7c3b41b656e668114e67af57e4cfd983c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 11:06:37.000000000","message":"Patch Set 19: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18c1142aab814a7abff41ec43e93ba36 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d464670b3af7491d82d2269331b8e408 : SUCCESS in 4m 27s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/7a05500ddbd44e8a9170dc44b2f8071a : SUCCESS in 1h 34m 09s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/b03034f9cf4b4d8a9964fb732642a299 : SUCCESS in 1h 17m 05s","accounts_in_message":[],"_revision_number":19},{"id":"a67ee39ec36b84fa1d8a0f3b055d4f5c0ccc51a6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 11:06:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":19},{"id":"0e4b108c4e7c5d28070995823bbffafbeb7288c3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 11:09:40.000000000","message":"Patch Set 19:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dffed04cff246178ae2b53f14e45040 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":19}],"current_revision_number":19,"current_revision":"2ffdd2f47560087a242a3731f831dfb51ed2b149","revisions":{"07ec0de1db88c290233b8aa79c9b556b47bdd212":{"kind":"REWORK","_number":1,"created":"2021-12-21 08:57:50.000000000","uploader":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"ref":"refs/changes/09/822409/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/09/822409/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/1"}}},"commit":{"parents":[{"commit":"3d79470f23b51d37c26e2e6178bf402c946cbde1","subject":"Change get_readiness_state() return type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/3d79470f23b51d37c26e2e6178bf402c946cbde1"}]}],"author":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-21 08:51:53.000000000","tz":0},"committer":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-21 08:55:14.000000000","tz":0},"subject":"Add test_service_with_not_ready_endpoints test","message":"Add test_service_with_not_ready_endpoints test\n\nThis test case automates the issue described in\nBZ https://bugzilla.redhat.com/show_bug.cgi?id\u003d1980957.\n\nThe new test case:\n - creates a deployment with a dummy readiness probe so all\n   the pods are not ready\n - creates a service for the deployment (where the endpoints\n   are not ready)\n - it checks Kuryr pods are not restarted\n\nChange-Id: Id19f85625901b35900f9441aace8fe8938cb1d68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/07ec0de1db88c290233b8aa79c9b556b47bdd212"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/07ec0de1db88c290233b8aa79c9b556b47bdd212"}]},"branch":"refs/heads/master"},"b4b28f0a18615c6328511beaacedc00c0d6e9a2c":{"kind":"REWORK","_number":2,"created":"2021-12-21 08:59:57.000000000","uploader":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"ref":"refs/changes/09/822409/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/09/822409/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/2"}}},"commit":{"parents":[{"commit":"3d79470f23b51d37c26e2e6178bf402c946cbde1","subject":"Change get_readiness_state() return type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/3d79470f23b51d37c26e2e6178bf402c946cbde1"}]}],"author":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-21 08:51:53.000000000","tz":0},"committer":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-21 08:58:49.000000000","tz":0},"subject":"Add test_service_with_not_ready_endpoints test","message":"Add test_service_with_not_ready_endpoints test\n\nThis test case automates the issue described in\nBZ https://bugzilla.redhat.com/show_bug.cgi?id\u003d1980957.\n\nThe new test case:\n - creates a deployment with a dummy readiness probe so all\n   the pods are not ready\n - creates a service for the deployment (where the endpoints\n   are not ready)\n - it checks Kuryr pods are not restarted\n\nChange-Id: Id19f85625901b35900f9441aace8fe8938cb1d68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/b4b28f0a18615c6328511beaacedc00c0d6e9a2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/b4b28f0a18615c6328511beaacedc00c0d6e9a2c"}]},"branch":"refs/heads/master"},"a7276cc3919e0641968a66f276856e579629a91d":{"kind":"REWORK","_number":3,"created":"2021-12-21 15:54:09.000000000","uploader":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"ref":"refs/changes/09/822409/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/09/822409/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/3"}}},"commit":{"parents":[{"commit":"3d79470f23b51d37c26e2e6178bf402c946cbde1","subject":"Change get_readiness_state() return type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/3d79470f23b51d37c26e2e6178bf402c946cbde1"}]}],"author":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-21 08:51:53.000000000","tz":0},"committer":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-21 15:52:55.000000000","tz":0},"subject":"Add test_service_with_not_ready_endpoints test","message":"Add test_service_with_not_ready_endpoints test\n\nThis test case automates the issue described in\nBZ https://bugzilla.redhat.com/show_bug.cgi?id\u003d1980957.\n\nThe new test case:\n - creates a deployment with a dummy readiness probe so all\n   the pods are not ready\n - creates a service for the deployment (where the endpoints\n   are not ready)\n - it checks Kuryr pods are not restarted\n\nChange-Id: Id19f85625901b35900f9441aace8fe8938cb1d68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/a7276cc3919e0641968a66f276856e579629a91d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/a7276cc3919e0641968a66f276856e579629a91d"}]},"branch":"refs/heads/master"},"bcf42eb852ac4e8ed8c622fd5804979dff4a716e":{"kind":"REWORK","_number":4,"created":"2021-12-21 22:49:36.000000000","uploader":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"ref":"refs/changes/09/822409/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/09/822409/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/4"}}},"commit":{"parents":[{"commit":"3d79470f23b51d37c26e2e6178bf402c946cbde1","subject":"Change get_readiness_state() return type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/3d79470f23b51d37c26e2e6178bf402c946cbde1"}]}],"author":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-21 08:51:53.000000000","tz":0},"committer":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-21 22:42:45.000000000","tz":0},"subject":"Add test_service_with_not_ready_endpoints test","message":"Add test_service_with_not_ready_endpoints test\n\nThis test case automates the issue described in\nBZ https://bugzilla.redhat.com/show_bug.cgi?id\u003d1980957.\n\nThe new test case:\n - creates a deployment with a dummy readiness probe so all\n   the pods are not ready\n - creates a service for the deployment (where the endpoints\n   are not ready)\n - it checks Kuryr pods are not restarted\n\nChange-Id: Id19f85625901b35900f9441aace8fe8938cb1d68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/bcf42eb852ac4e8ed8c622fd5804979dff4a716e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/bcf42eb852ac4e8ed8c622fd5804979dff4a716e"}]},"branch":"refs/heads/master"},"7ebf510af3c8239eef0da9fbfec2016c49683d47":{"kind":"REWORK","_number":5,"created":"2021-12-22 10:27:18.000000000","uploader":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"ref":"refs/changes/09/822409/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/09/822409/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/5"}}},"commit":{"parents":[{"commit":"a57c4ff178cd61b5c18c2688daed8ba454fa9421","subject":"Change get_readiness_state() return type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/a57c4ff178cd61b5c18c2688daed8ba454fa9421"}]}],"author":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-21 08:51:53.000000000","tz":0},"committer":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-22 10:26:04.000000000","tz":0},"subject":"Add test_service_with_not_ready_endpoints test","message":"Add test_service_with_not_ready_endpoints test\n\nThis test case automates the issue described in\nBZ https://bugzilla.redhat.com/show_bug.cgi?id\u003d1980957.\n\nThe new test case:\n - creates a deployment with a dummy readiness probe so all\n   the pods are not ready\n - creates a service for the deployment (where the endpoints\n   are not ready)\n - it checks Kuryr pods are not restarted\n\nChange-Id: Id19f85625901b35900f9441aace8fe8938cb1d68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/7ebf510af3c8239eef0da9fbfec2016c49683d47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/7ebf510af3c8239eef0da9fbfec2016c49683d47"}]},"branch":"refs/heads/master"},"aa0dee47ee206213ce04031f3768a431e3c4e5dd":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-12-22 10:59:05.000000000","uploader":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"ref":"refs/changes/09/822409/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/09/822409/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/6"}}},"commit":{"parents":[{"commit":"c94e308f222e68b977a98bfc0b692ee908d69eed","subject":"Change get_readiness_state() return type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c94e308f222e68b977a98bfc0b692ee908d69eed"}]}],"author":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-21 08:51:53.000000000","tz":0},"committer":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-22 10:58:05.000000000","tz":0},"subject":"Add test_service_with_not_ready_endpoints test","message":"Add test_service_with_not_ready_endpoints test\n\nThis test case automates the issue described in\nBZ https://bugzilla.redhat.com/show_bug.cgi?id\u003d1980957.\n\nThe new test case:\n - creates a deployment with a dummy readiness probe so all\n   the pods are not ready\n - creates a service for the deployment (where the endpoints\n   are not ready)\n - it checks Kuryr pods are not restarted\n\nChange-Id: Id19f85625901b35900f9441aace8fe8938cb1d68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/aa0dee47ee206213ce04031f3768a431e3c4e5dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/aa0dee47ee206213ce04031f3768a431e3c4e5dd"}]},"branch":"refs/heads/master"},"5fe5c8aba8504419df0f4b85444f459ededd778d":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-12-22 22:54:08.000000000","uploader":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"ref":"refs/changes/09/822409/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/09/822409/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/7"}}},"commit":{"parents":[{"commit":"8dbe6be09c5f5bd3e2872db1088cd628ba70939c","subject":"Change get_readiness_state() return type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8dbe6be09c5f5bd3e2872db1088cd628ba70939c"}]}],"author":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-21 08:51:53.000000000","tz":0},"committer":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-22 22:52:55.000000000","tz":0},"subject":"Add test_service_with_not_ready_endpoints test","message":"Add test_service_with_not_ready_endpoints test\n\nThis test case automates the issue described in\nBZ https://bugzilla.redhat.com/show_bug.cgi?id\u003d1980957.\n\nThe new test case:\n - creates a deployment with a dummy readiness probe so all\n   the pods are not ready\n - creates a service for the deployment (where the endpoints\n   are not ready)\n - it checks Kuryr pods are not restarted\n\nChange-Id: Id19f85625901b35900f9441aace8fe8938cb1d68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/5fe5c8aba8504419df0f4b85444f459ededd778d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/5fe5c8aba8504419df0f4b85444f459ededd778d"}]},"branch":"refs/heads/master"},"67874b1f92059792a0e33c87897280914bad8dc2":{"kind":"REWORK","_number":8,"created":"2021-12-23 15:11:24.000000000","uploader":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"ref":"refs/changes/09/822409/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/09/822409/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/8"}}},"commit":{"parents":[{"commit":"8dbe6be09c5f5bd3e2872db1088cd628ba70939c","subject":"Change get_readiness_state() return type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8dbe6be09c5f5bd3e2872db1088cd628ba70939c"}]}],"author":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-21 08:51:53.000000000","tz":0},"committer":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-23 15:06:56.000000000","tz":0},"subject":"Add test_service_with_not_ready_endpoints test","message":"Add test_service_with_not_ready_endpoints test\n\nThis test case automates the issue described in\nBZ https://bugzilla.redhat.com/show_bug.cgi?id\u003d1980957.\n\nThe new test case:\n - creates a deployment with a dummy readiness probe so all\n   the pods are not ready\n - creates a service for the deployment (where the endpoints\n   are not ready)\n - it checks Kuryr pods are not restarted\n\nChange-Id: Id19f85625901b35900f9441aace8fe8938cb1d68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/67874b1f92059792a0e33c87897280914bad8dc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/67874b1f92059792a0e33c87897280914bad8dc2"}]},"branch":"refs/heads/master"},"5a62cea2574c72ca435f4a75b21cc8ca13d30a5b":{"kind":"REWORK","_number":9,"created":"2022-01-03 11:09:13.000000000","uploader":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"ref":"refs/changes/09/822409/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/09/822409/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/9"}}},"commit":{"parents":[{"commit":"8dbe6be09c5f5bd3e2872db1088cd628ba70939c","subject":"Change get_readiness_state() return type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8dbe6be09c5f5bd3e2872db1088cd628ba70939c"}]}],"author":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-21 08:51:53.000000000","tz":0},"committer":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2022-01-03 11:07:14.000000000","tz":0},"subject":"Add test_service_with_not_ready_endpoints test","message":"Add test_service_with_not_ready_endpoints test\n\nThis test case automates the issue described in\nBZ https://bugzilla.redhat.com/show_bug.cgi?id\u003d1980957.\n\nThe new test case:\n - creates a deployment with a failing readiness probe so all\n   the pods are not ready\n - creates a service for the deployment (where the endpoints\n   are not ready)\n - it checks Kuryr pods are not restarted\n\nChange-Id: Id19f85625901b35900f9441aace8fe8938cb1d68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/5a62cea2574c72ca435f4a75b21cc8ca13d30a5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/5a62cea2574c72ca435f4a75b21cc8ca13d30a5b"}]},"branch":"refs/heads/master"},"316dd38b5dcd69cfe62df769462a9ab8c1b22ac9":{"kind":"REWORK","_number":10,"created":"2022-01-03 15:43:37.000000000","uploader":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"ref":"refs/changes/09/822409/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/09/822409/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/10"}}},"commit":{"parents":[{"commit":"8dbe6be09c5f5bd3e2872db1088cd628ba70939c","subject":"Change get_readiness_state() return type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8dbe6be09c5f5bd3e2872db1088cd628ba70939c"}]}],"author":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-21 08:51:53.000000000","tz":0},"committer":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2022-01-03 15:41:39.000000000","tz":0},"subject":"Add test_service_with_not_ready_endpoints test","message":"Add test_service_with_not_ready_endpoints test\n\nThis test case automates the issue described in\nBZ https://bugzilla.redhat.com/show_bug.cgi?id\u003d1980957.\n\nThe new test case:\n - creates a deployment with a failing readiness probe so all\n   the pods are not ready\n - creates a service for the deployment (where the endpoints\n   are not ready)\n - it checks Kuryr pods are not restarted\n\nChange-Id: Id19f85625901b35900f9441aace8fe8938cb1d68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/316dd38b5dcd69cfe62df769462a9ab8c1b22ac9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/316dd38b5dcd69cfe62df769462a9ab8c1b22ac9"}]},"branch":"refs/heads/master"},"3c28a3c789bd91cea05281b72585fe0d9b37c6c0":{"kind":"REWORK","_number":11,"created":"2022-01-05 13:09:17.000000000","uploader":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"ref":"refs/changes/09/822409/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/09/822409/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/11"}}},"commit":{"parents":[{"commit":"8dbe6be09c5f5bd3e2872db1088cd628ba70939c","subject":"Change get_readiness_state() return type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8dbe6be09c5f5bd3e2872db1088cd628ba70939c"}]}],"author":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-21 08:51:53.000000000","tz":0},"committer":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2022-01-05 13:07:12.000000000","tz":0},"subject":"Add test_service_with_not_ready_endpoints test","message":"Add test_service_with_not_ready_endpoints test\n\nThis test case automates the issue described in\nBZ https://bugzilla.redhat.com/show_bug.cgi?id\u003d1980957.\n\nThe new test case:\n - creates a deployment with a failing readiness probe so all\n   the pods are not ready\n - creates a service for the deployment (where the endpoints\n   are not ready)\n - it checks Kuryr pods are not restarted\n\nChange-Id: Id19f85625901b35900f9441aace8fe8938cb1d68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/3c28a3c789bd91cea05281b72585fe0d9b37c6c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/3c28a3c789bd91cea05281b72585fe0d9b37c6c0"}]},"branch":"refs/heads/master"},"a8ba5a7e5e180a260b7e43638ab302cdd6bba89a":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2022-01-10 15:37:27.000000000","uploader":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"ref":"refs/changes/09/822409/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/09/822409/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/12"}}},"commit":{"parents":[{"commit":"804a54584e002df6e1bef118d91e8335a07a7e31","subject":"Merge \"Remove LB_TIMEOUT in favor of lb_build_timeout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/804a54584e002df6e1bef118d91e8335a07a7e31"}]}],"author":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-21 08:51:53.000000000","tz":0},"committer":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2022-01-10 15:35:19.000000000","tz":0},"subject":"Add test_service_with_not_ready_endpoints test","message":"Add test_service_with_not_ready_endpoints test\n\nThis test case automates the issue described in\nBZ https://bugzilla.redhat.com/show_bug.cgi?id\u003d1980957.\n\nThe new test case:\n - creates a deployment with a failing readiness probe so all\n   the pods are not ready\n - creates a service for the deployment (where the endpoints\n   are not ready)\n - it checks Kuryr pods are not restarted\n\nChange-Id: Id19f85625901b35900f9441aace8fe8938cb1d68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/a8ba5a7e5e180a260b7e43638ab302cdd6bba89a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/a8ba5a7e5e180a260b7e43638ab302cdd6bba89a"}]},"branch":"refs/heads/master"},"d77f34f3f3266e30cd6bc3f7643b5058b60fa34a":{"kind":"REWORK","_number":13,"created":"2022-01-10 15:54:03.000000000","uploader":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"ref":"refs/changes/09/822409/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/09/822409/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/13"}}},"commit":{"parents":[{"commit":"804a54584e002df6e1bef118d91e8335a07a7e31","subject":"Merge \"Remove LB_TIMEOUT in favor of lb_build_timeout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/804a54584e002df6e1bef118d91e8335a07a7e31"}]}],"author":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-21 08:51:53.000000000","tz":0},"committer":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2022-01-10 15:51:51.000000000","tz":0},"subject":"Add test_service_with_not_ready_endpoints test","message":"Add test_service_with_not_ready_endpoints test\n\nThis test case automates the issue described in\nBZ https://bugzilla.redhat.com/show_bug.cgi?id\u003d1980957.\n\nThe new test case:\n - creates a deployment with a failing readiness probe so all\n   the pods are not ready\n - creates a service for the deployment (where the endpoints\n   are not ready)\n - it checks Kuryr pods are not restarted\n\nChange-Id: Id19f85625901b35900f9441aace8fe8938cb1d68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d77f34f3f3266e30cd6bc3f7643b5058b60fa34a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d77f34f3f3266e30cd6bc3f7643b5058b60fa34a"}]},"branch":"refs/heads/master"},"275bb2a6650afeb8c05f9b5e10d572b6abd5ec6a":{"kind":"REWORK","_number":14,"created":"2022-01-11 17:37:49.000000000","uploader":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"ref":"refs/changes/09/822409/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/09/822409/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/14"}}},"commit":{"parents":[{"commit":"804a54584e002df6e1bef118d91e8335a07a7e31","subject":"Merge \"Remove LB_TIMEOUT in favor of lb_build_timeout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/804a54584e002df6e1bef118d91e8335a07a7e31"}]}],"author":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-21 08:51:53.000000000","tz":0},"committer":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2022-01-11 17:35:45.000000000","tz":0},"subject":"Add test_service_with_not_ready_endpoints test","message":"Add test_service_with_not_ready_endpoints test\n\nThis test case automates the issue described in\nBZ https://bugzilla.redhat.com/show_bug.cgi?id\u003d1980957.\n\nThe new test case:\n - creates a deployment with a failing readiness probe so all\n   the pods are not ready\n - creates a service for the deployment (where the endpoints\n   are not ready)\n - it checks Kuryr pods are not restarted\n\nChange-Id: Id19f85625901b35900f9441aace8fe8938cb1d68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/275bb2a6650afeb8c05f9b5e10d572b6abd5ec6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/275bb2a6650afeb8c05f9b5e10d572b6abd5ec6a"}]},"branch":"refs/heads/master"},"6492f096578ad4cc786c23d9480fc90e8dba1329":{"kind":"REWORK","_number":15,"created":"2022-01-11 21:23:53.000000000","uploader":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"ref":"refs/changes/09/822409/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/09/822409/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/15"}}},"commit":{"parents":[{"commit":"804a54584e002df6e1bef118d91e8335a07a7e31","subject":"Merge \"Remove LB_TIMEOUT in favor of lb_build_timeout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/804a54584e002df6e1bef118d91e8335a07a7e31"}]}],"author":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-21 08:51:53.000000000","tz":0},"committer":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2022-01-11 21:21:36.000000000","tz":0},"subject":"Add test_service_with_not_ready_endpoints test","message":"Add test_service_with_not_ready_endpoints test\n\nThis test case automates the issue described in\nBZ https://bugzilla.redhat.com/show_bug.cgi?id\u003d1980957.\n\nThe new test case:\n - creates a deployment with a failing readiness probe so all\n   the pods are not ready\n - creates a service for the deployment (where the endpoints\n   are not ready)\n - it checks Kuryr pods are not restarted\n\nChange-Id: Id19f85625901b35900f9441aace8fe8938cb1d68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/6492f096578ad4cc786c23d9480fc90e8dba1329"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/6492f096578ad4cc786c23d9480fc90e8dba1329"}]},"branch":"refs/heads/master"},"5fcc05065ccb2f613474d7fb75e1bce6e521c191":{"kind":"REWORK","_number":16,"created":"2022-01-12 14:55:11.000000000","uploader":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"ref":"refs/changes/09/822409/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/09/822409/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/16"}}},"commit":{"parents":[{"commit":"804a54584e002df6e1bef118d91e8335a07a7e31","subject":"Merge \"Remove LB_TIMEOUT in favor of lb_build_timeout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/804a54584e002df6e1bef118d91e8335a07a7e31"}]}],"author":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-21 08:51:53.000000000","tz":0},"committer":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2022-01-12 14:53:04.000000000","tz":0},"subject":"Add test_service_with_not_ready_endpoints test","message":"Add test_service_with_not_ready_endpoints test\n\nThis test case automates the issue described in\nBZ https://bugzilla.redhat.com/show_bug.cgi?id\u003d1980957.\n\nThe new test case:\n - creates a deployment with a failing readiness probe so all\n   the pods are not ready\n - creates a service for the deployment (where the endpoints\n   are not ready)\n - it checks Kuryr pods are not restarted\n\nChange-Id: Id19f85625901b35900f9441aace8fe8938cb1d68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/5fcc05065ccb2f613474d7fb75e1bce6e521c191"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/5fcc05065ccb2f613474d7fb75e1bce6e521c191"}]},"branch":"refs/heads/master"},"aea7ae2a35dbee2d6cc21e4a0da1969628373fee":{"kind":"REWORK","_number":17,"created":"2022-01-13 12:52:00.000000000","uploader":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"ref":"refs/changes/09/822409/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/09/822409/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/17"}}},"commit":{"parents":[{"commit":"3a8ccf5b4f0e16998d0b5007bfc71a7c81295ab3","subject":"Merge \"Tune curl repetitions value\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/3a8ccf5b4f0e16998d0b5007bfc71a7c81295ab3"}]}],"author":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-21 08:51:53.000000000","tz":0},"committer":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2022-01-13 12:49:23.000000000","tz":0},"subject":"Add test_service_with_not_ready_endpoints test","message":"Add test_service_with_not_ready_endpoints test\n\nThis test case automates the issue described in\nBZ https://bugzilla.redhat.com/show_bug.cgi?id\u003d1980957.\n\nThe new test case:\n - creates a deployment with a failing readiness probe so all\n   the pods are not ready\n - creates a service for the deployment (where the endpoints\n   are not ready)\n - it checks Kuryr pods are not restarted\n\nChange-Id: Id19f85625901b35900f9441aace8fe8938cb1d68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/aea7ae2a35dbee2d6cc21e4a0da1969628373fee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/aea7ae2a35dbee2d6cc21e4a0da1969628373fee"}]},"branch":"refs/heads/master"},"00bb809b4feb1890dbeddf7b4bd750aed3da9bf3":{"kind":"REWORK","_number":18,"created":"2022-01-13 18:13:34.000000000","uploader":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"ref":"refs/changes/09/822409/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/09/822409/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/18"}}},"commit":{"parents":[{"commit":"3a8ccf5b4f0e16998d0b5007bfc71a7c81295ab3","subject":"Merge \"Tune curl repetitions value\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/3a8ccf5b4f0e16998d0b5007bfc71a7c81295ab3"}]}],"author":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-21 08:51:53.000000000","tz":0},"committer":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2022-01-13 18:11:08.000000000","tz":0},"subject":"Add test_service_with_not_ready_endpoints test","message":"Add test_service_with_not_ready_endpoints test\n\nThis test case automates the issue described in\nBZ https://bugzilla.redhat.com/show_bug.cgi?id\u003d1980957.\n\nThe new test case:\n - creates a deployment with a failing readiness probe so all\n   the pods are not ready\n - creates a service for the deployment (where the endpoints\n   are not ready)\n - it checks Kuryr pods are not restarted\n\nChange-Id: Id19f85625901b35900f9441aace8fe8938cb1d68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/00bb809b4feb1890dbeddf7b4bd750aed3da9bf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/00bb809b4feb1890dbeddf7b4bd750aed3da9bf3"}]},"branch":"refs/heads/master"},"2ffdd2f47560087a242a3731f831dfb51ed2b149":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2022-01-14 18:24:17.000000000","uploader":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"ref":"refs/changes/09/822409/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/09/822409/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/09/822409/19"}}},"commit":{"parents":[{"commit":"d9a93bcba9dd3360ce3de3fc81bef10452bf40c5","subject":"Amend check_sg_rules_for_np() method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d9a93bcba9dd3360ce3de3fc81bef10452bf40c5"}]}],"author":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-21 08:51:53.000000000","tz":0},"committer":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2022-01-14 18:24:17.000000000","tz":0},"subject":"Add test_service_with_not_ready_endpoints test","message":"Add test_service_with_not_ready_endpoints test\n\nThis test case automates the issue described in\nBZ https://bugzilla.redhat.com/show_bug.cgi?id\u003d1980957.\n\nThe new test case:\n - creates a deployment with a failing readiness probe so all\n   the pods are not ready\n - creates a service for the deployment (where the endpoints\n   are not ready)\n - it checks Kuryr pods are not restarted\n\nChange-Id: Id19f85625901b35900f9441aace8fe8938cb1d68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2ffdd2f47560087a242a3731f831dfb51ed2b149"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2ffdd2f47560087a242a3731f831dfb51ed2b149"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
