)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~822754","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~Id9e6389bfafc89a4d407952ee0625a932b51ef38","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"curl_repetitions","attention_set":{},"removed_from_attention_set":{"11600":{"account":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"last_update":"2022-01-12 17:06:19.000000000","reason":"Michał Dulko replied on the change"},"27032":{"account":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"last_update":"2022-01-13 08:02:15.000000000","reason":"Change was submitted"},"13692":{"account":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"last_update":"2022-01-13 08:02:15.000000000","reason":"Change was submitted"},"27086":{"account":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"last_update":"2022-01-13 08:02:15.000000000","reason":"Change was submitted"},"23567":{"account":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"last_update":"2022-01-13 08:02:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id9e6389bfafc89a4d407952ee0625a932b51ef38","subject":"Tune curl repetitions value","status":"MERGED","created":"2021-12-22 17:53:49.000000000","updated":"2022-01-13 08:04:46.000000000","submitted":"2022-01-13 08:02:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"822754-curl_repetitions","meta_rev_id":"8826c46009a108d50af22470c038c2b0b7633cc9","_number":822754,"virtual_id_number":822754,"owner":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-13 08:02:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2022-01-12 17:06:19.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":1,"date":"2022-01-10 11:59:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":2,"date":"2022-01-12 13:43:27.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":1,"date":"2022-01-12 17:06:19.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"reviewers":{"REVIEWER":[{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-22 19:08:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-24 10:30:37.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2021-12-26 07:25:37.000000000","updated_by":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"reviewer":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"state":"CC"},{"updated":"2021-12-26 07:25:37.000000000","updated_by":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2022-01-03 11:35:07.000000000","updated_by":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"},{"updated":"2022-01-03 11:35:07.000000000","updated_by":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"},{"updated":"2022-01-10 11:59:20.000000000","updated_by":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"reviewer":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"state":"REVIEWER"}],"messages":[{"id":"c653ee32cf01d9a22f90faede971a7c0061092ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"date":"2021-12-22 17:53:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4447d070c57fb4b7965c0654314fcb9ccfc4249a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 19:08:58.000000000","message":"Patch Set 1: Verified-1\n\n(20 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94a32edf7c49402995547fcee1960e8d : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd2b1e85ada148849ecc85915a53209c : FAILURE in 3m 39s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/0b2f188358f940b68356cd24d033b7b6 : FAILURE in 54m 57s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/a9e831778eae4a049881cba9dbc9def2 : FAILURE in 50m 41s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/b1b3194af5f147fe8f7dea5d35e44035 : FAILURE in 54m 34s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/872c4ed7afb14202b3fe490ab375998d : FAILURE in 43m 34s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/53cb109f69814bd9bb205540d720bbda : FAILURE in 52m 52s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/ecc6c74943264fe5a09ca658aac20f66 : FAILURE in 1h 11m 38s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/20ecc35c71784faf8ee900b892d1bb46 : FAILURE in 1h 04m 18s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/8d586aab87284ff689c43284e20c9f1c : FAILURE in 1h 11m 45s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/db174253e73d4a37828533a8f8ab4c74 : FAILURE in 1h 07m 43s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/7a661c7ff71d4596abb4df9f555e6514 : FAILURE in 45m 47s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/f31a49a71375416f839aec5d73cef0c3 : FAILURE in 52m 03s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/a09874f2f61a41c09d74428a2f8d92fe : POST_FAILURE in 43m 23s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/9a9a884622ee4dc5bde21d347615f35c : FAILURE in 49m 03s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/ca3c219efb744c7297d1d7c0b4a3135e : FAILURE in 35m 30s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/044be39f85054168a5e3a25f63d3907e : FAILURE in 41m 49s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/efd8425a71c64b14848b92500b58a912 : FAILURE in 36m 12s","accounts_in_message":[],"_revision_number":1},{"id":"65a72cca1b64b3dcfd08a581a79b5c1634174b94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"date":"2021-12-22 22:36:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"72210738a21edabebb7f9869acb60ac4114a75fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 01:11:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff36f08d356842a2aa1846e16e554948 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46137c71344c4c83ba8df9c101e26e44 : SUCCESS in 4m 16s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/f52d3b28efe143c8bad2e4f0f30f3ce7 : FAILURE in 1h 50m 00s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/203914b064e0474a9b3c291015aa19c1 : SUCCESS in 1h 44m 42s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/e3f3c018c0f3452ab65ec79670b71d4c : SUCCESS in 1h 39m 51s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/907236681fc1418ca7a446d5b129d80d : SUCCESS in 1h 43m 17s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/044a39065a3c4666976ef266be606145 : FAILURE in 1h 59m 45s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/0f7c1c4cc16e4a9299e7af1b318c4d07 : SUCCESS in 2h 01m 03s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/fbd4e2e30c5c45b5bc4c00e0634d103c : SUCCESS in 1h 12m 28s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/969e32c8451d497f8bc0fce67dfa8404 : SUCCESS in 2h 09m 33s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/6a514c0a744642c8bcdecebbf6df2dcd : SUCCESS in 1h 42m 12s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/66a5fd61b0c746b5b94656de311a0a39 : FAILURE in 2h 33m 27s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/43455a91837c402fbc1737362083d5dc : SUCCESS in 2h 31m 38s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/604e9b869940455a82ce6c92ee987666 : POST_FAILURE in 1h 20m 33s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/127049c9840b47be83ecbc5a77f6c954 : SUCCESS in 1h 44m 52s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/ec0711789f284f2ca9732501aebceb01 : SUCCESS in 1h 43m 07s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/60c72698469749d889e065846ead5ed8 : SUCCESS in 1h 21m 18s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/825c1039eb7d4d8eaa9a42759cbee0ce : FAILURE in 58m 58s","accounts_in_message":[],"_revision_number":2},{"id":"b7f478589d401703741c0f033d5eaed143e2aa23","author":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"date":"2021-12-23 06:28:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8b1cdad422b137fd530144128f18bc65c4da3665","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 09:19:55.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f10f855870fb492587dfc18f8b82b861 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/652c5c714a2248f2a1548ea3e7d5d848 : SUCCESS in 3m 10s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/8a66f328dafb4615a9f0e2e1c2b1f3ed : SUCCESS in 2h 31m 53s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/4ae5d194afed4d3bae00b64af7373b17 : SUCCESS in 1h 47m 05s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/35216a1884384a4ea651cb1e73c5ff61 : SUCCESS in 2h 09m 55s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/09a6d6ce4619439da6ae28e870b4cfc4 : FAILURE in 2h 04m 49s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/05cb69d366164e78b65e55ba97ef483b : SUCCESS in 2h 05m 56s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/02ebc91c5304499c965516349fd8026c : SUCCESS in 1h 32m 16s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/50820477d27d4219b2685b815aba1d29 : SUCCESS in 1h 20m 50s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/c25b1954d20b423593e0ebcac4c011f9 : SUCCESS in 1h 48m 31s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/9a481e3dde2747de8288222b2415947d : FAILURE in 1h 44m 08s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/7edff0686562437e8f50ac6505926180 : FAILURE in 2h 46m 37s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/ccee41252c0d4e8997f5fba54533a1a5 : FAILURE in 2h 22m 32s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/e46e379f182d4b5b9b090ef428f7e53e : POST_FAILURE in 58m 28s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/6b4dbe475df6480ca9c4860efc5480cc : SUCCESS in 1h 17m 55s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/67e0d16c497b4b188f9a77fd8fe811f1 : SUCCESS in 1h 07m 05s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/4624db2ca4a844e3a672b6256bfbf6e9 : SUCCESS in 1h 07m 14s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/2875bb1d6fad444d902658478651faae : SUCCESS in 55m 33s","accounts_in_message":[],"_revision_number":2},{"id":"b358245c4fd21d5ca04922ed09d8ff39cc41dd07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"date":"2021-12-23 13:16:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c3428d5c4170784251b84f04af820652ef021b81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 15:58:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3524eb82f0464b9facf007662152436d : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b3fd450c0834fa7b2ccd96f2a763218 : SUCCESS in 3m 22s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/ce807aa4d53640839656bc1f53f32daf : SUCCESS in 2h 28m 31s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/592f00d8a0c34a7bbef497665c8d8013 : SUCCESS in 2h 01m 55s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/03697b7a1dac47d1a3e5f1ba2c594716 : SUCCESS in 2h 23m 00s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/1b8f533f12b041949e73fb2877972694 : SUCCESS in 1h 42m 00s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/08975675ee87416ebacd7d676fc79586 : SUCCESS in 1h 40m 18s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/272904ab736941729979679a6f6deab9 : SUCCESS in 1h 10m 21s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/eb4c6fdc9a3543a68fdbd4b42444513b : SUCCESS in 1h 15m 17s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/88dd3a87b47a4de2b348dc9da70240c8 : SUCCESS in 1h 49m 26s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/1907ff83340b4f11ac9496d47565d28e : SUCCESS in 1h 32m 51s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/012ef996cf5845acac5db05617453f94 : SUCCESS in 2h 40m 12s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/f9ffd58cd6c740e3a2f7556434d398ff : FAILURE in 2h 13m 55s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/ec65996f28584252a587c4460a40fb1b : POST_FAILURE in 41m 12s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/f47859516da843c2a8e933c567504ca8 : SUCCESS in 1h 23m 06s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/8babc21362bc4726a70edbface44e174 : SUCCESS in 1h 31m 48s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/52e9d8facf134fe49d8db4537075c1e3 : SUCCESS in 1h 09m 51s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/8473759e8a474c1c90e35a2ab621ba8f : SUCCESS in 56m 35s","accounts_in_message":[],"_revision_number":3},{"id":"2971a8a09a55ebc1d3d6376586371900b5dfb575","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-12-24 10:30:37.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"49a529db0f922ee753102f1fe368010c63988f83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"date":"2022-01-03 11:25:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8dca47f9a0e7dab827dc28900e4652b9fc28e561","author":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"date":"2022-01-03 11:34:50.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"e4239700094f991be2a01c9534957b30616f9271","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 13:39:35.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6ae1adbfe9e49d3b17a4a13a3206ff1 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f247d2787c14955955c8d12dfa84e0e : SUCCESS in 4m 08s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/35ee45c9166c462fa6a912924253cbab : SUCCESS in 1h 11m 23s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/7e85058f3cc542a0836ed839d3f6733c : SUCCESS in 1h 31m 06s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/a4582a1275b445bcab9382e05ce0f4ac : FAILURE in 1h 40m 46s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/509278f8e9eb4ff3941d0415b6a15032 : FAILURE in 1h 17m 56s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/9722293bd8954bd28f3f5d564d6e6081 : SUCCESS in 1h 39m 25s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/6aa0a6e43fae443bb5e4dae899b17dcb : SUCCESS in 1h 14m 26s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/52e54c0453cb47d7b12b705e795bff8f : SUCCESS in 1h 20m 32s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/e4708149991e436fa6f655f92dd3cd6a : FAILURE in 1h 32m 30s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/a09bd016b30f42bba9441162a5cf3cf6 : FAILURE in 1h 15m 43s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/c246dd921d8841e0869feedc23cbb769 : FAILURE in 2h 08m 02s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/0fa542d1e2a24719a7c58db8b960a511 : SUCCESS in 1h 54m 01s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/7dc8a4d5a0084f4bb2fa902db918c6b6 : POST_FAILURE in 45m 33s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/5bcd7966623b4c189c3b8142252f31cc : SUCCESS in 1h 20m 08s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/859431cb9bcb4a9eb74942e0fef6f009 : SUCCESS in 1h 12m 13s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/bf588eb31b8c4ce0b649b1f8f8adaa5f : SUCCESS in 1h 05m 35s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/f12477b821ed48b0bda21b9dbdadcb6b : SUCCESS in 1h 10m 42s","accounts_in_message":[],"_revision_number":4},{"id":"5d1af42c6a70a8bde49d7b03ef2ae801be520a86","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2022-01-03 15:12:48.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2419511de66905eff627de236225aed48c8b202f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"date":"2022-01-03 15:38:19.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"306890a002dd7f700c9f5882bfc17b7a680cb06f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 17:40:55.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3c6823bfca24bd89e5da2760decb9e2 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91101a4d19ee44cf8eb8f1e0e448ec2d : SUCCESS in 4m 43s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/5a1f2171e00c447db9af59828a1485a5 : SUCCESS in 1h 27m 30s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/f55f406b0aa940598f9a6be8c30fccd3 : SUCCESS in 1h 23m 03s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/9265d92224454c29823d02df91d0500c : FAILURE in 1h 50m 40s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/cada97ac385942a1a0ea836ffc4813af : SUCCESS in 1h 29m 07s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/1522deb042424a14813266ef198630d0 : SUCCESS in 1h 27m 51s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/eb3895d1391f405c8cd514d4dbfa788a : SUCCESS in 1h 22m 04s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/3d1ae1ed9e2d4f0fb4b072e4f38745f7 : SUCCESS in 1h 18m 56s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/6308e5690a7543cb89f88970ddd78135 : FAILURE in 1h 27m 53s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/fdbc92620387478e8063101d3c09dd31 : FAILURE in 1h 28m 23s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/1cf5bd1a163b4c0d980d1273029a23f3 : FAILURE in 2h 00m 00s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/9d3a7103e778456fa093a808803f9e93 : FAILURE in 1h 57m 15s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/08decff12b57482fa7907aec23b47cdf : POST_FAILURE in 42m 16s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/5af964054c294fcf8ec36e76d8fa3f51 : SUCCESS in 1h 16m 43s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/586105b3be58443ebf2bb751424025ee : SUCCESS in 1h 20m 04s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/6ce4588df1284abb8b17f35e8b038c0b : SUCCESS in 47m 27s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/91e8d226245349b6a92b99af1cef3514 : SUCCESS in 50m 45s","accounts_in_message":[],"_revision_number":5},{"id":"66fd35b5dcf8b22f7a1a8c28abb6f804cf202730","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"date":"2022-01-04 07:41:12.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6db9f3dbde1b8b0eea365745119227350cb72a3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 09:59:20.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c9233ef16224b468b4406a6de759e61 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0edfee77c72d499dbd47b4aed1fd87ae : SUCCESS in 4m 10s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/f9b823cc770d40b395cfec6165d4ec49 : SUCCESS in 1h 20m 16s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/1c6f3b1abc24495b9bf0d3ab65e1bcc1 : SUCCESS in 1h 25m 29s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/39efb9e4289b42358adc2dc4d6e02391 : FAILURE in 1h 59m 41s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/eb22312280eb4730950946d801f46f98 : FAILURE in 1h 23m 20s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/a8d7485d09d3444dbf42b96b15af0589 : SUCCESS in 1h 19m 38s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/514b8d71490444bea50826dfe4366514 : SUCCESS in 1h 17m 40s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/f37999cadbd6400a8b9f408afdc68a2e : SUCCESS in 1h 14m 26s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/62b527b29e00470bbb1b80b0699dce5e : FAILURE in 1h 48m 27s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/985b5aaec9f6401e858a28266a0f7085 : SUCCESS in 1h 41m 24s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/9cb7df54a88b4495bd83ad50be78296b : FAILURE in 2h 16m 29s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/0dd32a7cc8324f209999c7c79960cb74 : FAILURE in 2h 12m 00s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/c250f01e3d0c4913b379706c0cca050a : POST_FAILURE in 42m 09s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/de1f35719b9c45ec8d4ce2ca92e41595 : SUCCESS in 1h 52m 45s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/c33dc82f533d4fc9a4d78c2b6104a5f9 : SUCCESS in 1h 08m 49s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/cc21f59de82146a0b01538cce01ca378 : SUCCESS in 1h 09m 55s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/daa954d33579405d9bbe358177560ed3 : SUCCESS in 1h 02m 08s","accounts_in_message":[],"_revision_number":6},{"id":"a39498016865ec03f44cee0ebbabe5987fd1670a","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2022-01-04 13:20:24.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"859d05e294b903317a61a95571a1fbc3867f5514","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2022-01-04 20:56:27.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"517fca6e8abca76584a3f038495f25182c2a0a8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 23:17:42.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b41645ecdc3b4c0ca3d6feef80f7dbe3 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b854e360438144bea1647348ecd6596a : SUCCESS in 3m 49s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/fd646705e2b74e30a0a833e562d12921 : SUCCESS in 1h 44m 40s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/e4384e40462d4c79bebfdcc826681bd7 : SUCCESS in 1h 36m 38s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/0be5eedde3bb49b6b1d329d6b6b0c59c : FAILURE in 2h 15m 14s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/3d738bb929264acdac953170a5ee90ef : FAILURE in 1h 09m 43s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/02871c7875c342d3999f67ec503224f7 : SUCCESS in 1h 27m 49s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/72f21044ded74e98857dc9d6a512a297 : SUCCESS in 1h 29m 59s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/fac87c904dc84770a41cd99aba6d39c0 : SUCCESS in 1h 22m 52s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/2a2611589de94f51aeebb986952ed6d7 : FAILURE in 1h 47m 02s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/c7a45785471e4d8794c1a0c2bc5c1708 : FAILURE in 1h 33m 53s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/4153859e53f04ac4b2de5c91e6879e87 : FAILURE in 1h 58m 00s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/937075742997424bbca6419d7d88eca3 : FAILURE in 1h 49m 39s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/2689ee27a1d241cd8bb3edbb30088cf7 : POST_FAILURE in 54m 54s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/8558362ec944435fbb895470481fd570 : SUCCESS in 1h 24m 25s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/d9de80649c794a8ca08c24942051a4c9 : SUCCESS in 1h 15m 51s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/67a67630e8974b4cb7ba8688225ffde4 : SUCCESS in 1h 01m 34s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/edfbc982b0524849883676acc556799c : SUCCESS in 1h 06m 45s","accounts_in_message":[],"_revision_number":6},{"id":"6b406ca9487a5021a07b3a5cd2c4e05829bc8950","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2022-01-05 17:39:38.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"c2a8afc7c0d2226522d9d6c823c1fdb50969e7a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 19:51:49.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75516fc919674a7fb66d560c2e364129 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cd584fa4cb644c6a7059541d89e5ec4 : SUCCESS in 3m 53s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/0bb0788205314335908ba3d159410f3e : SUCCESS in 1h 23m 22s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/0bece82bee1f4cc0a04d0e37e779fddb : SUCCESS in 1h 38m 01s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/31250f112aa1485d950d1adcc884f45f : FAILURE in 1h 34m 47s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/a4e22c3358744147bc4eed26acdf9a6e : SUCCESS in 1h 23m 52s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/865988d24c2340008f96774d612b3d5b : FAILURE in 1h 39m 46s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/854030679a0c426fb05893decbce5166 : SUCCESS in 1h 13m 15s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/f5a6d03165354f7997b4947cf7b84248 : SUCCESS in 1h 12m 30s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/52e24114bb8c4c2fa98f76c11096b5b0 : SUCCESS in 1h 24m 08s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/1a9178d8e64f4f50918abb7377928ac9 : SUCCESS in 1h 45m 04s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/974e8da855ad4662be92d311b3c3d217 : FAILURE in 1h 51m 39s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/dd9124675b174e1b8189ba073c89b1d4 : FAILURE in 2h 09m 28s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/8f3047ace80c48deb41fe2ff40474b4d : POST_FAILURE in 48m 31s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/be13bc4779704c06911e72d48b4116be : SUCCESS in 1h 41m 42s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/5db4e46108b840de987c9e63b77d97cb : SUCCESS in 1h 03m 32s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/e0d46f6ce7fc476384d1727c056a98a2 : SUCCESS in 1h 14m 24s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/18bbfbba6dab4ab9a9787c9d3fc7da7f : SUCCESS in 1h 04m 38s","accounts_in_message":[],"_revision_number":7},{"id":"1f2f22dc95d162f1f1e80be47dbe5f7cc110cbec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2022-01-06 23:07:48.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"023bdcb6ebe5b3ea776bd78a4ace01a123c1b42d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 01:54:42.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92c222fe4fe8446db5df713d4d48e085 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99ece499829b4b0d901d8f2934ebe281 : SUCCESS in 3m 18s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/6ddeee0103f44f429e7c8d72421bc5a4 : SUCCESS in 1h 38m 21s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/3faa72460ca74b338af0501f04b9bcab : FAILURE in 1h 19m 53s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/a59755784df7402f9aa5d8c136a51e1e : FAILURE in 2h 39m 49s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/e5a6a872704441f99a7d4d761ce4407c : FAILURE in 1h 08m 20s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/2a31b13378b34cd4bbd39c328f74b144 : SUCCESS in 1h 41m 47s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/2136e23dcebe4b94aab560e717b7f3bf : SUCCESS in 1h 27m 44s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/eea29a2560d54499a915603af63341fa : SUCCESS in 1h 23m 56s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/7f620e699c1c45fba4150491ccaec609 : FAILURE in 1h 35m 26s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/acf4176f723640edb31732f3a9b8ea58 : FAILURE in 1h 23m 13s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/387f0826e45d41259416db96ccd33f1e : SUCCESS in 1h 21m 38s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/534c2619ba28457d9a8780c4e18bda47 : SUCCESS in 1h 26m 58s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/52ffbc1e061d4cdcba0100662d13a0e7 : POST_FAILURE in 57m 04s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/e40fa8b6352044f5a2c52950439e660e : SUCCESS in 1h 42m 25s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/77f5df1c9cca4054b9d6663d4f30c442 : SUCCESS in 49m 48s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/13c73f718492429a94b29ecbc715c753 : SUCCESS in 1h 08m 28s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/f23ee6caf2dd4d64be867b5a69d62368 : SUCCESS in 1h 04m 56s","accounts_in_message":[],"_revision_number":8},{"id":"8253899c7b2bc9c453a76eb45a39eaa3fbbde8bb","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2022-01-07 02:08:49.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"2787b6245331325bc67ebb5dc765703c5c08d4da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2022-01-07 02:16:08.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"f742732a014897154876cdcf0b70a7d3cb2302a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 04:24:18.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0ff175e18b8454da3f17ff271905c03 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bae1c75ae3224e1c8d1028495ee26c30 : SUCCESS in 4m 50s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/ed23b72a34c343b7ac235633850182c0 : SUCCESS in 1h 39m 16s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/de8ecfa61d7a43309f450f3979fa7880 : FAILURE in 1h 27m 13s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/3d144d7dfa2f449794784ba9b2f1dd3f : SUCCESS in 1h 48m 42s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/acafbf18d61c47a28140b8ce7e03994b : FAILURE in 1h 20m 49s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/cd3669b7679143c2954968c52d9c0e0b : SUCCESS in 1h 32m 20s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/1a37db10a3484c108281cba62f8403b1 : SUCCESS in 1h 24m 12s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/00f7490aca9645069109a5797f75bb99 : SUCCESS in 1h 32m 13s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/d565fbcf11e94852b6f7957dd7e6db13 : SUCCESS in 1h 31m 30s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/7cf64f8965c04347ad5c929a8fec8aa3 : SUCCESS in 1h 27m 40s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/0969741bfebc42658b415c8db6da90df : SUCCESS in 1h 28m 06s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/0376b0f886e84b56bb36a5afda6dce6a : FAILURE in 1h 25m 10s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/f6b8baf3300945fe80e9eecc4e6cfbfa : POST_FAILURE in 59m 27s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/e647bba543fd4fa2b86b5faedb0a47c8 : SUCCESS in 1h 38m 16s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/35808d0b644243369c6ae81353b6ef94 : SUCCESS in 1h 06m 00s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/1169f38dd7424e03aad20efe40ad4d6d : SUCCESS in 1h 08m 31s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/581ed83f606f4fe5adc6f015be4dfd33 : SUCCESS in 1h 05m 34s","accounts_in_message":[],"_revision_number":9},{"id":"23d76032ce918b6b0c7df7dcb60303da4f7ed47e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2022-01-07 12:50:29.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"1e10c8003accb61a43a3833c9b350462b5d13d4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2022-01-07 12:50:52.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"608caa186da43b8cd22926a5c133ed56f648e4d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 14:48:51.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30d30063a4664a2c9cf4676c2527fa11 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe54a25a5e1a4ab0bdd33145e1d480dd : SUCCESS in 3m 28s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/666d25fbb2c446b696a2fca5aa1dcf98 : SUCCESS in 1h 23m 13s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/755d3b4f38a54fb0b1d24f54c6cf9682 : SUCCESS in 1h 23m 53s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/eef008ff772643749c07fcb8d72ccd41 : SUCCESS in 1h 43m 22s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/af1d512e219e446dab64b3520abeb3ac : SUCCESS in 1h 28m 36s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/cdc2c007397b48488ab819594b9e770f : SUCCESS in 1h 36m 49s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/3a7685ca49cd4d7eb83d4b76fde16afd : SUCCESS in 1h 17m 06s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/986dd89705d84a5d886f1e1030ca4041 : SUCCESS in 1h 11m 32s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/a50d51290b024574b03752bf4f0fdfa0 : SUCCESS in 1h 53m 19s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/fdb08a63c46d4894ab088676872c648a : SUCCESS in 1h 21m 27s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/10bd2e98148f45b78e9ee3120aab98b8 : SUCCESS in 1h 22m 29s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/49c4403c0ba74775ae1f9d1e14c215f8 : SUCCESS in 1h 26m 21s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/5782ea3a3b7741209be5975240829fae : POST_FAILURE in 1h 02m 35s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/471268f2c34345918274250c5c9f0118 : SUCCESS in 1h 30m 25s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/aaac0335df57435c83f61929b51a76fe : SUCCESS in 1h 12m 32s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/298e6d51c56e417ba18f9b0cd1f382bc : SUCCESS in 1h 15m 51s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/67f7338afe0543f5963f8f937055d746 : SUCCESS in 1h 06m 45s","accounts_in_message":[],"_revision_number":11},{"id":"9b772531aaca883d84268321f88614e8e24de171","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2022-01-10 11:59:20.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"a869581e6d512c062e919cbe2dc500e6b4348d3c","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2022-01-12 13:43:27.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"f824c093156be6e7d5854905f6ae359b3ca5257d","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-01-12 17:06:19.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"5576e44f38a00f31c0a2c4074a2a335dbd48423f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 06:31:52.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"dca9abc10b937cebfefc5b9a7d6a6aed0b6d24cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 08:02:14.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e636a4f2d47d475585c5a9177b8e520a : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5b6a5a602e74600876debf0c2e54892 : SUCCESS in 3m 50s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/86ccfad6b51749f09b67fa73e1b6b5e0 : SUCCESS in 1h 26m 41s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/175aefd3b8f44af58bca7846c27a9a98 : SUCCESS in 1h 07m 23s","accounts_in_message":[],"_revision_number":11},{"id":"09e5332a36009c24ba14bc8b14969f58d9e8f190","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 08:02:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"8826c46009a108d50af22470c038c2b0b7633cc9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 08:04:46.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64a15e9c2e3048abb7f1f8480ae5f3c6 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"3408ab72d7ef0aff0d4f0d9a041e6ae260f2ad6d","revisions":{"18c5649329b8d870d32ced3edf320a541b6f5db7":{"kind":"REWORK","_number":1,"created":"2021-12-22 17:53:49.000000000","uploader":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"ref":"refs/changes/54/822754/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/54/822754/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/1"}}},"commit":{"parents":[{"commit":"c62a4a8be41192081d631cc837879e15cfb6dc8f","subject":"Merge \"Small improvement in method definition and reference\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c62a4a8be41192081d631cc837879e15cfb6dc8f"}]}],"author":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-22 17:48:54.000000000","tz":0},"committer":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-22 17:48:56.000000000","tz":0},"subject":"[WIP] Tune curl repetitions value","message":"[WIP] Tune curl repetitions value\n\nFor svc and route endpoint connectivity checks there is a repetition\nparameter that can be tunned to be relative to the expected number\nof different endpoint responses.\n\nFor each expected different backend 6 repetitions seems to be enough\nand it will safe significant tests execution time.\n\nChange-Id: Id9e6389bfafc89a4d407952ee0625a932b51ef38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/18c5649329b8d870d32ced3edf320a541b6f5db7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/18c5649329b8d870d32ced3edf320a541b6f5db7"}]},"branch":"refs/heads/master"},"ed6ca8d6933bc3a2a56e69880c34d78b812f5e09":{"kind":"REWORK","_number":2,"created":"2021-12-22 22:36:17.000000000","uploader":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"ref":"refs/changes/54/822754/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/54/822754/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/2"}}},"commit":{"parents":[{"commit":"c62a4a8be41192081d631cc837879e15cfb6dc8f","subject":"Merge \"Small improvement in method definition and reference\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c62a4a8be41192081d631cc837879e15cfb6dc8f"}]}],"author":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-22 17:48:54.000000000","tz":0},"committer":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-22 22:33:40.000000000","tz":0},"subject":"[WIP] Tune curl repetitions value","message":"[WIP] Tune curl repetitions value\n\nFor svc and route endpoint connectivity checks there is a repetition\nparameter that can be tuned to be proportional to the expected number\nof different endpoint responses.\n\nFor each expected different backend 6 repetitions seems to be enough\nand it will safe significant tests execution time.\n\nChange-Id: Id9e6389bfafc89a4d407952ee0625a932b51ef38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ed6ca8d6933bc3a2a56e69880c34d78b812f5e09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ed6ca8d6933bc3a2a56e69880c34d78b812f5e09"}]},"branch":"refs/heads/master"},"83cd1bc8c120bf4f9839ef8d31ceb65b49b5c5de":{"kind":"REWORK","_number":3,"created":"2021-12-23 13:16:11.000000000","uploader":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"ref":"refs/changes/54/822754/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/54/822754/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/3"}}},"commit":{"parents":[{"commit":"c62a4a8be41192081d631cc837879e15cfb6dc8f","subject":"Merge \"Small improvement in method definition and reference\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c62a4a8be41192081d631cc837879e15cfb6dc8f"}]}],"author":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-22 17:48:54.000000000","tz":0},"committer":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-23 13:15:02.000000000","tz":0},"subject":"[WIP] Tune curl repetitions value","message":"[WIP] Tune curl repetitions value\n\nFor svc and route endpoint connectivity checks there is a repetition\nparameter that can be tuned to be proportional to the expected number\nof different endpoint responses.\n\nFor each expected different backend 6 repetitions seems to be enough\nand it will safe significant tests execution time.\n\nChange-Id: Id9e6389bfafc89a4d407952ee0625a932b51ef38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/83cd1bc8c120bf4f9839ef8d31ceb65b49b5c5de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/83cd1bc8c120bf4f9839ef8d31ceb65b49b5c5de"}]},"branch":"refs/heads/master"},"570c3d6443672b7d3b7e2ee6b71d9d78a78f09a0":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2022-01-03 11:25:29.000000000","uploader":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"ref":"refs/changes/54/822754/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/54/822754/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/4"}}},"commit":{"parents":[{"commit":"13e740911c448105f9748527e3f6c21265640fd7","subject":"Merge \"Change get_readiness_state() return type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/13e740911c448105f9748527e3f6c21265640fd7"}]}],"author":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-22 17:48:54.000000000","tz":0},"committer":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2022-01-03 11:23:01.000000000","tz":0},"subject":"Tune curl repetitions value","message":"Tune curl repetitions value\n\nFor svc and route endpoint connectivity checks there is a repetition\nparameter that can be tuned to be proportional to the expected number\nof different endpoint responses.\n\nFor each expected different backend 6 repetitions seems to be enough\nand it will save significant tests execution time.\n\nChange-Id: Id9e6389bfafc89a4d407952ee0625a932b51ef38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/570c3d6443672b7d3b7e2ee6b71d9d78a78f09a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/570c3d6443672b7d3b7e2ee6b71d9d78a78f09a0"}]},"branch":"refs/heads/master"},"7fe3abf1978d1c86ce341cf49c43b158d1b5e01f":{"kind":"REWORK","_number":5,"created":"2022-01-03 15:38:19.000000000","uploader":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"ref":"refs/changes/54/822754/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/54/822754/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/5"}}},"commit":{"parents":[{"commit":"13e740911c448105f9748527e3f6c21265640fd7","subject":"Merge \"Change get_readiness_state() return type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/13e740911c448105f9748527e3f6c21265640fd7"}]}],"author":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-22 17:48:54.000000000","tz":0},"committer":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2022-01-03 15:36:16.000000000","tz":0},"subject":"Tune curl repetitions value","message":"Tune curl repetitions value\n\nFor svc and route endpoint connectivity checks there is a repetition\nparameter that can be tuned to be proportional to the expected number\nof different endpoint responses.\n\nFor each expected different backend 8 repetitions seems to be enough\nand it will save significant tests execution time.\n\nChange-Id: Id9e6389bfafc89a4d407952ee0625a932b51ef38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/7fe3abf1978d1c86ce341cf49c43b158d1b5e01f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/7fe3abf1978d1c86ce341cf49c43b158d1b5e01f"}]},"branch":"refs/heads/master"},"c10e0a8ccd02dfd70ccb3f69c71fc1d04bb89fbc":{"kind":"REWORK","_number":6,"created":"2022-01-04 07:41:12.000000000","uploader":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"ref":"refs/changes/54/822754/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/54/822754/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/6"}}},"commit":{"parents":[{"commit":"804a54584e002df6e1bef118d91e8335a07a7e31","subject":"Merge \"Remove LB_TIMEOUT in favor of lb_build_timeout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/804a54584e002df6e1bef118d91e8335a07a7e31"}]}],"author":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-22 17:48:54.000000000","tz":0},"committer":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2022-01-04 07:39:12.000000000","tz":0},"subject":"Tune curl repetitions value","message":"Tune curl repetitions value\n\nFor svc and route endpoint connectivity checks there is a repetition\nparameter that can be tuned to be proportional to the expected number\nof different endpoint responses.\n\nFor each expected different backend 10 repetitions seems to be enough\nand it will save significant tests execution time.\n\nChange-Id: Id9e6389bfafc89a4d407952ee0625a932b51ef38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c10e0a8ccd02dfd70ccb3f69c71fc1d04bb89fbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c10e0a8ccd02dfd70ccb3f69c71fc1d04bb89fbc"}]},"branch":"refs/heads/master"},"5564833296540458938ad4f21ec7b65ff7f0f3ed":{"kind":"REWORK","_number":7,"created":"2022-01-05 17:39:38.000000000","uploader":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"ref":"refs/changes/54/822754/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/54/822754/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/7"}}},"commit":{"parents":[{"commit":"804a54584e002df6e1bef118d91e8335a07a7e31","subject":"Merge \"Remove LB_TIMEOUT in favor of lb_build_timeout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/804a54584e002df6e1bef118d91e8335a07a7e31"}]}],"author":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-22 17:48:54.000000000","tz":0},"committer":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2022-01-05 17:41:13.000000000","tz":0},"subject":"Tune curl repetitions value","message":"Tune curl repetitions value\n\nFor svc and route endpoint connectivity checks there is a repetition\nparameter that can be tuned to be proportional to the expected number\nof different endpoint responses.\n\nFor each expected different backend 10 repetitions seems to be enough\nand it will save significant tests execution time.\n\nChange-Id: Id9e6389bfafc89a4d407952ee0625a932b51ef38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/5564833296540458938ad4f21ec7b65ff7f0f3ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/5564833296540458938ad4f21ec7b65ff7f0f3ed"}]},"branch":"refs/heads/master"},"6ce2822320fcdd9ebcbcaa873bea24b5e070f7e1":{"kind":"REWORK","_number":8,"created":"2022-01-06 23:07:48.000000000","uploader":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"ref":"refs/changes/54/822754/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/54/822754/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/8"}}},"commit":{"parents":[{"commit":"804a54584e002df6e1bef118d91e8335a07a7e31","subject":"Merge \"Remove LB_TIMEOUT in favor of lb_build_timeout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/804a54584e002df6e1bef118d91e8335a07a7e31"}]}],"author":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-22 17:48:54.000000000","tz":0},"committer":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2022-01-06 23:09:26.000000000","tz":0},"subject":"Tune curl repetitions value","message":"Tune curl repetitions value\n\nFor svc and route endpoint connectivity checks there is a repetition\nparameter that can be tuned to be proportional to the expected number\nof different endpoint responses.\n\nFor each expected different backend 10 repetitions seems to be enough\nand it will save significant tests execution time.\n\nChange-Id: Id9e6389bfafc89a4d407952ee0625a932b51ef38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/6ce2822320fcdd9ebcbcaa873bea24b5e070f7e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/6ce2822320fcdd9ebcbcaa873bea24b5e070f7e1"}]},"branch":"refs/heads/master"},"30ed4ccd0febe1bdcc83fafb35f56b674d9124a3":{"kind":"REWORK","_number":9,"created":"2022-01-07 02:16:08.000000000","uploader":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"ref":"refs/changes/54/822754/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/54/822754/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/9"}}},"commit":{"parents":[{"commit":"804a54584e002df6e1bef118d91e8335a07a7e31","subject":"Merge \"Remove LB_TIMEOUT in favor of lb_build_timeout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/804a54584e002df6e1bef118d91e8335a07a7e31"}]}],"author":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-22 17:48:54.000000000","tz":0},"committer":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2022-01-07 02:17:27.000000000","tz":0},"subject":"Tune curl repetitions value","message":"Tune curl repetitions value\n\nFor svc and route endpoint connectivity checks there is a repetition\nparameter that can be tuned to be proportional to the expected number\nof different endpoint responses.\n\nFor each expected different backend 10 repetitions seems to be enough\nand it will save significant tests execution time.\n\nChange-Id: Id9e6389bfafc89a4d407952ee0625a932b51ef38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/30ed4ccd0febe1bdcc83fafb35f56b674d9124a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/30ed4ccd0febe1bdcc83fafb35f56b674d9124a3"}]},"branch":"refs/heads/master"},"2023af8e657a4c9654690326a884cae2c8008e20":{"kind":"REWORK","_number":10,"created":"2022-01-07 12:50:29.000000000","uploader":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"ref":"refs/changes/54/822754/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/54/822754/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/10"}}},"commit":{"parents":[{"commit":"804a54584e002df6e1bef118d91e8335a07a7e31","subject":"Merge \"Remove LB_TIMEOUT in favor of lb_build_timeout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/804a54584e002df6e1bef118d91e8335a07a7e31"}]}],"author":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-22 17:48:54.000000000","tz":0},"committer":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2022-01-07 12:51:58.000000000","tz":0},"subject":"Tune curl repetitions value","message":"Tune curl repetitions value\n\nFor svc and route endpoint connectivity checks there is a repetition\nparameter that can be tuned to be proportional to the expected number\nof different endpoint responses.\n\nFor each expected different backend 10 repetitions seems to be enough\nand it will save significant tests execution time.\n\nChange-Id: Id9e6389bfafc89a4d407952ee0625a932b51ef38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2023af8e657a4c9654690326a884cae2c8008e20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2023af8e657a4c9654690326a884cae2c8008e20"}]},"branch":"refs/heads/master"},"3408ab72d7ef0aff0d4f0d9a041e6ae260f2ad6d":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2022-01-07 12:50:52.000000000","uploader":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"ref":"refs/changes/54/822754/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/54/822754/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/54/822754/11"}}},"commit":{"parents":[{"commit":"46ddd75cf7d32987f08cbb9de2ef0f7b56142337","subject":"Ensure lb is active before checking connectivity","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/46ddd75cf7d32987f08cbb9de2ef0f7b56142337"}]}],"author":{"name":"Jon Uriarte","email":"juriarte@redhat.com","date":"2021-12-22 17:48:54.000000000","tz":0},"committer":{"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","date":"2022-01-07 12:50:52.000000000","tz":0},"subject":"Tune curl repetitions value","message":"Tune curl repetitions value\n\nFor svc and route endpoint connectivity checks there is a repetition\nparameter that can be tuned to be proportional to the expected number\nof different endpoint responses.\n\nFor each expected different backend 10 repetitions seems to be enough\nand it will save significant tests execution time.\n\nChange-Id: Id9e6389bfafc89a4d407952ee0625a932b51ef38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/3408ab72d7ef0aff0d4f0d9a041e6ae260f2ad6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/3408ab72d7ef0aff0d4f0d9a041e6ae260f2ad6d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
