)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~822997","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~I2af9e1d2be84ace03ccb7224a0c9357500123a20","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"port-pool","attention_set":{},"removed_from_attention_set":{"11600":{"account":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"last_update":"2022-01-12 16:47:53.000000000","reason":"Michał Dulko replied on the change"},"4727":{"account":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"last_update":"2022-01-17 14:09:43.000000000","reason":"Change was submitted"},"27032":{"account":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"last_update":"2022-01-17 14:09:43.000000000","reason":"Change was submitted"},"13692":{"account":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"last_update":"2022-01-10 06:50:57.000000000","reason":"Roman Dobosz replied on the change"}},"hashtags":[],"change_id":"I2af9e1d2be84ace03ccb7224a0c9357500123a20","subject":"Fix test_port_pool when namespace is created","status":"MERGED","created":"2021-12-27 07:50:54.000000000","updated":"2022-01-17 14:13:03.000000000","submitted":"2022-01-17 14:09:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"822997-port-pool","meta_rev_id":"2c3d21a22c509f15668c2012c299f0a062cd64d1","_number":822997,"virtual_id_number":822997,"owner":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-17 14:09:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2022-01-17 12:26:50.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":2,"date":"2022-01-04 12:43:18.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":2,"date":"2022-01-10 06:50:57.000000000","_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":1,"date":"2022-01-17 12:26:50.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-27 10:30:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-03 07:33:39.000000000","updated_by":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"},{"updated":"2022-01-03 07:33:39.000000000","updated_by":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"},{"updated":"2022-01-03 07:33:39.000000000","updated_by":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"}],"messages":[{"id":"607c4f77a50b4661bf6154c21a37f11f818046e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2021-12-27 07:50:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a85d0edc200216208d45a0bc588330bc4ba6fc45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-27 10:30:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28333d3668fd43ad83c382087b1839a3 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b7484d249e34a32a1ab8bf0470926c9 : SUCCESS in 2m 49s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/df52f19a6c394aa0967834537aa4d7f0 : SUCCESS in 1h 44m 20s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/e37e2f3c6bd84ae8a0d39e6ac32ec50c : SUCCESS in 1h 45m 27s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/6d7f92dbb5d74fc281b9c7faf92809bc : SUCCESS in 1h 29m 59s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/28089aa1a4814f729bac81a31b408389 : FAILURE in 27m 44s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/f5c80c62a3a644de9719af7fa14b69b5 : SUCCESS in 1h 40m 32s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/caf660e0cb7149d5953fb8da95a25536 : SUCCESS in 1h 16m 03s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/f2f3f301f2b144f0b1c999e4c72ca2e5 : SUCCESS in 1h 20m 26s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/5c29ffb3e8e94d9fb08ec31e1a886c16 : SUCCESS in 2h 04m 15s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/c23e23f4b3194109937552e089ba6f0d : SUCCESS in 1h 36m 24s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/50e2f9fbbd564182999c27232a206631 : FAILURE in 2h 34m 50s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/90025d6d1f25400d8b4091ac8c77a7cc : SUCCESS in 2h 02m 58s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/18f2112aab184a828999e236944ad0c3 : POST_FAILURE in 54m 18s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/5f4f5fcd4861457a95dbcdb7d85c623b : SUCCESS in 1h 47m 41s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/69c2c1236e374c45890bec39c31d1460 : SUCCESS in 1h 09m 16s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/a644ae03abec41fda6fa30d6402bafc8 : SUCCESS in 1h 09m 02s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/c606c3d0f20d4270a373f8e7f94f63ac : SUCCESS in 1h 16m 24s","accounts_in_message":[],"_revision_number":1},{"id":"a1413481b31887cbdb7ec404df14698a879bb141","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2022-01-03 07:33:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2ec373d8bc9b20070f5dc4a66973c477f8bacf44","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2022-01-03 07:33:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c143b03be5e06d07be576910b364071370bab721","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 09:24:26.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\n\n- kuryr-kubernetes-tempest-l2 https://zuul.opendev.org/t/openstack/build/5d648fc7d9e04782beb3f81dd3d6e635 : SUCCESS in 1h 43m 43s\n- kuryr-kubernetes-tempest-pools-namespace https://zuul.opendev.org/t/openstack/build/0f887681dfad4265b1676f9a9fdf645d : FAILURE in 1h 36m 27s\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/ac7c541f290b436e87374b03e36fe973 : SUCCESS in 1h 15m 56s (non-voting)\n- kuryr-kubernetes-tempest-dual-stack https://zuul.opendev.org/t/openstack/build/e613bdc21c0d4a6c8f2515af514c0d5e : SUCCESS in 1h 39m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2bd50967ff2ee45af0d903aa0953b1e517b39404","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 10:02:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98707c9a1c00407d8079237318da33a6 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3930288040dd42dc9174d35b39269557 : SUCCESS in 3m 19s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/7989d34ddb3c497b813923d1313ff7f3 : SUCCESS in 1h 54m 33s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/f69c98a64bf9464193aa9b3a3dc297c7 : SUCCESS in 1h 55m 17s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/2b64c783aa69459f93f27de40347af58 : SUCCESS in 1h 50m 34s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/c53f17803d1945a8998058ae762b1ef6 : SUCCESS in 1h 37m 28s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/07f82b67de24431bb9a6a8a84be8736f : SUCCESS in 1h 46m 46s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/da30166cb2704793959ea1d3af35822a : SUCCESS in 1h 20m 10s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/484c223541584b97b2ad806cfa8f9915 : SUCCESS in 1h 10m 46s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/826ac0e39a334a2bb9b8cd82def43909 : SUCCESS in 1h 36m 25s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/61d235c7c7974cfdafcc5ba1d1c3ce7c : SUCCESS in 1h 22m 07s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/83ac54ff7a7e46879a92a16edfa02e5c : FAILURE in 2h 26m 57s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/d3c6b0e45df548c2b51bf04ea14bcc92 : FAILURE in 1h 59m 56s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/a4ae2426f65240199feda2234b2fcea7 : POST_FAILURE in 51m 32s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/aedca9276bf949cdab5c7b84c450f86a : SUCCESS in 1h 42m 36s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/58d782e9f9134fe1b30bf1174dbec5f2 : SUCCESS in 1h 24m 35s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/7e5054df6aa04b849c305781b49f9739 : SUCCESS in 1h 34m 07s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/0e074f8d80e74122bf308932d530536f : SUCCESS in 1h 03m 36s","accounts_in_message":[],"_revision_number":1},{"id":"a908f6a6c43369ac3d881026fcb0787e4871b1fc","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2022-01-03 19:02:28.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"9f65ec842991ee133658d1fd8818567220965a02","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 20:40:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-l2 https://zuul.opendev.org/t/openstack/build/8220e3fff3944f808b61128858926c9e : SUCCESS in 1h 27m 12s\n- kuryr-kubernetes-tempest-pools-namespace https://zuul.opendev.org/t/openstack/build/129f23252b374ad6a427f6b0dba7828b : SUCCESS in 1h 34m 01s\n- kuryr-kubernetes-tempest-multinode-ha https://zuul.opendev.org/t/openstack/build/4676f29193ee4721a27c55e284b7d03b : SUCCESS in 1h 07m 32s (non-voting)\n- kuryr-kubernetes-tempest-dual-stack https://zuul.opendev.org/t/openstack/build/fb83c30467474c149e4a08952fce6415 : SUCCESS in 1h 33m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"18c50d3b50b3a7f5cdb2d0aa3c3c20e1a64e8351","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2022-01-04 12:43:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"77de103f5cd542f8130ec4cdcb61c4417889fb06","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2022-01-10 06:50:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"689204064a69ffd45bede90fcf7b61aae7fd04a6","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-01-10 08:44:44.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d4e7eacd8f65c006d82665a92e1558e6bcad9eeb","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2022-01-10 11:03:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a4bd3e8603eff9f11231781498832254fb099702","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-01-12 16:47:53.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8ac2f1371b0716cfc3a7dcb92f938c434ff91631","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-01-17 12:26:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5e59958e949b50fc81bb779d5f09aa271d92bf6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 12:27:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fec2ec47ce1ef64eb6bc521ad64772e682012396","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 14:09:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4f6af6afc474ce6b8af805a7f37005b : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4a05fe6858740169a885c5d91a56609 : SUCCESS in 4m 27s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/3e7c311cab054231ac9562301e1c65d2 : SUCCESS in 1h 40m 14s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/928df03713d54bf9b617a372579057f1 : SUCCESS in 1h 23m 05s","accounts_in_message":[],"_revision_number":1},{"id":"34ba608ca4c739b761477b3d2e2b80e5538f9433","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 14:09:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2c3d21a22c509f15668c2012c299f0a062cd64d1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 14:13:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbcc947fd73a41ef8596c85f3e3ea282 : SUCCESS in 1m 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"97af3a8d2e31c2bdab20bd685cd3ed9b988abaa6","revisions":{"97af3a8d2e31c2bdab20bd685cd3ed9b988abaa6":{"kind":"REWORK","_number":1,"created":"2021-12-27 07:50:54.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/97/822997/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/97/822997/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/97/822997/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/97/822997/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/97/822997/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/97/822997/1"}}},"commit":{"parents":[{"commit":"13e740911c448105f9748527e3f6c21265640fd7","subject":"Merge \"Change get_readiness_state() return type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/13e740911c448105f9748527e3f6c21265640fd7"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2021-12-27 07:44:53.000000000","tz":120},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2021-12-27 07:50:31.000000000","tz":120},"subject":"Fix test_port_pool when namespace is created","message":"Fix test_port_pool when namespace is created\n\nA pod is needed to be created also for the case where trigger_namespace_upon_pod\u003dFalse\n\nChange-Id: I2af9e1d2be84ace03ccb7224a0c9357500123a20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/97af3a8d2e31c2bdab20bd685cd3ed9b988abaa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/97af3a8d2e31c2bdab20bd685cd3ed9b988abaa6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
