)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~823792","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~I935cc510d8d96ca031f30c4a3dbdfec6a6238e92","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"ensure-lb-active","attention_set":{},"removed_from_attention_set":{"11600":{"account":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"last_update":"2022-01-12 16:49:06.000000000","reason":"Michał Dulko replied on the change"},"27032":{"account":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"last_update":"2022-01-13 08:02:13.000000000","reason":"Change was submitted"},"13692":{"account":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"last_update":"2022-01-13 06:31:06.000000000","reason":"Roman Dobosz replied on the change"},"23567":{"account":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"last_update":"2022-01-13 08:02:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I935cc510d8d96ca031f30c4a3dbdfec6a6238e92","subject":"Ensure lb is active before checking connectivity","status":"MERGED","created":"2022-01-07 12:37:09.000000000","updated":"2022-01-13 08:03:41.000000000","submitted":"2022-01-13 08:02:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823792-ensure-lb-active","meta_rev_id":"76488dbe45f522d8640b5b3f0c17100b0e081c13","_number":823792,"virtual_id_number":823792,"owner":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-13 08:02:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2022-01-12 16:49:06.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":1,"date":"2022-01-09 10:09:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":2,"date":"2022-01-13 06:31:06.000000000","_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-13 07:25:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":1,"date":"2022-01-07 15:56:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":1,"date":"2022-01-13 06:31:06.000000000","_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"reviewers":{"REVIEWER":[{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-07 15:16:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-07 15:56:28.000000000","updated_by":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"reviewer":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"state":"REVIEWER"},{"updated":"2022-01-09 10:09:40.000000000","updated_by":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"reviewer":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"state":"REVIEWER"},{"updated":"2022-01-11 12:04:55.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2022-01-11 12:04:55.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"},{"updated":"2022-01-11 12:04:55.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2022-01-13 07:25:25.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"6e52087f5073b35aea54ec7e8c3f93c85ef9f2ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2022-01-07 12:37:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8db1adabc4205b5aab18b9338c78be226e6d4288","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 15:16:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e32bba5ad2554c82ad2ffc2d5e867d53 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e3f4067b24d49d59dfcef4e6b4e731f : SUCCESS in 3m 17s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/3337f96b1eb247d0be419a1a2cd7f454 : SUCCESS in 2h 07m 55s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/6c98ab11d68e496583e01465fdd689bd : SUCCESS in 1h 36m 24s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/551e6a9a1a4c43fb89c12893a46095d4 : SUCCESS in 1h 50m 34s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/12ad5b9b140a4b68890e6acc7d7f0f9d : SUCCESS in 1h 37m 21s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/9ec65a9620ea4918805386f05e0d6c7d : SUCCESS in 1h 44m 18s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/3f0a28498bc2472f8728bdfe763ffab3 : SUCCESS in 1h 33m 21s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/a5dfd610ffc94a169baa488ea415d5bd : SUCCESS in 1h 18m 12s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/5f431ed5f6cb486e9d0a1bcfb3123614 : SUCCESS in 1h 35m 46s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/c7c12e4066014a438e606f0fe53ef8bc : SUCCESS in 2h 05m 29s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/4abf20369778412286851d7242d244ef : SUCCESS in 1h 25m 55s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/613493cf8fe242cbbf1a3468c57ccfc6 : SUCCESS in 2h 36m 29s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/7154aeaa47e14d25bfe0d47ac2b37056 : POST_FAILURE in 55m 33s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/c3088274c8904a0bbdfb9782c50be053 : SUCCESS in 1h 39m 17s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/42981b4b3360474681dd35547468f39f : SUCCESS in 1h 17m 17s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/41a95cf976e44c0b85fcffb83e5672ad : SUCCESS in 1h 11m 34s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/4d0a29bc0c334532bffa2ae8faf9ba7b : SUCCESS in 1h 14m 08s","accounts_in_message":[],"_revision_number":1},{"id":"25587eb409410998fbd3ea6d31a0c6d5e68da9f8","author":{"_account_id":27086,"name":"Jon Uriarte","email":"juriarte@redhat.com","username":"juriarte"},"date":"2022-01-07 15:56:28.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"92946e0637f0bfa08c66e5594c1ba5a8d02867a0","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2022-01-09 10:09:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"947b9ff01d81718bbde73797f2ace1a38bff2150","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-01-12 16:49:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"775a48dc24de26e3d0b2b0d71218642341166aa5","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-01-12 16:49:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1186b659316e61bfbe65b8ca74a08bdbffc4a950","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2022-01-13 06:31:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5fb9cf8f6d227249fbb01b53461fbabf97a0100f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 06:31:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"72f90517a73c9d685d36288d27f2bbca04c84f30","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-01-13 07:25:25.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"dc2720aa129b78d2625cdd2431cd1f50d6b2a0c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 08:02:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa382ef0600b46848731ef1a575120a5 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bfb62c1d8ba409d9e4fac62b47a6f1b : SUCCESS in 3m 24s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/565c22351c834c15b72cab19fa668b42 : SUCCESS in 1h 25m 57s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/b1056f0afbbe48e7a8daeebe47fb6d65 : SUCCESS in 1h 30m 14s","accounts_in_message":[],"_revision_number":1},{"id":"70bcebf1c1228541d156d441903a05cff9a3f10f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 08:02:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"76488dbe45f522d8640b5b3f0c17100b0e081c13","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 08:03:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f6ad94745044160a96ae29827a1ed30 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"46ddd75cf7d32987f08cbb9de2ef0f7b56142337","revisions":{"46ddd75cf7d32987f08cbb9de2ef0f7b56142337":{"kind":"REWORK","_number":1,"created":"2022-01-07 12:37:09.000000000","uploader":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"ref":"refs/changes/92/823792/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/92/823792/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/92/823792/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/92/823792/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/92/823792/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/92/823792/1"}}},"commit":{"parents":[{"commit":"804a54584e002df6e1bef118d91e8335a07a7e31","subject":"Merge \"Remove LB_TIMEOUT in favor of lb_build_timeout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/804a54584e002df6e1bef118d91e8335a07a7e31"}]}],"author":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2022-01-07 12:36:44.000000000","tz":0},"committer":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2022-01-07 12:36:44.000000000","tz":0},"subject":"Ensure lb is active before checking connectivity","message":"Ensure lb is active before checking connectivity\n\nThis commit ensures the load-balancer is active before checking\nconnectivity to it for the scenario of test_pod_service_curl.\n\nChange-Id: I935cc510d8d96ca031f30c4a3dbdfec6a6238e92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/46ddd75cf7d32987f08cbb9de2ef0f7b56142337"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/46ddd75cf7d32987f08cbb9de2ef0f7b56142337"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
