)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~830384","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~Ifc2dfce2405429bbcae8c01f13f06d4e9fae9c8a","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"limit-pod-requests","attention_set":{},"removed_from_attention_set":{"11600":{"account":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"last_update":"2022-03-03 19:11:20.000000000","reason":"Change was submitted"},"27032":{"account":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"last_update":"2022-03-03 17:39:29.000000000","reason":"Maysa de Macedo Souza replied on the change"},"13692":{"account":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"last_update":"2022-03-03 15:07:05.000000000","reason":"Roman Dobosz replied on the change"},"23567":{"account":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"last_update":"2022-03-03 19:11:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifc2dfce2405429bbcae8c01f13f06d4e9fae9c8a","subject":"Limit number of K8s API requests","status":"MERGED","created":"2022-02-22 11:42:41.000000000","updated":"2022-03-03 19:13:06.000000000","submitted":"2022-03-03 19:11:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830384-limit-pod-requests","meta_rev_id":"41982f989ad70bd99ceab9db9987771bed7e16d6","_number":830384,"virtual_id_number":830384,"owner":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-03 19:11:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"all":[{"value":2,"date":"2022-03-03 17:39:29.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-03 15:07:05.000000000","_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"all":[{"value":1,"date":"2022-03-03 17:39:29.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"reviewers":{"REVIEWER":[{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-22 17:32:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-23 13:26:59.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2022-02-23 13:26:59.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"},{"updated":"2022-02-23 13:26:59.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"}],"messages":[{"id":"799918ae36b5557b4c31553d92ba061961ed7f36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-02-22 11:42:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22134b781af953d9c6a4d1ae522e18f07192358c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-02-22 14:22:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"22b44d140ada7ab7ebecb45d294cf36eab636dee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 17:32:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddfdd2a82cbe47e79f68c8a15681a720 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cc0f0a361ac4ce686c33ec59f2b1f94 : SUCCESS in 3m 46s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/59d88d50cc30419181095b511b16d0f9 : SUCCESS in 1h 43m 40s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/368e7e360c84429eb69ae86caea654fd : SUCCESS in 1h 33m 36s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/5668c14e61454ab09838e3af3d26830c : SUCCESS in 1h 59m 44s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/c336529833554f74855972325f83ee41 : SUCCESS in 1h 20m 01s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/fc0e258d31b14248a9bd250417fc5ead : SUCCESS in 1h 46m 19s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/767667ff03eb48e1866e7c6da51e3bd5 : SUCCESS in 1h 39m 27s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/15e18b1e83934665a965fbecc02fd0d3 : SUCCESS in 1h 29m 40s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/efa6b99692a84878a7222b506a6abbf8 : SUCCESS in 2h 12m 00s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/a6af6dc1164740efabf40346b232322c : SUCCESS in 1h 56m 44s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/bb53668b3f714c00a9e76408acf6899d : SUCCESS in 1h 54m 13s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/a877ec0241d64daa882ef13bbcfd47ae : FAILURE in 2h 02m 22s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/dbb4150464ee474faad1795e9cee44a3 : POST_FAILURE in 3h 01m 42s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/524e86362e0a443c99928cf45286b65d : FAILURE in 41m 30s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/1f92eb5d836747329adb186dcb8827d5 : SUCCESS in 1h 27m 57s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/f2567e22c6f74ee9a219bcfa6a810439 : SUCCESS in 1h 22m 30s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/184387a9ccd54af581074420d55c469e : SUCCESS in 1h 20m 58s","accounts_in_message":[],"_revision_number":2},{"id":"eda806335998a77f7174b6f8a62328bdbf8d75ff","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-02-23 13:23:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9bb31037cc5c4825c390c443d9d02007c41f8de2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 16:16:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05f4158f7c834a0d9824f433d0fc497e : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba9289beb73f452582136a635ce64209 : SUCCESS in 3m 30s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/74c22d7ea605403791af3232470cdc3d : SUCCESS in 1h 30m 52s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/c0709fbf4d6f4ef697aa98f88322a59f : SUCCESS in 1h 35m 56s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/259cdad8380540f2a441352703aa48e9 : SUCCESS in 2h 04m 21s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/220f168500f346148b65bf5023f687e1 : SUCCESS in 1h 21m 13s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/d6f2e0d0852048af82b399da6fb2b7f6 : SUCCESS in 1h 29m 47s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/8b46a861817d438b976a959ded198c97 : SUCCESS in 1h 24m 42s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/a54083bc6a20479888c4f2b889e357ba : SUCCESS in 1h 14m 48s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/84f589ccd2cf4e36ad5460becae18a17 : FAILURE in 1h 45m 49s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/7144bbb017f9444f88995014ca428352 : SUCCESS in 1h 34m 41s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/e638bfa273fc4ccabeea2528f22a09ee : FAILURE in 2h 11m 42s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/726a33922c04462aa7a29615024f12cc : FAILURE in 2h 11m 05s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/4624d1ff9e4e4f629a9a7f6d175f9f2c : POST_FAILURE in 2h 47m 43s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/2da86af5f3484dfa8df38620e1c46021 : FAILURE in 35m 03s (non-voting)\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/da90590eacd74db5adfc203acd7a1e6a : SUCCESS in 1h 20m 28s (non-voting)\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/e364ef4de0744ed6ba581ab12c50bd27 : SUCCESS in 1h 27m 31s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/28a272afd0b34df8a0c63b7afddc6702 : SUCCESS in 1h 03m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5b24d552e8babebd13fdddda751abd35860b3964","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-02-24 10:04:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"09e6492ca22053a99c1f3af49a9d300531133950","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2022-02-24 10:17:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c6b210cdf6bdac9dc006622ecb8a0994a12307ce","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-02-24 10:24:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"414e170689b6f62446553472071b2e2bb2d4d897","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-03-01 12:18:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a8a0f55e5dd55640ddfc1fd3ef4edc65d40a2059","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 15:01:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d378f2c0e46a457684061f903b8e328f : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d6c43fa69d64cb08438784ea4feb85e : SUCCESS in 4m 44s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/e647e4ac22e942a69b7d32fb1973348c : SUCCESS in 1h 24m 00s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/fdf85393ef014b3e90003a039ce8120c : SUCCESS in 1h 19m 39s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/7e9cc4e9b4594ed5a0dcdb2f180def8e : SUCCESS in 1h 43m 35s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/c803c737d6f2426cb0fb9acf0bb628ce : SUCCESS in 1h 24m 28s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/5624a0e38e4b41a283d58743fa02ebf9 : SUCCESS in 1h 22m 46s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/b876d76b17ba4e0dbd432ad5e09c8c1d : SUCCESS in 1h 20m 41s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/e22625d547b34688a427c200303d67da : SUCCESS in 1h 03m 00s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/7349015023004544a539c8c75f2c5e5c : FAILURE in 1h 58m 11s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/2800a2b9621c4056a883669c8a465c2d : SUCCESS in 1h 39m 52s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/eceb5eca0e1a4f0fb88206dd7d2416ed : FAILURE in 2h 21m 15s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/b540455f03074e08833776d448effdb6 : SUCCESS in 1h 31m 55s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/27d0d0611d3647cc803318e7f604ffa0 : SUCCESS in 2h 40m 20s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/2fb38e1c5bc441ff8cd2d34240c25b1f : FAILURE in 27m 52s (non-voting)\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/10c2003e48cd4b79a6f961043ec17657 : SUCCESS in 1h 05m 24s (non-voting)\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/ecd6005701724cf6a8c0d45aeb78d265 : SUCCESS in 1h 06m 52s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/d70a6bd55d1e444b918c4e493e8590ab : SUCCESS in 56m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f3ab4530fe18f84b68ba18f84462ce22dfb68896","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2022-03-03 15:07:05.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"aafcf08be5c8b0fe51ada79189930f214b9f3a3f","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2022-03-03 17:39:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"bd43cadb6bdc123537afaf19f916d7ac24d1b148","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 17:40:31.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7e1b9aa87f9d5e14f59579d5d89a698f70dd7163","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 19:11:20.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f8f21588fba45e381acfa2cca9ffe77 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8dbe20dd084047ef910ca65760dea759 : SUCCESS in 6m 17s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/d27fa3eac6f3491e87c35a24c138533c : SUCCESS in 1h 25m 45s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/c0896dc85bc14305950146c297e9eaaf : SUCCESS in 1h 29m 38s","accounts_in_message":[],"_revision_number":3},{"id":"2bf2c0c8040f77498ae655286d5d9c50da9d39b9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 19:11:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"41982f989ad70bd99ceab9db9987771bed7e16d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 19:13:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bf61e285959486f981036b2f2a49002 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"86423cc26cd44234948b0a544a51df5aa835289b","revisions":{"849fb2698bc3e1b2926e12044b6fe63bf6c482f0":{"kind":"REWORK","_number":1,"created":"2022-02-22 11:42:41.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/84/830384/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/84/830384/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/84/830384/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/84/830384/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/84/830384/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/84/830384/1"}}},"commit":{"parents":[{"commit":"68b3f43b258cabdd24da536d45b89e801e3d2b95","subject":"Amend NP CRD pod selector check loop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/68b3f43b258cabdd24da536d45b89e801e3d2b95"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2022-02-22 11:39:46.000000000","tz":60},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2022-02-22 11:39:46.000000000","tz":60},"subject":"Limit number of pod requests","message":"Limit number of pod requests\n\nAs the K8s client is logging every response to every request, we\u0027re\nstarting to hit limits of tox\u0027 or Zuul\u0027s log side limit. This commit\nattempts to limit number of requests by making sure that in case of\nstatus checks we\u0027re iterating a list of pods instead of calling API for\nevery pod.\n\nChange-Id: Ifc2dfce2405429bbcae8c01f13f06d4e9fae9c8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/849fb2698bc3e1b2926e12044b6fe63bf6c482f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/849fb2698bc3e1b2926e12044b6fe63bf6c482f0"}]},"branch":"refs/heads/master"},"92901baa286352a637d125d5a7acd0b844667050":{"kind":"REWORK","_number":2,"created":"2022-02-22 14:22:50.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/84/830384/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/84/830384/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/84/830384/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/84/830384/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/84/830384/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/84/830384/2"}}},"commit":{"parents":[{"commit":"68b3f43b258cabdd24da536d45b89e801e3d2b95","subject":"Amend NP CRD pod selector check loop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/68b3f43b258cabdd24da536d45b89e801e3d2b95"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2022-02-22 11:39:46.000000000","tz":60},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2022-02-22 14:22:43.000000000","tz":60},"subject":"Limit number of pod requests","message":"Limit number of pod requests\n\nAs the K8s client is logging every response to every request, we\u0027re\nstarting to hit limits of tox\u0027 or Zuul\u0027s log side limit. This commit\nattempts to limit number of requests by making sure that in case of\nstatus checks we\u0027re iterating a list of pods instead of calling API for\nevery pod.\n\nChange-Id: Ifc2dfce2405429bbcae8c01f13f06d4e9fae9c8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/92901baa286352a637d125d5a7acd0b844667050"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/92901baa286352a637d125d5a7acd0b844667050"}]},"branch":"refs/heads/master"},"86423cc26cd44234948b0a544a51df5aa835289b":{"kind":"REWORK","_number":3,"created":"2022-03-01 12:18:37.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/84/830384/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/84/830384/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/84/830384/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/84/830384/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/84/830384/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/84/830384/3"}}},"commit":{"parents":[{"commit":"85807e55a997012511267726052b01a8238dc4bc","subject":"Merge \"Improve test_egress_np_to_service_wo_selectors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/85807e55a997012511267726052b01a8238dc4bc"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2022-02-22 11:39:46.000000000","tz":60},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2022-03-01 12:17:52.000000000","tz":60},"subject":"Limit number of K8s API requests","message":"Limit number of K8s API requests\n\nAs the K8s client is logging every response to every request, we\u0027re\nstarting to hit limits of tox\u0027 or Zuul\u0027s log side limit. This commit\nattempts to limit number of requests by making sure that in case of\nstatus checks we\u0027re iterating a list of pods instead of calling API for\nevery pod. Moreover many `time.sleep()` occurences are modified to\nreduce the number of the requests made.\n\nChange-Id: Ifc2dfce2405429bbcae8c01f13f06d4e9fae9c8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/86423cc26cd44234948b0a544a51df5aa835289b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/86423cc26cd44234948b0a544a51df5aa835289b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
