)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~830435","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~I04c3906225877caafe26487696991e9e3dbaf082","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"fix-np-test","attention_set":{},"removed_from_attention_set":{"11600":{"account":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"last_update":"2022-02-24 09:55:55.000000000","reason":"Michał Dulko replied on the change"}},"hashtags":[],"change_id":"I04c3906225877caafe26487696991e9e3dbaf082","subject":"Improve test_egress_np_to_service_wo_selectors","status":"MERGED","created":"2022-02-22 14:33:25.000000000","updated":"2022-02-24 11:41:57.000000000","submitted":"2022-02-24 11:40:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830435-fix-np-test","meta_rev_id":"67cd8aaf71eb4cb636658c7684129106522474ba","_number":830435,"virtual_id_number":830435,"owner":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-24 11:40:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"all":[{"value":2,"date":"2022-02-24 09:55:32.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":2,"date":"2022-02-24 07:33:31.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"all":[{"value":1,"date":"2022-02-24 09:55:32.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-22 14:39:18.000000000","updated_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"},{"updated":"2022-02-22 17:20:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-23 07:16:04.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2022-02-24 09:55:32.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"}],"messages":[{"id":"3464f7ff83ac9b4e7eea8f891c8f5f35e626c2da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-02-22 14:33:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1042e0b7484a8945ba028a023e5273a8fa93fb37","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2022-02-22 14:39:18.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9dd0d9f189a5b4fad256e73d4e2367990cb2e521","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 17:20:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b9ead5cffa0430bbab5ef1d24345094 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c68e16d278de4b7581322c2867d0bc92 : SUCCESS in 3m 51s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/abf5a4fe658e43d99c989dec4cfc15dd : SUCCESS in 1h 29m 05s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/65c9ab226ddd4722b72561072182df36 : SUCCESS in 1h 27m 29s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/ed702cc2b0f14cf6a2ff25f7c0989d9b : SUCCESS in 1h 47m 19s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/636ea2529edb4ab18e85b37aee79668e : SUCCESS in 1h 23m 51s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/8e3e8fcf6cb044209c2adc24c7b54d8b : SUCCESS in 1h 40m 40s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/01654b034a324c0bb0b0b7aa537ba43e : SUCCESS in 1h 45m 13s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/3fc9c4c46f684bda8a563a95e534ba41 : SUCCESS in 1h 16m 04s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/cd49bf5b834348429b45bb35cf7896ce : SUCCESS in 1h 52m 00s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/220e7d9d61f9490ca2045708ab8c3303 : SUCCESS in 1h 37m 06s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/19783f0575f7409092e68141b12969ef : SUCCESS in 2h 07m 11s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/c0c50ccd495d49e6a2787106d25876c1 : FAILURE in 2h 12m 16s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/38ac33fd92f94d538cce8f84120095ec : POST_FAILURE in 2h 41m 20s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/00e2c2fb62a448eb986b8aeba1327129 : FAILURE in 32m 08s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/d728f1a15f6d485c8ef1c43cf7ff8ace : SUCCESS in 1h 10m 33s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/c4fefc698c9f4a94a18539b33a61d6ff : SUCCESS in 1h 19m 52s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/3fda4fb3148f447bb32fbad7139524db : SUCCESS in 1h 10m 12s","accounts_in_message":[],"_revision_number":1},{"id":"e5de3c3cc8c8a85686d571de6fccc3e1432e2c01","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2022-02-23 07:16:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f250c30ed91e99b5594bda5e7fd84c1184eb1c37","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-02-23 13:23:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"983f8ad3f14d5c970e71b06b8c3895ab1abd14e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-02-23 13:25:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"93cde77d1342c136edc06ba7bebbeca96ba79cf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 16:34:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1823bfda71c9403aafdbfc886a4f41ca : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6ed777160e54eac875e53655fcd3ebe : SUCCESS in 3m 31s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/bfe9a951166b43f093f623aac995ffe7 : SUCCESS in 1h 30m 06s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/212d0c76f1f14b649d19ee37c8a0adfc : SUCCESS in 1h 25m 49s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/44ca455ec27b44cf93f99333737b021f : SUCCESS in 2h 05m 35s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/1925606ac492449d943be68a2ba59a26 : SUCCESS in 1h 24m 27s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/950e489c828449cb83a2b43babb96fed : SUCCESS in 1h 24m 12s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/a8de0c522237438da5d2d0d04bbca635 : SUCCESS in 1h 47m 12s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/d9571a19a2b1463082ee84a41e88a96c : SUCCESS in 1h 00m 51s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/347b8e82194f4240bd06dfa5bd9d969e : SUCCESS in 1h 17m 12s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/84385be9e2f84b449c6f4db7091aa0bf : SUCCESS in 1h 31m 54s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/07d36ad4ba8749b892221ed451a9bdf3 : SUCCESS in 1h 36m 50s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/7bc5ff4a5fa34ae8b5ce1336ac55077e : SUCCESS in 2h 10m 49s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/74ca2f92a05242fdb4bb5aaec80c3787 : POST_FAILURE in 3h 01m 44s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/5e9971c94544488bb15b51e472ab044a : FAILURE in 21m 54s (non-voting)\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/edb4e3899d89423dbd7fe39b09978e94 : SUCCESS in 56m 10s (non-voting)\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/5b854ac9421f42ba82ece638ec7190c7 : SUCCESS in 1h 18m 29s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/f2d4c64a1ac340d2b2bc84f3e7314216 : SUCCESS in 1h 14m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b775a24bad69d77c2a3e6edb8e077202cee9787d","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2022-02-24 07:33:31.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"18b97f54dce7f4334688f41621ab4de8a88a21b9","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2022-02-24 09:55:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2f6b3434994c622c84e359a391a9c47c7204a4db","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-02-24 09:55:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"116786fe5051477d2b25622cea1db812e46d682b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 09:56:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cab8a7dcda4a7aa15df2d3adc345d22ca2eee0c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 11:40:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25adc1d8ba9d4429b6b182f36908b4f1 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/250f8e88206e49e1aba114d92403a94b : SUCCESS in 3m 17s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/5d21fb5538bc4bb68c68815c65e633f3 : SUCCESS in 1h 42m 49s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/605940eff28b4765b62ece16d77e8e9a : SUCCESS in 1h 20m 55s","accounts_in_message":[],"_revision_number":2},{"id":"18a469f02079dd51149120443eccd210cbb4655d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 11:40:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"67cd8aaf71eb4cb636658c7684129106522474ba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 11:41:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27fc2adeb6ce4b368f6ecedf5e91ce4a : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5368b60c5f15d3cfe702be761858a2b24c1e2c63","revisions":{"37f9da48977dfd21ce1e061865ec2d5fb8ff3651":{"kind":"REWORK","_number":1,"created":"2022-02-22 14:33:25.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/35/830435/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/35/830435/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/35/830435/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/35/830435/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/35/830435/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/35/830435/1"}}},"commit":{"parents":[{"commit":"68b3f43b258cabdd24da536d45b89e801e3d2b95","subject":"Amend NP CRD pod selector check loop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/68b3f43b258cabdd24da536d45b89e801e3d2b95"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2022-02-22 14:32:08.000000000","tz":60},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2022-02-22 14:33:16.000000000","tz":60},"subject":"Improve test_egress_np_to_service_wo_selectors","message":"Improve test_egress_np_to_service_wo_selectors\n\nSeems like the test in question was only waiting for the NP to get the\npodSelector written into the status and not for the expected and\nexisting SG rules to match. This commit fixes that.\n\nChange-Id: I04c3906225877caafe26487696991e9e3dbaf082\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/37f9da48977dfd21ce1e061865ec2d5fb8ff3651"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/37f9da48977dfd21ce1e061865ec2d5fb8ff3651"}]},"branch":"refs/heads/master"},"5368b60c5f15d3cfe702be761858a2b24c1e2c63":{"kind":"REWORK","_number":2,"created":"2022-02-23 13:25:41.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/35/830435/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/35/830435/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/35/830435/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/35/830435/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/35/830435/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/35/830435/2"}}},"commit":{"parents":[{"commit":"68b3f43b258cabdd24da536d45b89e801e3d2b95","subject":"Amend NP CRD pod selector check loop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/68b3f43b258cabdd24da536d45b89e801e3d2b95"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2022-02-22 14:32:08.000000000","tz":60},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2022-02-23 13:25:34.000000000","tz":60},"subject":"Improve test_egress_np_to_service_wo_selectors","message":"Improve test_egress_np_to_service_wo_selectors\n\nSeems like the test in question was only waiting for the NP to get the\npodSelector written into the status and not for the expected and\nexisting SG rules to match. This commit fixes that.\n\nChange-Id: I04c3906225877caafe26487696991e9e3dbaf082\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/5368b60c5f15d3cfe702be761858a2b24c1e2c63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/5368b60c5f15d3cfe702be761858a2b24c1e2c63"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
