)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~830503","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~I5e8fccec80b1a5593e9e32f8dadf8bc220d0424a","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"disable-stable-gates","attention_set":{},"removed_from_attention_set":{"11600":{"account":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"last_update":"2022-02-23 11:38:51.000000000","reason":"Change was submitted"},"27032":{"account":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"last_update":"2022-02-23 09:58:41.000000000","reason":"Maysa de Macedo Souza replied on the change"},"13692":{"account":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"last_update":"2022-02-23 11:38:51.000000000","reason":"Change was submitted"},"23567":{"account":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"last_update":"2022-02-23 07:14:07.000000000","reason":"Luis Tomas Bolivar replied on the change"}},"hashtags":[],"change_id":"I5e8fccec80b1a5593e9e32f8dadf8bc220d0424a","subject":"Temporarily disable stable jobs","status":"MERGED","created":"2022-02-22 17:00:11.000000000","updated":"2022-03-08 16:42:39.000000000","submitted":"2022-02-23 11:38:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830503-disable-stable-gates","meta_rev_id":"484ed4c519ad50cc13b2957cd8341524e01c7d35","_number":830503,"virtual_id_number":830503,"owner":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-23 11:38:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"all":[{"value":2,"date":"2022-02-23 09:58:41.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-23 07:14:07.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"all":[{"value":1,"date":"2022-02-23 09:58:41.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}],"reviewers":{"REVIEWER":[{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-22 19:59:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-23 07:06:27.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"},{"updated":"2022-02-23 07:06:38.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2022-02-23 07:06:47.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"}],"messages":[{"id":"59c475ae9f52e7173a1fe3f26262f0709d5e13f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-02-22 17:00:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c46449dd30e4de1d3a0a497ac792a3fc838d24ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 19:59:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58a8505b14c64b74a34e613a4e9884bd : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6e0b40fd3984d26ac4859b245c96d6d : SUCCESS in 3m 47s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/6f8c87314e204ec6b1b6709708e97c5a : SUCCESS in 1h 31m 31s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/1210894a439649c58f5495bc1ad41df7 : SUCCESS in 1h 25m 36s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/46def317bf3f4772a86ea7022a7ff62f : SUCCESS in 1h 40m 54s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/58c122db92dd42b1bc55bfc35df6d8d3 : SUCCESS in 1h 19m 49s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/4d25f0486e8b474da175a9a508446703 : SUCCESS in 1h 27m 10s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/9d0ddc6b8cf04c8fa57c831c653e036a : SUCCESS in 1h 25m 28s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/fa7d647f943448b1b0847ccd01ebbd2d : SUCCESS in 1h 10m 10s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/62c021c170ca4d23825c9f3399a786cf : SUCCESS in 1h 36m 54s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/c3827502aa184ed3a80f9c0bff11f0e3 : SUCCESS in 1h 29m 42s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/87cabc60d3af463c80e3bb21dd238ff5 : FAILURE in 1h 53m 15s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/9109c50eb59a4ed5b1625e7e1698f110 : FAILURE in 2h 03m 05s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/de7a44be8dfe4f398d758dcc0d3e01f8 : POST_FAILURE in 2h 51m 28s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/f9a38b9614ba480e8e7e5fbb793ce7f3 : FAILURE in 35m 01s (non-voting)\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/40c88396a4fe450c9ef6f8cf9bac3a5f : SUCCESS in 1h 30m 07s (non-voting)\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/809bc10196644c1c926fac46be8282c9 : SUCCESS in 1h 35m 28s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/5d729a1f54634d58b26f8b813323e8c5 : SUCCESS in 1h 33m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"31ac28bc712b8c95deccccba0c35b3603970e6de","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2022-02-23 07:14:07.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"410dcbc68f701a78218b333c830e04daa7e97bb7","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2022-02-23 09:58:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d9dec382ad1da0e90c0e63cc13325cfe9edb9bb4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 09:59:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"58d40d0946c6f46b72b5a829c5edc01602e5b289","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 11:38:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6aac4d953044f288887cce022fb3407 : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9267696f1bf4cfd8678d67faef97dd7 : SUCCESS in 4m 51s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/dd47cf35a1294a9a8c9ff8a0eaf4e059 : SUCCESS in 1h 20m 36s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/4ba490e03cc4442088fa20e221601d3d : SUCCESS in 1h 37m 44s","accounts_in_message":[],"_revision_number":1},{"id":"a0c3c053547e06acfb014acb60cc34c00fd2a1d6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 11:38:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"98b1d295c4cb235b90652f0cef41a2949e8904ee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 11:44:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d6069021d504bd9b095bd7f1fe7777a : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1},{"id":"484ed4c519ad50cc13b2957cd8341524e01c7d35","tag":"autogenerated:gerrit:revert","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-03-08 16:42:39.000000000","message":"Created a revert of this change as I42c208812862c43d4b43b87c8331218b8ba890fd","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"06b6d31ee1dc1c266d582bbe73d596b907f40030","revisions":{"06b6d31ee1dc1c266d582bbe73d596b907f40030":{"kind":"REWORK","_number":1,"created":"2022-02-22 17:00:11.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/03/830503/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/03/830503/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/03/830503/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/03/830503/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/03/830503/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/03/830503/1"}}},"commit":{"parents":[{"commit":"68b3f43b258cabdd24da536d45b89e801e3d2b95","subject":"Amend NP CRD pod selector check loop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/68b3f43b258cabdd24da536d45b89e801e3d2b95"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2022-02-22 16:55:51.000000000","tz":60},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2022-02-22 16:55:51.000000000","tz":60},"subject":"Temporarily disable stable jobs","message":"Temporarily disable stable jobs\n\nWe have multiple CI stability issues going on at the moment.  In order\nto be able to merge master fixes, we need to disable broken stable jobs.\nOnce kuryr-tempest-plugin is stabilized we should be able to merge fixes\nto the stable jobs and then we\u0027ll be able to revert this patch and make\nthem voting again.\n\nChange-Id: I5e8fccec80b1a5593e9e32f8dadf8bc220d0424a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/06b6d31ee1dc1c266d582bbe73d596b907f40030"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/06b6d31ee1dc1c266d582bbe73d596b907f40030"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
